)]}'
{"id":"openstack%2Fswift~789129","triplet_id":"openstack%2Fswift~master~I68dabf94024957c2fb8c93050d473c6ae02f3f46","project":"openstack/swift","branch":"master","topic":"wip-lonely-frags-node-iter","hashtags":[],"change_id":"I68dabf94024957c2fb8c93050d473c6ae02f3f46","subject":"Quarantine stale EC fragments - test follow up.","status":"ABANDONED","created":"2021-05-03 04:48:57.000000000","updated":"2021-05-07 01:23:39.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"dd2ba74aa7370986cfae487e89b9a7080a39fe0d","_number":789129,"virtual_id_number":789129,"owner":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-05-03 05:28:40.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-03 05:28:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-06 09:42:38.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"CC"}],"messages":[{"id":"4c98fa6371ab7dfd673d1df3b0c31f7ab29b0dab","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2021-05-03 04:48:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1f949f75fb7acc40fe071532ff3504b1abdfc734","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-03 05:28:40.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a4f57623b32448d683c186862fa2e135 : SUCCESS in 8m 02s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a814fd99855d4d06a7fce283d4d1eee1 : SUCCESS in 8m 30s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/ca08e5263a354c4199007ec29ba1cc48 : SUCCESS in 5m 34s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/4b0785cf21564601b329d90fa207db04 : SUCCESS in 16m 07s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/9b89bf938c42444ca8d73b7cccbffecf : SUCCESS in 15m 31s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/c5800ef4976a4058b261ecd739ce6a1b : SUCCESS in 16m 44s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/d614b3eeea9f46f7acd07ac5d500523e : SUCCESS in 15m 00s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/41f993486562471aac430c2be92284c0 : SUCCESS in 15m 30s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/7d247f7ef115499e940b0e9b3d2d2817 : SUCCESS in 16m 11s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/b6d61a3f0ea347d59eae6a9fb43fa7af : SUCCESS in 18m 04s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/54ba04247bf94e02a7b8ed6027aff082 : SUCCESS in 15m 34s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/fa785f5d5de645a9b58251e3dc928e3f : SUCCESS in 17m 23s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/4b3bc1e248584acb8ae597cf3e435adc : SUCCESS in 13m 12s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/1166aaa728094821bd46df6eea22db59 : SUCCESS in 18m 25s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/5dcbfd6068824b5999dcc821277c738e : SUCCESS in 38m 17s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/50d913ab659f4b58bf390316d0217d19 : SUCCESS in 25m 53s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7aa3c27c9a4e407189ee372b3cefd897 : SUCCESS in 16m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3bf9900d8b1346fbaa1d8568dcfa08a1 : FAILURE in 7m 24s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a2bd2fa379834264ac239b30f999ff4d : FAILURE in 29m 30s","accounts_in_message":[],"_revision_number":1},{"id":"9da6fe2ad27146965fe4d0d11b3d7034465a64ac","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-05-06 09:42:38.000000000","message":"Patch Set 1:\n\nThanks Matt - squashed this in","accounts_in_message":[],"_revision_number":1},{"id":"dd2ba74aa7370986cfae487e89b9a7080a39fe0d","tag":"autogenerated:gerrit:abandon","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2021-05-07 01:23:39.000000000","message":"Abandoned\n\nsqaushed into parent patch","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0c24c641bb93b5500175f206a3a29da822680298","revisions":{"0c24c641bb93b5500175f206a3a29da822680298":{"kind":"REWORK","_number":1,"created":"2021-05-03 04:48:57.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/29/789129/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/29/789129/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/789129/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/789129/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/789129/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/29/789129/1"}}},"commit":{"parents":[{"commit":"42b4856da1e87c9a51d3f9d125fe39c724a3b362","subject":"Quarantine stale EC fragments after checking handoffs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/42b4856da1e87c9a51d3f9d125fe39c724a3b362"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2021-05-03 04:47:26.000000000","tz":600},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2021-05-03 04:47:26.000000000","tz":600},"subject":"Quarantine stale EC fragments - test follow up.","message":"Quarantine stale EC fragments - test follow up.\n\nChange-Id: I68dabf94024957c2fb8c93050d473c6ae02f3f46\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0c24c641bb93b5500175f206a3a29da822680298"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0c24c641bb93b5500175f206a3a29da822680298"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
