)]}'
{"id":"openstack%2Fswift~789471","triplet_id":"openstack%2Fswift~master~Iab41cdf086c47a0df9fe4721529ca696a82861f0","project":"openstack/swift","branch":"master","topic":"add_extra_cleave_active_states","attention_set":{"7233":{"account":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"last_update":"2022-03-14 16:23:57.000000000","reason":"Alistair Coles replied on the change"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"Iab41cdf086c47a0df9fe4721529ca696a82861f0","subject":"sharding: Add rows_cleaved to CleaveContext","status":"NEW","created":"2021-05-04 02:31:55.000000000","updated":"2022-03-14 16:23:57.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":2,"unresolved_comment_count":2,"has_review_started":true,"meta_rev_id":"a0bb2026fec22bd0feca55dd412c5bf398275a9a","_number":789471,"virtual_id_number":789471,"owner":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:check","value":1,"date":"2022-02-17 06:58:07.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":-1,"date":"2022-03-14 16:23:57.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-04 03:46:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-14 16:23:57.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"}],"messages":[{"id":"e06e6adb548dc00a3a6ac1965cde3f6b366b0a05","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2021-05-04 02:31:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cf9a657dc6e4ec25183cb298d8f3f31f4c846499","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-04 03:46:27.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1484086a34d64cf88490fb31f5ceb443 : SUCCESS in 8m 45s\n- grenade https://zuul.opendev.org/t/openstack/build/91d361280805435cb05c44694dcbd6e9 : SUCCESS in 1h 09m 11s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/660619e1aee94566be5c5dd34c382290 : SUCCESS in 45m 49s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/120c2080a3bd4a8b8a5059d59912b247 : SUCCESS in 9m 34s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/f43174c646db484aaab70113703fbe72 : SUCCESS in 7m 36s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/004882c979774f8fa443e7a9ec64ac62 : FAILURE in 17m 20s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/64ee9a48cbcc4ca1a8ab0c34178ebe63 : FAILURE in 14m 22s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/db92668f9f86404797b088400d683c26 : FAILURE in 18m 06s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/b279673268694a8ea21d704055960307 : FAILURE in 14m 12s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/86db7bc0fad9492ab0f59fee266bf986 : FAILURE in 19m 38s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/33c622cdcb0e40fe9d1d3ebf99eafd0e : SUCCESS in 15m 40s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/ea221a400ab542c7bd0ec9bf0d9f74e1 : SUCCESS in 16m 20s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/3f05f6d70d714c5da2bfaabe6874d53d : SUCCESS in 14m 05s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/dc1d865d1c7649a19f5f6a02fa6d5013 : SUCCESS in 17m 35s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/221a6d31ea4d473a931425d386342491 : SUCCESS in 18m 54s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/b7145e2c7c064539b9f8c91d6838a01e : SUCCESS in 16m 48s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d9d33d24303646d99db8b06ea1326012 : SUCCESS in 12m 27s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f259e47b1f994a589af482920f13b504 : FAILURE in 22m 34s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/3d638e1ce6be4b7cb54cccd25e9b565d : SUCCESS in 36m 27s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/a13eb271f6684e9dae93a208d37921a3 : SUCCESS in 48m 14s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4e775478e7c54d28b745825707dc2805 : SUCCESS in 38m 52s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/00c46120ec03404fafea378692796fd3 : SUCCESS in 42m 10s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3b9091108b294507adf343b7d52426b5 : FAILURE in 11m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c143c7ad153e42bf9de4e86e94dc4272 : SUCCESS in 6m 06s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/dc3440967baf4745b8fb3c3ab7e96440 : SUCCESS in 19m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/844c36895d8f48a0af5652979dab6eaf : SUCCESS in 47m 59s","accounts_in_message":[],"_revision_number":1},{"id":"de77425b63621ff34290392ca061e8dda167cb02","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2021-05-04 22:04:46.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"29adcc7ac9dded816fe3c01402e17295f0c939f6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-04 23:27:35.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6647fe83bab947708c13cdedb41c1c8a : SUCCESS in 12m 06s\n- grenade https://zuul.opendev.org/t/openstack/build/011cd1e7c65b47b4b094c808279ccb1a : SUCCESS in 43m 03s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/994d2773449c431984a72097f572b5d7 : SUCCESS in 41m 03s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/6c219c8bce9746c1b5329ca1d25720fa : SUCCESS in 7m 54s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/03fb1ebbf9ea4949b03e8a5176b2d4ce : SUCCESS in 6m 42s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/56059de0d0ca43419b2b06315d9f6868 : SUCCESS in 22m 19s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/35ee3531e9264bf0a2d6bfdb7734255e : SUCCESS in 16m 50s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/fb05ab383afd4201b8a8125c7a5dfb22 : SUCCESS in 12m 42s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/8d40096c403a4084b3163f83d2455040 : SUCCESS in 21m 29s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/d12363464826478e9e650454e430d4a3 : SUCCESS in 12m 12s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/49ba52c037a74b35b1de10335528718c : SUCCESS in 13m 25s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/3f9cc3d7312f4fafb63d49aaea625271 : SUCCESS in 14m 05s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/d866a92c438140a7af410eb618374eb2 : SUCCESS in 25m 11s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/8c5e97757eb4426cae2fbbae241341a3 : SUCCESS in 13m 46s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/792f099bde2440cb93e81e5c48fc27ff : SUCCESS in 13m 39s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/97fed334e71b49bba5146d8d85a19910 : SUCCESS in 13m 54s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/64a66fda8cee4cd6bf697b3f884b438f : SUCCESS in 6m 06s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b0f42b25b2194867ae430fe8edc4bdc6 : FAILURE in 41m 30s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/dd8da831c4874fb3b497770cf78de490 : SUCCESS in 49m 30s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/cc72a27900394214a3960a9de5a85aab : SUCCESS in 1h 19m 40s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/2c62a18e8b3049ed94539e7557ed6162 : SUCCESS in 39m 53s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/eb6920a4ffe14600b089c96d28fd51bc : SUCCESS in 1h 00m 27s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7faa09ca4c3847a980cc576e349fe1f4 : SUCCESS in 21m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/65dd26a54e41463797ad38b99d5a6454 : SUCCESS in 5m 36s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/fdf7ff114b39477090b386a38af9f3b0 : FAILURE in 13m 34s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8e32fef1d4a94dcc86f578ef90cf379f : SUCCESS in 1h 06m 23s","accounts_in_message":[],"_revision_number":2},{"id":"a7172cee9b1e376c73121f0b1f45f85c1c193399","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2021-05-11 05:40:06.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"228218778e03ec55540eaa07e5043b14c6814561","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-11 06:46:36.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/693a3cdb6e704423ba4523561475daec : SUCCESS in 9m 01s\n- grenade https://zuul.opendev.org/t/openstack/build/b405a8f89ac846b489f0b8a0fb82c14e : SUCCESS in 54m 44s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c3751ad24bba4a6daa25ff1d822f86ee : SUCCESS in 40m 21s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/cdcd27661fa645d3a8a0e98fd1019ab9 : SUCCESS in 10m 23s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/1f1fc800b8db480a8734da4c5487dfd4 : SUCCESS in 8m 09s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/64314b77d03549bdb1b21ad77c5adf7a : SUCCESS in 18m 27s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/8cc0e0fb0cf24193b7ff43d03606dfb8 : SUCCESS in 17m 29s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/62e6fc9fda5b44209f2ba15115862c3c : SUCCESS in 21m 35s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/38e647dd0d4a410786231749993e1204 : SUCCESS in 15m 03s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/acd60f548f9040f4a898b46decc36bef : SUCCESS in 14m 36s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/467dc2b5619342b0a81bc1caf30d3b2d : SUCCESS in 23m 10s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/763bc73eaeb048c2b76028249df73550 : SUCCESS in 17m 49s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/5993798ead024bebab7f730fa7195cef : SUCCESS in 17m 45s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/c5154d2b85ed42c9bbb30e50e4cca3af : SUCCESS in 23m 10s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/730b31ac6c2441a6aee1a06462952881 : SUCCESS in 26m 07s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/af6bb0798afa4927a6f8d26eaf279000 : SUCCESS in 25m 48s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/ca7abcc0df4e49059521565ea9bad064 : SUCCESS in 6m 04s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/2d4232f6919f45669f00e53cd64c830f : SUCCESS in 23m 03s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/3af864218cce474bb63f8a087594b638 : SUCCESS in 1h 01m 51s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/c167fa4b385549d4b2b90bffb2c9429e : SUCCESS in 53m 43s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a62a4bed9474445ab891e6df45e2cfc8 : SUCCESS in 41m 40s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/067ed81e48484aa89eb28b3182b89b1b : SUCCESS in 41m 08s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/43e2dae709794d6aa163110795c354a7 : SUCCESS in 15m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4477769d0fe94c6680ccac4ddaa0e65c : SUCCESS in 5m 57s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/0d7a64b5075c45fabb18e6bb085bbe6e : SUCCESS in 32m 29s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/78a4e6c37fcd41b78a13610e5e7007f9 : SUCCESS in 57m 50s","accounts_in_message":[],"_revision_number":3},{"id":"4016ffa83f84719384bb983942179e351cb7b99f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2021-05-28 05:19:00.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"95f8749889d0c5e77653b0c65ef155e33af1a9a4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-28 06:30:56.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/74109778478f49b3866885b8a775eecc : SUCCESS in 8m 54s\n- grenade https://zuul.opendev.org/t/openstack/build/41cf1ca90237415ab7e790aec88d05be : SUCCESS in 1h 07m 32s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c7a3a57f27834400b43dd2abec42fcb7 : SUCCESS in 45m 51s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b4ce8e218c29440480615970fd5c919c : SUCCESS in 9m 48s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/8c6c8f7dff3347c1aa75967fbf14644e : SUCCESS in 7m 26s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/3e8154bd7f1946ef871bd1be35c6ccf8 : SUCCESS in 16m 38s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/e6e9b8568de545ecbecc719b385f83c9 : SUCCESS in 16m 29s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/8f2afbd9bb844a089fa0fe721799c988 : SUCCESS in 15m 23s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/4a0efdb971fe4b76bce56cbc5bc985c7 : SUCCESS in 15m 46s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/a05ae73622ee4277a19587ed01f5435c : FAILURE in 16m 49s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/621b4039c2e644ebade8104f71499594 : SUCCESS in 17m 55s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/dc354a238ee643349b494008758305ca : SUCCESS in 21m 23s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/3319592ac94b469da0eca09866afc684 : SUCCESS in 17m 55s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/105c7639cb70419286380652fc2d7103 : SUCCESS in 19m 25s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/fcc75bd2fc3b412882517c3126101bcb : SUCCESS in 17m 12s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/9c9431d538a74425a78c879955b2b925 : SUCCESS in 19m 24s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/78401b165ab744e49249dd02fb8a3318 : SUCCESS in 6m 39s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/3def985c037847fea5fd6d0b256eae8a : SUCCESS in 31m 44s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/d61c51563972495fad4fa638814b6cb8 : SUCCESS in 53m 45s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/e388f459dcaf4a20be85f337faf12cc1 : SUCCESS in 50m 31s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a78b7b2208654575bb5d41cef4bbc43e : SUCCESS in 56m 11s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/034da4d4fdb64df7ab9d86fa7bdd1095 : SUCCESS in 37m 42s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/25022dd5552847ef826ad058b18e36f5 : SUCCESS in 16m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b2bfaeeae3634e3eb3aafb5f23d05d4a : SUCCESS in 5m 40s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9404c98d838b4e61a2e4fa23f1dfd527 : SUCCESS in 30m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a4a0c57d100a4b73bd2eb4cd07b7de7f : SUCCESS in 1h 04m 05s","accounts_in_message":[],"_revision_number":4},{"id":"3bf849c9ae7e476b107c35afb2d3c391dd994508","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2022-02-17 05:59:22.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"b802631f0493549ace70271fbcf50c38a61d9917","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-17 06:58:07.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/328581cc619341038ee0d155ec346991 : SUCCESS in 9m 25s\n- grenade https://zuul.opendev.org/t/openstack/build/433efb81789b479195c09235ff3d7e3a : SUCCESS in 49m 24s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/79e103f2f3c24840ad3cf30b1bf5fc99 : SUCCESS in 44m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6d7aa1af65e8400fb2f514171bb3cca0 : SUCCESS in 46m 27s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/8cb1a96a11d146c98f92db60881f8b5b : SUCCESS in 23m 20s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/29a8cdfae5f34158a7290d4c6278a8de : SUCCESS in 22m 07s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/adc7c27ae5fe481fae80901e58f8e20a : SUCCESS in 23m 40s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/27dcbb0e4f7243ed85ca3de2b44d436a : SUCCESS in 7m 49s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/30930c04ec5a4313ad385c63750f4c0e : SUCCESS in 6m 55s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/190104d8af5e440a8cd4dd42a71cc3e6 : SUCCESS in 16m 15s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/90409483584449528b9c842708ef8f1f : SUCCESS in 16m 41s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/548ec42436c746638869166461001d05 : SUCCESS in 15m 08s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/ad864bd7a6ff4da49874ae57c24237f8 : SUCCESS in 18m 44s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/b5da1f50989647b9ab83412ab98fa781 : SUCCESS in 18m 53s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/7876c76898914420a059a3558c1b415e : SUCCESS in 20m 42s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/13b8500b48f94990b220d1b9b3586b3c : SUCCESS in 16m 27s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/57b3e741f92746e3b83deb576e64b316 : SUCCESS in 17m 45s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/3a6bc0683ee44f4393b8f08540376ac4 : SUCCESS in 17m 40s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/9ec733f2647e42e18f454d4c31dde449 : SUCCESS in 7m 13s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/ed08ca726c224943b5831f8dcee8e9c4 : SUCCESS in 23m 43s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/c2bec1df99784e1092847efe89b5b1d1 : SUCCESS in 49m 12s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/0baf6cbdd9f54049b48d7f9ab977550a : SUCCESS in 52m 46s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/233ed81020e44549b20bc224a13c159b : SUCCESS in 41m 30s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/990cc9bc5cec47a58c594d3aaef9607b : SUCCESS in 40m 03s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/71f2ceb3c43a4b7bbec1b5ef0ef0c54d : SUCCESS in 17m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bdc0b6541b9b49989bce5c041d3f18f1 : SUCCESS in 5m 55s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/69fce5cb549240098b59fc9357306c04 : SUCCESS in 25m 58s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0ad225d81b4347d88694cda5ed295c3e : SUCCESS in 53m 33s","accounts_in_message":[],"_revision_number":5},{"id":"abbd267e09500cdcc3df1b54737df791eac85af5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-17 10:27:45.000000000","message":"Patch Set 5:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3924cc88aca74bf18b3299633394e18e : SUCCESS in 54m 57s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5afec175db8f40f895f6903b37c032bf : SUCCESS in 1h 02m 10s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/d35ee63b950b4176a3a03f0289c674e8 : TIMED_OUT in 3h 01m 21s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/8e3f82b7d9b8430fa84a82539cf9e1f9 : SUCCESS in 55m 50s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/d93b1516352845018f02bf699c991a5a : SUCCESS in 1h 28m 44s","accounts_in_message":[],"_revision_number":5},{"id":"a0bb2026fec22bd0feca55dd412c5bf398275a9a","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-03-14 16:23:57.000000000","message":"Patch Set 5: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"6ca79c0edef16e6ba764b860c74bb65845da102c","revisions":{"10888dc6e4db7285b719fac006aa8b296a3389ce":{"kind":"REWORK","_number":1,"created":"2021-05-04 02:31:55.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/71/789471/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/71/789471/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/789471/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/789471/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/789471/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/71/789471/1"}}},"commit":{"parents":[{"commit":"c5fe114c9622783b9d6e00ac9bf50e9ffe6dd17b","subject":"Merge \"sharder: stall cleaving at shard range gaps\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c5fe114c9622783b9d6e00ac9bf50e9ffe6dd17b"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2021-05-04 02:28:45.000000000","tz":600},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2021-05-04 02:28:45.000000000","tz":600},"subject":"sharding: Add rows_cleaved to CleaveContext","message":"sharding: Add rows_cleaved to CleaveContext\n\nThis patch adds a rows_cleaved member variable to the CleaveContexts\nthat tracks how many object rows were cleaved.\n\nThis will allow us to to make better shrinking decisions as we can tell\nif a completed context was from a small container or whatnot.\n\nChange-Id: Iab41cdf086c47a0df9fe4721529ca696a82861f0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/10888dc6e4db7285b719fac006aa8b296a3389ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/10888dc6e4db7285b719fac006aa8b296a3389ce"}]},"branch":"refs/heads/master"},"60ffc94d15631842f4a020a74843dea8ea6cb4b6":{"kind":"REWORK","_number":2,"created":"2021-05-04 22:04:46.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/71/789471/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/71/789471/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/789471/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/789471/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/789471/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/71/789471/2"}}},"commit":{"parents":[{"commit":"c5fe114c9622783b9d6e00ac9bf50e9ffe6dd17b","subject":"Merge \"sharder: stall cleaving at shard range gaps\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c5fe114c9622783b9d6e00ac9bf50e9ffe6dd17b"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2021-05-04 02:28:45.000000000","tz":600},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2021-05-04 22:04:21.000000000","tz":600},"subject":"sharding: Add rows_cleaved to CleaveContext","message":"sharding: Add rows_cleaved to CleaveContext\n\nThis patch adds a rows_cleaved member variable to the CleaveContexts\nthat tracks how many object rows were cleaved.\n\nThis will allow us to to make better shrinking decisions as we can tell\nif a completed context was from a small container or whatnot.\n\nChange-Id: Iab41cdf086c47a0df9fe4721529ca696a82861f0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/60ffc94d15631842f4a020a74843dea8ea6cb4b6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/60ffc94d15631842f4a020a74843dea8ea6cb4b6"}]},"branch":"refs/heads/master"},"ffae789bdcf9e39c0e39484c5a62004f9ce4a520":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-05-11 05:40:06.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/71/789471/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/71/789471/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/789471/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/789471/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/789471/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/71/789471/3"}}},"commit":{"parents":[{"commit":"4ce907a4ae97b150fa994237ebffd73ee8ac694d","subject":"relinker: Add /recon/relinker endpoint and drop progress stats","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4ce907a4ae97b150fa994237ebffd73ee8ac694d"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2021-05-04 02:28:45.000000000","tz":600},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2021-05-11 05:35:52.000000000","tz":600},"subject":"sharding: Add rows_cleaved to CleaveContext","message":"sharding: Add rows_cleaved to CleaveContext\n\nThis patch adds a rows_cleaved member variable to the CleaveContexts\nthat tracks how many object rows were cleaved.\n\nThis will allow us to to make better shrinking decisions as we can tell\nif a completed context was from a small container or whatnot.\n\nChange-Id: Iab41cdf086c47a0df9fe4721529ca696a82861f0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ffae789bdcf9e39c0e39484c5a62004f9ce4a520"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ffae789bdcf9e39c0e39484c5a62004f9ce4a520"}]},"branch":"refs/heads/master"},"bf741c764a092344399867a9671d0855140b7369":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2021-05-28 05:19:00.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/71/789471/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/71/789471/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/789471/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/789471/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/789471/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/71/789471/4"}}},"commit":{"parents":[{"commit":"1ba17f63540c2c34cfbb1c5880f22d7294bf0593","subject":"Get TestDarkDataQuarantining passing when policy-0 is erasure-coded","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1ba17f63540c2c34cfbb1c5880f22d7294bf0593"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2021-05-04 02:28:45.000000000","tz":600},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2021-05-28 05:18:34.000000000","tz":600},"subject":"sharding: Add rows_cleaved to CleaveContext","message":"sharding: Add rows_cleaved to CleaveContext\n\nThis patch adds a rows_cleaved member variable to the CleaveContexts\nthat tracks how many object rows were cleaved.\n\nThis will allow us to to make better shrinking decisions as we can tell\nif a completed context was from a small container or whatnot.\n\nChange-Id: Iab41cdf086c47a0df9fe4721529ca696a82861f0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bf741c764a092344399867a9671d0855140b7369"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bf741c764a092344399867a9671d0855140b7369"}]},"branch":"refs/heads/master"},"6ca79c0edef16e6ba764b860c74bb65845da102c":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2022-02-17 05:59:22.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/71/789471/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/71/789471/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/789471/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/789471/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/789471/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/71/789471/5"}}},"commit":{"parents":[{"commit":"713b944eb47f4a5f0e740fd37e9c582c4a42b462","subject":"Merge \"Proxy: override user_agent with backend_user_agent\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/713b944eb47f4a5f0e740fd37e9c582c4a42b462"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2021-05-04 02:28:45.000000000","tz":600},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2022-02-17 00:57:19.000000000","tz":660},"subject":"sharding: Add rows_cleaved to CleaveContext","message":"sharding: Add rows_cleaved to CleaveContext\n\nThis patch adds a rows_cleaved member variable to the CleaveContexts\nthat tracks how many object rows were cleaved.\n\nThis will allow us to to make better shrinking decisions as we can tell\nif a completed context was from a small container or whatnot.\n\nChange-Id: Iab41cdf086c47a0df9fe4721529ca696a82861f0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6ca79c0edef16e6ba764b860c74bb65845da102c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6ca79c0edef16e6ba764b860c74bb65845da102c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
