)]}'
{"id":"openstack%2Fswift~790200","triplet_id":"openstack%2Fswift~master~I5691bf644ade9fd7f7490f91add388a1d5a4aaa2","project":"openstack/swift","branch":"master","topic":"relinker_cli_auto","hashtags":[],"change_id":"I5691bf644ade9fd7f7490f91add388a1d5a4aaa2","subject":"relinker: let --workers accept auto","status":"MERGED","created":"2021-05-07 02:19:37.000000000","updated":"2021-05-07 16:42:31.000000000","submitted":"2021-05-07 16:41:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"790200-relinker_cli_auto","meta_rev_id":"a56ea933f149dc518a0719144041a218874a86e3","_number":790200,"virtual_id_number":790200,"owner":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-05-07 16:40:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-05-07 15:43:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-05-07 15:43:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-07 02:28:11.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2021-05-07 03:27:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"58ee3c8641f836552483a9a1fc68e632e827f392","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2021-05-07 02:19:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"027603cea8e02f0df0025e53e34e370a5764c172","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-05-07 02:28:11.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)\n\nMight want to actually have a\n\n def auto_or_int(value):\n     return config_auto_int_value(value, default\u003d\u0027auto\u0027)","accounts_in_message":[],"_revision_number":1},{"id":"0cfe8e2e35264ccce33ef1f0dbc5cec1431b25b9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-07 03:27:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/52db1b398ba64015ae5356944006a107 : SUCCESS in 10m 01s\n- grenade https://zuul.opendev.org/t/openstack/build/2ba30f06f0814e88afb35cfa70c4cc76 : SUCCESS in 1h 06m 31s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/5c4b049dd7d847c2af7ce7a7f9ee6e85 : SUCCESS in 50m 34s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a976afa183354bf2b74dfed14ebda278 : SUCCESS in 7m 56s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/cccd3d9d85f64323a91f5d6ca27ee99a : SUCCESS in 7m 23s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/43f9a389eb4b4adc97d6e730b1afecf8 : SUCCESS in 43m 07s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/588d8b8e8f9a4398885c17bf590e626c : SUCCESS in 17m 07s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/cec0814908324bf09d71fad1a369a87e : SUCCESS in 57m 27s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/d920f3d1cb384b7a92b1851cb065922f : SUCCESS in 30m 08s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/83ce468b76684d49b09b82779dfd9761 : SUCCESS in 13m 36s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/5d290d27ac584f9a9ecf4a7cdac196c7 : SUCCESS in 18m 15s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/ed2c7510cabc4b6ba88ee013cff7c0a2 : SUCCESS in 19m 22s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/40800773386944ae9a4f71f1821dae2f : SUCCESS in 19m 37s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/c7da42c0d8c446d1a5d44ae8701b42a9 : SUCCESS in 26m 23s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/ce6eac91d86f4cfdb4ba8f03af3cef7a : SUCCESS in 17m 06s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/22b1f29beb544aceae9b13d0e22da42e : SUCCESS in 27m 22s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/4c351960f7004badb4399c9c69daa98e : SUCCESS in 7m 28s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/9e2eaf30a6434eb4a5bc55b084becba1 : FAILURE in 24m 50s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/332640958d5d4a2baaf7b524e6523029 : SUCCESS in 44m 18s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/387d74c891b34d17a4c6ba5d3001a489 : SUCCESS in 54m 02s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e1c4050880ba4f649dc9f73ab9165771 : SUCCESS in 29m 35s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d8418c95b53a4f6bb4da6549e5a1194a : SUCCESS in 36m 49s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/297dc5a521c94718874c49f9e65607c0 : SUCCESS in 14m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/18e06287522646578ee03b943d3db7b1 : SUCCESS in 6m 57s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/570d220abfb74fef89992da2eee7f4c3 : SUCCESS in 23m 43s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ef12be01256143bb9215fb2fee799dad : SUCCESS in 46m 00s","accounts_in_message":[],"_revision_number":1},{"id":"14831c40868f9c8532d4389768785748724fec1c","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2021-05-07 04:35:45.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1: Code-Review+1\n\u003e \n\u003e (1 comment)\n\u003e \n\u003e Might want to actually have a\n\u003e \n\u003e  def auto_or_int(value):\n\u003e      return config_auto_int_value(value, default\u003d\u0027auto\u0027)\n\noh good idea, that looks cleaner 😊","accounts_in_message":[],"_revision_number":1},{"id":"020b916c2ef36bbc911b043cae4887c34418b80f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2021-05-07 04:51:49.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"4a4be189460e8cbf38666a7dc2eaa8ba8956bb94","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2021-05-07 05:05:55.000000000","message":"Patch Set 2:\n\nswift-object-relinker: error: argument --workers: invalid auto_or_int value: \u0027sdkf\u0027\n\nLooks better then before.\n\nLol, I could do something like:\n\n  def auto_or_int(value):\n      return config_auto_int_value(value, default\u003d\u0027auto\u0027)\n\n\n  auto_or_int.__name__ \u003d \"\u0027auto\u0027 or integer\"\n\nor\n\n  def auto_or_int(value):\n      auto_or_int.__name__ \u003d \"\u0027auto\u0027 or integer\"\n      return config_auto_int_value(value, default\u003d\u0027auto\u0027)\n\nAnd:\n\nswift-object-relinker: error: argument --workers: invalid \u0027auto\u0027 or integer value: \u0027sdkf\u0027","accounts_in_message":[],"_revision_number":2},{"id":"7bd649986f558b40a1bfe3498bbc7b3798b22666","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-07 05:53:05.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/488c5251b8a24be496c239e604740875 : SUCCESS in 8m 20s\n- grenade https://zuul.opendev.org/t/openstack/build/bfbdbc0b352e4ad289868715023215ee : SUCCESS in 1h 00m 07s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/8da8937228c347c4ba8bba8803a8889c : SUCCESS in 41m 22s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ed7bf44b299f45eead540918cdb6ca7d : SUCCESS in 9m 12s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/34cdbf36d237461099ac0b872153e963 : SUCCESS in 6m 19s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/92565b8fa7e24a6cbbdefea3834bb133 : SUCCESS in 19m 38s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/065f9b6d8c604736992bd5629a6c81f2 : SUCCESS in 15m 06s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/5772a34c688944c6b3f51a2aca534c52 : SUCCESS in 16m 01s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/34cdddc346574a8e8e5cefba16e5567b : SUCCESS in 14m 13s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/fac4869bb8694502a6b418aab3b11fca : SUCCESS in 19m 19s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/2a97d73752bd491faecf6caa7f0fe151 : SUCCESS in 19m 56s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/503b3345b4e04bd1b2923a51793b0337 : SUCCESS in 20m 45s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/12d56b504dd14d4084f358d3997ecae2 : SUCCESS in 15m 28s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/e57816d5c21642988e2af25915b69507 : SUCCESS in 23m 28s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/c1476324dad143f7870827de402908af : SUCCESS in 21m 44s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/70f3c9ce92f64be68e830750b1b187c7 : SUCCESS in 22m 47s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/4a1d285c5c1a4b8b99b6d6b7605eb75b : SUCCESS in 5m 57s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/5e523727f1ec4add8e1d9e4e3cce3804 : FAILURE in 27m 29s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/23f07e353c4140fdaa7e9e027dcfc011 : SUCCESS in 51m 28s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/75a7d303b32e42afbe0bd56b405c7732 : SUCCESS in 42m 09s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/34cfb7d73dfe4b9d94c4647a29ae1b05 : SUCCESS in 39m 51s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4cf14176f5e04fa6bde8ea77dadf124e : SUCCESS in 42m 58s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6a84b82778c249fdacc5895ec01537e4 : SUCCESS in 16m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a1a5967db4204fd5bc136c61dd490bb6 : SUCCESS in 6m 35s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b6d584d9c6c244b482f7bcf70c0ea627 : SUCCESS in 22m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/76eaaa8ae9b54dbc8f8e5e2cdb000297 : SUCCESS in 42m 40s","accounts_in_message":[],"_revision_number":2},{"id":"0bbe67bbb13da8d3f021f7d3335fc037a72ac69a","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-05-07 15:43:51.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"9fd67a7d456fb0b21d2a5b0edffd1ffbbdb2f769","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-07 15:44:17.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"14cd4652c0d95912084fca4586c8d65827c48490","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-07 16:40:59.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7fa10d3e07d249538836935eefe7e575 : SUCCESS in 8m 50s\n- grenade https://zuul.opendev.org/t/openstack/build/336fd988d5ba4c678f09a9b51c7e7589 : SUCCESS in 54m 54s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/cf7b3336dc5e498890473226a9d2515e : SUCCESS in 29m 17s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/f1a2e9800aba471eb2173fae0a3d68bd : SUCCESS in 10m 18s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/389710ea85b747c3b0852f1caa7bb2f5 : SUCCESS in 8m 00s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/2b20c0698bcf4857b67aedd5db404178 : SUCCESS in 12m 59s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/0c531b69f95b44cf8051f03b3197c13b : SUCCESS in 12m 52s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/dec38d0a288440a9b01d4705dcaaf16e : SUCCESS in 15m 55s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/48541d11ce7c47898c0f876fe2ea00c8 : SUCCESS in 18m 44s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/5d821b70bf524a25918ea0489d8249a3 : SUCCESS in 13m 26s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/1f5a52be6e8c45dd9777c620b0be3de6 : SUCCESS in 23m 39s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/db9e0498e013471d976237af619dc688 : SUCCESS in 18m 36s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/e11163038130462aab0ccd823e1b3872 : SUCCESS in 25m 15s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/19d5ed1c01084da9b21ffe492a05a9ad : SUCCESS in 18m 06s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/1bc5f508f9b7450099f3cc0e9db4ea57 : SUCCESS in 16m 01s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/d7e9845617564b29be19549120584946 : SUCCESS in 20m 28s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/65b359115bb84ef1a5dd9263b9ce43d7 : SUCCESS in 6m 33s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/0ff8b87c0a134184bd5e92eefc7caf42 : SUCCESS in 52m 58s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/845da37ff04f4e5492b835e75428752b : SUCCESS in 43m 42s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c37d7843f2ec4a2b866566c994ef5778 : SUCCESS in 41m 27s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ac5840b25e5a4956a0ed69449fb1e4e8 : SUCCESS in 40m 01s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ac307feb69bc4603a974b3cd3a45be76 : SUCCESS in 17m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/971843978dd145738efbdde59eced8f8 : SUCCESS in 6m 27s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/32007be5b92e45f0a5e7b5b2e4c2ca73 : SUCCESS in 30m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b495a55f44444e5bbae3ee4de95376c6 : SUCCESS in 45m 27s","accounts_in_message":[],"_revision_number":2},{"id":"76a490b8310d64bb2a8df7bfd79f32818ea89ca9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-07 16:41:03.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"a56ea933f149dc518a0719144041a218874a86e3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-07 16:42:31.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea8712d046b04f75b5a19a03448a079a : SUCCESS in 59s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/9b71c9f29e16402d9500fc96fa7a97c3 : FAILURE in 48s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/39baa78e58cd4e5499d8d247bbbedf9c : SUCCESS in 53s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"f140de75e4e63ae0d101a2903a737e5727e7d794","revisions":{"eae71b4e364eb8d897fd0985efd1d338a6026fc0":{"kind":"REWORK","_number":1,"created":"2021-05-07 02:19:37.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/00/790200/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/00/790200/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/790200/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/790200/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/790200/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/00/790200/1"}}},"commit":{"parents":[{"commit":"770c8037c93153f38a7cb3ea52032bee90ece250","subject":"Merge \"Remove test-requirement on fixtures\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/770c8037c93153f38a7cb3ea52032bee90ece250"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2021-05-07 02:11:25.000000000","tz":600},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2021-05-07 02:11:25.000000000","tz":600},"subject":"relinker: let --workers accept auto","message":"relinker: let --workers accept auto\n\nCurrently from the relinker cli we allow a --workers parameter but it\ncurrently _must_ be an int. However, if someone wants to override a\nstatic int worker value from the config back to the \u0027auto\u0027 this results\nin a VALUE error.\n\nThis patch changes the argparse type check for the worker param to use\nconfig_auto_int_value, so a value of \u0027auto\u0027 will be accepted.\n\nBecause the argparse type validation function expects a single param to\ncheck functools.partial is used to specify an \u0027auto\u0027 default. The\nparallel_process function is in charge of translating \u0027auto\u0027 to\nlen(devices) to this works perfectly.\n\nCloses-Bug: #1926757\nChange-Id: I5691bf644ade9fd7f7490f91add388a1d5a4aaa2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/eae71b4e364eb8d897fd0985efd1d338a6026fc0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/eae71b4e364eb8d897fd0985efd1d338a6026fc0"}]},"branch":"refs/heads/master"},"f140de75e4e63ae0d101a2903a737e5727e7d794":{"kind":"REWORK","_number":2,"created":"2021-05-07 04:51:49.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/00/790200/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/00/790200/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/790200/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/790200/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/790200/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/00/790200/2"}}},"commit":{"parents":[{"commit":"770c8037c93153f38a7cb3ea52032bee90ece250","subject":"Merge \"Remove test-requirement on fixtures\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/770c8037c93153f38a7cb3ea52032bee90ece250"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2021-05-07 02:11:25.000000000","tz":600},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2021-05-07 04:50:30.000000000","tz":600},"subject":"relinker: let --workers accept auto","message":"relinker: let --workers accept auto\n\nCurrently from the relinker cli we allow a --workers parameter but it\ncurrently _must_ be an int. However, if someone wants to override a\nstatic int worker value from the config back to the \u0027auto\u0027 this results\nin a VALUE error.\n\nThis patch changes the argparse type check for the worker param to use\nconfig_auto_int_value, via a small shim, so a value of \u0027auto\u0027 will be\naccepted.\n\nCloses-Bug: #1926757\nChange-Id: I5691bf644ade9fd7f7490f91add388a1d5a4aaa2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f140de75e4e63ae0d101a2903a737e5727e7d794"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f140de75e4e63ae0d101a2903a737e5727e7d794"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
