)]}'
{"id":"openstack%2Fswift~790859","triplet_id":"openstack%2Fswift~master~I433bad1cb3f52ad60a81dcbf3209681f7068ed55","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2022-05-27 15:59:24.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"I433bad1cb3f52ad60a81dcbf3209681f7068ed55","subject":"staticweb: Allow empty listings at the root of a container","status":"MERGED","created":"2021-05-11 22:50:30.000000000","updated":"2022-05-27 18:36:44.000000000","submitted":"2022-05-27 18:32:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"790859","meta_rev_id":"3c1e0c4ed45ae8271e02458e90519f4262da4e98","_number":790859,"virtual_id_number":790859,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-05-27 18:32:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"all":[{"value":2,"date":"2022-05-19 17:54:20.000000000","_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-05-27 15:59:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-11 23:49:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-19 17:54:20.000000000","updated_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"reviewer":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"state":"REVIEWER"}],"messages":[{"id":"6ef03019b10d894ac5a6b30af03ac0a96a4f948f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-05-11 22:50:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3a51329331fc2cb481f81fb1ab5cacd1e9f352e1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-11 23:49:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a0e3673d2a7488ebf8ba6088c4df4f9 : SUCCESS in 10m 04s\n- grenade https://zuul.opendev.org/t/openstack/build/1a243b79ef354a7a8fb305f9ba6263f2 : SUCCESS in 58m 07s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/ca0f5efb60c8419da74e374cc7a91c6a : SUCCESS in 35m 01s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e9f8439255cc4016a3f80bb640871a39 : SUCCESS in 7m 28s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/2a1f0839dab84621a25de569c9868374 : SUCCESS in 6m 51s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/4f5a07a96d424a3eb68a719ed9dcdfe5 : SUCCESS in 16m 26s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/84d1e6ebe1f84bc693e5f8ca3b5eb341 : SUCCESS in 16m 15s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/b5db433604c941339321932ec55cbaa8 : SUCCESS in 19m 43s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/b2e38eb15f6147b79b31925a5534c48f : SUCCESS in 15m 21s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/bd546d13776045b1ad6eae46b6f07ee9 : SUCCESS in 16m 53s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/0692a0b3356a4a4da429d0fd3fbf6eda : SUCCESS in 28m 00s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/9d2b2fd64c2444fc8d14e35389c91ff4 : SUCCESS in 17m 57s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/34fc3fb374854247b419fc31747b9e40 : SUCCESS in 29m 22s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/08fc15c52e0f4c77bc626a37c50a7844 : SUCCESS in 28m 33s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/8b24105b95344b3bb88f421eb1427d7e : SUCCESS in 31m 32s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/5a6d95346c394b8ea8a163fe4c768b01 : SUCCESS in 32m 50s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a99126793a504e5aa51312b8e6de4e82 : SUCCESS in 6m 47s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/7a40b6dbb749416d838aa87bceaa6bbd : FAILURE in 19m 51s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/2d87c192ab67493b983d6bc35bd7fdd0 : SUCCESS in 46m 11s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/7dfd5152b7bd49e980fcb7d5c9bd6589 : SUCCESS in 48m 14s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/9ade80b25a7b4e25b84c0903db7c4873 : SUCCESS in 32m 21s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/6cd8829f2eb94aa18dc45ff2af34b83d : SUCCESS in 33m 54s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/653678ec655e400a87a9f202e064bd72 : SUCCESS in 30m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1493c89dd1fe4257af17a0b6d1781ea7 : SUCCESS in 6m 09s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/10256231466648f787ed659277b394f4 : SUCCESS in 23m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/770615bcd2984cba8f3475714ae3cd23 : SUCCESS in 51m 04s","accounts_in_message":[],"_revision_number":1},{"id":"733d8649c8e636888bd2e1919261dc1255c7b8d1","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2022-05-19 17:54:20.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"19acd4d73b2c4c2004674fc0bc409211d4337758","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-05-27 15:59:24.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"82f9a83acb8339b64f35ee95cc48cf156e7b8be1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-27 16:01:45.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7c0cf739814a6392076492b8b0c6a7260ea1ae60","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-27 18:32:33.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b8e39754735a4c808847b2854aaa1c27 : SUCCESS in 10m 13s\n- grenade https://zuul.opendev.org/t/openstack/build/dfc2b2f0550f4be08b9d73cd2cb93a3d : SUCCESS in 1h 02m 03s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/6bf7752f30084bbf90f68961da3fdbe9 : SUCCESS in 47m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ee99389aa53147d1894a83128f2d5465 : SUCCESS in 47m 42s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/d44947214f9847b8bb42130b65aca945 : SUCCESS in 8m 58s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/22162216bf7c435386f5599585a33c6e : SUCCESS in 7m 55s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/0318c406ab44453dbe5ec35f837a3c43 : SUCCESS in 20m 13s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/f94674436b104bfe95cdf79e07e12e0d : SUCCESS in 16m 36s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/f0c87ae73ec840b3b7853524987bea43 : SUCCESS in 16m 56s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/9b1dea76d1e9417eb25cbb859026f178 : SUCCESS in 20m 54s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/bfc19e69a2e04dc78c733d94d894552d : SUCCESS in 20m 00s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/87196713356542908396ece379f8fdf6 : SUCCESS in 19m 14s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/bc880defde6b42e8bcc19470564ce0b1 : SUCCESS in 19m 08s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/68cbd848477f463883c75492fc15b697 : SUCCESS in 20m 34s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/750c6e0aeb2b42f0a88d589ba576439b : SUCCESS in 18m 28s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/1353c754dca94450b6719a7e521b478e : SUCCESS in 7m 31s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/e4b8adacca1e400fb92b509d31a50237 : SUCCESS in 52m 07s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/ede02eb66c674a24a800223a80baa08a : SUCCESS in 56m 09s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/281cae9c1d2b40cbb24f72b5285c4399 : SUCCESS in 42m 15s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/2114dca85dbf4603a27f72be8aa63e27 : SUCCESS in 51m 44s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e29124dfc8484f6486dcfee9e2d44612 : SUCCESS in 18m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/906cdd85399247d2adef6be6dd1d7a17 : SUCCESS in 7m 25s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a3d1b301afc44b04b646c9e458345837 : SUCCESS in 24m 23s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3f3a626f5d4e4c3ba570235ae8019259 : SUCCESS in 55m 59s","accounts_in_message":[],"_revision_number":1},{"id":"06931ad61c79b39838ac70e2a3a5e13328a4adc0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-27 18:32:34.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"3c1e0c4ed45ae8271e02458e90519f4262da4e98","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-27 18:36:44.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2290a452bd654146b20e3ca28483d9f4 : SUCCESS in 55s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/33b33674dc0c48abaf4363a7c724f027 : SUCCESS in 47s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/972e9cc707d246e2b2b25b04ce2a9595 : SUCCESS in 1m 02s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"10767e482e7f773754b2974e1862b8bb7d120036","revisions":{"10767e482e7f773754b2974e1862b8bb7d120036":{"kind":"REWORK","_number":1,"created":"2021-05-11 22:50:30.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/59/790859/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/59/790859/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/790859/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/790859/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/790859/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/59/790859/1"}}},"commit":{"parents":[{"commit":"c5fe114c9622783b9d6e00ac9bf50e9ffe6dd17b","subject":"Merge \"sharder: stall cleaving at shard range gaps\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c5fe114c9622783b9d6e00ac9bf50e9ffe6dd17b"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-05-11 22:45:08.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-05-11 22:46:45.000000000","tz":-420},"subject":"staticweb: Allow empty listings at the root of a container","message":"staticweb: Allow empty listings at the root of a container\n\nThis is useful for testing that staticweb metadata was properly set\nbefore putting data into the container.\n\nChange-Id: I433bad1cb3f52ad60a81dcbf3209681f7068ed55\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/10767e482e7f773754b2974e1862b8bb7d120036"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/10767e482e7f773754b2974e1862b8bb7d120036"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
