)]}'
{"id":"openstack%2Fswift~791305","triplet_id":"openstack%2Fswift~master~I36e9cb777219162105912ea32945153f5808507f","project":"openstack/swift","branch":"master","topic":"p-sharder-conf-loading","hashtags":[],"change_id":"I36e9cb777219162105912ea32945153f5808507f","subject":"s-m-s-r: use argparse mutually exclusive group","status":"MERGED","created":"2021-05-13 18:47:25.000000000","updated":"2021-05-14 18:06:19.000000000","submitted":"2021-05-14 18:02:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"791305-p-sharder-conf-loading","meta_rev_id":"6a439e5c76d78da917b73e279161a5e82fee115e","_number":791305,"virtual_id_number":791305,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-05-14 18:02:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":2,"date":"2021-05-14 18:06:19.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":2,"date":"2021-05-14 13:46:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":1,"date":"2021-05-14 13:46:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-13 19:47:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-14 07:49:37.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"}],"messages":[{"id":"37ff380001ae96f5a9ffb9ac2a41cec37ab67830","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2021-05-13 18:47:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e4b12304b93d21ef54dbbc49257003364089cc15","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-13 19:47:35.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b9016561d58f4e66b0c5e3fd7c2e5503 : SUCCESS in 9m 13s\n- grenade https://zuul.opendev.org/t/openstack/build/6cb15c9ee2bf4c329f2f654941d3297f : SUCCESS in 57m 21s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a59262f6fb634e7b9c3a70b3e688ebb6 : SUCCESS in 32m 30s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/5dc2eb9083d1421e8476a2032b3d5b88 : SUCCESS in 8m 23s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/2dbd4b9e9b21405eb550ab0ac6c95b05 : SUCCESS in 6m 40s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/ccb944208ca24e7c9fc528ac90ffeb03 : SUCCESS in 22m 09s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/9c4f17774b6e4dfe8be625b6f6871a49 : SUCCESS in 18m 46s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/79f5218e8c164b40bbe617a4016b0714 : SUCCESS in 23m 33s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/f3698a9703324fdfb6aae6aa076c4870 : SUCCESS in 41m 51s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/36d081a998ae4f679f9b2b052ee40bca : SUCCESS in 15m 46s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/bf003c78344a4c628bbf30c38c2be999 : SUCCESS in 41m 13s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/5d602efc82d24116827564353d751bd2 : SUCCESS in 16m 14s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/9224f476ec6742e0b6c7d3709cabcbe2 : SUCCESS in 25m 52s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/38b8c3b4a0754a38b6fae68362bdce7f : SUCCESS in 18m 06s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/efba4232b15f42b4b0d112d564134629 : SUCCESS in 18m 19s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/700ffd18b79b4508bb06db0bc803341e : SUCCESS in 18m 36s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/7c90dfa36cfe4466a1aab551ee1325a4 : SUCCESS in 8m 08s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/51fd117d2823408c91b3410b891eda6d : SUCCESS in 18m 25s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/6cbaab8738824659b2c63ff23c3111d4 : SUCCESS in 41m 34s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/10f5dc12657948fda7966e579dc41e77 : SUCCESS in 42m 28s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c328028652984f5ab5db77799b603b13 : SUCCESS in 32m 34s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/dcb070b8df6f49efba07162d567e17b9 : SUCCESS in 35m 40s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0579b074f68d4ccb809cb159d693f5e4 : SUCCESS in 15m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/df325c763c1742e8887b6715f3bd86b5 : SUCCESS in 6m 37s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/950ae5c09a794a10ba204aab23bd4630 : FAILURE in 22m 24s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6499d3164077463580c3625621774f9c : SUCCESS in 43m 12s","accounts_in_message":[],"_revision_number":1},{"id":"0600c9315c0221f033cab685fb4d6f3df39e229f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-05-14 07:49:37.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"aeb72042ff646edc81578502d812e2f05f437984","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-05-14 07:54:39.000000000","message":"Patch Set 2: Code-Review+2\n\nThanks Clay! I wasn\u0027t aware argparse handled exclusion. #shouldhavesearched\n\nLeaving for you to +A if you happy with my change.\n\nIn pycharm I get:\n\n  AssertionError: \u0027usage: nosetests path_to_file repair [-h] [--yes | --dry-run]\u0027 !\u003d \u0027usage: _jb_nosetest_runner.py path_to_file repair [-h] [--yes | --dry-run]\u0027\n- usage: nosetests path_to_file repair [-h] [--yes | --dry-run]\n?                ^\n+ usage: _jb_nosetest_runner.py path_to_file repair [-h] [--yes | --dry-run]\n?        ++++        ^^^^^^^^^^\n\n...but I suspect we won\u0027t be using nosetests for ever http://eavesdrop.openstack.org/meetings/swift/2021/swift.2021-05-12-21.00.log.html#l-14","accounts_in_message":[],"_revision_number":2},{"id":"b7d2f8414bb61e3393574f0fac47b1b9cde2346b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-14 08:55:59.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb50d708a01e4427a9bb97dacec91ea7 : SUCCESS in 8m 59s\n- grenade https://zuul.opendev.org/t/openstack/build/b80a3617e52e46cc96c4d0a6cc0727fd : SUCCESS in 1h 02m 04s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/77ecfb62882540a9ac766b52b941c2aa : SUCCESS in 42m 30s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/883121b825a046409cf906cfbecc00f3 : SUCCESS in 10m 50s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/8c2869c02f45470e9c7ce1e08297609e : SUCCESS in 6m 42s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/b13755e9a20746d9ad30d4a0d7e2fb3c : SUCCESS in 16m 16s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/7a79e1c7e51a40c99ef2ff89f7787a41 : SUCCESS in 16m 12s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/226d635d29c547a99029c0db4c2f13bd : SUCCESS in 16m 59s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/6eded0b4664c4315b7f198aea1515fbb : SUCCESS in 15m 48s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/2b8c5d1f35c24ba58f0a9b719c9b99f5 : SUCCESS in 19m 52s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/47b8ddfa514e41239e0bd982cb4b87a9 : SUCCESS in 18m 36s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/97f2b7aecea2402eb5e0d3e07f735f03 : SUCCESS in 19m 00s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/57f394784f7045769298326b81f4aa36 : SUCCESS in 22m 33s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/a326a219a7854023aa6d38164ee19e62 : SUCCESS in 18m 33s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/687ac274bde743cf8a2a96d9385ef60a : SUCCESS in 21m 46s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/e71f3397a94042ba9a724e5ef85dc277 : SUCCESS in 18m 48s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/392b66d5a05a4b84a9fa903fe32d79d3 : SUCCESS in 7m 27s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/281729a14f3d497d812a84f4863a460a : SUCCESS in 24m 36s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/c25c1c892e7b4481b499dbed35a7fdb6 : SUCCESS in 48m 14s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/0504238ba3094553896acaff56b48979 : SUCCESS in 51m 39s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/fd9b2a1d178a4231a318d77cc778d72a : SUCCESS in 38m 44s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b028e68eb3af4855bff8264c07730316 : SUCCESS in 39m 36s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bbd8e600b83a47d899e36ce7e73fba94 : SUCCESS in 19m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6fa777c234a6450086205d61a4170c69 : SUCCESS in 6m 23s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/5750053338f0450aaee96829f5a83eea : SUCCESS in 23m 24s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/022ae62546cc4670bf81d4a9464183ec : SUCCESS in 58m 48s","accounts_in_message":[],"_revision_number":2},{"id":"4cfdbe16987bae33564b677142a9751b2ce33f82","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-05-14 09:52:06.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"278ed79a711a46c03c8757e145454c969f362b73","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-05-14 09:53:04.000000000","message":"Patch Set 3:\n\nrebased onto master to enable absolute conf value patches to be rebase on this and eliminate merge conflicts between them all!","accounts_in_message":[],"_revision_number":3},{"id":"6a5a730cc4ee295e1a9671a8a97cb7149b2b3b2e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-14 10:52:53.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d960c35a9c5445d1b868623dbec53de6 : SUCCESS in 9m 42s\n- grenade https://zuul.opendev.org/t/openstack/build/32332286add74d4dac69e8131fa487f3 : SUCCESS in 59m 15s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/155f6894e819468a83902902e6f689e3 : SUCCESS in 36m 53s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/5c83046b1aca449f9c72a261b51f1a72 : SUCCESS in 11m 02s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/3b5256a5a0934365892dffa1a562e742 : SUCCESS in 10m 11s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/70d3909e33b048d9a233d8a410f64847 : SUCCESS in 17m 21s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/b27fe8c7ebc1411c892379ab3a4ee3bf : SUCCESS in 22m 49s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/f910745ead8a430e8f3a845eb90ef1c0 : SUCCESS in 17m 05s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/eb66ff5cf814420faabf8ca88470a3cb : SUCCESS in 19m 51s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/5d32672f444e42e38826dd1d5b874667 : SUCCESS in 13m 57s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/137d3341db0e4c34927470e3e97a9dc0 : SUCCESS in 18m 41s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/6f8116eae92549f6a50f60a83e251cdf : SUCCESS in 17m 14s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/534a0206b40a4467b603bef38d44c436 : SUCCESS in 24m 31s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/111a5cad4521496c8885c4e7ac29e905 : SUCCESS in 16m 54s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/5f005240cd52498d98db608b7f77cf63 : SUCCESS in 19m 58s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/ab1169ca63cc4ce5b2f00ce63092c441 : SUCCESS in 19m 22s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/47b5be9141ab4e08bf854a07ff089628 : SUCCESS in 7m 38s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/c5408a1d1136431cbdf32c0cc4005561 : SUCCESS in 18m 33s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/34c92f817b7b4f7d9a2e7f43390b5e96 : SUCCESS in 53m 33s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/5e19922fa5104d62925ae2ddd1fa4053 : SUCCESS in 50m 55s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/299dcb7288eb42c5b83a7fc13bb9f783 : SUCCESS in 30m 43s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e03333f3e50141798a26b18350073b00 : SUCCESS in 38m 29s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/80ede64375af444ea4f59a41d7ec6464 : SUCCESS in 19m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c685d74629743b7a9060593cc929a88 : SUCCESS in 7m 43s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/4ffdea21898b43619a6ddf18eed1933f : SUCCESS in 25m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7266953a7e7a4d3aad47557a4bd29bb3 : SUCCESS in 53m 43s","accounts_in_message":[],"_revision_number":3},{"id":"c68be4bae42022929cb80f9c2f57e124220ed2bb","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2021-05-14 13:46:31.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"7ccddd8cbcd0f3cdfcf3300f3d9df6b17d2ff3ca","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-14 13:47:07.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"8862c6a58f6e1b712c5b21731ed870b4819a71d2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-14 18:02:18.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b849fd53a9c48e381b38d4cfbe82e06 : SUCCESS in 9m 02s\n- grenade https://zuul.opendev.org/t/openstack/build/d97fd74c69c64d8284f57b09f5b19501 : SUCCESS in 56m 12s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c236727ab0d4415e917d4322f4d02da0 : SUCCESS in 54m 34s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/920ac33c1a384081b5e6bbb650e6a43c : SUCCESS in 9m 51s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/38c860a5a118480ba2b70df8b349f501 : SUCCESS in 9m 14s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/dbab382895e34e6e969064b171d2c15b : SUCCESS in 20m 45s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/0b056b11aa354441a5fa2e47c0a36085 : SUCCESS in 15m 47s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/7be69f74964a465ab58b68d49d9f6a4d : SUCCESS in 16m 22s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/84d7407e5a7e4f238b4423116737f17d : SUCCESS in 20m 11s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/2d48eb01fe3045cc93bef01eb0511b64 : SUCCESS in 15m 52s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/fbbcf6874195460db53801bceab9d6a5 : SUCCESS in 16m 27s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/fdbb18871c6e49d1a9dd1d3b44e60ff6 : SUCCESS in 23m 29s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/e339e4e22d0b43a09e351b4e44e5d7e9 : SUCCESS in 19m 29s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/848b563ecf154eab926edab321dc1b1e : SUCCESS in 20m 22s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/20c5b35a5e1c4802b6240bd90d735fe0 : SUCCESS in 19m 08s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/0c44aa57305841c3999c9962fc343490 : SUCCESS in 32m 30s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/756620cc6caa4e4aab1cae39fc0f0d54 : SUCCESS in 6m 50s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/6964675c36a44996ab12b2c20a48f8e4 : SUCCESS in 52m 19s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/433873e9c0c04a16a19a4b7a46602343 : SUCCESS in 1h 12m 44s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/9d972b330d6445489f0383ef5684d55a : SUCCESS in 37m 30s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/0abeec4cd64b41f3b095dec63475bc6c : SUCCESS in 37m 15s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/88f9f98f51a74b9da0acb0ca665603b8 : SUCCESS in 16m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/77bdff01caa24dd18b11b55007e8fe79 : SUCCESS in 5m 52s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/63e3a48b881246c882db321f0f56a31d : SUCCESS in 28m 58s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/be0cbffe8c4e4ce291c99fbfb2e66227 : SUCCESS in 53m 20s","accounts_in_message":[],"_revision_number":3},{"id":"4952dfcc49fb8c56c44c86f831048c5acf79ace6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-14 18:02:31.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"312c3763a1acf0fc603c1a834de1d5eddae852b2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-14 18:06:19.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/63c980f59d2e40d1bcdc0417fbe91940 : SUCCESS in 58s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/cadf4bf6c5494b0089d42282260c519c : SUCCESS in 44s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/471ebfc42bce443b98dc391b9e00e5e8 : SUCCESS in 42s (non-voting)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"bbe189b5377509192eae660b6b6f5eb623d06912","revisions":{"e4fc961bc567bff604571513395b1d649497c1f0":{"kind":"REWORK","_number":1,"created":"2021-05-13 18:47:25.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/05/791305/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/05/791305/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/791305/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/791305/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/791305/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/05/791305/1"}}},"commit":{"parents":[{"commit":"568036f1b74e1057462401492ed323958d9d8ff5","subject":"manage-shard-ranges: add --dry-run option for compact and repair","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/568036f1b74e1057462401492ed323958d9d8ff5"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2021-05-13 18:42:46.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2021-05-13 18:47:14.000000000","tz":-300},"subject":"s-m-s-r: use argparse mutually exclusive group","message":"s-m-s-r: use argparse mutually exclusive group\n\nChange-Id: I36e9cb777219162105912ea32945153f5808507f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e4fc961bc567bff604571513395b1d649497c1f0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e4fc961bc567bff604571513395b1d649497c1f0"}]},"branch":"refs/heads/master"},"cc1e892e43d909c988bb2cb59e9ae8faa707148d":{"kind":"REWORK","_number":2,"created":"2021-05-14 07:49:37.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/05/791305/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/05/791305/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/791305/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/791305/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/791305/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/05/791305/2"}}},"commit":{"parents":[{"commit":"568036f1b74e1057462401492ed323958d9d8ff5","subject":"manage-shard-ranges: add --dry-run option for compact and repair","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/568036f1b74e1057462401492ed323958d9d8ff5"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2021-05-13 18:42:46.000000000","tz":-300},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-05-14 07:49:27.000000000","tz":60},"subject":"s-m-s-r: use argparse mutually exclusive group","message":"s-m-s-r: use argparse mutually exclusive group\n\nChange-Id: I36e9cb777219162105912ea32945153f5808507f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cc1e892e43d909c988bb2cb59e9ae8faa707148d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cc1e892e43d909c988bb2cb59e9ae8faa707148d"}]},"branch":"refs/heads/master"},"bbe189b5377509192eae660b6b6f5eb623d06912":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-05-14 09:52:06.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/05/791305/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/05/791305/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/791305/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/791305/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/791305/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/05/791305/3"}}},"commit":{"parents":[{"commit":"9edcfad22fbdee643860f562eb7e55d0bb7465f2","subject":"Merge \"swift-manage-shard-ranges: output total compactible sequences\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9edcfad22fbdee643860f562eb7e55d0bb7465f2"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2021-05-13 18:42:46.000000000","tz":-300},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-05-14 09:49:55.000000000","tz":60},"subject":"s-m-s-r: use argparse mutually exclusive group","message":"s-m-s-r: use argparse mutually exclusive group\n\nChange-Id: I36e9cb777219162105912ea32945153f5808507f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bbe189b5377509192eae660b6b6f5eb623d06912"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bbe189b5377509192eae660b6b6f5eb623d06912"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
