)]}'
{"id":"openstack%2Fswift~791485","triplet_id":"openstack%2Fswift~master~Ie15c9f40a132374c89337f0009fb5cf5a8e62c51","project":"openstack/swift","branch":"master","topic":"p-sharder-missing-overlaps","hashtags":[],"change_id":"Ie15c9f40a132374c89337f0009fb5cf5a8e62c51","subject":"sharder: report perfectly overlapping shard ranges","status":"MERGED","created":"2021-05-14 13:44:04.000000000","updated":"2021-05-17 17:35:43.000000000","submitted":"2021-05-17 08:15:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":15,"unresolved_comment_count":5,"has_review_started":true,"submission_id":"791485-p-sharder-missing-overlaps","meta_rev_id":"dcf7396ce2eb976adae88f7082d11f273331bfc7","_number":791485,"virtual_id_number":791485,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-05-17 08:15:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"all":[{"value":2,"date":"2021-05-17 07:01:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":2,"date":"2021-05-14 13:54:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-05-17 17:35:43.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"all":[{"value":1,"date":"2021-05-17 07:01:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-14 13:54:09.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"},{"updated":"2021-05-14 14:47:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-14 20:45:12.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2021-05-17 07:01:15.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"REVIEWER"}],"messages":[{"id":"854c3ee63db3092cf1588359ba69a700aa9252ae","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-05-14 13:44:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d93563a9b37fe738517ee7012dcfcc589fc06a99","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-05-14 13:48:35.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fe0d491fce2e313b2cca9db60a8353f515ed6cc9","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-05-14 13:50:46.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9a511dcb9ad571060e2e361d87dc29139a906f6f","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2021-05-14 13:54:09.000000000","message":"Patch Set 1: Code-Review+2\n\n(2 comments)\n\nthis all makes sense!","accounts_in_message":[],"_revision_number":1},{"id":"3c2911ff8379af7d2b4d2098aa2aff8740121b3f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-14 14:47:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/824dcf9b8d3e4e8ca117f89b3b2f8867 : SUCCESS in 9m 32s\n- grenade https://zuul.opendev.org/t/openstack/build/f496bd2c0c174a0dbada565220d450a1 : SUCCESS in 58m 44s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/bf199c5b4ae54ab4943c2743cd2ed899 : SUCCESS in 49m 09s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/0637ba6ca5174abdbc248f86a008d9de : SUCCESS in 9m 44s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/0348dcf63d83401790c63e3e8266bab7 : SUCCESS in 7m 24s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/d3c2ed6fcff0454b8e0ddb6f0e2ed5f0 : SUCCESS in 16m 23s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/8e0b192629904bb5b0f9c8c4bff7033c : SUCCESS in 20m 11s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/066294379fe341a4a4c3d769c5a33b23 : SUCCESS in 20m 41s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/c39c0348663d4c72acf8ad3bfaa4dbe6 : SUCCESS in 15m 54s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/593e65418cfa4f65aab91ed388f7244a : SUCCESS in 15m 34s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/607343bdb3a84303953370727bafccfb : SUCCESS in 26m 39s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/f3292b49c59947cdb79bb1be44cdab77 : SUCCESS in 18m 29s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/e1943957741b43aaba3bc62459d38862 : SUCCESS in 20m 04s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/6121fc54065e41d0956f15a86abfe981 : SUCCESS in 18m 58s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/e036e54c1ef846e588e983c6674c12c9 : SUCCESS in 18m 30s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/3f9c4f0e903040cd912528e471748b09 : SUCCESS in 17m 41s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/980d122336a84fc79debee54974c8481 : SUCCESS in 7m 01s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/cb4c78315a954af2b696d13f3491453d : SUCCESS in 22m 01s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/c7dbd7a931514feabed1d3eac194aa57 : SUCCESS in 58m 06s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/49f6a964ec09439d97f220a7300ff72b : SUCCESS in 49m 11s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/71269567dbef475ba2e5bc96032af26f : SUCCESS in 35m 39s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/fe893229b24a4cee86321d0974cdbc7a : SUCCESS in 42m 16s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9186368d46084ebc96a8f5d5294bb422 : SUCCESS in 19m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b083ea07872f40fea2791d17e99c1bdc : SUCCESS in 6m 24s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e432b881bb6b4dbd878b89810a2b9bc8 : SUCCESS in 21m 08s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c1a8097f4b624c1ea1bf64ed7280c976 : SUCCESS in 47m 40s","accounts_in_message":[],"_revision_number":1},{"id":"fca6d5d957bf758c08395f1d33f346cda04e0fdd","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-05-14 20:45:12.000000000","message":"Patch Set 1: Code-Review+1\n\n(3 comments)\n\nCouple other patches in a similar vein:\n\n* https://review.opendev.org/c/openstack/swift/+/772621\n* https://review.opendev.org/c/openstack/swift/+/721377\n\nThis seems pretty well-targeted, though. I might see if I can further improve that quadratic behavior.","accounts_in_message":[],"_revision_number":1},{"id":"5bbddf53ac36a7aaab5ad7b18bd00ae132d04539","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2021-05-17 07:01:15.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(3 comments)\n\nLooks great. And I look forward to having a play with Tims speedup follow up tomorow when I get a chance.","accounts_in_message":[],"_revision_number":1},{"id":"492dbc0bd9550694a75d89bbebd59b2aa1db9213","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-17 07:01:40.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a1b0bc4329baa9440f0a038cea7253301bbc3745","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-17 08:15:13.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a352d4c0501c45abb350e8ece2b5064e : SUCCESS in 9m 29s\n- grenade https://zuul.opendev.org/t/openstack/build/39f2d3f9200c4393a1d65156f6f985d1 : SUCCESS in 1h 08m 12s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/52ea33459f99447499eb512450eda643 : SUCCESS in 50m 58s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/ab3e994302ca45bd8363f017aeb6baaf : SUCCESS in 10m 18s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/1e6544ea3acb4b2dade7623b8ecbb386 : SUCCESS in 8m 05s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/7117d8241898420ead91bee4995812e2 : SUCCESS in 17m 44s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/196f51d6479a4b49bef5b9f3076646d8 : SUCCESS in 16m 49s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/d051903941db480caf1f691d5781cb85 : SUCCESS in 17m 48s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/b6f8921a9d8b47b5bf07d73cdbb20beb : SUCCESS in 16m 06s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/dab3a4f47f4f417c9d2bfc60d45dbcb6 : SUCCESS in 17m 34s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/c60be08265a9445dbd192a3050c5b3fe : SUCCESS in 15m 56s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/48bda0ecc4a746788faf98ae9023c9df : SUCCESS in 18m 39s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/f4f969451de84beba956292c6589581e : SUCCESS in 17m 30s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/010f246b03d54f9a8c087d279809045e : SUCCESS in 17m 10s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/f955292e6e1b4e96878fcb57ff6f6c37 : SUCCESS in 18m 14s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/02a58313413d4752a4082b2ff7514a4f : SUCCESS in 19m 07s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/3ca48f74110d4cc483cfebb51c017dae : SUCCESS in 6m 51s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/583ea47425ad49de982f2c4a279dff55 : SUCCESS in 47m 06s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/4127152340c7460ea812992603c11feb : SUCCESS in 49m 58s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ccb0c98bb76641b9969087b177f6acdd : SUCCESS in 43m 31s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/36bae17e657b499dad5e919dfab9097c : SUCCESS in 32m 33s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/52bf13ae80d94a0cad12abb148d1b552 : SUCCESS in 14m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/72652e6e9367457ea9d07a618c67de67 : SUCCESS in 5m 40s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c006df77a568451897be7907e4d9376c : SUCCESS in 22m 02s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2e72a0e4f6e0467dacf48fc3a8e3d94f : SUCCESS in 51m 51s","accounts_in_message":[],"_revision_number":1},{"id":"4579b3bd8013e005167330de6d39b9fb4212c2c5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-17 08:15:14.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"a8b46c875efed1bafd2215193df5028b65356db4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-17 08:16:52.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f48c9db1debb4d2cab5fdefb39f14313 : SUCCESS in 1m 10s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/b22ae42a859e41efa65770a27e0c743b : SUCCESS in 56s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/2f9c8ce5788649f7a5a0067622848168 : SUCCESS in 59s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8c4074a904595d2b7f970b8d6617d6235618edee","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-05-17 08:29:05.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"dcf7396ce2eb976adae88f7082d11f273331bfc7","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-05-17 17:35:43.000000000","message":"Patch Set 1: Code-Review+2\n\n(2 comments)\n\nLog line length was the main thing I was worried about, but now I see that it\u0027s not really any worse than before.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1b183f221d258819c82ee58961afc765529314ac","revisions":{"1b183f221d258819c82ee58961afc765529314ac":{"kind":"REWORK","_number":1,"created":"2021-05-14 13:44:04.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/85/791485/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/85/791485/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/791485/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/791485/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/791485/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/85/791485/1"}}},"commit":{"parents":[{"commit":"9edcfad22fbdee643860f562eb7e55d0bb7465f2","subject":"Merge \"swift-manage-shard-ranges: output total compactible sequences\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9edcfad22fbdee643860f562eb7e55d0bb7465f2"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-05-14 12:59:37.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-05-14 13:05:29.000000000","tz":60},"subject":"sharder: report perfectly overlapping shard ranges","message":"sharder: report perfectly overlapping shard ranges\n\nPreviously, shard ranges that overlapped perfectly (i.e. their bounds\nwere equal) were erroneously excluded from overlapping shard reports.\nWith this patch they will be reported in logs and cause shard audit\nfailures.\n\nAlso, the format of the audit log line is simplified to avoid\nunnecessary duplication of information.\n\nChange-Id: Ie15c9f40a132374c89337f0009fb5cf5a8e62c51\nCloses-Bug: #1928459\nRelated-Bug: #1913332\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1b183f221d258819c82ee58961afc765529314ac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1b183f221d258819c82ee58961afc765529314ac"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
