)]}'
{"id":"openstack%2Fswift~791487","triplet_id":"openstack%2Fswift~master~I55a9bd5ca63c0ece701714f76f8a6331198dfdd9","project":"openstack/swift","branch":"master","topic":"p-smsr-analyze-populations","hashtags":[],"change_id":"I55a9bd5ca63c0ece701714f76f8a6331198dfdd9","subject":"smsr: enable analyze command to open a db file","status":"NEW","created":"2021-05-14 14:01:17.000000000","updated":"2026-09-04 18:58:24.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"5295b7e269300bddbfbf1ee2e618d36999a17183","_number":791487,"virtual_id_number":791487,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-09-04 18:58:24.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-14 15:04:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"959815387248793f65b22858317a680081dcd0e9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-05-14 14:01:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"465f4c59e34a21fff779231c7502c36343f4f6fd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-14 15:04:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eac51f2211934aeebdab9a4453ba17ea : SUCCESS in 11m 35s\n- grenade https://zuul.opendev.org/t/openstack/build/6b4f03aff27f45949524c585e7320abd : SUCCESS in 1h 01m 58s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e515c02a7e924023aa4ea0c0dc6f5bc7 : SUCCESS in 46m 10s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/8a492ddb97a54d6e977022250c09afc7 : SUCCESS in 9m 10s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/45d26b0c1e36490d8c40f9e3dbe90689 : SUCCESS in 7m 28s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/f900c9c9bdc940cca197b9c1afd6cb20 : SUCCESS in 16m 57s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/39b601a887ee41c3937a5044c5b4446b : SUCCESS in 18m 10s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/2a37620c18384aa5ba6c72b98c1162b1 : SUCCESS in 14m 41s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/410f763ad8324a5b9fd0b61808bb2624 : SUCCESS in 15m 53s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/dc53a17142e14abeae7ebbb316762bf9 : SUCCESS in 12m 14s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/186b6af661964932997f9a68bc300f5b : SUCCESS in 17m 03s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/548f0e3bc99b4bfba94ab7e95c701204 : SUCCESS in 16m 18s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/6112b6f0b2de4e63857b396a6f43b844 : SUCCESS in 14m 59s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/4734f242b79d49fca4ef29d491977da4 : SUCCESS in 16m 30s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/e1e652d95455489286fe376dc97d541e : SUCCESS in 14m 53s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/36710ac292a544f5bc08aa0613e0db36 : SUCCESS in 16m 11s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/64fb48ad22e74b7e8fe95766f1d3c0cb : SUCCESS in 6m 30s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f19b2650afde43a98f357ca600e3e71b : SUCCESS in 20m 51s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/5d464182d8dc499b87ea46ef5c34922d : SUCCESS in 45m 51s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/a757812ae02f46db8bc4eb3784b5718b : SUCCESS in 58m 44s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/3d1de8496ab44d2882946fc017516a99 : SUCCESS in 43m 50s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/aa88b3c9c88948969a30939a65406e2b : SUCCESS in 35m 45s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/15caf95c53b64776bab722d52613406f : SUCCESS in 15m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0e61b83655254f0a847ce9c00621c257 : SUCCESS in 6m 31s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9a8852028e894635aabe23e207427a94 : SUCCESS in 21m 54s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c6107da2aba3426bb1275824dfb15b14 : SUCCESS in 50m 51s","accounts_in_message":[],"_revision_number":1},{"id":"9631432d03e656c95f56e2df7bfe373cd3f22c1b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2026-09-04 17:29:13.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"7f6e03ad0478c139c805d60ecb5c7f00a151b904","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2026-09-04 17:32:12.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"0d5125858bde7d9cd5ee7dbd540ce28e4a21fbbb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2026-09-04 17:43:45.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"be63bcf567f81af1720b5a1af034fc7cf494c654","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-04 18:38:09.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/405333ace9c64c5bb68b2f6934d0a095\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/8bfd7608174249f39919979a561b73f0 : SUCCESS in 52m 43s","accounts_in_message":[],"_revision_number":4},{"id":"5295b7e269300bddbfbf1ee2e618d36999a17183","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-04 18:58:24.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e0d9d135cabc476986c5b7349d00aa0e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff22a4bc60d04239bc86e15fab8c7b46 : SUCCESS in 10m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d72ece78048f48958b70a15dea748f67 : SUCCESS in 9m 41s\n- grenade https://zuul.opendev.org/t/openstack/build/3e89d2f4c50c4f3896ef6efa6a21cdac : SUCCESS in 44m 08s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e34b4ad86fa04b43944230308c0630f4 : SUCCESS in 1h 04m 37s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/1dbe1d23f8574cc5b3e7c5f21565fbb5 : SUCCESS in 51m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2274708187c6477094e9ce803dcd2eca : SUCCESS in 1h 08m 06s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2cbf611b34454a79b2f3bb4a0580d2cb : SUCCESS in 10m 58s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3577333005174a6c8523a9fc2189eb84 : SUCCESS in 18m 30s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0773e62933b548809438d41bd8953701 : SUCCESS in 20m 00s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/9c3d3c14e55a4b8880e07b2f3dd551dc : SUCCESS in 8m 43s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/2c94a1a22af64c84b19bf6d9c8d466c4 : SUCCESS in 28m 07s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/0895b824563341d1af6b0831b1960d74 : SUCCESS in 24m 21s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/940bd64b75424f4383082e9f00865c6a : SUCCESS in 35m 17s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/12b7055b1d234ae28393cbcebb4dc9c5 : SUCCESS in 17m 05s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/3df86f58068742efa3b8d264f6c4d635 : SUCCESS in 15m 51s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/19fcac6c7831469c94c9afaca52c6a37 : SUCCESS in 15m 15s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/f8a73f9bacc34e2f9c93e2dd0105888b : SUCCESS in 15m 44s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/cc50be4952b44bb4ab533b60fc45dab4 : SUCCESS in 17m 14s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/c2710300521744cba8af49a6c780e98b : SUCCESS in 5m 54s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f3192730e2d6441494c4ea1939a86e05 : FAILURE in 30m 13s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/74ede0ede95d4537a7a3b8192f68f17e : SUCCESS in 9m 20s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/33d27bcdb1ed45c4963e5c0210920f8a : SUCCESS in 1h 00m 38s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c9caea3955f84c7ab07d07f17612aea8 : SUCCESS in 35m 29s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/39771707e75246418c98d1fbff23d6a2 : SUCCESS in 1h 00m 15s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4ddca70b0c1d4ab2b5f138f64f8e8267 : SUCCESS in 28m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0710c4b01a064a50a7ab0a4247597875 : SUCCESS in 4m 02s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/70a4128643ea4f6393e7bbdfa8715e33 : SUCCESS in 20m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4dc9d5732eb34f9cb4c9b45ff9c33eae : SUCCESS in 53m 45s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"627e6be13009ed312c7c98a376bb249053992c77","revisions":{"647371a1c8dfd7708369181709037276a77da6f2":{"kind":"REWORK","_number":1,"created":"2021-05-14 14:01:17.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/87/791487/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/87/791487/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/87/791487/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/87/791487/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/87/791487/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/87/791487/1"}}},"commit":{"parents":[{"commit":"bbe189b5377509192eae660b6b6f5eb623d06912","subject":"s-m-s-r: use argparse mutually exclusive group","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bbe189b5377509192eae660b6b6f5eb623d06912"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-05-14 10:57:39.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-05-14 10:59:27.000000000","tz":60},"subject":"smsr: enable analyze command to open a db file","message":"smsr: enable analyze command to open a db file\n\nAllow either json file or db file to be passed in for analyze.\n\nChange-Id: I55a9bd5ca63c0ece701714f76f8a6331198dfdd9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/647371a1c8dfd7708369181709037276a77da6f2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/647371a1c8dfd7708369181709037276a77da6f2"}]},"branch":"refs/heads/master"},"8f4b23409f0ece6433c023b269d7658142e206e3":{"kind":"REWORK","_number":2,"created":"2026-09-04 17:29:13.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/87/791487/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/87/791487/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/87/791487/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/87/791487/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/87/791487/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/87/791487/2"}}},"commit":{"parents":[{"commit":"5cab258d88fd56fc341ec8768682f96be1e738ab","subject":"Merge \"trivial: Remove some py2 comments and validation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5cab258d88fd56fc341ec8768682f96be1e738ab"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-05-14 10:57:39.000000000","tz":60},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2026-09-04 15:54:29.000000000","tz":-300},"subject":"smsr: enable analyze command to open a db file","message":"smsr: enable analyze command to open a db file\n\nAnalyze currently accepts only JSON shard-range data, so examining ranges in a\ncontainer DB requires extracting that data first. Allow the command to read\neither a DB or JSON file, preserving JSON fallback.\n\nAssisted-By: Codex:gpt-5\nChange-Id: I55a9bd5ca63c0ece701714f76f8a6331198dfdd9\nSigned-off-by: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8f4b23409f0ece6433c023b269d7658142e206e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8f4b23409f0ece6433c023b269d7658142e206e3"}]},"branch":"refs/heads/master"},"6b48e940bbcd3245735c481c75153d0c2a18087f":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2026-09-04 17:32:12.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/87/791487/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/87/791487/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/87/791487/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/87/791487/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/87/791487/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/87/791487/3"}}},"commit":{"parents":[{"commit":"5cab258d88fd56fc341ec8768682f96be1e738ab","subject":"Merge \"trivial: Remove some py2 comments and validation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5cab258d88fd56fc341ec8768682f96be1e738ab"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-05-14 10:57:39.000000000","tz":60},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2026-09-04 17:30:31.000000000","tz":-300},"subject":"smsr: enable analyze command to open a db file","message":"smsr: enable analyze command to open a db file\n\nAnalyze currently accepts only JSON shard-range data, so operators must\nextract ranges from a container DB before examining them. Allow the command\nto read either a DB or JSON file while preserving JSON fallback.\n\nAssisted-By: Codex:gpt-5\nChange-Id: I55a9bd5ca63c0ece701714f76f8a6331198dfdd9\nSigned-off-by: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6b48e940bbcd3245735c481c75153d0c2a18087f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6b48e940bbcd3245735c481c75153d0c2a18087f"}]},"branch":"refs/heads/master"},"627e6be13009ed312c7c98a376bb249053992c77":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2026-09-04 17:43:45.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/87/791487/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/87/791487/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/87/791487/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/87/791487/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/87/791487/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/87/791487/4"}}},"commit":{"parents":[{"commit":"5cab258d88fd56fc341ec8768682f96be1e738ab","subject":"Merge \"trivial: Remove some py2 comments and validation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5cab258d88fd56fc341ec8768682f96be1e738ab"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-05-14 10:57:39.000000000","tz":60},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2026-09-04 17:43:45.000000000","tz":0},"subject":"smsr: enable analyze command to open a db file","message":"smsr: enable analyze command to open a db file\n\nAnalyze currently accepts only JSON shard-range data, so operators must\nextract ranges from a container DB before examining them. Allow the\ncommand to read either a DB or JSON file while preserving JSON fallback.\n\nAssisted-By: Codex:gpt-5\nChange-Id: I55a9bd5ca63c0ece701714f76f8a6331198dfdd9\nSigned-off-by: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/627e6be13009ed312c7c98a376bb249053992c77"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/627e6be13009ed312c7c98a376bb249053992c77"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
