)]}'
{"id":"openstack%2Fswift~791488","triplet_id":"openstack%2Fswift~master~I7d8a5d9618d57b56faa5578e159e081bf4621789","project":"openstack/swift","branch":"master","topic":"p-smsr-analyze-populations","hashtags":[],"change_id":"I7d8a5d9618d57b56faa5578e159e081bf4621789","subject":"WIP swift-manage-shard-ranges: add shard population analysis","status":"ABANDONED","created":"2021-05-14 14:01:17.000000000","updated":"2021-10-29 10:37:38.000000000","total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"455666f85add1a960fa3d03f294c06bb6ca12ca4","_number":791488,"virtual_id_number":791488,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2021-05-14 16:23:22.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-14 15:00:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c2385526b2b7901289b3e2e7605657742c3808d7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-05-14 14:01:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"86d9746417650d7d0fa67fba0827fb658faedc2f","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-05-14 14:03:38.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9756fc7e3e33eea30fd57007c15518f5bcddbb7a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-14 15:00:03.000000000","message":"Patch Set 1: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c5edf4b118649368535b89f9ff84ef3 : SUCCESS in 9m 18s\n- grenade https://zuul.opendev.org/t/openstack/build/05adc1f454f945fb8a9f3b0180154fc5 : SUCCESS in 57m 04s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/1d94ef99496f43f4ab2938183566459b : SUCCESS in 39m 34s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/19d13c43bdc643eb966bc4285f040cde : SUCCESS in 9m 16s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/86d722dd2f4143c2977939d65e6bcda5 : SUCCESS in 8m 55s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/a34494c1f630460eaa6b2e65821915a3 : SUCCESS in 16m 44s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a3848fcc8fe4409fb138812514b47851 : SUCCESS in 20m 20s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/dfae4bf602484270979afc0cab865abf : SUCCESS in 17m 17s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/cbb66808b9a94f15b42987ca8a2c30a2 : SUCCESS in 15m 30s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/76be4d89c55e4cf9bbea3f2328d762d9 : SUCCESS in 16m 16s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/3030c85c8772424c97b3318ceb12e614 : SUCCESS in 15m 57s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/2afc67579f9341938e6a30d950dc8d0c : SUCCESS in 21m 30s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/af31c22ddd4b43cf9d42f5211126eda1 : SUCCESS in 16m 05s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/8a2d08d2eb0749b788964a53fa754da5 : SUCCESS in 16m 51s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/3b5efbd59c404c31a625bb1dc3f6e882 : SUCCESS in 16m 56s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/d837459d11b04dabba88be33c31a086d : SUCCESS in 21m 03s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/53da736f72bb454a8d6f6ecc924abcc2 : SUCCESS in 7m 27s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/c1dad800f02b4e358b98c55113a1efcc : SUCCESS in 21m 20s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/7dc32f57cb1e4fd8ba4185183d010224 : SUCCESS in 46m 15s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/70cb63c1c16e4c2a8b6c78f447cdfd1d : SUCCESS in 50m 57s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/aa3621b6ba6c464f9dc7a792db2b20bf : SUCCESS in 32m 07s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/7767177e062d43aebef6ffeae55771cb : SUCCESS in 36m 18s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9e2a938320f8496f81be205f059683b0 : SUCCESS in 22m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6bf95bbe02424ef4bc8d2f97d220c081 : FAILURE in 5m 43s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c4498ce02ec74839a9000027f884cb71 : SUCCESS in 30m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dce29dc9fccf495e844eaaf34bcd0b79 : SUCCESS in 50m 15s","accounts_in_message":[],"_revision_number":1},{"id":"0f49b7e2f3042effd30f171ab47fbab0b969685e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-05-14 15:23:34.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a7c782e87d7494eac63f7e2d08f310c3dfe7adf4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-14 16:23:22.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/845170e0db4c4da9aaaea4619e26fba3 : SUCCESS in 9m 06s\n- grenade https://zuul.opendev.org/t/openstack/build/46bb0f39f85b4234a1cda1a99500899f : SUCCESS in 54m 29s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/4818ed2b107e46e58638c795c212d031 : SUCCESS in 45m 53s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/85d0fe5e658548cea90d96701864af75 : SUCCESS in 9m 20s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/438594d91a544c238ef64067619a2e01 : SUCCESS in 7m 08s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/0c2ff2cc8cc94537bf16676fa0f4d23b : SUCCESS in 17m 40s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/05b8c31e611440ee92c9c9ca112dfb5a : SUCCESS in 17m 08s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/0c1f284b4b4f47d8b9c0fc0b0439f69f : SUCCESS in 18m 30s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/e5af08be900d4284a1a2422b9b6148f4 : SUCCESS in 16m 09s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/beb8582b31714bc9964f2ddee28b1d90 : SUCCESS in 14m 22s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/7bfbe94b158c4eb4a34a50f2940a2301 : SUCCESS in 17m 57s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/a593c4fe1ab04b02b9c03b3705660916 : SUCCESS in 34m 55s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/a811b7245de14478b849580339237084 : SUCCESS in 20m 37s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/a8cd7c4f41fc4d2fb036c24468b332af : SUCCESS in 34m 58s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/09c794e09e824bf4b1bfec47ce6b1f09 : SUCCESS in 18m 41s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/2800f3fcdc8e4762b7a0a9dc7a7eeb0c : SUCCESS in 25m 17s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/7ffc190a55a74fff93095538924e3d02 : SUCCESS in 7m 09s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/3518ff3d4efc43e3a844e60b4d5fb10a : SUCCESS in 24m 12s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/d05f0b8e6966474bbabbc2c052a197e0 : SUCCESS in 54m 20s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/1b487af4e1104d4ca96454d21e130c82 : SUCCESS in 52m 01s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ba3499e0129847be91f464eaae3e224e : SUCCESS in 36m 25s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/96e4bf6dd5e54f228bb3c2d66c9609e5 : SUCCESS in 46m 42s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1871c20498e945559f591d27c5c40ccd : SUCCESS in 17m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/27e15117e1704625b0ccd9490dceca01 : SUCCESS in 5m 55s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/71f9a56d8fcf4d4a8ac182242343f053 : SUCCESS in 21m 56s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7f403f3375ec46ada0e1add7b05bdc59 : SUCCESS in 43m 24s","accounts_in_message":[],"_revision_number":2},{"id":"455666f85add1a960fa3d03f294c06bb6ca12ca4","tag":"autogenerated:gerrit:abandon","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-10-29 10:37:38.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c52e28f97183afdb6f8696e1cfb07b91bac8cee5","revisions":{"52dd88c262f7697981746db071eb1607677d1c9e":{"kind":"REWORK","_number":1,"created":"2021-05-14 14:01:17.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/88/791488/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/88/791488/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/88/791488/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/88/791488/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/88/791488/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/88/791488/1"}}},"commit":{"parents":[{"commit":"647371a1c8dfd7708369181709037276a77da6f2","subject":"smsr: enable analyze command to open a db file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/647371a1c8dfd7708369181709037276a77da6f2"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-05-14 07:42:44.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-05-14 13:56:03.000000000","tz":60},"subject":"WIP swift-manage-shard-ranges: add shard population analysis","message":"WIP swift-manage-shard-ranges: add shard population analysis\n\nTODO: needs tests\n\nChange-Id: I7d8a5d9618d57b56faa5578e159e081bf4621789\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/52dd88c262f7697981746db071eb1607677d1c9e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/52dd88c262f7697981746db071eb1607677d1c9e"}]},"branch":"refs/heads/master"},"c52e28f97183afdb6f8696e1cfb07b91bac8cee5":{"kind":"REWORK","_number":2,"created":"2021-05-14 15:23:34.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/88/791488/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/88/791488/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/88/791488/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/88/791488/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/88/791488/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/88/791488/2"}}},"commit":{"parents":[{"commit":"647371a1c8dfd7708369181709037276a77da6f2","subject":"smsr: enable analyze command to open a db file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/647371a1c8dfd7708369181709037276a77da6f2"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-05-14 07:42:44.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-05-14 15:23:18.000000000","tz":60},"subject":"WIP swift-manage-shard-ranges: add shard population analysis","message":"WIP swift-manage-shard-ranges: add shard population analysis\n\nTODO: needs tests\n\nChange-Id: I7d8a5d9618d57b56faa5578e159e081bf4621789\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c52e28f97183afdb6f8696e1cfb07b91bac8cee5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c52e28f97183afdb6f8696e1cfb07b91bac8cee5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
