)]}'
{"id":"openstack%2Fswift~791559","triplet_id":"openstack%2Fswift~master~If0d36145dd59047c5bfc7e92879e3bbf55447331","project":"openstack/swift","branch":"master","topic":"p-sharder-missing-overlaps","hashtags":[],"change_id":"If0d36145dd59047c5bfc7e92879e3bbf55447331","subject":"Improve find_overlapping_ranges performance","status":"NEW","created":"2021-05-14 21:31:34.000000000","updated":"2021-05-14 22:31:01.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"2cc8043b3d0df0f1756033ec0a72694145000010","_number":791559,"virtual_id_number":791559,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2021-05-14 22:31:01.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-14 22:31:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"943936dbec656e4ac339eb6dbad2983ac7f87679","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-05-14 21:31:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2cc8043b3d0df0f1756033ec0a72694145000010","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-14 22:31:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/939c9d6ebdc949d0a63f26e12c55f50d : SUCCESS in 9m 35s\n- grenade https://zuul.opendev.org/t/openstack/build/47c79166be4b4f92a590065c3f9e6d26 : SUCCESS in 54m 17s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/73e0857f532341fd9f4fdb07ce886eb9 : SUCCESS in 39m 48s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d5c0c08dfe044a918f0083660409b57c : SUCCESS in 8m 44s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/da0acaf4d4f1498b9ce94b32210f3ca5 : SUCCESS in 7m 16s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/da2c209678c74310b33a6a6ea4672341 : SUCCESS in 16m 16s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/67607aad1736443884fd53541546413f : SUCCESS in 15m 27s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/0c7baf0d6b1c48c4b3982ccd4af37eb9 : SUCCESS in 16m 01s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/9d57fa7004df4c25a2c309539562a0dc : SUCCESS in 15m 52s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/eeb952c8a10444cab4bde2d2e6caffde : SUCCESS in 15m 54s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/7fc5d593f2ec453a96e9be7acdc331f9 : SUCCESS in 16m 13s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/2f63f8b796d54e47a30e4ca8c35e21c9 : SUCCESS in 16m 04s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/0863dd678de54d119b0bc3b93aca416e : SUCCESS in 16m 44s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/8900e788179842b087f57a3c23654dfb : SUCCESS in 16m 52s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/00e130cf35a14134af32d31a5c7ecfba : SUCCESS in 17m 12s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/c3e37197f70d4370b81cdc2d5fb6b9af : SUCCESS in 18m 39s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/cbb2bee7702746ed8147c4c6209b4e9e : SUCCESS in 7m 03s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/cab511d27e114101b29e37ed3ca13232 : SUCCESS in 19m 11s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/6d345b0c30af4060978a5cd5658b95b9 : SUCCESS in 58m 24s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/9652373fcc8340b9a3cf6c54619cb3e0 : SUCCESS in 57m 14s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/5db818cd1d3f45bb99a533427f97f7f6 : SUCCESS in 36m 48s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ce513d24fe7f46bb8eeb12600bc59dc5 : SUCCESS in 36m 35s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fe896941f63e4b249b4167ff0a49788d : SUCCESS in 15m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/64f2eb32274d4361a8e9cb92466d13d6 : SUCCESS in 5m 40s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/fe2c7bbcd76146b88e4706656edd5992 : SUCCESS in 22m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e061cf1f5aba4ce9b341061961ce09e0 : SUCCESS in 55m 27s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7641388643973d673c37b5f30e514be3e23d5350","revisions":{"7641388643973d673c37b5f30e514be3e23d5350":{"kind":"REWORK","_number":1,"created":"2021-05-14 21:31:34.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/59/791559/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/59/791559/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/791559/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/791559/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/791559/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/59/791559/1"}}},"commit":{"parents":[{"commit":"1b183f221d258819c82ee58961afc765529314ac","subject":"sharder: report perfectly overlapping shard ranges","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1b183f221d258819c82ee58961afc765529314ac"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-05-14 21:31:08.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-05-14 21:31:08.000000000","tz":-420},"subject":"Improve find_overlapping_ranges performance","message":"Improve find_overlapping_ranges performance\n\n...at least, in the common case.\n\nChange-Id: If0d36145dd59047c5bfc7e92879e3bbf55447331\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7641388643973d673c37b5f30e514be3e23d5350"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7641388643973d673c37b5f30e514be3e23d5350"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
