)]}'
{"id":"openstack%2Fswift~792307","triplet_id":"openstack%2Fswift~master~Ib6eb866ddbd9edb6139996d0361cd40f415a1e89","project":"openstack/swift","branch":"master","hashtags":[],"change_id":"Ib6eb866ddbd9edb6139996d0361cd40f415a1e89","subject":"DNM: make sure we really can turn off strict locks","status":"ABANDONED","created":"2021-05-20 00:52:23.000000000","updated":"2021-05-25 22:39:47.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"f7f540d1ab351374db185d14982ff6013d8d8e2e","_number":792307,"virtual_id_number":792307,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-05-20 02:30:03.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-20 02:30:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"32be96e19d91656ec78ede376b730dd3790b83d1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-05-20 00:52:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e2d26b70c0b876a4577dd937709298c3df51e8e2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-20 02:30:03.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c1dd4cd7c9ce4bb09b002c0755c03f87 : SUCCESS in 7m 55s\n- grenade https://zuul.opendev.org/t/openstack/build/5bd915e8bd254ad290ad9a0555126eae : FAILURE in 1h 04m 56s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a9e03c63552b4e1e936097db9bfcb44d : FAILURE in 21m 26s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/259b1d3f652b45a291b721b4563c6312 : SUCCESS in 8m 25s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/08ad9bf60e974cad83553c432f487ebc : SUCCESS in 5m 46s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/f68cc8c0f91c462cb5120f72ebd19c27 : TIMED_OUT in 1h 31m 06s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/3cc31996ee84439da76eec818e5d3a68 : TIMED_OUT in 1h 30m 57s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/7454225bc29e41f09f48191959265079 : TIMED_OUT in 1h 31m 00s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/bc75783023fd4bca999379d4d3d1922a : TIMED_OUT in 1h 31m 08s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/c0e63598f07d4768a8f088ea752a7040 : TIMED_OUT in 1h 31m 51s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/c3dd8afe95a642618c9bb5b654f19339 : FAILURE in 5m 26s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/2e56167926a9470caf05e15618adfb7c : FAILURE in 4m 39s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/269f80c00a154906896171f7ac8eeff9 : FAILURE in 4m 37s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/1ce078694abf4946986ca1009bf3c8cb : FAILURE in 6m 44s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/1f9a010fd6c640809c974aa087fddfb7 : FAILURE in 4m 34s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/8c9f068778aa48708ab32c0e0671da9a : FAILURE in 5m 12s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/7fa38ad9d10b481cbe9fc17b2763241a : SUCCESS in 7m 19s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/3b63be7233434e4998a43d31e740605c : SUCCESS in 19m 34s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/facb3a6150f04ef8938f1d14badb3ef1 : FAILURE in 54m 16s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/e891fab4912f441abdd42ddcb0b1557d : FAILURE in 49m 08s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/7f5500513ab94b8084093e456372278e : TIMED_OUT in 1h 32m 06s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/36d6aa80233f49ec98a2b05c2dec5dbe : TIMED_OUT in 1h 32m 08s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c0332b583ace453f9bdaade633ca4e36 : TIMED_OUT in 40m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4d35557c81154aee9dfc98515e9c963c : SUCCESS in 6m 38s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/609e7c936a224ef7b60c3867622826e0 : TIMED_OUT in 1h 31m 00s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4dcb52f68625446e9a99cfcef791649a : FAILURE in 30m 54s","accounts_in_message":[],"_revision_number":1},{"id":"f7f540d1ab351374db185d14982ff6013d8d8e2e","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-05-25 22:39:47.000000000","message":"Abandoned\n\nI think it served its purpose -- we could at least get a decent number of probe tests passing... looks like it\u0027s mostly swift-manage-shard-ranges still having trouble.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"58f9881dc3b89d5e1aba2ef5f3c0a66f03f6fd46","revisions":{"58f9881dc3b89d5e1aba2ef5f3c0a66f03f6fd46":{"kind":"REWORK","_number":1,"created":"2021-05-20 00:52:23.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/07/792307/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/07/792307/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/07/792307/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/07/792307/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/07/792307/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/07/792307/1"}}},"commit":{"parents":[{"commit":"3bb0ea26900cd972e547dc3b830a729f0264a5ef","subject":"lock_path: Prevent aquiring deleted/moved locks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3bb0ea26900cd972e547dc3b830a729f0264a5ef"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-05-20 00:51:24.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-05-20 00:51:24.000000000","tz":-420},"subject":"DNM: make sure we really can turn off strict locks","message":"DNM: make sure we really can turn off strict locks\n\nChange-Id: Ib6eb866ddbd9edb6139996d0361cd40f415a1e89\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/58f9881dc3b89d5e1aba2ef5f3c0a66f03f6fd46"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/58f9881dc3b89d5e1aba2ef5f3c0a66f03f6fd46"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
