)]}'
{"id":"openstack%2Fswift~792405","triplet_id":"openstack%2Fswift~master~Ia536c09534ccadf26ead5eea08bab075156e49e0","project":"openstack/swift","branch":"master","topic":"malei","hashtags":[],"change_id":"Ia536c09534ccadf26ead5eea08bab075156e49e0","subject":"Use a less bogus credit for Melissa Ma Lei","status":"MERGED","created":"2021-05-20 15:58:26.000000000","updated":"2021-06-01 06:53:05.000000000","submitted":"2021-06-01 06:49:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"792405-malei","meta_rev_id":"a3e69730063538c87ec3f25fa7a19612a5155703","_number":792405,"virtual_id_number":792405,"owner":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-01 06:49:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-06-01 05:04:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-06-01 05:04:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-20 15:59:16.000000000","updated_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"reviewer":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"state":"CC"},{"updated":"2021-05-20 16:17:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-01 05:04:12.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"CC"},{"updated":"2021-06-01 05:04:21.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"93d51d8935b0a634dfac48e060ff147793e45b75","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2021-05-20 15:58:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"42f57df5a1ff3c46f51eead0f42bc7bb7c54d7a7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-20 16:17:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8406e8b2a822465fa6a9b060291473ad : SUCCESS in 9m 40s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d97421e7cfb74e5ea09aa9284c7c1212 : SUCCESS in 10m 32s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/4482d10ae9b744b78d26102950a4c05f : SUCCESS in 7m 38s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/b268efa7b44440a79829bc456ad9d7c5 : SUCCESS in 17m 01s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/233dacc7348546688c82de53581b40e8 : SUCCESS in 17m 08s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/ae28590705f64e628fcd705af62d57fc : SUCCESS in 14m 18s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/569c61f268e148b19414261c033de607 : SUCCESS in 13m 18s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/77605dfe73f24ae7be24f930000be220 : SUCCESS in 16m 04s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/69e574f8d0384588b95ba8d547896a07 : SUCCESS in 14m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/700ef47442bb4909b2c1f1a691c1ab9b : SUCCESS in 7m 38s","accounts_in_message":[],"_revision_number":1},{"id":"b2798048a74815eadfb0017f94d633ff4ef23ea6","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-06-01 05:04:12.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"4eba6402e020028cecf96a6ccf6537ab3bc1f6dd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-06-01 05:04:21.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"9751e818dc2aefc14780c93ec9368468c9616308","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-06-01 05:04:50.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"93c610a0c09fb33f25e6b415361432f9142c0727","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-01 05:22:24.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df6a63a567b049d0a06baad52b5ae6c6 : SUCCESS in 11m 04s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/9463a83c3bf84918b9ba6c014bbbb285 : SUCCESS in 9m 31s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/f41b6eae765f41318961fdfce1ba3923 : SUCCESS in 11m 56s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/145c27b38c7b42e88869f2456a1d1d0f : SUCCESS in 14m 38s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/342d9258da6540e0ba796760ef812d39 : SUCCESS in 15m 37s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/24dce35875e145fc839ecd3654cb7e84 : SUCCESS in 14m 54s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/9286cea4a8f34edf8ca464ff7e57fcd2 : SUCCESS in 14m 27s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/00422af472ef46bf8208e1e64b00340d : SUCCESS in 15m 57s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/690d24da18454a6e8dfc22bae2aa9fa1 : SUCCESS in 16m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/449d9d1a7be743e381b3422892ce0ad4 : SUCCESS in 6m 57s","accounts_in_message":[],"_revision_number":2},{"id":"356d62e26886298cb607cc36976c3f1b2e0fb542","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-01 05:22:41.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"c3411ebc2a4e16a177afeef7a0f84032d56b19e1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-01 06:49:38.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e4fff220dfe47f181319053958950b2 : SUCCESS in 9m 15s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/2f547b595e3346bbbc9f7416bcfd70c3 : SUCCESS in 10m 54s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/2df6e41b6a5c4f38b7bb7ded447fac55 : SUCCESS in 9m 51s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/d2e85432fa5b466580b97325cb534b88 : SUCCESS in 14m 53s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/5714f6f5e37446968a682b888c909c6a : SUCCESS in 24m 24s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/0628a87360e54e3caf9bfea167111c83 : SUCCESS in 15m 47s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/fee91046cf944d51abba275ad6bbeb67 : SUCCESS in 15m 54s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/b3af5498f7a94df087946408abe0064f : SUCCESS in 16m 20s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/c8d8a6e9d3c34ac5a2824ee17a0608dc : SUCCESS in 21m 50s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/0d458310c79446a4b1cfdad35b35ef44 : SUCCESS in 40m 18s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/f887d95f96c344f280922a718499448d : SUCCESS in 17m 36s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/40fbae70a26142dd8b65d91ece56cd6d : SUCCESS in 17m 26s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/dd0fce8ba4cc4af398280a8e47cf2969 : SUCCESS in 22m 51s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/99a93797546c4c73be4a00ba0814ead6 : SUCCESS in 25m 06s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/363dccaa8c6d4f9cbcf7fa4786a3aecc : SUCCESS in 6m 55s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/264132060d46401cbbd634cf434a5c03 : SUCCESS in 14m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dc34239292e84c398dda460fda267c0b : SUCCESS in 5m 39s","accounts_in_message":[],"_revision_number":2},{"id":"be3a836d3afb6b012baf733953e919ffd5936fe7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-01 06:49:44.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"a3e69730063538c87ec3f25fa7a19612a5155703","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-01 06:53:05.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b5563e57c7354eccae8204420f0a12a0 : SUCCESS in 1m 13s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/4d6c99bd1b5641369a5aa5a3a0f83c9d : SUCCESS in 49s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/dd553d4c45f641ac9b987c5956e2653c : SUCCESS in 49s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"79b263f5e929e797c3554fb401c797470f0779cf","revisions":{"85bb52d724bad689a0afa58799ad2010033ff998":{"kind":"REWORK","_number":1,"created":"2021-05-20 15:58:26.000000000","uploader":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"ref":"refs/changes/05/792405/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/05/792405/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/792405/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/792405/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/792405/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/05/792405/1"}}},"commit":{"parents":[{"commit":"8cc1ef92558bd3f7c345c955f386498b936bace1","subject":"Merge \"relinker: Add policy to relinker progress output\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8cc1ef92558bd3f7c345c955f386498b936bace1"}]}],"author":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2021-03-25 14:26:48.000000000","tz":-300},"committer":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2021-03-25 14:26:48.000000000","tz":-300},"subject":"Use a less bogus credit for Melissa Ma Lei","message":"Use a less bogus credit for Melissa Ma Lei\n\nChange-Id: Ia536c09534ccadf26ead5eea08bab075156e49e0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/85bb52d724bad689a0afa58799ad2010033ff998"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/85bb52d724bad689a0afa58799ad2010033ff998"}]},"branch":"refs/heads/master"},"79b263f5e929e797c3554fb401c797470f0779cf":{"kind":"REWORK","_number":2,"created":"2021-06-01 05:04:21.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/05/792405/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/05/792405/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/792405/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/792405/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/792405/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/05/792405/2"}}},"commit":{"parents":[{"commit":"8cc1ef92558bd3f7c345c955f386498b936bace1","subject":"Merge \"relinker: Add policy to relinker progress output\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8cc1ef92558bd3f7c345c955f386498b936bace1"}]}],"author":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2021-03-25 14:26:48.000000000","tz":-300},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-06-01 05:03:08.000000000","tz":-420},"subject":"Use a less bogus credit for Melissa Ma Lei","message":"Use a less bogus credit for Melissa Ma Lei\n\nChange-Id: Ia536c09534ccadf26ead5eea08bab075156e49e0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/79b263f5e929e797c3554fb401c797470f0779cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/79b263f5e929e797c3554fb401c797470f0779cf"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
