)]}'
{"id":"openstack%2Fswift~792485","triplet_id":"openstack%2Fswift~master~Ic2f4487b6caf81fef3455ce03dda2cc144ae24ec","project":"openstack/swift","branch":"master","hashtags":[],"change_id":"Ic2f4487b6caf81fef3455ce03dda2cc144ae24ec","subject":"symlink: Ensure headers are returned as a list on py3","status":"MERGED","created":"2021-05-20 22:33:24.000000000","updated":"2021-05-21 09:52:40.000000000","submitted":"2021-05-21 09:50:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"792485","meta_rev_id":"900a9c2390049fb8e559ed661209471be8343f78","_number":792485,"virtual_id_number":792485,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-05-21 09:50:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"all":[{"value":2,"date":"2021-05-21 08:14:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"all":[{"value":1,"date":"2021-05-21 08:14:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-20 23:49:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-21 08:14:11.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"REVIEWER"}],"messages":[{"id":"40c351c07eed80024f2a614d368d5955159a80b9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-05-20 22:33:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"27b744ae22d0745de6e97db922960655f4e4b1d7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-20 23:49:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/424fc7a86d074ad9b1abab86a6ca56dc : SUCCESS in 8m 45s\n- grenade https://zuul.opendev.org/t/openstack/build/80f6bcef91b24434991c4007d9a51528 : SUCCESS in 1h 10m 29s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/17a935ad1e714d11a0eac5763e62b366 : SUCCESS in 31m 24s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f098ca0058b84b0cba256af96c5ee9bf : SUCCESS in 9m 55s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/82530b8547934355859d4c823f2c31ef : SUCCESS in 6m 32s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/0f6d26c03395419fbf749ee9d895a4ef : SUCCESS in 15m 33s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/e93a240ff6ec41cbaa67b806931f9bf7 : SUCCESS in 14m 42s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/23951c7f809a456a9d71e75d5a41909e : SUCCESS in 15m 54s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/684c862df3cc4aadb80f68fdcc32e31c : SUCCESS in 15m 21s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/f9cb5b48b6aa4db2a2035efae05a914a : SUCCESS in 17m 51s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/3abca8a9aefc4a4f9a2c257322112d96 : SUCCESS in 16m 23s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/eff989570b4e40d484a1b8361b872245 : SUCCESS in 18m 06s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/4f2faecf00c84e55b7425a4b3b78eb57 : SUCCESS in 17m 37s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/fb4d10e126814e9c874b83c39921d5c3 : SUCCESS in 18m 12s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/d9d281174f86460e8aa3b9cd3bc85e99 : SUCCESS in 17m 14s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/99c953fcbac34c18b7802e8671330bb9 : SUCCESS in 17m 46s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/27ee42acfaf04c71b8774bfbca9e1079 : SUCCESS in 6m 55s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/7710dbcadd3d4f4290c0289ce8d90701 : SUCCESS in 20m 05s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/d14f1311a84c460597be572c2ebe8588 : SUCCESS in 43m 28s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/066f3f9e6ae54578aedcb8cf591cf6d9 : SUCCESS in 49m 50s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/01d41979c03a4226989965f2d36bd651 : SUCCESS in 31m 52s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/1b69b3b356be40feb793a15c0865708f : SUCCESS in 41m 45s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3cdbd8963da44633908d43844c543514 : SUCCESS in 14m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/73a47b4e87ef46a5bbc142fc0c81a0ec : SUCCESS in 6m 00s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a8383ec29fa24fde8369c6c315d38079 : SUCCESS in 24m 08s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/41ba1c25ee8d45b7a8238e66eb253ea6 : SUCCESS in 46m 29s","accounts_in_message":[],"_revision_number":1},{"id":"f4b7d6f1d45b37c11ffa3f5a72712f9ed2de7bac","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2021-05-21 08:14:11.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nfun of py3 list like objects.","accounts_in_message":[],"_revision_number":1},{"id":"c27fbbd0a86a0d91d1cb4fb8f446531d8766d8ca","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-21 08:14:37.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fa63d7d6845ed5f7bdb7ca15954cb240327fa745","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-21 09:50:27.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ba6b21df8eba4facaf5011db978c4cfc : SUCCESS in 11m 21s\n- grenade https://zuul.opendev.org/t/openstack/build/b4d377d8a9364d1f83378aea18bf2a11 : SUCCESS in 54m 33s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/7079020eaa294cca8824b750318ea052 : SUCCESS in 36m 02s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/05ed130779864c33b7e9424be8ba949a : SUCCESS in 13m 18s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/a163f906bf2d4148bdb31d956f964186 : SUCCESS in 11m 14s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/5f5aa8470d6146a9a5e836b7aa1c1a08 : SUCCESS in 16m 11s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d5c3375d75e347cc96652f39c6353c56 : SUCCESS in 20m 04s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/236841ffc69d40f9bd901be479fc0e7b : SUCCESS in 18m 16s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/82cc0faa43ce4ad9b68f8e804a049b8c : SUCCESS in 16m 02s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/79150f6c35be4d74bf1e8d5c754c7a00 : SUCCESS in 19m 33s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/cc08d4dbb346493391cc9f9aa19fcba3 : SUCCESS in 38m 02s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/3ee7985defea4b88a4c7e5ea647c1df6 : SUCCESS in 20m 13s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/697a73542fb54534ae06c98ad3ae9171 : SUCCESS in 22m 09s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/ebbc873de61144f0ae967daa60c9fec2 : SUCCESS in 19m 46s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/57d05a9803504717b87a0eb964a81a75 : SUCCESS in 19m 25s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/0a4fbc6a449d4635ada96b395391fd62 : SUCCESS in 20m 49s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/da9b39a8db3d4c00a711127551c03560 : SUCCESS in 7m 48s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/c4d3fc3b3a8b4a27b0e267de1440237a : SUCCESS in 48m 18s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/2c4591b559e34f2db99705c11ca9f241 : SUCCESS in 52m 22s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f365eaced01c4e29b4ad2ed9a2fbc807 : SUCCESS in 36m 24s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/5e10312b50a744da9a8ac279e3cffd27 : SUCCESS in 47m 42s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/75bfcf4a1052414ea27842da06d54515 : SUCCESS in 17m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/24b3e0df7f8e426ab374887a3f1cea34 : SUCCESS in 7m 45s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/dda86461b5e14814b0aafa59af90e79a : SUCCESS in 32m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/70f1b5165d17489d9ded8afe8bfbb036 : SUCCESS in 50m 17s","accounts_in_message":[],"_revision_number":1},{"id":"21d3fc083cd9bbb4b455e59b1a7b21dd3c101373","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-21 09:50:33.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"900a9c2390049fb8e559ed661209471be8343f78","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-21 09:52:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/20c506b843ce4caf889ce7b238a04c87 : SUCCESS in 1m 07s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/fc418e676d4d4dac82913467738de7d1 : SUCCESS in 58s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/c790d009f4f548c6964ee2c8fdbd5175 : SUCCESS in 53s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6bfd93d88644f0a7ff979dc9d6c3d85fff42f632","revisions":{"6bfd93d88644f0a7ff979dc9d6c3d85fff42f632":{"kind":"REWORK","_number":1,"created":"2021-05-20 22:33:24.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/85/792485/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/85/792485/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/792485/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/792485/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/792485/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/85/792485/1"}}},"commit":{"parents":[{"commit":"0a58574109e0e329ddb6ee78b7ed82b3a9c42c61","subject":"Merge \"sharder: report perfectly overlapping shard ranges\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0a58574109e0e329ddb6ee78b7ed82b3a9c42c61"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-05-20 22:32:50.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-05-20 22:32:50.000000000","tz":-420},"subject":"symlink: Ensure headers are returned as a list on py3","message":"symlink: Ensure headers are returned as a list on py3\n\nChange-Id: Ic2f4487b6caf81fef3455ce03dda2cc144ae24ec\nRelated-Bug: #1929083\nCo-Authored-By: Walter Doekes \u003cwalter+github@wjd.nu\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6bfd93d88644f0a7ff979dc9d6c3d85fff42f632"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6bfd93d88644f0a7ff979dc9d6c3d85fff42f632"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
