)]}'
{"id":"openstack%2Fswift~792657","triplet_id":"openstack%2Fswift~stable%2Fvictoria~I5c398b6008716b64c668737e4201ba3b6ab3320b","project":"openstack/swift","branch":"stable/victoria","topic":"bug/1929083-stable/wallaby-stable/victoria","hashtags":[],"change_id":"I5c398b6008716b64c668737e4201ba3b6ab3320b","subject":"tempurl: Fix PUT upload to temp url on py3","status":"MERGED","created":"2021-05-21 18:41:23.000000000","updated":"2021-05-26 04:41:40.000000000","submitted":"2021-05-26 04:38:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"792657-bug/1929083-stable/wallaby-stable/victoria","cherry_pick_of_change":792655,"cherry_pick_of_patch_set":1,"meta_rev_id":"0b6de0fa096846223f01d70111520e32cfa9a0ef","_number":792657,"virtual_id_number":792657,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31241,"name":"wdoekes","email":"walter+ubuntu@wjd.nu","username":"wdoekes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-05-26 04:38:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"recommended":{"_account_id":31241,"name":"wdoekes","email":"walter+ubuntu@wjd.nu","username":"wdoekes"},"all":[{"value":1,"date":"2021-05-22 20:17:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31241,"name":"wdoekes","email":"walter+ubuntu@wjd.nu","username":"wdoekes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-05-25 22:23:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":31241,"name":"wdoekes","email":"walter+ubuntu@wjd.nu","username":"wdoekes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-05-25 22:23:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31241,"name":"wdoekes","email":"walter+ubuntu@wjd.nu","username":"wdoekes"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-21 18:41:23.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":31241,"name":"wdoekes","email":"walter+ubuntu@wjd.nu","username":"wdoekes"},"state":"REVIEWER"},{"updated":"2021-05-21 20:01:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b32d52ca6658c67de35a44b420f9601840c8d7cf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-05-21 18:41:23.000000000","message":"Patch Set 1: Cherry Picked from branch stable/wallaby.","accounts_in_message":[],"_revision_number":1},{"id":"04667ab688775705e0d6fd607d95adf6362ab76c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-21 20:01:41.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a6a0e3d0afc1494eab82effc6f531c10 : FAILURE in 8m 10s\n- grenade https://zuul.opendev.org/t/openstack/build/54aa71fe35e84d6f8d87bed007f71f1d : SUCCESS in 1h 15m 18s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/7800ebc5c6ae4062b6712457090e224a : SUCCESS in 53m 29s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ee4cd84c00f1443d9c08fe3a93cbf4b0 : SUCCESS in 9m 12s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/aa9059ae32e74fa5832b7c0ce0cb4767 : SUCCESS in 6m 19s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/4f3b2d94ebff4f1b8e7108f33ae1cbef : SUCCESS in 14m 15s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/f905da49eb6d4043ab6335627049d16b : SUCCESS in 13m 54s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/4a7efa7d583246899c6552181ebc4610 : SUCCESS in 14m 39s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/f995235f2b664ff3b567d407e8a437c3 : SUCCESS in 15m 35s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/49379ee3caf74fe198daa47a0dbb6394 : SUCCESS in 20m 27s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/19ac5cd7e04e4af2a37a30118159f408 : SUCCESS in 20m 57s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/f19c139dfbb44e9c97dce8ccfea76419 : SUCCESS in 19m 51s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/c1cb3d36c2db4b60bc4e03377d343182 : SUCCESS in 17m 37s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/834133c23bfb4393845bf4c3e2593216 : SUCCESS in 17m 57s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/dbb5d9093dd940d9bf943f3199708a67 : SUCCESS in 19m 29s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/5620c9798fec45578cedab14c27d6996 : SUCCESS in 21m 11s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/978fac32c6ac4fc48ada64f2b0277863 : SUCCESS in 51m 00s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ac3f8a785a954ee6a4838651be169483 : SUCCESS in 51m 24s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/7ccf0082815248649cb8e05a6278daa2 : SUCCESS in 45m 25s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b0ca4894fb3148099e53750f0d4c76fa : SUCCESS in 13m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7f75bb4cfb6446df91260af9a1c80dbb : SUCCESS in 7m 24s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b0b548eed2084556b5e41c39fedc2c29 : FAILURE in 10m 34s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/117526ef181b483590b922d2dcc82db2 : SUCCESS in 1h 01m 37s","accounts_in_message":[],"_revision_number":1},{"id":"dcf84192253bf4c4ea88c1ec95de755849dd9635","author":{"_account_id":31241,"name":"wdoekes","email":"walter+ubuntu@wjd.nu","username":"wdoekes"},"date":"2021-05-22 20:17:22.000000000","message":"Patch Set 1: Code-Review+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"e899b5da92bdb5473d8803414d869d0c804344ba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-22 21:23:03.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a450fefed4ec4ce3b9763306441d6caa : FAILURE in 8m 10s\n- grenade https://zuul.opendev.org/t/openstack/build/9ee7a77c13904fd3a4272a6004c492c4 : SUCCESS in 54m 00s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/0e6ba3452a6b40bb86404cb355e60a8a : SUCCESS in 43m 32s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/594cc8b092e54855855484edc218eeb4 : SUCCESS in 8m 20s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/bd0452bba28c4b2c877db681c4f5e299 : SUCCESS in 6m 05s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/a423678b44bb42e5ba973a07f8d9a978 : SUCCESS in 13m 38s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/3cd59f1da12848398f551f7c385f5942 : SUCCESS in 14m 21s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/899927353c914176b89ef10a51611684 : SUCCESS in 14m 03s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/f4b4203dda6b4df3a9505265ded1c629 : SUCCESS in 14m 37s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/0a2b2813485149aeadd1b0d473e8f317 : SUCCESS in 17m 00s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/0485228bb2434fbcbc080c145891fee4 : SUCCESS in 16m 46s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/f09114c3cd5d417980679ade1453df54 : SUCCESS in 17m 57s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/ccfdd6597f48480c8307659ed57995af : SUCCESS in 16m 56s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/c63e173070984331aab1b6dc741d174a : SUCCESS in 17m 54s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/cc5c17e868f645d2b8f4cffcfbf00dc6 : SUCCESS in 18m 19s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e6032c844b384b35bb6d14f90f2e92f7 : SUCCESS in 20m 30s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/ce93bfa4b71045628d5683ae1256835e : SUCCESS in 46m 21s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/5bdaa9ac644e48cc9c86a7f046376267 : SUCCESS in 46m 59s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/dddcd67179c14a5897389b5f9ba4d840 : SUCCESS in 41m 39s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/07ac9d8d6b86444290a4f33fa33363ce : SUCCESS in 13m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2f0412664c44488f8188acdbbacc0d9f : SUCCESS in 6m 12s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/78ca9880ecbb468eb9d1b4183b1c693e : FAILURE in 9m 27s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/085e9a94521643e194115fc969d4f9ca : SUCCESS in 1h 00m 03s","accounts_in_message":[],"_revision_number":1},{"id":"06c6ffaa3019db6f30a388f23dc8b53770c552f8","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-05-25 21:14:45.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c567b728da4ac88f0bc261aa7daeff50c6896685","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-25 22:12:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b09187038edd4403b2dc010a5825ad64 : SUCCESS in 9m 56s\n- grenade https://zuul.opendev.org/t/openstack/build/e18b1342d9d7425ab51f4629037b0948 : SUCCESS in 55m 42s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/db28cba8d5ae4b86af947088ddb2c7f6 : SUCCESS in 33m 17s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b92e3ca03fe44e14836984311c009f33 : SUCCESS in 8m 27s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/5c245aec22ec434788b5a445e6b1448c : SUCCESS in 6m 07s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/f1e63f57191e46599f2056aa246db115 : SUCCESS in 10m 52s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/1d0dd9b9e4f94bdabd70997af52ca680 : SUCCESS in 10m 08s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/0f766d7eb41b466989b5f9a158d5630a : SUCCESS in 15m 31s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/452cbad6d8a2496e8f48ac0676ca5be3 : SUCCESS in 15m 45s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/df17eefcd7694a19b34d7836bba2db1c : SUCCESS in 34m 49s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/df83a3908db142b4935f3c6a4c6c9d30 : SUCCESS in 22m 21s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/2fff9451b1184413922f25020cf3e44b : SUCCESS in 29m 08s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/b5c38bec0ed94ead8d8503a951b17a58 : SUCCESS in 13m 23s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/3b7223cb8c41407c92f6083ac202b082 : SUCCESS in 33m 18s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/d1290bdafc474f40926598b84337579f : SUCCESS in 19m 20s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/d53d2797516a41b09f6a915bbc8f6a0e : SUCCESS in 16m 47s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/657e023e2b8343169689fd50f3293504 : SUCCESS in 36m 26s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ae81e2a48c844c3784113bcca881fc21 : SUCCESS in 40m 57s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a1d3aee74fc745d79adff0a9153fcb1c : SUCCESS in 44m 51s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/330d108bf78b4777b17e269cb885da4d : SUCCESS in 14m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fc5b8563585546a1b5645251f8a47e61 : SUCCESS in 5m 17s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/30255867a77742f3bae958213ed15b85 : FAILURE in 10m 55s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/687d5911b91e4e6b9fd34b0af862b997 : SUCCESS in 53m 34s","accounts_in_message":[],"_revision_number":1},{"id":"8890fc3f489e6313718a94f4c06f67c6dcfc5010","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-05-25 22:23:49.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e948b332f6becb0d36ad2507061cb60c5d1b0a35","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-25 22:24:07.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"99e03e04ca5be3b5ef45d855a64f6d36ca615e3d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-26 04:38:09.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/959f066d74bf4abc9e9b93126020ba7f : SUCCESS in 10m 59s\n- grenade https://zuul.opendev.org/t/openstack/build/ae49e11027dd4e31a31e069f67d57226 : SUCCESS in 1h 07m 42s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/4278ecbd91d440e598b52d0c67f8fdd3 : SUCCESS in 48m 46s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/21eb0c8dbe504cacb0455ae97fb232e2 : SUCCESS in 11m 07s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/ed0f554fce2244a3b5f370f07429d322 : SUCCESS in 8m 44s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/c3cd7bedeba8450ca648d52605caabde : SUCCESS in 15m 16s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a45ddf3a441a44aeb63e1b3d6cba325a : SUCCESS in 16m 16s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/a957b1a39ebf4a7abe765f0b1e0af983 : SUCCESS in 16m 22s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/1568ce5d79a84bf9ab1efb5913e0ee1d : SUCCESS in 16m 12s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/371b7ba67340485f8c677c4437c474a8 : SUCCESS in 25m 29s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/ea9a0762747d4accba7684b74d6a00b9 : SUCCESS in 30m 48s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/050b86740f5b4420b236a213ebf207b4 : SUCCESS in 19m 45s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/c46f76ef6b114504bd5304af60c1f1c0 : SUCCESS in 29m 46s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/735498c20fba45a1b480d82e83ec20dc : SUCCESS in 20m 30s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/891d8aa9be7c47ea9e2b4f76ee748468 : SUCCESS in 49m 43s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e3fbb6a28d9a4938a818a50994a3029b : SUCCESS in 47m 00s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ce649cf0850e450eb95c80e370927c91 : SUCCESS in 44m 36s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/40726780032946b59aaf64a5938a5400 : SUCCESS in 15m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/54c12f4444a0458689fcd529a27f4570 : SUCCESS in 7m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0a41a440c170428ca1442450b8dda13c : SUCCESS in 1h 05m 16s","accounts_in_message":[],"_revision_number":1},{"id":"95792351e606170659ecae0fa65a1cb6af2c7798","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-26 04:38:09.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"0b6de0fa096846223f01d70111520e32cfa9a0ef","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-26 04:41:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be5b3983ad544d9cbfbf28a897258183 : SUCCESS in 1m 07s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/e6fb9b188f6a4f759b003a2b86df79a0 : SUCCESS in 36s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/6149c3965bf2422ebd39d68c64087e9a : SUCCESS in 38s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a8c2b46da624a2e133e5d6289529ba8ac63b2557","revisions":{"a8c2b46da624a2e133e5d6289529ba8ac63b2557":{"kind":"REWORK","_number":1,"created":"2021-05-21 18:41:23.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/57/792657/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/57/792657/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/792657/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/792657/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/792657/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/57/792657/1"}}},"commit":{"parents":[{"commit":"bdb4510e05863942d0452db6e1d399b1921eeff7","subject":"Merge \"Avoid loops when gathering container listings from shards\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bdb4510e05863942d0452db6e1d399b1921eeff7"}]}],"author":{"name":"Walter Doekes","email":"walter+github@wjd.nu","date":"2021-05-20 15:06:25.000000000","tz":120},"committer":{"name":"Tim Burke","email":"tburke@nvidia.com","date":"2021-05-21 18:41:23.000000000","tz":0},"subject":"tempurl: Fix PUT upload to temp url on py3","message":"tempurl: Fix PUT upload to temp url on py3\n\nWithout this change, the self._response_headers gets a dict_items() type\non PUT. But the rest of the code assumes that it is a list.\n\nBug manifestation:\n\n    File \"swift/common/middleware/catch_errors.py\", line 120,\n        in handle_request\n      self._response_headers.append((\u0027X-Trans-Id\u0027, trans_id))\n  AttributeError: \u0027dict_items\u0027 object has no attribute \u0027append\u0027\n\nCloses-Bug: #1929083\nChange-Id: I5c398b6008716b64c668737e4201ba3b6ab3320b\n(cherry picked from commit 77530136f13f1fc0d8625d43e1689427d4ee2fad)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a8c2b46da624a2e133e5d6289529ba8ac63b2557"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a8c2b46da624a2e133e5d6289529ba8ac63b2557"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
