)]}'
{"id":"openstack%2Fswift~793018","triplet_id":"openstack%2Fswift~stable%2Fvictoria~I599d4ea430cc1e65bb50d0481617b005febf6de2","project":"openstack/swift","branch":"stable/victoria","hashtags":[],"change_id":"I599d4ea430cc1e65bb50d0481617b005febf6de2","subject":"Dep\u0027s should be restricted by upper-constraints","status":"MERGED","created":"2021-05-25 17:51:30.000000000","updated":"2021-05-25 21:15:46.000000000","submitted":"2021-05-25 21:14:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"793018","meta_rev_id":"362970d6f75b5f14bcfe28aed7e3f8b1cc8c5094","_number":793018,"virtual_id_number":793018,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-05-25 21:14:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-05-25 19:51:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-05-25 19:51:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"}],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-25 17:51:30.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"state":"REVIEWER"},{"updated":"2021-05-25 19:33:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4f11692aa0c8ebdf6a40cd7b4704f57c25d9e0f1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-05-25 17:51:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2910319adb3505e06c3a90f54b32ccb38030299a","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-05-25 18:17:21.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5c1dd4e0c075c36366786724e4ea623772d110c9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-05-25 18:18:39.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"f66d9b2a73845fb0813024b2cfd1c0b3ecd49198","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-05-25 18:25:17.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"7f4377fa7ef3f4414dcbad1ba890dca2553d1e8d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-25 19:33:39.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc59c3cbce9642e097796c77ac8c4abc : SUCCESS in 9m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1314f84939b341ac9d00b6f4cfc38b60 : SUCCESS in 6m 14s\n- grenade https://zuul.opendev.org/t/openstack/build/99f39c721dd8498f91e2c09becd9dbc2 : SUCCESS in 1h 02m 48s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/40b47bb2779649078cc955100ff479eb : SUCCESS in 42m 55s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/1af509f4200c4f0e83238a64557a5308 : FAILURE in 7m 36s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/279d574b290b4c1abd4302406ca7ef0e : SUCCESS in 6m 46s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/30f781374f5b49bc9bf3084598d0069e : SUCCESS in 16m 38s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/19c9121e7b91471a92f06d3160a4fcc2 : SUCCESS in 17m 46s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/da31dcc4e806448f9fb1596b0ed2cb35 : SUCCESS in 17m 33s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/e68ea1bcaa624e418b667f8cd5811ac3 : SUCCESS in 17m 39s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/76ea039567f243d0ac355d12e03503d0 : SUCCESS in 25m 11s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/d1195befe0164d62ad2e130970662ad2 : SUCCESS in 17m 57s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/36bd13eebc7c48cdba63887774454b85 : SUCCESS in 25m 43s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/9c1b3ca9e1ad45b890d69ad2cd0d38ca : SUCCESS in 18m 07s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/110706d4b1f941338629f1440e6e074f : SUCCESS in 26m 40s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/674597a534654243a453c59b5fd72a31 : SUCCESS in 28m 26s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/642c521143c44a6aaef9c1865cf4a843 : SUCCESS in 25m 47s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/e1f143e69c5b470cbbd0d9d04988eae0 : SUCCESS in 48m 44s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4015a6aa93da491fb2660f59d166150e : SUCCESS in 43m 43s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/22c4e31c392e40d2835f3e06008d3e57 : SUCCESS in 42m 13s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/53057e3eb96040f297f4f9136dbb57b6 : SUCCESS in 16m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/51f1032d6778455691dbee41cfe7ca27 : SUCCESS in 6m 37s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/1515feffd24a4e3290f39af4c5ba2f9f : FAILURE in 11m 46s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3e676afcdd0e4f2ea0ef9b64d1ce2dae : SUCCESS in 56m 51s","accounts_in_message":[],"_revision_number":3},{"id":"723bb6d51734bd04584c42d657926201b9786932","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-05-25 19:51:24.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\nWooo! Working gate again!","accounts_in_message":[],"_revision_number":3},{"id":"12d17982979e464c9d4b5ff67cdb9e168a17dd77","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-25 19:51:43.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"1a78b1e3d1212313ead0012709efbb506e4afd8b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-25 21:14:04.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0bd8bd75da414b5883a06c338a4860d6 : SUCCESS in 8m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/aef7f98253e74cbcacb0f2d9e8192790 : SUCCESS in 5m 55s\n- grenade https://zuul.opendev.org/t/openstack/build/7e70e04b99054215a59592ba8eaaa351 : SUCCESS in 1h 01m 12s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/7c90251eafbf458dbdb441b0b0e21092 : SUCCESS in 45m 27s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/91b925439bed4ce194b7b667fd80a68d : SUCCESS in 9m 41s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/59f7de9cc36c475694024ba030c20fd7 : SUCCESS in 7m 31s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/4b0983063c2346e78cdd0a7cecec8362 : SUCCESS in 13m 59s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/9a3e252089d24cf2b0fb2d74a17b322e : SUCCESS in 19m 10s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/d53d4b5737df470cb085a5b6b9914fba : SUCCESS in 14m 16s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/7e785a602f2d4a33b905b68c6c0da019 : SUCCESS in 16m 51s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/bce5c074f4bc42f6b2ac29a9eaa83387 : SUCCESS in 23m 07s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/34a6992127de46039df7d645dbe406a4 : SUCCESS in 35m 19s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/64894785f94e4e0eb84f64b94c215fa4 : SUCCESS in 30m 32s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/691a4f9c68a345dcb2a72782b10cb88f : SUCCESS in 24m 03s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/b9e09e2f29c84c56b8b75d393ed0b9d4 : SUCCESS in 18m 14s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/447288d743b441e997dd15c692cfe261 : SUCCESS in 50m 23s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/52c16808a8594cc4a1fd373b372744f1 : SUCCESS in 59m 39s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/2b807b4f544a43b1acc51859550e10e7 : SUCCESS in 34m 29s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4833085518b740b3a5e0904c0745ee11 : SUCCESS in 17m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fce647d15ac24b25a1f8739186fa39d0 : SUCCESS in 7m 00s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/928a93c24d9e4b8e83ee38d7b77f40d5 : SUCCESS in 45m 36s","accounts_in_message":[],"_revision_number":3},{"id":"85d6e018e5a181afc5bdaaaf8da9dfe335b805c2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-25 21:14:05.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"362970d6f75b5f14bcfe28aed7e3f8b1cc8c5094","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-25 21:15:46.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e86496bba036411ea4a4e8f0ed9fdbc9 : SUCCESS in 1m 20s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8e24d78533034283934d286b9a4c734b : SUCCESS in 51s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/d49a6f86967945fdb55feb3af50ee7c3 : SUCCESS in 42s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/74eedf7e0dd44b2e8490c2c493caa0c7 : SUCCESS in 45s (non-voting)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"c0bb5b2f3f9e0cf838ba1f6b3364ec7211eeee60","revisions":{"333bdebdf5a417c12efc0c6b37816b703f6fe506":{"kind":"REWORK","_number":1,"created":"2021-05-25 17:51:30.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/18/793018/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/18/793018/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/793018/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/793018/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/793018/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/18/793018/1"}}},"commit":{"parents":[{"commit":"bdb4510e05863942d0452db6e1d399b1921eeff7","subject":"Merge \"Avoid loops when gathering container listings from shards\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bdb4510e05863942d0452db6e1d399b1921eeff7"}]}],"author":{"name":"wu.shiming","email":"wushiming@yovole.com","date":"2020-11-11 10:30:27.000000000","tz":480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-05-25 17:50:05.000000000","tz":-420},"subject":"Dep\u0027s should be restricted by upper-constraints","message":"Dep\u0027s should be restricted by upper-constraints\n\nTox trying to install latest versions for building docs which may\nnot be supported by stable and lower branches, so should be\nrestricted by respective version\u0027s upper-constraints.txt\n\nChange-Id: I599d4ea430cc1e65bb50d0481617b005febf6de2\n(cherry picked from commit ee641dffc9b2ac835530efade91b1b72cee12e89)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/333bdebdf5a417c12efc0c6b37816b703f6fe506"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/333bdebdf5a417c12efc0c6b37816b703f6fe506"}]},"branch":"refs/heads/stable/victoria"},"ba182e3c3f9d41b8749c19d2bd9d1090437adee6":{"kind":"REWORK","_number":2,"created":"2021-05-25 18:18:39.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/18/793018/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/18/793018/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/793018/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/793018/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/793018/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/18/793018/2"}}},"commit":{"parents":[{"commit":"bdb4510e05863942d0452db6e1d399b1921eeff7","subject":"Merge \"Avoid loops when gathering container listings from shards\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bdb4510e05863942d0452db6e1d399b1921eeff7"}]}],"author":{"name":"wu.shiming","email":"wushiming@yovole.com","date":"2020-11-11 10:30:27.000000000","tz":480},"committer":{"name":"Tim Burke","email":"tburke@nvidia.com","date":"2021-05-25 18:18:07.000000000","tz":0},"subject":"Dep\u0027s should be restricted by upper-constraints","message":"Dep\u0027s should be restricted by upper-constraints\n\nTox trying to install latest versions for building docs which may\nnot be supported by stable and lower branches, so should be\nrestricted by respective version\u0027s upper-constraints.txt\n\nChange-Id: I599d4ea430cc1e65bb50d0481617b005febf6de2\n(cherry picked from commit ee641dffc9b2ac835530efade91b1b72cee12e89)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ba182e3c3f9d41b8749c19d2bd9d1090437adee6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ba182e3c3f9d41b8749c19d2bd9d1090437adee6"}]},"branch":"refs/heads/stable/victoria"},"c0bb5b2f3f9e0cf838ba1f6b3364ec7211eeee60":{"kind":"REWORK","_number":3,"created":"2021-05-25 18:25:17.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/18/793018/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/18/793018/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/793018/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/793018/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/793018/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/18/793018/3"}}},"commit":{"parents":[{"commit":"bdb4510e05863942d0452db6e1d399b1921eeff7","subject":"Merge \"Avoid loops when gathering container listings from shards\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bdb4510e05863942d0452db6e1d399b1921eeff7"}]}],"author":{"name":"wu.shiming","email":"wushiming@yovole.com","date":"2020-11-11 10:30:27.000000000","tz":480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-05-25 18:23:34.000000000","tz":-420},"subject":"Dep\u0027s should be restricted by upper-constraints","message":"Dep\u0027s should be restricted by upper-constraints\n\nTox trying to install latest versions for building docs which may\nnot be supported by stable and lower branches, so should be\nrestricted by respective version\u0027s upper-constraints.txt\n\nChange-Id: I599d4ea430cc1e65bb50d0481617b005febf6de2\n(cherry picked from commit ee641dffc9b2ac835530efade91b1b72cee12e89)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c0bb5b2f3f9e0cf838ba1f6b3364ec7211eeee60"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c0bb5b2f3f9e0cf838ba1f6b3364ec7211eeee60"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
