)]}'
{"id":"openstack%2Fswift~793019","triplet_id":"openstack%2Fswift~stable%2Fussuri~I599d4ea430cc1e65bb50d0481617b005febf6de2","project":"openstack/swift","branch":"stable/ussuri","hashtags":[],"change_id":"I599d4ea430cc1e65bb50d0481617b005febf6de2","subject":"Dep\u0027s should be restricted by upper-constraints","status":"MERGED","created":"2021-05-25 17:53:05.000000000","updated":"2021-05-26 04:41:38.000000000","submitted":"2021-05-26 04:38:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"793019","meta_rev_id":"b6463f2e5f0af3e710e49369efb20e5989edec6b","_number":793019,"virtual_id_number":793019,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-05-26 04:38:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-05-25 19:51:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-05-25 19:51:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"}],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-25 17:53:05.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"state":"REVIEWER"},{"updated":"2021-05-25 19:46:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"da364ecdf6c214379cf1fb8a927f55e7b7b5032c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-05-25 17:53:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7166215c46fcf1ccf0c9ece907ec0a7be5d56772","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-05-25 18:26:52.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"9f70168bbe016fdcf539658d55ddd79772754656","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-25 19:46:45.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3674febdeb684ff1af477232540d8da7 : SUCCESS in 9m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/028773fefb654ea3ae82e4d42f747de4 : SUCCESS in 6m 30s\n- grenade https://zuul.opendev.org/t/openstack/build/de4ed27616094245ae62ccb619a87578 : SUCCESS in 1h 14m 09s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c96c2a59b383465ebf9eea3a51d5c6c4 : SUCCESS in 49m 58s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d8ef0ad54e724d5390bbd268c7596666 : SUCCESS in 9m 12s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/5c2cc4947432440faf100adabff35ade : SUCCESS in 6m 48s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/97c7d2b8fae94021a636737f92c3f225 : SUCCESS in 14m 38s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/12541dfe3a364b9b89cfa26a59a3d806 : SUCCESS in 14m 58s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/de87eafe2cb7424fbf1c5bba3200807d : SUCCESS in 15m 22s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/dbb27bf5d13948b4a1ae8f31e6454bcb : SUCCESS in 15m 13s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/197185bc9b374c3abf184a1556264f44 : SUCCESS in 12m 12s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/3b7051db978a4632a2b207989bab8fd1 : SUCCESS in 13m 24s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/c5e4cc69e6a245ce94a1e61a7f4f5958 : SUCCESS in 11m 08s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/9f94f127d8c64d4784daa973cb30db28 : SUCCESS in 12m 10s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/13ee138889624115bea6c4c882a33664 : SUCCESS in 16m 05s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/605275c3e3664171831af71a722a789b : SUCCESS in 10m 20s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/bd864a90dcc14fb7a86591c5679af152 : SUCCESS in 14m 14s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/e9be31269bc14a419ae243ea588bf5f9 : SUCCESS in 11m 36s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/5ec6c243334b4f4fa2684bc70b541f42 : SUCCESS in 12m 26s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/fd54a7dc374646a0b1d1c5aa5019b1c4 : SUCCESS in 17m 39s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/31687e92fcf34d45b7f1e50c0772cf4b : SUCCESS in 22m 29s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/4144086631a04345ba6c8d356b357dec : SUCCESS in 45m 53s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/bd76bc73d37c4424b190490711e674f2 : SUCCESS in 35m 56s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/ed2fb024c8154825abe8a062cee8e142 : SUCCESS in 36m 43s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b12b26eb8c71490a8da9d154b9d211c1 : SUCCESS in 30m 03s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d41f822663094255994a0830e9eb2d6c : SUCCESS in 14m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/82e033c64f52420898eb7f22eae6baa5 : SUCCESS in 7m 08s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/127e84597e554a2baa47255cf86f964f : FAILURE in 11m 21s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/09c57460cb614355aca0bd81c745f219 : SUCCESS in 54m 15s","accounts_in_message":[],"_revision_number":2},{"id":"714ac742f18cda84465d149686136112185e9149","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-05-25 19:51:30.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"cf43773a845b476f9f6c7e46b3809af0202de91f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-25 19:51:49.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b9a04b71df6da29dafafed3a84795b9ac1a3814f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-25 21:46:22.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d8f63f78952469983495befb3213c41 : SUCCESS in 8m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a1b9f30e3a194f0e8b17daa761af55f0 : SUCCESS in 5m 27s\n- grenade https://zuul.opendev.org/t/openstack/build/6d9ea64bcad54af989bd1a950fc88b35 : SUCCESS in 1h 17m 21s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/aafe84d18c7b4e7cbeec45bbc1932e1a : SUCCESS in 41m 25s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/0d98964ccead412b9b5abe5e89796df7 : POST_FAILURE in 8m 44s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/304bfd23274947598e9e89299b447244 : SUCCESS in 7m 00s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/7feded3ea44c48559cb01fab8bebfe62 : SUCCESS in 27m 59s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/52c3ef260e3b49b8bb4f16e199b89d51 : SUCCESS in 29m 07s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/c9ec639c5aef46debbd8b4ad21cafa97 : SUCCESS in 29m 36s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/9f4b91f433834635bd317a2645937f84 : SUCCESS in 20m 28s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/bc540ae122e24bd9a7a4ef540dde18a4 : SUCCESS in 16m 00s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/cbb6e4676f334c7099eca58b2a2b771f : SUCCESS in 19m 10s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/f1db5b3acc5d408286e7288654066f34 : SUCCESS in 17m 45s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/2365f50d8cc04eef9894cf46f7018e27 : SUCCESS in 18m 25s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/3bf40c95d0734a5899e673988c771ff1 : SUCCESS in 16m 56s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/5e60f2df847f418881ad81d577933f1e : SUCCESS in 13m 24s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/30fc7db39ba74cabbd25a2341109422e : SUCCESS in 13m 54s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/601f6aa23c85410184f4b4f8f42c34eb : SUCCESS in 25m 09s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/677567a320e6478984d842eeade03fa5 : SUCCESS in 30m 55s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/9a9cf5cac46541e2827edcc7217ea245 : FAILURE in 1h 53m 42s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4f697fc154584831a2195badb5f7ca88 : SUCCESS in 33m 04s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/f611b2459cd34c73a79dcf3023b85b38 : SUCCESS in 26m 37s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/650dd7d199e94b909a3a590d9851e37f : SUCCESS in 26m 56s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/190e32739fed41e0b030ccb4d15077fc : SUCCESS in 29m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f6b6a76ede2642b4b662ce723b6fbebb : SUCCESS in 6m 14s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/48276a0de25c4f9ab94fbb5876e905a1 : SUCCESS in 52m 50s","accounts_in_message":[],"_revision_number":2},{"id":"6a3db52cbb4a9ce798b70f8f64f312ec35295959","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-05-25 22:23:12.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"77a6833bb834f091085cc1283486f2624b3fbb2d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-25 23:22:48.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a681e83a20f4a529022db2e5d808bd4 : SUCCESS in 14m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ad108e63b1fb49a280dd46c720dd1b7a : SUCCESS in 5m 54s\n- grenade https://zuul.opendev.org/t/openstack/build/46f4299eac5b4456957322f862c459c7 : SUCCESS in 52m 51s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/5f1313301c7447488c9130f3bf15f885 : SUCCESS in 28m 37s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/fb2eda01998b45b0ab3cfed5bc136346 : SUCCESS in 7m 37s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/b98729e290e44a5584142ff8baf2718b : SUCCESS in 6m 02s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/da7e33ad46b64947ad5b2227b3fecd10 : SUCCESS in 11m 18s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/0f0185c1da644c4794553746ebe09f34 : SUCCESS in 12m 17s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/e720a91f29fb43efbabf3298dcda2a9c : SUCCESS in 26m 32s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/dc20cd65d4b94347b3c8f7553593014d : SUCCESS in 14m 09s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/2075f8939cfe476a8bf36d56a40dde00 : SUCCESS in 10m 11s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/a4b9c57b3fae471ab50f0807b58b564b : SUCCESS in 11m 35s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/068197f3ec8a48338a98e341edeffe0e : SUCCESS in 9m 18s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/fc5c4143f4b84d5a8bc7b6f595f263db : SUCCESS in 11m 08s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/63822d12430749f4a0f52b8371a0f0ea : SUCCESS in 14m 14s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/db1b5e5816ca4d429829227b5889830d : SUCCESS in 10m 14s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/b4a243e4987c459fb8b63f83391ee8bc : SUCCESS in 12m 40s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/5ca234bff085440fade25a9fb8f9150d : SUCCESS in 9m 57s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/2fce02f5f702489babc58d9a992ebee4 : SUCCESS in 9m 39s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/93926a53820b41f6906baab891aa7fbe : SUCCESS in 13m 48s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/00e4149e040048949f852cc2a0f10c84 : SUCCESS in 20m 37s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/d552eac1decd4cf69f3cc9753e85df08 : SUCCESS in 47m 15s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/5801136a21124362bc4f735f01e7ccb1 : SUCCESS in 32m 01s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/36913e39f49d4ae2ac47364281b7067c : SUCCESS in 22m 29s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/fb7468f7f84d473ba8ebe6264735a662 : SUCCESS in 19m 49s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/57a8f6c440364f7a88a19f44666f2814 : SUCCESS in 12m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/029bb7ae010549bb99ad98f6e5ae1580 : SUCCESS in 6m 13s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ca2d60668006458dbbbaad84f4938dc9 : FAILURE in 11m 35s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4291ffc165ce466a91858960c0de3037 : SUCCESS in 42m 31s","accounts_in_message":[],"_revision_number":2},{"id":"bfd80447530cb21cbb39aed7f7ad204c379e234c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-25 23:23:09.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"e17d999d1f22fcc82b807682e47325665b81f738","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-26 04:38:17.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f424abdb519340edbb250f6eb1eb1847 : SUCCESS in 11m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f767a38c1ee042f7ba3df9b525b4f4c4 : SUCCESS in 6m 33s\n- grenade https://zuul.opendev.org/t/openstack/build/2fc1ddaca9cf46bfa3b9b8cb8f681dbb : SUCCESS in 1h 16m 18s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/4dfed28039dc4291bf9807773de81f46 : SUCCESS in 45m 31s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/004d83b45a4d40b299cae3990445c85e : SUCCESS in 10m 06s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/2d49a5383e95428b8c95666b96f8b407 : SUCCESS in 7m 41s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/648bb6f77dee46d488dee05d5f47b162 : SUCCESS in 17m 14s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/4c7514ac32184edeb4c925d0e74d7d78 : SUCCESS in 14m 55s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/ac4b55f9e665446985dcc1a9d0b5bd9f : SUCCESS in 15m 48s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/75a70921cbca488aafdd0e69c3b3bdf7 : SUCCESS in 14m 45s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/3bc008139de0465fb3d4c385cda10846 : SUCCESS in 12m 39s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/b18afd2a61fb4d0ba7d0f3b0425466ea : SUCCESS in 15m 15s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/9d06851a8d84460abc028906172a9695 : SUCCESS in 11m 07s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/853eec66e205496ca5040764315501ec : SUCCESS in 11m 31s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/c3407f7ce8c24dd5bd99a00e7d77679f : SUCCESS in 16m 17s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/253757b538a04032ac956b656b43b3ae : SUCCESS in 12m 36s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/689512b42f044a329a72e2da564542fb : SUCCESS in 10m 20s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/5cdd53e80a4c44b984582bb1b60d6d5a : SUCCESS in 12m 13s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/51aa11ec05f44a069d1bdb4e6c36929e : SUCCESS in 14m 20s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/c97525624b5e4314a00531a845048428 : SUCCESS in 45m 47s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/936923989f394c0b9ed68afbe90429a7 : SUCCESS in 29m 49s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/31bb51517f434110aba7fa8a59f6468f : SUCCESS in 35m 30s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/5e0d7c63bd884117b3c01f8be5b789ef : SUCCESS in 26m 24s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2d19ec80e22445bda776e12b2e0c9dbf : SUCCESS in 14m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fce434ab135c483cb43a90b2a50de463 : SUCCESS in 7m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/43ad4bf7ffce49598ec5fc79d59790b7 : SUCCESS in 57m 50s","accounts_in_message":[],"_revision_number":2},{"id":"089c032de297efd3993fc94e46f3539803977cca","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-26 04:38:18.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"b6463f2e5f0af3e710e49369efb20e5989edec6b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-26 04:41:38.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6c5cae74a493401eb9a5739785e4a0c8 : SUCCESS in 58s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/22f78c40606746cfaf92996b96c0c5a0 : SUCCESS in 1m 05s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/6c4567c3d87c42a8bfe5cd6b2f07fad4 : SUCCESS in 38s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/4919a6ae44884caea558d831683f598c : SUCCESS in 42s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"29eefcd0dee62f151a1541074870d58120c83edc","revisions":{"3a0fc1cdcdfd7f8ec8440cc6c6d029b65e2b6709":{"kind":"REWORK","_number":1,"created":"2021-05-25 17:53:05.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/19/793019/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/19/793019/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/793019/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/793019/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/793019/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/19/793019/1"}}},"commit":{"parents":[{"commit":"e69b3c60feb8c60cae5ec27508fd6f860d8c14e0","subject":"Avoid loops when gathering container listings from shards","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e69b3c60feb8c60cae5ec27508fd6f860d8c14e0"}]}],"author":{"name":"wu.shiming","email":"wushiming@yovole.com","date":"2020-11-11 10:30:27.000000000","tz":480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-05-25 17:52:38.000000000","tz":-420},"subject":"Dep\u0027s should be restricted by upper-constraints","message":"Dep\u0027s should be restricted by upper-constraints\n\nTox trying to install latest versions for building docs which may\nnot be supported by stable and lower branches, so should be\nrestricted by respective version\u0027s upper-constraints.txt\n\nChange-Id: I599d4ea430cc1e65bb50d0481617b005febf6de2\n(cherry picked from commit ee641dffc9b2ac835530efade91b1b72cee12e89)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3a0fc1cdcdfd7f8ec8440cc6c6d029b65e2b6709"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3a0fc1cdcdfd7f8ec8440cc6c6d029b65e2b6709"}]},"branch":"refs/heads/stable/ussuri"},"29eefcd0dee62f151a1541074870d58120c83edc":{"kind":"REWORK","_number":2,"created":"2021-05-25 18:26:52.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/19/793019/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/19/793019/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/793019/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/793019/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/793019/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/19/793019/2"}}},"commit":{"parents":[{"commit":"e69b3c60feb8c60cae5ec27508fd6f860d8c14e0","subject":"Avoid loops when gathering container listings from shards","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e69b3c60feb8c60cae5ec27508fd6f860d8c14e0"}]}],"author":{"name":"wu.shiming","email":"wushiming@yovole.com","date":"2020-11-11 10:30:27.000000000","tz":480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-05-25 18:26:36.000000000","tz":-420},"subject":"Dep\u0027s should be restricted by upper-constraints","message":"Dep\u0027s should be restricted by upper-constraints\n\nTox trying to install latest versions for building docs which may\nnot be supported by stable and lower branches, so should be\nrestricted by respective version\u0027s upper-constraints.txt\n\nChange-Id: I599d4ea430cc1e65bb50d0481617b005febf6de2\n(cherry picked from commit ee641dffc9b2ac835530efade91b1b72cee12e89)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/29eefcd0dee62f151a1541074870d58120c83edc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/29eefcd0dee62f151a1541074870d58120c83edc"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
