)]}'
{"id":"openstack%2Fswift~793148","triplet_id":"openstack%2Fswift~master~I61ab1dbea9326a625d7fce90457dc104d0a25690","project":"openstack/swift","branch":"master","topic":"bug/1925782","hashtags":[],"change_id":"I61ab1dbea9326a625d7fce90457dc104d0a25690","subject":"More test assertions for dark data grace age","status":"ABANDONED","created":"2021-05-26 11:52:16.000000000","updated":"2021-06-02 03:46:31.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"e0bd9738884124874a87a7912437dac0781ee4ab","_number":793148,"virtual_id_number":793148,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-05-26 12:31:50.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-26 12:31:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fb41fba6d4fa5d9f26a30e0dcdcd55ca1f3753e4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-05-26 11:52:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e5e66467e40ca3658b60090aa8e3f86799f09c35","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-05-26 11:55:29.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e35d0b550976f26ed7aa4e63b0907bf449a3356b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-26 12:31:50.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/44f73a24a81d4f288da64a3f64d5a63c : SUCCESS in 8m 15s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/9aa19402639347458a61ff4478a49525 : SUCCESS in 8m 02s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/2c5829501c764722af61f1d0159b7698 : SUCCESS in 6m 00s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/fd19b8de4b274fb68f57f96ef9f9ff3e : SUCCESS in 13m 40s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/1b1fda8264b6460fa4c97a5c2310db31 : SUCCESS in 13m 52s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/3d5651bcac3c45269b84f8229a0639db : SUCCESS in 13m 15s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/68c9d960c8664e1d9d703f617f95fc2d : SUCCESS in 12m 34s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/484423e9b10f4b21b6983bff5923e789 : SUCCESS in 13m 32s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/9e9d59bd042c464a95121cf4ddcfe7ae : SUCCESS in 18m 38s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/98c1bc1081534200b60589820e1d4a08 : SUCCESS in 23m 47s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/8d324ef7cd3f470ea56cd255f699a3ab : SUCCESS in 24m 39s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/a8fd2f28a53348488a1774913cef9695 : SUCCESS in 18m 13s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/d8461eded6374562b6d20e25b80bd227 : SUCCESS in 29m 16s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/a0dd8d41fadb4ee1895911405e466abb : SUCCESS in 25m 19s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/7004d67290e645819487138f1fe486fd : SUCCESS in 34m 52s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/25770a18165c43e08f9c1bfbc426bc17 : SUCCESS in 30m 03s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/29041688d1b54c65b9ee1dfd21a360b9 : SUCCESS in 15m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bc1985c4dd5d42a9bdf7f6ffa1891a33 : FAILURE in 5m 34s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/be4f9458686a4670bc0ed187fbf81a1e : SUCCESS in 23m 35s","accounts_in_message":[],"_revision_number":2},{"id":"e0bd9738884124874a87a7912437dac0781ee4ab","tag":"autogenerated:gerrit:abandon","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2021-06-02 03:46:31.000000000","message":"Abandoned\n\nSquashed into Ieafa72a496328f7a487ca7062da6253994a5a07d","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"9710bc20c7a4c6d9f4a6eb933461d023363a99f0","revisions":{"57457c516bb3cf5b2cb820a44dcc52461db196a0":{"kind":"REWORK","_number":1,"created":"2021-05-26 11:52:16.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/48/793148/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/48/793148/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/793148/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/793148/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/793148/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/48/793148/1"}}},"commit":{"parents":[{"commit":"9ceae2dfd258b95684449737c34cd171f436dd76","subject":"Make dark data watcher ignore the newly updated objects","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9ceae2dfd258b95684449737c34cd171f436dd76"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-05-26 11:51:35.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-05-26 11:51:35.000000000","tz":60},"subject":"More test assertions for dark data grace age","message":"More test assertions for dark data grace age\n\n...and adding an ec fragment to the test object pool.\n\nChange-Id: I61ab1dbea9326a625d7fce90457dc104d0a25690\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/57457c516bb3cf5b2cb820a44dcc52461db196a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/57457c516bb3cf5b2cb820a44dcc52461db196a0"}]},"branch":"refs/heads/master"},"9710bc20c7a4c6d9f4a6eb933461d023363a99f0":{"kind":"REWORK","_number":2,"created":"2021-05-26 11:55:29.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/48/793148/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/48/793148/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/793148/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/793148/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/793148/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/48/793148/2"}}},"commit":{"parents":[{"commit":"9ceae2dfd258b95684449737c34cd171f436dd76","subject":"Make dark data watcher ignore the newly updated objects","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9ceae2dfd258b95684449737c34cd171f436dd76"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-05-26 11:51:35.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-05-26 11:53:05.000000000","tz":60},"subject":"More test assertions for dark data grace age","message":"More test assertions for dark data grace age\n\n...and adding an ec fragment to the test object pool.\n\nChange-Id: I61ab1dbea9326a625d7fce90457dc104d0a25690\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9710bc20c7a4c6d9f4a6eb933461d023363a99f0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9710bc20c7a4c6d9f4a6eb933461d023363a99f0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
