)]}'
{"id":"openstack%2Fswift~793248","triplet_id":"openstack%2Fswift~master~I023334ac65a3a255dc30a258b5fce613f3813ccc","project":"openstack/swift","branch":"master","hashtags":[],"change_id":"I023334ac65a3a255dc30a258b5fce613f3813ccc","subject":"DNM: Sanity-check that install-script failures fail image builds","status":"ABANDONED","created":"2021-05-26 22:07:27.000000000","updated":"2021-05-27 21:48:09.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"36ec7caeb9847ab49dafaafe61b36cc79d39a230","_number":793248,"virtual_id_number":793248,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2021-05-26 23:26:28.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-26 22:07:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1e596824ca3afb282b81eba18d6f9e1c57f79d12","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-05-26 22:07:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fe4bb847c44e82356d24af7efc41d76b6ae21d70","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-26 22:07:47.000000000","message":"Patch Set 1: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/swift on branch master.  The error was:\n\n  found undefined alias\n  in \"openstack/swift/.zuul.yaml@master\", line 552, column 31","accounts_in_message":[],"_revision_number":1},{"id":"a55f32290386bd70323d17fa05b245bd4d2c726c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-05-26 22:09:59.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"5454427a36d5547a732ac0bccd3b8233c5614f08","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-26 22:19:41.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- swift-build-image https://zuul.opendev.org/t/openstack/build/4ab9501a6eff418ca420884becff7fa3 : FAILURE in 4m 05s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/27324e71c3b841b8bda51f7b632a7236 : FAILURE in 4m 04s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d159644b931bd09662fd398e129f0624ed8279d4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-05-26 22:28:09.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"af4a8608d993f59074b50e92de54094851ea1e16","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-26 22:36:25.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- swift-build-image https://zuul.opendev.org/t/openstack/build/27c3bc133f8648a4b9a2521c70765d0c : FAILURE in 4m 01s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/75fd1b61d32e4d6e96711e8fe86fcc79 : FAILURE in 4m 17s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"a6e788e92d8c58d139940db49b3f080515bba3ac","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-05-26 23:17:31.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"8b2fca5dba1ef27d658f976ba9295174a9165f5a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-26 23:26:28.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d09cb4f4a4914ed58ff577e48c546696 : FAILURE in 7m 58s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/f859074a2e614de89463a22b29d0b41f : FAILURE in 5m 36s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"36ec7caeb9847ab49dafaafe61b36cc79d39a230","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-05-27 21:48:09.000000000","message":"Abandoned\n\nSo *that* sure looks like a pip bug:\n\n \u001b[0m\u001b[91mERROR: Could not find a version that satisfies the requirement xattr\u003e\u003d0.4 (from versions: ...)\n \u001b[0m\u001b[91mERROR: No matching distribution found for xattr\u003e\u003d0.4\n \u001b[0mtesting `set -e`\n\nWhy\u0027d pip still exit 0 after the error??\n\nw/e; I think https://review.opendev.org/c/openstack/swift/+/793483 is how we ought to fix it.","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"b3a5191998d554d2ce38c11ce2b7dc4e391457d9","revisions":{"9c44aca0edf202f02d67f5e409cb33081c34f2dc":{"kind":"REWORK","_number":1,"created":"2021-05-26 22:07:27.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/48/793248/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/48/793248/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/793248/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/793248/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/793248/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/48/793248/1"}}},"commit":{"parents":[{"commit":"f7f1553edb7e3dbe57507113b730ee1f77472928","subject":"Merge \"Use ContainerSharderConf class in sharder and manage-shard-ranges\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f7f1553edb7e3dbe57507113b730ee1f77472928"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-05-26 22:05:58.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-05-26 22:07:06.000000000","tz":-420},"subject":"DNM: Sanity-check that RUN failures fail image builds","message":"DNM: Sanity-check that RUN failures fail image builds\n\nChange-Id: I023334ac65a3a255dc30a258b5fce613f3813ccc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9c44aca0edf202f02d67f5e409cb33081c34f2dc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9c44aca0edf202f02d67f5e409cb33081c34f2dc"}]},"branch":"refs/heads/master"},"d8dd4ba17d06edbb35f3b723de93e2623a31edc9":{"kind":"REWORK","_number":2,"created":"2021-05-26 22:09:59.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/48/793248/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/48/793248/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/793248/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/793248/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/793248/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/48/793248/2"}}},"commit":{"parents":[{"commit":"f7f1553edb7e3dbe57507113b730ee1f77472928","subject":"Merge \"Use ContainerSharderConf class in sharder and manage-shard-ranges\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f7f1553edb7e3dbe57507113b730ee1f77472928"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-05-26 22:05:58.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-05-26 22:09:51.000000000","tz":-420},"subject":"DNM: Sanity-check that RUN failures fail image builds","message":"DNM: Sanity-check that RUN failures fail image builds\n\nChange-Id: I023334ac65a3a255dc30a258b5fce613f3813ccc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d8dd4ba17d06edbb35f3b723de93e2623a31edc9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d8dd4ba17d06edbb35f3b723de93e2623a31edc9"}]},"branch":"refs/heads/master"},"2967d3e134782500bd4d42c2e36fe38206c120b7":{"kind":"REWORK","_number":3,"created":"2021-05-26 22:28:09.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/48/793248/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/48/793248/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/793248/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/793248/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/793248/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/48/793248/3"}}},"commit":{"parents":[{"commit":"f7f1553edb7e3dbe57507113b730ee1f77472928","subject":"Merge \"Use ContainerSharderConf class in sharder and manage-shard-ranges\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f7f1553edb7e3dbe57507113b730ee1f77472928"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-05-26 22:05:58.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-05-26 22:27:43.000000000","tz":-420},"subject":"DNM: Sanity-check that install-script failures fail image builds","message":"DNM: Sanity-check that install-script failures fail image builds\n\nChange-Id: I023334ac65a3a255dc30a258b5fce613f3813ccc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2967d3e134782500bd4d42c2e36fe38206c120b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2967d3e134782500bd4d42c2e36fe38206c120b7"}]},"branch":"refs/heads/master"},"b3a5191998d554d2ce38c11ce2b7dc4e391457d9":{"kind":"REWORK","_number":4,"created":"2021-05-26 23:17:31.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/48/793248/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/48/793248/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/793248/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/793248/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/793248/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/48/793248/4"}}},"commit":{"parents":[{"commit":"f7f1553edb7e3dbe57507113b730ee1f77472928","subject":"Merge \"Use ContainerSharderConf class in sharder and manage-shard-ranges\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f7f1553edb7e3dbe57507113b730ee1f77472928"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-05-26 22:05:58.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-05-26 23:17:20.000000000","tz":-420},"subject":"DNM: Sanity-check that install-script failures fail image builds","message":"DNM: Sanity-check that install-script failures fail image builds\n\nChange-Id: I023334ac65a3a255dc30a258b5fce613f3813ccc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b3a5191998d554d2ce38c11ce2b7dc4e391457d9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b3a5191998d554d2ce38c11ce2b7dc4e391457d9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
