)]}'
{"id":"openstack%2Fswift~793268","triplet_id":"openstack%2Fswift~master~I51390aacc4b73a957556b8b7a5b50a11b2f19f2b","project":"openstack/swift","branch":"master","topic":"error_limited_testing","hashtags":[],"change_id":"I51390aacc4b73a957556b8b7a5b50a11b2f19f2b","subject":"POC: add reseller_admin info endpoint to grab error_limited info","status":"NEW","created":"2021-05-27 05:33:41.000000000","updated":"2021-05-27 06:47:23.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"57beabc2e3c91d6dee61868779acb3239e30a79d","_number":793268,"virtual_id_number":793268,"owner":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2021-05-27 06:47:23.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-27 06:47:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"68d5e9d2e06c5c1c6bb8e8f76dc2531a859f7d8c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2021-05-27 05:33:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"57beabc2e3c91d6dee61868779acb3239e30a79d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-27 06:47:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28dd362a12624db88bca3f91331dbc44 : SUCCESS in 9m 07s\n- grenade https://zuul.opendev.org/t/openstack/build/28886844813f451fa3bf2fcea916ce36 : SUCCESS in 1h 09m 35s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e204aabcfba241d6892f401f66eb0880 : SUCCESS in 44m 39s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/8b3ff08d8304412580f0faff077f42e8 : SUCCESS in 8m 57s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/425d67a3a8bf4ab88492bff8d098f9d9 : SUCCESS in 6m 29s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/62d10312edc7496dbd4522b3c753338a : SUCCESS in 15m 46s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/64f7f42e70d94fcda30a0a48308ea0e4 : SUCCESS in 14m 17s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/57305d801eb249cdbc52892e95148a5b : SUCCESS in 16m 49s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/b6df9e2a11cf40b0b0f7e2edd0894452 : SUCCESS in 14m 14s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/d69e4a1968fa4e3eacb64ff2981ea445 : SUCCESS in 12m 44s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/e279147203db4613b14ba0ecd1528335 : SUCCESS in 23m 16s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/20e8bbb50e7f4da6aece4dbdc5dc08d7 : SUCCESS in 27m 06s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/1c543eedc31c4eff99bd807333035943 : SUCCESS in 25m 21s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/934eaf80372a44eb945285f60b4417dd : SUCCESS in 24m 21s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/859d7ff9c84f4da9b59902824f42a445 : SUCCESS in 28m 55s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/cefa1feffaec47e7b56c1dba3b4590ac : SUCCESS in 25m 40s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/336f65aadb284a778cc0949b206ddf9b : SUCCESS in 8m 10s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/54385c857f7e463bbc10a51a4e544c10 : FAILURE in 26m 08s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/f114bd94f584455eb2d7ad0b56f8bb18 : SUCCESS in 41m 42s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/b5bd9670f25842409bc1c96a6ed39678 : SUCCESS in 1h 00m 10s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/561bc881bdbc444bb1dbd3845fedf85f : SUCCESS in 31m 21s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/647d32507c954fd1b137ac12e4c67a07 : SUCCESS in 39m 59s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1fc2aafa121641b7b5f0182f608c1fdc : SUCCESS in 16m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/613a6dc173644a97a26fa70907500319 : SUCCESS in 5m 54s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/563b035409774d1aa167aca058f44359 : SUCCESS in 20m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8fb86cfea2964e8f8e4f4f86953ea3ab : SUCCESS in 1h 03m 57s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e24fdbd142a8ca142c17297ba49aa9adb1779338","revisions":{"e24fdbd142a8ca142c17297ba49aa9adb1779338":{"kind":"REWORK","_number":1,"created":"2021-05-27 05:33:41.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/68/793268/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/68/793268/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/68/793268/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/68/793268/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/68/793268/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/68/793268/1"}}},"commit":{"parents":[{"commit":"b3def185c66dd70a85c7755b168da014529d2428","subject":"Merge \"Allow floats for all intervals\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b3def185c66dd70a85c7755b168da014529d2428"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2021-05-27 05:27:39.000000000","tz":600},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2021-05-27 05:33:24.000000000","tz":600},"subject":"POC: add reseller_admin info endpoint to grab error_limited info","message":"POC: add reseller_admin info endpoint to grab error_limited info\n\nCurrently there isn\u0027t an easy way to see the current state of what the\nproxies have in their error_limiting dict in memory.\n\nThis patch allows you to get a json dump of it by hitting the\n`/info/error_limited` endpoint. But this endpoint is only available if\nyou are a reseller admin.\n\nIt\u0027s kinda like a recon, and maybe I should just make a proxy recon\nendpoint in a controller and rename it to /recon/error_limited, but I\nwanted to get something to see if it worked first.\n\nChange-Id: I51390aacc4b73a957556b8b7a5b50a11b2f19f2b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e24fdbd142a8ca142c17297ba49aa9adb1779338"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e24fdbd142a8ca142c17297ba49aa9adb1779338"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
