)]}'
{"id":"openstack%2Fswift~793503","triplet_id":"openstack%2Fswift~master~I25eafa0f1eee24b728dd55ad05589703a99f3393","project":"openstack/swift","branch":"master","hashtags":[],"change_id":"I25eafa0f1eee24b728dd55ad05589703a99f3393","subject":"probe: Make sure we know what version of python we\u0027re testing in the gate","status":"ABANDONED","created":"2021-05-27 20:51:51.000000000","updated":"2022-08-20 00:22:15.000000000","total_comment_count":2,"unresolved_comment_count":2,"has_review_started":true,"meta_rev_id":"de8930743ef3c21a29621e6aca8a9d5c3a15fb2b","_number":793503,"virtual_id_number":793503,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2022-03-23 23:53:27.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-27 22:47:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-01 01:53:48.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"CC"}],"messages":[{"id":"db0517be36811251f01e444544590f5b1400ea42","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-05-27 20:51:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"45e7fb3b1252379ac8ee463af482b6519a8ca673","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-05-27 21:35:11.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"42554dc7b95db857aa6f46800a052a5a7d4aba6c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-27 22:47:28.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/16134c957b124eed8ba1231a97f90cc2 : SUCCESS in 9m 46s\n- grenade https://zuul.opendev.org/t/openstack/build/1db1f3dadb474f038278e3eec55d22d7 : SUCCESS in 1h 06m 53s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/7e0c78a659294174aee6596004333da3 : SUCCESS in 44m 39s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/536b76928ac847178ab3965275df1a82 : SUCCESS in 9m 10s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/993c65a1e548485a9fd6ffb4f53de9c4 : SUCCESS in 6m 48s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/f290bcfe227b429bb0251040a1431008 : SUCCESS in 16m 04s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/01b30bb0c602450c84651cc0c9724e2e : SUCCESS in 16m 29s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/c3509f9d70eb40739a34f6c64eda2b53 : SUCCESS in 41m 35s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/3d31add5f2af4b05a2c61eb38a550376 : SUCCESS in 17m 04s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/8ee2f7937da941d98e6d9658560a5ae9 : FAILURE in 16m 49s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/7e245b8dd03246988046c57db350f0ef : SUCCESS in 18m 05s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/f1d8f16a61de4835a64dde4ae140cda0 : SUCCESS in 19m 06s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/ca8e96f425be43e99fa15e2c5d796ed1 : SUCCESS in 18m 14s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/7b53ec85035f4732a968d3dce7f5dbdb : SUCCESS in 26m 57s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/bc6bc730a8c9480793ec4f25e72ea423 : SUCCESS in 20m 02s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/8266cb02f2ab43be8b5f643d472dbafe : SUCCESS in 18m 00s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/0134619611b446bfbd400d22e66e0cd8 : SUCCESS in 6m 49s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/6e3af56fd1d64dbaad191c5ec942e049 : SUCCESS in 22m 53s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/dc3c2344f23d45489e464df68de00346 : SUCCESS in 48m 51s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/27ea2ccb5f094cd79e10886d80897c46 : SUCCESS in 48m 36s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a6c107138f1c4be3a2acea36ba05620f : SUCCESS in 34m 41s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/2698f6e37af34d6c9d1286e66f3131b3 : SUCCESS in 44m 09s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5edf7217a3494619b788c70ac1a8911b : SUCCESS in 16m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1597237d349043ff86fce64a8f0e421c : SUCCESS in 6m 01s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/1268df4aa0d04888a849d9dc8939bbd1 : SUCCESS in 24m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c2d8b6c66121422796dae255618d9d92 : SUCCESS in 53m 36s","accounts_in_message":[],"_revision_number":2},{"id":"3fcd18f268aa8f7c78bb19c448288a9b93be5270","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-06-01 01:46:11.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"f1bb903ef575c9cb07e5f95ce03a8d1453c55129","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-06-01 01:53:13.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d19307467a2326500621230bb69d46ba0bc3ad7f","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2021-06-01 01:53:48.000000000","message":"Patch Set 2:\n\nLooks like centos 8 is using 3.6 too:\n\n  /bin/sh: python: command not found\n  Python 3.6.8\n\nDo we want to look at centos 9 if we want something later?","accounts_in_message":[],"_revision_number":2},{"id":"60e1797755745087a3e15b483eb25e0d6b577eed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-01 02:49:21.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/99037fcbc0f945c293645ca390846372 : SUCCESS in 8m 38s\n- grenade https://zuul.opendev.org/t/openstack/build/db4ad773dc9446ed8ba42072c917bbf9 : SUCCESS in 1h 02m 30s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e6020bfb1a2842038ffae444f5d5f543 : SUCCESS in 48m 02s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a0a197df9e6543f88b1aa0f7f113a859 : SUCCESS in 9m 03s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/b1e9f1c515de4458a2485e67745775aa : SUCCESS in 7m 54s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/1ddc3d305e954982ad934d6eb1a7eb1b : SUCCESS in 20m 30s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/bda7da559459427dbffc7d543d496806 : SUCCESS in 21m 29s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/856698a3b7b54b5aa42dceb1c9e139bc : SUCCESS in 20m 22s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/bbb260d07b27416ab0c38b99cdf94fa6 : SUCCESS in 20m 31s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/ae941f900eaa4d748f56a1ccb96206b7 : SUCCESS in 20m 00s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/441f100916af453ba1f0474fd08a7baa : SUCCESS in 18m 34s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/668f96dc342e49e68097f510a9337564 : SUCCESS in 21m 05s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/7ded3e54f9284b50a8b5136e2accf2e2 : SUCCESS in 22m 30s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/ed9328986715425689124735c018386e : SUCCESS in 21m 57s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/5a985e31f5e54e6986519f1102da21f1 : SUCCESS in 20m 30s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/a514a5df0a7c4daa9f3c58fd13c9b234 : SUCCESS in 19m 19s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/7a3844a26b634a08a051cf9aac32eae0 : SUCCESS in 6m 25s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/a427ac299a114892893eae1c6947536e : FAILURE in 23m 17s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/c7c259c3e6894b598f87118060c45be2 : SUCCESS in 49m 17s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/b9ea1b06863a40339ceca35d67f39779 : SUCCESS in 47m 20s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/23eaf20e4ca14487ad5d7052d149e3ed : SUCCESS in 41m 52s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/6937ddda97284a6993b0a81dbf703c19 : SUCCESS in 40m 34s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/76b7990ad29b409fb25571eec33a9b0e : SUCCESS in 19m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8e28b73f0f224ae9a2d8c78e6720fe7d : SUCCESS in 6m 03s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/dffff7d9f27b4f5ba33596655c051d7d : SUCCESS in 30m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bd2003b74e834845abf34b3697c2aabc : SUCCESS in 48m 08s","accounts_in_message":[],"_revision_number":2},{"id":"e450d72da7085b8f662e92bba9ecb6d8ed909ca2","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-06-08 22:14:45.000000000","message":"Patch Set 2:\n\n3.6 is fine -- the biggest question in my mind was what version was being used for hte ARM jobs. I suspect it\u0027s also 3.6, but given the 3.7 func test jobs that were actually running on 3.6, I figured it was worth confirming.","accounts_in_message":[],"_revision_number":2},{"id":"80fdd9339a1ce18394210403c4ea317f4aa5c180","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-06-08 22:16:26.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"c2e74c0aa8933dccc26d283dec9086155de3be0c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-08 23:15:46.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f682ee029d8f4eeba058e736574fba39 : SUCCESS in 9m 39s\n- grenade https://zuul.opendev.org/t/openstack/build/d08cfb9aa543456e86737ab8a3c52078 : SUCCESS in 55m 45s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/59c4fef64b9a4b9fb0b1ed28756b9006 : SUCCESS in 40m 25s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/4bb4d6ff89334a528f61b5e3e2e93a10 : SUCCESS in 8m 21s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/36e671e51c9f4d06bc7f9a5e3c4802d0 : SUCCESS in 9m 02s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/c0ee5567e2a14283a50d15216c554899 : SUCCESS in 15m 31s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/2be969a9563845b7ba584409843f5f64 : SUCCESS in 15m 06s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/c63b803b8ebf43da9c8e0ddf8bc96abb : SUCCESS in 15m 08s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/cab6948a09b44d6db3d6f93e0ce5d606 : SUCCESS in 15m 10s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/82466ca1b1a646979160f0a619debe91 : SUCCESS in 19m 17s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/82f9f248b8de404ab9c1f93c57a16ee8 : SUCCESS in 20m 30s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/63cf23c10e8d4acbac8f47fc12e70ffe : SUCCESS in 16m 47s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/bb0c1cf1f8934d55a5692f3d1557c3ba : SUCCESS in 17m 09s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/b03e3b13dbc94fd6b0a8d124a352bb7a : SUCCESS in 16m 36s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/331a713b64274dcdbb1109484329abb4 : SUCCESS in 17m 58s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/4d0d37a5669345d697a12e7e9c5d9c47 : SUCCESS in 17m 47s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/63df0d7243d94b5b880f59131e302f6e : SUCCESS in 7m 24s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/63626bc684794ad3ab6f1033dcdc0d16 : SUCCESS in 21m 34s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/f3c26f76ac1e4d6983f8369313a26504 : SUCCESS in 46m 35s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/905945ebaf584be4a35dd91f5d57b07c : SUCCESS in 45m 49s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/1258f2ed66034aebac04c0baf85e0fff : SUCCESS in 37m 38s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/3a1c4da838d14b8abb4e37600f292a2a : SUCCESS in 37m 34s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a094880563d641f086e0eb98ec060623 : SUCCESS in 15m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b68baac451bc4a5f97fb7367468383f7 : SUCCESS in 6m 10s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a62ef3027bad45a481c7ce610e41a9ec : SUCCESS in 26m 48s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/089798f4af0940768b8f6e56163bf7a6 : SUCCESS in 55m 28s","accounts_in_message":[],"_revision_number":3},{"id":"8e655b2be5025792231cd4eb98f1f17cd0e747c5","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-06-12 17:07:39.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"4b276128f8a7a6a80663c985c3b77baa508c6862","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-03-23 22:37:45.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"ec389ac13131788df14248c160b750a565682da5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-23 23:53:27.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b454835989f4237bfdab2cb526861bd : SUCCESS in 12m 21s\n- grenade https://zuul.opendev.org/t/openstack/build/70c30e0d3ea440d792d6d1b66c699bde : SUCCESS in 54m 43s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/08650cf6168b444e9a09297c6baeca94 : SUCCESS in 48m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8797f0f81f7f4d67af89a5e136cf218f : SUCCESS in 50m 50s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/b7f6746b2d8c4ddf8a55c1898b0c385a : SUCCESS in 22m 32s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/72ca16b35bab4e34bf48e02fc77e3dea : SUCCESS in 32m 24s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/40771144167341b181f63cd0bcc2ac9b : SUCCESS in 24m 11s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/33f8c05352944010af8d280a56148d0e : SUCCESS in 10m 04s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/bb31b3032a3d4321b53381be501f5dd4 : SUCCESS in 7m 17s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/7313544b7b1b4310afb7176b143af9af : SUCCESS in 15m 55s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a6101c0b98b64be29ef2b91cc5eb92b2 : SUCCESS in 29m 36s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/ab0bd1fe35f74d2fabaa8b836654b714 : SUCCESS in 17m 18s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/73e1f96cb7a7470a909fa251fc5590bf : SUCCESS in 26m 29s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/60d2d0622259486db775572f77e3a180 : SUCCESS in 18m 22s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/badaf8b0fdb0453b820d6c9c8543efb0 : SUCCESS in 20m 34s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/4151410e95734f49abc687003eb3ab37 : SUCCESS in 18m 30s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/7d2e3a0a55a04eb2a3c531d75a1ef6d9 : SUCCESS in 16m 41s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/0e6f74328b03475c84c2bfd9dd044c19 : SUCCESS in 21m 07s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/bc5d074049564730a6c6b01a5185dae8 : SUCCESS in 9m 24s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/1479b66ab44e4aa9bca959aa5f0d3ddf : SUCCESS in 24m 01s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/553ff32cafba4f32966717b8b2eb245e : SUCCESS in 41m 44s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/25bfa9c5640b4a3f95eb38a2be5c0c1a : SUCCESS in 46m 10s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d9b85ea602204e16b190e6045670520b : SUCCESS in 37m 04s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/aba62f8c49eb42c3abc216bb6cd313d8 : SUCCESS in 36m 58s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f6878a424daf45239518eb9297aa2ded : SUCCESS in 16m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d47f475360c84af8b110c601c2604499 : SUCCESS in 7m 16s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e4bcf758603c4de08a4783d352bcddcf : SUCCESS in 24m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b9921a2575bc4e928ee48a233115c34f : SUCCESS in 1h 08m 17s","accounts_in_message":[],"_revision_number":4},{"id":"f95731186de431713063da1cce98d6229ae474de","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-24 01:37:51.000000000","message":"Patch Set 4:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/64eb7448a574498189ca9e957a49a92e : SUCCESS in 37m 41s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d48eedb774704629af32586ded0829e8 : SUCCESS in 49m 38s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/744cd10dd0ef4a129c21595a075fbac4 : FAILURE in 2h 57m 48s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/e825978f28a444a99cdc29c28b79692e : SUCCESS in 48m 16s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/6054dc87cec14711a8bd5114381fd8a8 : SUCCESS in 42m 21s","accounts_in_message":[],"_revision_number":4},{"id":"de8930743ef3c21a29621e6aca8a9d5c3a15fb2b","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-08-20 00:22:15.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"363934c061bdcfb5dab72707e47ee231d7a3d4af","revisions":{"4ce2e01a190a70fed348b74d28690c2a65bb213d":{"kind":"REWORK","_number":1,"created":"2021-05-27 20:51:51.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/03/793503/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/03/793503/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/03/793503/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/03/793503/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/03/793503/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/03/793503/1"}}},"commit":{"parents":[{"commit":"f7f1553edb7e3dbe57507113b730ee1f77472928","subject":"Merge \"Use ContainerSharderConf class in sharder and manage-shard-ranges\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f7f1553edb7e3dbe57507113b730ee1f77472928"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-05-27 20:51:30.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-05-27 20:51:30.000000000","tz":-420},"subject":"probe: Make sure we know what version of python we\u0027re testing in the gate","message":"probe: Make sure we know what version of python we\u0027re testing in the gate\n\nChange-Id: I25eafa0f1eee24b728dd55ad05589703a99f3393\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4ce2e01a190a70fed348b74d28690c2a65bb213d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4ce2e01a190a70fed348b74d28690c2a65bb213d"}]},"branch":"refs/heads/master"},"59046fb1f906af0baf921c84a37f6a1eeffa1c06":{"kind":"REWORK","_number":2,"created":"2021-05-27 21:35:11.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/03/793503/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/03/793503/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/03/793503/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/03/793503/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/03/793503/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/03/793503/2"}}},"commit":{"parents":[{"commit":"f7f1553edb7e3dbe57507113b730ee1f77472928","subject":"Merge \"Use ContainerSharderConf class in sharder and manage-shard-ranges\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f7f1553edb7e3dbe57507113b730ee1f77472928"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-05-27 20:51:30.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-05-27 21:35:01.000000000","tz":-420},"subject":"probe: Make sure we know what version of python we\u0027re testing in the gate","message":"probe: Make sure we know what version of python we\u0027re testing in the gate\n\nChange-Id: I25eafa0f1eee24b728dd55ad05589703a99f3393\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/59046fb1f906af0baf921c84a37f6a1eeffa1c06"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/59046fb1f906af0baf921c84a37f6a1eeffa1c06"}]},"branch":"refs/heads/master"},"ad259d1637dcb31e0d8124490af037cd28cde4c3":{"kind":"REWORK","_number":3,"created":"2021-06-08 22:16:26.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/03/793503/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/03/793503/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/03/793503/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/03/793503/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/03/793503/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/03/793503/3"}}},"commit":{"parents":[{"commit":"f7f1553edb7e3dbe57507113b730ee1f77472928","subject":"Merge \"Use ContainerSharderConf class in sharder and manage-shard-ranges\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f7f1553edb7e3dbe57507113b730ee1f77472928"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-05-27 20:51:30.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-06-08 22:16:16.000000000","tz":-420},"subject":"probe: Make sure we know what version of python we\u0027re testing in the gate","message":"probe: Make sure we know what version of python we\u0027re testing in the gate\n\nChange-Id: I25eafa0f1eee24b728dd55ad05589703a99f3393\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ad259d1637dcb31e0d8124490af037cd28cde4c3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ad259d1637dcb31e0d8124490af037cd28cde4c3"}]},"branch":"refs/heads/master"},"363934c061bdcfb5dab72707e47ee231d7a3d4af":{"kind":"REWORK","_number":4,"created":"2022-03-23 22:37:45.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/03/793503/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/03/793503/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/03/793503/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/03/793503/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/03/793503/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/03/793503/4"}}},"commit":{"parents":[{"commit":"f7f1553edb7e3dbe57507113b730ee1f77472928","subject":"Merge \"Use ContainerSharderConf class in sharder and manage-shard-ranges\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f7f1553edb7e3dbe57507113b730ee1f77472928"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-05-27 20:51:30.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-03-23 22:37:36.000000000","tz":-420},"subject":"probe: Make sure we know what version of python we\u0027re testing in the gate","message":"probe: Make sure we know what version of python we\u0027re testing in the gate\n\nChange-Id: I25eafa0f1eee24b728dd55ad05589703a99f3393\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/363934c061bdcfb5dab72707e47ee231d7a3d4af"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/363934c061bdcfb5dab72707e47ee231d7a3d4af"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
