)]}'
{"id":"openstack%2Fswift~793504","triplet_id":"openstack%2Fswift~stable%2Fwallaby~Idf6bf6a2c2700ecd19924dd71219d50ef04ef863","project":"openstack/swift","branch":"stable/wallaby","hashtags":[],"change_id":"Idf6bf6a2c2700ecd19924dd71219d50ef04ef863","subject":"DNM: test gate jobs","status":"ABANDONED","created":"2021-05-27 21:13:10.000000000","updated":"2021-05-27 22:24:32.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"866eb1f4ab503bcd088553d8e0e5799f48ab6d2c","_number":793504,"virtual_id_number":793504,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-05-27 22:21:06.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-27 22:21:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"38f39a4f83a9bd61ceba02ac7484e8366bb7e5b8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-05-27 21:13:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"11efd0d76967f96e859d2a5837da1d6a170e6e29","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-27 22:21:06.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a9a30e186594cc7a516ab8f7f7efbd4 : SUCCESS in 9m 46s\n- grenade https://zuul.opendev.org/t/openstack/build/13ac85cb1de943f8accb999cbb2628d3 : SUCCESS in 1h 03m 02s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/74f1a460ed834739b0bb9a72e5285ccc : SUCCESS in 45m 33s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/8f6c2c4f4345401ebbdbfe865b4afd17 : SUCCESS in 9m 23s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/c4b154d3784543b396b324a125c0e820 : SUCCESS in 6m 55s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/7e357f3c3de649868a82ae42e9ba8bdb : SUCCESS in 14m 36s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/ecc7e9cc55d44c54b40e4566785053be : SUCCESS in 15m 09s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/29fddd779eee44fc8273eb05e5d53a07 : SUCCESS in 14m 11s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/b87a69b040a047d9a2ef98f92f216b3c : SUCCESS in 15m 27s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/321e4936c8e84507b7eaaa14a6245eb7 : FAILURE in 15m 47s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/1b363f64ae84492a88d25a2f44bd067d : SUCCESS in 24m 31s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/0bb9d643a4ad450293f22e71684bc7a9 : SUCCESS in 20m 03s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/7c690adda3a0423c991a1a6e7422dc05 : SUCCESS in 25m 02s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/59328e0697aa4dd491c38c6f34ad71cb : SUCCESS in 17m 52s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/6a84472009a44ceaa34eb209d29ba082 : SUCCESS in 19m 07s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/82e9daf3c7534bb19913aa65b1253e10 : SUCCESS in 17m 59s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/819a5d0da3a44ad29d87aabdd310dd8b : SUCCESS in 6m 14s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/41ee7e33d53944639119256492feb59e : FAILURE in 24m 50s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/83fdcdbfc1364b2589e603d4c6cc0041 : SUCCESS in 42m 36s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/a260242c7c114baaa88fa0250f77dab7 : SUCCESS in 44m 13s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/8242c288753649c3952d2250e931bd86 : SUCCESS in 40m 53s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d71df358fb6c418486f5f64a046ce468 : SUCCESS in 27m 32s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3581adbbe0cc4ffaa00117115f547a84 : SUCCESS in 20m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0a8657c0ecf74ef1941b969020fc8790 : SUCCESS in 5m 34s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/2fbacb59ec434e08ba47105b98c3478d : SUCCESS in 27m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/02cdb994ba8340129c69fc1f50805e13 : SUCCESS in 59m 30s","accounts_in_message":[],"_revision_number":1},{"id":"866eb1f4ab503bcd088553d8e0e5799f48ab6d2c","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-05-27 22:24:32.000000000","message":"Abandoned\n\nSure enough. https://review.opendev.org/c/openstack/swift/+/793439 should fix it.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"bcd2507993b04efc858b2a3d6ca9d0d5335a3f2f","revisions":{"bcd2507993b04efc858b2a3d6ca9d0d5335a3f2f":{"kind":"REWORK","_number":1,"created":"2021-05-27 21:13:10.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/04/793504/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/04/793504/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/04/793504/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/04/793504/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/04/793504/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/04/793504/1"}}},"commit":{"parents":[{"commit":"f9ae900c32fd72f3bc8fd0bad9c0eb78f9eddab3","subject":"Merge \"symlink: Ensure headers are returned as a list on py3\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f9ae900c32fd72f3bc8fd0bad9c0eb78f9eddab3"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-05-27 21:12:15.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-05-27 21:12:24.000000000","tz":-420},"subject":"DNM: test gate jobs","message":"DNM: test gate jobs\n\nI suspect py39 unit tests will fail.\n\nChange-Id: Idf6bf6a2c2700ecd19924dd71219d50ef04ef863\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bcd2507993b04efc858b2a3d6ca9d0d5335a3f2f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bcd2507993b04efc858b2a3d6ca9d0d5335a3f2f"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
