)]}'
{"id":"openstack%2Fswift~793888","triplet_id":"openstack%2Fswift~master~I527f398ad0c8c844f90cf91b9b151e66a9341f9e","project":"openstack/swift","branch":"master","topic":"p-lonely-frags-v2","hashtags":[],"change_id":"I527f398ad0c8c844f90cf91b9b151e66a9341f9e","subject":"reconstructor: add skip_partition_count option","status":"NEW","created":"2021-06-01 06:37:34.000000000","updated":"2021-08-13 08:57:10.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"b0dfad080ef45de680552ce04bc247fab09ac3d4","_number":793888,"virtual_id_number":793888,"owner":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-08-13 08:03:29.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-01 07:43:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0808e8eba82764ee5ee4a44f87938f94cbc88255","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2021-06-01 06:37:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dfb547e2f3941d724249bf14279a8f44948d3b26","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-01 07:43:05.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de931908ac884a3eb677c02ba4c2f5ca : SUCCESS in 11m 57s\n- grenade https://zuul.opendev.org/t/openstack/build/241ebad4ac73495e8c260616e2bced68 : SUCCESS in 1h 02m 24s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/b3443239cc60407b9be5413143da90b3 : SUCCESS in 42m 47s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d3f16883631a4bbfa14a7b18b476bbff : SUCCESS in 12m 23s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/d4015afe3fde42e5a84cbdd1ce2aea21 : SUCCESS in 9m 55s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/22511cb419d94242bfabe0b3cb714c05 : SUCCESS in 17m 15s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/99a994fc97864710b18d367f143697e9 : SUCCESS in 15m 53s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/8493010442cf49d0b6554d66d006f6a7 : SUCCESS in 15m 48s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/4f8bb5745c9f42afb48d0285b26a1798 : SUCCESS in 16m 41s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/c6b93ce38c28436d99af34237d477e41 : SUCCESS in 22m 34s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/764aa5d3bca64eb19e49d0f00ec8edf8 : SUCCESS in 19m 38s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/104baa44882e4654955f3f5a510f3c52 : SUCCESS in 19m 51s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/8bfa01d3ac214ec385147a826e0dfb8a : SUCCESS in 17m 57s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/27987f69bb464d7da9a83f16823d417f : SUCCESS in 20m 53s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/163819f9726848a8b2e5597d14a9fb09 : SUCCESS in 18m 39s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/8b714d939dbe49819d56d3a8379eb2e2 : SUCCESS in 19m 51s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/5d31af75638243688ce16d188ae020ad : SUCCESS in 10m 02s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/0f95418cd1ed4d6b9c7166c9028d67b3 : SUCCESS in 21m 14s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/08001a0b97bd4fe6990b73c4e81cabad : SUCCESS in 57m 02s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/22153ac4c69e45c093814b600aef24bf : SUCCESS in 50m 52s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/df6984e40b764e92b56d9da40092f1e1 : SUCCESS in 37m 29s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f770e9cb279a49b8b404b5418ba65c6e : SUCCESS in 42m 52s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6a8ba638876940219551af829652dced : SUCCESS in 18m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/533636d41a7e495fa7097b1035fb855f : FAILURE in 7m 51s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/393df59e654e4d9b9eb144b78ef5187a : SUCCESS in 19m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d398a408f5474e018fb399f2d2044980 : SUCCESS in 54m 13s","accounts_in_message":[],"_revision_number":1},{"id":"dd289e1745523a5967bda85657ddf8c191b7a9bf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2021-06-07 03:35:54.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"15527eb60f01625722df4d53efca7a8bfeb071e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-07 04:44:24.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/03133f2cbfa546b480ec4dbe2429d7c6 : SUCCESS in 10m 18s\n- grenade https://zuul.opendev.org/t/openstack/build/0c94ef4791574663bc2729ada3714bdc : SUCCESS in 1h 04m 12s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/177d7094e998415d88b80fe3440b0ecc : SUCCESS in 47m 24s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f2340f7eaa834e83acac28c33d1b6b21 : SUCCESS in 7m 29s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/be876cff80c448cb8e3d4b344620c495 : SUCCESS in 9m 50s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/70563136157246fca82da1840910618a : SUCCESS in 14m 15s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/4a293c717364483ead80a275590efac9 : SUCCESS in 18m 15s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/5a8a478180614ee99d556a62fb6f6a32 : SUCCESS in 15m 38s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/024dbb4b33994fc0b609c1490cff9ae4 : SUCCESS in 14m 42s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/1e5ecb6e78354f93b2901488666c531e : SUCCESS in 29m 14s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/7efa66df8eb442e2829f8bc2d49b63d5 : SUCCESS in 16m 32s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/f382c2aab1944b9e94b4e5998f6be536 : SUCCESS in 33m 15s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/cd9d41be90af4ced9802aa9f1b93924a : SUCCESS in 18m 23s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/45c427a91f5e4a52bfdb9786f0671571 : SUCCESS in 16m 55s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/0e3d7261606b499aa49cfc79c71d6028 : SUCCESS in 32m 38s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/323c786074a84dc293f4b0612eae7a0d : SUCCESS in 19m 50s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/072bdb95d68a47e98bd51bc4ad3a1123 : SUCCESS in 6m 32s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/a5c8e76d7293437e9ffe56fdc2a5535f : FAILURE in 23m 22s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/310b31b24deb47888b0808dc9f666c60 : SUCCESS in 46m 44s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/59489c0880cf4ea4a09a40a10bffe8fd : SUCCESS in 48m 44s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/add5eb75d9e94d9185eb7f541debdb3d : SUCCESS in 40m 21s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/dd49dc73a5a64a24b7913f11d0848db5 : SUCCESS in 37m 30s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/484779f1bb1747cf8460f723ffefdaf4 : SUCCESS in 17m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ac08a32d94c343a99b4bc950998abfa0 : FAILURE in 5m 16s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c7599bb82519402dbe49c34e7955ee83 : SUCCESS in 28m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/76e9d2820ab844f08a858e1ba1035d73 : SUCCESS in 49m 08s","accounts_in_message":[],"_revision_number":2},{"id":"6aa7b4015fb737f68bcb5c199af0531be5ded2fa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2021-08-06 01:40:17.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"8ff25797bd49eb308c1bab0cf20430cf961abbee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-06 03:07:59.000000000","message":"Patch Set 3: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4da73affbb5b4b1eb6a542b8ab949f5b : SUCCESS in 9m 14s\n- grenade https://zuul.opendev.org/t/openstack/build/012734d2844844378d05dde69d5b78a1 : SUCCESS in 57m 27s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/ad3b10b287a24971b0767f72ee9b994b : FAILURE in 38m 41s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/6c570de0f3d4498bb868d38d2d777999 : SUCCESS in 8m 12s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/6215b434288e47ef997eab32e08bc2b0 : SUCCESS in 8m 23s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/2390900f48a24d66a85a755a36f21dd2 : SUCCESS in 15m 44s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/8c8ea809939b439fb7324f35f80afd18 : SUCCESS in 17m 49s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/4c33705a5fb54861baa8121bc3721ed9 : SUCCESS in 17m 48s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/68703b6c38064cdf9a232014bb8e6f07 : SUCCESS in 16m 45s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/a2c4ea99a43642d2b3d230b7ef6067d0 : SUCCESS in 18m 24s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/11098c420703434bafc484d560b6a460 : SUCCESS in 16m 42s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/ec00cd96f3e34035854c155880565de8 : SUCCESS in 18m 49s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/96fcfcec6d5945599a458df3a6b72482 : SUCCESS in 17m 48s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/e20b04c5cd954a009da5d242858fa63c : SUCCESS in 15m 59s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/7a0f347eb74f4676a6724883a11b45c6 : SUCCESS in 17m 00s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/0bbeef0c45cf4a6d90ff7e8e51baaace : SUCCESS in 16m 57s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e0caab31a03546ae93c5b96ba8305665 : SUCCESS in 6m 36s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/782deb582d154097b78a7133b9f6e57e : SUCCESS in 20m 02s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/1fff67f3c2234ef7a3fbe2b518dab794 : SUCCESS in 53m 12s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/bfd0232ae711422fb9e7ce893b0add86 : SUCCESS in 50m 37s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/57ac36741e704768b529e76381e141dc : SUCCESS in 40m 31s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/6b1c638109964c70aaf9fd97b47616b0 : SUCCESS in 41m 37s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/afbb353aaee948469514d8ec7008900c : SUCCESS in 20m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f7829a5f75c14655b4039cb4ee5b42b6 : FAILURE in 5m 49s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/1eef31085c8943d19dfe28370b211235 : SUCCESS in 23m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f2720363eb60416798ff1afe6199e0a2 : SUCCESS in 46m 39s","accounts_in_message":[],"_revision_number":3},{"id":"0faa4bda36aa48f1a66512c38256c4e47baedf5a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-06 03:44:34.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/807a299f8e154b8aa0a751f415c339db : SUCCESS in 48m 53s (non-voting)\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5adf5e49f6244228832f18022953fd33 : SUCCESS in 44m 22s (non-voting)\n- swift-probetests-centos-8-arm64 https://zuul.opendev.org/t/openstack/build/075cefd5da7444559e5ba46ef7a74d6b : TIMED_OUT in 2h 01m 22s (non-voting)\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/99f6fc3d02a744b38d6cae3e5a8d42ce : SUCCESS in 57m 30s (non-voting)\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/6ad08bababee4ff1893c0c5a93b6e66e : SUCCESS in 1h 03m 17s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"64b0bb6dcfda95dd696c9537e14d2c26e88a9b03","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2021-08-13 07:02:48.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"49030f1e9cc2cba8da209f34ed547efa3dce01c0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-13 08:03:29.000000000","message":"Patch Set 4: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/35aa44314bb4466fb93e08776cb2d5eb : SUCCESS in 9m 14s\n- grenade https://zuul.opendev.org/t/openstack/build/ccdffa8b1aee479facf2a09c7b6e72cc : SUCCESS in 57m 20s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/fedd285c5cc8427aac6f4b4cc7d5fb3b : SUCCESS in 38m 46s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/56db2f024f1148ca80ee6377314038a7 : SUCCESS in 8m 05s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/aa4f9f1f73b54e5db945ac0ccbc2a170 : SUCCESS in 10m 05s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/57de9efd079d4218a44221d1a57e0a07 : SUCCESS in 20m 27s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/e80d124f15874c468b8f73a1dc5975b9 : SUCCESS in 21m 54s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/02b79d49fa264771addd3f606c49131a : SUCCESS in 17m 19s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/776b6b5f99d14cf19d74f6328f136cde : SUCCESS in 15m 28s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/2325113b04184028a142cea93e03d81b : SUCCESS in 17m 13s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/de6abada864248538206e94ec2e895ec : SUCCESS in 20m 22s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/1fb5a879c05a430589148493bb07034f : SUCCESS in 18m 35s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/f0134557d9e141a9ae3f4a87d43170b5 : SUCCESS in 18m 46s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/8ebde4d7788a469b890c206ed1321721 : SUCCESS in 18m 49s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/48c7dea55b824fd0aca5c809353f2284 : SUCCESS in 16m 32s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/ff72e874bcf34012a9c0543eda5e33d8 : SUCCESS in 18m 29s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/18557a27b3084d4b8b5158c76c0563d5 : SUCCESS in 7m 01s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/922dd653933c4deba326dcfd2321cf79 : SUCCESS in 21m 56s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/7150d071dcf4498599cb71deb5da28d3 : SUCCESS in 55m 50s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/b3450d9ac07f4873b253b82925ff79cf : SUCCESS in 48m 32s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/36dd94bfb86446bf907f1655031548b0 : SUCCESS in 37m 44s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/db53db1169fa48fc8e2a32363989c39d : SUCCESS in 39m 20s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/75307f70764f4b03b32639d2baa041e4 : SUCCESS in 20m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9c8716f64c0d4dffa6803f0b59c4ffd1 : FAILURE in 6m 11s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/01ef3b2dac0e42a295ed65c7c61834ff : SUCCESS in 28m 48s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/094657638b8f4341a7f0b70d37d9a9e0 : SUCCESS in 37m 42s","accounts_in_message":[],"_revision_number":4},{"id":"b0dfad080ef45de680552ce04bc247fab09ac3d4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-13 08:57:10.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/21e1cc9641874bd58614e37654972af5 : SUCCESS in 30m 42s (non-voting)\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c6f54b5484bd4dfe9a8f69e910ce3a95 : SUCCESS in 44m 24s (non-voting)\n- swift-probetests-centos-8-arm64 https://zuul.opendev.org/t/openstack/build/5c98af03800f43c88e06398afbea25e7 : SUCCESS in 1h 14m 23s (non-voting)\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/f749feb03c354b5b85fb7f79a6cc23de : SUCCESS in 30m 37s (non-voting)\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/a2114cf8c5974557b41fadcfdce1ef25 : SUCCESS in 38m 09s (non-voting)","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"3554a0dd47324f9597764979e7de56718e93cc28","revisions":{"f9d06247a5172cae2d733e8770aef721a163d167":{"kind":"REWORK","_number":1,"created":"2021-06-01 06:37:34.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/88/793888/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/88/793888/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/88/793888/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/88/793888/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/88/793888/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/88/793888/1"}}},"commit":{"parents":[{"commit":"e3fb14a33f2d59b8549ab6f42065e3436b22b1d2","subject":"WIP: Reconstructor: Use past node and abort to handoff","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e3fb14a33f2d59b8549ab6f42065e3436b22b1d2"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2021-06-01 06:28:36.000000000","tz":600},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2021-06-01 06:37:14.000000000","tz":600},"subject":"reconstructor: add skip_partition_count option","message":"reconstructor: add skip_partition_count option\n\nAfter a rebalance, the next run on of the constructor may find frags missing\nfrom the partition that was moved. This means everytime we go rebuild a\nfragment we\u0027ll end up finding the missing fragment on the old primary.\nIn this circumstance it seems useless to continue looking for existing\nfrags. It would be better to track the number of fragements have been\nfound on a partition and if we get to a certain number, short circut and\nskip that partition.\n\nThat\u0027s what this patch does. It adds a `skip_partition_count` option,\ndefaulting to 0 (disabled), which when a partition reaches above this\ncount we start raising SsyncAbortDatafile\u0027s for that partition.\nThereby leaving it to the handoff to push its data to the new location.\n\nChange-Id: I527f398ad0c8c844f90cf91b9b151e66a9341f9e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f9d06247a5172cae2d733e8770aef721a163d167"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f9d06247a5172cae2d733e8770aef721a163d167"}]},"branch":"refs/heads/master"},"2cbf13b93559068d8501a10071afe6f8c30ae564":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-06-07 03:35:54.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/88/793888/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/88/793888/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/88/793888/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/88/793888/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/88/793888/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/88/793888/2"}}},"commit":{"parents":[{"commit":"c299c9208882293382fc1d73e1dd95698bbc184c","subject":"WIP: Reconstructor: Use past node and abort to handoff","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c299c9208882293382fc1d73e1dd95698bbc184c"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2021-06-01 06:28:36.000000000","tz":600},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2021-06-07 03:35:29.000000000","tz":600},"subject":"reconstructor: add skip_partition_count option","message":"reconstructor: add skip_partition_count option\n\nAfter a rebalance, the next run on of the constructor may find frags missing\nfrom the partition that was moved. This means everytime we go rebuild a\nfragment we\u0027ll end up finding the missing fragment on the old primary.\nIn this circumstance it seems useless to continue looking for existing\nfrags. It would be better to track the number of fragements have been\nfound on a partition and if we get to a certain number, short circut and\nskip that partition.\n\nThat\u0027s what this patch does. It adds a `skip_partition_count` option,\ndefaulting to 0 (disabled), which when a partition reaches above this\ncount we start raising SsyncAbortDatafile\u0027s for that partition.\nThereby leaving it to the handoff to push its data to the new location.\n\nChange-Id: I527f398ad0c8c844f90cf91b9b151e66a9341f9e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2cbf13b93559068d8501a10071afe6f8c30ae564"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2cbf13b93559068d8501a10071afe6f8c30ae564"}]},"branch":"refs/heads/master"},"7851c9c8e874cfb8498bbfcf9453927ddd4869d5":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-08-06 01:40:17.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/88/793888/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/88/793888/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/88/793888/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/88/793888/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/88/793888/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/88/793888/3"}}},"commit":{"parents":[{"commit":"c67491d15d5d550cd9def61b72a6623d167e915e","subject":"WIP: Reconstructor: Use past node and abort to handoff","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c67491d15d5d550cd9def61b72a6623d167e915e"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2021-06-01 06:28:36.000000000","tz":600},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2021-08-06 01:39:49.000000000","tz":600},"subject":"reconstructor: add skip_partition_count option","message":"reconstructor: add skip_partition_count option\n\nAfter a rebalance, the next run on of the constructor may find frags missing\nfrom the partition that was moved. This means everytime we go rebuild a\nfragment we\u0027ll end up finding the missing fragment on the old primary.\nIn this circumstance it seems useless to continue looking for existing\nfrags. It would be better to track the number of fragements have been\nfound on a partition and if we get to a certain number, short circut and\nskip that partition.\n\nThat\u0027s what this patch does. It adds a `skip_partition_count` option,\ndefaulting to 0 (disabled), which when a partition reaches above this\ncount we start raising SsyncAbortDatafile\u0027s for that partition.\nThereby leaving it to the handoff to push its data to the new location.\n\nChange-Id: I527f398ad0c8c844f90cf91b9b151e66a9341f9e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7851c9c8e874cfb8498bbfcf9453927ddd4869d5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7851c9c8e874cfb8498bbfcf9453927ddd4869d5"}]},"branch":"refs/heads/master"},"3554a0dd47324f9597764979e7de56718e93cc28":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2021-08-13 07:02:48.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/88/793888/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/88/793888/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/88/793888/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/88/793888/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/88/793888/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/88/793888/4"}}},"commit":{"parents":[{"commit":"bb98190eca2332a102fdf99e43360b837d652d78","subject":"WIP: Reconstructor: Use past node and abort to handoff","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bb98190eca2332a102fdf99e43360b837d652d78"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2021-06-01 06:28:36.000000000","tz":600},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2021-08-13 07:02:33.000000000","tz":600},"subject":"reconstructor: add skip_partition_count option","message":"reconstructor: add skip_partition_count option\n\nAfter a rebalance, the next run on of the constructor may find frags missing\nfrom the partition that was moved. This means everytime we go rebuild a\nfragment we\u0027ll end up finding the missing fragment on the old primary.\nIn this circumstance it seems useless to continue looking for existing\nfrags. It would be better to track the number of fragements have been\nfound on a partition and if we get to a certain number, short circut and\nskip that partition.\n\nThat\u0027s what this patch does. It adds a `skip_partition_count` option,\ndefaulting to 0 (disabled), which when a partition reaches above this\ncount we start raising SsyncAbortDatafile\u0027s for that partition.\nThereby leaving it to the handoff to push its data to the new location.\n\nChange-Id: I527f398ad0c8c844f90cf91b9b151e66a9341f9e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3554a0dd47324f9597764979e7de56718e93cc28"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3554a0dd47324f9597764979e7de56718e93cc28"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
