)]}'
{"id":"openstack%2Fswift~794666","triplet_id":"openstack%2Fswift~stable%2Ftrain~I10f94c03dbc50acd9e4dad853677bd484ca52538","project":"openstack/swift","branch":"stable/train","hashtags":[],"change_id":"I10f94c03dbc50acd9e4dad853677bd484ca52538","subject":"Authors/ChangeLog for 2.23.3","status":"MERGED","created":"2021-06-03 18:59:10.000000000","updated":"2021-06-03 22:23:42.000000000","submitted":"2021-06-03 22:21:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"794666","meta_rev_id":"6221746ba49e8d5caeeeb307565cfc8ede149fac","_number":794666,"virtual_id_number":794666,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-03 22:21:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-06-03 21:23:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-06-03 21:23:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-03 19:44:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-06-03 20:38:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"60d8bf04c9ef421b31c953eecac287ea4365b8fd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-06-03 18:59:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cd1637cebca05739214f7ed55b139867f28147c2","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-06-03 19:08:00.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"af87aa72edff54e8185907beddf0deb28a59ba7a","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-06-03 19:10:23.000000000","message":"Patch Set 1:\n\ncheck experimental","accounts_in_message":[],"_revision_number":1},{"id":"1e84958023a95094198f82abb3c1c2715aed4973","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-03 19:44:14.000000000","message":"Patch Set 1:\n\nBuild failed (experimental pipeline).\n\n- swift-tox-py27-centos-7 https://zuul.opendev.org/t/openstack/build/2dd6628cfdcc49bfab1f8504b2d818ae : SUCCESS in 17m 08s\n- swift-tox-func-py27-centos-7 https://zuul.opendev.org/t/openstack/build/b7d6205cc493453488dc8aeb82f12659 : SUCCESS in 14m 13s\n- swift-tox-func-encryption-py27-centos-7 https://zuul.opendev.org/t/openstack/build/5a964128307244a6a0805eeb998b01d9 : SUCCESS in 13m 45s\n- swift-tox-func-ec-py27-centos-7 https://zuul.opendev.org/t/openstack/build/246ba0476f224b6c9746181024c8d928 : SUCCESS in 13m 56s\n- swift-multinode-rolling-upgrade-queens https://zuul.opendev.org/t/openstack/build/b5d3acca5bab46f5bc101ed5009d7fc7 : FAILURE in 18m 47s\n- swift-multinode-rolling-upgrade-rocky https://zuul.opendev.org/t/openstack/build/a9893fcd1d6c4b42988369b5207bf038 : SUCCESS in 26m 35s\n- swift-multinode-rolling-upgrade-stein https://zuul.opendev.org/t/openstack/build/640e099151e1433e8b03ef92586bb3e9 : SUCCESS in 20m 36s","accounts_in_message":[],"_revision_number":1},{"id":"0bc1fbd931b5fb3cfe465b4fd1fe1e7bf6d6e8b8","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-06-03 20:16:30.000000000","message":"Patch Set 1:\n\nQueens rolling upgrade failure seems to just be an eventual-consistency issue.","accounts_in_message":[],"_revision_number":1},{"id":"e7a3a4c60c3c4bd595fb154c7d94376e5455b67b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-03 20:38:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d7827ae118424e648eb2e885bd575463 : SUCCESS in 5m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1223c16871054f53a3c4f0e8191dbd0e : SUCCESS in 5m 26s\n- grenade https://zuul.opendev.org/t/openstack/build/2afe4fe6e1c74891af64c74a8d3ee4e4 : SUCCESS in 48m 39s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/0b903e569e864e4a9943b5ff6daa0be7 : SUCCESS in 10m 20s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/eb861aec238c4f37aeaba2b35db7240f : SUCCESS in 13m 25s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/b1b3f662eb8b45e8972387ee82eb2d9b : SUCCESS in 20m 29s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/227972089c7f424aacf2bdbe8fdfdae5 : SUCCESS in 21m 40s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5f1c3205660a481080a99f7220fdc578 : SUCCESS in 17m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/81b294f57aea49f28078576764491674 : SUCCESS in 7m 31s","accounts_in_message":[],"_revision_number":1},{"id":"539d2e91838530f5a127b256e1b32cc765df8725","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-06-03 21:23:21.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3f050bb61d4cf278e832292a8d2a2c9267d4123e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-03 21:23:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2f128faccb9c344dec8970d8d551657ce78d3069","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-03 22:21:47.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/831003016a4546e1973acb5f420775b5 : SUCCESS in 5m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f4db27cafca344b5a3f50e38dc6c6e07 : SUCCESS in 5m 18s\n- grenade https://zuul.opendev.org/t/openstack/build/430458d25b224ff0954e9d5612df3acb : SUCCESS in 56m 27s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/25f2dcbc4c2f4b3c8ecca4ccf6b21ede : SUCCESS in 9m 53s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/7d3ec45b7a794cde8cb2fa6db28d035b : SUCCESS in 16m 13s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/8174e0bfb37745858bbf48f58b003359 : SUCCESS in 16m 16s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/828e7e00212446ceb65381854d1286de : SUCCESS in 16m 08s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/c5ffe53c171542a2ba07d2ecb9f2fb77 : SUCCESS in 12m 11s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/2100643662ed45c1a19ec7cbac90b292 : SUCCESS in 13m 28s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/fa20569bfc594556a31dee6908eef7f1 : SUCCESS in 13m 37s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/802daa2b27954eb0a7c7e0ff3118eb5b : SUCCESS in 13m 52s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/675a76afd50d410aaafc07baed78290b : SUCCESS in 16m 22s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/1d78b1b48ab4486589b1c2a43c74f588 : SUCCESS in 12m 06s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/b49761ef1be54a7fbac30ee96a4a6336 : SUCCESS in 11m 41s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/1a8b9b081bbf44b1bafa678113b070c7 : SUCCESS in 8m 56s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/ae8df75e9e4a45e0b603960d7d3a61d0 : SUCCESS in 15m 02s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7f6b1a14e08844d4bf9c6868ef40cd83 : SUCCESS in 14m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/faa76cb70bac4872aaec90f05881ceb6 : SUCCESS in 6m 30s","accounts_in_message":[],"_revision_number":1},{"id":"be94f8b487f45c4dd7daec09191cdbe5b83381ec","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-03 22:21:55.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"6221746ba49e8d5caeeeb307565cfc8ede149fac","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-03 22:23:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/48ce9d06863b42c5b0e8bdc9eecaa724 : SUCCESS in 1m 00s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cfd23e12b9ed4fad98a36b66a0cb0c39 : SUCCESS in 1m 05s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/ca5747fd16784278b176c26c16bd7848 : SUCCESS in 48s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"81d845f34e4d350c90af05117fcb0152958b4b3d","revisions":{"81d845f34e4d350c90af05117fcb0152958b4b3d":{"kind":"REWORK","_number":1,"created":"2021-06-03 18:59:10.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/66/794666/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/66/794666/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/794666/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/794666/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/794666/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/66/794666/1"}}},"commit":{"parents":[{"commit":"a989b47675688b761722e91e0ebc0dc85b1a074b","subject":"Merge \"tempurl: Fix PUT upload to temp url on py3\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a989b47675688b761722e91e0ebc0dc85b1a074b"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-06-03 18:57:21.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-06-03 18:57:37.000000000","tz":-420},"subject":"Authors/ChangeLog for 2.23.3","message":"Authors/ChangeLog for 2.23.3\n\nChange-Id: I10f94c03dbc50acd9e4dad853677bd484ca52538\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/81d845f34e4d350c90af05117fcb0152958b4b3d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/81d845f34e4d350c90af05117fcb0152958b4b3d"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
