)]}'
{"id":"openstack%2Fswift~795427","triplet_id":"openstack%2Fswift~master~Ie300e3d0602e6a5b154761886d7a7c6c11e5538a","project":"openstack/swift","branch":"master","topic":"extra_unrecoverable_ec_logging","hashtags":[],"change_id":"Ie300e3d0602e6a5b154761886d7a7c6c11e5538a","subject":"Make frags_with_data handling a little more explicit","status":"ABANDONED","created":"2021-06-08 23:57:47.000000000","updated":"2022-03-23 22:40:18.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"07b3ccf81b8d6c47380b6c70e235b3c41f966752","_number":795427,"virtual_id_number":795427,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:check","value":1,"date":"2021-06-09 05:33:44.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":-1,"date":"2021-06-14 18:46:43.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-09 00:53:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-06-09 01:48:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-14 18:46:43.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"}],"messages":[{"id":"d0c793b9e09da81d6c20fb61399054960ab8a3fb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-06-08 23:57:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8a2b66af9005e3e15231e06a86c625d3aad4d270","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-09 00:53:11.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/eb6fc3dceb8e4199bc3ccfc9c7469fc5 : SUCCESS in 45m 34s (non-voting)\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d9f92294a336444cbef3b8f8c53ff2b5 : SUCCESS in 53m 32s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a4aff601e6947a59464f0077ee78bd6d73d838bc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-09 01:48:40.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/801cf43164a5478e956263d864d8a990 : SUCCESS in 8m 52s\n- grenade https://zuul.opendev.org/t/openstack/build/008958358c3b4c3da82e6b1b02173d1d : SUCCESS in 1h 01m 58s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/9f0f712b6e8240fbaa13c8bd0629933f : SUCCESS in 45m 49s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/71394e2d7af6447b86d5fbdbc472a5d0 : SUCCESS in 7m 52s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/3a1b3fe4bba74b5b8274145dc92d1cb6 : SUCCESS in 8m 21s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/ff09ec81754b413e9be4f694d68bf072 : SUCCESS in 15m 54s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d7543eab5bf64759920e9ed1c62512c5 : SUCCESS in 16m 49s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/2c0300358281498aaa7529d97a13e587 : SUCCESS in 17m 15s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/65042699f0c348e6ae3a045f19573239 : SUCCESS in 17m 29s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/c064cf19f3d146548012bfbaad89a0ba : SUCCESS in 16m 11s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/caea332947884b32a4e9e3a7ad49ab18 : SUCCESS in 17m 07s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/102d49af688842f88f3718ff6ad7aedc : SUCCESS in 18m 03s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/ef5552cc5e874c668a703f4e4b9a7798 : SUCCESS in 17m 39s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/2e0c026912944dc19c96312e4bc189af : SUCCESS in 18m 08s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/0d93889d7d4c409684b067abb2f3dce4 : SUCCESS in 17m 56s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/578040e005e542c9812abeb825bfbe37 : SUCCESS in 26m 47s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/ba9f623315914d45904b76e90ebee2aa : SUCCESS in 6m 13s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/66d4e58c6c6e455794a7c8be33de51a4 : SUCCESS in 22m 14s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/90d61043eb4e49cfa4915506c8f02e3a : SUCCESS in 1h 07m 37s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/4e32de4237144807af7aedc5b12a741c : SUCCESS in 1h 07m 59s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/834540e56ed04442bf00a7ad326f3c15 : SUCCESS in 40m 42s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b4923a84ca7b4780a0c1a96ae2a7820e : SUCCESS in 39m 51s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a1622b97681240e1bc8d2bc0edeb5f31 : SUCCESS in 16m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c87813fd7bc441a985f54dd113b2e918 : SUCCESS in 5m 33s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ba02dba266884ee7aad48c8e21cdea83 : FAILURE in 28m 05s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/35753c2795a9408bbc63ba48f9a10dd9 : SUCCESS in 1h 03m 23s","accounts_in_message":[],"_revision_number":1},{"id":"dceb365764056ea0ca3b330507fc6b411ee64137","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-06-09 04:28:53.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"4a6708e6198ad4aac3d4d50f42abfd4caf91104b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-09 05:09:45.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c6788101e8b649b2b6e982ed8181e864 : SUCCESS in 38m 16s (non-voting)\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/62a8b0b779be4d5ebd58fb8106b0c528 : SUCCESS in 35m 51s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5e59e1b2354c475d9f6fef962fe0b9ae2658216f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-09 05:33:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4cbc28c158904387bc6723256c0c9eb4 : SUCCESS in 10m 04s\n- grenade https://zuul.opendev.org/t/openstack/build/fcc9e4ab27fb47acbf9594b5388c23df : SUCCESS in 58m 18s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/522ff233980a48738fa92ef416195bb9 : SUCCESS in 41m 50s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/4214a446496a46f58b1731cd99dbc56c : SUCCESS in 7m 41s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/a143cebf7e16410ab24557c0b0b50b95 : SUCCESS in 8m 21s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/20538c6baeac4f92a76fe5d503a4a979 : SUCCESS in 14m 11s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/0b2ac7ab578d44b985554bd4cb6531ce : SUCCESS in 17m 05s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/868d64f2a2964458b0cbb5c97fef179d : SUCCESS in 20m 32s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/38915cfc90cb4881aeb0a486aca202f9 : SUCCESS in 20m 11s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/596634f3b3e749058f013afdcf0f0f23 : SUCCESS in 16m 28s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/68a4d1b300a944c9a91da38c1c078dec : SUCCESS in 14m 26s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/3292d69794b2458792f5fc6c80e19dad : SUCCESS in 16m 07s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/dfa7408ed631451eaa97d93db3f6440f : SUCCESS in 19m 21s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/76dbe85436dd46ec9228ee9aad1570be : SUCCESS in 17m 35s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/e44d4c12846e4b2396146d8a380a3419 : SUCCESS in 16m 25s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/3e84d6da1d3242fe8640646181bba350 : SUCCESS in 21m 05s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/c416253633d94acf9a2f0aee5b376bbc : SUCCESS in 6m 28s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/6b21896aeb344ea0a06d0efea956a597 : SUCCESS in 24m 26s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/93e37f96202c40b69d7a2c1b947b3c77 : SUCCESS in 46m 24s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/50892210d73040d4ad5274d66923cb64 : SUCCESS in 49m 19s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/941d8257cb284f0fb6fe84c0bad3382f : SUCCESS in 40m 36s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c9d24f2319394e57a761a1b9f40d5f58 : SUCCESS in 40m 11s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/86f49e60cfec4925877fb5399c16b21d : SUCCESS in 16m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9c352fb4e4a1450a8ebbc2e66fadb4ed : SUCCESS in 6m 00s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b4fae8faf6f348d887cb3b7b7324c934 : SUCCESS in 30m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dc2dea6f04064a53887f99961ad71f3c : SUCCESS in 41m 57s","accounts_in_message":[],"_revision_number":1},{"id":"ae55a888de72b5c47189ee666c1f1f42f544dee3","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2021-06-14 18:46:43.000000000","message":"Patch Set 1: Code-Review-1\n\nif there\u0027s a technical reason for this change - i don\u0027t see it\n\nif it\u0027s just aesthetic - i mostly prefer the original phrasing\n\n1) I don\u0027t like explicit `\u003d\u003d 0: break` reminds me of `for (i \u003d 1; i \u003d\u003d 10; i++)``\n2) I don\u0027t like dangling elif, although the comment makes it slightly more bareable.","accounts_in_message":[],"_revision_number":1},{"id":"07b3ccf81b8d6c47380b6c70e235b3c41f966752","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-03-23 22:40:18.000000000","message":"Abandoned\n\nPurely aesthetic -- I didn\u0027t much care for the nested if\u0027s, but w/e","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fee2619c36a554102e36816ee5d6ee9473ed764e","revisions":{"fee2619c36a554102e36816ee5d6ee9473ed764e":{"kind":"REWORK","_number":1,"created":"2021-06-08 23:57:47.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/27/795427/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/27/795427/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/27/795427/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/27/795427/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/27/795427/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/27/795427/1"}}},"commit":{"parents":[{"commit":"e4c656bc59ddfda94be367053d762cdbbd1021fe","subject":"Add unit test for missing whole EC fragments","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e4c656bc59ddfda94be367053d762cdbbd1021fe"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-06-08 23:57:19.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-06-08 23:57:19.000000000","tz":-420},"subject":"Make frags_with_data handling a little more explicit","message":"Make frags_with_data handling a little more explicit\n\nChange-Id: Ie300e3d0602e6a5b154761886d7a7c6c11e5538a\nRelated-Change: If68e72d613685081a3149aedbfd07a69b8669d7f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fee2619c36a554102e36816ee5d6ee9473ed764e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fee2619c36a554102e36816ee5d6ee9473ed764e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
