)]}'
{"id":"openstack%2Fswift~796057","triplet_id":"openstack%2Fswift~master~If0b4cfc1ac9a8c66085eb4dc95366d43806d5ae2","project":"openstack/swift","branch":"master","topic":"fips-compatibiity","attention_set":{},"removed_from_attention_set":{"34120":{"account":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"last_update":"2022-01-27 00:11:07.000000000","reason":"Change was submitted"},"9914":{"account":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"last_update":"2022-01-27 00:11:07.000000000","reason":"Change was submitted"},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2022-01-26 19:36:25.000000000","reason":"Tim Burke replied on the change"}},"hashtags":[],"change_id":"If0b4cfc1ac9a8c66085eb4dc95366d43806d5ae2","subject":"Add FIPS CI jobs","status":"MERGED","created":"2021-06-11 19:35:42.000000000","updated":"2022-01-27 00:16:40.000000000","submitted":"2022-01-27 00:11:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":13,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"796057-fips-compatibiity","meta_rev_id":"cb7fb65d8e554906356adf715c302ebe233f0587","_number":796057,"virtual_id_number":796057,"owner":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-27 00:11:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-01-26 19:37:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-01-26 19:37:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"}],"CC":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-11 19:36:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-12 05:09:24.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"CC"},{"updated":"2021-10-18 18:07:06.000000000","updated_by":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"reviewer":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"state":"CC"},{"updated":"2021-10-19 13:22:21.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"CC"},{"updated":"2021-10-19 15:48:46.000000000","updated_by":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"reviewer":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"state":"CC"},{"updated":"2021-11-02 17:18:52.000000000","updated_by":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"reviewer":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"state":"REVIEWER"},{"updated":"2021-11-08 18:15:06.000000000","updated_by":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"CC"},{"updated":"2021-12-10 17:04:58.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2022-01-13 14:56:07.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"f8566dce5b77ec4cc940fbafc73119bf82d38e5f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2021-06-11 19:35:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0fd82c5841a8ee1aa9557b9142cce9251906cdf1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-11 19:36:04.000000000","message":"Patch Set 1: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/swift on branch master.  The error was:\n\n  found undefined alias\n  in \"openstack/swift/.zuul.yaml@master\", line 593, column 31","accounts_in_message":[],"_revision_number":1},{"id":"e96d9e7fcf466684a94e5f17b2f493fe2deec153","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-11 19:36:15.000000000","message":"Patch Set 1:\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/swift on branch master.  The error was:\n\n  found undefined alias\n  in \"openstack/swift/.zuul.yaml@master\", line 593, column 31","accounts_in_message":[],"_revision_number":1},{"id":"02daee1cc19f5a9e3c8bda65211b771a98ef9889","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2021-06-11 19:39:50.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"7dac8e2b4cf1907cc00f3a1200a073845778163d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-11 21:15:14.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/a83ca980b2ad4502888fc895e8eebce8 : FAILURE in 24m 48s\n- swift-tox-func-py36-centos-8-fips https://zuul.opendev.org/t/openstack/build/980acef81d4c43eeb635158f475a2b79 : SUCCESS in 13m 04s\n- swift-tox-func-encryption-py36-centos-8-fips https://zuul.opendev.org/t/openstack/build/3a1ac441c997443292a2a3bd067eb6f0 : TIMED_OUT in 1h 31m 11s\n- swift-tox-func-ec-py36-centos-8-fips https://zuul.opendev.org/t/openstack/build/428bf14aef3e4d60805a2c89d01716b7 : SUCCESS in 13m 18s","accounts_in_message":[],"_revision_number":2},{"id":"2ee3355f908b1997cd67a85ce22a9baa6e3d313f","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-06-12 05:09:24.000000000","message":"Patch Set 2:\n\n(1 comment)\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"1df6304ccfbaff72e5675acad01674643e3d931c","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2021-06-16 19:43:36.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"5bd9c3223098604c5c5e2d3095085939426d1595","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2021-06-16 19:45:32.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"0d0ba905e06e2b5b2ff56e2ae0611b2506ee76f7","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2021-06-16 19:46:18.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"a61906858106984cf2e7805bb4e1883e29ffe237","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-16 21:22:47.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/fee5cc89b0904b92ac14c12b6a083611 : SUCCESS in 18m 07s\n- swift-tox-func-py36-centos-8-fips https://zuul.opendev.org/t/openstack/build/8dd517ad2bb8456192596eccf6822f01 : SUCCESS in 13m 23s\n- swift-tox-func-encryption-py36-centos-8-fips https://zuul.opendev.org/t/openstack/build/8df39762bba64d399e09f93d98a0a8c7 : TIMED_OUT in 1h 31m 10s\n- swift-tox-func-ec-py36-centos-8-fips https://zuul.opendev.org/t/openstack/build/9704c04bb72a46d1a386f58a8d3a81ed : SUCCESS in 13m 30s","accounts_in_message":[],"_revision_number":3},{"id":"bb71329d14f24271271d4c81b42520a8cd951f81","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-06-18 19:14:28.000000000","message":"Patch Set 3:\n\ncheck experimental","accounts_in_message":[],"_revision_number":3},{"id":"c0bd52e473f446797e93c25153fed5713043ee0e","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-18 19:14:48.000000000","message":"Patch Set 3:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":3},{"id":"275fc2f73d0afe1a8168868fbf6c5aa0ce83baa7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2021-06-18 19:55:48.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"40bf5781b946dfbb90d4ebc138702b2fdea6de77","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-18 19:56:11.000000000","message":"Patch Set 4: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/swift on branch master.  The error was:\n\n  found undefined alias\n  in \"openstack/swift/.zuul.yaml@master\", line 607, column 31","accounts_in_message":[],"_revision_number":4},{"id":"eaec1b7b673310139acaad36707d3271277caab5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-18 19:56:24.000000000","message":"Patch Set 4:\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/swift on branch master.  The error was:\n\n  found undefined alias\n  in \"openstack/swift/.zuul.yaml@master\", line 607, column 31","accounts_in_message":[],"_revision_number":4},{"id":"1b3a25a5dc8045166b806e6b80780b8cc5f58425","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2021-06-18 20:05:57.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"7bb262b09fb048583598ee83a9991f6c7c2f46f7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-18 21:37:41.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- swift-tox-func-encryption-py36-centos-8 https://zuul.opendev.org/t/openstack/build/3373095d598d4c008b5fff8059a65c79 : SUCCESS in 20m 28s\n- swift-tox-func-py36-centos-8-fips https://zuul.opendev.org/t/openstack/build/a7e052dca0d94221a03e2a3618b89487 : SUCCESS in 17m 18s\n- swift-tox-func-encryption-py36-centos-8-fips https://zuul.opendev.org/t/openstack/build/e0aa824ee01f449387fbb13202897cf3 : TIMED_OUT in 1h 31m 12s\n- swift-tox-func-ec-py36-centos-8-fips https://zuul.opendev.org/t/openstack/build/f1245641e5624e75b7dc875a41d903b0 : SUCCESS in 24m 24s","accounts_in_message":[],"_revision_number":5},{"id":"d68d7f12dffe67a048e5a9f8347036b90605a54e","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-06-18 23:07:37.000000000","message":"Patch Set 5:\n\ncheck experimental","accounts_in_message":[],"_revision_number":5},{"id":"294232b74a7a1d29bbee0071698e03ff68aec6be","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-18 23:40:48.000000000","message":"Patch Set 5:\n\nBuild succeeded (experimental pipeline).\n\n- swift-tox-py27-centos-7 https://zuul.opendev.org/t/openstack/build/34fb891fb0c54392bd77da348bb50464 : SUCCESS in 17m 13s\n- swift-tox-func-py27-centos-7 https://zuul.opendev.org/t/openstack/build/47b7da2de07049f3a8b1c4ffbd97936c : SUCCESS in 19m 48s\n- swift-tox-func-encryption-py27-centos-7 https://zuul.opendev.org/t/openstack/build/bcedaadacb5e4dfba0435bd333333bbc : SUCCESS in 20m 36s\n- swift-tox-func-ec-py27-centos-7 https://zuul.opendev.org/t/openstack/build/2a463854a96e49a58c41430a9ee59822 : SUCCESS in 21m 42s\n- swift-tox-py36-centos-8 https://zuul.opendev.org/t/openstack/build/f319c651a81d4e8dba49b7a78e4ed5bf : SUCCESS in 17m 06s\n- swift-tox-func-py36-centos-8 https://zuul.opendev.org/t/openstack/build/f88e864e7b5b4dff82c07bb8c758a68d : SUCCESS in 18m 50s\n- swift-tox-func-encryption-py36-centos-8 https://zuul.opendev.org/t/openstack/build/4d98ef96af5f4f83aa65895b978e152b : SUCCESS in 18m 57s\n- swift-tox-func-ec-py36-centos-8 https://zuul.opendev.org/t/openstack/build/b1a719f419c6402c8209d67affa112dc : SUCCESS in 19m 48s\n- swift-multinode-rolling-upgrade-rocky https://zuul.opendev.org/t/openstack/build/9c4c284b8a4e4f56a07443d7d8371d42 : SUCCESS in 20m 18s\n- swift-multinode-rolling-upgrade-stein https://zuul.opendev.org/t/openstack/build/8b7de7b5b6c746e7b23ed3e6c2a84da2 : SUCCESS in 20m 19s\n- swift-multinode-rolling-upgrade-train https://zuul.opendev.org/t/openstack/build/d5db3d05882c45c49d49809896d456e2 : SUCCESS in 21m 21s\n- swift-multinode-rolling-upgrade-ussuri https://zuul.opendev.org/t/openstack/build/3e4cabc0e04442c4bf3f63e184b76b77 : SUCCESS in 25m 31s\n- swift-multinode-rolling-upgrade-victoria https://zuul.opendev.org/t/openstack/build/18cb71b605734b0787df03213bde864d : SUCCESS in 25m 26s\n- swift-multinode-rolling-upgrade-wallaby https://zuul.opendev.org/t/openstack/build/2693f468aa8e451e9f38bb15ca2ed617 : SUCCESS in 22m 45s\n- swift-multinode-rolling-upgrade-master https://zuul.opendev.org/t/openstack/build/0bc8c9cdf25d408881d64546446084cb : SUCCESS in 24m 39s","accounts_in_message":[],"_revision_number":5},{"id":"8d3cf2d05ffdad877ea6816a49e11c9aa2157e34","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2021-07-12 17:52:00.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"dd62425bcd18178c9b5957a02cf77cb0d3e98e36","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-12 19:25:15.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- swift-tox-func-encryption-py36-centos-8 https://zuul.opendev.org/t/openstack/build/8ead5860e4814167bb005ccb6f03e538 : SUCCESS in 18m 38s\n- swift-tox-func-py36-centos-8-fips https://zuul.opendev.org/t/openstack/build/acefc3dc83a94bc8bda40d7842371c0f : SUCCESS in 13m 25s\n- swift-tox-func-encryption-py36-centos-8-fips https://zuul.opendev.org/t/openstack/build/c374536121874ca1b746df24bb9c6c98 : TIMED_OUT in 1h 31m 33s\n- swift-tox-func-ec-py36-centos-8-fips https://zuul.opendev.org/t/openstack/build/172596ca0c21470b8d8a96d860011935 : SUCCESS in 13m 51s","accounts_in_message":[],"_revision_number":6},{"id":"fac7794f4c481566c9ccee24089211ca2ab17f17","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2021-08-17 21:53:46.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"8b3114fb820e6de1aedece8bacf1465c46728aed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-17 23:46:31.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- swift-tox-func-encryption-py36-centos-8 https://zuul.opendev.org/t/openstack/build/34902099446f4c8280c3493f333193cc : SUCCESS in 20m 06s\n- swift-tox-func-py36-centos-8-fips https://zuul.opendev.org/t/openstack/build/896b261e79df4f6c9332a484f89d0e73 : SUCCESS in 14m 10s\n- swift-tox-func-encryption-py36-centos-8-fips https://zuul.opendev.org/t/openstack/build/1465b65209c84e20b52a133c12a9e3aa : TIMED_OUT in 1h 31m 12s\n- swift-tox-func-ec-py36-centos-8-fips https://zuul.opendev.org/t/openstack/build/fb2f2b6631d2419fb8d47c5428ef6a07 : SUCCESS in 12m 58s","accounts_in_message":[],"_revision_number":7},{"id":"d86c94c8beda4261760d004acc4fa680fa4201d0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2021-08-23 17:41:51.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"826d0cf69adaba199cfac2f006f262fb02e8b7ae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-23 19:18:19.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- swift-tox-func-encryption-py36-centos-8 https://zuul.opendev.org/t/openstack/build/53d96443f659434292b8dc9c6bf48d64 : SUCCESS in 18m 13s\n- swift-tox-func-py36-centos-8-fips https://zuul.opendev.org/t/openstack/build/64c242273b8b4243a8d2e05c47dcdb8f : SUCCESS in 14m 25s\n- swift-tox-func-encryption-py36-centos-8-fips https://zuul.opendev.org/t/openstack/build/e37f5c0e6533423e8664a077cbbdb25e : TIMED_OUT in 1h 31m 24s\n- swift-tox-func-ec-py36-centos-8-fips https://zuul.opendev.org/t/openstack/build/1cca634d4e924bf6aba6437e6052806c : SUCCESS in 13m 38s","accounts_in_message":[],"_revision_number":8},{"id":"bfd0d18a310087c732e5d70d9f008e3ae9ce8618","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2021-10-04 15:02:32.000000000","message":"Patch Set 8:\n\nrecheck","accounts_in_message":[],"_revision_number":8},{"id":"7b44958f0176b9c4d09cb527a272eb9c1bc573c0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-04 16:38:07.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- swift-tox-func-encryption-py36-centos-8 https://zuul.opendev.org/t/openstack/build/9f71f576eb09468f99b151e35ce14df7 : SUCCESS in 21m 34s\n- swift-tox-func-py36-centos-8-fips https://zuul.opendev.org/t/openstack/build/299d9ae7884e4beab19957ecf9428eee : SUCCESS in 11m 52s\n- swift-tox-func-encryption-py36-centos-8-fips https://zuul.opendev.org/t/openstack/build/c1c1a44dcb7d42e08d16f7e929e00378 : TIMED_OUT in 1h 31m 29s\n- swift-tox-func-ec-py36-centos-8-fips https://zuul.opendev.org/t/openstack/build/ea56aa410b194076932404dd49f4dffb : SUCCESS in 15m 37s","accounts_in_message":[],"_revision_number":8},{"id":"c37b692c6881f0fe793f0cc0b32a5d12a418c8b6","author":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"date":"2021-10-18 18:07:06.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"eadf24eeee1ab368b148830b53aa176e63feb2ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-18 19:42:34.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- swift-tox-func-encryption-py36-centos-8 https://zuul.opendev.org/t/openstack/build/0e134a620d9e4b2cb04143b23b3903c7 : SUCCESS in 17m 14s\n- swift-tox-func-py36-centos-8-fips https://zuul.opendev.org/t/openstack/build/ac6d3af4a2214afab138beb0a350c248 : SUCCESS in 14m 47s\n- swift-tox-func-encryption-py36-centos-8-fips https://zuul.opendev.org/t/openstack/build/8fefe2da3d754c9484f2cdd2090eb484 : TIMED_OUT in 1h 31m 14s\n- swift-tox-func-ec-py36-centos-8-fips https://zuul.opendev.org/t/openstack/build/07c036ca17b24f2ea10292a291660dd0 : SUCCESS in 13m 18s","accounts_in_message":[],"_revision_number":8},{"id":"68f9ea5e39ceabe64a7af485371a7f6383883911","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2021-10-19 13:22:21.000000000","message":"Patch Set 8:\n\nrecheck again since there\u0027s now and autohold set for the swift-tox-func-encryption-py36-centos-8-fips job","accounts_in_message":[],"_revision_number":8},{"id":"49649395772739c8c5a063f5ad5eba5c1cfee01e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-19 14:58:37.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- swift-tox-func-encryption-py36-centos-8 https://zuul.opendev.org/t/openstack/build/a5e0aed0f73f4a569d4d646237da11fc : SUCCESS in 19m 28s\n- swift-tox-func-py36-centos-8-fips https://zuul.opendev.org/t/openstack/build/693b68bb6c4b41138540490c172c18a9 : SUCCESS in 15m 56s\n- swift-tox-func-encryption-py36-centos-8-fips https://zuul.opendev.org/t/openstack/build/1b8a078112174a7c88e6527b716399c6 : TIMED_OUT in 1h 31m 48s\n- swift-tox-func-ec-py36-centos-8-fips https://zuul.opendev.org/t/openstack/build/73b72216ac9447159bcf13f932d8acf8 : SUCCESS in 13m 40s","accounts_in_message":[],"_revision_number":8},{"id":"d4e066099ddb944d54dccbbf9f488074717e2441","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"date":"2021-11-02 17:02:56.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"3bd182cdcb49208b2761071a008365a48e40948d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"date":"2021-11-02 17:06:57.000000000","message":"Patch Set 10: Commit message was updated.","accounts_in_message":[],"_revision_number":10},{"id":"17f6f9a841322bdb36bc665632c94002d2afae41","author":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"date":"2021-11-02 17:18:52.000000000","message":"Patch Set 10: Code-Review-1","accounts_in_message":[],"_revision_number":10},{"id":"b03a17f29a8e4b881d0f30f96749213ef435f348","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"date":"2021-11-02 18:03:53.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"a60e6ecbf47b1e5e0bd7f67075e3db1ae63d06de","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"date":"2021-11-02 18:04:31.000000000","message":"Patch Set 12: Commit message was updated.","accounts_in_message":[],"_revision_number":12},{"id":"84c6502cae415bde039c2387abba27e968081259","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-02 18:52:37.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- swift-tox-func-encryption-py36-centos-8 https://zuul.opendev.org/t/openstack/build/343c5758a4c5498aaa2f763ed443bb8b : SUCCESS in 18m 22s\n- swift-tox-func-py36-centos-8-fips https://zuul.opendev.org/t/openstack/build/ff9057dc05e44f0a807952e0afab6857 : SUCCESS in 21m 44s\n- swift-tox-func-encryption-py36-centos-8-fips https://zuul.opendev.org/t/openstack/build/c9b332aa05f340d4b016399bda800220 : SUCCESS in 22m 26s\n- swift-tox-func-ec-py36-centos-8-fips https://zuul.opendev.org/t/openstack/build/c40b695d243f4d1080fea6e486cf2445 : SUCCESS in 31m 24s","accounts_in_message":[],"_revision_number":12},{"id":"b8ef519216347d90b04735352004ee52dd952ce6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"date":"2021-11-05 15:37:27.000000000","message":"Patch Set 13: Commit message was updated.","accounts_in_message":[],"_revision_number":13},{"id":"81c2cd355cb307a9fbff9507fb65fff1e68b87b6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-05 16:04:51.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\n\n- swift-tox-func-encryption-py36-centos-8 https://zuul.opendev.org/t/openstack/build/38176e1da45343e28d83dc8b84ebf0da : SUCCESS in 21m 17s\n- swift-tox-func-py36-centos-8-fips https://zuul.opendev.org/t/openstack/build/642c4aa5eed54420b402333c56d8e2d4 : SUCCESS in 17m 32s\n- swift-tox-func-encryption-py36-centos-8-fips https://zuul.opendev.org/t/openstack/build/c860dfff94a94fe7a31af3489074be69 : SUCCESS in 19m 30s\n- swift-tox-func-ec-py36-centos-8-fips https://zuul.opendev.org/t/openstack/build/b455727dfdab4c0c8b912a794fa1cb7b : SUCCESS in 18m 49s","accounts_in_message":[],"_revision_number":13},{"id":"4f0c2979338b25fb40f63c9312ca1734732c763e","author":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"date":"2021-11-05 16:23:27.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"c52656e8036a4c2bc32fc594e55b2a5438a82b1f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-05 16:23:55.000000000","message":"Patch Set 13: Verified-1\n\nThis change depends on a change with an invalid configuration.","accounts_in_message":[],"_revision_number":13},{"id":"9eba41927e0db2409b16b9bf99c0ea89930557d9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-05 16:24:30.000000000","message":"Patch Set 13:\n\nThis change depends on a change with an invalid configuration.","accounts_in_message":[],"_revision_number":13},{"id":"c9742645521d736b5a2852a95122d905e80abc5d","author":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"date":"2021-11-05 16:37:18.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"af72ff173645482b603366a9c232c596844a1914","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-05 17:07:22.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\n\n- swift-tox-func-encryption-py36-centos-8 https://zuul.opendev.org/t/openstack/build/2b7ab1eb548c4e41aa11367219eaab78 : SUCCESS in 18m 23s\n- swift-tox-func-py36-centos-8-fips https://zuul.opendev.org/t/openstack/build/160be56f1a934e578ecd4f44d7d7bbc5 : SUCCESS in 24m 06s\n- swift-tox-func-encryption-py36-centos-8-fips https://zuul.opendev.org/t/openstack/build/addc4e83a9a14f9b8c9c751299b632ff : SUCCESS in 20m 59s\n- swift-tox-func-ec-py36-centos-8-fips https://zuul.opendev.org/t/openstack/build/fba5a8e0115b49099a59db2f2d720269 : SUCCESS in 23m 12s","accounts_in_message":[],"_revision_number":13},{"id":"52422f0c3fffa6a5c873bcd0109ec7b5df4dd58a","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"date":"2021-11-17 17:16:06.000000000","message":"Topic change-796057 removed","accounts_in_message":[],"_revision_number":13},{"id":"7f4d2d98940cad78a30c36c154c634e6d1725a3c","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"date":"2021-11-17 17:16:11.000000000","message":"Topic set to add_fips_job","accounts_in_message":[],"_revision_number":13},{"id":"ea21dab27e30f6691e95d907976d7e02fd77a309","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-12-10 17:04:58.000000000","message":"Patch Set 13: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":13},{"id":"89a782f257cf52cac46478c9e9e53e0a10df85fe","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-12-10 22:02:39.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"db1e817c72584ff2a8d912e33a299f67f6989d9e","author":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"date":"2022-01-13 09:44:17.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"9242216c4541bd0e89aad4842528af3739ad7469","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"date":"2022-01-13 09:55:11.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"936ceb964213b78556b7b7da9635b2bba3add019","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"date":"2022-01-13 14:55:44.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"793c51f8c00fdf8f8f0e617317d9349e535fa0d4","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2022-01-13 14:56:07.000000000","message":"Patch Set 15:\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/swift on branch master.  The error was:\n\n  found duplicate anchor; first occurrence\n  in \"openstack/swift/.zuul.yaml@master\", line 631, column 31\nsecond occurrence\n  in \"openstack/swift/.zuul.yaml@master\", line 661, column 31","accounts_in_message":[],"_revision_number":15},{"id":"d62ef4bad4accd2f2dcb1d904aee1d1649fd59ec","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-13 14:56:46.000000000","message":"Patch Set 15:\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/swift on branch master.  The error was:\n\n  found duplicate anchor; first occurrence\n  in \"openstack/swift/.zuul.yaml@master\", line 631, column 31\nsecond occurrence\n  in \"openstack/swift/.zuul.yaml@master\", line 661, column 31","accounts_in_message":[],"_revision_number":15},{"id":"4b286ac3ef8f6046419e3c3576e7664dbe08e342","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-13 14:57:16.000000000","message":"Patch Set 15: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/swift on branch master.  The error was:\n\n  found duplicate anchor; first occurrence\n  in \"openstack/swift/.zuul.yaml@master\", line 631, column 31\nsecond occurrence\n  in \"openstack/swift/.zuul.yaml@master\", line 661, column 31","accounts_in_message":[],"_revision_number":15},{"id":"77c34902c8bbf2c171138e83f237d95b076e41ca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"date":"2022-01-13 14:57:20.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"bafd2f43171d110258240fae528a1cd8b3a8df89","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2022-01-13 14:57:40.000000000","message":"Patch Set 16:\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/swift on branch master.  The error was:\n\n  found duplicate anchor; first occurrence\n  in \"openstack/swift/.zuul.yaml@master\", line 631, column 31\nsecond occurrence\n  in \"openstack/swift/.zuul.yaml@master\", line 660, column 31","accounts_in_message":[],"_revision_number":16},{"id":"c6f47ce56d3f2bdaa8699c63b3adb7fbf38a93c3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-13 14:58:24.000000000","message":"Patch Set 16:\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/swift on branch master.  The error was:\n\n  found duplicate anchor; first occurrence\n  in \"openstack/swift/.zuul.yaml@master\", line 631, column 31\nsecond occurrence\n  in \"openstack/swift/.zuul.yaml@master\", line 660, column 31","accounts_in_message":[],"_revision_number":16},{"id":"609f39e159d1c577a1be6e3bedb3bb7afd6a7ebc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-13 14:58:56.000000000","message":"Patch Set 16: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/swift on branch master.  The error was:\n\n  found duplicate anchor; first occurrence\n  in \"openstack/swift/.zuul.yaml@master\", line 631, column 31\nsecond occurrence\n  in \"openstack/swift/.zuul.yaml@master\", line 660, column 31","accounts_in_message":[],"_revision_number":16},{"id":"cc1d679939a8d614230239524812a0114982ee38","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"date":"2022-01-13 15:01:40.000000000","message":"Patch Set 17: Commit message was updated.","accounts_in_message":[],"_revision_number":17},{"id":"ad211b329bcacdb552e5ccf1ee8bbb4d7d7e8957","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2022-01-13 15:02:01.000000000","message":"Patch Set 17:\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/swift on branch master.  The error was:\n\n  found duplicate anchor; first occurrence\n  in \"openstack/swift/.zuul.yaml@master\", line 631, column 31\nsecond occurrence\n  in \"openstack/swift/.zuul.yaml@master\", line 660, column 31","accounts_in_message":[],"_revision_number":17},{"id":"4593a2dddcdfcdc150e1316ff1f0f88b9ed8176f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-13 15:02:40.000000000","message":"Patch Set 17:\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/swift on branch master.  The error was:\n\n  found duplicate anchor; first occurrence\n  in \"openstack/swift/.zuul.yaml@master\", line 631, column 31\nsecond occurrence\n  in \"openstack/swift/.zuul.yaml@master\", line 660, column 31","accounts_in_message":[],"_revision_number":17},{"id":"4c7ff5a86ba408659f5e86b401ae9ce6f0f3935e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-13 15:03:10.000000000","message":"Patch Set 17: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/swift on branch master.  The error was:\n\n  found duplicate anchor; first occurrence\n  in \"openstack/swift/.zuul.yaml@master\", line 631, column 31\nsecond occurrence\n  in \"openstack/swift/.zuul.yaml@master\", line 660, column 31","accounts_in_message":[],"_revision_number":17},{"id":"d731b7ffeecc7f4380b0a6b4ee1881983e77d274","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"date":"2022-01-14 10:29:36.000000000","message":"Uploaded patch set 18.","accounts_in_message":[],"_revision_number":18},{"id":"8a5b99200657fdf1944b22ca80adeb58db019f23","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-14 11:43:25.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dfbd2bc3df4b464ba67cd6f0a2cf6c3d : SUCCESS in 10m 05s\n- grenade https://zuul.opendev.org/t/openstack/build/318d271731814bf2a7b9437fc6feb05d : SUCCESS in 1h 02m 33s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c00f38277bd144e6823d3d1bb15583de : SUCCESS in 1h 04m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9f91846275eb4cfe896010874b42e030 : SUCCESS in 1h 01m 07s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f401be4dfebe48a7b4969459dbf2ce38 : SUCCESS in 8m 23s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/f2a58a8357174156bc083414d58b93f4 : SUCCESS in 5m 36s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/c66dd62488ae4505ad185d7b2cc5ca35 : SUCCESS in 21m 51s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/f92cb60f13ea4727a03a98713643a18e : SUCCESS in 15m 20s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/aac0aadafff04c6ea9205034f7149756 : SUCCESS in 15m 06s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/576797cfa2aa4802b33009e533c9fe67 : SUCCESS in 15m 55s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/3878a6fcd4704e73ab8c9ace2703528e : SUCCESS in 22m 52s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/c2b91584407d4527903ae7d96844a272 : SUCCESS in 22m 39s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/1a0748cc126d4ec2a498c08facb96f4d : SUCCESS in 15m 36s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/c5e94fcfc89b4e4f8f59a05cf0058588 : SUCCESS in 24m 25s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/9d498d3f57ea477b979d00a1272d6b0c : SUCCESS in 15m 58s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/9d385e1489c34e208751d153c97dd446 : SUCCESS in 6m 20s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/8054bf1b45be4819ab9f5673619c1172 : SUCCESS in 20m 59s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/cc40258c254942f2a768dbdbd748486e : SUCCESS in 49m 44s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/659fea2ce3044cab83e45817f2873160 : SUCCESS in 52m 16s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/dbb26cdd4f9f4066a0109b1edbe0d483 : SUCCESS in 39m 43s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e2963a4b974e4e79bc843ff0fb3a01a2 : SUCCESS in 39m 22s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/99e6730d4847425f8c6d99d3d6046f0d : SUCCESS in 15m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/68c24ea33ce7450f8d66034c12e5bc80 : SUCCESS in 5m 40s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e308e44aab494e21ac81351ddf99504a : SUCCESS in 20m 06s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2fb473cdbcfb4fbdb338595d55106dc9 : SUCCESS in 1h 05m 28s","accounts_in_message":[],"_revision_number":18},{"id":"e5310218631e5c4e838c07cf344d4b184149df40","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-14 13:38:34.000000000","message":"Patch Set 18:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/81ea59f0c79a47e5925d2286aaefca2b : SUCCESS in 31m 50s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/48dc45da2bae40c7bdb01de5ae1f7719 : SUCCESS in 31m 13s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/b21b954c828f4f7e927f9b30e9b0f17b : TIMED_OUT in 3h 02m 28s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/35e069060e874692a46d61e312bc4ca4 : SUCCESS in 54m 13s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/7643b4243b2b4a47939f4daddfb63bbd : SUCCESS in 47m 31s","accounts_in_message":[],"_revision_number":18},{"id":"2034f25fecccfa1e2224646cea5858f24baf4854","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"date":"2022-01-14 15:06:44.000000000","message":"Uploaded patch set 19.","accounts_in_message":[],"_revision_number":19},{"id":"f570322444d3c8b25397b3b97e64b39776da6232","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-14 16:13:19.000000000","message":"Patch Set 19: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9eb54aa68f9c4f28802592526803c88e : SUCCESS in 10m 59s\n- grenade https://zuul.opendev.org/t/openstack/build/89502748516a4a209ad9a83a765d3df8 : SUCCESS in 1h 01m 24s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/db6a04869b854cfc809d8b8c9f56628d : SUCCESS in 51m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c0cfd082bd964eee8f871d4a01faa18e : SUCCESS in 46m 05s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/84032c36c2d5410b9c9fb2236180f28c : SUCCESS in 27m 52s\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/7e6996da579f42a8aa77ae78162d8bf0 : SUCCESS in 29m 06s\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/c3200f41e1b54a87903c902190daf9b9 : SUCCESS in 26m 45s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/08070194094944ea8cf5a097955847e1 : SUCCESS in 10m 30s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/81b5b3c0c680482eb379d6741f0be9a8 : SUCCESS in 7m 02s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/ad07b72888e2489cb297f3fbfd132270 : SUCCESS in 17m 01s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/8ba09c2c73d54417b5829a2d0c1311ea : SUCCESS in 23m 57s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/529eae3894d64b4796a8a9296b208a47 : SUCCESS in 20m 44s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/206e6f3155204e4db3bfbd2d2868bcff : SUCCESS in 24m 42s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/bee917176591433a9b0b23118482b3d2 : SUCCESS in 23m 08s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/52b3baa8ab864c52adb4d61fe16a9d79 : SUCCESS in 26m 20s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/623afebd52e649418a23ec8bdc23f0da : SUCCESS in 23m 17s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/ebd9e32e711a4f4bb2ca44fbfb6b9665 : SUCCESS in 23m 20s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/e7290a71bbe94d489d38d6fc07708600 : SUCCESS in 16m 45s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/74a0cd542c6f49368f34185b0843a854 : SUCCESS in 7m 20s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e7aaa0ff0d4b4c3d80499b60b6bc0d93 : SUCCESS in 21m 43s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/84f7afaf5942447198c0cd7cb654c12c : SUCCESS in 58m 06s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/bb0f054fb6ed4cb3816301c4403c1458 : SUCCESS in 57m 17s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/26100670e3a7463681cd814a728f77da : SUCCESS in 40m 55s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f88cab0d67a34fd682500886c384d860 : SUCCESS in 40m 55s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9a5361f198ee4fc2a97dea0ff9cb43ab : SUCCESS in 17m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c9c4734f38714b09900889a205a40a6b : SUCCESS in 8m 01s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b78fdb1bef2d4145ac22d6249f54e306 : SUCCESS in 21m 43s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b65a2cc95fef45569506ec96de700140 : SUCCESS in 48m 57s","accounts_in_message":[],"_revision_number":19},{"id":"d965f3796d0f51b58d39ba62bf56fd7122b96a12","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-14 18:22:15.000000000","message":"Patch Set 19:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b88e61dc1e14472aa34274e6566442c3 : SUCCESS in 41m 38s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/dd57fe258926422c9a37cb4fd1ee995c : TIMED_OUT in 1h 31m 56s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/9b9ad13ce5fe4317bfd35c7989d7fefb : TIMED_OUT in 3h 01m 37s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/685c65dc88f547bcb24d4025e445f999 : SUCCESS in 51m 39s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/d290bd782503435cb615b08a59cbecc0 : SUCCESS in 46m 02s","accounts_in_message":[],"_revision_number":19},{"id":"85ca30579dfba3cb9faf8e009a1d9e0f3a22f7da","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2022-01-14 19:39:04.000000000","message":"Patch Set 20: Commit message was updated.","accounts_in_message":[],"_revision_number":20},{"id":"1a0364ad0b0ae9c9b685da4331923ef954c42a68","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-14 21:08:35.000000000","message":"Patch Set 20: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb0ac06d5cc74ff1a4ccb41c0be2a267 : SUCCESS in 11m 08s\n- grenade https://zuul.opendev.org/t/openstack/build/aab4255d21f34a2d9b44b9db3c593508 : SUCCESS in 1h 00m 49s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/b76b01a2a8b14ee294acbd953001e92c : SUCCESS in 37m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/68b39e9ddf774d118ef9bc8402efea3b : SUCCESS in 43m 43s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/adc80a22fc7a4c77a6ce671f8f81e0c7 : SUCCESS in 18m 06s\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/86a7c3c8c7634e2f9280aec596de0b30 : SUCCESS in 24m 51s\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/81d8062701aa4a9c88054cbb6a7ea0b3 : SUCCESS in 22m 20s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/404bb8a815b84ea290983e14683747ec : SUCCESS in 10m 56s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/114db5fb65fe4cf99eeb1f52cc0067e8 : SUCCESS in 7m 31s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/54fa02f2d5aa44feb1fa72dac3ab6c3b : SUCCESS in 20m 19s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/71f8f057b3bc410691ef3bc5f367a052 : SUCCESS in 18m 18s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/315afde5d0d44493826a9616d2636580 : SUCCESS in 17m 23s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/3ec2be1e48fc45f3ae00461b8352be79 : SUCCESS in 18m 04s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/5f3ca42464794b939c5d99d3b713090c : SUCCESS in 24m 01s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/13894dc391964fd1ad02b202fe4a8483 : SUCCESS in 22m 25s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/1669692086cb4d0f9e880da77b7a9480 : SUCCESS in 16m 39s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/2c594d7fac8940f2bf1a893a85f077a1 : SUCCESS in 17m 42s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/cf0f48ed4fec4919ae2e4f855799d5fa : SUCCESS in 18m 21s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/4f4d7741628f4d7a8dbf4e8841d7e8a6 : SUCCESS in 7m 28s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/1fe56aa408e84637abf699127d2fd12e : SUCCESS in 23m 48s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/652294f707814bac8030cde553486418 : SUCCESS in 56m 46s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/b793e648698e4621a977ac2fd23558b3 : SUCCESS in 1h 04m 41s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/27a913b607cf436086a934b5b79c768b : SUCCESS in 43m 55s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/1db1ac789b5840afaee6ec02aa403928 : SUCCESS in 41m 21s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/33c0e620b9f7496889c32da36ccf7143 : SUCCESS in 17m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d6f239b9bb8a4005977c35797109b0a0 : SUCCESS in 8m 38s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/54deca51ec7e40eba3b5e3838b6fd551 : SUCCESS in 33m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1a8d0b382f0a4d1292344a96282f919e : SUCCESS in 42m 26s","accounts_in_message":[],"_revision_number":20},{"id":"e76b4224d7f1169b2a1cf6affd58ef44dca88b0d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-14 23:02:03.000000000","message":"Patch Set 20:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a9f32e4f8fd54494a9aac7ebc45f1a74 : SUCCESS in 45m 26s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/633684dcdb3c46d997db52408584d256 : SUCCESS in 1h 05m 07s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/0f969c66cd3e4b7ab4ca39ff3d5c54d1 : FAILURE in 2h 38m 35s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/e460f6646de3473cbcbfdf5d689e7e89 : SUCCESS in 53m 10s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/3fbec9a5c0ee4a3ba81e66cedbca70d5 : SUCCESS in 47m 55s","accounts_in_message":[],"_revision_number":20},{"id":"1d7bdbe1c46306253cc276a8cc728f9e1392dca7","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2022-01-17 20:04:00.000000000","message":"Topic change-796057 removed","accounts_in_message":[],"_revision_number":20},{"id":"34b219ce8c8c077538b239816f100165a55f45c1","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2022-01-17 20:04:14.000000000","message":"Topic set to fips-compatibiity","accounts_in_message":[],"_revision_number":20},{"id":"a9a5490ff97d49a561995268d0d28a68175b8506","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2022-01-19 21:13:51.000000000","message":"Patch Set 21: Published edit on patch set 20.","accounts_in_message":[],"_revision_number":21},{"id":"8793a9d41c160825385b986f3211a2a48509f2ae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-19 22:37:38.000000000","message":"Patch Set 21: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b79f946272f0492e8abb5e58fa871e56 : SUCCESS in 8m 56s\n- grenade https://zuul.opendev.org/t/openstack/build/a08b565e7b764fe1a00a0885586612b1 : SUCCESS in 56m 20s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e8720ba955dc4ee88fcf697ad55afcc1 : SUCCESS in 46m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ba0c42af2bd1461dbaaf830429c22f7d : SUCCESS in 51m 04s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/9bc9f29c21334663a5351543a5081d42 : SUCCESS in 19m 57s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/aa1df688c0dc4f4c98a19b2d63f70607 : SUCCESS in 20m 54s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/85d30d115cab4b95b4642216ddaa8d3e : SUCCESS in 19m 48s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/fb3ee086a41c4861a0bba9fa17e86c3d : SUCCESS in 8m 57s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/e0ed5218db704ac89967f682283b2e0c : SUCCESS in 6m 24s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/885077e1d6e249139e499d65e73b7364 : SUCCESS in 17m 06s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/5576e70a18c64ab0bc4b26e6188df17f : SUCCESS in 16m 47s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/65196b31c7b245169ddaba4eb8fa90d9 : SUCCESS in 16m 43s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/33eae828d8e54c18804384ae10f6280f : SUCCESS in 21m 17s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/54051657ebb047aeb6b2ff84012c4052 : SUCCESS in 18m 22s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/ad5a3cb1f44b4bd182f409e89c4cb502 : SUCCESS in 20m 44s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/b45b2d4cd211492ab8ddd5c757363bec : SUCCESS in 19m 35s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/8fce2ece643b4e9a98922ad9e2366c66 : SUCCESS in 15m 55s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/956edeb0f15440269c21a7c722ef88c1 : SUCCESS in 17m 09s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/16e71d22e4ad4f70ae74303e158013bb : SUCCESS in 7m 44s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/c6e850625adc4f1f9feab177652b1d68 : SUCCESS in 19m 45s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/aefdade537974d989fb5a62c2db56ac1 : SUCCESS in 50m 46s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/a0b28ad47f1d488e830e1c70bdc60fea : SUCCESS in 1h 04m 11s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/584e5d98a8c74c9897608cf043f6aa5e : SUCCESS in 28m 13s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/cd42a8953f5241739fd5f02a2cbdc39f : SUCCESS in 42m 16s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b85ddc97fe2c4ff883cc43de5c8f91c1 : SUCCESS in 18m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f0830795fcf24601acecfae3691c4597 : SUCCESS in 6m 27s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/4a4329bce85b4b1d983a2ad10fc3873a : SUCCESS in 24m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ae3cb827971248dca0ca193b134797b9 : SUCCESS in 54m 27s","accounts_in_message":[],"_revision_number":21},{"id":"a991d752369e603d32f50c7490ff5d75188c15c1","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-01-19 23:52:23.000000000","message":"Patch Set 21: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":21},{"id":"99bf203a0f5b3025031e7e984876bd330685969a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-20 00:47:19.000000000","message":"Patch Set 21:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f70b21173a0d4694a0604903fc7d9f1e : SUCCESS in 1h 13m 38s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/bcd084a0892e42aa968167932b4a74c4 : SUCCESS in 1h 15m 36s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/27b417d0add84efaab82aa9383f69691 : TIMED_OUT in 3h 01m 25s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/8acf118dab5441f29d022fd1286b2e09 : SUCCESS in 1h 16m 40s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/d344f46b5e804f86914150ae0772d1e0 : SUCCESS in 50m 59s","accounts_in_message":[],"_revision_number":21},{"id":"bf15b0db77e05c0451aa066387bff2125c7dd0b9","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2022-01-20 22:49:42.000000000","message":"Patch Set 21:\n\n(1 comment)","accounts_in_message":[],"_revision_number":21},{"id":"f5ce82ea12c6d5b706520e13bb39ff959a0564b7","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-01-25 20:08:18.000000000","message":"Patch Set 21:\n\n(1 comment)","accounts_in_message":[],"_revision_number":21},{"id":"2cde7ad9af989d5f263c32295d5b1fc57ecaec85","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-01-25 20:08:25.000000000","message":"Uploaded patch set 22.","accounts_in_message":[],"_revision_number":22},{"id":"1df6bd09c7bf841442d9893ea7a36062a36fc8ea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-25 22:16:56.000000000","message":"Patch Set 22: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3623abae165c4198813e33e9bca2562a : SUCCESS in 8m 39s\n- grenade https://zuul.opendev.org/t/openstack/build/1f066360fb194d26a1e9ecc6d2921d7a : SUCCESS in 45m 37s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e5ecf4654c274a62aeb4b5ae3c828829 : SUCCESS in 36m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c07a5ab72ab340d1a465f1f2b529938c : SUCCESS in 42m 14s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/32490e242b3e4026b765e96e55dc689d : SUCCESS in 16m 40s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/e072a2037df640b7b9e59748e282d05f : SUCCESS in 21m 24s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/6b0c00ac455e421c9f5e8683495af318 : SUCCESS in 17m 50s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e29d347029a542eca56340c14f6c3347 : SUCCESS in 8m 55s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/a38605a02e004c9292dbcc8a25769e11 : SUCCESS in 6m 35s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/3c4231997a5b40aa9b98194135d3e567 : SUCCESS in 15m 52s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/18d4b722958d44a79f56b11d25508cef : SUCCESS in 12m 40s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/be518e66ae5346cebead57db46088c38 : SUCCESS in 15m 44s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/24f11123588e428d9f4742984cfcf8fa : SUCCESS in 20m 32s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/047730e050d04b9895d9d84652c45200 : SUCCESS in 20m 05s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/57b33080eba94a608e2ed5a529b1ef6b : SUCCESS in 16m 51s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/3b02bc8041bd4255b91149a70a9660f3 : SUCCESS in 16m 35s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/ab5633bac0ff4f20b21d96ce9ea283a7 : SUCCESS in 15m 11s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/7dc7f480ee51425b968c2622640349fc : SUCCESS in 15m 57s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/8b050d1dba704fd89110026b16ad614b : SUCCESS in 6m 30s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/055aafdc22844de588fcb112504041b1 : SUCCESS in 20m 07s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/39e556cd57e34f79bc6d38bca405f3a2 : TIMED_OUT in 2h 01m 20s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/79d2905ee6ce40dc9a89fec0f5b43025 : SUCCESS in 51m 48s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e01990cc8c734d7fbb13d84c90b44639 : SUCCESS in 42m 12s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/05f553bd98164b1284553f019080e77d : SUCCESS in 28m 14s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8b6dc49cda8d4ca1918bd7efdb8ed7dc : SUCCESS in 15m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b354c7f6b3c04aed9cc2e9ce20ee045c : SUCCESS in 6m 35s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/23535266a0ac461cb8fff2819a71230e : SUCCESS in 26m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/214dc594f66c4aabbbbfaf2e81bc502c : SUCCESS in 54m 01s","accounts_in_message":[],"_revision_number":22},{"id":"e213e2250796cf86fc822e2ad644779246b8acce","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-25 23:17:36.000000000","message":"Patch Set 22:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/77d419c803914481b616435c72aa7766 : SUCCESS in 39m 15s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0e674f6051384362aa875f5db1798181 : SUCCESS in 43m 29s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/aff59acfe56b4000841dcbe2621c7350 : TIMED_OUT in 3h 01m 17s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/3c2d571366e7434b9e6bfe1254947f18 : SUCCESS in 43m 11s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/b3bdc21aa4d245e8a1eba0503eb1f419 : SUCCESS in 44m 15s","accounts_in_message":[],"_revision_number":22},{"id":"9c476d421842fe8bcc0a276cef6c30f9e417a592","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-01-26 19:36:25.000000000","message":"Patch Set 22:\n\n(1 comment)","accounts_in_message":[],"_revision_number":22},{"id":"a0afe97fcc6fd485c877d9fa9f317cda0d5aab4d","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-01-26 19:37:44.000000000","message":"Patch Set 22: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":22},{"id":"f22de2226111381b53fa2f31feaabe0ad5785bf2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-26 20:44:03.000000000","message":"Patch Set 22: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/40f3e4e2f4b044f495f12fa0cb4d741a : SUCCESS in 8m 52s\n- grenade https://zuul.opendev.org/t/openstack/build/8d89fbe07c7f45caa83867cec2f02a16 : SUCCESS in 1h 02m 31s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/2fbda51c78704862a4c9290d8e8dd1a4 : SUCCESS in 56m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c9cffb16dc3a44d98bc6a85f3384145a : SUCCESS in 51m 33s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/c53a6716a7674133a99685072e658657 : SUCCESS in 18m 44s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/2c6fdf69491447adac2bb2e20e347ff9 : SUCCESS in 21m 50s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/65b909b743344f0b84a0e1396fe13d73 : SUCCESS in 19m 44s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/6bce7141471a4a038fb9a8650af7f197 : SUCCESS in 9m 36s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/45d4d98443524d39aedbbafa314a83af : SUCCESS in 5m 43s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/9ad5c8ba894d4b679a70faa190aee3e4 : SUCCESS in 18m 14s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/33896e1e3a4a4dbaa306808af7b24f8a : SUCCESS in 16m 47s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/20585d0832284112876a6f13229d108f : SUCCESS in 16m 42s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/093fee0d5dd64220b6c079d4eea82865 : SUCCESS in 16m 03s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/51760480f2104532bfb538dfe0313e7c : SUCCESS in 19m 16s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/b242dafd870548dd85d6dbb146a9fb4e : SUCCESS in 18m 25s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/9ed24589394549deb3d629b6e5058877 : SUCCESS in 17m 30s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/a7e724006b024575944a871a6b1109c6 : SUCCESS in 18m 20s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/0fad0304f7b34ac08d369f060c3549a6 : SUCCESS in 18m 11s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/5f45866f9dc540e781eb928c7c9229c5 : SUCCESS in 7m 23s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/ecbfcb5078ae4c299ca9662ef9d3211f : SUCCESS in 17m 29s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/bdd50de3fba44710a7a41669c6704622 : SUCCESS in 48m 53s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/4ccbc8720852424ba9f230cb9b004635 : SUCCESS in 1h 03m 23s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4adef592996f4b81a22272d86e966ed3 : SUCCESS in 43m 15s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/dab8a342fe2b4aca91ca660a7ef4b258 : SUCCESS in 41m 43s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a061b9ce7d5a48939a20c7d0b61a7887 : SUCCESS in 15m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/85f185fa0b334d6c9d8f31cc44821b75 : SUCCESS in 7m 15s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/2f883984574e40fb846b02982adf4608 : SUCCESS in 29m 06s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f4a77fb7d8be45abadfc0342cd0d532e : SUCCESS in 47m 15s","accounts_in_message":[],"_revision_number":22},{"id":"4f4ab14357d83e5b60a0f9977d399379779a5974","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-26 20:54:52.000000000","message":"Patch Set 22: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":22},{"id":"a467a33eaf13d418a82e3334ea35d873c1d218da","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-26 22:40:36.000000000","message":"Patch Set 22:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d5f8dd45f0354fc0b12a73fdd122b09f : SUCCESS in 49m 01s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d2810bc193b84676afbd0c8eaff1737a : TIMED_OUT in 1h 32m 02s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/17fe8dcd25744713bc8fcb1eb2693536 : TIMED_OUT in 3h 01m 23s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/b051bf9beb40403d95ac37f9ad5afdf9 : SUCCESS in 56m 21s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/9883e32927454400acdcca89aa1b6a6a : SUCCESS in 59m 15s","accounts_in_message":[],"_revision_number":22},{"id":"fb255f1b9e03e6af6caf28b0cdf76c4779ed1d68","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-27 00:11:07.000000000","message":"Patch Set 22: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c896938679241d290f6c53f17d67f27 : SUCCESS in 8m 45s\n- grenade https://zuul.opendev.org/t/openstack/build/550b3134d4a648419859580adf41522b : SUCCESS in 1h 06m 42s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/4ec23d16320243fd9ddf524996baabbe : SUCCESS in 55m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3e6ef64174c94adc957e62d95aa55765 : SUCCESS in 50m 47s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/b0d40e5bbeb3471fa3c6c53530dcfbd4 : SUCCESS in 9m 21s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/bf1eaf9ddb2e44708504869bdf25fd98 : SUCCESS in 7m 34s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/669b9fcd4d3a487c8b5118343faf07aa : SUCCESS in 16m 39s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/8ac1b854ef25411b908541a76e33608c : SUCCESS in 17m 12s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/4847b1facb144abe93066b23e35bddaf : SUCCESS in 15m 47s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/2ab648e241e14fb7b6bfbfecfdcc3788 : SUCCESS in 17m 08s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/29723cf4b90c4e1ea2461c9d283289a4 : SUCCESS in 21m 04s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/2ff13d1e87404af0ad5cc90b7604ed58 : SUCCESS in 17m 27s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/9c66d86c6d3a4bc59a8b3bcf6aa0ca5b : SUCCESS in 14m 57s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/cbef274f69824c02bd080fa03216c3e7 : SUCCESS in 19m 40s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/58bb666ad8bd4dab88e7a0b35d0feef4 : SUCCESS in 20m 35s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/81915ad5bfe7413c8c9fd2cc57a4a594 : SUCCESS in 6m 34s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/b62164617fe24215bb12db0af81b5d1c : SUCCESS in 1h 03m 31s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/532a6f7bf30848399a195fefd2a0616e : SUCCESS in 57m 45s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/35e7ae7ea8184b4b901533505a9e2a0e : SUCCESS in 28m 14s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/0264606faffd4a0ebb0ee9ad3d1cbbe2 : SUCCESS in 47m 55s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/713d8ea0de0e427994df566bb7d87b44 : SUCCESS in 17m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3f43ef9782ab49bd9f9c8b4516b02b0e : SUCCESS in 5m 56s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/4f76d1e48512413e804f9e31cd3a6384 : SUCCESS in 30m 09s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1e22852afa7b4ae2abf95b34b6f5c072 : SUCCESS in 46m 48s","accounts_in_message":[],"_revision_number":22},{"id":"f6bdd48e1b769751a332e20533bda47a0056b1f5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-27 00:11:07.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":22},{"id":"cb7fb65d8e554906356adf715c302ebe233f0587","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-27 00:16:40.000000000","message":"Patch Set 22:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f275860c7fe49e3be2ca969f9861d33 : SUCCESS in 1m 52s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/0da20ab18574446aade71fed590b8298 : SUCCESS in 56s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/37e18144ffc741edbf27117d479c328f : SUCCESS in 57s (non-voting)","accounts_in_message":[],"_revision_number":22}],"current_revision_number":22,"current_revision":"03be71c44aee283fa940f9a43c31072cb6082301","revisions":{"ad88f0a702b7772b7376a1bb9b586ba62afa6957":{"kind":"REWORK","_number":1,"created":"2021-06-11 19:35:42.000000000","uploader":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"ref":"refs/changes/57/796057/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/57/796057/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/796057/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/796057/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/796057/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/57/796057/1"}}},"commit":{"parents":[{"commit":"1a5b42da9e43f4fef16ceba9eb19aec9344d7132","subject":"Merge \"Remove babel.cfg\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1a5b42da9e43f4fef16ceba9eb19aec9344d7132"}]}],"author":{"name":"Ade Lee","email":"alee@redhat.com","date":"2021-06-11 19:24:54.000000000","tz":-240},"committer":{"name":"Ade Lee","email":"alee@redhat.com","date":"2021-06-11 19:34:37.000000000","tz":-240},"subject":"WIP/DNM: Add FIPS CI jobs","message":"WIP/DNM: Add FIPS CI jobs\n\nAdded jobs to run the swift functional tests when FIPS is enabled.\nTo speed up testing, I have temporarily removed a bunch of check jobs.\n\nDepends-On: https://review.opendev.org/c/zuul/zuul-jobs/+/788778\nChange-Id: If0b4cfc1ac9a8c66085eb4dc95366d43806d5ae2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ad88f0a702b7772b7376a1bb9b586ba62afa6957"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ad88f0a702b7772b7376a1bb9b586ba62afa6957"}]},"branch":"refs/heads/master"},"4c39a7baf88e69c79fc8558bbc6723df89cc8962":{"kind":"REWORK","_number":2,"created":"2021-06-11 19:39:50.000000000","uploader":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"ref":"refs/changes/57/796057/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/57/796057/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/796057/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/796057/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/796057/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/57/796057/2"}}},"commit":{"parents":[{"commit":"1a5b42da9e43f4fef16ceba9eb19aec9344d7132","subject":"Merge \"Remove babel.cfg\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1a5b42da9e43f4fef16ceba9eb19aec9344d7132"}]}],"author":{"name":"Ade Lee","email":"alee@redhat.com","date":"2021-06-11 19:24:54.000000000","tz":-240},"committer":{"name":"Ade Lee","email":"alee@redhat.com","date":"2021-06-11 19:39:44.000000000","tz":-240},"subject":"WIP/DNM: Add FIPS CI jobs","message":"WIP/DNM: Add FIPS CI jobs\n\nAdded jobs to run the swift functional tests when FIPS is enabled.\nTo speed up testing, I have temporarily removed a bunch of check jobs.\n\nDepends-On: https://review.opendev.org/c/zuul/zuul-jobs/+/788778\nChange-Id: If0b4cfc1ac9a8c66085eb4dc95366d43806d5ae2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4c39a7baf88e69c79fc8558bbc6723df89cc8962"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4c39a7baf88e69c79fc8558bbc6723df89cc8962"}]},"branch":"refs/heads/master"},"ea6049292598a31167d1dad2c17bae82b45c0fb2":{"kind":"REWORK","_number":3,"created":"2021-06-16 19:45:32.000000000","uploader":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"ref":"refs/changes/57/796057/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/57/796057/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/796057/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/796057/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/796057/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/57/796057/3"}}},"commit":{"parents":[{"commit":"1a5b42da9e43f4fef16ceba9eb19aec9344d7132","subject":"Merge \"Remove babel.cfg\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1a5b42da9e43f4fef16ceba9eb19aec9344d7132"}]}],"author":{"name":"Ade Lee","email":"alee@redhat.com","date":"2021-06-11 19:24:54.000000000","tz":-240},"committer":{"name":"Ade Lee","email":"alee@redhat.com","date":"2021-06-16 19:45:09.000000000","tz":-240},"subject":"WIP/DNM: Add FIPS CI jobs","message":"WIP/DNM: Add FIPS CI jobs\n\nAdded jobs to run the swift functional tests when FIPS is enabled.\nTo speed up testing, I have temporarily removed a bunch of check jobs.\n\nDepends-On: https://review.opendev.org/c/zuul/zuul-jobs/+/788778\nChange-Id: If0b4cfc1ac9a8c66085eb4dc95366d43806d5ae2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ea6049292598a31167d1dad2c17bae82b45c0fb2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ea6049292598a31167d1dad2c17bae82b45c0fb2"}]},"branch":"refs/heads/master"},"3991524fabe880a4b802e310e9011323961debed":{"kind":"REWORK","_number":4,"created":"2021-06-18 19:55:48.000000000","uploader":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"ref":"refs/changes/57/796057/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/57/796057/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/796057/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/796057/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/796057/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/57/796057/4"}}},"commit":{"parents":[{"commit":"dfb59c7a15bd4e89879f4928bf34f505b4b2a17a","subject":"Merge \"Add async_pending_last time to object.recon\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/dfb59c7a15bd4e89879f4928bf34f505b4b2a17a"}]}],"author":{"name":"Ade Lee","email":"alee@redhat.com","date":"2021-06-11 19:24:54.000000000","tz":-240},"committer":{"name":"Ade Lee","email":"alee@redhat.com","date":"2021-06-18 19:55:17.000000000","tz":-240},"subject":"WIP/DNM: Add FIPS CI jobs","message":"WIP/DNM: Add FIPS CI jobs\n\nAdded jobs to run the swift functional tests when FIPS is enabled.\nTo speed up testing, I have temporarily removed a bunch of check jobs.\n\nDepends-On: https://review.opendev.org/c/zuul/zuul-jobs/+/788778\nChange-Id: If0b4cfc1ac9a8c66085eb4dc95366d43806d5ae2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3991524fabe880a4b802e310e9011323961debed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3991524fabe880a4b802e310e9011323961debed"}]},"branch":"refs/heads/master"},"5be68127b8e94f14d34c0812f9973f1f8504a788":{"kind":"REWORK","_number":5,"created":"2021-06-18 20:05:57.000000000","uploader":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"ref":"refs/changes/57/796057/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/57/796057/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/796057/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/796057/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/796057/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/57/796057/5"}}},"commit":{"parents":[{"commit":"dfb59c7a15bd4e89879f4928bf34f505b4b2a17a","subject":"Merge \"Add async_pending_last time to object.recon\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/dfb59c7a15bd4e89879f4928bf34f505b4b2a17a"}]}],"author":{"name":"Ade Lee","email":"alee@redhat.com","date":"2021-06-11 19:24:54.000000000","tz":-240},"committer":{"name":"Ade Lee","email":"alee@redhat.com","date":"2021-06-18 20:05:50.000000000","tz":-240},"subject":"WIP/DNM: Add FIPS CI jobs","message":"WIP/DNM: Add FIPS CI jobs\n\nAdded jobs to run the swift functional tests when FIPS is enabled.\nTo speed up testing, I have temporarily removed a bunch of check jobs.\n\nDepends-On: https://review.opendev.org/c/zuul/zuul-jobs/+/788778\nChange-Id: If0b4cfc1ac9a8c66085eb4dc95366d43806d5ae2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5be68127b8e94f14d34c0812f9973f1f8504a788"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5be68127b8e94f14d34c0812f9973f1f8504a788"}]},"branch":"refs/heads/master"},"6693ff41900a14051785688abcd8f9aa0b74cc21":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2021-07-12 17:52:00.000000000","uploader":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"ref":"refs/changes/57/796057/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/57/796057/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/796057/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/796057/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/796057/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/57/796057/6"}}},"commit":{"parents":[{"commit":"88d87afda0bee3e5f3f8d2fccdcdeae063598223","subject":"Merge \"Sharder: Add root_audit overlap stats\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/88d87afda0bee3e5f3f8d2fccdcdeae063598223"}]}],"author":{"name":"Ade Lee","email":"alee@redhat.com","date":"2021-06-11 19:24:54.000000000","tz":-240},"committer":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2021-07-12 17:52:00.000000000","tz":0},"subject":"WIP/DNM: Add FIPS CI jobs","message":"WIP/DNM: Add FIPS CI jobs\n\nAdded jobs to run the swift functional tests when FIPS is enabled.\nTo speed up testing, I have temporarily removed a bunch of check jobs.\n\nDepends-On: https://review.opendev.org/c/zuul/zuul-jobs/+/788778\nChange-Id: If0b4cfc1ac9a8c66085eb4dc95366d43806d5ae2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6693ff41900a14051785688abcd8f9aa0b74cc21"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6693ff41900a14051785688abcd8f9aa0b74cc21"}]},"branch":"refs/heads/master","description":"Rebase"},"a2af557452c3747a720ceedf5dc631af8b65904e":{"kind":"REWORK","_number":7,"created":"2021-08-17 21:53:46.000000000","uploader":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"ref":"refs/changes/57/796057/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/57/796057/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/796057/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/796057/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/796057/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/57/796057/7"}}},"commit":{"parents":[{"commit":"3cca6a11be695a2b8fa393fad638674203a157f3","subject":"Merge \"Add a project scope read-only role to keystoneauth\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3cca6a11be695a2b8fa393fad638674203a157f3"}]}],"author":{"name":"Ade Lee","email":"alee@redhat.com","date":"2021-06-11 19:24:54.000000000","tz":-240},"committer":{"name":"Ade Lee","email":"alee@redhat.com","date":"2021-08-17 21:53:29.000000000","tz":-240},"subject":"WIP/DNM: Add FIPS CI jobs","message":"WIP/DNM: Add FIPS CI jobs\n\nAdded jobs to run the swift functional tests when FIPS is enabled.\nTo speed up testing, I have temporarily removed a bunch of check jobs.\n\nDepends-On: https://review.opendev.org/c/zuul/zuul-jobs/+/788778\nChange-Id: If0b4cfc1ac9a8c66085eb4dc95366d43806d5ae2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a2af557452c3747a720ceedf5dc631af8b65904e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a2af557452c3747a720ceedf5dc631af8b65904e"}]},"branch":"refs/heads/master"},"c0300297c6efd089529132f88292fc3f78ddb7cf":{"kind":"REWORK","_number":8,"created":"2021-08-23 17:41:51.000000000","uploader":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"ref":"refs/changes/57/796057/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/57/796057/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/796057/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/796057/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/796057/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/57/796057/8"}}},"commit":{"parents":[{"commit":"f190ba2b7d02693ee2d7b59043b0baf9fddd203b","subject":"Merge \"use a context manager to apply --no-auto-shard option\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f190ba2b7d02693ee2d7b59043b0baf9fddd203b"}]}],"author":{"name":"Ade Lee","email":"alee@redhat.com","date":"2021-06-11 19:24:54.000000000","tz":-240},"committer":{"name":"Ade Lee","email":"alee@redhat.com","date":"2021-08-23 17:41:42.000000000","tz":-240},"subject":"WIP/DNM: Add FIPS CI jobs","message":"WIP/DNM: Add FIPS CI jobs\n\nAdded jobs to run the swift functional tests when FIPS is enabled.\nTo speed up testing, I have temporarily removed a bunch of check jobs.\n\nDepends-On: https://review.opendev.org/c/zuul/zuul-jobs/+/788778\nChange-Id: If0b4cfc1ac9a8c66085eb4dc95366d43806d5ae2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c0300297c6efd089529132f88292fc3f78ddb7cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c0300297c6efd089529132f88292fc3f78ddb7cf"}]},"branch":"refs/heads/master"},"ae8da263a72977929e137d076bd8c0cd8120b17c":{"kind":"REWORK","_number":9,"created":"2021-11-02 17:02:56.000000000","uploader":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"ref":"refs/changes/57/796057/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/57/796057/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/796057/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/796057/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/796057/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/57/796057/9"}}},"commit":{"parents":[{"commit":"f190ba2b7d02693ee2d7b59043b0baf9fddd203b","subject":"Merge \"use a context manager to apply --no-auto-shard option\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f190ba2b7d02693ee2d7b59043b0baf9fddd203b"}]}],"author":{"name":"Ade Lee","email":"alee@redhat.com","date":"2021-06-11 19:24:54.000000000","tz":-240},"committer":{"name":"Andre Aranha","email":"afariasa@redhat.com","date":"2021-11-02 17:01:49.000000000","tz":0},"subject":"WIP/DNM: Add FIPS CI jobs","message":"WIP/DNM: Add FIPS CI jobs\n\nAdded jobs to run the swift functional tests when FIPS is enabled.\nTo speed up testing, I have temporarily removed a bunch of check jobs.\n\nDepends-On: https://review.opendev.org/c/zuul/zuul-jobs/+/788778\nChange-Id: If0b4cfc1ac9a8c66085eb4dc95366d43806d5ae2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ae8da263a72977929e137d076bd8c0cd8120b17c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ae8da263a72977929e137d076bd8c0cd8120b17c"}]},"branch":"refs/heads/master"},"eac1cdba798c37560144a47dcc253db3de7b3270":{"kind":"NO_CODE_CHANGE","_number":10,"created":"2021-11-02 17:06:57.000000000","uploader":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"ref":"refs/changes/57/796057/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/57/796057/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/796057/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/796057/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/796057/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/57/796057/10"}}},"commit":{"parents":[{"commit":"f190ba2b7d02693ee2d7b59043b0baf9fddd203b","subject":"Merge \"use a context manager to apply --no-auto-shard option\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f190ba2b7d02693ee2d7b59043b0baf9fddd203b"}]}],"author":{"name":"Ade Lee","email":"alee@redhat.com","date":"2021-06-11 19:24:54.000000000","tz":-240},"committer":{"name":"Andre Aranha","email":"afariasa@redhat.com","date":"2021-11-02 17:06:57.000000000","tz":0},"subject":"WIP/DNM: Add FIPS CI jobs","message":"WIP/DNM: Add FIPS CI jobs\n\nAdded jobs to run the swift functional tests when FIPS is enabled.\nTo speed up testing, I have temporarily removed a bunch of check jobs.\n\nChange-Id: If0b4cfc1ac9a8c66085eb4dc95366d43806d5ae2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/eac1cdba798c37560144a47dcc253db3de7b3270"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/eac1cdba798c37560144a47dcc253db3de7b3270"}]},"branch":"refs/heads/master","description":"Edit commit message"},"a735f66dba52734ffe24ee55381c99329df12cda":{"kind":"REWORK","_number":11,"created":"2021-11-02 18:03:53.000000000","uploader":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"ref":"refs/changes/57/796057/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/57/796057/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/796057/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/796057/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/796057/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/57/796057/11"}}},"commit":{"parents":[{"commit":"f190ba2b7d02693ee2d7b59043b0baf9fddd203b","subject":"Merge \"use a context manager to apply --no-auto-shard option\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f190ba2b7d02693ee2d7b59043b0baf9fddd203b"}]}],"author":{"name":"Ade Lee","email":"alee@redhat.com","date":"2021-06-11 19:24:54.000000000","tz":-240},"committer":{"name":"Andre Aranha","email":"afariasa@redhat.com","date":"2021-11-02 18:02:46.000000000","tz":0},"subject":"WIP/DNM: Add FIPS CI jobs","message":"WIP/DNM: Add FIPS CI jobs\n\nAdded jobs to run the swift functional tests when FIPS is enabled.\nTo speed up testing, I have temporarily removed a bunch of check jobs.\n\nDepends-On: https://review.opendev.org/c/zuul/zuul-jobs/+/788778\nChange-Id: If0b4cfc1ac9a8c66085eb4dc95366d43806d5ae2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a735f66dba52734ffe24ee55381c99329df12cda"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a735f66dba52734ffe24ee55381c99329df12cda"}]},"branch":"refs/heads/master"},"a6be6f867d80d8ebddc6c223494206e1b8ca20df":{"kind":"NO_CODE_CHANGE","_number":12,"created":"2021-11-02 18:04:31.000000000","uploader":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"ref":"refs/changes/57/796057/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/57/796057/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/796057/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/796057/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/796057/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/57/796057/12"}}},"commit":{"parents":[{"commit":"f190ba2b7d02693ee2d7b59043b0baf9fddd203b","subject":"Merge \"use a context manager to apply --no-auto-shard option\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f190ba2b7d02693ee2d7b59043b0baf9fddd203b"}]}],"author":{"name":"Ade Lee","email":"alee@redhat.com","date":"2021-06-11 19:24:54.000000000","tz":-240},"committer":{"name":"Andre Aranha","email":"afariasa@redhat.com","date":"2021-11-02 18:04:31.000000000","tz":0},"subject":"WIP/DNM: Add FIPS CI jobs","message":"WIP/DNM: Add FIPS CI jobs\n\nAdded jobs to run the swift functional tests when FIPS is enabled.\nTo speed up testing, I have temporarily removed a bunch of check jobs.\n\nDepends-On: https://review.opendev.org/c/zuul/zuul-jobs/+/816385\nChange-Id: If0b4cfc1ac9a8c66085eb4dc95366d43806d5ae2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a6be6f867d80d8ebddc6c223494206e1b8ca20df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a6be6f867d80d8ebddc6c223494206e1b8ca20df"}]},"branch":"refs/heads/master","description":"Edit commit message"},"6255891f4db8a81c0226219d5f814b71f324b0db":{"kind":"NO_CODE_CHANGE","_number":13,"created":"2021-11-05 15:37:27.000000000","uploader":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"ref":"refs/changes/57/796057/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/57/796057/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/796057/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/796057/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/796057/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/57/796057/13"}}},"commit":{"parents":[{"commit":"f190ba2b7d02693ee2d7b59043b0baf9fddd203b","subject":"Merge \"use a context manager to apply --no-auto-shard option\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f190ba2b7d02693ee2d7b59043b0baf9fddd203b"}]}],"author":{"name":"Ade Lee","email":"alee@redhat.com","date":"2021-06-11 19:24:54.000000000","tz":-240},"committer":{"name":"Andre Aranha","email":"afariasa@redhat.com","date":"2021-11-05 15:37:27.000000000","tz":0},"subject":"WIP/DNM: Add FIPS CI jobs","message":"WIP/DNM: Add FIPS CI jobs\n\nAdded jobs to run the swift functional tests when FIPS is enabled.\nTo speed up testing, I have temporarily removed a bunch of check jobs.\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-zuul-jobs/+/816855\nChange-Id: If0b4cfc1ac9a8c66085eb4dc95366d43806d5ae2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6255891f4db8a81c0226219d5f814b71f324b0db"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6255891f4db8a81c0226219d5f814b71f324b0db"}]},"branch":"refs/heads/master","description":"Edit commit message"},"f89cdd74c4de2f50d7265bff07971186fd40363c":{"kind":"REWORK","_number":14,"created":"2022-01-13 09:55:11.000000000","uploader":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"ref":"refs/changes/57/796057/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/57/796057/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/796057/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/796057/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/796057/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/57/796057/14"}}},"commit":{"parents":[{"commit":"32da73f5c9eec1170718bcd7b1dc827e1affb124","subject":"Merge \"Finer grained ratelimit for update\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/32da73f5c9eec1170718bcd7b1dc827e1affb124"}]}],"author":{"name":"Ade Lee","email":"alee@redhat.com","date":"2021-06-11 19:24:54.000000000","tz":-240},"committer":{"name":"afariasa","email":"afariasa@redhat.com","date":"2022-01-13 09:55:03.000000000","tz":0},"subject":"WIP/DNM: Add FIPS CI jobs","message":"WIP/DNM: Add FIPS CI jobs\n\nAdded jobs to run the swift functional tests when FIPS is enabled.\nTo speed up testing, I have temporarily removed a bunch of check jobs.\n\nChange-Id: If0b4cfc1ac9a8c66085eb4dc95366d43806d5ae2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f89cdd74c4de2f50d7265bff07971186fd40363c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f89cdd74c4de2f50d7265bff07971186fd40363c"}]},"branch":"refs/heads/master"},"d24e81480dd204269f767ae6764ef75198b4810c":{"kind":"REWORK","_number":15,"created":"2022-01-13 14:55:44.000000000","uploader":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"ref":"refs/changes/57/796057/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/57/796057/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/796057/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/796057/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/796057/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/57/796057/15"}}},"commit":{"parents":[{"commit":"32da73f5c9eec1170718bcd7b1dc827e1affb124","subject":"Merge \"Finer grained ratelimit for update\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/32da73f5c9eec1170718bcd7b1dc827e1affb124"}]}],"author":{"name":"Ade Lee","email":"alee@redhat.com","date":"2021-06-11 19:24:54.000000000","tz":-240},"committer":{"name":"afariasa","email":"afariasa@redhat.com","date":"2022-01-13 14:55:34.000000000","tz":0},"subject":"WIP/DNM: Add FIPS CI jobs","message":"WIP/DNM: Add FIPS CI jobs\n\nAdded jobs to run the swift functional tests when FIPS is enabled.\nTo speed up testing, I have temporarily removed a bunch of check jobs.\n\nChange-Id: If0b4cfc1ac9a8c66085eb4dc95366d43806d5ae2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d24e81480dd204269f767ae6764ef75198b4810c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d24e81480dd204269f767ae6764ef75198b4810c"}]},"branch":"refs/heads/master"},"8609e741ec414e7593096290b37fa2c9b8c0e47c":{"kind":"REWORK","_number":16,"created":"2022-01-13 14:57:20.000000000","uploader":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"ref":"refs/changes/57/796057/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/57/796057/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/796057/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/796057/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/796057/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/57/796057/16"}}},"commit":{"parents":[{"commit":"32da73f5c9eec1170718bcd7b1dc827e1affb124","subject":"Merge \"Finer grained ratelimit for update\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/32da73f5c9eec1170718bcd7b1dc827e1affb124"}]}],"author":{"name":"Ade Lee","email":"alee@redhat.com","date":"2021-06-11 19:24:54.000000000","tz":-240},"committer":{"name":"afariasa","email":"afariasa@redhat.com","date":"2022-01-13 14:57:17.000000000","tz":0},"subject":"WIP/DNM: Add FIPS CI jobs","message":"WIP/DNM: Add FIPS CI jobs\n\nAdded jobs to run the swift functional tests when FIPS is enabled.\nTo speed up testing, I have temporarily removed a bunch of check jobs.\n\nChange-Id: If0b4cfc1ac9a8c66085eb4dc95366d43806d5ae2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8609e741ec414e7593096290b37fa2c9b8c0e47c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8609e741ec414e7593096290b37fa2c9b8c0e47c"}]},"branch":"refs/heads/master"},"82de0426642304b156f3f16c72c346cae94525f9":{"kind":"NO_CODE_CHANGE","_number":17,"created":"2022-01-13 15:01:40.000000000","uploader":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"ref":"refs/changes/57/796057/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/57/796057/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/796057/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/796057/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/796057/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/57/796057/17"}}},"commit":{"parents":[{"commit":"32da73f5c9eec1170718bcd7b1dc827e1affb124","subject":"Merge \"Finer grained ratelimit for update\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/32da73f5c9eec1170718bcd7b1dc827e1affb124"}]}],"author":{"name":"Ade Lee","email":"alee@redhat.com","date":"2021-06-11 19:24:54.000000000","tz":-240},"committer":{"name":"Andre Aranha","email":"afariasa@redhat.com","date":"2022-01-13 15:01:40.000000000","tz":0},"subject":"WIP/DNM: Add FIPS CI jobs","message":"WIP/DNM: Add FIPS CI jobs\n\nAdded jobs to run the swift functional tests when FIPS is enabled.\n\nChange-Id: If0b4cfc1ac9a8c66085eb4dc95366d43806d5ae2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/82de0426642304b156f3f16c72c346cae94525f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/82de0426642304b156f3f16c72c346cae94525f9"}]},"branch":"refs/heads/master","description":"Edit commit message"},"dcd20dd6dc88b89c9c70d1825edd15d7d3703251":{"kind":"REWORK","_number":18,"created":"2022-01-14 10:29:36.000000000","uploader":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"ref":"refs/changes/57/796057/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/57/796057/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/796057/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/796057/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/796057/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/57/796057/18"}}},"commit":{"parents":[{"commit":"32da73f5c9eec1170718bcd7b1dc827e1affb124","subject":"Merge \"Finer grained ratelimit for update\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/32da73f5c9eec1170718bcd7b1dc827e1affb124"}]}],"author":{"name":"Ade Lee","email":"alee@redhat.com","date":"2021-06-11 19:24:54.000000000","tz":-240},"committer":{"name":"afariasa","email":"afariasa@redhat.com","date":"2022-01-14 10:29:28.000000000","tz":0},"subject":"WIP/DNM: Add FIPS CI jobs","message":"WIP/DNM: Add FIPS CI jobs\n\nAdded jobs to run the swift functional tests when FIPS is enabled.\n\nChange-Id: If0b4cfc1ac9a8c66085eb4dc95366d43806d5ae2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/dcd20dd6dc88b89c9c70d1825edd15d7d3703251"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/dcd20dd6dc88b89c9c70d1825edd15d7d3703251"}]},"branch":"refs/heads/master"},"e6d4256721035c3856df4f225f799f2d3d20dd8c":{"kind":"REWORK","_number":19,"created":"2022-01-14 15:06:44.000000000","uploader":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"ref":"refs/changes/57/796057/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/57/796057/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/796057/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/796057/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/796057/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/57/796057/19"}}},"commit":{"parents":[{"commit":"32da73f5c9eec1170718bcd7b1dc827e1affb124","subject":"Merge \"Finer grained ratelimit for update\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/32da73f5c9eec1170718bcd7b1dc827e1affb124"}]}],"author":{"name":"Ade Lee","email":"alee@redhat.com","date":"2021-06-11 19:24:54.000000000","tz":-240},"committer":{"name":"afariasa","email":"afariasa@redhat.com","date":"2022-01-14 15:06:35.000000000","tz":0},"subject":"WIP/DNM: Add FIPS CI jobs","message":"WIP/DNM: Add FIPS CI jobs\n\nAdded jobs to run the swift functional tests when FIPS is enabled.\n\nChange-Id: If0b4cfc1ac9a8c66085eb4dc95366d43806d5ae2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e6d4256721035c3856df4f225f799f2d3d20dd8c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e6d4256721035c3856df4f225f799f2d3d20dd8c"}]},"branch":"refs/heads/master"},"754692e3b8bb55581506431e453b074deb688a88":{"kind":"NO_CODE_CHANGE","_number":20,"created":"2022-01-14 19:39:04.000000000","uploader":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"ref":"refs/changes/57/796057/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/57/796057/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/796057/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/796057/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/796057/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/57/796057/20"}}},"commit":{"parents":[{"commit":"32da73f5c9eec1170718bcd7b1dc827e1affb124","subject":"Merge \"Finer grained ratelimit for update\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/32da73f5c9eec1170718bcd7b1dc827e1affb124"}]}],"author":{"name":"Ade Lee","email":"alee@redhat.com","date":"2021-06-11 19:24:54.000000000","tz":-240},"committer":{"name":"Ade Lee","email":"alee@redhat.com","date":"2022-01-14 19:39:04.000000000","tz":0},"subject":"Add FIPS CI jobs","message":"Add FIPS CI jobs\n\nAdded jobs to run the swift functional tests when FIPS is enabled.\n\nChange-Id: If0b4cfc1ac9a8c66085eb4dc95366d43806d5ae2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/754692e3b8bb55581506431e453b074deb688a88"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/754692e3b8bb55581506431e453b074deb688a88"}]},"branch":"refs/heads/master","description":"Edit commit message"},"a35bc4b2c57d8af522cf4b11a041657f10b839a5":{"kind":"REWORK","_number":21,"created":"2022-01-19 21:13:51.000000000","uploader":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"ref":"refs/changes/57/796057/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/57/796057/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/796057/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/796057/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/796057/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/57/796057/21"}}},"commit":{"parents":[{"commit":"32da73f5c9eec1170718bcd7b1dc827e1affb124","subject":"Merge \"Finer grained ratelimit for update\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/32da73f5c9eec1170718bcd7b1dc827e1affb124"}]}],"author":{"name":"Ade Lee","email":"alee@redhat.com","date":"2021-06-11 19:24:54.000000000","tz":-240},"committer":{"name":"Ade Lee","email":"alee@redhat.com","date":"2022-01-19 21:13:38.000000000","tz":0},"subject":"Add FIPS CI jobs","message":"Add FIPS CI jobs\n\nAdded jobs to run the swift functional tests when FIPS is enabled.\nWe\u0027ll set these jobs to be non-voting initially, so that we can\nensure that they are stable.\n\nChange-Id: If0b4cfc1ac9a8c66085eb4dc95366d43806d5ae2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a35bc4b2c57d8af522cf4b11a041657f10b839a5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a35bc4b2c57d8af522cf4b11a041657f10b839a5"}]},"branch":"refs/heads/master"},"03be71c44aee283fa940f9a43c31072cb6082301":{"kind":"REWORK","_number":22,"created":"2022-01-25 20:08:25.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/57/796057/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/57/796057/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/796057/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/796057/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/796057/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/57/796057/22"}}},"commit":{"parents":[{"commit":"32da73f5c9eec1170718bcd7b1dc827e1affb124","subject":"Merge \"Finer grained ratelimit for update\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/32da73f5c9eec1170718bcd7b1dc827e1affb124"}]}],"author":{"name":"Ade Lee","email":"alee@redhat.com","date":"2021-06-11 19:24:54.000000000","tz":-240},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-01-25 20:07:09.000000000","tz":-480},"subject":"Add FIPS CI jobs","message":"Add FIPS CI jobs\n\nAdded jobs to run the swift functional tests when FIPS is enabled.\nWe\u0027ll set these jobs to be non-voting initially, so that we can\nensure that they are stable.\n\nChange-Id: If0b4cfc1ac9a8c66085eb4dc95366d43806d5ae2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/03be71c44aee283fa940f9a43c31072cb6082301"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/03be71c44aee283fa940f9a43c31072cb6082301"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
