)]}'
{"id":"openstack%2Fswift~797515","triplet_id":"openstack%2Fswift~master~I0d519ebaaade35249fb7b17bd5f419ffdaa616c0","project":"openstack/swift","branch":"master","topic":"p-min-nondurable-age","hashtags":[],"change_id":"I0d519ebaaade35249fb7b17bd5f419ffdaa616c0","subject":"reconstructor: Delay purging reverted non-durable datafiles","status":"MERGED","created":"2021-06-22 17:07:06.000000000","updated":"2021-07-02 19:11:29.000000000","submitted":"2021-07-02 19:10:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":6,"has_review_started":true,"submission_id":"797515-p-min-nondurable-age","meta_rev_id":"195287f35cc4b39006595192a859b0d46a601cc3","_number":797515,"virtual_id_number":797515,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-07-02 19:10:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-07-02 17:20:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-07-02 17:20:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-22 18:01:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-06-22 18:30:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-22 22:32:53.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"5a6eceda4c9d73cc172a130a8ecf9ef9eb4e7abf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-06-22 17:07:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b2956485d87a3bb8189f5513e4522a3c53b8a243","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-22 18:01:06.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1d9e65c437e04501ac753186a42edbae : SUCCESS in 38m 06s (non-voting)\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ed1cbb6f91dc4cc1938e60b63fa1aabf : SUCCESS in 40m 41s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8dec9669999fcb99e85d1dec58a5e4410e652ec4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-22 18:30:58.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/105bec7f6ddd41b68c4cd7e81c868281 : SUCCESS in 10m 53s\n- grenade https://zuul.opendev.org/t/openstack/build/89710c3c0b9649a880a96299a4f002fd : SUCCESS in 1h 01m 10s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/4540372eae6e45d8a59d42d2870f0b73 : SUCCESS in 47m 15s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/251cb9352e2345a386b8480c6fa95dfe : SUCCESS in 8m 45s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/3039f70f5ebb4e80b7ca125aefebbcb3 : SUCCESS in 10m 52s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/b05479a3355942c7a35ef90eb27a6902 : SUCCESS in 15m 58s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/52e31eaa155f4938ab3723f02aee0ac4 : SUCCESS in 18m 07s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/b96b3c9fa76b4bf29e448d3332cfead1 : SUCCESS in 17m 58s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/9e36a9afcd0149168cfffbacafa820f4 : SUCCESS in 17m 37s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/f549ba93278c42b8aed3c7f2ca2e23ea : SUCCESS in 32m 07s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/a846cd96da7845e1a537a60bdaadd3d5 : SUCCESS in 17m 59s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/c339d5bdbdac40318e504ab746e9bd2b : SUCCESS in 18m 49s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/6955d335a9f94e0e9271bd20622f0988 : SUCCESS in 18m 28s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/b73b0d1b2f4e4aa7a7ee27f07a5ad8bc : SUCCESS in 19m 56s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/ede93e652f38443f8db2c6d19fa022db : SUCCESS in 17m 58s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/460bfcfa183f4981ab67c3d930b87de5 : SUCCESS in 18m 31s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/75887c23da3340239243c90836684de5 : SUCCESS in 7m 09s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f524b7231ac545e5af11f6c5277e2611 : FAILURE in 24m 13s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/2b761f43d3a44eb8bf39f8921157f8f8 : FAILURE in 52m 01s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/e12e250e5fa348508499efabdf42b9f8 : FAILURE in 57m 27s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/24467dd2396d4d349c708928145ec930 : SUCCESS in 38m 23s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e15adef626584386b04e5e868c0f0f74 : SUCCESS in 33m 19s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4b96e22bcf0b4b7890400faa44b12664 : SUCCESS in 17m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af65a0293e1147c5a801f6745a1fca02 : SUCCESS in 7m 03s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/0bb2477cffec411990020eabc30004f0 : SUCCESS in 24m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/efb0b545651c474d8435a8810a70e152 : SUCCESS in 48m 09s","accounts_in_message":[],"_revision_number":1},{"id":"e43d324542080e6fa6568183ecf29dae5ff2cf9b","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-06-22 22:32:53.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)\n\nYeah, I think I like this more than my take.","accounts_in_message":[],"_revision_number":1},{"id":"b19db0685c9c0fa5a92b6f9ab0ebd920efad51b6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-06-23 18:10:51.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"3be214e03b3d1c86515513d27814ac894e76875d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-23 18:48:40.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/cf06c7a62d264b9d95cc2703dc94c456 : SUCCESS in 36m 03s (non-voting)\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/35ab053f1b5d401eb4e7b936b6b41fe8 : SUCCESS in 30m 33s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"40de99a96af779ee2a1dbdb19d2aad0105b95f53","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-23 19:26:27.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d511850c4f24ee8843bfa34ed2ef117 : SUCCESS in 10m 43s\n- grenade https://zuul.opendev.org/t/openstack/build/850d13676321455f9386f8484246fd57 : SUCCESS in 1h 03m 33s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/3158300a040b437dbb0a230eb97f4199 : SUCCESS in 42m 40s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/7b2f6bf5aff749de821f3b5b8a357795 : SUCCESS in 9m 59s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/4bc44e0f4e55409582a7deb7566df2f7 : SUCCESS in 10m 48s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/ff9a2fc37ac64b5cac5f79426fdc4866 : SUCCESS in 18m 30s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/cebb25cd13ec46c2b6fcde0ed0ee4cef : SUCCESS in 16m 22s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/9950dc40a2d748eeb2988927a4d1b41b : SUCCESS in 17m 42s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/0f17d20fa218427ba55e26f2f508b73d : SUCCESS in 17m 20s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/77c2423a17414ab1898ac6c30d7ec62f : SUCCESS in 25m 37s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/3a58fd4fd1c845d595e574059077c3cb : SUCCESS in 32m 02s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/822375b7123444488806a74cc41fb731 : SUCCESS in 17m 56s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/1c4a3d88da474795be4c3802aa465d45 : SUCCESS in 20m 48s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/54fbf6ef600048598b90867b17b49cc1 : SUCCESS in 20m 44s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/b0cddc8c5d6745b9a09612b6ed9569ac : SUCCESS in 19m 25s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/844502f87b1a472d83877442f4503bd8 : SUCCESS in 20m 20s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/aa5a4240bbf04ff9a4f15efde7098e6d : SUCCESS in 13m 52s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f48764cf3e33450f9e16f4ef8d648b7b : SUCCESS in 25m 34s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/7cba87ea22b04acc98c1581511607be2 : SUCCESS in 54m 51s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/248bad2f9221413b9cb6c85097a3d2e0 : SUCCESS in 49m 44s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/7eaa1e76fe714b74b23fd359585f12a3 : SUCCESS in 38m 06s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/50c7c07af92149caa2b88e56fc8eb74f : SUCCESS in 44m 40s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c450d4c1fc584f29908f87c9234434af : SUCCESS in 18m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/22c1398ce0b44b4cb4c36e162272f76f : SUCCESS in 7m 14s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f0908f2fccd84496a12aeecaed63a8ef : SUCCESS in 26m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/656b10890dbd4af5b554940e56d0782b : SUCCESS in 51m 06s","accounts_in_message":[],"_revision_number":2},{"id":"3667e76924342bc7425ef73d382444dc3a6a246c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-06-24 10:02:11.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"973b271e5a44a2d6d499de321f43b95e82a14ec8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-24 11:17:54.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65bc0b25a64047fda9d670b08811ab40 : SUCCESS in 9m 53s\n- grenade https://zuul.opendev.org/t/openstack/build/fb82f81b05f446bdbdda15a97bd38ccf : SUCCESS in 50m 03s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/9e070df6439a4885a9b727bb383e7fbf : SUCCESS in 42m 10s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ed5dbc9cf55247698617580bd5e2c95f : SUCCESS in 8m 42s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/1b47a2d537d643d7aaf449fa73c54894 : SUCCESS in 9m 25s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/f516b59fc53c4bb2a5dbda608b6d80e8 : SUCCESS in 15m 45s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/967f17af3a1940d3b4b83207690148f3 : SUCCESS in 16m 35s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/96d2dbc9ee704942a46507aadcf5a448 : SUCCESS in 16m 33s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/9313eee814344dada88e6e79ee10d654 : SUCCESS in 15m 53s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/c11b1aa427774a12b1c0f30aaab98584 : SUCCESS in 17m 52s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/88b2e7b5a79544709aa92efc1699b6f2 : SUCCESS in 17m 22s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/7330566db2494db8b9fecc3ccce56104 : SUCCESS in 31m 21s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/733c30be5e624fcb87c51214c5886a5f : SUCCESS in 19m 48s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/c88bf833208740f19cb64c117a0d5c91 : SUCCESS in 18m 41s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/0a813e17ab2943e4853f8941e823b377 : SUCCESS in 20m 19s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/baeac604799b4b6b84bf330232817afd : SUCCESS in 21m 30s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/5cb7b1750edf4f218b15e0be128af5a4 : SUCCESS in 8m 21s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/9190ee0a2a6e4a9995728b3fecccdfda : SUCCESS in 22m 26s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/4328453fb6674ff6930dfc8e0b571b55 : SUCCESS in 54m 15s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/c8b16c4bc3234d55b215974381818f38 : SUCCESS in 50m 19s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/10b064f265ca4a9d9e979b0c32fdf260 : SUCCESS in 41m 20s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/15215e44c4f346a68abee21ca0999486 : SUCCESS in 40m 15s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9f57be581c0547219ec7825bea2d7899 : SUCCESS in 21m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/86518a88d2cb4a19a95ad773c4a58ada : SUCCESS in 7m 00s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/bfe9a83419be487f86c2b390bb19a690 : SUCCESS in 25m 13s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/db67deec044c4c5e9abafdd7fabac2dd : SUCCESS in 50m 49s","accounts_in_message":[],"_revision_number":3},{"id":"dd0bebc189edcaec266794a55e82f009795c7be4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-24 11:18:12.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5877645647e848118a5cdde903a2bb47 : SUCCESS in 1h 08m 14s (non-voting)\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/df47b9fcd4c249deb5979dcedb5a3462 : SUCCESS in 1h 04m 56s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"8111efe3656dfe13b829b8a9235db46855f2cced","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-07-02 17:20:14.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(5 comments)","accounts_in_message":[],"_revision_number":3},{"id":"1de0817705524b5c3ddabef27c00144f3fdded05","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-02 17:20:37.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"a3f3a6d5233b3d682645f56099fa9da8c7461deb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-02 19:10:04.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5047356a664248dfa7ea49d5fc44ef45 : SUCCESS in 9m 31s\n- grenade https://zuul.opendev.org/t/openstack/build/088d7a7e5c8c47918c5e3a2cb6e7a41b : SUCCESS in 1h 13m 18s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/5d4e2c68a04548de851cdb32466ef949 : SUCCESS in 48m 40s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/037885ec8ce6479a99691f6c24b5d28a : SUCCESS in 9m 16s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/746ac1b89f5f45feb626748142033e9a : SUCCESS in 9m 43s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/f93208e8a8f4455c92508e3edfa5c6fa : SUCCESS in 19m 20s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/adcc03f243d74bfd875205ec10a46a27 : SUCCESS in 17m 56s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/388715269b8a4e45a1ef75db21bcafc0 : SUCCESS in 17m 38s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/18287c4e0939447c893c229fc4c5b739 : SUCCESS in 18m 24s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/1650e1a3a67241198fddc404cb85a479 : SUCCESS in 18m 15s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/65d5dcc78c08437b952bbea84173e707 : SUCCESS in 18m 04s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/95967af6fa994d1e86fe22240272b51c : SUCCESS in 18m 32s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/b3f0e07a0401405d8c94b6cb902327d6 : SUCCESS in 21m 20s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/249b0d4e9f774283a8eff36eed414b65 : SUCCESS in 18m 42s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/94b95188a0674593ba708a4159bdcb5b : SUCCESS in 18m 58s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/61f801411e2441c3b0ba860f307a7d91 : SUCCESS in 19m 54s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/bf9e5af585ac4d3e96e7b003f2f1d50d : SUCCESS in 7m 00s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/1743c8a0885546db83ac93079a491cc9 : SUCCESS in 50m 47s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/b829cc2044d047d0aa201b51c033641c : SUCCESS in 52m 52s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a1c234161a8540878fc366ed72f0488d : SUCCESS in 45m 58s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/976804eb6eef43f9b36ee69a149f08a6 : SUCCESS in 41m 29s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7af2fc83dc2941f2ba4095129429a186 : SUCCESS in 19m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a7710b8f77a14e228f271a38ac98ee28 : SUCCESS in 6m 21s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/877c9975079c47ae845e93fd9c21a191 : SUCCESS in 25m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5c3cecabe4ff49d3a9f945a2da7794f1 : SUCCESS in 44m 43s","accounts_in_message":[],"_revision_number":3},{"id":"fe1546be3b8387100a6a5e9e3ff7746a9243ca83","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-02 19:10:04.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"195287f35cc4b39006595192a859b0d46a601cc3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-02 19:11:29.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/02b55ca97d814548a36d2e9e300a0da4 : SUCCESS in 59s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/35c320acd45c4295a51222895e7b9b31 : SUCCESS in 51s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/e11cb1785436479088e13dd489c03b98 : FAILURE in 41s (non-voting)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"2934818d608e6cedd30ecb81900d02969476275c","revisions":{"85a01c8aa1fa93edc953263baa0b1fccaa99b3ec":{"kind":"REWORK","_number":1,"created":"2021-06-22 17:07:06.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/15/797515/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/15/797515/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/15/797515/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/15/797515/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/15/797515/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/15/797515/1"}}},"commit":{"parents":[{"commit":"977f1bcc61077833c8473f7dee7a53e5605def2f","subject":"Merge \"Add unit test for missing whole EC fragments\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/977f1bcc61077833c8473f7dee7a53e5605def2f"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-06-22 15:48:59.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-06-22 17:06:43.000000000","tz":60},"subject":"WIP: reconstructor: Delay purging reverted non-durable datafiles","message":"WIP: reconstructor: Delay purging reverted non-durable datafiles\n\nTODO:\n- unit test is_file_older\n- fix probe test\n\nThe reconstructor may revert a non-durable datafile concurrently with\nan object server PUT to the handoff that is about to make the datafile\ndurable.  This could previously lead to the reconstructor deleting the\nrecently written datafile before the object-server attempts to rename\nit to a durable datafile, and a traceback in the object server.\n\nThe reconstructor will now only remove reverted nondurable datafiles\nthat are older (according to mtime) than a period set by a new\nnondurable_purge_delay option (defaults to 60 seconds). More recent\nnondurable datafiles may be made durable or will remain on the handoff\nuntil a subsequent reconstructor cycle.\n\nChange-Id: I0d519ebaaade35249fb7b17bd5f419ffdaa616c0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/85a01c8aa1fa93edc953263baa0b1fccaa99b3ec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/85a01c8aa1fa93edc953263baa0b1fccaa99b3ec"}]},"branch":"refs/heads/master"},"9a9cba500ef8d0cfaef7c9d14ba5e5b1a1a5cdfc":{"kind":"REWORK","_number":2,"created":"2021-06-23 18:10:51.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/15/797515/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/15/797515/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/15/797515/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/15/797515/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/15/797515/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/15/797515/2"}}},"commit":{"parents":[{"commit":"2b21f85cb0793f4aa41be114f775d67bca6479fb","subject":"Capture logs when running custom daemons in probe tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2b21f85cb0793f4aa41be114f775d67bca6479fb"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-06-22 15:48:59.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-06-23 18:10:39.000000000","tz":60},"subject":"reconstructor: Delay purging reverted non-durable datafiles","message":"reconstructor: Delay purging reverted non-durable datafiles\n\nThe reconstructor may revert a non-durable datafile on a handoff\nconcurrently with an object server PUT that is about to make the\ndatafile durable.  This could previously lead to the reconstructor\ndeleting the recently written datafile before the object-server\nattempts to rename it to a durable datafile, and consequently a\ntraceback in the object server.\n\nThe reconstructor will now only remove reverted nondurable datafiles\nthat are older (according to mtime) than a period set by a new\nnondurable_purge_delay option (defaults to 60 seconds). More recent\nnondurable datafiles may be made durable or will remain on the handoff\nuntil a subsequent reconstructor cycle.\n\nChange-Id: I0d519ebaaade35249fb7b17bd5f419ffdaa616c0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9a9cba500ef8d0cfaef7c9d14ba5e5b1a1a5cdfc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9a9cba500ef8d0cfaef7c9d14ba5e5b1a1a5cdfc"}]},"branch":"refs/heads/master"},"2934818d608e6cedd30ecb81900d02969476275c":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-06-24 10:02:11.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/15/797515/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/15/797515/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/15/797515/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/15/797515/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/15/797515/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/15/797515/3"}}},"commit":{"parents":[{"commit":"40aace89f0070a06d1654edcfacadccb140c65b4","subject":"Capture logs when running custom daemons in probe tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/40aace89f0070a06d1654edcfacadccb140c65b4"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-06-22 15:48:59.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-06-24 08:33:06.000000000","tz":60},"subject":"reconstructor: Delay purging reverted non-durable datafiles","message":"reconstructor: Delay purging reverted non-durable datafiles\n\nThe reconstructor may revert a non-durable datafile on a handoff\nconcurrently with an object server PUT that is about to make the\ndatafile durable.  This could previously lead to the reconstructor\ndeleting the recently written datafile before the object-server\nattempts to rename it to a durable datafile, and consequently a\ntraceback in the object server.\n\nThe reconstructor will now only remove reverted nondurable datafiles\nthat are older (according to mtime) than a period set by a new\nnondurable_purge_delay option (defaults to 60 seconds). More recent\nnondurable datafiles may be made durable or will remain on the handoff\nuntil a subsequent reconstructor cycle.\n\nChange-Id: I0d519ebaaade35249fb7b17bd5f419ffdaa616c0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2934818d608e6cedd30ecb81900d02969476275c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2934818d608e6cedd30ecb81900d02969476275c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
