)]}'
{"id":"openstack%2Fswift~797735","triplet_id":"openstack%2Fswift~master~I25da3aa81018c5de7b63e5584ac6a9dbb73243db","project":"openstack/swift","branch":"master","topic":"p-log-capture","hashtags":[],"change_id":"I25da3aa81018c5de7b63e5584ac6a9dbb73243db","subject":"Capture logs when running custom daemons in probe tests","status":"MERGED","created":"2021-06-23 18:10:51.000000000","updated":"2021-06-25 02:41:52.000000000","submitted":"2021-06-25 02:40:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":12,"unresolved_comment_count":7,"has_review_started":true,"submission_id":"797735-p-log-capture","meta_rev_id":"ea33248a3625a09c7b3916aa40f3ea2f5380ae01","_number":797735,"virtual_id_number":797735,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-25 02:40:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":2,"date":"2021-06-24 15:06:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":1,"date":"2021-06-24 15:06:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-23 18:48:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-06-23 19:25:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-23 21:33:32.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"},{"updated":"2021-06-24 16:58:43.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"CC"}],"messages":[{"id":"98a4d14f92ca386781b0f11f73622433ae2fafd2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-06-23 18:10:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a7f8c08abbc38e941b3956315ca9ab85417a3d22","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-23 18:48:18.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/dfe9dceec6a641dbb9f4bfa778cff43b : SUCCESS in 31m 56s (non-voting)\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/70ec06e90fd64e35bc7a54b5c0d1a362 : SUCCESS in 30m 59s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"eace1fbeac17cded4e54da3666cda6df5b09eb2e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-23 19:25:18.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ecfd382990094a3d9bdd3ae691bb9712 : SUCCESS in 12m 16s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/9028548454ac4047aa8db3930f2355b3 : SUCCESS in 8m 57s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/255d9f3a797d4ee9b7db3a3a2b20e013 : SUCCESS in 10m 13s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/e27ed8e050414caebce708bb1bc089f0 : SUCCESS in 18m 37s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/7c844e53491546fa83fad441d275183b : SUCCESS in 17m 52s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/82fa9e003b774189b1d6e09ea219ba46 : SUCCESS in 16m 50s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/4f22d6d87d194c8a8fe08fdd43c44de0 : SUCCESS in 25m 17s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/d4714b852d084b11a1a39f80f40198d6 : SUCCESS in 24m 51s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/655b955d0413443fbaf16e902f69d85e : SUCCESS in 20m 36s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/0269c070bb594f68883bc310a9d2c882 : SUCCESS in 19m 00s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/c27b51a017af46e09b6e429a76c70d6a : SUCCESS in 20m 55s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/f1fa29448c844631b30a8997027975e7 : SUCCESS in 23m 45s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/95fce22adcbb4192acc628fcb2b0751c : SUCCESS in 19m 33s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/aa1bdc29fdb6417fa63ee1a357081a2b : SUCCESS in 20m 39s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/174b1c633faf49f1948423783aeb56b3 : SUCCESS in 7m 04s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/c12e6361025643cf88bf29c2a2fba045 : SUCCESS in 21m 35s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/5b1c956e2375444dbcfded625ed3c80f : SUCCESS in 52m 09s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/210d1e1fa61144978f8fad708e628c29 : SUCCESS in 51m 00s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/07a8fad89865451e814158f1d913324a : SUCCESS in 42m 09s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e7ae8814e3f44e1f9b7d66b6950806ad : SUCCESS in 42m 03s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/50fc7f6f7b4d490dac8aaa895f961e70 : TIMED_OUT in 41m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2628c867a73e4b658fbd625da30fab62 : SUCCESS in 7m 49s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b4b84fb9a0fe4ff6a563f5ff0048f4e2 : SUCCESS in 31m 55s","accounts_in_message":[],"_revision_number":1},{"id":"e86b80881346f28177d2be60e98025679ad62966","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-06-23 21:23:38.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"4616c5f22780f12c663ed1c6bfdeaca4d6243920","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2021-06-23 21:33:32.000000000","message":"Patch Set 1: Code-Review+1\n\n(6 comments)\n\ngoal seems entirely reasonable - probably some different ways to structure everything depending on how we expect it to get used.","accounts_in_message":[],"_revision_number":1},{"id":"c02a79b3c1b1b7a62fd278bdffc5e7e571b4b1d2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-23 22:13:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/207d478e15b14437888783099baec164 : SUCCESS in 9m 11s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/15cc0ba591ca4d32b606490ae35b4f68 : SUCCESS in 8m 38s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/b02e720b9adf4db8bc6a394999ad4a08 : SUCCESS in 8m 56s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/6ea438cd3724445fb589d7779f7018ed : SUCCESS in 14m 49s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c75fe2752e424df28e43c066aa827a0d : SUCCESS in 15m 11s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/b1d062f1277840b58409a96921e590a7 : SUCCESS in 16m 04s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/1fda27b173c74858acf0e39187bcc4e1 : SUCCESS in 15m 31s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/117687476a744960a33aed4731a8230c : SUCCESS in 16m 01s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/5c1def4c7c83439d9448827722a9f165 : SUCCESS in 16m 56s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/4e5a9be3f07b4677b769213ff50a85b7 : SUCCESS in 18m 01s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/271c284120504a1e9609ec9d920e658d : SUCCESS in 17m 04s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/f6689184c9d44155823df14914dbb3d2 : SUCCESS in 16m 26s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/ef42a20a302b49c2b069d0d0d100720a : SUCCESS in 17m 11s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/23e79d5b3a5941d6a5dc485ae44a4836 : SUCCESS in 19m 30s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/609addfcfc47400894ca00e6dcf79434 : SUCCESS in 6m 54s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e659e14267ad480a9edcedb8331e30ae : FAILURE in 21m 47s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/2bbd1d230a614f2f8f3c13d80e926c5a : SUCCESS in 45m 21s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/f8ae8e5459a34f808c2269c6ab8f1b35 : SUCCESS in 47m 28s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/afcc2a05c5a441649ab159ac610eee78 : SUCCESS in 34m 17s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/6522a95d111c4158948c1a73f8a10029 : SUCCESS in 37m 45s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ae8dbfd646364268800118f487e515fb : SUCCESS in 15m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/09b539a19d794356a95c5e29be9fd505 : SUCCESS in 6m 25s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/76171d71bdae472c81ec2be75cb32be5 : SUCCESS in 24m 02s","accounts_in_message":[],"_revision_number":1},{"id":"140dace44bdece3b5ba865eb0f96fa160d94d8c0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-23 22:20:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/77e5982f653f490293ef18bab4bcbdce : SUCCESS in 54m 57s (non-voting)\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/422a4095ad044f68b7b1a9f5161e3054 : SUCCESS in 51m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"46bba744f2133d8904fb41c1e0f68a23d5c00d49","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-06-24 10:00:59.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"38552c45d5b2ffbbc63f83fbfe6839f4c97cbd4c","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-06-24 10:09:04.000000000","message":"Patch Set 2:\n\n(5 comments)\n\nChanged to use a context manager - nice idea @Clay!\n\nFurther refactoring here:\nhttps://review.opendev.org/c/openstack/swift/+/797735","accounts_in_message":[],"_revision_number":2},{"id":"e8b516e891af703de0b0b1e57f0ad45ca99ae419","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-24 10:58:13.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/fc5bfc9e2a514a15b01865cbe1664b3a : SUCCESS in 55m 05s (non-voting)\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b66c25542c124f8d805995f79bfd9c59 : SUCCESS in 50m 35s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"85c441c2379dc784adffb146c3d0596c620fbb57","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-24 11:15:06.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea0d4a01e185447288f6eb0e8bcee13e : SUCCESS in 9m 35s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/0b610fab33ad4d249b67e61b2317d54f : SUCCESS in 8m 36s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/e6f91ead4ed041bb848376d0af054316 : SUCCESS in 8m 47s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/d928a398a8fe4bf5ba301f927d98bbc3 : SUCCESS in 19m 55s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/bb8bd5861ede48fc8a561160bfc643c5 : SUCCESS in 17m 11s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/2fb931300dba4c349caa4a7970abbdf7 : SUCCESS in 20m 03s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/57e75dd904f04bfcb007de1c2cd2cb45 : SUCCESS in 16m 22s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/e76920ebef9d496c8dce0868300c6dcc : SUCCESS in 32m 50s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/6bd8418a68c94394b960912316aa9811 : SUCCESS in 17m 59s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/79710a5a47c04a93a03b3d0464fb4524 : SUCCESS in 18m 25s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/9806e224e95a4859ab02df7cb7918020 : SUCCESS in 23m 30s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/65e8b3e3a3db4deaa287f3fb5e3d2333 : SUCCESS in 17m 32s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/3e3f536a22ce4697a32e53ce9b0a0df4 : SUCCESS in 18m 20s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/99ac046671b54e9f819717fb9c66e04a : SUCCESS in 19m 27s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/9a05435c35fd4547b59003d75e014282 : SUCCESS in 7m 16s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b34d086c8aba4868b2e8c7f591b25df8 : FAILURE in 25m 25s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/f8f408690701462a847d80a93b9d14a9 : SUCCESS in 54m 14s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/62523a75b6c54fcf934ee977490a479c : SUCCESS in 52m 22s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/0e5054219d1246c9bfd20682701784ef : SUCCESS in 41m 12s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/8064abec4c224465818d88266b2e0c82 : SUCCESS in 40m 05s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/385e23521e134f679d29ac70242cb9ff : SUCCESS in 16m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d681e1b8f21242ecb359438980381cd8 : SUCCESS in 6m 40s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/2189cad4863243f1af4b4b94fd80295c : SUCCESS in 27m 42s","accounts_in_message":[],"_revision_number":2},{"id":"48a9a7e86fcd0f2dd42bfc88507be1449360e0b6","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2021-06-24 15:06:06.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)\n\nI\u0027m not 100% sure about https://review.opendev.org/c/openstack/swift/+/797864\n\n... but this looks great.","accounts_in_message":[],"_revision_number":2},{"id":"bb292e53e6790a14b326083f24e2fbc08997c95c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-24 15:07:39.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"5062903952b6e254c5557560196b728388a68e12","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-24 16:44:59.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc5b7c5604994ad7a0db67178900ca77 : SUCCESS in 8m 58s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/21743948f2794933985175c07bc01dec : RETRY_LIMIT in 6m 03s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/e9c1fa7afde64a5d9609a54bb44ced84 : RETRY_LIMIT in 6m 46s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/93b58ae73a344435a1c011bce52f1d10 : SUCCESS in 16m 50s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/207e9d22c93542cabd07e8c68d180e18 : SUCCESS in 14m 56s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/b0ee080d4b45445abacb766d13963a42 : SUCCESS in 21m 32s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/c05ab3adb02644a4aec3108efb7e0d76 : SUCCESS in 14m 23s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/f36284d3fd344ab6833bbf32d7528869 : SUCCESS in 21m 34s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/e1ab201a8a244376be2e49a5cb91c375 : SUCCESS in 24m 56s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/b4e5660ac7d14937b6b091749e395780 : SUCCESS in 21m 51s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/0dda67d36db84aaabbb7a55c2a83c35f : SUCCESS in 15m 31s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/d8ddba482e534eb1ab6baf6d113fa5da : SUCCESS in 16m 00s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/0bf6fd9423294380833168202a3c7d31 : SUCCESS in 22m 03s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/f2c15492ec964f199df0f1bbe0f6307d : SUCCESS in 15m 53s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/185ef079f5d74934b5e8ce43727e5bc2 : SUCCESS in 6m 45s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/ac6d89eaa77c482486b56d65a5a92cdf : SUCCESS in 47m 49s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/56d1a10225064c1abd2183883d3f1286 : SUCCESS in 49m 13s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/617577acf7c841bda77e7a2cdde0aeed : SUCCESS in 40m 45s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b40bed116cc54a29a2419b07cbef9541 : SUCCESS in 34m 00s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f5734212998c4c068f8274655451b60d : SUCCESS in 15m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d3284346e1c740789177c7eb8bfb4df8 : SUCCESS in 5m 34s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a37b46f112104f72b813a394f51f739d : FAILURE in 28m 59s","accounts_in_message":[],"_revision_number":2},{"id":"72c02b20fdbe24d28607a41fb23bdcb7850688a0","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-06-24 16:58:43.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"34e7e826f90df8cf89e4312dc0bcfc7b1ef597ef","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-24 17:58:11.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f1296350d78a42d48d1b0de10a0abc85 : SUCCESS in 57m 23s (non-voting)\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/00f71599f46a43ba825ca807dbdc4710 : SUCCESS in 38m 08s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e32c7d051bf2b37dca5183e6bde80a0d4b17088d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-24 18:35:47.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/87c2f7fe3f274d31a7046c675aa62258 : SUCCESS in 11m 59s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/0da11465452f4fc2b4033bcc1b795225 : RETRY_LIMIT in 5m 46s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/bd2ac80e321d4b19a9469dff1c527dce : RETRY_LIMIT in 6m 45s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/0ed91f3a3a224d45b2c558b213f3226f : SUCCESS in 17m 56s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/13df413b408b44afa20e62340acd6b8d : SUCCESS in 18m 14s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/9f113ca08a7f488dbb5f84c84366caae : SUCCESS in 20m 08s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/a8808f42ecc944ef999d2a45a99e6456 : SUCCESS in 17m 46s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/ebc2d82dd9c6410a955f390f5cb09a61 : SUCCESS in 19m 17s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/d4976f1c4e2746ed89f0c9020e267612 : SUCCESS in 20m 03s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/8da0a3903b0544d68218818bd1c94bf8 : SUCCESS in 20m 51s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/2525042473b2465d92313f8cff0a82ea : SUCCESS in 19m 47s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/4a1f4dea66734ba484d73267bdc6f83f : SUCCESS in 19m 57s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/1dc194988aef46649792c2df4c41c91b : SUCCESS in 18m 41s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/62346cec0b07492c937598ad90aafe7f : SUCCESS in 20m 03s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/dbf862a5529e45de86bf78994afd5eb7 : SUCCESS in 10m 44s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/3774dddb9f354224bfe931bc493fe49d : SUCCESS in 23m 02s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/b58b089e0dcc499a8089bed411c7b5e6 : SUCCESS in 54m 27s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/81a2b91ac64f4afb8203c344c37289ea : SUCCESS in 53m 37s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/5de42f3f290a4c2ab7ebe90d46347106 : SUCCESS in 42m 30s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/8c83e44b3d87467f971bc1a78dbfc6d5 : SUCCESS in 47m 44s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1929ff741adb4b5dbc7efe98fbe1eddd : SUCCESS in 17m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/63354b2b8f0942188b10b0b644efd947 : SUCCESS in 7m 50s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b9e5766770004140ba45975d84f957fc : SUCCESS in 27m 42s","accounts_in_message":[],"_revision_number":2},{"id":"b041fb2ebdf85ab9f7d847002c5d1cbed119c616","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-24 18:36:22.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"6cd66c246e3d652ccb32a4095758d533d97fba37","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-25 02:40:17.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d36ca0005ea45399f8d7d5b9e74b2ba : SUCCESS in 13m 01s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/6e1b539746c34d2abee48a1d1c862d1c : SUCCESS in 11m 34s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/c3ccebfb7e8d4c4182e75cc4eda10cc4 : SUCCESS in 11m 55s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/42b62c3d7883433f94ad2e0965e2d81a : SUCCESS in 19m 18s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/06554c6f055e45fca819184cfd5cd283 : SUCCESS in 19m 27s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/939f7eb535ce468a95c93ef02d57a99f : SUCCESS in 20m 07s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/83864eeec97f46b28c3745f1939355bd : SUCCESS in 22m 48s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/83aa6a1f3b4d4c24b90b7e86c614850d : SUCCESS in 22m 38s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/5b0a308ef38142bdb295705be7c3e901 : SUCCESS in 19m 16s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/560ee25884c042139872dec4d38ddbb0 : SUCCESS in 19m 17s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/6dd014f555e04895a374f30648707633 : SUCCESS in 20m 20s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/c5e74bd65ab4438594a9a2717576330a : SUCCESS in 19m 56s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/b6aa962d22ee4dfbb37ed5f5c1789223 : SUCCESS in 19m 25s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/1ab8ca5d54dd44688eec89c48d5a6880 : SUCCESS in 21m 04s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/0b24286281df4333ab965e18c94d64c2 : SUCCESS in 9m 53s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/02ecde45a09e488c8d8c79949e0555f5 : SUCCESS in 54m 18s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/c80ffc6346de4018b70707a74ae19023 : SUCCESS in 52m 10s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/7fa27df123fc4e83b52b5cd626e10ca8 : SUCCESS in 44m 37s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/21b8742dbb784c6292bf16e45bb03fec : SUCCESS in 42m 26s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1ee54d14093e48c6ba106f2f365f7784 : SUCCESS in 19m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bfe20c9f7ce745aa8cd55eaa96bab0c0 : SUCCESS in 9m 16s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/018927761faa4d9cb9f5df927327be23 : SUCCESS in 26m 21s","accounts_in_message":[],"_revision_number":2},{"id":"5133a6e67e54d0788c3c565f03a0ab808b2f97cc","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-25 02:40:20.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"ea33248a3625a09c7b3916aa40f3ea2f5380ae01","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-25 02:41:52.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c4d91edd75464aeb9e9c03658c2ac31a : SUCCESS in 1m 00s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/ebd67c399df14eacafb1d64cdb38e596 : SUCCESS in 45s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/bcd873bcb86a4afe82c01daf64beecbe : SUCCESS in 47s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"40aace89f0070a06d1654edcfacadccb140c65b4","revisions":{"2b21f85cb0793f4aa41be114f775d67bca6479fb":{"kind":"REWORK","_number":1,"created":"2021-06-23 18:10:51.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/35/797735/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/35/797735/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/797735/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/797735/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/797735/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/35/797735/1"}}},"commit":{"parents":[{"commit":"977f1bcc61077833c8473f7dee7a53e5605def2f","subject":"Merge \"Add unit test for missing whole EC fragments\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/977f1bcc61077833c8473f7dee7a53e5605def2f"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-06-23 17:21:35.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-06-23 18:01:02.000000000","tz":60},"subject":"Capture logs when running custom daemons in probe tests","message":"Capture logs when running custom daemons in probe tests\n\nPreviously a DebugLogger was used when probe tests ran \u0027custom\ndaemons\u0027, which provided a means to inspect captured logs but logged\nto console. This patch replaces that DebugLogger with a \u0027normal\u0027 swift\nlogger that is adapted to also capture logs and support log\ninspection.\n\nChange-Id: I25da3aa81018c5de7b63e5584ac6a9dbb73243db\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2b21f85cb0793f4aa41be114f775d67bca6479fb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2b21f85cb0793f4aa41be114f775d67bca6479fb"}]},"branch":"refs/heads/master"},"40aace89f0070a06d1654edcfacadccb140c65b4":{"kind":"REWORK","_number":2,"created":"2021-06-24 10:00:59.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/35/797735/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/35/797735/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/797735/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/797735/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/797735/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/35/797735/2"}}},"commit":{"parents":[{"commit":"977f1bcc61077833c8473f7dee7a53e5605def2f","subject":"Merge \"Add unit test for missing whole EC fragments\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/977f1bcc61077833c8473f7dee7a53e5605def2f"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-06-23 17:21:35.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-06-24 08:32:38.000000000","tz":60},"subject":"Capture logs when running custom daemons in probe tests","message":"Capture logs when running custom daemons in probe tests\n\nPreviously a DebugLogger was used when probe tests ran \u0027custom\ndaemons\u0027, which provided a means to inspect captured logs but logged\nto console. This patch replaces that DebugLogger with a \u0027normal\u0027 swift\nlogger that is adapted to also capture logs and support log\ninspection.\n\nChange-Id: I25da3aa81018c5de7b63e5584ac6a9dbb73243db\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/40aace89f0070a06d1654edcfacadccb140c65b4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/40aace89f0070a06d1654edcfacadccb140c65b4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
