)]}'
{"id":"openstack%2Fswift~797961","triplet_id":"openstack%2Fswift~master~I2d4dab1d56c31f904161e5338f1bb89a03fd88f7","project":"openstack/swift","branch":"master","topic":"p-sharder-validation","hashtags":[],"change_id":"I2d4dab1d56c31f904161e5338f1bb89a03fd88f7","subject":"sharder: add more validation checks on config","status":"MERGED","created":"2021-06-24 15:31:04.000000000","updated":"2021-07-15 01:31:09.000000000","submitted":"2021-07-15 01:29:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":3,"has_review_started":true,"submission_id":"797961-p-sharder-validation","meta_rev_id":"c6b9426a442e7d33a87e7808c8ff1772260e4cd7","_number":797961,"virtual_id_number":797961,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-07-15 01:29:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-07-14 22:13:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-07-14 22:13:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-24 16:33:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-06-24 16:46:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-28 19:57:08.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"7319e1b639eebb6b11c9ee547c1f0bb34387bae3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-06-24 15:31:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"53ae1ea039d51575cf35533acd6c834fa0672dc7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-24 16:33:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2d6bb083c5cb4e77bd062a007cef26ed : FAILURE in 58m 53s (non-voting)\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/662dd9d7a60947908dd93191a9d68590 : FAILURE in 59m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"cbff86671f575284a919baef3dfc679c0f93f959","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-24 16:46:46.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/64b7ce6d73c048c390974e2040bab6f7 : SUCCESS in 10m 32s\n- grenade https://zuul.opendev.org/t/openstack/build/a6027aa3c0384268b2207a037e6e3431 : SUCCESS in 1h 09m 48s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/946652f6a00046e389fc2e6df9082bd7 : SUCCESS in 47m 30s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/de90945e7190491390cc5f2d35b4935d : RETRY_LIMIT in 5m 30s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/99184a3bac044b82825d230f14f27185 : RETRY_LIMIT in 6m 43s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/338e7b2b42274a18be50d826d5780a38 : FAILURE in 18m 54s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/b460a9dc8dae4f9c89db1c329de845a4 : FAILURE in 18m 55s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/945639f8c4af4f42ac9cd46e7bdfcd69 : FAILURE in 18m 29s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/43fdf920473040a8a1582ec8c3f66dfe : FAILURE in 18m 58s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/353d86cbc95b488484b32fa554ec6eeb : FAILURE in 19m 06s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/80be05d22fae47058335c557042e47c9 : SUCCESS in 18m 33s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/14f428badfb04533a25be77e2cce958f : SUCCESS in 20m 01s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/9b611f3151e846c5a7c8b68484c0f1e6 : SUCCESS in 21m 32s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/447880b6f627482c9c8344e4e8ecc834 : SUCCESS in 19m 01s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/6342714cf54b4b368c572554464663f7 : SUCCESS in 19m 00s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/53093a18076d48348d05d6c12c44d60a : SUCCESS in 19m 24s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/01101b21e0724e8b9d1e534aff6e72ef : SUCCESS in 7m 45s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/d0eaa18444be41ec8d0cd597f62ec76c : SUCCESS in 25m 46s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/ec4d1e7e268c4afb8cf6524f8986f21d : SUCCESS in 54m 08s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/dcefdde5b86e4f1fadc5909ce14368b0 : SUCCESS in 56m 03s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/3c942df8a8d74a26848db2ff9714cd40 : SUCCESS in 43m 53s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4ee0bcbb5bbe47d7bf9f8589b28739c6 : SUCCESS in 44m 15s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c308278f8f554efea8b081afcdeb8083 : FAILURE in 21m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b6f775d91e014fc59d093e96c95e3186 : SUCCESS in 7m 53s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/015f7ada062b4ef6a5f110db0dab0ad0 : SUCCESS in 25m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/107baaa150a74933bbaba1a2951a310d : SUCCESS in 52m 21s","accounts_in_message":[],"_revision_number":1},{"id":"4d395406ab32d17f0e7fcbea3e5cd4cd66cdcc54","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-06-28 11:10:23.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"4168d21f4623dd96cde5efd6a1a4c6ef27cc6657","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-28 12:11:50.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2da70793c38543c0b55540fca260f694 : SUCCESS in 11m 05s\n- grenade https://zuul.opendev.org/t/openstack/build/720c001581ae4fe9aad716624c139a47 : SUCCESS in 59m 37s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/bab690dde7d640d0b4b3d1bb0ccce4c9 : SUCCESS in 45m 24s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/63cb71e321124786b4595a5099f3a389 : SUCCESS in 11m 09s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/c6df801a66cf44be849993fc87efb4e2 : SUCCESS in 10m 55s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/baac933a714d4d10a0505907b7e985c1 : SUCCESS in 17m 58s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/f67e7cc07a05485c9ef95fdb1f69ca29 : SUCCESS in 18m 00s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/f3136493a7ec432a80d4ce7ca18ff6ba : SUCCESS in 18m 14s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/39370c0e7da34eedacf6ce6d411ee232 : SUCCESS in 17m 57s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/ca79e03bacf44af8a3a1467b289946cf : SUCCESS in 16m 45s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/70a515fb4e0642338242b126a8440876 : SUCCESS in 18m 41s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/20930606d41541f6a45439615b29b232 : SUCCESS in 18m 44s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/27551f20cbf044c2a6c1e420f9c9fd74 : SUCCESS in 19m 52s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/c0ca791b4ce8402895a1055554fc9af1 : SUCCESS in 19m 14s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/5c936ba797c14ce7931b6e7173ab64a8 : SUCCESS in 18m 47s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/e7e123cd1d074f0dba2c8965179f5485 : SUCCESS in 19m 28s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/0fca5c73daed403781e5aa8553109f81 : SUCCESS in 9m 29s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/84c58c2915b94bc589d5c1233ee8dd6f : SUCCESS in 24m 05s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/c6c386edb8294d089bab0ac46f52495e : SUCCESS in 51m 20s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/fb33d2c074b74f61a8ea1cdc841dcba2 : SUCCESS in 50m 06s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/1277829ccef845298dd954fc929c11da : SUCCESS in 40m 55s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ce698adf97cd43b9b084b3be4372d8b5 : SUCCESS in 40m 02s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2e3252f1a234420bbb8bbb59e8a1fa84 : SUCCESS in 17m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1425d820d9cc40f2b38d7ddf05c0f752 : SUCCESS in 8m 11s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/003f265619b84e6a8896ed0a36dcaba3 : FAILURE in 26m 31s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0fba2b0ce112466788b7ea6073daeeea : SUCCESS in 48m 19s","accounts_in_message":[],"_revision_number":2},{"id":"f6dbbd912f231f0849932ee81a547f8579ee8eea","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-28 12:35:47.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3620a17b265747dba8969d98a39ce0e3 : SUCCESS in 1h 04m 16s (non-voting)\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0c9330f8d2914641806319c7fdba02a1 : SUCCESS in 1h 02m 35s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d8074f341236981e4f641c190c41f3524cf4ffb2","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-06-28 19:57:08.000000000","message":"Patch Set 2: Code-Review+1\n\n(2 comments)\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"d9853be572430d4e26465ae7979d01c44c697eda","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-07-08 11:54:54.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"d0f37b4de241209a8cec10c5232e5933c9161ad9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-08 12:33:57.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/443f493f05264610b1b21dba413a1ceb : SUCCESS in 37m 17s (non-voting)\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/506efcca41fa47a6a927196ee5628568 : SUCCESS in 36m 46s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"955570777e7974a08590946a624a732413d0d69c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-08 13:26:51.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d22d05c2c3d640fdb6158cc1ec4b707b : SUCCESS in 9m 48s\n- grenade https://zuul.opendev.org/t/openstack/build/51f6d7d11aeb43d2aaf91433b3e7f278 : SUCCESS in 1h 10m 47s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/2497ddb1947c492a8d6607330b9f9f93 : SUCCESS in 49m 08s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c1d7dc9ab28a4b0e8e79df19f5e2c354 : SUCCESS in 7m 48s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/2747af628577424ea9734186e2bcebc4 : SUCCESS in 10m 29s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/7e8e79a7df094dad8d2f2a454938b3e2 : SUCCESS in 20m 58s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/08bd04dfe1c0481cb8639584c0c2fc40 : SUCCESS in 15m 55s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/c18030907cac4099ad09df0cf4080f0d : SUCCESS in 16m 44s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/8ad38ac1c624493baba5686d99c969bf : SUCCESS in 16m 51s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/ddccb31173244fe89e9a7326d8c73291 : SUCCESS in 16m 18s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/f6e76134ffe44ed787bafde1aa52dfbd : SUCCESS in 17m 28s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/9bb9e93830aa4c3593193ca728ad88df : SUCCESS in 18m 16s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/222c4d0158e244a8ae55059de0cecf82 : SUCCESS in 19m 09s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/deb4b5fe73e64666883e16f470404a12 : SUCCESS in 17m 37s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/7177c83b08954dbbbc594b7bcc576e5a : SUCCESS in 21m 01s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/150ea72a8e4847d8825697272e96ba2a : SUCCESS in 16m 59s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/8ed3b6bc3c7f4b8480a4598ed10b9948 : SUCCESS in 7m 07s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/9578a9af4f47488a9d477632baed90e9 : SUCCESS in 22m 16s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/fcb3e88d01b045129828cbd4ee0a3954 : SUCCESS in 54m 56s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/0000b2b30b7e429e9e8c92cfa03e78cb : SUCCESS in 51m 03s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/9c160bdaec8343d9992e1ad71461ea45 : SUCCESS in 39m 22s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/566a0e7af20e4fc58515d29d935af3b0 : SUCCESS in 41m 29s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0802c56980f14794b6f7f9e2a6c842ec : SUCCESS in 17m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c67bf7042f4b408da939819e7154f86b : SUCCESS in 6m 52s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/76dd60304bc04a3780c06b78d7f22487 : SUCCESS in 31m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/65010afecd7b46e48af5b52b3b89d434 : SUCCESS in 41m 19s","accounts_in_message":[],"_revision_number":3},{"id":"c9b42108c3f9453a38b6b0967b0a23509916f764","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-07-14 22:13:38.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"4936fc09425468413e345f616e8cccc0e85a9bd2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-14 22:13:56.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"c23d845e18915a046bf6088adf68997e988c109b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-14 23:12:38.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a4f5a8cf7ffa4cf595b1583ea2463f30 : SUCCESS in 9m 15s\n- grenade https://zuul.opendev.org/t/openstack/build/42ce7535be2e4fc986b73da9f2e79282 : SUCCESS in 58m 06s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c0a17f5c81f64142ae5386f4b6479dc9 : SUCCESS in 45m 41s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/97703b6ac0c34d6eba712402f2f6ff7a : SUCCESS in 12m 07s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/2fbe8b5398394fbc8debcc710637baef : SUCCESS in 8m 23s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/a44068a2cf054ef18b361dde0e29d3b1 : SUCCESS in 16m 06s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/fedb0b877fec4ec9ab7836be455ce725 : SUCCESS in 15m 10s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/a048766ade884602982657090e5a56e5 : SUCCESS in 16m 17s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/f2843f826aac4c79ae6f80a273178004 : SUCCESS in 14m 12s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/5d5b9e75781244608ed5a50f32f7e4ca : SUCCESS in 16m 27s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/8f234052e6674227a8dafaeea92a0c36 : SUCCESS in 20m 43s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/8013b48e49c645c092395f04a0e06b17 : SUCCESS in 17m 45s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/f05fed79f8564bbb988025a1e1516666 : SUCCESS in 21m 24s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/cb26f0b81ee746668a615d5a9d76178a : SUCCESS in 17m 48s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/0e0cfcf83bd6478bb631f0f3db29c125 : SUCCESS in 22m 06s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/6fe9868d0251416ea6cccc5821b7126f : SUCCESS in 17m 43s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/827259dd78b648adbd16afb120c18dbd : SUCCESS in 6m 34s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/e2fa7034bdc14ac686bfb0922b10d43d : SUCCESS in 50m 25s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/387299fcbba24d20a784835690980ef5 : FAILURE in 50m 12s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/75a3f41c6b85476197301062efe84bbc : SUCCESS in 38m 26s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/1c6fce223f884feaa208579abdf38601 : SUCCESS in 34m 19s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ac0be09b8db34a09bc3151c3a0ab4191 : SUCCESS in 15m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4ae66c450fed4185ae4aac48074625fb : SUCCESS in 7m 56s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ff08bc351eee4494b0f56f1d91fc744b : SUCCESS in 28m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/09b4bb9a0cd04ba68841999e0022cc9b : FAILURE in 23m 38s","accounts_in_message":[],"_revision_number":3},{"id":"873727513fb95d4c21c612d4da2f4958abc9556c","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-07-14 23:20:07.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"bf1955e9bfd738e0bf93e9f052ea295f808a8f2c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-14 23:53:06.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/78b881d6fd3747ad9ec350372a6b615f : SUCCESS in 29m 10s (non-voting)\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ec41b7773e7c4e09bdf072151673afc5 : SUCCESS in 28m 40s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"60b9e4d62f211585153d82973a96b5b13b650be0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-15 00:14:00.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f4389bc6aa14c19a30ed794af2bd1db : SUCCESS in 9m 05s\n- grenade https://zuul.opendev.org/t/openstack/build/11f170ae355e47cebbd544bb33ef2691 : SUCCESS in 53m 29s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/b0fcca754ce94aa3a6881dd218f154f5 : SUCCESS in 44m 37s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/640fa312b7084a718d767f0714bce004 : SUCCESS in 7m 48s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/4fe8af3c382240d497b0c74ade0e1fc4 : SUCCESS in 8m 24s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/5b3b29225f3445109e03c81bdba8fb2f : SUCCESS in 22m 04s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/0e593b2b13d445be87989d746954d995 : SUCCESS in 17m 19s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/47c7358827c343889207668b15ff390e : SUCCESS in 21m 15s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/6a2d4914d19541f6968dfaefcee5b426 : SUCCESS in 20m 43s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/503255c7f5d64474839bd7cb7da61afa : SUCCESS in 20m 55s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/8aa45c4f62fb4a12a6134f0ba71e3e55 : SUCCESS in 18m 30s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/a2a18b4c997642d6ae2bab83acb7f85e : SUCCESS in 21m 35s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/abd32874471742b098a8e5d91c95f673 : SUCCESS in 20m 07s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/a184f3daf4fc4b27817da00559e1b6bd : SUCCESS in 19m 00s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/6de14b6b59e64b99bd488b6508661047 : SUCCESS in 18m 16s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/c08a8744bf5a44fb9c1dad3bdda53215 : SUCCESS in 22m 20s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a89085aa49d943038229a0a6d59b6f8a : SUCCESS in 8m 01s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e12fe01f1dc34085934105a3d10bd6ec : SUCCESS in 26m 19s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/18644976c6a14a5985c717e4992e0c56 : SUCCESS in 49m 30s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/e165eeec944740adbc1553a6f3556c00 : SUCCESS in 50m 21s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/7f664fa488794f719f8bb8f5fa3b6223 : SUCCESS in 37m 19s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b468ad8e4c334ab9854e625f6376e438 : SUCCESS in 41m 39s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/276aeb36beab4be8b87f64b17857ea2c : SUCCESS in 16m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5656fafdda884181af14d7a7938a0a0b : SUCCESS in 6m 26s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ef726fda89a84b2a9004fdecfd115642 : SUCCESS in 38m 57s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4f8caa4cfe704fc69627019159102969 : SUCCESS in 43m 49s","accounts_in_message":[],"_revision_number":3},{"id":"d9ab6aa21bec55c466e666b21146ec690a993738","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-15 00:14:18.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"b5c7f4d81d42ded12458a4076bffe18582d11187","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-15 01:29:37.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f3662a9c700444dab61d5aaa3e3e39e4 : SUCCESS in 9m 07s\n- grenade https://zuul.opendev.org/t/openstack/build/ea2e0d97dd5b401187a1459628d75959 : SUCCESS in 1h 09m 32s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/b26b71c9154d4d8db0c5a4f9234f9fa4 : SUCCESS in 38m 19s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/f639d96d87714b408ba5bc35434a165f : SUCCESS in 8m 17s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/dce33da4a1fd43e6a7f60c69b3ddda2d : SUCCESS in 9m 10s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/2b2e08dc708c427da27133f56f711181 : SUCCESS in 17m 21s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/61483019512540b4b182e49789b21116 : SUCCESS in 22m 34s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/5125a39fa7204d3f93d7af49f1e6240a : SUCCESS in 20m 54s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/2ac977f5dac54e42b647128111d8b7d8 : SUCCESS in 17m 31s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/d62064bac4c84a8bb21a146367517dc6 : SUCCESS in 17m 18s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/5eb2ac1fa8b84b08800ac43d28cc48f0 : SUCCESS in 21m 28s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/2dd9acd431b947f6b0a66255ed00cb53 : SUCCESS in 18m 50s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/14f71b860aae41fca68de7ebdcc80160 : SUCCESS in 22m 05s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/29a057569f1f4c2fb97922aa04381dbc : SUCCESS in 15m 58s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/5d7e50ef90574f948ec503803aa70aa8 : SUCCESS in 21m 40s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/a9e1c78ef51c499e8c72701fa3395d06 : SUCCESS in 17m 17s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f78ce500691f4e1498691813c79d0832 : SUCCESS in 6m 45s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/3db8e6158a654dda878b87a51ef76f83 : SUCCESS in 52m 12s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/49c584f110ec493ca90e52a6e26c81c7 : SUCCESS in 50m 12s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f3030f78d16b4b2ab48fc6427898cabc : SUCCESS in 41m 22s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c4d70b3b8f1f4170a7706e10ff9aa831 : SUCCESS in 40m 20s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/46d74a036aaf445dac0f17602821d38e : SUCCESS in 16m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c1de4510f583454c862c4b79ff77e1be : SUCCESS in 6m 27s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/035dc100d579455584b4870d97796d34 : SUCCESS in 23m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/15efd4742d604637a54c1fc58d420b4d : SUCCESS in 42m 57s","accounts_in_message":[],"_revision_number":3},{"id":"d30d66338a1e881860d6405bad17fb6533aa38c9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-15 01:29:38.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"c6b9426a442e7d33a87e7808c8ff1772260e4cd7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-15 01:31:09.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1acc6b70a8834cd482b7ab8d174a6e37 : SUCCESS in 1m 13s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/2e69cbe3fa4c44668ceb6df81f351ed2 : SUCCESS in 49s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/3ff19bcb3b5d40e686c7db494d650108 : SUCCESS in 47s (non-voting)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"9b94c278f19781467cc86df4298011f47667e997","revisions":{"a9ca9d4a7901c198086f201971811456aa892643":{"kind":"REWORK","_number":1,"created":"2021-06-24 15:31:04.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/61/797961/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/61/797961/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/61/797961/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/61/797961/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/61/797961/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/61/797961/1"}}},"commit":{"parents":[{"commit":"5923417b1442fd2a0921f574a53c228acc586bc2","subject":"sharder: avoid small tail shards","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5923417b1442fd2a0921f574a53c228acc586bc2"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-06-24 13:54:53.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-06-24 15:28:13.000000000","tz":60},"subject":"sharder: add more validation checks on config","message":"sharder: add more validation checks on config\n\nAdd checks that more of the sharder config values make sense with\nrespect to each other.\n\nFor the daemon, all config options are checked. For\nswift-manage-shard-ranges only those options relevant to the current\nsubcommand are checked.\n\nChange-Id: I2d4dab1d56c31f904161e5338f1bb89a03fd88f7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a9ca9d4a7901c198086f201971811456aa892643"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a9ca9d4a7901c198086f201971811456aa892643"}]},"branch":"refs/heads/master"},"cd1e59f372331e5cca4480d9f16f697a23ad8cde":{"kind":"REWORK","_number":2,"created":"2021-06-28 11:10:23.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/61/797961/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/61/797961/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/61/797961/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/61/797961/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/61/797961/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/61/797961/2"}}},"commit":{"parents":[{"commit":"89fdce188f5efd8c04ac855c1f2ac644f3ff6866","subject":"sharder: avoid small tail shards","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/89fdce188f5efd8c04ac855c1f2ac644f3ff6866"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-06-24 13:54:53.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-06-28 10:01:25.000000000","tz":60},"subject":"sharder: add more validation checks on config","message":"sharder: add more validation checks on config\n\nAdd checks that more of the sharder config values make sense with\nrespect to each other.\n\nFor the daemon, all config options are checked. For\nswift-manage-shard-ranges only those options relevant to the current\nsubcommand are checked.\n\nChange-Id: I2d4dab1d56c31f904161e5338f1bb89a03fd88f7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cd1e59f372331e5cca4480d9f16f697a23ad8cde"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cd1e59f372331e5cca4480d9f16f697a23ad8cde"}]},"branch":"refs/heads/master"},"9b94c278f19781467cc86df4298011f47667e997":{"kind":"REWORK","_number":3,"created":"2021-07-08 11:54:54.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/61/797961/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/61/797961/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/61/797961/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/61/797961/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/61/797961/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/61/797961/3"}}},"commit":{"parents":[{"commit":"2a593174a5d74426e7e771fb65bc5ebd9de08c60","subject":"sharder: avoid small tail shards","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2a593174a5d74426e7e771fb65bc5ebd9de08c60"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-06-24 13:54:53.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-07-08 11:42:16.000000000","tz":60},"subject":"sharder: add more validation checks on config","message":"sharder: add more validation checks on config\n\nAdd checks that more of the sharder config values make sense with\nrespect to each other.\n\nFor the daemon, all config options are checked. For\nswift-manage-shard-ranges only those options relevant to the current\nsubcommand are checked.\n\nChange-Id: I2d4dab1d56c31f904161e5338f1bb89a03fd88f7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9b94c278f19781467cc86df4298011f47667e997"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9b94c278f19781467cc86df4298011f47667e997"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
