)]}'
{"id":"openstack%2Fswift~798275","triplet_id":"openstack%2Fswift~master~Ib70479ab6663b551b21121d1595956a0fce51d3b","project":"openstack/swift","branch":"master","topic":"dnspython2","attention_set":{"29349":{"account":{"_account_id":29349,"name":"Simeon Gourlin","email":"simeon.gourlin@infomaniak.com","username":"6meon"},"last_update":"2021-12-21 02:42:00.000000000","reason":"Pete Zaitcev replied on the change"},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2021-12-21 02:42:00.000000000","reason":"Pete Zaitcev replied on the change"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"Ib70479ab6663b551b21121d1595956a0fce51d3b","subject":"Adding dnspython2 and tcp option on the cname_lookup middleware.","status":"NEW","created":"2021-06-28 09:38:40.000000000","updated":"2021-12-21 02:42:00.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":4,"unresolved_comment_count":3,"has_review_started":true,"meta_rev_id":"bd0ea01c06a7110da3eb8a5f5dce796b1f4928e6","_number":798275,"virtual_id_number":798275,"owner":{"_account_id":29349,"name":"Simeon Gourlin","email":"simeon.gourlin@infomaniak.com","username":"6meon"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2021-10-13 20:50:43.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-28 11:06:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-28 17:31:39.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"CC"},{"updated":"2021-12-21 02:42:00.000000000","updated_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"reviewer":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"state":"CC"}],"messages":[{"id":"8d51d6601830577fda8beb1084ac14ca1f23696c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29349,"name":"Simeon Gourlin","email":"simeon.gourlin@infomaniak.com","username":"6meon"},"date":"2021-06-28 09:38:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"38462a24a59a2f03cc08fc42aa18fd5d79c8a92a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-28 11:06:30.000000000","message":"Patch Set 1: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/967fc5bb222742fbb3baddb26537085b : SUCCESS in 11m 43s\n- grenade https://zuul.opendev.org/t/openstack/build/9195d999579a43318da6b3036807225b : SUCCESS in 1h 21m 37s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/ceb013201bcb459bbbdc4eafe4b2e9f0 : SUCCESS in 49m 57s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/33ae70ac967648b5bb5ddf1a6a26332a : SUCCESS in 10m 13s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/c23f7dc533784e10bb04b6785ea8a9bc : SUCCESS in 11m 28s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/b7ea99282c41484fa3bf1faac511d64f : FAILURE in 19m 05s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/ab6e68bd25ec4a689e0ffff53ee9ad9b : FAILURE in 22m 03s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/8b3ec914e2d3484593fde305c676b1a0 : FAILURE in 20m 25s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/36da8e69cb2f4d888a9dc23044ed2ebe : FAILURE in 19m 34s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/7d87b60fa3e548b0922e789c15bc8c07 : FAILURE in 18m 15s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/3e526e8331e948029bcaf164d75459c0 : SUCCESS in 21m 10s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/51746afae9d9408888f9d7e136c110f5 : SUCCESS in 27m 24s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/ceafaa28eb5545d5a5c9a9ab16925380 : SUCCESS in 25m 38s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/49500d66340f426296f7a17252d92962 : SUCCESS in 21m 46s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/7601f6cbe027485c8ee829a63c837944 : SUCCESS in 21m 49s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/d52b637404fb4470b737fcee363cf8ed : SUCCESS in 23m 41s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/433e6c755c45433aaf9fdfea84826f53 : SUCCESS in 9m 13s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/62f7b6d592cb4beb87dfd4ae5d036ec8 : SUCCESS in 23m 25s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/f2f2a6eb41f94c5aada444212970d45a : SUCCESS in 51m 48s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/5786f60800194ed2b99968a962ec550b : SUCCESS in 54m 14s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/0915212f1baf4e809681050d1058469c : SUCCESS in 45m 28s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/dbc6ac63eae644bc89bfa131d77e2542 : SUCCESS in 39m 35s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9741f74dc2fa4998b876f3d5fc8c0fd5 : FAILURE in 17m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/841bb959d5264b298d27962e6a791371 : FAILURE in 8m 57s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a0c9c15dc615476c993264f66633e1bb : SUCCESS in 23m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8c07851e77f6487abd2f6c611b0c26ee : SUCCESS in 51m 32s","accounts_in_message":[],"_revision_number":1},{"id":"839b541077145e00b56d264ddbc4fbc382927fd7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-28 11:18:53.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/61a973b716954f1ebde1a49cfe7095d1 : TIMED_OUT in 1h 33m 04s (non-voting)\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f0aaa51dc1a143aabbc3e8696945a721 : FAILURE in 1h 02m 07s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b8bcfee2cd32dcd9a9d6c1b5455aa1ef026c7771","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-06-28 17:31:39.000000000","message":"Patch Set 1:\n\n(3 comments)\n\nLooks like our unit tests have a few mocks that will need to be updated to allow for the new tcp argument.","accounts_in_message":[],"_revision_number":1},{"id":"2e07644510be7b5265b5a474a58b34fc80de79e6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-10-13 19:46:09.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"99fa641a1d972d25d03068d9a3dd79832f2c7755","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-13 20:50:43.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b608dab389bc4827818ba540d40084d0 : SUCCESS in 9m 12s\n- grenade https://zuul.opendev.org/t/openstack/build/e4743f98703c4799908df25a3049f568 : SUCCESS in 59m 01s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e1df7379ef9c43a0b137c413516910e1 : SUCCESS in 42m 36s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/32e3a4f4176e4ecf98621a644c33c4e3 : SUCCESS in 8m 49s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/f0bf8552bef2421abd16ce162a64571b : SUCCESS in 9m 56s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/89514ba36ec0427d8de85f7a7b84f127 : SUCCESS in 23m 09s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/1eb85af4af9a4506a94c38d38c398a24 : SUCCESS in 17m 29s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/2b416e4c02a044b09f3db40eb42f6dcd : SUCCESS in 17m 50s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/d06d03538a1d4fc1bd9cae71020e5917 : SUCCESS in 17m 58s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/0fbc27118592484f87d739829583f1a5 : SUCCESS in 17m 28s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/07862cd343f04cafbc837c94176f46ee : SUCCESS in 17m 55s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/c17a1374a8314fe096317293d12e6700 : SUCCESS in 18m 04s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/47a792154d4e40a99bce58e0c78be7b3 : SUCCESS in 19m 16s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/e07a5df130db4c8e90c97036eab9752c : SUCCESS in 38m 55s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/478c40de33b34024a6829975b2fece6c : SUCCESS in 16m 22s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/a82aa38fc7794668aaf4593d06cae3b6 : SUCCESS in 18m 17s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/49df0420909b4005aefa32ddbb19cfa3 : SUCCESS in 6m 52s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/46f5ca2da420401a90feb37faca2ad05 : SUCCESS in 24m 22s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/df0bee6e400c4d248c035da24483cfc2 : SUCCESS in 42m 13s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/9650efe408fa4ade8598016757a9b427 : SUCCESS in 1h 00m 08s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/97e96b40714f412fbdfb8744630dc8af : SUCCESS in 39m 33s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/590870ab03cf4e59ba07cb8e43f996d4 : SUCCESS in 47m 22s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/29788b93efd744bca0a25195ea9fe69c : SUCCESS in 17m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/72f249f3e6b644c586c6abf05a443755 : SUCCESS in 6m 15s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/fcbb9b2039da4a1a9fa9f261bd4378c5 : SUCCESS in 47m 02s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/193e9c4029ed43c288baa247472b1ab8 : SUCCESS in 50m 09s","accounts_in_message":[],"_revision_number":2},{"id":"6ba52db3beeb60747f7d2b66895c472e7073e816","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-13 21:55:45.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4d8c7178412c4e2a9ac032d2b80fb561 : SUCCESS in 58m 22s (non-voting)\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4932407013e24484b7c30810103dcca5 : SUCCESS in 1h 02m 08s (non-voting)\n- swift-probetests-centos-8-arm64 https://zuul.opendev.org/t/openstack/build/ae3329722d104412b9599b50e820e0ff : FAILURE in 2h 06m 27s (non-voting)\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/e0bc47c379cf4e148a1024a8c65cc30b : SUCCESS in 1h 11m 46s (non-voting)\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/a870535144b5428ea09057ece9a731a6 : SUCCESS in 58m 48s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"bd0ea01c06a7110da3eb8a5f5dce796b1f4928e6","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2021-12-21 02:42:00.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"d09883eb8007c377061ae147cc2b378228e61ed5","revisions":{"8816e73f3af8fddab9853f0c63bf0beeb3ae9261":{"kind":"REWORK","_number":1,"created":"2021-06-28 09:38:40.000000000","uploader":{"_account_id":29349,"name":"Simeon Gourlin","email":"simeon.gourlin@infomaniak.com","username":"6meon"},"ref":"refs/changes/75/798275/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/75/798275/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/75/798275/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/75/798275/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/75/798275/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/75/798275/1"}}},"commit":{"parents":[{"commit":"da40d7d89bdcb1714affb996354f2d786f0f8ef7","subject":"Merge \"Quiet EPIPE tracebacks\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/da40d7d89bdcb1714affb996354f2d786f0f8ef7"}]}],"author":{"name":"Simeon Gourlin","email":"simeon.gourlin@infomaniak.com","date":"2021-06-28 09:31:41.000000000","tz":120},"committer":{"name":"Simeon Gourlin","email":"simeon.gourlin@infomaniak.com","date":"2021-06-28 09:38:32.000000000","tz":120},"subject":"Adding dnspython2 and tcp option on the cname_lookup middleware.","message":"Adding dnspython2 and tcp option on the cname_lookup middleware.\n\nThis patch test dnspython major version and uses the new resolver.resole()\nmethod for the 2.x version of dnspython.\n\nIt also add a new \"tcp\" option to config file for cname_lookup middleware,\nif you need to query your DNS with TCP instead of UDP. Option is set to False\nby default to not modify current behaviour.\n\nChange-Id: Ib70479ab6663b551b21121d1595956a0fce51d3b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8816e73f3af8fddab9853f0c63bf0beeb3ae9261"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8816e73f3af8fddab9853f0c63bf0beeb3ae9261"}]},"branch":"refs/heads/master"},"d09883eb8007c377061ae147cc2b378228e61ed5":{"kind":"REWORK","_number":2,"created":"2021-10-13 19:46:09.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/75/798275/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/75/798275/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/75/798275/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/75/798275/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/75/798275/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/75/798275/2"}}},"commit":{"parents":[{"commit":"32cc19962485d2a46d9c5a5f46cfb82234c8b6d9","subject":"Merge \"sharder: Make stats interval configurable\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/32cc19962485d2a46d9c5a5f46cfb82234c8b6d9"}]}],"author":{"name":"Simeon Gourlin","email":"simeon.gourlin@infomaniak.com","date":"2021-06-28 09:31:41.000000000","tz":120},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-10-13 19:44:52.000000000","tz":-420},"subject":"Adding dnspython2 and tcp option on the cname_lookup middleware.","message":"Adding dnspython2 and tcp option on the cname_lookup middleware.\n\nThis patch tests dnspython major version and uses the new resolver.resole()\nmethod for the 2.x version of dnspython.\n\nIt also adds a new \"tcp\" option to config file for cname_lookup middleware,\nif you need to query your DNS with TCP instead of UDP. Option is set to False\nby default to not modify current behaviour.\n\nChange-Id: Ib70479ab6663b551b21121d1595956a0fce51d3b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d09883eb8007c377061ae147cc2b378228e61ed5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d09883eb8007c377061ae147cc2b378228e61ed5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
