)]}'
{"id":"openstack%2Fswift~799818","triplet_id":"openstack%2Fswift~master~I820e133a4e24400ed1e6a87ebf357f7dac463e38","project":"openstack/swift","branch":"master","topic":"p-sharder-tiny-tail","hashtags":[],"change_id":"I820e133a4e24400ed1e6a87ebf357f7dac463e38","subject":"sharder: support rows_per_shard in config file","status":"MERGED","created":"2021-07-07 12:59:57.000000000","updated":"2021-07-07 23:07:44.000000000","submitted":"2021-07-07 23:06:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"799818-p-sharder-tiny-tail","meta_rev_id":"c9d3d495cbd69d088283d3afe15577f3ce5b4e56","_number":799818,"virtual_id_number":799818,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-07-07 23:06:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-07-07 17:28:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-07-07 17:28:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-07 13:47:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-07-07 14:01:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-07 17:28:19.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"3936def4cdb3763230577329dbf772f593b05580","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-07-07 12:59:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b7e1a82ce8e5da84b00949ebef461b91f1ccb45f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-07 13:47:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3f90de9a784b4de191a14f293e81f9eb : SUCCESS in 40m 59s (non-voting)\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/15777b6346564ba4aa7f1d8d1fc6dcad : SUCCESS in 40m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ba75dbc7168b20b02c49f927bbbb3c9b28063ffc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-07 14:01:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bace4320aa024a758a50e56155440ef7 : SUCCESS in 9m 17s\n- grenade https://zuul.opendev.org/t/openstack/build/93746961487648ebb2c67b10eeff20c0 : SUCCESS in 59m 44s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e9a632891af84887a89dc4d82aeacc82 : SUCCESS in 38m 47s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f66d6450f6ae4a3c82eb69a25fd0db51 : SUCCESS in 9m 28s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/fcdbf5f4cac0408b9f61ff13dc3d7c2b : SUCCESS in 8m 26s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/6d715d0a240342b7b8e20ede29cc98ad : SUCCESS in 18m 31s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/88c2f2e6d4d74802b5cd3343491547c0 : SUCCESS in 17m 59s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/2dea8c7b984941f78b3b4d45d90895e1 : SUCCESS in 16m 37s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/7b281a0f47b442f1832c83e8d825aa49 : SUCCESS in 15m 48s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/8007e833e1e948719dc1af08c3758e8e : SUCCESS in 17m 03s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/859ee448451a40a495b558d13983fe36 : SUCCESS in 17m 32s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/2662496348ec493eadb52f492e73f9b3 : SUCCESS in 19m 38s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/e7c3dbf813fd494db882e4a5383d0719 : SUCCESS in 19m 24s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/1d819657950e4f8bae0a4608b52c967c : SUCCESS in 21m 02s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/c135b78f4595432286a20023ef629d53 : SUCCESS in 18m 02s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/2459177f7d3e47f1ad58ee71d1b5c595 : SUCCESS in 22m 41s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/caa936c694a042078393073b51ab92cc : SUCCESS in 7m 39s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/7737e9d28d564943ab3a21362671b281 : SUCCESS in 20m 43s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/39d8cd0187404825b09fa2c518d77e72 : SUCCESS in 47m 31s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/1f87c52c03424fa3811fde2913789922 : SUCCESS in 51m 42s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/7cb8bb26393c4e7b95982ce8fff92d35 : SUCCESS in 44m 14s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/622191743f424e16b8837bb7cf368d2f : SUCCESS in 41m 31s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b077eed8b94a4054bd613d182ac7c8b3 : SUCCESS in 15m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2dc8b255a51741f6a671719e7604544c : SUCCESS in 6m 44s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f30e220e5d494b959974b4cdff16192d : SUCCESS in 24m 13s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0ddd41e200d34f1fac84370d4476a502 : SUCCESS in 54m 40s","accounts_in_message":[],"_revision_number":1},{"id":"aac029cd10989a7ba9b052bdb9e7a9cd93ecdad3","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-07-07 17:28:19.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nYeah, that all seem reasonable.","accounts_in_message":[],"_revision_number":1},{"id":"d41c96acd9022e31e6918fe9c9914ec4fab066f0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-07 17:28:42.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3f0e47718861f8ce425a5b1a717e1fbb4cc7ef85","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-07 18:41:56.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e2f785b51d043d28f026be29ebc5918 : SUCCESS in 9m 32s\n- grenade https://zuul.opendev.org/t/openstack/build/fd7e39ff5ecd465c8520bbbf21ace370 : SUCCESS in 1h 07m 50s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/9a8aeed7853c43f9931f574efd30f122 : SUCCESS in 49m 27s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/6ea8295e419d48cdad72a0e22ed908f8 : SUCCESS in 9m 17s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/a0401790aeaa4486a6b76ead8b9e3b93 : SUCCESS in 10m 39s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/18ce74f4d9744010a2c6c636f474fa20 : SUCCESS in 16m 31s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/11bfa82a4ef74e7096d1fbd60523a6f4 : SUCCESS in 20m 23s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/0a3c98b4b8a94748907511fbbd0b0232 : SUCCESS in 17m 24s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/b99dde7c7bcd450b849945d9a09cb515 : SUCCESS in 20m 45s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/e914657b804f4b55aacb9525fbd62ee9 : SUCCESS in 18m 38s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/cc6fdaf4907749bdaa2f60335354292d : SUCCESS in 18m 42s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/6299531c83cd42b48704125602337ee4 : SUCCESS in 18m 15s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/2f6d22f1757a40308d3d89bf2d8fac07 : SUCCESS in 19m 03s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/c54d46541d1745e494661efe4d97aeb2 : SUCCESS in 18m 04s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/063ed6a67efc4c1b9d342454d19b93f5 : SUCCESS in 17m 17s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/522ca693b50f4ddb9c3ccfb31ac95c5a : SUCCESS in 17m 50s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/0b7b020104b64b52a244ab692229abe5 : SUCCESS in 8m 04s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/c7155aad827440f09263c2aa335b92a1 : SUCCESS in 47m 48s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/3f1dfdfbf13340f09d824ef004bf7c94 : SUCCESS in 50m 07s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/dfc1c628172d40e59b4e600db40bd59c : SUCCESS in 40m 49s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a2933627d3804b39933432d2213f3bdc : SUCCESS in 40m 11s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d0d6c82735a04330acd0cb44533dd041 : SUCCESS in 17m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/739e39be9f0c4991a77a826da9622d67 : SUCCESS in 7m 10s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/1f2a5780c1d243bfaa9f5556c0a7a341 : FAILURE in 23m 57s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/829174c9bc6c4669960022d670c10cab : SUCCESS in 54m 27s","accounts_in_message":[],"_revision_number":1},{"id":"2a7a4c30bf57e4fa6a2e7d190a65d77b8dcef827","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-07-07 19:31:48.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"9e37be44a52c8ada16eac744e543c8a74f51cd71","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-07 20:42:29.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/864ac4178e4a4f20ad74b2feda227a0e : SUCCESS in 56m 50s (non-voting)\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8813e3b2bdab4e0cb798bc048ddb40d4 : SUCCESS in 1h 08m 23s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"abdcd29d31e16ff31b6ae004f3c35fecb3d581e7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-07 20:51:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0784cf2758d84ef2ba62c8daeceec0d2 : SUCCESS in 9m 23s\n- grenade https://zuul.opendev.org/t/openstack/build/0ec79e03f708422388ee0e11cd08c6cb : SUCCESS in 1h 17m 58s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/65cbe355cd8d46f1878d50d1441791a0 : SUCCESS in 56m 05s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/9e90f1c6df9340d09adbbb29f35544af : SUCCESS in 8m 02s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/d1518a8dcdc44928a7e6824b35d9368d : SUCCESS in 8m 18s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/1d1132415d8e4f6aa39db8695183f8c1 : SUCCESS in 16m 03s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/4f536fc0eee84140a4cef2a65e306139 : SUCCESS in 14m 47s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/0b3b5a322ae04309bdec0ba4e8fcd00b : SUCCESS in 14m 29s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/64cd1b190f7845e989666d66443ecebf : SUCCESS in 14m 24s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/a6bc07e3f4174504821edc0a281c599d : SUCCESS in 26m 24s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/1a1cacf0c29b49fdbf3e78dd6f3a3bb9 : SUCCESS in 15m 43s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/8b20ae4f13d24bccac78131eb6408773 : SUCCESS in 22m 13s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/643d28e433e1451c87dd5f3cd154f7f0 : SUCCESS in 16m 41s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/074956a8f401446d814dbdaefed75668 : SUCCESS in 19m 59s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/c66618e27f5b452a8cd01df05da4bbee : SUCCESS in 16m 00s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/c498660f014a486684c5bbdf8a32fb17 : SUCCESS in 23m 13s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/9a3464ec9b1e4658bbef40ef16202d93 : SUCCESS in 6m 56s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/23a17ee3468e47d1b75876103f7d7dc7 : SUCCESS in 22m 41s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/da6424865bde4baeae161c1493430103 : SUCCESS in 51m 16s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/962dcf4e6f334efcb14497115da871d4 : SUCCESS in 46m 58s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/26055dd7c2024b2f84c1c36cbc787f47 : SUCCESS in 40m 42s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/0d08957649284091ba1b1938210a7c22 : SUCCESS in 36m 36s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/558c678f1245459aa7ad64491b90822e : SUCCESS in 14m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9246c8b2d1b146d3ad0043e1cf898c88 : SUCCESS in 6m 38s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a72d14dd69d84fdbb2d18d0dae63b687 : SUCCESS in 31m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/48f3f1743436435b9a2ce2796c10811d : SUCCESS in 59m 13s","accounts_in_message":[],"_revision_number":1},{"id":"4ce7801e7169ce11d98d489144fd16a3f595721d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-07 20:52:01.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"64d8e5706735a26817632abc9633dac7aa4c577c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-07 23:06:06.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d2236fff104d42bdb3e66b3902f40ae7 : SUCCESS in 10m 46s\n- grenade https://zuul.opendev.org/t/openstack/build/79434eb1077249b2b3dba24a30d5cde2 : SUCCESS in 1h 11m 50s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/41e44fc6b1234031bad9a83676d2de19 : SUCCESS in 48m 13s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/2da63c412b0146bba185d3eef3521b7b : SUCCESS in 9m 59s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/51ec711d3d274c09826c6082c104744b : SUCCESS in 10m 23s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/1e697ecfee364da8ac73e6968f820928 : SUCCESS in 18m 57s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/ad6a24b6185847bc9f2047f8c313a644 : SUCCESS in 17m 47s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/3268e627d6884dbdb334c084aeeb3b8d : SUCCESS in 18m 41s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/23f562e9c41547369dc0b56eaac9897b : SUCCESS in 17m 34s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/a4546c896c9e441eb6aaa139e249e37f : SUCCESS in 17m 14s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/baaac624301e443eaf462a2ed35737a2 : SUCCESS in 18m 06s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/d05d59ee517e4d078289419d3ab2acf0 : SUCCESS in 19m 41s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/b5af529c07ef417e97be13d4dca61f2d : SUCCESS in 19m 52s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/f625d2bef9d84aec92d8b1b65ac62459 : SUCCESS in 18m 09s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/e36126e8c24a4bf1b47dab34333aa66c : SUCCESS in 18m 48s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/35cb1b82321143c4ae013a342c6304ca : SUCCESS in 19m 08s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/023473f1b13949a7a06185fd2d8e7155 : SUCCESS in 8m 11s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/40bc5085de244474b53bb8b78d285eaf : SUCCESS in 55m 16s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/4a878f56292840bd8c8ede7aa5a3df92 : SUCCESS in 49m 29s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c32bde000b1240baa8661913cb227267 : SUCCESS in 43m 04s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/96fa1270059e4bef94885f6e9a7da60c : SUCCESS in 44m 15s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0a2c8ef3b6cc4c38b7ec7b185aac658b : SUCCESS in 22m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c1c99bcd75dd4194957e8ec799b4273d : SUCCESS in 6m 53s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/faa15f2ec1dc460a9123b753331523b4 : SUCCESS in 23m 46s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f6c6c5afe9174c6b81d136cb329e21c0 : SUCCESS in 55m 24s","accounts_in_message":[],"_revision_number":1},{"id":"51938bf23b5133597ad62e9e0c56eb751d66fe6e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-07 23:06:08.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"c9d3d495cbd69d088283d3afe15577f3ce5b4e56","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-07 23:07:44.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/49d668608fca47e2be16837b1ec4ad12 : SUCCESS in 1m 01s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/f5dd50fd5a03489a83adfabb098ba2f7 : SUCCESS in 46s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/ae7c2ff3a6c44eeb838228412efc3773 : SUCCESS in 48s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a87317db6e55beee0f57cf01a8a29d8a78db0615","revisions":{"a87317db6e55beee0f57cf01a8a29d8a78db0615":{"kind":"REWORK","_number":1,"created":"2021-07-07 12:59:57.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/18/799818/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/18/799818/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/799818/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/799818/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/799818/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/18/799818/1"}}},"commit":{"parents":[{"commit":"4f96fe5a84c187eed2abaca2181f63a1b125a2a8","subject":"Merge \"relinker: Remove replication locks for empty parts\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4f96fe5a84c187eed2abaca2181f63a1b125a2a8"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-07-07 11:55:50.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-07-07 12:59:36.000000000","tz":60},"subject":"sharder: support rows_per_shard in config file","message":"sharder: support rows_per_shard in config file\n\nMake rows_per_shard an option that can be configured\nin the [container-sharder] section of a config file.\n\nFor auto-sharding, this option was previously hard-coded to\nshard_container_threshold // 2.\n\nThe swift-manage-shard-ranges command line tool already supported\nrows_per_shard on the command line and will now also load it from a\nconfig file if specified. Any value given on the command line takes\nprecedence over any value found in a config file.\n\nChange-Id: I820e133a4e24400ed1e6a87ebf357f7dac463e38\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a87317db6e55beee0f57cf01a8a29d8a78db0615"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a87317db6e55beee0f57cf01a8a29d8a78db0615"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
