)]}'
{"id":"openstack%2Fswift~799866","triplet_id":"openstack%2Fswift~master~I65919f6e45f537b304f7fda3e015771cec2a9cae","project":"openstack/swift","branch":"master","attention_set":{"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2025-03-06 02:12:20.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},"34892":{"account":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"last_update":"2025-03-06 02:12:20.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"removed_from_attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2024-12-02 21:57:45.000000000","reason":"Change was marked work in progress"}},"hashtags":[],"change_id":"I65919f6e45f537b304f7fda3e015771cec2a9cae","subject":"Let WSGI apps signal that a client connection should be closed","status":"NEW","created":"2021-07-07 16:37:27.000000000","updated":"2025-03-06 02:12:20.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":8,"unresolved_comment_count":5,"has_review_started":true,"meta_rev_id":"97f0019c0a494461fc6534daff9a3f5855601490","_number":799866,"virtual_id_number":799866,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2024-12-03 00:27:58.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-07 17:19:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-07-07 17:37:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-07 22:02:03.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"CC"},{"updated":"2024-12-02 21:52:52.000000000","updated_by":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"reviewer":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"state":"CC"}],"messages":[{"id":"5027267d5d36a007dc4fa1689c141b21e124ec40","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-07-07 16:37:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f0a7518047ea89222f261407c324ae109187b054","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-07 17:19:20.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c11af348956046a1b146a61eba00418e : SUCCESS in 39m 57s (non-voting)\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/50e3e50868f5450996a383cfa1752c54 : SUCCESS in 32m 24s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"cd66e4e55903880935f5a3555ed5974d8dad8d01","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-07 17:37:44.000000000","message":"Patch Set 1: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7442d76fc319428288e7f4b9f14e6741 : SUCCESS in 9m 57s\n- grenade https://zuul.opendev.org/t/openstack/build/b4393cc253a243d6a15a8d9e461848f7 : SUCCESS in 58m 45s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/8f51e91f1ef646d8b8b9afee5fc154ea : SUCCESS in 39m 24s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/84dbed0a3f0f43beb7934dac5565d496 : SUCCESS in 8m 50s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/978ed8100a5c4051a765fa2f17199cf4 : SUCCESS in 10m 13s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/7550b964fad646418ace86c8b1baef92 : FAILURE in 17m 42s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/5225ff060f244321acea025f2aaebe6d : SUCCESS in 20m 22s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/2c1047582cd84fa0a03ea8afb6c13202 : SUCCESS in 16m 56s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/e99a700e7f774328b82c4eb28511a922 : SUCCESS in 18m 08s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/90ca1b81f049437084c28137a9b81408 : SUCCESS in 19m 36s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/512d8100e09942ad9b3bb68903aca8f1 : SUCCESS in 17m 29s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/6b05dc786947491a88dec891a4339f93 : SUCCESS in 28m 56s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/842ec0c0c56c44eaae73f717ca2043fe : SUCCESS in 19m 34s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/1bf6bc1e620d49a690597d5df522f53b : SUCCESS in 21m 37s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/3f94dca8a0aa4c11ac2831d2e8011c22 : SUCCESS in 18m 58s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/9e01eb4319e34cd0afc400826bae47af : SUCCESS in 21m 27s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/153f38dd94c34e76b4d9b753c057982b : SUCCESS in 8m 23s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/34a0d729db6f4032a9af24e14af2c560 : SUCCESS in 24m 03s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/64ad18780ae94e18a43a5e87c764cd92 : SUCCESS in 50m 47s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/5444a36a94574afca7e3e360d12ba547 : SUCCESS in 52m 41s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/3686e082915c4d56b5934e720263a085 : SUCCESS in 39m 25s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/42e442d773414d429eeca011a5f2d7fc : SUCCESS in 42m 07s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8037912435da4c28a4b231b0a2487610 : FAILURE in 16m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/04a4742384d94ccfac75af4f722b051a : FAILURE in 6m 51s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/58ba8a8a2b574e4e844da6356ca73c8f : SUCCESS in 27m 41s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2e28d5f3cd0d4e719136987bccd868e2 : SUCCESS in 45m 10s\n\nWarning:\n  Comments left for invalid file test/unit/common/test_wsgi.py\n  Comments left for invalid file test/unit/proxy/test_server.py","accounts_in_message":[],"_revision_number":1},{"id":"e94d28276a9fa15d57beaea206458bc0eb89efa1","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2021-07-07 22:02:03.000000000","message":"Patch Set 1:\n\n(4 comments)\n\nthis is a cool idea!  don\u0027t love the corolocals!  ;)\n\nhttps://review.opendev.org/c/openstack/swift/+/799920","accounts_in_message":[],"_revision_number":1},{"id":"b489942a7eb44a4e66a357af53a31eb151d97706","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-07-07 22:07:20.000000000","message":"Patch Set 1:\n\n(2 comments)\n\nStupid py2","accounts_in_message":[],"_revision_number":1},{"id":"775c34ebebf3498282ffe47faeea01b608891ecb","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-04-25 17:57:30.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0be4ecee425045990dbe9fa56ad012d01830647b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-04-25 20:08:49.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"538bd6ad47841b48b2fcd6168196db47a2f38b7b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-25 22:19:32.000000000","message":"Patch Set 2: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f18314329d86416cbd1edc4b16a77f10 : SUCCESS in 7m 12s\n- grenade https://zuul.opendev.org/t/openstack/build/381ea48bc1ed46f39432555980bb19c4 : SUCCESS in 57m 22s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/19dce12824e243838956fb5c0722582d : SUCCESS in 55m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f250efb8ba2c4d128e2d2e051fb51c1d : SUCCESS in 41m 53s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/0b736cda712e477bbd1a77d6e7811fd6 : SUCCESS in 15m 23s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/588a0c16f0f94e9a8d73d4c9fd2ab8fa : SUCCESS in 17m 36s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/df7f0e0dd43c49039fe16c9a8ab0afa3 : SUCCESS in 16m 58s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/1f15e913a61e4faca4b79af473a28324 : SUCCESS in 7m 00s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/b0f33fff45bc49f2923da37523822cf6 : SUCCESS in 5m 57s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/1cbe330429c44891adefc23b9707c87f : SUCCESS in 14m 54s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/79cd2c20c5b84c0da14543e258476990 : SUCCESS in 12m 19s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/8c92c2d919f44f8181c90cd74da07692 : SUCCESS in 13m 01s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/d94ebcef455549ddb4cf0c6c23130039 : SUCCESS in 14m 06s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/510f3ee0e1264500a3add0d3c2aefb45 : SUCCESS in 16m 27s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/3eccfafba77b4d2a8134fa8a324693ab : SUCCESS in 17m 41s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/9440d829300147d98aad46316281c417 : SUCCESS in 12m 44s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/a13cabc6160c42f8ad8df644393076f9 : SUCCESS in 20m 45s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/8bb150a4878140c58bb159ffa80a0ac9 : SUCCESS in 13m 06s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/008ed793e64f4c8cb35f5651bca7b286 : SUCCESS in 6m 22s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/8e35220494ee43de95655356fc0bd06e : SUCCESS in 19m 37s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/0830c78e33a841fca29be5dd9791163f : TIMED_OUT in 2h 01m 21s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/a22b9711972c4e6b88d77a46d3d66828 : SUCCESS in 1h 00m 25s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/eb1ff7b1ec65437ab2a2a0b349eff6ac : SUCCESS in 28m 26s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/913339c8c7a44ea1a395aa9e71e523e8 : SUCCESS in 40m 38s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/42ede874de60450cb52b39a05a09719c : SUCCESS in 19m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4ef3f41c2455481d9685af7609a21985 : FAILURE in 5m 56s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ae9168ef1fc4409b9f81a153ac705e69 : SUCCESS in 21m 21s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ca9dadacac0645cea951b71696dfbe99 : SUCCESS in 40m 31s\n\nWarning:\n  Comments left for invalid file test/unit/common/test_wsgi.py\n  Comments left for invalid file test/unit/proxy/test_server.py","accounts_in_message":[],"_revision_number":2},{"id":"35509f5c640444288719cab5477365d396002a4f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-25 22:45:51.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/efe971378d884085a86a967052dd76e9 : SUCCESS in 53m 54s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d39825bd1d204a8cb7a0b454d7b0bea6 : SUCCESS in 37m 42s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/adbf52cc302246e7840c433fae256e8a : FAILURE in 2h 34m 27s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/2df36a54e68040a7894b720b1901a69e : SUCCESS in 59m 48s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/00f56d2c04fa4e69a6de0e170c9d551c : SUCCESS in 53m 22s","accounts_in_message":[],"_revision_number":2},{"id":"63ab5e142a07bbe62a4085428c6240fbcf8a7ad1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-04-26 18:13:19.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"66a56ca43394b7bc0f278f11498f9e292480ab8c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-26 20:21:05.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/520bc9eab70b451aa3a8fa5e9d3ea19b : SUCCESS in 8m 33s\n- grenade https://zuul.opendev.org/t/openstack/build/f7762a9f6ea04928b5a038d689ef3f60 : SUCCESS in 54m 13s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/3509f42053db43449c5e0fc58816e5b7 : SUCCESS in 36m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f5ec746cf55745aab7b5cb93c7756969 : SUCCESS in 52m 55s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/47f50fa0f6c74f1aab8b14e22336675b : SUCCESS in 20m 08s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/643cd3c22e2646feb39eba11dc6b6cdb : SUCCESS in 22m 48s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/177561edcb0149709474f83022d4ed72 : SUCCESS in 21m 06s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/fa265f462dcd4d44abe833098146a781 : SUCCESS in 8m 15s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/c769e27857a64bf388fc786781f58462 : SUCCESS in 7m 33s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/411f7362e7c5452781bac936079ab7be : SUCCESS in 16m 05s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/618f5dab585a4a38918c7c94a81135b9 : SUCCESS in 18m 01s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/f01ef34975124156b3ca9dc0970d052a : SUCCESS in 17m 34s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/6847238285544da782beb9a1fdf3c5da : SUCCESS in 16m 43s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/b58ed50e006e4d0b94eb31fce3362526 : SUCCESS in 19m 02s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/21cd48c62ec0427e8774278e507d4221 : SUCCESS in 19m 32s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/33a5547ea0154aa09e15ea60657925bc : SUCCESS in 17m 18s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/6a080df0d0204032ae733d0e2520c25c : SUCCESS in 20m 32s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/c22d9de4cc2b432dbbd4a51ae2ff98fb : SUCCESS in 18m 09s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/9c1a7764ae63462eb4afe5876ad85ac4 : SUCCESS in 7m 53s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/0fca3431166241d9943eaafa216e424a : SUCCESS in 19m 25s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/7a87c255fc5d4f758fdc558620719e90 : SUCCESS in 55m 00s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/f63535559031460aa4e5a711a548df51 : TIMED_OUT in 2h 01m 08s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/1c37d55753e14603aa78f99345e7f431 : SUCCESS in 41m 40s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/3aadd1a692e047c498faf4a0c1175bd1 : SUCCESS in 39m 55s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4ab706d3bdcc4e8498233cf621bb2172 : SUCCESS in 18m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/86459e56db374a758a9769a83915a8e2 : SUCCESS in 5m 59s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/4f9f20d605294f5882c4b0d1fb18caa4 : SUCCESS in 20m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/22c7bfc5ad2a4edeb8ecd7653036e8cb : SUCCESS in 1h 00m 16s","accounts_in_message":[],"_revision_number":3},{"id":"0f5f9b05e17b283993d621071aef8335d9de9b84","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-26 21:18:03.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1d9fcd3012364e8dbe3c22ccbff1dd4a : SUCCESS in 41m 09s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/27eb3ec57ff848d2903fea2748f94a01 : SUCCESS in 40m 50s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/66821661114e47e986adbb9f13f2ccc6 : TIMED_OUT in 3h 01m 37s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/a3590517840f46078b58f73d1dff9b03 : SUCCESS in 54m 04s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/6928e5f87ec74b619b4d52a2ac6df69f : SUCCESS in 46m 45s","accounts_in_message":[],"_revision_number":3},{"id":"da6062aaa7ed5b226b25203c579e4417dbd14afd","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-12-02 21:57:45.000000000","message":"Patch Set 4: Patch Set 3 was rebased\n\nThe following files contain Git conflicts:\n* swift/common/http_protocol.py\n* swift/proxy/controllers/obj.py\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"acdb5b83c9c5dc92ab13d2977212c435817b6fa5","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-12-02 22:01:09.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"339115834430f017d3c98f4db03ea7d91463ef7b","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-12-02 23:06:15.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"4b5394bb9bf13f9aea1a70070a19b7869760e26d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-03 00:27:58.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a9086cdb6d064e888e3e4d02c0acf02d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f465f1516734c068e5a15d380198598 : SUCCESS in 4m 55s\n- grenade https://zuul.opendev.org/t/openstack/build/0b5036d621cc41bc9b3f074c91bc8d0c : SUCCESS in 36m 25s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/5cbd10ddf1724ad49f58ea3c0c89079f : SUCCESS in 34m 17s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/4ce3aeb8cf1545a5ade27e84b93bc822 : SUCCESS in 38m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cdb5754613404bc7aafe81e572bd671b : SUCCESS in 38m 59s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/502c0a483b604175af709b3cdf1330a3 : SUCCESS in 18m 43s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e0aaaac0f96a441094e1f9b39d2948ed : SUCCESS in 12m 23s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9b92a8e57f9343cd8c94bf0c2527d6d8 : SUCCESS in 20m 46s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/0b2023497d764685adc62fdb684355f1 : SUCCESS in 4m 12s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/90f53862f53344eea1d9d21bd3fe0c26 : SUCCESS in 22m 47s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/5f2ca74346ec4747a4f2e976e71bf6ea : SUCCESS in 21m 11s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/ec4da2423f4e4cb7b2eb8d6bebc77a29 : SUCCESS in 23m 00s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/e382dc8fa91e45c283fda7cc2ffd25d4 : SUCCESS in 23m 20s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/303e41586000401a9b27b35f24c95f5b : SUCCESS in 21m 57s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/62a12f1e249c46a397dad82a541a6db1 : SUCCESS in 23m 00s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/e00e88a7757c4da993d7c4d3c61e2334 : SUCCESS in 23m 10s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/354283a01f984310a02d866a8dee3dc2 : SUCCESS in 20m 09s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/f839c04e9a89424290d4bc85cd6cb34e : SUCCESS in 19m 28s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/10c89a67f5df452f888390d6477d44cb : SUCCESS in 20m 41s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/c6fe94623d4741828f34faa93b689f83 : SUCCESS in 5m 43s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/a6e4cffc7afe4aefad9e074315b2aefb : SUCCESS in 12m 10s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/6cb41b80b0ed4768a7930e70977826c8 : SUCCESS in 6m 13s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/fe9c06838a7241eda86ca33e764ecb89 : SUCCESS in 50m 44s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/fe2c4f53e1044e969b2dee02de5ef020 : SUCCESS in 37m 31s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a09356693a88490fb443533e1207a813 : SUCCESS in 30m 42s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9f14d34489fe4d3aa240d80eb6002bd2 : SUCCESS in 23m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c4a7d1d9db9c48bb99bc06d5e41c15c0 : SUCCESS in 4m 06s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/05186613b1b54f25acc474e3e402ea3a : SUCCESS in 23m 05s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f547dfb6661c4fa6adbec161ca25485a : SUCCESS in 1h 04m 41s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6284931ab5d443b38071201e92ebc827 : SUCCESS in 37m 08s","accounts_in_message":[],"_revision_number":6},{"id":"d1cf4e6d18b1142370481c91e178fae3c8881fbd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-03 00:35:43.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4d717ec8a0844023b9626d98ba24c2cd\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/46ece7731cdd435d91cffceeb6c40eff : SUCCESS in 1h 27m 16s","accounts_in_message":[],"_revision_number":6},{"id":"97f0019c0a494461fc6534daff9a3f5855601490","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-03-06 02:12:20.000000000","message":"Patch Set 6:\n\n(1 comment)\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"e4660893cf99b41dfdeccce92c424530cb12e07c","revisions":{"96656824d1c6eae5389ee60adbe3f80a37258dfd":{"kind":"REWORK","_number":1,"created":"2021-07-07 16:37:27.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/66/799866/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/66/799866/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/799866/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/799866/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/799866/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/66/799866/1"}}},"commit":{"parents":[{"commit":"cd196dc8399948299c5e17fee7164e5d4de1b7bf","subject":"Extract SwiftHttpProtocol to its own module","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cd196dc8399948299c5e17fee7164e5d4de1b7bf"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-07-06 22:43:54.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-07-07 16:18:14.000000000","tz":-420},"subject":"Let WSGI apps signal that a client connection should be closed","message":"Let WSGI apps signal that a client connection should be closed\n\nThe WSGI spec makes eventlet (as WSGI server) pretty clearly responsible\nfor connection management, and it errs on the side of allowing\nconnection reuse, even when the client wants to send a *massive* amount\nof data.\n\nSince we\u0027ve already got our own SwiftHttpProtocol that we pass down\nto eventlet, we can use some of eventlet\u0027s existing machinery to take\non this decision ourselves. We might want to do this if we expect the\nclient to send us a *lot* of data that we\u0027d just discard, or if we\u0027re\nworried about request framing issues.\n\nUpdate the proxy server to close out the connection whenever it would\nrespond:\n\n   * 408 Request Timeout,\n   * 411 Length Required,\n   * 413 Request Entity Too Large, or\n   * 499 Client Disconnect\n\nChange-Id: I65919f6e45f537b304f7fda3e015771cec2a9cae\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/96656824d1c6eae5389ee60adbe3f80a37258dfd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/96656824d1c6eae5389ee60adbe3f80a37258dfd"}]},"branch":"refs/heads/master"},"a67f768c433f089318e8ee478f9ee14bc76a9e2c":{"kind":"REWORK","_number":2,"created":"2022-04-25 20:08:49.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/66/799866/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/66/799866/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/799866/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/799866/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/799866/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/66/799866/2"}}},"commit":{"parents":[{"commit":"466cc0b517321171c3c2086baa3ea555e79519ac","subject":"Extract SwiftHttpProtocol to its own module","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/466cc0b517321171c3c2086baa3ea555e79519ac"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-07-06 22:43:54.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-04-25 20:08:37.000000000","tz":-420},"subject":"Let WSGI apps signal that a client connection should be closed","message":"Let WSGI apps signal that a client connection should be closed\n\nThe WSGI spec makes eventlet (as WSGI server) pretty clearly responsible\nfor connection management, and it errs on the side of allowing\nconnection reuse, even when the client wants to send a *massive* amount\nof data.\n\nSince we\u0027ve already got our own SwiftHttpProtocol that we pass down\nto eventlet, we can use some of eventlet\u0027s existing machinery to take\non this decision ourselves. We might want to do this if we expect the\nclient to send us a *lot* of data that we\u0027d just discard, or if we\u0027re\nworried about request framing issues.\n\nUpdate the proxy server to close out the connection whenever it would\nrespond:\n\n   * 408 Request Timeout,\n   * 411 Length Required,\n   * 413 Request Entity Too Large, or\n   * 499 Client Disconnect\n\nChange-Id: I65919f6e45f537b304f7fda3e015771cec2a9cae\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a67f768c433f089318e8ee478f9ee14bc76a9e2c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a67f768c433f089318e8ee478f9ee14bc76a9e2c"}]},"branch":"refs/heads/master"},"f6c5346f0d93623ecef53be68a3fa1d82d10a4c6":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2022-04-26 18:13:19.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/66/799866/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/66/799866/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/799866/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/799866/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/799866/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/66/799866/3"}}},"commit":{"parents":[{"commit":"d2682d2b36165821bb609ff262e19b73b21b37d2","subject":"Extract SwiftHttpProtocol to its own module","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d2682d2b36165821bb609ff262e19b73b21b37d2"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-07-06 22:43:54.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-04-26 18:09:07.000000000","tz":-420},"subject":"Let WSGI apps signal that a client connection should be closed","message":"Let WSGI apps signal that a client connection should be closed\n\nThe WSGI spec makes eventlet (as WSGI server) pretty clearly responsible\nfor connection management, and it errs on the side of allowing\nconnection reuse, even when the client wants to send a *massive* amount\nof data.\n\nSince we\u0027ve already got our own SwiftHttpProtocol that we pass down\nto eventlet, we can use some of eventlet\u0027s existing machinery to take\non this decision ourselves. We might want to do this if we expect the\nclient to send us a *lot* of data that we\u0027d just discard, or if we\u0027re\nworried about request framing issues.\n\nUpdate the proxy server to close out the connection whenever it would\nrespond:\n\n   * 408 Request Timeout,\n   * 411 Length Required,\n   * 413 Request Entity Too Large, or\n   * 499 Client Disconnect\n\nChange-Id: I65919f6e45f537b304f7fda3e015771cec2a9cae\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f6c5346f0d93623ecef53be68a3fa1d82d10a4c6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f6c5346f0d93623ecef53be68a3fa1d82d10a4c6"}]},"branch":"refs/heads/master"},"8a9f9f23d2714a65e5320f68d44756c36b1025ed":{"kind":"REWORK","_number":4,"created":"2024-12-02 21:57:45.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/66/799866/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/66/799866/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/799866/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/799866/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/799866/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/66/799866/4"}}},"commit":{"parents":[{"commit":"2dad4dcbbd71191d3ec014f93eeefd4d2013d293","subject":"Merge \"Update gate jobs as per the 2025.1 cycle testing runtime\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2dad4dcbbd71191d3ec014f93eeefd4d2013d293"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-07-06 22:43:54.000000000","tz":-420},"committer":{"name":"ASHWIN A NAIR","email":"nairashwin952013@gmail.com","date":"2024-12-02 21:57:45.000000000","tz":0},"subject":"Let WSGI apps signal that a client connection should be closed","message":"Let WSGI apps signal that a client connection should be closed\n\nThe WSGI spec makes eventlet (as WSGI server) pretty clearly responsible\nfor connection management, and it errs on the side of allowing\nconnection reuse, even when the client wants to send a *massive* amount\nof data.\n\nSince we\u0027ve already got our own SwiftHttpProtocol that we pass down\nto eventlet, we can use some of eventlet\u0027s existing machinery to take\non this decision ourselves. We might want to do this if we expect the\nclient to send us a *lot* of data that we\u0027d just discard, or if we\u0027re\nworried about request framing issues.\n\nUpdate the proxy server to close out the connection whenever it would\nrespond:\n\n   * 408 Request Timeout,\n   * 411 Length Required,\n   * 413 Request Entity Too Large, or\n   * 499 Client Disconnect\n\nChange-Id: I65919f6e45f537b304f7fda3e015771cec2a9cae\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8a9f9f23d2714a65e5320f68d44756c36b1025ed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8a9f9f23d2714a65e5320f68d44756c36b1025ed"}]},"branch":"refs/heads/master","description":"Rebase"},"d0722eba9845d35e488414cbefd465857b00e74b":{"kind":"REWORK","_number":5,"created":"2024-12-02 22:01:09.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/66/799866/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/66/799866/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/799866/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/799866/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/799866/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/66/799866/5"}}},"commit":{"parents":[{"commit":"2dad4dcbbd71191d3ec014f93eeefd4d2013d293","subject":"Merge \"Update gate jobs as per the 2025.1 cycle testing runtime\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2dad4dcbbd71191d3ec014f93eeefd4d2013d293"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-07-06 22:43:54.000000000","tz":-420},"committer":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-12-02 22:01:01.000000000","tz":-480},"subject":"Let WSGI apps signal that a client connection should be closed","message":"Let WSGI apps signal that a client connection should be closed\n\nThe WSGI spec makes eventlet (as WSGI server) pretty clearly responsible\nfor connection management, and it errs on the side of allowing\nconnection reuse, even when the client wants to send a *massive* amount\nof data.\n\nSince we\u0027ve already got our own SwiftHttpProtocol that we pass down\nto eventlet, we can use some of eventlet\u0027s existing machinery to take\non this decision ourselves. We might want to do this if we expect the\nclient to send us a *lot* of data that we\u0027d just discard, or if we\u0027re\nworried about request framing issues.\n\nUpdate the proxy server to close out the connection whenever it would\nrespond:\n\n   * 408 Request Timeout,\n   * 411 Length Required,\n   * 413 Request Entity Too Large, or\n   * 499 Client Disconnect\n\nChange-Id: I65919f6e45f537b304f7fda3e015771cec2a9cae\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d0722eba9845d35e488414cbefd465857b00e74b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d0722eba9845d35e488414cbefd465857b00e74b"}]},"branch":"refs/heads/master"},"e4660893cf99b41dfdeccce92c424530cb12e07c":{"kind":"REWORK","_number":6,"created":"2024-12-02 23:06:15.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/66/799866/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/66/799866/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/799866/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/799866/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/799866/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/66/799866/6"}}},"commit":{"parents":[{"commit":"2dad4dcbbd71191d3ec014f93eeefd4d2013d293","subject":"Merge \"Update gate jobs as per the 2025.1 cycle testing runtime\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2dad4dcbbd71191d3ec014f93eeefd4d2013d293"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-07-06 22:43:54.000000000","tz":-420},"committer":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-12-02 23:06:09.000000000","tz":-480},"subject":"Let WSGI apps signal that a client connection should be closed","message":"Let WSGI apps signal that a client connection should be closed\n\nThe WSGI spec makes eventlet (as WSGI server) pretty clearly responsible\nfor connection management, and it errs on the side of allowing\nconnection reuse, even when the client wants to send a *massive* amount\nof data.\n\nSince we\u0027ve already got our own SwiftHttpProtocol that we pass down\nto eventlet, we can use some of eventlet\u0027s existing machinery to take\non this decision ourselves. We might want to do this if we expect the\nclient to send us a *lot* of data that we\u0027d just discard, or if we\u0027re\nworried about request framing issues.\n\nUpdate the proxy server to close out the connection whenever it would\nrespond:\n\n   * 408 Request Timeout,\n   * 411 Length Required,\n   * 413 Request Entity Too Large, or\n   * 499 Client Disconnect\n\nChange-Id: I65919f6e45f537b304f7fda3e015771cec2a9cae\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e4660893cf99b41dfdeccce92c424530cb12e07c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e4660893cf99b41dfdeccce92c424530cb12e07c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
