)]}'
{"id":"openstack%2Fswift~799920","triplet_id":"openstack%2Fswift~master~Ided663c5d7f42d6cdfb307fb893948cfedac930f","project":"openstack/swift","branch":"master","attention_set":{"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2024-12-03 01:38:57.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},"34892":{"account":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"last_update":"2024-12-03 01:38:57.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"removed_from_attention_set":{},"hashtags":[],"change_id":"Ided663c5d7f42d6cdfb307fb893948cfedac930f","subject":"WIP: playing with close_connection","status":"NEW","created":"2021-07-07 21:58:29.000000000","updated":"2024-12-03 01:38:57.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":6,"unresolved_comment_count":4,"work_in_progress":true,"has_review_started":true,"meta_rev_id":"ace325e2315c6030682987174ee8e224d380cebf","_number":799920,"virtual_id_number":799920,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-12-03 01:12:04.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{"CC":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"reviewer_updates":[{"updated":"2021-07-07 22:32:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-07-08 00:04:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-12-03 01:38:57.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"CC"}],"messages":[{"id":"767978215261da292694890b10bf122b25f2aabf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2021-07-07 21:58:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b68cd12eab9c17e7f6cdb487612855c8232ee475","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2021-07-07 22:01:35.000000000","message":"Patch Set 1: Code-Review-1 Workflow-1\n\n(2 comments)\n\ni like the idea of using environ better than using globals - this might work on py3 since there\u0027s no such thing as old style classes and types.InstanceType will attribute error\n\nbut it doesn\u0027t work with upstream eventlet on py2 any better than the dependent change","accounts_in_message":[],"_revision_number":1},{"id":"73a0e166199038e0094d1ee1dff16c476e6fdf86","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-07 22:32:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/efe0ead504ac4016926c1ee8eccbb5b9 : FAILURE in 29m 46s (non-voting)\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0a16fe4b8e2a4e3c9a223c1d02270e84 : FAILURE in 29m 51s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5f7b0340332e32c7278741c07c8c85f83b9778ab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-08 00:04:04.000000000","message":"Patch Set 1: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/779062eb32f04ce195460388809249e4 : SUCCESS in 9m 27s\n- grenade https://zuul.opendev.org/t/openstack/build/85d36aa491994cb09a40f46089f323f8 : SUCCESS in 57m 38s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c1705616201440a5a7300407d2b8e611 : SUCCESS in 37m 01s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/9a76120143b44298bba2d585b1cecb94 : SUCCESS in 9m 16s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/657bc13ba4624213b3f99df45e3fbf38 : SUCCESS in 9m 30s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/be1874a63d064a13b93f772060e0ba79 : FAILURE in 14m 16s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/331c44e8e3cb4fb984602bb74fd07c32 : FAILURE in 15m 51s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/116a47e1e2e94cfb919d4d1fc573f09c : FAILURE in 21m 21s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/890004545086450fa4d1bf8f5394e6ee : FAILURE in 16m 32s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/f6891438794b452e9e12954f103cc43c : FAILURE in 21m 34s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/dafa84fcd0254611bb2e2fba37adc1e9 : FAILURE in 6m 00s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/f8db694aab2a4c1d81e1c3b900e33798 : FAILURE in 7m 21s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/c98a6faa56d64267bb75963a4c02708e : FAILURE in 6m 04s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/bccccf024bfd413383eb225ce8f539e7 : FAILURE in 42m 24s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/b6f3ef39bbc544d5b44034b9b7612341 : FAILURE in 55m 43s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/1e485fb262264c1381c7ecbd5669f62e : FAILURE in 20m 43s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/ef43d053cfee4ae4929934ac57986e4d : FAILURE in 7m 13s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/64e5c2d76fc245e7933b3a1adb2975d7 : FAILURE in 12m 34s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/3a8b8ccddf02476fa2adfdd0e8661aa0 : TIMED_OUT in 2h 01m 17s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/35c2319a30b946798dd1e40e7a8bd68f : FAILURE in 1h 39m 26s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d9a9dbe9a8c54e08b1bf3d003333ce23 : FAILURE in 28m 40s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f04a02892a3c473fa3e393811ed67322 : FAILURE in 27m 34s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0b1c0f15f0b64166ae3482b0e958a437 : FAILURE in 14m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3f85aafa222345a79400fdb0a50e4b43 : FAILURE in 6m 39s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/1958909bb3144aaa9372bbe0dbaf3619 : TIMED_OUT in 1h 31m 17s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f1ea33892a894faeac05eeaa01e7958c : SUCCESS in 39m 43s\n\nWarning:\n  Comments left for invalid file test/unit/common/test_wsgi.py\n  Comments left for invalid file test/unit/proxy/test_server.py","accounts_in_message":[],"_revision_number":1},{"id":"8ee689c86a5acee620ee29f99890a82e293d4103","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-12-02 22:15:20.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nThe following files contain Git conflicts:\n* swift/proxy/controllers/obj.py\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":2},{"id":"af86fabb4e22a7ff81676df6515bb254793c2983","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-02 22:25:57.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/557b54ae4abd47e0b9dbe4c144ec6e8b\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/06c0e20554de43a5853f727646a47795 : FAILURE in 8m 40s","accounts_in_message":[],"_revision_number":2},{"id":"322190679c2b14689d40fa3a72b342827345cc89","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-12-02 22:38:04.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"4412631c5069b716519b70435a11874488420cd3","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-12-02 23:06:27.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"4009a9389865fccc5a544c2b1cd9e87b7aa333b8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-03 01:08:13.000000000","message":"Patch Set 4:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9048df6b6904497cbc8e7dac3594d30e\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/7b0061b26c01475e9847b4fb49bb6209 : FAILURE in 1h 59m 37s","accounts_in_message":[],"_revision_number":4},{"id":"ba7dc9a2a9baf08dffdc89ee9527a38d6a51203d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-03 01:12:04.000000000","message":"Patch Set 4: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f6b39f02e301410aa18430e8266b2d2f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0d41d9c3c594fd792b09aa711de19be : SUCCESS in 8m 22s\n- grenade https://zuul.opendev.org/t/openstack/build/96909060aaf44e17983364ec9a7a0376 : FAILURE in 36m 41s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c4c73e88034540b2b9233f485f74b5e6 : FAILURE in 39m 23s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/1f21905d81364d92900dcd8cc1f9cc45 : FAILURE in 21m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0ea6cf9c1ee14598ad220dab6d112f94 : POST_FAILURE in 22m 25s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fb85caef0c35481fb27b6cd81efc0cf9 : FAILURE in 6m 07s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9e0d9565ac914ae4adc3477bebafcaf0 : FAILURE in 4m 51s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3740a1f75dc04a5d993c4a9f03bf8129 : FAILURE in 6m 47s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/8feb4a277ec84ef4be367b66dbfab882 : SUCCESS in 4m 40s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/b22252e6f86346b1882deba6d4a9f2d0 : FAILURE in 37m 50s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/ccad777c04184144bc4879128919a91f : FAILURE in 40m 16s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/c033e868138144519cdbc19be5bd5979 : FAILURE in 20m 59s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/3d61346bea33421a8ea3e423f3f1dcde : FAILURE in 1h 01m 50s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/1f62e9e2b6bd44a9b39af104a751f23c : FAILURE in 4m 12s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/d8bd3f0d841f4d03a3eb01aeda35f8d0 : FAILURE in 5m 31s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/936038cea9a640d9be912c19dfd42dd5 : FAILURE in 2m 45s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/a4d0a4f0eeac423e806442f0f6706233 : FAILURE in 4m 07s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/81d887560f7d460ba237142b4a8530c5 : FAILURE in 3m 57s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/a8726ff54f214b29953564b1d79a4752 : FAILURE in 5m 41s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/86e01f75be8043c8a1b06d0a1da760bb : FAILURE in 7m 47s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/54d479b14a77470a99a5e711af2cd46d : FAILURE in 8m 37s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/db365739ccdc4b8bb03635669ae2329a : FAILURE in 14m 12s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/ed5f296b394649e09097b2a65ec4a5cd : TIMED_OUT in 2h 01m 10s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/1a6258dedc4e4fcfa02a20d653c1ae0b : FAILURE in 12m 02s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b0c9d504e36c47daa0e0ae61e0a85066 : FAILURE in 12m 06s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8c9587342b5c462c8a2c78baa750087f : FAILURE in 27m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0851d28ab368471e8eeff17bb6fd6182 : FAILURE in 3m 38s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/05d8a8a5a1b944439981d1286f65e37a : TIMED_OUT in 1h 30m 58s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d324f74c34db439cabb4aeadcdd7e2f8 : FAILURE in 20m 33s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5af939a51bf04970a7363077007c3b4f : FAILURE in 28m 39s","accounts_in_message":[],"_revision_number":4},{"id":"ace325e2315c6030682987174ee8e224d380cebf","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-12-03 01:38:57.000000000","message":"Patch Set 4:\n\n(4 comments)","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"f7793db0f0efc95cdc8133095a9bc90593a19e6b","revisions":{"3f0e96f59332586f020d9a4a4bfc7f65b54a6153":{"kind":"REWORK","_number":1,"created":"2021-07-07 21:58:29.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/20/799920/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/20/799920/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/20/799920/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/20/799920/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/20/799920/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/20/799920/1"}}},"commit":{"parents":[{"commit":"96656824d1c6eae5389ee60adbe3f80a37258dfd","subject":"Let WSGI apps signal that a client connection should be closed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/96656824d1c6eae5389ee60adbe3f80a37258dfd"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2021-07-07 21:58:20.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2021-07-07 21:58:20.000000000","tz":-300},"subject":"WIP: playing with close_connection","message":"WIP: playing with close_connection\n\nChange-Id: Ided663c5d7f42d6cdfb307fb893948cfedac930f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3f0e96f59332586f020d9a4a4bfc7f65b54a6153"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3f0e96f59332586f020d9a4a4bfc7f65b54a6153"}]},"branch":"refs/heads/master"},"02ea1bebbe8895a389a0b4643c8aefcc4bfd2f33":{"kind":"REWORK","_number":2,"created":"2024-12-02 22:15:20.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/20/799920/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/20/799920/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/20/799920/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/20/799920/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/20/799920/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/20/799920/2"}}},"commit":{"parents":[{"commit":"d0722eba9845d35e488414cbefd465857b00e74b","subject":"Let WSGI apps signal that a client connection should be closed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d0722eba9845d35e488414cbefd465857b00e74b"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2021-07-07 21:58:20.000000000","tz":-300},"committer":{"name":"ASHWIN A NAIR","email":"nairashwin952013@gmail.com","date":"2024-12-02 22:15:20.000000000","tz":0},"subject":"WIP: playing with close_connection","message":"WIP: playing with close_connection\n\nChange-Id: Ided663c5d7f42d6cdfb307fb893948cfedac930f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/02ea1bebbe8895a389a0b4643c8aefcc4bfd2f33"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/02ea1bebbe8895a389a0b4643c8aefcc4bfd2f33"}]},"branch":"refs/heads/master","description":"Rebase"},"92722b228a43feed40aa1c78faeaea69e7a849be":{"kind":"REWORK","_number":3,"created":"2024-12-02 22:38:04.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/20/799920/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/20/799920/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/20/799920/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/20/799920/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/20/799920/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/20/799920/3"}}},"commit":{"parents":[{"commit":"d0722eba9845d35e488414cbefd465857b00e74b","subject":"Let WSGI apps signal that a client connection should be closed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d0722eba9845d35e488414cbefd465857b00e74b"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2021-07-07 21:58:20.000000000","tz":-300},"committer":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-12-02 22:37:30.000000000","tz":-480},"subject":"WIP: playing with close_connection","message":"WIP: playing with close_connection\n\nChange-Id: Ided663c5d7f42d6cdfb307fb893948cfedac930f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/92722b228a43feed40aa1c78faeaea69e7a849be"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/92722b228a43feed40aa1c78faeaea69e7a849be"}]},"branch":"refs/heads/master"},"f7793db0f0efc95cdc8133095a9bc90593a19e6b":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2024-12-02 23:06:27.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/20/799920/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/20/799920/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/20/799920/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/20/799920/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/20/799920/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/20/799920/4"}}},"commit":{"parents":[{"commit":"e4660893cf99b41dfdeccce92c424530cb12e07c","subject":"Let WSGI apps signal that a client connection should be closed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e4660893cf99b41dfdeccce92c424530cb12e07c"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2021-07-07 21:58:20.000000000","tz":-300},"committer":{"name":"ASHWIN A NAIR","email":"ashnair@nvidia.com","date":"2024-12-02 23:06:27.000000000","tz":0},"subject":"WIP: playing with close_connection","message":"WIP: playing with close_connection\n\nChange-Id: Ided663c5d7f42d6cdfb307fb893948cfedac930f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f7793db0f0efc95cdc8133095a9bc90593a19e6b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f7793db0f0efc95cdc8133095a9bc90593a19e6b"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
