)]}'
{"id":"openstack%2Fswift~800494","triplet_id":"openstack%2Fswift~master~I5cf2b3dd6d6965b0d56c89ee681228d6e241cb63","project":"openstack/swift","branch":"master","topic":"ppi_aware_reconciler","hashtags":[],"change_id":"I5cf2b3dd6d6965b0d56c89ee681228d6e241cb63","subject":"squash: reconciler: PPI aware reconciler","status":"ABANDONED","created":"2021-07-12 15:00:06.000000000","updated":"2021-07-14 17:42:00.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"19a0d0184ec4e99325acd567798ec12ea824f46c","_number":800494,"virtual_id_number":800494,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2021-07-12 16:35:05.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-12 16:25:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-07-12 16:35:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-13 06:13:50.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"CC"}],"messages":[{"id":"de32a7018b45ddf20bd123dabbf2160cad434ddd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-07-12 15:00:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cc1a7276278aa19c998648124371b310c6cb846f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-12 16:25:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/51adfae3de4d468eabd0e1e379d2b19a : SUCCESS in 1h 23m 52s (non-voting)\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ed435cce19624141bb6e17de6bbba677 : SUCCESS in 1h 15m 52s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"59ed0a69f8cf9df2904ae161447cbf96d624f1d0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-12 16:35:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a31f299442244f888980b565e864b91 : SUCCESS in 9m 07s\n- grenade https://zuul.opendev.org/t/openstack/build/32011abef4da4cf699b816efd4b57f73 : SUCCESS in 1h 30m 19s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/9f5916699ea94284826d5a9ff2d50223 : SUCCESS in 42m 52s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e8321581e511492caf263b1c24e7c165 : SUCCESS in 13m 57s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/7b9ca1cab07f4d1d8eb3f732ef9606c2 : SUCCESS in 8m 51s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/936d72e5234545fa86aef1edbc4c832d : SUCCESS in 19m 53s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/85b8a7cf426546c881afc8cb2282ca26 : SUCCESS in 16m 50s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/a50d48f8fdd947808a2d513ee4cc9d76 : SUCCESS in 26m 49s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/475d3133816946d2aa2787ad86bb7638 : SUCCESS in 17m 01s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/59abc9b4faf24142a01cf1ab9c314ad4 : SUCCESS in 22m 03s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/5df82fba014744768ce830209a36af87 : SUCCESS in 16m 20s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/2f0bf5c47e0b4645867fb72e50c3c3b6 : SUCCESS in 22m 33s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/52a87d4aa3ce40eda5b5ef8ad1e85510 : SUCCESS in 17m 13s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/4ae5e9a9791f4ff79cd9735d6644bbd5 : SUCCESS in 21m 58s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/568f26608d3b4185ab7eddbe89eaca56 : SUCCESS in 16m 51s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/0214ef4bf33142139af811f19c16c600 : SUCCESS in 17m 09s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/aa52a08e663c47519d63bed1d9c0fda3 : SUCCESS in 8m 14s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/1576ff9f21104f4e9d16ce8f944754eb : SUCCESS in 19m 57s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/b2fe525cadb54eac9ffc344ea566fefd : SUCCESS in 48m 14s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/8e86f846a96949d7a7143a63abd8b8fd : SUCCESS in 45m 08s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/9e77b7bf21c44f3c9b4e56223c93913c : SUCCESS in 38m 58s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a5d3af10055a4ec2b408979d45a9f9e9 : SUCCESS in 38m 38s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cba2ce2991ae4ed68ced698dc260c2ea : SUCCESS in 16m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7698533a743a4968bbbd794b132b2921 : SUCCESS in 6m 07s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/214f96d8a8e34df79336bf108d210b60 : SUCCESS in 24m 43s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/657cd52cb82c447695d2a3c52e63732c : SUCCESS in 43m 56s","accounts_in_message":[],"_revision_number":1},{"id":"1d3e2a8bd0c378c7f33aa21bc168b6f5193c3253","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2021-07-13 06:13:50.000000000","message":"Patch Set 1:\n\nsquashed into parent patch. This can be abandoned.","accounts_in_message":[],"_revision_number":1},{"id":"19a0d0184ec4e99325acd567798ec12ea824f46c","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-07-14 17:42:00.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a58a06c89116dd5521b46a9c05ce20e208a65b9d","revisions":{"a58a06c89116dd5521b46a9c05ce20e208a65b9d":{"kind":"REWORK","_number":1,"created":"2021-07-12 15:00:06.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/94/800494/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/94/800494/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/94/800494/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/94/800494/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/94/800494/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/94/800494/1"}}},"commit":{"parents":[{"commit":"7f4e1d9e6d64dacccc06351a604e186b73eaf885","subject":"reconciler: PPI aware reconciler","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7f4e1d9e6d64dacccc06351a604e186b73eaf885"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-07-12 14:55:53.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-07-12 14:59:48.000000000","tz":-420},"subject":"squash: reconciler: PPI aware reconciler","message":"squash: reconciler: PPI aware reconciler\n\nChange-Id: I5cf2b3dd6d6965b0d56c89ee681228d6e241cb63\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a58a06c89116dd5521b46a9c05ce20e208a65b9d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a58a06c89116dd5521b46a9c05ce20e208a65b9d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
