)]}'
{"id":"openstack%2Fswift~800701","triplet_id":"openstack%2Fswift~master~Ib5cec0e552cd51592653e9e100729fcf91f99415","project":"openstack/swift","branch":"master","topic":"fix-rebase","attention_set":{"7233":{"account":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"last_update":"2023-07-25 03:51:58.000000000","reason":"A robot voted negatively on a label"},"34930":{"account":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"last_update":"2026-02-10 05:54:54.000000000","reason":"A robot voted negatively on a label"},"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2024-06-04 22:19:58.000000000","reason":"A robot voted negatively on a label"},"34892":{"account":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"last_update":"2025-04-11 00:21:38.000000000","reason":"A robot voted negatively on a label"},"35790":{"account":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"last_update":"2026-07-22 23:58:01.000000000","reason":"Vote got outdated and was removed: Code-Review+1"}},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2026-03-25 10:00:41.000000000","reason":"\u003cGERRIT_ACCOUNT_7847\u003e replied on the change","reason_account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2025-05-20 19:15:17.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"Ib5cec0e552cd51592653e9e100729fcf91f99415","subject":"Delete s3api MPU segments when expiring the manifest","status":"NEW","created":"2021-07-13 17:47:53.000000000","updated":"2026-09-11 21:58:04.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":147,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"fa557a6226b3708c42b776b8f57fe47e440f592e","_number":800701,"virtual_id_number":800701,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-09-11 21:58:04.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-13 18:23:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-07-13 18:47:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-15 00:12:09.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"CC"},{"updated":"2021-12-22 06:32:26.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"CC"},{"updated":"2021-12-22 11:16:08.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"CC"},{"updated":"2022-01-10 11:22:58.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"},{"updated":"2022-01-11 23:06:41.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"REVIEWER"},{"updated":"2022-06-08 00:58:56.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2023-10-24 16:45:35.000000000","updated_by":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"reviewer":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"state":"REVIEWER"},{"updated":"2025-07-14 18:58:34.000000000","updated_by":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"reviewer":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"state":"REVIEWER"},{"updated":"2026-03-03 16:21:03.000000000","updated_by":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"reviewer":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"state":"CC"},{"updated":"2026-03-03 17:42:11.000000000","updated_by":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"reviewer":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"state":"REVIEWER"}],"messages":[{"id":"056f30318140a52bc13757d5a42e60a33e7d826b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2021-07-13 17:47:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ffdecaa4fc2e516ae10bd5210bc7bd7c780b13c2","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2021-07-13 18:05:36.000000000","message":"Patch Set 1: Workflow-1\n\n(7 comments)\n\nI could write up some tests for this POC and we could carry it until s3api is using ALOs\n\n... or we could try to tease out some of the s3api specific bits from the proxy and object server such that it\u0027s doing something generic with sysmeta and then call it a real improvement to the expirer to handle s3api DELETE responses.  Maybe even s3api itself could benifit (no more HEAD before every DELETE?), or somehow even PUT overwrite responses could take advantage of similar \"best effort\" cleanup.","accounts_in_message":[],"_revision_number":1},{"id":"7f028cc948647840eb6c51728ec18e583e0ec4f9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-13 18:23:52.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/50bfdf7cfb524ee6aab8f27553cc0a5c : FAILURE in 29m 52s (non-voting)\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/40f5541b2fda4fe19028bbd0f2831010 : FAILURE in 30m 11s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"32ab506a72d5d370a7083fffe7bcae0646dc2eeb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-13 18:47:27.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/77ae275fac374d31af46761da11ca621 : SUCCESS in 8m 49s\n- grenade https://zuul.opendev.org/t/openstack/build/f49476625e1643f498c12d4464d91c18 : SUCCESS in 57m 55s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c809d7be46694d6eab7cd13f1cc69fcb : SUCCESS in 47m 28s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/90224ee6ddbf46aeb80fd8bced524219 : SUCCESS in 7m 59s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/21edae74acdb466bb122aec1e2e3e031 : SUCCESS in 8m 40s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/24eb58961f1e48d290cc1bce2f9a3758 : FAILURE in 17m 51s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/8a6a820aa093460baa37a528310a4a7e : FAILURE in 21m 20s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/382af45e56994d46a22aaddbe8409ba3 : FAILURE in 19m 48s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/1e601de4574d4e4e81c4705e8ed98dd6 : FAILURE in 16m 35s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/5c4cd5ad81dc41238a78ab67a5953273 : FAILURE in 17m 06s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/c5608cbc95b549fa93c5a7e2a123a09f : SUCCESS in 19m 46s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/398cecd5e85e42b7b00ef496ed8f62eb : SUCCESS in 16m 35s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/ccf4477af9da4481a6702d9c4dba28f8 : SUCCESS in 17m 42s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/884deb4232494c5cbeed607fc2b33487 : SUCCESS in 19m 03s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/21ce1f88c2ae42538525e508835dbbe1 : SUCCESS in 25m 58s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/52b71480eacd445392287fea35962ffe : SUCCESS in 19m 02s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/dcc60a5a1f8440d7acaf8509638a89d7 : SUCCESS in 6m 17s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/84f4cc5a382b4f95b9299f12de6800e7 : SUCCESS in 27m 17s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/9b529b94116b46af8b7d9975708e08c2 : SUCCESS in 53m 01s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/6bdc80f4317146f68ca6aed2f7926329 : SUCCESS in 55m 16s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/8025a99b167742f5bd2688153e95a826 : SUCCESS in 41m 54s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a4455eca78d5401ebf3228468e70e031 : SUCCESS in 51m 57s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9370a441752c45ddaeafac99cbaa8a73 : FAILURE in 16m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0106630a93894318bd9f3f16d1802fa1 : SUCCESS in 7m 00s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e81d01241aca4c39acbfc8058b94b973 : SUCCESS in 27m 39s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/333efede091c49d3a8c854769bae4cb7 : SUCCESS in 50m 36s","accounts_in_message":[],"_revision_number":1},{"id":"98f453900bdf43044753978046da00e34aab5af8","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-07-15 00:12:09.000000000","message":"Patch Set 1:\n\n(6 comments)\n\nI\u0027m not real happy with how much s3api is bleeding into the proxy app, but I also know there are other times I wish the proxy was a bit less opaque with what happened on the backend. I\u0027m not sure what the right generalized thing to do would be, though. Maybe push a new key into the request env like swift.backend_response_headers that\u0027s a list of what all we got? :-/","accounts_in_message":[],"_revision_number":1},{"id":"be46d6d1862f3ebbeb351d11362c7b4bb438bd3e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2021-12-17 18:21:06.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"2cafcb57888cc9119baabcb9f23097fa6b8f23c2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-17 19:26:43.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eae60bc5f7bd4876986d914d88fc259a : SUCCESS in 9m 31s\n- grenade https://zuul.opendev.org/t/openstack/build/3a2f25077b8341cc8fa8c3b7205452e7 : SUCCESS in 57m 52s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/3646f3ba80da45749ad3ff61fb9df3e2 : SUCCESS in 46m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/880c79fc97f944c7a80d76a20639ad8e : SUCCESS in 48m 33s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/6ac5718173834709a4b9c52ecee0d4f4 : SUCCESS in 8m 02s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/f8fa347bc47b40f4969c84dba58efc87 : SUCCESS in 5m 51s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/be53125e52cd46e08e8cfd22f43b9867 : SUCCESS in 17m 51s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/365cd18a5cb845bca408665142665365 : SUCCESS in 16m 14s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/bb4859bc794c4865a543ebd16bd304e5 : SUCCESS in 19m 00s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/3bdf5d475bf44d5ba6b7c13711ce7f3f : SUCCESS in 17m 29s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/5ccc65e3dab24e33b9ced2ce2093eaf8 : SUCCESS in 18m 43s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/aa18ac417f57490383e4023c1a0633ad : SUCCESS in 31m 54s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/75c59e9e728148f88297781a643ab5bc : SUCCESS in 31m 09s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/3bf9120bcaee41ae8b5a191eaa19de54 : SUCCESS in 21m 15s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/058bc51a39f14820833252b38df80822 : SUCCESS in 12m 41s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/82c580cc6fe546bf93035ed3ee9340d8 : SUCCESS in 30m 39s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/454a80ebf9b745d4b45684f54ac076ab : SUCCESS in 37m 11s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/338aace4fdb94bb3afe31b9629a276e1 : SUCCESS in 7m 00s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/090281417f39484dbf6c4cd565001621 : SUCCESS in 22m 01s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/6214877bf39540bcb544149a236bf21b : SUCCESS in 57m 43s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/564f305099b34aca9329a1bbcf4ed44b : SUCCESS in 1h 02m 23s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/7767128d14ac46b29f03b3aa729c7836 : SUCCESS in 37m 01s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/03182156c408476d9001723629dcb25d : SUCCESS in 43m 02s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/16d0ee46dfec46f0a7240f3b85ccb217 : SUCCESS in 19m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/295fa92b3bd24ced9c57b686897eb5e2 : SUCCESS in 5m 51s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/407a59bc7e3e44b4bf79f40d885d29fa : SUCCESS in 24m 37s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/04e7546bc8724458b51833f73d62f8c6 : SUCCESS in 45m 21s","accounts_in_message":[],"_revision_number":2},{"id":"8cb6eafcc33bd1aef7018118f95096a008661953","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-17 20:00:56.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f8a1b6f94c744baaa552fe5a7683d9ea : SUCCESS in 36m 45s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1581a926888f4ac192e39358ab8e2e1c : SUCCESS in 36m 44s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/c299f02ea1724513906187dff4a5a855 : SUCCESS in 1h 31m 13s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/732c461c07e04fa49b3912e4e87e2a0a : SUCCESS in 46m 40s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/23c178c4e1a4435bacef604f528b70c7 : SUCCESS in 31m 23s","accounts_in_message":[],"_revision_number":2},{"id":"596230ef908816610b2378061a92e16dde7b1593","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2021-12-20 21:06:55.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"a18c299dfa901970ee5c5edf21a7fed3bdf71a89","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-20 22:44:56.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b02eaeedcd7743da85101acde946993f : FAILURE in 44m 16s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/bced3de977da43e094ca0774f6551aca : TIMED_OUT in 1h 31m 08s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/da7c0418cd554760aed5bddd37180ec7 : RETRY_LIMIT in 11m 13s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/320c950c420f4b61b045d8efada3f81f : SUCCESS in 38m 20s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/bb294583283d49d48886f43246f52d21 : SUCCESS in 50m 57s","accounts_in_message":[],"_revision_number":3},{"id":"463b38b5c039bb44666b092a45f8730f96d8bbbf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-20 22:45:02.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/32906f80c71442528404b4ffda094ee3 : SUCCESS in 9m 16s\n- grenade https://zuul.opendev.org/t/openstack/build/0a4564dd19b94d53b871db4da224453e : SUCCESS in 1h 01m 07s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/63f7b9fe05444927b51df0a4279ac62b : SUCCESS in 52m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/42db36fc1e464b7094a433aa97d6130c : SUCCESS in 54m 21s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/5331ea01935b4666a185f88fb43a6761 : SUCCESS in 8m 21s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/5d6e49b9bc3046de812994c7407f2f18 : SUCCESS in 6m 01s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/882d2b96db984ce7bfb9dde2caf691c5 : FAILURE in 20m 20s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/b42db1707a3b4c308026d31b7b7a03cc : TIMED_OUT in 1h 31m 26s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/ddaa9691aa804f01bf4ae60924893488 : FAILURE in 17m 32s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/adc56695a2b14094b6c8189dc4d9b1d4 : FAILURE in 16m 59s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/eba3355c2aba4e538f40e5878207ae84 : TIMED_OUT in 1h 31m 13s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/79b409f09fc94d199aa428c4f3c3fd8a : SUCCESS in 24m 04s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/36f8fd4c2c7e47a6b261cbf2d49e784a : SUCCESS in 21m 32s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/d84e36d2324a45aa947a75857271dcb5 : SUCCESS in 21m 42s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/2f2e86f113be4a2baeb2dad5f0b522c5 : SUCCESS in 20m 39s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/ddc2b44a843349ccb55d37f8e77fd930 : SUCCESS in 19m 25s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/585b94c390e3426ca9df868f1484819c : SUCCESS in 21m 14s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/be2b3ab1a28f4135bb84e375c2f8199c : SUCCESS in 6m 57s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/61bcdd04c1ec48478b9ea65911b7d2ed : SUCCESS in 25m 31s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/20b7e9afc86b4421b8f028b9651deaec : SUCCESS in 57m 51s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/246bd96184fe4a04a94900b7c261c90e : RETRY_LIMIT in 5m 38s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/985ecef7ef134e8987f6c57631a6b9ac : SUCCESS in 41m 01s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/2592e34696a84a658e92f5be7b47982f : SUCCESS in 41m 21s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/63695f4dc75e421086591bcbab43f966 : FAILURE in 16m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4bd0bb9759524ef7bf1a45a254615e2b : SUCCESS in 6m 15s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e85ab5824c9648c3b1b50a050c241a7a : SUCCESS in 28m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/97d808948b5f491996bbeb89f0ec89f3 : SUCCESS in 58m 51s","accounts_in_message":[],"_revision_number":3},{"id":"0613730445c411ec691d5a241502c2603ecc9b86","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2021-12-21 02:30:44.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"a32fc5d24cf537deb563a2c42faccbf5d4e79991","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-21 03:56:52.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/662770ca00294ecaba5645b723b7956b : SUCCESS in 11m 49s\n- grenade https://zuul.opendev.org/t/openstack/build/2fc74725c1ed47fcb417d416cf1748ac : SUCCESS in 51m 18s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/0859348668134dffa943a6a1b7113c1d : SUCCESS in 36m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/32980d32700a4fa0a5f6124c870d4917 : SUCCESS in 38m 19s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/3008813ef13f4baab645575363500444 : SUCCESS in 9m 26s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/68edfa4519f74aac9db23a5473c04a66 : SUCCESS in 7m 53s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/6242e837b494441ab2edee40ef94598c : FAILURE in 17m 04s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/b0b3efd2c2474d21ae50b5dfbfb0af52 : FAILURE in 21m 48s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/6aad8f5002b44efa8d872ec9112a7d1b : FAILURE in 18m 15s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/9be1e99058b84d1aa5ed3352a160350d : FAILURE in 17m 38s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/15c7664533fb47fb9818df80b2e5d99e : FAILURE in 18m 18s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/2f45a490c65540e09b9d74f8a6bede93 : SUCCESS in 22m 35s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/f4667a9e33cd44c68f2fcf9aeaf36283 : SUCCESS in 18m 15s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/d712922a907a4f2c838eb40ab6f7cf4e : SUCCESS in 18m 39s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/4d985df34e734a1e94f2fac9f2584203 : SUCCESS in 24m 42s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/c0f3c7cdbf9f4d42a4fe7486582b8324 : SUCCESS in 24m 37s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/c3ce715bbe5d4f6a824a3ee28ec61db8 : SUCCESS in 23m 33s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/661bb92e086b444b8eb39db766464742 : SUCCESS in 8m 08s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e1dd1ef82314406487accc12b84f338d : SUCCESS in 21m 48s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/debb971cf9874690aeea5ecf260ce7c3 : SUCCESS in 51m 27s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/c27326b5c35b4d429b632cd72978ce13 : FAILURE in 51m 51s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/981d4b76a64b47aaa37ca8e4baedb81b : SUCCESS in 47m 24s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/27f81696d1f04d888f895dce12e53bc7 : SUCCESS in 47m 21s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/05d755e4645d43e9b0a28c1e1fa8c62d : FAILURE in 21m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b87f6476565049e285db6b34d88796b5 : SUCCESS in 8m 45s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/faa2517cf0df4456b9e5187fc9d66ab0 : SUCCESS in 23m 13s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7108791e42f74770bae4fcd95e8d5a18 : SUCCESS in 53m 36s","accounts_in_message":[],"_revision_number":4},{"id":"644da6a8bbd26faae7dcb47aea422343d72fd5ce","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-21 04:46:31.000000000","message":"Patch Set 4:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5b6ab4ec30f3411dbd983b81146e6710 : FAILURE in 59m 34s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/30ccf7ed1f4a48198f99ec45c46784fe : TIMED_OUT in 1h 31m 52s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/d5d52dc3a5ae40ee9ec5a5f400e7aa0f : FAILURE in 1h 50m 06s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/cbd6e7ede77c4e248cdfc19eaa06e1c0 : SUCCESS in 1h 00m 37s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/6bf95a6cb1044c53a0ecde0de446c266 : SUCCESS in 59m 44s","accounts_in_message":[],"_revision_number":4},{"id":"1185ebca1da8b3a2048fb0eeb09f083ca3f8e93b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2021-12-21 16:59:53.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"62fe00b05e672034a29dfd261414529397eeb18c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-21 18:41:41.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/26b23c7eec3c4ff7b33ecb102b090c2f : SUCCESS in 8m 34s\n- grenade https://zuul.opendev.org/t/openstack/build/121a54a488b24e87b7755bc17ad5aff2 : SUCCESS in 1h 05m 56s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/ebe3b707f0c84c259258b77bf338bbae : SUCCESS in 48m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/46e04e6bdfd941dbbbe4d6b7b739493d : SUCCESS in 54m 18s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/816c925fdd174af8a18c2583afe62b8f : SUCCESS in 7m 53s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/3571f1a128cc4983b9b57813b838e968 : SUCCESS in 7m 17s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/47f7f611a91a41e9a8b0448a4d490d85 : FAILURE in 16m 24s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/59a330075dac4f0eb8bc816e69bc1343 : FAILURE in 17m 20s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/04e977fa23a34ec7b01cc53a56694986 : FAILURE in 15m 54s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/c5d4d866502c47738a6d491a8b154889 : FAILURE in 15m 37s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/fad3af02db4244f585cc3ca1b5cd349c : TIMED_OUT in 1h 31m 25s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/3c43e3c3e4eb4eb58a1b7b0a4b4820d1 : SUCCESS in 19m 55s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/ac128865442442ea911edbd29cc65282 : SUCCESS in 23m 41s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/fbf43814727040ba936b58c126a9a69f : SUCCESS in 47m 52s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/fed7ab88aba646b4b4f8bbffe50b2e21 : SUCCESS in 17m 39s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/544e9c856f7c4bdebfd1c9d5d3bc2ae9 : SUCCESS in 15m 37s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/a33dbf4a017348a5b0db7539da3d9615 : SUCCESS in 20m 44s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/fbdf0c05f11d4920be6a84fc792d9e7c : SUCCESS in 6m 49s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/571d9dec4a964bd7a60bee954d58339a : SUCCESS in 22m 49s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/35241a67e83544759e213f50a82c73e4 : SUCCESS in 55m 38s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/afa7ca9940d54a5fbe3059b4544275c9 : SUCCESS in 1h 16m 13s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/5639f7a94614468fbc5224e93c6ba229 : SUCCESS in 43m 07s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f9da6720c1cc4993a590074c137aaf4e : SUCCESS in 45m 01s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7891ec2d4bdc4a228b922c688e05911c : FAILURE in 18m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/03827edbf8c8440791d0dd847afb20e3 : SUCCESS in 5m 53s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6301ecc7104e449fba8256cb4b6e7a1e : SUCCESS in 19m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3d4733f3d51141ac848ab1c4eafd04c7 : SUCCESS in 1h 03m 40s","accounts_in_message":[],"_revision_number":5},{"id":"f333e9a2aee6b1895cf31e47367ace918df6aeb4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-21 18:46:34.000000000","message":"Patch Set 5:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3da29f1181364d64b151cf5f5cfb3dd8 : FAILURE in 43m 45s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6172c3b4f7fb4415a6cd05bbae1f780a : FAILURE in 39m 31s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/3eeee464b16040399e44274ea0609b40 : SUCCESS in 1h 33m 47s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/5b594cdd3cfa4fb0879c21cee1c91b75 : SUCCESS in 40m 12s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/b8dc2b47662a44148384e65335f27889 : TIMED_OUT in 1h 32m 08s","accounts_in_message":[],"_revision_number":5},{"id":"db34c487c171aaecaa4cdba105e3985458ddcc8f","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2021-12-21 22:56:28.000000000","message":"Patch Set 5:\n\n(7 comments)","accounts_in_message":[],"_revision_number":5},{"id":"c07149c2000581a0a9bebd9fcc74ba3f94cd2c57","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2021-12-22 06:32:26.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"f0e5cb531eb8cb29d551aa987dcb37d51511150a","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-12-22 11:16:08.000000000","message":"Patch Set 5:\n\n(5 comments)","accounts_in_message":[],"_revision_number":5},{"id":"a50987c8b76e31409557445e168a1b4736863cca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2021-12-22 17:17:32.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"065ab846bb8995aa8769cbb6a8a57319e340d191","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-22 18:19:50.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4567c151d7114ce6aa7eee1750ab6240 : SUCCESS in 11m 26s\n- grenade https://zuul.opendev.org/t/openstack/build/fdcd078989e24e27aad2fc306ca669b3 : SUCCESS in 55m 36s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/ef0a3ea2ddb842c58014ac26c4042d0a : SUCCESS in 49m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/53cb2e1ae21546b4b52dc5566ee7e515 : SUCCESS in 43m 03s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/6542a60c4e8649a0be162669bf0473c1 : SUCCESS in 9m 01s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/262a75d190314a78be7ab0e845950264 : SUCCESS in 7m 32s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/fa31d764825f4c6e9efbfacf31a89f6c : SUCCESS in 21m 22s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/fe68d89727ca488e85a1eaaab02acd4e : SUCCESS in 18m 01s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/95f0f0b21ffc47b0b0ebd4c12eca3fc8 : SUCCESS in 18m 52s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/b15ab96d4c19453faf52e1b521d1cd35 : SUCCESS in 18m 50s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/aaeef1514ce74fb6ace4973709097fb9 : SUCCESS in 24m 07s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/8cd8da6efe3e4bfb97c88c5fd7c65c33 : SUCCESS in 23m 12s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/d26f5c36852440369b01262023d8f5bd : SUCCESS in 22m 05s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/ace226aff8a942748bc000989afc762a : SUCCESS in 22m 35s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/c6e2443525854f8cb9f91dcabe266087 : SUCCESS in 21m 34s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b166c08807ad476a89ce7291d19f7035 : SUCCESS in 7m 34s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/aea4a128976b415ba10f24747feff3da : SUCCESS in 27m 50s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/58da0be749bd4612bf577fd1f4412c58 : SUCCESS in 53m 52s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/1aa1bc5c6a1c4f22a0bc3ddca49e7524 : SUCCESS in 58m 14s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/51571b3c58554b7d9af3cf3e36809e92 : SUCCESS in 43m 48s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a4e941a81eb44c5e9eb09cec9fc01a08 : SUCCESS in 41m 10s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a5f7644a135e437888df0ff93c970a29 : SUCCESS in 18m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/94439cab06fe4c8c9618d42acdbb651c : SUCCESS in 7m 48s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/fababa517a0a4f139169320af3f854ea : SUCCESS in 30m 09s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8afe9a05160141f7ac60e427ccb5aecf : SUCCESS in 47m 42s","accounts_in_message":[],"_revision_number":6},{"id":"05705ba5085965250392ee7c0fb3163de265bd43","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-22 19:28:31.000000000","message":"Patch Set 6:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/94e85e98ef1d49e391a7475b87dbb438 : SUCCESS in 51m 59s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9685af060f3e43598cc0282fe7035606 : SUCCESS in 47m 54s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/06cce5ff799a44ec9024164606b72ee8 : FAILURE in 2h 08m 25s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/1e771374a407469ea8b79a4d7948476a : SUCCESS in 44m 49s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/b7de01585b8049db8d6955f8e65c521c : SUCCESS in 47m 16s","accounts_in_message":[],"_revision_number":6},{"id":"0022cda8871cc4370f241452743f1d19d6cec369","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2022-01-06 22:16:28.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"72fec77f1ce1f87ab55f97ddc21a1a785014e080","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-06 23:18:42.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/78314da2cb444df186034d3d9bcff33c : SUCCESS in 8m 32s\n- grenade https://zuul.opendev.org/t/openstack/build/0b14d8e2a5b54b319cbf91bdba7aa228 : SUCCESS in 58m 58s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/b929c6ca31fe4703a6363591461ad467 : SUCCESS in 52m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3b379549cd6d4a97ae3972cb0a58b216 : SUCCESS in 49m 35s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/53ff66a91da94223af58997d883c36c5 : SUCCESS in 8m 14s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/78cbdbba4d7042db827ba460ce5ad942 : SUCCESS in 5m 47s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/5123dbcd931343ed92be4672d32f236c : SUCCESS in 15m 40s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/91c4127b58b44073818b3505e5c57253 : FAILURE in 14m 39s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/b619b77cc82c4de3872e17b09ee2c9fd : FAILURE in 17m 05s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/a4790e8c5c73416c87c9518777f51402 : SUCCESS in 17m 40s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/3882407f6b8b4ec3b88649b5761cbe1a : SUCCESS in 18m 35s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/6228cde9a5044e20ae6519d84779a5f6 : SUCCESS in 18m 45s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/9109f979a9c246c1af56932680eaf0e9 : SUCCESS in 17m 16s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/9639dfaaf70e46bca9bb0d377f7d09f4 : SUCCESS in 24m 58s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/512c6497f27f495b91d453912e6e4c46 : SUCCESS in 17m 06s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b68faeae883b42a0904efda4c23a0665 : SUCCESS in 6m 52s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/3ad2d01b80544d56a5a022ca06518d97 : SUCCESS in 21m 37s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/148f1da25de24b9eaccb738ffa00a899 : SUCCESS in 46m 34s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/67fb543de7334a20bd2596491cdfba51 : SUCCESS in 57m 43s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/1716fdb84a794a76aa508ba4de497e30 : SUCCESS in 41m 35s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/2bbb0823a0864a1183e8e3e1a53df3fd : SUCCESS in 36m 50s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/393cf45fb9d34987ad0d35c1a717f01c : SUCCESS in 17m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ffff566dc3444408a688a563c396fdfa : SUCCESS in 6m 33s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/0ce83be61af04ddea2c3b2a83c4a7479 : SUCCESS in 28m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1d05778367eb416593ce305fcabe7eb7 : SUCCESS in 48m 28s","accounts_in_message":[],"_revision_number":7},{"id":"055ff4e8720b16fc252afb7b046b3f2673f198eb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-07 00:19:45.000000000","message":"Patch Set 7:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/65552aced92c4cc59a435ac01a5f8da6 : FAILURE in 49m 30s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1263598a4117499fbc86d126cf942746 : FAILURE in 51m 22s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/71774350102e4118ad8ef1f3e31e7921 : FAILURE in 2h 00m 19s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/d04e0782c8324341995dfe46ef63ed32 : SUCCESS in 59m 31s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/6ff4d0f6d694473da416a54089275dea : SUCCESS in 29m 33s","accounts_in_message":[],"_revision_number":7},{"id":"08b432a77e28f7ba231ee0da9482b0790b33416e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2022-01-07 22:11:31.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"4cc4c16e1f31c11be229ffe35aaf2d1df6f6551c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-07 23:48:08.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d069313b4f64128880a7c05baa364c7 : SUCCESS in 9m 53s\n- grenade https://zuul.opendev.org/t/openstack/build/add594dbb2eb4f62b3ef4a0750b934ad : SUCCESS in 56m 47s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/4abc0d26a309470796b6574ccd9f0b70 : SUCCESS in 47m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3758bb9e46e446fa860f6d17ec1b5dd3 : SUCCESS in 39m 51s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d1be866d42074bf687e224435217bf77 : SUCCESS in 8m 28s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/471d876ac259459489a4c5f82a751490 : SUCCESS in 6m 41s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/68a5c85e592d4a8cb68c52daa9da4c24 : FAILURE in 17m 22s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/0109672fc3914d10a51ef0437bab86fd : FAILURE in 22m 37s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/ae2f3c0b813043e2be7dc976006b419f : TIMED_OUT in 1h 31m 25s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/4f220c2b954e4e5a90957cfdaa1feca5 : SUCCESS in 23m 55s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/4f14f240668b4570a99ebc4ab7bfde1c : SUCCESS in 21m 09s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/a03b22a7875a4a0c87b87b7aac706aed : SUCCESS in 22m 27s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/25b2261466d0447ca7f79d71380b088c : SUCCESS in 30m 27s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/3585c0016dfd4140b3c72591a66cf65c : SUCCESS in 27m 06s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/631c5b31cb3540dfbbf90754eead698a : SUCCESS in 23m 26s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/73fa41b04892400d96f91f37dcfc184b : SUCCESS in 7m 54s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f6ea0e62ab494883b461f2882e726ae0 : SUCCESS in 29m 19s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/c8711d076169420b954c4a6eece48dd5 : FAILURE in 56m 45s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/bffc246fdcbf4ccd9de46d7d0168860e : FAILURE in 56m 13s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b7f870472305459cbd91e77cc394dfe9 : SUCCESS in 42m 14s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e90ead7aa7144eb6af9efa7f5db48e39 : SUCCESS in 43m 08s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/68be0fa281ec44828e33da6c0c862080 : FAILURE in 17m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f9e5b3958ed94fab873a98bf2bf09122 : SUCCESS in 6m 40s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/1ba02ac584db452b9bcd8b8258720c28 : SUCCESS in 28m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/afca3142b6944e8eb51ca01a640492bc : SUCCESS in 48m 55s","accounts_in_message":[],"_revision_number":8},{"id":"5a215d3e6434766f85320b108ac4f3d95f954925","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-08 01:34:01.000000000","message":"Patch Set 8:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/deb8111f75e647ab97ed82495295a46e : FAILURE in 42m 33s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5c279218404c4836a29ca634ce005dd8 : TIMED_OUT in 1h 31m 24s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/83bfc9f361964939abc999d72038bfe6 : FAILURE in 2h 02m 09s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/8aa33404dede44b6bc8745adc9051087 : SUCCESS in 1h 14m 08s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/f2cca5c6ec194d288fc2912a7573faa8 : SUCCESS in 1h 04m 54s","accounts_in_message":[],"_revision_number":8},{"id":"990fc81e9b23c543dd9bedd5cc8bcead16058b4e","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-01-10 11:22:58.000000000","message":"Patch Set 8: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":8},{"id":"c9f4264667243283be27a67274a2458bfe24f355","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2022-01-10 16:43:26.000000000","message":"Patch Set 8: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":8},{"id":"c45f1c9e919933c428ee3af25b8de29cf129f3b4","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-01-10 18:22:09.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"97f4140ff05b79189c0da60770184f935f0b4352","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2022-01-11 20:40:16.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"638d351a698e4014097119f3367fa29c0a063e3e","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2022-01-11 20:45:35.000000000","message":"Patch Set 9: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"cf2b36e8e4302463b72c98cef2e1987abe6000ca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-11 21:44:42.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c98718ddaa3426880c898a92ca89a49 : SUCCESS in 10m 52s\n- grenade https://zuul.opendev.org/t/openstack/build/acfdd148315643f18ee34895140cd8d1 : SUCCESS in 1h 00m 24s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/9beda70107b74342b4586f2c2b83dacd : SUCCESS in 46m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d6986f31c4e24a8c93cba9d96909ca0f : SUCCESS in 45m 18s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/49cc0f9501714e6a8e7b5f080b676c2e : SUCCESS in 8m 14s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/bad1db3e68de465cbe613690c998a193 : SUCCESS in 8m 02s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/beac5e204ded4afb8cb4735cd3268240 : SUCCESS in 21m 36s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/8c0e42c2f73a46c6b1c4c06caa6fc1c1 : SUCCESS in 19m 01s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/61beeb05923f459994538aab35f3304d : SUCCESS in 16m 21s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/daa0ef194662408f8bea6c969f119946 : SUCCESS in 21m 00s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/3a18bb517594496b9e13c5f095e03bca : SUCCESS in 19m 42s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/f39b23dadf2645c78c8d8f9c253a1c16 : SUCCESS in 26m 45s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/001a455b02e9413e8879671ec62251b2 : SUCCESS in 17m 03s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/f90daa01b06b43ebaccea22903297e0f : SUCCESS in 19m 30s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/55681ee4588947e3a33d3e52f0dc21fe : SUCCESS in 19m 23s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/17995eef3b6a4a98a22a406e24e167e1 : SUCCESS in 8m 23s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b29a790caf8b472aad1dafb2f2c2347e : SUCCESS in 26m 43s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/691c229be9d1431cbb4d41707397fd74 : SUCCESS in 56m 57s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/584b0553f8494795a42edc7521b902f6 : SUCCESS in 57m 40s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/6e33dc7bafcb48d6ab7618ec1b663a20 : SUCCESS in 43m 01s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ff75b79d32b04cb498e4c012757ac997 : SUCCESS in 38m 30s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3ab4f8d6ebfb47aab866cf8540e752a5 : SUCCESS in 17m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ceb0e3341dc8482682a3a8733bacf207 : SUCCESS in 6m 46s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/37a1582957b747fba9b375369cc4ad09 : SUCCESS in 26m 48s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/88ac306056a2426b868d6ec18b025b08 : SUCCESS in 47m 21s","accounts_in_message":[],"_revision_number":9},{"id":"9cf89523b86da8896fa951d8bd733489671c4ff6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-11 22:31:29.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f299bfae2c284416b73b4e74f0e49b7c : SUCCESS in 31m 59s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/91031aafe7974dee923997d51c3ecd74 : SUCCESS in 37m 30s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/52964992126842e784565fdcbec44473 : SUCCESS in 1h 48m 12s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/95526e90ade441d085297e625c309081 : SUCCESS in 58m 21s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/1e4ade255aed46c991fe13d4c575e3a0 : SUCCESS in 51m 41s","accounts_in_message":[],"_revision_number":9},{"id":"e6628b0bbea2d2dab74204b8404d123534ebbfe6","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2022-01-11 23:06:41.000000000","message":"Patch Set 9: Code-Review+1\n\n(4 comments)","accounts_in_message":[],"_revision_number":9},{"id":"4d91fc0634159dd6f7280e8c041bfef8ad1732ae","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-01-12 15:20:12.000000000","message":"Patch Set 9: Code-Review-1\n\n(10 comments)","accounts_in_message":[],"_revision_number":9},{"id":"82174345e79114195880bd5c2d33eeb12fc638c3","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2022-01-13 06:12:25.000000000","message":"Patch Set 9: -Code-Review Workflow-1\n\n(15 comments)","accounts_in_message":[],"_revision_number":9},{"id":"34a2db67478de61fe78d3de2310fef4f0d51ec08","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2022-01-13 06:13:03.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"9ffd88414aa356a28ac9dcabf5ffd46374302fbe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-13 07:25:49.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f51553e80c754d77ad1cb1c0f0cafe80 : SUCCESS in 9m 24s\n- grenade https://zuul.opendev.org/t/openstack/build/4010536d04a0440a9f912dd127e7d99b : SUCCESS in 51m 09s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/0636f3b9932b4b60820ff2635b340b7e : SUCCESS in 41m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9c0ecdf4e5884be1bed8c331923a0026 : SUCCESS in 41m 33s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/3a309b9ed52e4e8981df018e4120aeb0 : SUCCESS in 7m 22s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/3997804137774d2ebb79bcace25e0e54 : SUCCESS in 5m 39s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/9cc1916d1f914546afbe06938fbeeb57 : FAILURE in 15m 21s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d7dda64d1c574d59b363226929a035b7 : FAILURE in 22m 19s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/c60df34d25f44790b4e430903bbcbdba : FAILURE in 16m 15s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/c419decfe8ed45d0a14cb85710f6f274 : SUCCESS in 24m 38s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/8928ba1da2a94a4a905dc1b65173c243 : SUCCESS in 20m 49s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/fae15fe434724546b4c90a5516238068 : SUCCESS in 25m 33s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/ef3a4d7944d64002b6489c9e40d1ecd3 : SUCCESS in 14m 49s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/59ab2da13cc94fd39954ef6631125e8f : SUCCESS in 22m 34s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/c5c042c75acd4f98a1c6a287eb4b7c9e : SUCCESS in 24m 31s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/02b802b3732547438544fd27688d335b : SUCCESS in 6m 41s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b152184481d04180ba695ff25580066c : SUCCESS in 27m 50s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/422a45a5344244818b5a811894a80812 : SUCCESS in 58m 48s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/5efba07c2e4a4c0993763db7e4e6cd3a : SUCCESS in 1h 01m 34s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/5ffa7954d068449395b5e2be3b5b8752 : SUCCESS in 41m 10s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/71ae96d392474afcaca62c03361b3802 : SUCCESS in 36m 54s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0d4879dce96545a3947ca0ddad8b9523 : FAILURE in 22m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c28b2a52deb64ec59062f19dfb80c971 : SUCCESS in 6m 16s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/168b4b8da782498b93acb021188393d0 : FAILURE in 23m 08s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1279bdd1cc0048ea8e670b2775bdf506 : SUCCESS in 1h 06m 23s","accounts_in_message":[],"_revision_number":10},{"id":"c7f1a4b91636e0eeff1ea7d145ee37ec8fd45a5d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-13 07:41:36.000000000","message":"Patch Set 10:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7e4f9cdac5f84967b59c3b27296da826 : FAILURE in 35m 17s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/31120d479bf444599672e94366f0a649 : FAILURE in 27m 57s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/b44a480cc83e4798b3dbb3aec1f71ff7 : SUCCESS in 1h 22m 15s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/286c3387ceaf40b7b088da86acd611a2 : SUCCESS in 29m 03s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/5cdf38c9b8d7492294d1c3696df0a81c : SUCCESS in 28m 18s","accounts_in_message":[],"_revision_number":10},{"id":"b96c188205c90e07552c474006a924584a5171ee","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-01-13 13:00:18.000000000","message":"Patch Set 10: Code-Review+1\n\n(6 comments)","accounts_in_message":[],"_revision_number":10},{"id":"5a89fda07e5c03104f631bc15434e6a157c71661","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2022-01-13 20:13:06.000000000","message":"Uploaded patch set 11: Commit message was updated.","accounts_in_message":[],"_revision_number":11},{"id":"4f9bef78bb87bb5054b8cecb858147caaae81c94","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-13 21:18:49.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/282f2526124c42299efe25779532a272 : SUCCESS in 8m 34s\n- grenade https://zuul.opendev.org/t/openstack/build/a88406a3118448579899bd0acd8ce17f : SUCCESS in 52m 03s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/83fc6ebd3b5141c280f63cc612b9e04d : SUCCESS in 47m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fbc0400d753a443888d71b3e9601e5b2 : SUCCESS in 46m 51s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b98b67fc5b3e4ec09803d60cc1235365 : SUCCESS in 8m 20s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/d15ebaf568fa4cf387cf74946d8f3805 : SUCCESS in 6m 40s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/befec7249435433bae92b2f2d7af3b65 : FAILURE in 17m 55s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/aa4bef221c2e4a98ba209a8e6f206aa6 : FAILURE in 18m 03s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/d14d4c9334534aaaa99089b1a9326649 : FAILURE in 21m 43s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/c2b20324500d4e128d4f5a45b1e9cccf : SUCCESS in 23m 58s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/c9c81480bac44c3491cf87f106be85ee : SUCCESS in 23m 11s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/6c69b7294aec4056a81e73091df099e1 : SUCCESS in 19m 59s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/e0ecaf163d534d0e8de272a6f7f7dc06 : SUCCESS in 20m 25s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/74fc86e62a004365a6b5bdb2eace0d59 : SUCCESS in 24m 33s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/6a6d6a27116747ec99d2ac27104e661e : SUCCESS in 21m 01s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/92d0f5c9cfa448ff82c818bad4933469 : SUCCESS in 8m 36s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/6fcc50f78ead43afa5d3c45c5de01cbf : SUCCESS in 27m 18s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/8a40239b05c44b18bb8359267856a9d4 : SUCCESS in 1h 00m 16s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/85b151167c854a1cb0c4dbde0179d96c : SUCCESS in 57m 37s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/514c239421ae4e98b7bbfb00e6d270cc : SUCCESS in 41m 40s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/aac528d1abf44cdfb8b3826bfb50d208 : SUCCESS in 41m 22s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c673b33d04c84221b86477243ff741f9 : FAILURE in 19m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/621bfb707fe24264aafbc389c0d89aa0 : SUCCESS in 6m 41s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/5e0ac3f17d9d4c3da36fd9fbe4625b8a : SUCCESS in 22m 38s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f8537d658105469eb5fd3d69496a1f0c : SUCCESS in 53m 41s","accounts_in_message":[],"_revision_number":11},{"id":"e7eb36c7067756fff894cefee95e5df4fb7c6237","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-13 23:19:25.000000000","message":"Patch Set 11:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/93779b5432bb46a4986a67aba13e0efa : FAILURE in 31m 03s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4fbe5e3c604941cc9a666c8544aa1282 : TIMED_OUT in 1h 31m 30s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/10758ec4a0fa4243bf928a7742a79be1 : TIMED_OUT in 3h 01m 07s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/a87ee7cbef064baea613eba2e483f9a2 : SUCCESS in 32m 09s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/00b39eec75274621911bf4577eb8447c : SUCCESS in 37m 45s","accounts_in_message":[],"_revision_number":11},{"id":"24733f6095c319e978d99292dd420034cab953de","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2022-01-14 13:46:26.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"a78407917fe1c03feac92c5f0e473ec9f8babc90","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-14 14:55:49.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/205f3fad75c74c6887c0ce8d7f3fe4ad : SUCCESS in 9m 48s\n- grenade https://zuul.opendev.org/t/openstack/build/47d19a7c4c9f4c618f47b2f47dc74e66 : SUCCESS in 59m 24s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/3ab07dd40104439a9f4af430f950f01d : SUCCESS in 44m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/af6e367df34549beb9df1f4e5f7abd91 : SUCCESS in 49m 15s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/aa95f5232e574ea088ba7b5a3367e8bf : SUCCESS in 8m 42s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/8ac02a22eca94fa4ad0a5eb83d67d5c5 : SUCCESS in 7m 46s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/55eb32451e2f4c6cae7fff807345cf59 : SUCCESS in 17m 57s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/b5cf2a8d05904810a7b450d3b0aaf35a : FAILURE in 18m 15s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/69da2b184c2a4ba28187f956b9c5d908 : FAILURE in 16m 15s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/1bb145409116435fafb7cb0438df58b9 : SUCCESS in 23m 15s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/dfb86628ef88492ea59fee19dfdae6a6 : SUCCESS in 23m 06s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/fe1e943697ef4b3a93ccb00eb3c947b7 : SUCCESS in 29m 58s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/81ac0a086f8d4a3f8a7f1571e18453e5 : SUCCESS in 22m 24s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/b1421fca3def41fbb35bf3a3d5563661 : SUCCESS in 22m 32s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/cd5de02ca93c4e4cbe5ca71025559e76 : SUCCESS in 21m 40s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/96256c5bfe424c6c8b1024d2a3cd7d0c : SUCCESS in 6m 38s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/61cafe3f2a0547ef92a2c106f11e445b : SUCCESS in 26m 38s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/ac5b2f9d74c7492390fd6e361cc95c1b : SUCCESS in 50m 57s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/24a7a8046df44dea82ab66bbb7a21a1b : SUCCESS in 1h 04m 00s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/90e8cf4f42564338a5003f8080a6c55a : SUCCESS in 40m 03s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/313eba01494646e69ec7de132509af7b : SUCCESS in 39m 21s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d615433661964a0aac8da24f651531a7 : SUCCESS in 17m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/be06a177787346edaa75825875b50067 : SUCCESS in 6m 47s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ed3fc050afaf411e83e68f91afa8ecb3 : SUCCESS in 21m 46s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/43fc542c014a4c529c318f6484a1a15a : SUCCESS in 49m 47s","accounts_in_message":[],"_revision_number":12},{"id":"6269f48c6066e43ad0fce5c39a2935fde5a0213e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2022-01-14 15:03:32.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"4fd59205ab14f628cdab73a6438eef04a25ff1dc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-14 16:15:33.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/005e302f713642be993d448df3284d74 : SUCCESS in 12m 25s\n- grenade https://zuul.opendev.org/t/openstack/build/362ad785871e46d08a7e8bcc545e6e12 : SUCCESS in 1h 09m 10s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/1752fc9dcce14cf89bc828b11272303c : SUCCESS in 43m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/49b2be4ff72c412093717a038a9b27a2 : SUCCESS in 49m 01s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/4025f8f5552a49f8b38036dc3ac02124 : SUCCESS in 8m 08s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/38fe1a40e2d64c4880b19d7d2895574a : SUCCESS in 7m 59s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/8113128058844b3ba57cebb94bcfcac1 : SUCCESS in 19m 01s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/e9bee6621f4b4a75873cc512f4ab2ddf : SUCCESS in 19m 11s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/dd3c3525d73e4bd7bdb20ee49e68f3d8 : SUCCESS in 17m 25s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/06a2979e66574de3b73f5abc0165312d : SUCCESS in 18m 12s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/051f68c096ca45f7b256e877bfc89077 : SUCCESS in 19m 19s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/ae8e1566c553491ea7e36edc58da50ce : SUCCESS in 24m 19s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/e08b403ace9141c1b63dc18b42b7d3c6 : SUCCESS in 22m 30s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/a9831d6de2bb41e98ccc53ee4cf16478 : SUCCESS in 19m 58s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/9d710cac9ac4423bbdc7c5896667c237 : SUCCESS in 32m 06s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/ac6354b9e4b84f95aa848198f7ef317f : SUCCESS in 8m 29s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/7d9b02b25d9c4d05b2bb5d6d3df12073 : SUCCESS in 37m 51s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/e3b0df1277934c678a78d1018f106c58 : SUCCESS in 55m 16s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/b2368dbd2abd475ab09b2bab77146a4e : SUCCESS in 55m 22s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e2568d31ef0c49a087350f25c419ff4e : SUCCESS in 41m 09s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/7f3bd7849f194c17a746642357eed9e5 : SUCCESS in 49m 57s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0de3aa98daf6434b97066138018d5374 : SUCCESS in 16m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/917c9cf697ef47579625df33e535c6e3 : SUCCESS in 7m 55s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/178d10e870554cb09e7ea46c4131b1ab : SUCCESS in 23m 14s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9c9d321b6cd8457681583ae90f8a1245 : SUCCESS in 54m 45s","accounts_in_message":[],"_revision_number":13},{"id":"7e4e8208f678de4edf69529f027107fecf4055f5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-14 18:17:18.000000000","message":"Patch Set 13:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/702422f165464723a0c70f566640d146 : SUCCESS in 42m 27s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/42ce998f08dc4b49b032559cf63a9276 : SUCCESS in 53m 45s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/60ba6232941246248a0cdfe757f85071 : TIMED_OUT in 3h 01m 35s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/f8e31c08c30146b69e1c015b8f4521ce : SUCCESS in 1h 02m 42s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/5132914d724848b793afc5ebda19f502 : SUCCESS in 51m 14s","accounts_in_message":[],"_revision_number":13},{"id":"f34f646c09cd8813f1693855a1d656d9a0bf8b7a","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-01-17 11:08:56.000000000","message":"Patch Set 13: Code-Review+1","accounts_in_message":[],"_revision_number":13},{"id":"8f66374e0d244fd4d788a5496d12439be6d328c7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2022-01-28 15:57:22.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"ed8ca3d8709b0a8d2770ee94fbbc03feaeceeb96","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-28 17:06:53.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f97901eed55488bbd808cd12e2d0576 : SUCCESS in 10m 01s\n- grenade https://zuul.opendev.org/t/openstack/build/dbf8e50023b64abd8ab17b0d968d8822 : SUCCESS in 1h 02m 47s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/09496c973f244969ae94085b6f0e8536 : SUCCESS in 54m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b0b3801962f041cfb6b8700a545b3be3 : SUCCESS in 49m 40s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/433cbe486f274e4abae385e6c9adb57b : SUCCESS in 23m 03s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/690ed5633dfc425bba36fba1582e7ee0 : SUCCESS in 24m 43s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/a00c4c924cf94296939fc319a10f56c5 : SUCCESS in 21m 45s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f71b703dbce047bb927c46ca9000db04 : SUCCESS in 8m 20s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/63cd855c92154cb58e95badc523af7df : SUCCESS in 7m 26s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/41dfb581b32948ee98b9755608da44ad : SUCCESS in 17m 27s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/b1bb5e83e7fd45edb0e120afd6e8eee5 : SUCCESS in 17m 37s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/e5a481bfad864a09acff086b4ec8f943 : SUCCESS in 16m 47s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/a735f88d78c64ec491cf58601207df4c : SUCCESS in 20m 08s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/e157d886d0684fee847e542eda67c1f7 : SUCCESS in 23m 27s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/076a74fac52f4aa096b09d4e6dd34279 : SUCCESS in 22m 01s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/86aa1cdbd42d477aa3de124a2b89ec3b : SUCCESS in 19m 45s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/8cf60037808541f18d3e9254bcb49a4c : SUCCESS in 21m 35s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/fc5100da09574ceea774bbfb3dc7ad8e : SUCCESS in 17m 45s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/9642a22349db4437825d5e855379df7e : SUCCESS in 8m 15s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/18df204dad064a91803cfcfab261f33f : SUCCESS in 24m 46s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/7a41561ef9a945939da04561dbbda456 : SUCCESS in 51m 38s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/d544d10098b648088894bcb7c8c7af67 : SUCCESS in 58m 51s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/cb148726867d4d2593aa28e1c6d5640f : SUCCESS in 46m 04s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/902d5bac2e0540d280404082e38047bc : SUCCESS in 41m 43s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0a1b99b20e13426d97e69cf0ff5cfd80 : SUCCESS in 17m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1aa6d24ce53d4cb285b123658c7bca1c : SUCCESS in 7m 29s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/2c8633d24117464d948bdd78ed7dd44a : SUCCESS in 30m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/00476d91e3c740a1abe0643be6bd0da6 : SUCCESS in 54m 24s","accounts_in_message":[],"_revision_number":14},{"id":"1c1079460c2b49c3c7b58698e53631b794a9dbcb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-28 19:08:38.000000000","message":"Patch Set 14:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d07139db27c14ae7879b98ec0bf9263a : SUCCESS in 47m 28s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/bb230e5c14394dcbb29cbbdde5b1f482 : SUCCESS in 53m 43s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/7d4c802bd2a04de99aa943536fa63025 : TIMED_OUT in 3h 03m 11s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/141806c362ae4be1a8b41b3f92e09b45 : SUCCESS in 37m 02s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/ce66246c3d8b4c9a85684295e255c593 : SUCCESS in 1h 23m 16s","accounts_in_message":[],"_revision_number":14},{"id":"e2d531cf56a1730d26caf5db6e29840716c62cd9","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-02-11 17:32:22.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"91b9c6b64dadfd714b92ffc3609ab492e36c8e4c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2022-02-14 18:18:32.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"706152b0dba42613c6140ffc7d8323e0f2255041","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-14 19:16:52.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f97394d90a1741008fca00a7463d23bd : SUCCESS in 7m 04s\n- grenade https://zuul.opendev.org/t/openstack/build/b34c1e8cfb194d2c8e3311ddfba8f4f8 : SUCCESS in 42m 45s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/241a58d5af2d4762b9e04f0df5f3ed8f : SUCCESS in 38m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9f525e40708a498cbf1243db0e80a49b : SUCCESS in 32m 03s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/d842957e329340769c2e9ff827c8966f : SUCCESS in 16m 42s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/61d6aa00b1224277a601f7a5f74169f5 : SUCCESS in 14m 25s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/6a8e4ab51f7f40458a8739e438c1f7c2 : SUCCESS in 18m 33s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/82705d6154da481bae049acf697e009c : SUCCESS in 8m 23s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/828ea7b9b20141028b36d7030b349410 : SUCCESS in 5m 45s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/75160533eb3f475da61e8d5f161fa9dc : SUCCESS in 13m 14s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/4fd862170e7d43af94dbe262fc81af8c : SUCCESS in 14m 46s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/0a9c35884a754b8baaecf6cf22939211 : SUCCESS in 12m 52s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/9a8f511925d34ff9bd3a4382462b4bbc : SUCCESS in 15m 13s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/6b5aaa7cfbb545c9af0ec3e849e70bc6 : SUCCESS in 14m 54s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/ee997dfcb2ae4ab5b9a58c6b0abfa60e : SUCCESS in 16m 30s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/35c7d19e00db4fdea855079efcae3df8 : SUCCESS in 15m 30s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/9151fc1a29fb4210bfdf021970c90ba2 : SUCCESS in 13m 33s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/37e3f2818f254e7f800d3c012e2dd762 : SUCCESS in 17m 19s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/57652a18186e454590055abd46c4f2a0 : SUCCESS in 5m 49s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/139725ca98b2452598096ffc3f3fa151 : SUCCESS in 23m 24s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/e93c146a2f22471aaba87150793dd88a : SUCCESS in 45m 24s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/476f220a84e942278c6816578e53b85b : SUCCESS in 55m 41s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/67ad01f64e2445b39308f8b1ada51269 : SUCCESS in 41m 16s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/120133c5b78b46888ff17e183cfb3638 : SUCCESS in 41m 08s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c7aa0c85b0c64eb2bc1271190be33a9f : SUCCESS in 16m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8816be423f7c4063a855a5b7b53f95b4 : SUCCESS in 6m 45s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/952781f2549142b4b17885da9c3af08c : SUCCESS in 28m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ed120755f6fb46c48ec44d70f2ae7bad : SUCCESS in 46m 57s","accounts_in_message":[],"_revision_number":15},{"id":"f571b16885ec11e6e302dd43a8fc8c8db0b5c238","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-14 23:03:14.000000000","message":"Patch Set 15:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/dfe047798bdf4894ba8ba43754053759 : TIMED_OUT in 1h 31m 30s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5613b144afc6426da17058ecb4ee9f10 : TIMED_OUT in 1h 31m 31s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/9b639b137c1f4fd69674ce52a35b796c : TIMED_OUT in 3h 01m 37s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/ac6d20316a2644f4a00b9fed876102c2 : FAILURE in 1h 13m 57s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/399f87c6b59f49af86f5659194ba9e7d : SUCCESS in 1h 25m 12s","accounts_in_message":[],"_revision_number":15},{"id":"b46fc8a8bb91ecfdac6c294d1b76490e4b783671","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2022-05-09 20:07:05.000000000","message":"Patch Set 15: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"151b3132c480c107dad4e788640ae09f1257ff90","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-06-08 00:58:56.000000000","message":"Patch Set 15: Code-Review+1\n\n(12 comments)","accounts_in_message":[],"_revision_number":15},{"id":"b1e1e5275cf1e6789491f0cd07ec47945ff8d3a3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2022-10-17 18:21:53.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"7be83c697bfbf8ec85905aec6dd5e87b6275b581","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-17 19:24:44.000000000","message":"Patch Set 16:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/28a49829166d41508285da3415cb20d4\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/040c454ae2b340a486d56cd40fd6e437 : SUCCESS in 43m 34s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f5ce1928fab6483d9bbf54f389d1af67 : SUCCESS in 28m 07s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/32d2255f5869469b8a33bbe407d2fe4a : SUCCESS in 27m 44s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/194a2b1f46694c2c802cc063b9c0e8aa : SUCCESS in 1h 00m 53s","accounts_in_message":[],"_revision_number":16},{"id":"f20b9cc673d7f91f9c38468871f43a51c77547f6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-17 19:36:45.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c897b10e51c448f389e0755df5f66c44\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/85981e57f74645e5b7e3a9b01591f7ca : SUCCESS in 8m 33s\n- grenade https://zuul.opendev.org/t/openstack/build/ad5fe28b925e4085a107345b5c9d08a9 : FAILURE in 38m 47s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/81dcd06ea4ea41f8aa8d93831d049ad8 : SUCCESS in 49m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d211e729ed6b40b98740ff4b0798db31 : SUCCESS in 54m 19s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/058e05a8233e4442a69d84b2651d4a2c : FAILURE in 17m 44s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/87239e5c37b044ee80d33459beab0e50 : FAILURE in 18m 13s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d48e12fd202344318b7f312d70c101d7 : FAILURE in 18m 11s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c173b06a829e41a3898c8a4d7779522e : SUCCESS in 8m 48s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/ce040631f05c4f96a3f247142bca037b : SUCCESS in 5m 00s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/4f1e834b29d54deb826950bb0a97ca07 : SUCCESS in 16m 55s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/1a936cd0b37f4381b26d83b08c7428fe : SUCCESS in 15m 29s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/9ade8fe6f93d473798bb14c4d55e2ac2 : SUCCESS in 14m 55s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/08bd561b75b34e2a84ab16d1ca2789bf : SUCCESS in 16m 38s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/50c732865e04445ba832bddb3ad4a011 : SUCCESS in 18m 06s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/8df0499c2cbd4f1ba0a049607d6e5a0c : SUCCESS in 17m 51s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/03fd0b3a94e04503b86457b69631d26c : SUCCESS in 16m 03s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/1896ea0b7c6c4cc4871119c7cd700299 : SUCCESS in 14m 03s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/07a6fbe14a954e38ac52673a6de31678 : SUCCESS in 16m 11s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f8676cfb9a1e4724b51d308925190bbf : SUCCESS in 8m 09s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/a884a377b5644e33a21b9f80928cece8 : SUCCESS in 18m 34s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/0046b7d525de4acba5f78be186934ac5 : SUCCESS in 8m 30s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/456332da330a498cba0d514f12e9c631 : SUCCESS in 58m 00s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/be3193f33fcf4382b1626eafa8431cfe : SUCCESS in 1h 07m 55s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/9af9c0aae25c48049a6d4118ef444904 : SUCCESS in 43m 51s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/afef0e04a72f4a5ab76e991a93ae8fcd : SUCCESS in 44m 40s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/58024e83457f40d3be2f816a1595611e : SUCCESS in 15m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/97409944921843e1ab577847455d6545 : SUCCESS in 5m 08s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a56df4513d2c4c6da801f69827de52e2 : SUCCESS in 21m 21s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/36731a9c2c034f299aae0dc678c9bb86 : SUCCESS in 42m 53s","accounts_in_message":[],"_revision_number":16},{"id":"fc1ee420e817d0b0152a2210853e45869c91e806","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2022-11-04 04:15:22.000000000","message":"Uploaded patch set 17.","accounts_in_message":[],"_revision_number":17},{"id":"3a18dbf2aae6520e1c3089d886247da840286336","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-04 04:57:54.000000000","message":"Patch Set 17:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/801bb03cff0241b7ae4a30971b153080\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4700427b53754cfa9be821f13eb6c097 : SUCCESS in 36m 00s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e4adec58a7f64902aff86cd426be1ff9 : SUCCESS in 35m 42s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/ef5a304916304c4796ec16923e8954e0 : SUCCESS in 36m 13s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/f6b0f154367a44f0b8062636a5c54117 : SUCCESS in 34m 14s","accounts_in_message":[],"_revision_number":17},{"id":"8287a9eab3009f271f451e868bf21f3ab671a0f6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-04 05:32:48.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f617c370a02042e788ec98d3f695791c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ccee077232814de2903d32ec66d56419 : SUCCESS in 7m 51s\n- grenade https://zuul.opendev.org/t/openstack/build/866a5cfb67c6413286e96f1c3da1f533 : SUCCESS in 1h 11m 30s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f79ec92a0497499e975811735bd867ed : SUCCESS in 49m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/37c53677eb4142cfa0610e7b01591f7d : SUCCESS in 48m 54s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f44dd1c3908c4a01ae17b36109bbe138 : FAILURE in 19m 08s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f78829fade574053bfd2e15f3fcf55c7 : FAILURE in 16m 53s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b1a39dc366e34b438919003b087ea1f7 : FAILURE in 21m 24s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c11816513dc24bcaadd60cc0b33d681a : SUCCESS in 8m 55s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/98af465873724ff597b61727f31359f6 : SUCCESS in 5m 23s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/c635a286735846c78c01eaaed983177c : SUCCESS in 15m 35s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/75505f3388a54d9597ba1aed5ee419ff : SUCCESS in 16m 01s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/75b06438be1d4972bd83a4857ce572c6 : SUCCESS in 13m 55s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/7eb40f1f830941d8b345c72f74d577b9 : SUCCESS in 17m 59s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/b4302a1092b8471bbccbc1b9c4804a4a : SUCCESS in 18m 24s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/c68166c1f41743129267282daa5990fe : SUCCESS in 17m 24s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/f86b8ea8230042839b01bd7bc351fc7e : SUCCESS in 13m 55s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/c52382c24b7240a59002eafcb9114d68 : SUCCESS in 18m 27s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/8997ff801a7543beaab04d10b0f23568 : SUCCESS in 16m 05s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d1fa2e0dbbe8493a88e8185a65965f0c : SUCCESS in 6m 51s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/41e92f508162473cbf80383e93a699be : SUCCESS in 18m 00s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/dd4e0b48c7234cf99ed0628d30d99fe3 : SUCCESS in 6m 45s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/5ab58c16f07442718cc1f206acb2f45e : SUCCESS in 54m 37s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/3bd251d823714d33ad9894948cd9db8b : SUCCESS in 55m 35s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d4ac0a24aa794dfeaedeee664a68bc42 : SUCCESS in 48m 43s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/8404546e5c0b411fb87f7aa79d467e50 : SUCCESS in 45m 52s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/588f0a6641574565a1a42b8943f59e1a : SUCCESS in 15m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d9ce388c12934e369fc1e8288ff697aa : SUCCESS in 5m 26s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/dab677ff8a6d46dcad5f3c348d760a56 : SUCCESS in 18m 21s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4ffad9fcdfad4a77b93fe480fd69b7b2 : SUCCESS in 49m 23s","accounts_in_message":[],"_revision_number":17},{"id":"48a4efb41c6e62940b7b16733b57afb36277352e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-01-03 18:31:03.000000000","message":"Uploaded patch set 18.","accounts_in_message":[],"_revision_number":18},{"id":"9feea8181ac75229c5e5982e7cf158a7d1d313aa","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-03 18:45:25.000000000","message":"Patch Set 18:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b8002f5fbace464989b042fce78e3195\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/06e5e634730d48cb9d58ebc18b90d16b : FAILURE in 9m 17s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c1f7a2d4459245c58a56cc7a3953d903 : FAILURE in 9m 44s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/a19ff73c00e84939a4a1be028cdd11df : FAILURE in 5m 56s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/f16f2d3ed1f74a9283e0782c1a228af6 : FAILURE in 6m 03s","accounts_in_message":[],"_revision_number":18},{"id":"07b6a6636d05f55f539bfb35fdeb75f695c3f85e","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-01-03 19:21:07.000000000","message":"Patch Set 18:\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"8c564707de1690e5a7835a326dbee5cb78c0f11f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-03 20:37:32.000000000","message":"Patch Set 18: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a976bcd554c346ceb1e4b20b4d172263\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/41ba6a045f294b25bfc06ba6262a2217 : FAILURE in 3m 43s\n- grenade https://zuul.opendev.org/t/openstack/build/42c3d565b5f24f019042ed58754528de : SUCCESS in 1h 12m 06s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/ba8b423776ca4e2992d9fa7923f5c75c : SUCCESS in 41m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8e1529edb70e420d81c4074512b1bdaa : SUCCESS in 44m 06s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/36a88c7b0298458ca4f0628dcc893938 : FAILURE in 5m 34s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/974f7c8fedb14fe59a73aaf7d3d329c2 : FAILURE in 5m 18s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/bedef63f603145f694eef18c3c49c6e3 : FAILURE in 5m 40s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/1892bdf003924b7997e212504c43cb30 : SUCCESS in 8m 08s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/602b5fa24e2649a8a7342a8450a60070 : SUCCESS in 4m 54s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/c46ef1ab38c84a5189936b87cf1cf0c3 : SUCCESS in 16m 33s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/bd9a51a51078425f90b9f6a8739e27b7 : SUCCESS in 25m 58s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/decd6dade03249a8b432022c01d07d36 : FAILURE in 2m 50s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/0b486664288841989e60490ab85d2ef4 : FAILURE in 2m 51s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/05adeebd698a4fc08cad018dbf81c298 : SUCCESS in 27m 12s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/2b96e34e2ae4446d80b309cdd1452750 : SUCCESS in 18m 13s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/c8b43aa44b0549e886e4d1a82e2146b0 : SUCCESS in 17m 55s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/cf600e9ef3614db5bc5b291df74f8c8b : FAILURE in 2m 39s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/7c205a17b4b04ef7a241f35437ca677b : FAILURE in 2m 58s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/b9a8b2358bf54886b051f7b1709fb7e0 : FAILURE in 3m 26s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/52aa56bdf43c45c291b35f6017776461 : SUCCESS in 8m 06s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/4a048ec6220e4edd86af98b167e3f758 : SUCCESS in 17m 50s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/33dc5340e7684cf79869abe5725c42cd : SUCCESS in 9m 12s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/548ffd237f1942daac52a72fe87bd832 : SUCCESS in 1h 21m 50s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/e81f4f0030eb494097b404e7e4af4956 : TIMED_OUT in 2h 01m 07s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/72cd45cfa0834ac78ccc6aabe8bd4dcc : FAILURE in 12m 26s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/be1a9c2d7f4b4cf29ec8b7434bd51919 : FAILURE in 11m 54s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1bdd9c26310e42b4921bfbc88e9f5352 : SUCCESS in 26m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ccf2cefeeabb4c5ca4d5dd23f2c06109 : FAILURE in 3m 27s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/78f6c5dbbd674454af33e3fb4adf62e5 : FAILURE in 32m 59s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d66f4bfd6cd94527975875225bf6aa2f : SUCCESS in 1h 04m 46s","accounts_in_message":[],"_revision_number":18},{"id":"7087898574a541cc966128b706ba620a758470ff","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-02-09 15:11:45.000000000","message":"Uploaded patch set 19.","accounts_in_message":[],"_revision_number":19},{"id":"141f12d7157abbd892b9a8732ea0e50d47b06555","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-09 15:34:06.000000000","message":"Patch Set 19:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1bb821226ac44de1bb1599925c766306\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9f061ea7945c4be2b33d2395dfa73028 : SUCCESS in 15m 56s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/680522eb628f4a3bb582720409f99b6a : SUCCESS in 14m 52s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/e26c4af093a94afdafdf57a689361b13 : SUCCESS in 20m 46s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/034d243b4a7941dc916fb698a56f70f5 : SUCCESS in 17m 39s","accounts_in_message":[],"_revision_number":19},{"id":"d86a7d8774613733a405176709a5afbe30953b0c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-09 17:16:35.000000000","message":"Patch Set 19: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d7f2ab5dba0d4466a1e12123c1b2590b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e056a032727041a29578576ccf8f95d1 : SUCCESS in 8m 23s\n- grenade https://zuul.opendev.org/t/openstack/build/f4c10bd75f244467b6624fbdde05ae6c : SUCCESS in 58m 36s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/85b634f613954c70a5d7baefcd462c81 : SUCCESS in 48m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/87d2c7d8ce6a40a2b9c8ba61d94d9dfb : SUCCESS in 1h 02m 44s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e0e82557c8e942ed8f70869bbd836a07 : SUCCESS in 17m 47s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/785ce6951a7e4120ab68431c4bde89c2 : SUCCESS in 19m 58s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/af1bcff76e9f425cbdfadd55de2e758b : SUCCESS in 18m 57s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b9f7af99623a46049c5b4007371049b6 : SUCCESS in 8m 26s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/0f4bdce3489c48679e1962e7f4757bdb : SUCCESS in 5m 18s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/80ff81c352444b04a771166ccccad170 : SUCCESS in 17m 53s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/80d0f384b00d4aa18ada55b1ff1110fd : SUCCESS in 15m 07s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/98918ac12d384001bbc269e49fdbe445 : SUCCESS in 32m 04s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/25f6a0627d034f4c9b598d83080ce672 : SUCCESS in 37m 12s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/99007c8be1084b66ad9f943ec51c0956 : SUCCESS in 17m 47s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/6173223f57544185a5a2feef0ae442c6 : SUCCESS in 17m 40s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/3d1b0b76f21c4960a72082551b22d5ff : SUCCESS in 15m 52s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/ad8457790edc4c1688588dad0a798d4a : SUCCESS in 14m 37s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/5bb69cdf9d304fabb5877c58f7308d90 : SUCCESS in 13m 28s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/b9ff6060eb4b43d2a881e8791a365fd8 : SUCCESS in 15m 00s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e3bdbb16d2c048519e94a7a5ed60ffb1 : SUCCESS in 7m 21s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/dfcbf4e1290e40b6940e9d66661ff361 : SUCCESS in 17m 16s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/1b850275c5084f4faef147a1ebc39804 : SUCCESS in 6m 45s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/a2cc437f9e644ca78c00e6e62bbae643 : SUCCESS in 53m 44s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/02442b57bd8c4fe68b780978d55d2517 : SUCCESS in 57m 30s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e4f386fbbbea457eb01bb494bb1a3153 : SUCCESS in 48m 18s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/0694008a3e1f49f49020605d586f1a04 : SUCCESS in 45m 16s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/73accca18bd44a80a8e8c4199fedf306 : SUCCESS in 16m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/71b33094c04445b9ae7abcf8ac7a07cf : SUCCESS in 5m 54s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/1f8ef30943ca4ca6b038595b7dc3926f : SUCCESS in 22m 49s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7f89334e5c7f427da3b912616b0dafda : SUCCESS in 45m 03s","accounts_in_message":[],"_revision_number":19},{"id":"7ba6195836d46fe556883238fab3f46640df2f96","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-03-28 00:41:07.000000000","message":"Patch Set 19: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"b08ceaee61c69e22fbd2b612c0c492739bad01fb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2023-07-25 01:41:14.000000000","message":"Uploaded patch set 20.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":20},{"id":"896c195c7f3ec647e24cecc74a0116241cda5598","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2023-07-25 01:41:37.000000000","message":"Patch Set 20:\n\n(1 comment)","accounts_in_message":[],"_revision_number":20},{"id":"9af49846b977e02760ac159c3392b22d2d80c004","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-25 03:14:59.000000000","message":"Patch Set 20:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6463332dd1b04d1298ec376e334701d7\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/44f23278977d48a4a107bf6d4415572b : TIMED_OUT in 1h 31m 13s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/37587f32903f4044900a0a8eb8301315 : SUCCESS in 15m 41s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/47f169e4dcec4dbf97f8fcaac5827a3a : TIMED_OUT in 1h 31m 04s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/974cc7b6e6a94348874eae0e3c02fc31 : SUCCESS in 18m 37s","accounts_in_message":[],"_revision_number":20},{"id":"d693d8ebeaa774e1477065a69286108779eb447d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-25 03:51:58.000000000","message":"Patch Set 20: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/be75d4effdd543bd8024f08e8a5f14c2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c3bb2906ca54346a2592c3d2e09deec : SUCCESS in 9m 05s\n- grenade https://zuul.opendev.org/t/openstack/build/b6926fe388d74ceeba60fde05f5ccc9e : SUCCESS in 54m 15s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/987394748c324ac395353ee10d39906a : SUCCESS in 44m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e39936f058874bad99d7d54c96ca8758 : SUCCESS in 57m 32s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/257bfeec411b43029803bdb5ece91e7e : SUCCESS in 16m 16s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f7c062b14eb24c46bddaa0b32a08a467 : SUCCESS in 19m 13s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/af6ad202e2644c328635e3a9f03726cd : SUCCESS in 19m 57s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e8911fdc1ef4468e9b89126aa5ec32a4 : SUCCESS in 10m 01s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/9ec6f6ee5f964557a77f8679bb0bd82f : SUCCESS in 5m 36s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/e8ad3bd8ab0f4efab47283d152e0c02a : SUCCESS in 18m 45s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/cd173dbfcf714177aa5e19d05e78cabc : SUCCESS in 18m 33s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/6a7c46fd3b66471f96a0af12896927f4 : SUCCESS in 11m 36s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/c50cf97e88914da6af8d18e419da20ba : SUCCESS in 11m 13s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/9a63303036094683942b8b0c99dd874f : SUCCESS in 16m 42s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/7f2c3b9c33084e2e8e6d3917cda2801c : SUCCESS in 17m 12s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/bd28e93c766e439e9b1669af7129d45a : SUCCESS in 18m 19s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/52f9d28debd347fbba7dc81d9acc0501 : SUCCESS in 13m 23s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/faa5256b28cb4f8faba3b7a801ceb1a5 : SUCCESS in 10m 29s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/2bb27b31cd3d46cdaf4a91d675fbb373 : SUCCESS in 12m 15s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/f6b92bdeb7ed4267b42c8a4f6b44fd3a : SUCCESS in 11m 36s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/0ca4a4765baa49ab92f7aa3a8183be0f : SUCCESS in 5m 57s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/86463fd089fd42a4be461a0c66f51021 : SUCCESS in 12m 22s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/6c876fbf14c8478486a6700c9897ff6a : SUCCESS in 8m 06s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/c659ccc23d5047288495f950cc09ec12 : TIMED_OUT in 2h 01m 06s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/48eae20ae8764d188cca485ef7d7b721 : SUCCESS in 1h 08m 58s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/df8d737ad45f487e9ec4e304ed570fe7 : SUCCESS in 47m 51s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ceb062a15ca94394ac0da1b3c5ae3223 : SUCCESS in 47m 24s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fb743b2811a4417d8258f47b61aab20a : SUCCESS in 21m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a4d93fd15f0e404cbb8febd90a079e6d : SUCCESS in 5m 52s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/fce67765b7cc48f1916b585cecb5175e : SUCCESS in 25m 03s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/80e80743ebe9450da6d8439ae6551bf1 : SUCCESS in 46m 46s","accounts_in_message":[],"_revision_number":20},{"id":"15217651c7404f2a348a2565ab16c28a2beec701","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-09-28 22:43:53.000000000","message":"Uploaded patch set 21.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":21},{"id":"c97c16bfc15e6e38942cba01f570f22450d648fc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-28 23:55:29.000000000","message":"Patch Set 21: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c2b4d4bccfb540228448129df27b6f80\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/69969235412d4dca83551588e1b37ab5 : SUCCESS in 8m 59s\n- grenade https://zuul.opendev.org/t/openstack/build/dbc1bcc1ddea4e47a033799b6f2ac732 : SUCCESS in 48m 55s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c3bbccb6116041b0bebf8330809b7a67 : SUCCESS in 48m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3e30870a9fd344ea96134e78a01f01b8 : SUCCESS in 1h 01m 44s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f3c1d6100a21435c91595fdc01db8a92 : SUCCESS in 18m 43s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b6b41d6c6253421883358be2e25c7ab0 : SUCCESS in 15m 39s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/89bf1e32f4894aa08a536cfd699cc274 : SUCCESS in 19m 51s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/92217d1acb04413f8a8f8bc3b611c410 : SUCCESS in 9m 21s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/1bbee7b5ced44ad8b52e13d976d24497 : SUCCESS in 6m 06s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/2eb3c49e2de24e9c8a8ea06ca8f0c862 : SUCCESS in 17m 32s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/9faa97b914bd4449b67138dc4b044139 : SUCCESS in 16m 31s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/f3e74886ff5f4ddfa731c09bcb842192 : SUCCESS in 13m 26s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/534092f3bd6c43c2b4e2198941a85f5c : SUCCESS in 16m 12s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/fd6ede9b1cea4a6cb9d529b25da5a62c : SUCCESS in 17m 01s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/9220dfa4ca7e4c4f8b12461795c357f7 : SUCCESS in 17m 55s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/4ddd447506164887a9e30a6c2f021f19 : SUCCESS in 21m 49s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/386944807c19455ca74dc691ce995941 : SUCCESS in 15m 57s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/0a0c10d48e5343d0a6f82bb5de58d226 : SUCCESS in 17m 15s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/ddae7f8c0cbc4d63a3c6b089c3e43772 : SUCCESS in 15m 39s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/ca4a1ba76edc47c1bc4cc6d23f4ff858 : SUCCESS in 16m 32s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f09a6885679146938683927542796d18 : SUCCESS in 7m 17s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/2b52cccc177d4ebea7948c89e9598053 : SUCCESS in 12m 48s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/bf0148d8d9d44a308b0642bb0f23764c : SUCCESS in 8m 52s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/f8ddec04f920418496934f6a412dac14 : SUCCESS in 56m 43s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/7b4662fc53ff49a6996e728dc06dec12 : SUCCESS in 1h 06m 06s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/58ebf38576064027aa499a7900963ab1 : SUCCESS in 54m 38s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/031dc4465fbb4a2c953595d71900d218 : SUCCESS in 49m 46s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/37ce665794dd4b5cb8afb2e63b8e7c70 : SUCCESS in 17m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/74a379157d9d453f90aadee1e249269c : SUCCESS in 4m 57s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/585e4a8b9f0b42a4a3d14bf9d00a7e58 : SUCCESS in 20m 51s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/85165a6f8e59432487d1a21b4fe234a5 : SUCCESS in 46m 39s","accounts_in_message":[],"_revision_number":21},{"id":"73baa0aced9aad694258b0239122c23db377318b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-29 00:10:34.000000000","message":"Patch Set 21:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/25ddbffb1d714e43b3b21a59f0544d1b\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/77358515f3d44f2abd8d9726155d01a6 : SUCCESS in 53m 53s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/69ca6b00fa0643638bcb613f29bd5b5f : SUCCESS in 1h 24m 16s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/86bb7eb350f048b392c6d025ef28350b : SUCCESS in 1h 04m 51s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/42be47b433f6486287fd6348f617bd66 : SUCCESS in 52m 51s","accounts_in_message":[],"_revision_number":21},{"id":"7d6af29be350e7b10c75c248cf24c46f5f3909af","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-10-16 22:42:45.000000000","message":"Uploaded patch set 22.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":22},{"id":"b211374202abfd83911fed963af19e3dfc7a63c2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-17 00:00:31.000000000","message":"Patch Set 22: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/74e358dbeea840588719f58b9a1fe472\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c4ea9b2097f34b7aaba6610e0e03bea1 : SUCCESS in 9m 53s\n- grenade https://zuul.opendev.org/t/openstack/build/56e86ae40c6c43b889c81aaf487b09e4 : SUCCESS in 55m 17s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/eec354376a044555856ea1a3bfc6c2ec : SUCCESS in 1h 01m 49s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/4a5c5f8239ac4c0087395f061d288276 : SUCCESS in 47m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/88d55a492b5941d18e85f7808a9fc67c : SUCCESS in 1h 08m 05s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/266467b538b14324afbb3ac1d554fa45 : SUCCESS in 19m 23s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3d551872e4f649cba05589b8cf44729b : SUCCESS in 19m 41s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c1bc06dd5d13490cad1392464051f452 : SUCCESS in 18m 02s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c257c8e6d9944277bc2d8d12670a998b : SUCCESS in 9m 31s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/1bd53fa32e2b4b2c978d3692a0808d6b : SUCCESS in 5m 17s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/8cb8090183a744858bbd1169587e5cce : SUCCESS in 13m 53s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/b193342176c0470e8d2e07ac44618189 : SUCCESS in 15m 01s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/c668f96d5f834c8abe4f41faafae7c18 : SUCCESS in 16m 19s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/0867f1a84a2c41f685d4fb37f3b0931b : SUCCESS in 18m 56s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/22cd82ff2bd64137a1f7cb2d9addc4a9 : SUCCESS in 16m 07s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/2af63e062847434e94d068c8eeb6fe09 : SUCCESS in 20m 10s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/b386ceee1b5e4f1b8fb3b7e502075d02 : SUCCESS in 18m 58s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/0ac9ddb6baf244d4bd8b3fc07330fffc : SUCCESS in 19m 24s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/95e0ff9541bc442491e461e0856cb1c4 : SUCCESS in 19m 41s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/341aa27e56f34716935b21a3ad038d78 : SUCCESS in 14m 43s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/b0d1e57f6cdc406584028ecd5746a850 : SUCCESS in 13m 07s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/39a12357fdd54ff093c85e9baca1d46c : SUCCESS in 8m 04s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/52e29e4c26e4494b865a1ebb64fdfb91 : SUCCESS in 15m 06s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/9aadf6c3503c47d6a8bdd024d17a3b54 : SUCCESS in 7m 08s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/87bfbe191f8841359ced8f148b6a404c : FAILURE in 59m 42s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/46713a4c86164e7aa10daaea0406b7c7 : SUCCESS in 59m 04s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d0b5bb8cdedc4f6faae3d362932ce1f4 : SUCCESS in 58m 42s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/fd0f5b5d965042de8fa66629ef7b0144 : SUCCESS in 57m 12s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9005eb48796c46b8aec3d3e2d64deff1 : SUCCESS in 16m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/07bf0111f9aa438b947f1b43d8aaf45c : SUCCESS in 4m 58s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e51b5a5bf20b447aaa1c4129f940b01d : SUCCESS in 20m 37s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/09a12d259df2419a8e1bf5a8d359ca37 : SUCCESS in 1h 07m 33s","accounts_in_message":[],"_revision_number":22},{"id":"743b0ef5423c826bc530196a305a5499f6fe89f3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-17 02:27:15.000000000","message":"Patch Set 22:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4141df594f0e4d799a6f3947f80d6c87\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/deaeec762dc94c6eb00c391ce8ac624e : SUCCESS in 1h 23m 02s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d697fa92cd964149b8d105c78a0d7634 : SUCCESS in 1h 05m 58s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/c9dd2cbb2dce47d78e6ad2d5e925b682 : SUCCESS in 1h 03m 25s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/1e4feb2918f94df89eed7d1ee9ccd099 : SUCCESS in 53m 27s","accounts_in_message":[],"_revision_number":22},{"id":"5eaece4628fa3390c94f6da878ea89380cbc8df5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-10-17 18:10:55.000000000","message":"Uploaded patch set 23: Patch Set 22 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":23},{"id":"c27c1023a560478893462463fcbbc49b708dae8c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-17 19:27:16.000000000","message":"Patch Set 23: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ffe4aa68962e488e9f54350aa1ed9b69\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c5c4e79a975748f9a6940623f621d38a : SUCCESS in 10m 37s\n- grenade https://zuul.opendev.org/t/openstack/build/1d333ad6e6be49f1b4ff84f96cdd31f0 : SUCCESS in 1h 10m 13s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/be441ba1ddd94b9187c8d0b5054c9751 : SUCCESS in 1h 05m 25s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/5fc6fc475b8c4ff4850d08cd72d70fae : SUCCESS in 37m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bde771374d51425cb3e0258387818dac : SUCCESS in 58m 59s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7a2a32de42644c06992b40eab92c25f1 : SUCCESS in 17m 21s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/58c572a9ded045c9a7a7965c25a3ff74 : SUCCESS in 20m 25s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5638c73defad468fa46003da47e07dab : SUCCESS in 20m 52s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/72682a07308040a8b4270f1d52c04c0c : SUCCESS in 9m 00s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/0bb4c28bdbf64d85baaf060d436b4993 : SUCCESS in 6m 47s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/0d2e90d01e754056888ad0c268807eef : SUCCESS in 18m 18s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/4172aa1b4cac45e9b1907978d8a84c59 : SUCCESS in 25m 28s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/ec6bc9f9818840909ef1f752a4a1d508 : SUCCESS in 14m 24s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/943d47b226c24fc38367ca7f94cbcf50 : SUCCESS in 16m 18s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/be4979a2897345f59ffcd420a42e0d03 : SUCCESS in 15m 34s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/0ab0316127e442a5877b9a35c0e81556 : SUCCESS in 18m 17s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/41acc7146d6a46d1b8336e9e446fff33 : SUCCESS in 16m 47s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/7be840a099b1403690aafd9cc7c9663d : SUCCESS in 22m 47s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/4b0390b61f0b4ac18add720e6d8dc886 : SUCCESS in 16m 29s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/a78133d2485542a6b4fb623a94effdc2 : SUCCESS in 26m 14s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/8f95e5a953ed4d5684c0d9c9b093fda8 : SUCCESS in 17m 42s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/6f0382b8a4e84a268461a563f741d411 : SUCCESS in 9m 03s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/8121c15dbac8499c98d836d83f8fedcf : SUCCESS in 23m 50s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/519e6a8f54bb483bb83e2a016d12d131 : SUCCESS in 9m 22s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/fdc60cf715ac4d97a06a3d0a40889ff3 : FAILURE in 55m 55s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/4377397c18c94aefa82707c42d82f6c1 : SUCCESS in 57m 25s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4949d6fafed247fab1c249d1f4f73eea : SUCCESS in 35m 20s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/6cb1da6140114e06bf14ac408f9e4b68 : SUCCESS in 55m 59s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3b8bce86c6284eafb55ed32fb511074c : SUCCESS in 14m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f915afbe71d5470099fae4f7ea0a6924 : SUCCESS in 4m 30s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6b751f11e92d4dc3941b7f80c302c6be : SUCCESS in 20m 09s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4442269b9a3b4e0db3be8ff2893bda19 : SUCCESS in 41m 38s","accounts_in_message":[],"_revision_number":23},{"id":"5a6731327779724a51cb7196349a1c7119113a40","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-17 20:23:08.000000000","message":"Patch Set 23:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/47bb0e3640ef46cda21ca2fa5277f762\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d9358ff26438480fa19a4da40a5c88af : SUCCESS in 48m 27s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/178d145507ef4510b138f74f9a93e096 : SUCCESS in 1h 13m 30s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/1e6d51e9509f4a0c9ce5f6afd2e72a1b : TIMED_OUT in 1h 31m 10s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/0209a4623fdd46d991cb764e889baac5 : SUCCESS in 58m 06s","accounts_in_message":[],"_revision_number":23},{"id":"afebfd4ae67aa1fec782bac69b685805c04ff4eb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-10-18 09:53:30.000000000","message":"Uploaded patch set 24: Patch Set 23 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":24},{"id":"6d57c7c07acaae66af82a7ee96f93bc0d6952deb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-10-18 10:27:17.000000000","message":"Uploaded patch set 25: Patch Set 24 was rebased.","accounts_in_message":[],"_revision_number":25},{"id":"eb5f56080bbae25cc52be0d219eb3fd169e66e2e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-18 11:45:53.000000000","message":"Patch Set 25: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8b22cd46eb3a4326a232101a6bff8d84\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ec012c96bcb47b89d654bb8059d67b1 : SUCCESS in 9m 10s\n- grenade https://zuul.opendev.org/t/openstack/build/48d16f8219ab448b907aa734792462e2 : SUCCESS in 1h 07m 20s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/beaf84d6bb114464b72c5fe62012d714 : SUCCESS in 1h 04m 47s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/5e46cbd3d85d424ca0dee997c9c2ea75 : SUCCESS in 51m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/33cbf7ab004b43409a1417f6d8d332f4 : SUCCESS in 1h 06m 35s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/eb70b9d13bb4440780daf3c20f274027 : SUCCESS in 20m 23s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fe2bccb2d1b74f02b77349c6b3418a7f : SUCCESS in 19m 12s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/19e66d98a2c141b595b814cc7268b6be : SUCCESS in 16m 33s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ec9b852e55a64c618022f0c8ac5f7bd7 : SUCCESS in 8m 15s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/cb0b6669eedb4293bae8f107fad9ae85 : SUCCESS in 5m 08s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/d1db40de4d7948a5ada262ab74aa7987 : SUCCESS in 17m 15s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/ada7281be6414a80a7b0e79250913fe9 : SUCCESS in 16m 16s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/f13f847de43745c99c543390ae6383fb : SUCCESS in 16m 59s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/8a8b27064cf8448581dc70dccf66f13f : SUCCESS in 16m 46s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/4a766bb83e5c49cda3d7f0c39f045844 : SUCCESS in 14m 38s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/e0e8f7006cc8467c84561557c68c315c : SUCCESS in 17m 55s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/58c0cc5ad7fe4964ad8c1a7c359817f4 : SUCCESS in 23m 29s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/e4a612d25bd54946b075a59fb7a11b06 : SUCCESS in 19m 45s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/6fa9cf20f9b9421aa5fb854eb9a3da88 : SUCCESS in 15m 46s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/93117f02c62447b6bd247882dbde9fa7 : SUCCESS in 22m 45s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/adbafb598bde4bc8b4e9592cfd8c56d1 : SUCCESS in 16m 25s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/0fbabb41144b4b2ab51779038c97b68a : SUCCESS in 8m 22s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/9242c512c15a4b68aec7c4382d7fa9b0 : SUCCESS in 21m 03s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/644ffb6c9ef646c0a8328b826c854a79 : SUCCESS in 8m 05s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/f738b34d4e94405da46510c1e470202d : SUCCESS in 1h 11m 00s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/2c1b5009aa424e6d9a933eda8b0aab83 : SUCCESS in 1h 16m 32s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/3da8bd9414244e87a3bc61c10032af90 : SUCCESS in 1h 00m 19s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/af13751b5d5645bcbeeb502bba8d6ffc : SUCCESS in 57m 26s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/651a2d53c96b4e7c9f4adf832b956f1c : SUCCESS in 24m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/25023eacf06a46ee9ae61a1f7c99e311 : SUCCESS in 5m 03s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d50be03027844f7e955620413a52797c : SUCCESS in 24m 28s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f7fe11c1b5d74b8ba2fee4a031208cab : SUCCESS in 1h 09m 19s","accounts_in_message":[],"_revision_number":25},{"id":"ac11ec0e40fe1bd437aea4dad5bbba719572a1bb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-18 12:01:21.000000000","message":"Patch Set 25:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3083233120db412e832fe0a707abc2f9\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4157ba1011ea48c6931c73870b396869 : SUCCESS in 58m 19s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/01ac52a526e842c0a97561d8426684f3 : TIMED_OUT in 1h 31m 30s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/e15a9e5e2bc54ba6b3c62ed975ee1db2 : TIMED_OUT in 1h 30m 57s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/2c9b21a86b3643a8939a37b8be2d7494 : SUCCESS in 1h 13m 36s","accounts_in_message":[],"_revision_number":25},{"id":"123199bd3e007444f95e5d9a8108537752de9786","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2023-11-13 18:50:47.000000000","message":"Patch Set 26: Patch Set 25 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":26},{"id":"7a38898d161d10da003fa8477cd6f39126fc5ff1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-13 19:20:08.000000000","message":"Patch Set 26:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/84055ef3c157488fa75efa24a130d8e1\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/43a6128480be4733af31baec355ac8e9 : SUCCESS in 19m 40s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b2f8ba45a4c849c19832b0a2fe39902d : SUCCESS in 19m 44s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/ce74950a78fa44bb97e45a57d53b08af : SUCCESS in 27m 48s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/0ab524a15f8b4796aaae0ed1c0c4712a : SUCCESS in 23m 39s","accounts_in_message":[],"_revision_number":26},{"id":"62aaf84dd6977efd2df58dbcbc7e68fa6265f742","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-13 20:02:05.000000000","message":"Patch Set 26: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7d39a0d3cae642d6b2b52fd7f63020c8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ebce90b87ad4ee1942030cc7a2d39f0 : SUCCESS in 9m 39s\n- grenade https://zuul.opendev.org/t/openstack/build/b26942885dc54b2e83665c2cd6809546 : SUCCESS in 1h 04m 31s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/9097ac1262a544cdbae742e9e08c8fec : SUCCESS in 1h 01m 50s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/9647a27a29c3432e93ae3bf482d1a945 : SUCCESS in 35m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/15143121d92d40dbaff9d51da709729c : SUCCESS in 56m 49s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b1c340ad06594502a1ac0b4e865db4cb : SUCCESS in 14m 29s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5f49777b4c4a4775814d858efe6e73fe : SUCCESS in 19m 59s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f0d94b5560c348bda360aa36b244d575 : SUCCESS in 20m 23s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f1aa9e157ece470b9ea15f19b5c6a786 : SUCCESS in 10m 00s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/c636b6e58cbf4f1eb759dccf97f9a428 : SUCCESS in 8m 08s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/7d0f7b2fb16c4ddaa6314678673a6681 : SUCCESS in 15m 09s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/158eef839afc406fa68aa02f29f181b9 : SUCCESS in 17m 57s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/7a37e1ed7f8046f8b6895703d754fc17 : SUCCESS in 15m 47s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/5f8592d21d304089907a4d9f03dc8a4d : SUCCESS in 17m 20s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/ef9d419b1a3642bda0437277d1f664ad : SUCCESS in 16m 59s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/cca1fee11c564e27a638cbd3fa998f84 : SUCCESS in 17m 56s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/defd8c0fe15e4e90bc38b5bb38296a79 : SUCCESS in 18m 16s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/571b087400e340ffafd79096b797d07d : SUCCESS in 18m 51s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/496bd0602d47402d89b03b1614a57724 : SUCCESS in 15m 22s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/a0310d911cd44a46a8a15e5e7bccb69e : SUCCESS in 17m 21s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/653f68375b6340789923e7b4ac666429 : SUCCESS in 16m 48s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/8d58057db6ff405197886b58fe4f622f : SUCCESS in 7m 12s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/6294291e83ca451c839974cf4fe60c66 : SUCCESS in 15m 43s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/d2ebd8030ea743fea77aea714d306c35 : SUCCESS in 8m 04s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/51c58ee89b5f4fb0a2523d1e35de9bc7 : SUCCESS in 58m 47s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/83187fb9ee0f4c0483aa6414fe855e00 : SUCCESS in 57m 01s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/6c07dfc01ee74bd88e5d64cbc45503c4 : SUCCESS in 41m 40s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/6798564f6d564ff9b0efc356bbd57384 : SUCCESS in 39m 48s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/21b72de6b290454da2127afd3f70d267 : SUCCESS in 17m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c488bd9cb5e54556bfd86b6b7b70da78 : SUCCESS in 4m 30s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/cd49e68e7a064d8ab7c3ddfe9fb47059 : SUCCESS in 21m 05s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4931a53a834a42e8a2c989ea6e7b6c67 : SUCCESS in 39m 26s","accounts_in_message":[],"_revision_number":26},{"id":"1d2144b32f41b62bbe6cbad2aa96237f88ed7ff3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2023-11-14 13:26:04.000000000","message":"Patch Set 27: Patch Set 26 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":27},{"id":"2c9de3a65faf5662ba795032d8ea9f99813c8240","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-14 14:37:58.000000000","message":"Patch Set 27: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7d85a2b2387f442cb272a2d1ca6f1c7a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4cca74f6adfa46c8992aa2f319e4bdf2 : SUCCESS in 8m 49s\n- grenade https://zuul.opendev.org/t/openstack/build/186f38da76fb480dbcbac5f943d70a24 : SUCCESS in 55m 39s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/553b1e7d240a474b9aec52737c75c876 : SUCCESS in 1h 01m 16s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/5a30f078d1ba489a9c6ae63f5498b87b : SUCCESS in 40m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/28464bbc7617421395a251c8d6953970 : SUCCESS in 1h 04m 18s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/26a2ebce51344bcd9c98e130846fa1c2 : SUCCESS in 19m 14s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/763fc89de22b4fb78ebd810fc15b2ecb : SUCCESS in 19m 11s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/104d238ba7da4978ba1891af97062964 : SUCCESS in 18m 59s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/020930680f5c4341ac6a2224a9103e49 : SUCCESS in 9m 04s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/1f49772441414b92a4e37f8f6878df6e : SUCCESS in 5m 20s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/de425ef5b1ee47bcbc4e271f06e7c334 : SUCCESS in 16m 23s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/e78b28a5b73a4db4a20986567f433045 : SUCCESS in 24m 02s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/3b91a383b83646f8a1d4180c7c4c98be : SUCCESS in 15m 54s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/7da78b7182a14454ab8a854345d5a9cb : SUCCESS in 16m 08s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/428200da16b5440d969f011b4165658f : SUCCESS in 16m 00s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/239f46e8f81242388c6ff2e69baede33 : SUCCESS in 14m 27s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/c2842bcbdcb24c16b0ff70abbface62c : SUCCESS in 18m 20s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/b2072e58a19d40f399cd036559cf42ac : SUCCESS in 26m 58s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/0395ad94dca44653991f25c09b5a98bb : SUCCESS in 15m 53s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/188900b64dfb41e0ac9e23a14a3fe696 : SUCCESS in 15m 23s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/34b24fb160e344d588bcbdb2fc8f544f : SUCCESS in 25m 51s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e666524ea2104b94b38203dc3662ce88 : SUCCESS in 7m 05s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/ee2c103d34a24d328df4e811aa73f23f : SUCCESS in 22m 07s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/0dcbdb5db69f4003b83d7aaa15e2b025 : SUCCESS in 7m 05s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/9aa94539cf50471b9aba61345a7095ee : SUCCESS in 53m 23s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/3182e374b3f04f029228201e43a91587 : SUCCESS in 1h 10m 15s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/7d08732a614a4e45a0cbc9b8e55e3e05 : SUCCESS in 43m 21s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/5b3fefc94b6d443bbc0fae2adf7d5a09 : SUCCESS in 47m 58s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/064499cf70d74248b539cd4ec6405420 : SUCCESS in 17m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c50066eb50b94bb2b8e8bd08f33d3732 : FAILURE in 4m 37s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/93d7ffc176ba46a1b3f25efc751575bd : SUCCESS in 24m 56s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6902aea36c4c454eb937208557e63fc4 : SUCCESS in 59m 33s\n\nWarning:\n  Comments left for invalid file swift/common/middleware/s3api/controllers/multi_upload.py","accounts_in_message":[],"_revision_number":27},{"id":"8cb0c666434d9c0620ae9fb119fd3c907d32e38d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-14 15:00:19.000000000","message":"Patch Set 27:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/db9fb652f1a04c2d8c8a7663f46236e2\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f46ce4e8e9fa441492cd938c59b7e4b2 : SUCCESS in 15m 26s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e4af450a3fff48d1b87bdba85d99f5c5 : SUCCESS in 16m 02s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/e3dcff1f566a4519abedaff27642fb8a : TIMED_OUT in 1h 31m 05s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/8008e3d886d341f59846b1c5fbdc53b9 : SUCCESS in 19m 02s","accounts_in_message":[],"_revision_number":27},{"id":"9ea9b78ba0d65b37b34af0df5c6ff5b8e6014a3c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-11-14 15:20:57.000000000","message":"Uploaded patch set 28: Patch Set 27 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":28},{"id":"3ea4d3fb854355e79e6c55b6f892d512676837d1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-14 16:41:44.000000000","message":"Patch Set 28: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3adaebab40234e42aed7a123c2268b3e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28758f2e01ce41bf8737bff04bcdc4f6 : SUCCESS in 9m 30s\n- grenade https://zuul.opendev.org/t/openstack/build/e25f69213beb4428bb93267748ebff1e : SUCCESS in 50m 35s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/4327baad62b740f1b49cd0dc09897678 : SUCCESS in 55m 06s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c7a3d39acb5d47ddb5caa97b8080fa44 : SUCCESS in 43m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0527fb6010384f889b1bba908d20544a : SUCCESS in 56m 26s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e3e3642505124a6cbae6ce97ea968894 : SUCCESS in 18m 14s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3bded57bc59e4148ade0ae21a369c2f7 : SUCCESS in 19m 11s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ce648a86d63b4fe69fa6a90ca5c7af18 : SUCCESS in 21m 34s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b1cff599217345f9afea18a38fd39bf9 : SUCCESS in 9m 30s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/b43c47f746a946588974f21f2e7d1099 : SUCCESS in 6m 01s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/db559156e99a4fb68d4c2ec09a14de61 : SUCCESS in 18m 58s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/5e4f2de4fe50489fb9a131c9e954c032 : SUCCESS in 17m 27s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/0aa62524f5314975824a15ea6a462352 : SUCCESS in 15m 37s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/9ddae6eb5b0d4718a3dab07cd0b55d0e : SUCCESS in 20m 45s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/5ed9993d210a4be398335f51e24d6abd : SUCCESS in 23m 43s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/9204e0797c3e47aa8aa0a7cc5e7ee2fc : SUCCESS in 20m 20s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/00644c90a3544944b908bbf8da1f974d : SUCCESS in 25m 34s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/23a38b35bb2d4cd2b9b49bfae7de9bd9 : SUCCESS in 19m 05s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/47629ac74720405cac1be30005857255 : SUCCESS in 15m 57s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/4e7dba6a48bb41e4b53af2e1292ea7a7 : SUCCESS in 22m 48s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/507143565c894bf891e28ef12bcd4e6f : SUCCESS in 16m 34s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/c86c410f956849748954ab3db0064114 : SUCCESS in 8m 39s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/4517e844fa9e441fa9d5edb99135c614 : SUCCESS in 17m 50s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/eaec45f0bcc845bcadfaeee240ebf719 : SUCCESS in 7m 31s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/380bfc23912f45b6b0e47652c44bac86 : SUCCESS in 53m 44s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/04cad5b9c12449ba9615de37344f26cb : SUCCESS in 1h 02m 57s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b8c0de3d688344a989664136b714195d : SUCCESS in 42m 33s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/64a8fede9af54bf29a9932c642a9c88b : SUCCESS in 59m 17s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9f4b98109ec64aae83b231b73ef125f3 : SUCCESS in 19m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d041cbf7c764faa825119fe33319940 : SUCCESS in 5m 04s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/553240ac17174efebeb737cb1f3a07c6 : SUCCESS in 21m 29s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/799c43d04d1a4b7fbb9c4c8184606aeb : SUCCESS in 1h 05m 39s","accounts_in_message":[],"_revision_number":28},{"id":"6729343710ffb982fe78dd27025de087322dc52d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-14 16:54:47.000000000","message":"Patch Set 28:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5763b07f926646849ac5d5bc84b23ff9\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/cf242027b5db46bbae6abc289e4fc982 : TIMED_OUT in 1h 31m 13s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6c34c83551c844a8a3d47ddf6c1e5bdf : SUCCESS in 1h 17m 08s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/6a116c58d8274c73b97622cc51169699 : SUCCESS in 18m 58s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/d49c4b5119af42a4b3a2b22cd8a034f6 : TIMED_OUT in 1h 31m 03s","accounts_in_message":[],"_revision_number":28},{"id":"b863c4f8d679fe56f6416f0a7766a2669a6690f0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"real_author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2023-12-20 17:22:56.000000000","message":"Patch Set 29: Patch Set 28 was rebased on behalf of \u003cGERRIT_ACCOUNT_7847\u003e\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}],"_revision_number":29},{"id":"c1e0552341c95de32f4442769df2426fe18768c9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-12-20 18:21:31.000000000","message":"Uploaded patch set 30: Patch Set 29 was rebased.","accounts_in_message":[],"_revision_number":30},{"id":"d852fe006f078e88a6be4e3d921a6330563c2451","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-20 18:44:57.000000000","message":"Patch Set 30:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4d01122793314f599e52f542dc24c3b8\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3e2d8debed9d429980e993bbaf5eab70 : SUCCESS in 14m 49s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/24910e29d3d44e34b1d9c67bb1dd409b : SUCCESS in 14m 53s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/dd58a89f4d1848babc0b1d4eb145c08e : SUCCESS in 22m 25s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/44d7c30796b2459cbd16b8f96d7ee296 : SUCCESS in 19m 02s","accounts_in_message":[],"_revision_number":30},{"id":"bb77d99b8e2e788f07dcd9cf913aa17ada98efaa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-20 20:00:49.000000000","message":"Patch Set 30: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7f552165d977460b82a93ce4752477ea\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/15bc1d0841e342bd9886397c391e416e : SUCCESS in 8m 37s\n- grenade https://zuul.opendev.org/t/openstack/build/0a83bdc8e27d4891ab431db78db7ade8 : SUCCESS in 44m 09s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/c3225580717b498cb0f1acbf55b62991 : SUCCESS in 1h 12m 01s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/6309dd320325441ca792904f5cb0f7a8 : SUCCESS in 33m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1232b41b2f28401da9c7542ec4da279e : SUCCESS in 1h 08m 57s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/82d781cdc5974e5cb99dc3945bf4f551 : SUCCESS in 15m 54s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/32076d2c230746aca310f8813ce8b992 : SUCCESS in 22m 10s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e9fbc1328ce84e879cfe3608b5730fe3 : SUCCESS in 22m 55s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/4c6d00a76f814ba68276cc337312961b : SUCCESS in 7m 57s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/265c1b1d183f4eaab3fc42f0b0b3dc04 : SUCCESS in 5m 38s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/78f9381eb4464b5396bf3608027a815c : SUCCESS in 16m 37s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/22a458a602a84b689cd88eb921ee8699 : SUCCESS in 17m 58s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/46fd86a71c374f1f9c6a719dc2184fba : SUCCESS in 15m 18s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/a0e07cab734143c2879dbe44401c9c16 : SUCCESS in 14m 11s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/08ee1e5a34294470a114243bcde83455 : SUCCESS in 17m 24s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/f3ef36d8d3f54e42a55645e308a9c911 : SUCCESS in 17m 48s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/1ffe713bef6449409c459dd832ca7d33 : SUCCESS in 19m 27s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/a3328cd450294e26b3985c37ec200a10 : SUCCESS in 14m 30s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/d1a24628556e4342ad799079373f53b6 : SUCCESS in 12m 53s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/4cab7fbec7db4e26952d3997b14963b1 : SUCCESS in 16m 13s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/d3bb41199e2c42e385806d5308fe9d35 : SUCCESS in 19m 32s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/5e4325ba626446e996855848e161ddd6 : SUCCESS in 7m 29s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/cc849d757dbf4b0498e71910796c9905 : SUCCESS in 19m 30s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/4fda5a8a72b94a94a6196d2b5e848711 : SUCCESS in 10m 50s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/3c0d96e36c2b4ec7af7099d0a8ddb61b : SUCCESS in 1h 01m 56s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/65f557209bb249b69b9c9eb3c425531e : SUCCESS in 51m 39s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/92581d01adfa466b9ac6fe8989948b78 : SUCCESS in 50m 42s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/bb2931ab0276483ba77b1fe3afa3358f : SUCCESS in 51m 03s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d969e1855ad34960b909f58e31c74041 : SUCCESS in 18m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/72c48371d43e4771b0b103b103389bff : SUCCESS in 4m 49s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/bb75ffae353f4b17811ce89db3f3baac : SUCCESS in 20m 57s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/648371e9d56d4038912c24df5b2d9cb1 : SUCCESS in 1h 02m 09s","accounts_in_message":[],"_revision_number":30},{"id":"7f575317344056af488a9bab9f8438106acb2891","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"real_author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-01-03 19:20:13.000000000","message":"Patch Set 31: Patch Set 30 was rebased on behalf of \u003cGERRIT_ACCOUNT_7847\u003e\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}],"_revision_number":31},{"id":"5b65109f98f2a77ede5d595b7aaea0f0f98a82e6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-03 20:33:03.000000000","message":"Patch Set 31:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f04aa0dff32f49ad945a0f1e996b5850\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/01c45e49f8cb4bd4bb03319f31867af6 : SUCCESS in 16m 42s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5083e3c1df9a42a1a6d22e495a1910db : SUCCESS in 1h 10m 29s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/c54d65fa998a4b3d8f7f8d20958c5c58 : SUCCESS in 21m 01s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/3fd50766042b4fcb833b793016101fce : SUCCESS in 1h 03m 04s","accounts_in_message":[],"_revision_number":31},{"id":"edd4acc6c5d93b002bc67da2fab9f9d181f44815","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-03 21:02:56.000000000","message":"Patch Set 31: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0b2417ed4f004739a8646d3f4923715c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/22ade1b145a64eb283ddf7f2e799a738 : SUCCESS in 8m 05s\n- grenade https://zuul.opendev.org/t/openstack/build/b070bd3bc3634c988f5416f61dc1b48d : SUCCESS in 55m 37s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/09af53e744574ed584e992219d6e11a5 : SUCCESS in 1h 05m 23s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d81b6f6de33c47a697402ec0c4de0155 : SUCCESS in 50m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/212fb6cbfe474f87842c6d4a7c41f8e8 : SUCCESS in 1h 03m 16s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0121503dc284461198b782b1f892d0f4 : SUCCESS in 16m 21s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9c7dd36ecaf1458994003b69023ed7d5 : SUCCESS in 21m 31s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8aec5cb6cc3944f89d0c5c9ecf2ca9de : SUCCESS in 20m 21s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/fbbaa801debd47ee846803a1e68a3c59 : SUCCESS in 9m 45s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/673e0335cd7b4a24a61c73d6faef662e : SUCCESS in 6m 42s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/7ce2af22b8b14f56b1d70f6013d84fb5 : SUCCESS in 15m 58s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/1f504cc8094c440e8ac201045f6e1a6b : SUCCESS in 12m 04s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/cea71541822f4b13906db263737cac6f : SUCCESS in 16m 57s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/fb3cf53dfe7e4ecabca44c5f905ca52b : SUCCESS in 17m 33s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/aa9bba3ca8ca4e4eab4771f92bfdaa58 : SUCCESS in 16m 35s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/7f72abf2621c4d80bf948b37c522ea0c : SUCCESS in 17m 57s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/672e6debc222410cb04f4338ce7a7c93 : SUCCESS in 19m 47s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/b73f9c4e27bf4047aa821515ad74d510 : SUCCESS in 26m 17s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/4d71170c262945ecb46dd4c6b124141e : SUCCESS in 15m 48s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/ff5c8c604cd845eebaafba75491f6c4f : SUCCESS in 16m 39s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/0723fcb4734e4622a4bf080f4ba5b72a : SUCCESS in 22m 04s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/2f4003e3f0d04b24be6bff19be432fe3 : SUCCESS in 6m 06s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b6186b3994a646aeb47154d11ab72e52 : SUCCESS in 17m 26s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/4faf2c2a7e044cf2b3fbe2b9dca62b1d : SUCCESS in 8m 01s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/91bbbf4ff9cf4470b214ae4d408ccb4f : SUCCESS in 58m 25s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/205ea0cb9a9a43d1bfd2acdc6769cfc0 : SUCCESS in 1h 05m 00s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/8912d4b2302647ef9e472bf35e9f2515 : SUCCESS in 39m 50s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f5a04870a0a74c89a18ff9b779275b18 : SUCCESS in 45m 51s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7c911cf356414ebd97704e46515e0de1 : SUCCESS in 17m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c33079a715f94baf966c8a48c342e917 : SUCCESS in 5m 35s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/4834203d57604e979d338b75c233574f : SUCCESS in 22m 36s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ed194f7d293f4cddb290f0d0e266e7fd : SUCCESS in 1h 06m 13s","accounts_in_message":[],"_revision_number":31},{"id":"8718237b242498b36051969ae28c53744bf69f90","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-01-11 14:30:16.000000000","message":"Uploaded patch set 32: Patch Set 31 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":32},{"id":"3f80fcf1bd158a8e746a5511b091124a03bf8852","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-11 15:58:09.000000000","message":"Patch Set 32:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0fd8163a332e4b2991e8ada1ec133cc4\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6f0e9fbebc1e40f882474d33d363e058 : SUCCESS in 21m 51s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1a27c3232f934680b2db682b77e6869d : SUCCESS in 54m 06s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/1f5918ab9f0c4ea4b18d758837d18516 : SUCCESS in 23m 16s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/73fe44b70f0a402a862e96dbe8b33417 : SUCCESS in 1h 25m 10s","accounts_in_message":[],"_revision_number":32},{"id":"0d217005b3e927de72f4109dfdcc665e90edffa0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-11 15:58:36.000000000","message":"Patch Set 32: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e50815c433524d64a4c67083ce387c6d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e9c9ebe736254d5b81422a12c36221f8 : SUCCESS in 8m 47s\n- grenade https://zuul.opendev.org/t/openstack/build/2b8d56f6ff474f7091ab0f5430bc91c0 : SUCCESS in 1h 03m 02s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/4335604ffd2940e5bfe35a7091bfb983 : SUCCESS in 47m 22s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/3672846c602e4115a8f391d394cdcc65 : SUCCESS in 42m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3de23f478ddf4851a2f1f413793744ad : SUCCESS in 41m 49s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2a7e728a9fd6430380b01ea522f5aa46 : SUCCESS in 19m 41s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2ee461c530f64ac78e15197c8b8ac253 : SUCCESS in 20m 45s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/499762392b844bb0a0c8c50944acdd8a : SUCCESS in 20m 28s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/19c6bea3d39141d2b0eb947a5704e7bb : SUCCESS in 11m 57s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/3736f993529d441b98abe5697bdd365f : SUCCESS in 6m 13s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/553eecfd944f4624ab517688d2f656ce : SUCCESS in 19m 03s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/2a39625db9a143c09eda3eeff74c6948 : SUCCESS in 15m 54s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/8f37038d38f64acfbd2a32f17da97ac6 : SUCCESS in 15m 51s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/b05f711205a340fa8fa8c114e0d45587 : SUCCESS in 18m 33s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/eb9e843dc33b4ca280e88535ddb5a0f1 : SUCCESS in 16m 57s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/d2dda91b62c44723840caca63b24b17f : SUCCESS in 19m 07s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/f77fecfe978f460e832fc51f16dec7fc : SUCCESS in 19m 17s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/2240e9fc91d240ec974d2f5a5eceab6a : SUCCESS in 17m 56s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/5f032ebbd19c4b0ea71cdcb03fcfda76 : SUCCESS in 16m 52s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/6493cfa3206b4b8ea458652b50345135 : SUCCESS in 15m 30s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/5832463f137b4b9da87273ed3d2e85d4 : SUCCESS in 17m 36s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b3b47467614c4856a9a453a32eeaf5a4 : SUCCESS in 7m 44s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/1bb6778ab4d84e04ad7a57fc19eaf56e : SUCCESS in 19m 45s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/ddf81277efb242c187fa72517b8ed8f1 : SUCCESS in 8m 33s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/f7a9c3a04b9c4a7ca43a4764fd1eb2ea : SUCCESS in 1h 02m 22s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/09cd0e3498144d1eb044ac57cf5c0e1b : SUCCESS in 1h 10m 34s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/25d3d21f48c3448b9c46671b73e1b683 : SUCCESS in 45m 05s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ab06fef609cc499b83ad4e4514bf6e43 : SUCCESS in 1h 01m 50s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9c4a7dc03b0241358d906f6a1e3b33c7 : SUCCESS in 19m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aa5b3f784c4944d98cb663af2a8f94ba : SUCCESS in 5m 14s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/2faa29c4a7d0401c910d4f06139ab16d : SUCCESS in 22m 09s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8d117198737f402fa52cd0e2304debba : SUCCESS in 44m 16s","accounts_in_message":[],"_revision_number":32},{"id":"d813622596130874558a16025daa22a1234aba01","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-03-20 19:09:29.000000000","message":"Patch Set 32: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":32},{"id":"b8646d93ad0d33a29cc8013484bfa10f9f499286","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-03-21 10:48:42.000000000","message":"Uploaded patch set 33: Patch Set 32 was rebased.\n\nCopied Votes:\n* Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":33},{"id":"8f0b389e7884cde6d26c064b727f321725f2b433","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-21 11:46:30.000000000","message":"Patch Set 33:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f92627b5326540cab0c152400db3ab14\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/efe2b36a277c400e96cbe740f88602b9 : SUCCESS in 14m 36s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/41aa398ca77f43049e94234a17eabff0 : SUCCESS in 49m 02s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/8ac834e8eb0f4e45a56cf6c46cd32187 : SUCCESS in 18m 13s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/7ebed3f476564e1aa55c39c95320a83b : SUCCESS in 55m 53s","accounts_in_message":[],"_revision_number":33},{"id":"6c80d74e3f7864f1280268dcd4a88e9aec411747","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-21 12:14:44.000000000","message":"Patch Set 33: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a94ad8aaf43a474d819b91595f3f0c16\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e66f85e4956f4f9bb1a42aa185fab37d : SUCCESS in 8m 43s\n- grenade https://zuul.opendev.org/t/openstack/build/2e04fc4414244b0a9fbad436298ab824 : SUCCESS in 50m 10s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/a288953617154bd3901b6f52b21ea5bd : SUCCESS in 1h 00m 52s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/6176993960394bdc85afba8217163e8a : SUCCESS in 50m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/734f10aef5a4471595029f86576429ec : SUCCESS in 59m 13s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/703afe21a796487f90120f0df403560a : SUCCESS in 24m 29s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b49e5603d92847e5b4eb5e395c72cb08 : SUCCESS in 24m 57s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/27baa7d3a5e74eaa85792d9539c13ead : SUCCESS in 18m 04s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/532ba6ff4b1243a988d51203a19e7c8c : RETRY_LIMIT in 2m 21s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/df6cfcc7fdb846bcb6c0b8362804c52d : RETRY_LIMIT in 2m 28s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/e12049e018fa4cef9d42e9debddd2d01 : SUCCESS in 21m 15s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/cb7f6f6ee854410aa229318c53fbde6a : SUCCESS in 18m 13s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/f20c7726f8934daea3f2798b193630ea : SUCCESS in 16m 48s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/8f66aca9d2594b43b67d03c6f8c5842e : SUCCESS in 14m 49s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/f436eb455f364c49a2c1711c9453e0ad : SUCCESS in 19m 59s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/063a3f35797f40ac968e02cdb4e3e885 : SUCCESS in 18m 14s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/8ee2ed1e8e7946ec84036ad523084187 : SUCCESS in 19m 21s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/23e9bdfdb51f49f19891863b67b5c6f4 : SUCCESS in 12m 47s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/5a6d7e713c3641e8848fca0a686766b3 : SUCCESS in 18m 05s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/a6048a618d82493696da4401fcd6363c : SUCCESS in 15m 47s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/7254f9d8745c47d8ac68518152c0d11e : SUCCESS in 6m 23s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e4b1d8c126174a0782b2aa8e47219811 : SUCCESS in 16m 35s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/a7dfb4d5c7c740bb947d2718ea680b31 : SUCCESS in 7m 19s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/0a01013736934a4fb37380ca8056e755 : SUCCESS in 1h 01m 37s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/2621adf9e879425fb5d51f5a34d2f6a1 : SUCCESS in 53m 17s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/0435f0b0853c4196bf31dbf2ffc70d07 : SUCCESS in 52m 54s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8039c2e2965b402396e00c68d75107bd : SUCCESS in 21m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c496660116ac4c59808c4c942e13c242 : SUCCESS in 4m 05s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/7ddd4545a3fa4b3fa29354aae43d8ccf : SUCCESS in 20m 41s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/525e0baa78b04e6ebecee4d5c55b468d : SUCCESS in 1h 02m 42s","accounts_in_message":[],"_revision_number":33},{"id":"c10bf1aa97d761a17ad46af381321a03075a55eb","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-04-01 14:05:21.000000000","message":"Patch Set 33:\n\n(27 comments)","accounts_in_message":[],"_revision_number":33},{"id":"0444eac59cc1aa05389de054baa0905b9fcd2db5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-04-01 14:48:16.000000000","message":"Uploaded patch set 34.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":34},{"id":"dd9fb6ac0c8bdb1218dc438a51e93efbe19edc2f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-01 16:21:08.000000000","message":"Patch Set 34:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a09373ed7b094e1a8a17fc830274fcd2\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f2ef255b5df94d6da3e948ac9bc5c68b : SUCCESS in 14m 48s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b3fcd2d5084e41269d692e436a508871 : TIMED_OUT in 1h 30m 52s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/76a65d4d789f460086c6b4bc18ba3d30 : SUCCESS in 22m 28s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/5d39b5059aac456d886e5db5f31af3d0 : SUCCESS in 1h 25m 02s","accounts_in_message":[],"_revision_number":34},{"id":"d837afda7f3672a74be5135beeeecd2df2ef6e88","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-01 16:32:01.000000000","message":"Patch Set 34: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e776c42d4eae40f9b49957c1739bafcf\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/742cb1fcac94483e98ff52103ffa1118 : SUCCESS in 8m 25s\n- grenade https://zuul.opendev.org/t/openstack/build/6f047f82c8164521a51d798ada83957d : SUCCESS in 57m 50s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/85267f301bd449f289fdced6370e90a0 : SUCCESS in 56m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/40649660cfdb46bfa1259a7b434c2009 : SUCCESS in 39m 47s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f980c41c5c264e33b5118c2f6c5128f8 : SUCCESS in 19m 24s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8fa5e2cb289c4199b9061d6181bed6c6 : SUCCESS in 17m 25s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f6019a95d09a40ecb31a189d7f0dd12a : SUCCESS in 19m 19s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f1234b2fe867435aa760de0c4818de06 : SUCCESS in 8m 51s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/832b6ccc82c64ed882ba1d36d16dcd5c : SUCCESS in 5m 02s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/183b71af6e124207ba63ecd90088deba : SUCCESS in 18m 08s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/e37e1d22bb7a4766984bb2834a25e503 : SUCCESS in 19m 19s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/ef8d6da736c74e6c896475521248a716 : SUCCESS in 16m 07s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/d27f6895817f43e3b95680ace9746c19 : SUCCESS in 16m 10s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/c496d432a1904a7b91cebd6d1378a5aa : SUCCESS in 17m 51s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/9c33bd12c9ba45c69d63217a7e4ca2d2 : SUCCESS in 19m 01s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/b6291cbe4529484399d1445a55116ac4 : SUCCESS in 18m 40s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/d74f1e98b2cd43e3b43e98126eedc2b7 : SUCCESS in 15m 17s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/1780ecf852be44b2b1e5119262d4d6a8 : SUCCESS in 17m 21s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/2799eced8f4c46c794c0d7758966ef62 : SUCCESS in 17m 12s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/6137c69686514c269aaef3e13ed9d52d : FAILURE in 5m 41s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/9d8b8375d251486bbb0c20c974ef8954 : SUCCESS in 15m 08s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/2de85a3912d64a4abf1f1631a68167a5 : SUCCESS in 7m 39s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/c4945f72cefa4cfebd1e8ddb1f8344cf : SUCCESS in 47m 49s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/996381052a284ee6a8fb55cf699acabb : SUCCESS in 41m 29s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4192ba4b499d439d8c9759cf83fc97c3 : SUCCESS in 53m 49s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/be4b198c825a4fffa5795e46629822f9 : SUCCESS in 21m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e7058eeadafe409b9f907a2d42abb133 : SUCCESS in 4m 04s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/76c0ffc306b04f8a8163a066419900de : SUCCESS in 16m 31s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a5c3cb28862f4f2183a74f99f89425f6 : SUCCESS in 36m 45s","accounts_in_message":[],"_revision_number":34},{"id":"70b8c0d3f18ddd36bbb3ee048c3dbd81f9ff97cb","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-04-04 18:18:09.000000000","message":"Patch Set 34:\n\n(8 comments)","accounts_in_message":[],"_revision_number":34},{"id":"5b86a4a7e9c5868b150beaf97b6b443927bed67d","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-04-04 18:34:07.000000000","message":"Patch Set 34:\n\n(1 comment)","accounts_in_message":[],"_revision_number":34},{"id":"8e0d4617ca9854d94ca05570003aed1e8adfcaad","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-04-04 19:07:05.000000000","message":"Patch Set 34: Code-Review+1\n\n(10 comments)","accounts_in_message":[],"_revision_number":34},{"id":"84e94ce2d1a6192a6edae06a41a54dc9a22b585d","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-04-08 16:36:59.000000000","message":"Patch Set 34:\n\n(1 comment)","accounts_in_message":[],"_revision_number":34},{"id":"c9edf1cd82cd170772dc636f4ed0729d5a2a6e7e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-08 17:38:38.000000000","message":"Patch Set 34: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ef25898c23284df9a583a94250cc8cba\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65aec9baf3c84f49a85f7fbd0618c420 : SUCCESS in 7m 05s\n- grenade https://zuul.opendev.org/t/openstack/build/e06912c662e348068d6f0f8cc967c600 : SUCCESS in 51m 47s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/7d692a781654453e8c813424593c2cfa : SUCCESS in 38m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/949c9ebe1da44f929618e713df9dbe94 : SUCCESS in 50m 42s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e3589da2c88a46448c9e8b0eadc56a9a : SUCCESS in 18m 42s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b3b6b61b9f174d10a52437b3775f5b60 : SUCCESS in 22m 25s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/74de1e013dac49dfbfcdfb301ded47da : SUCCESS in 19m 48s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/28c22fb51c83442ab8a1d1043a65fb9b : SUCCESS in 11m 18s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/90ee8387af5e4f5799bf589ff103b310 : SUCCESS in 6m 07s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/413410795cfd40699402260ac3d0fbc2 : SUCCESS in 21m 11s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/97491a0fe6544e27b3f851437f8ef529 : SUCCESS in 18m 50s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/43776f90d97947999c330ba7969c3bc7 : SUCCESS in 21m 00s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/d7a000a4a1ad40b2be099bfc91ec677a : SUCCESS in 19m 20s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/6ad2f82eb2ed4615a291f5d466e27480 : SUCCESS in 20m 32s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/90f1718b16934d85bd7e003f9b8eea83 : SUCCESS in 19m 25s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/32ea1db1415a4059a028fb3a992258b6 : SUCCESS in 21m 59s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/28cd815651394b38800849f3f38b8137 : SUCCESS in 16m 10s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/05ec41e4db62483294173ba537a63fd8 : SUCCESS in 20m 23s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/186ecd39fcea4b81abe968255a5a34b2 : SUCCESS in 16m 26s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/7a1cab26a26145b3a14facca4db5532a : SUCCESS in 5m 34s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/dec6dc16cfa6437482111170704c2c0b : SUCCESS in 19m 10s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/a55d4a09b9db49459fa08ab8e89aeed2 : SUCCESS in 7m 19s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/d295710eb6d6417bb05575ee861e0a3f : SUCCESS in 54m 39s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/25156f7a1d8e4927be22cba1d64ed21b : SUCCESS in 40m 13s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f9b0bdf3969d4c45848b385219b4992b : SUCCESS in 41m 08s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6a98732638074c6aa91e6de903ad1717 : SUCCESS in 19m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/349f9f4a4204463f9eaf233f4517a04f : SUCCESS in 4m 05s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/13b12904da814f339cb1b38b223f8522 : SUCCESS in 24m 13s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ef55dd0863ef4af8ae16be1ddeafaa0a : SUCCESS in 44m 24s","accounts_in_message":[],"_revision_number":34},{"id":"8d9d1742e21362789043bb4c8f136ccb459e8629","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-08 18:02:37.000000000","message":"Patch Set 34:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fa8f4f6471ca40309c7000995043b2a5\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1b362b6897ea440fa38c40c7c053d4cf : SUCCESS in 52m 52s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/86a161834eaf4d7691c90808fa2800b3 : SUCCESS in 1h 19m 44s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/be97a9bfed5043c3b0dee47c70e50bf9 : SUCCESS in 1h 12m 24s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/5c8f7f5f05cf42d29d8ec9659d9215b2 : SUCCESS in 17m 48s","accounts_in_message":[],"_revision_number":34},{"id":"eb0ad7b5c3601a660332d197f72681c5e70572cc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-04-09 19:31:47.000000000","message":"Uploaded patch set 35.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":35},{"id":"d2df8a14f5ae5c5c016c3312d93941ae735cecca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-09 20:57:17.000000000","message":"Patch Set 35: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/31dd7d617d4a4aa69f351a5e9c69ee62\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0da4e6d04a484b3e96fa38704bb97908 : SUCCESS in 8m 45s\n- grenade https://zuul.opendev.org/t/openstack/build/1da8d11b4af14e238cec92600bf82dd3 : SUCCESS in 1h 03m 05s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/73d45235b3c34f2eb3e2300a280d2b5f : SUCCESS in 50m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b45aa6495b924281a055d2eda8d6484e : SUCCESS in 1h 06m 21s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f2bf1005fc234b5299254e5715c21fdf : SUCCESS in 18m 36s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b46cada6d1ae4b00a02cae6f65ea2c43 : SUCCESS in 20m 18s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8a5f0ede636a45ff9e591268111b0124 : SUCCESS in 19m 27s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/2c7069b820ea428b8284f754a2935cb9 : SUCCESS in 10m 37s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/7acd93e1240a41049f7d2fe2cbcb1cbc : SUCCESS in 6m 13s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/5a30594b7c844791bd05ff97bbef6fa7 : SUCCESS in 22m 03s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c2e613b2fdfb4aa7ae39634cc2800daf : SUCCESS in 20m 44s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/1fb17b652dc04356862a7cf59d9f7565 : SUCCESS in 19m 19s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/ab810633f8624a859fd98b9796cec47a : SUCCESS in 19m 19s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/ee03b1af0ff845fb8ea1f82b9ddd15b5 : SUCCESS in 18m 10s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/e0c4eafb48284117990911593559dc8a : SUCCESS in 18m 29s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/b409dd2018a743eca20ded26066011a9 : SUCCESS in 19m 04s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/7a86d61b026745bab10a3913e51ffbe9 : SUCCESS in 14m 45s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/6926e1d10c0a4c2eaa8e5f1465842be8 : SUCCESS in 15m 49s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/b9a9eda47cc94df2b5defa4d1b2fb193 : SUCCESS in 16m 23s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/0493d29ed9354b2eaf4ae46a4f58f5cd : SUCCESS in 5m 33s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/a86d7ce181de4ca9b39954653bd74faa : SUCCESS in 14m 16s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/a859d0a2f3ef499babcd3a6e44e0b552 : SUCCESS in 7m 15s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/95d94f6c5fc74bc0b091850a6387c65e : SUCCESS in 48m 06s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e316be2b0c7a42b7b32dbc51fea949de : SUCCESS in 36m 10s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/52a221bd4041438fa67e47caaecce256 : SUCCESS in 55m 58s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e44b787f89434feaa685e936dc6cb466 : SUCCESS in 26m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f9237fc7f984397aa340494ea3698b0 : SUCCESS in 4m 20s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/394a190ecd7a4fc99874ccf6927e5571 : SUCCESS in 19m 21s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/23448a1d0bf9479da69d6ecaabaa3ca8 : SUCCESS in 57m 27s","accounts_in_message":[],"_revision_number":35},{"id":"47ee709a797988fab018ec7b863d8dd51c8bef0c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-09 21:04:36.000000000","message":"Patch Set 35:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a3e937ee40d7441187133e2606830048\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/37f4939909d243619dca1d6d5a193fee : SUCCESS in 1h 13m 27s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/46a29f91f36142edbd46ebd8e458658e : TIMED_OUT in 1h 30m 57s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/d7b1caab220143cb95ab7958eb5ece3d : SUCCESS in 1h 10m 01s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/745d1d71c235467fb0da5be7d9f81fb4 : SUCCESS in 21m 38s","accounts_in_message":[],"_revision_number":35},{"id":"4204bfe10f8db4ea789cd1279a8915dcd5a0ac5c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-05-01 18:42:11.000000000","message":"Uploaded patch set 36: Patch Set 35 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":36},{"id":"23b7b03c83d4372687fa082ccf4792a967274a64","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-01 19:59:00.000000000","message":"Patch Set 36: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/486bcfb26e11438b94edeaf7fbd6ce2a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8ea66f371634b3983e337ba07c08332 : SUCCESS in 7m 03s\n- grenade https://zuul.opendev.org/t/openstack/build/58fa7f87076444788ad54c1d1d221792 : SUCCESS in 1h 09m 05s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/512b3c8c66094fa197d5afb2eb66edb2 : SUCCESS in 49m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/86ee686c2705481d84a70dc42a5dea3c : SUCCESS in 1h 07m 50s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8f9a2256b8df46b794f729a162bf59b6 : SUCCESS in 22m 42s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f409d7af487c4fddbe9841d40d705e2e : SUCCESS in 19m 41s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/446d0e86a0ff4406a5783aa8694024c1 : SUCCESS in 18m 57s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e5c6bad847b04629aa7fcdfe630c8ae3 : SUCCESS in 10m 20s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/9694968182b045b8860873bb2ff0c6ed : SUCCESS in 5m 15s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/8f661c6b2c0a4caa981122863bea2c49 : SUCCESS in 22m 07s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a55c00645dc1488f911f2219667991d6 : SUCCESS in 18m 18s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/248a3184cb584d9f86497c0039fcae37 : SUCCESS in 17m 15s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/666190993b4b4a9aa1fef8952c90b8e8 : SUCCESS in 17m 42s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/c0a5e347dcf3478fa66acafbf388ca3b : SUCCESS in 15m 18s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/6906d99a3c09446f81ac42b4503b5ffe : SUCCESS in 20m 41s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/391fc32f4c8a4145ac153b67effabdcd : SUCCESS in 18m 36s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/8bda8074a42145b489ea466ac10c3546 : SUCCESS in 18m 12s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/0c8b5eeb6ade4c8a9cca33a669e794bb : SUCCESS in 16m 43s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/035b198008ef47fbb4cdf81a66ed9aed : SUCCESS in 16m 58s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f872a15e7c764096b8b814872969accc : SUCCESS in 6m 56s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/69239f43393b4737a96449a2b9891067 : SUCCESS in 17m 09s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/d91760dfcf1c474e97e63fb86d297f3d : SUCCESS in 7m 14s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/757cee769ede48e7a14c97c5e08bdac1 : SUCCESS in 54m 54s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4e22cce278104eab9553c08ea18737da : SUCCESS in 54m 35s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/703bc09d79c6427ebfb99caa6bc6f8d5 : SUCCESS in 48m 45s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ff3508090a3641ec86f0f9d31dd95036 : SUCCESS in 18m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/09043ae4d02a4c9190d4418e58eecdf3 : SUCCESS in 4m 36s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/7560b7bcf56e4669a98c77118101986c : SUCCESS in 20m 01s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4d0d1cd5edaf4a949f1d54fafc1f601b : SUCCESS in 54m 42s","accounts_in_message":[],"_revision_number":36},{"id":"826bec3aa347b9717fb8e2068128ede31e5b5e7d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-01 19:59:25.000000000","message":"Patch Set 36:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4d6c0459de894602b2cb08aec8dfc2dc\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4d15b5103b644de9b820e86df6f17433 : SUCCESS in 1h 15m 32s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/acf16a3875224c9d912f8aef5f8b177b : SUCCESS in 17m 40s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/b0bedb7440c448c396c310767b574ee0 : SUCCESS in 44m 04s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/7295eaa328f6470b861d890e11b7f36c : SUCCESS in 19m 15s","accounts_in_message":[],"_revision_number":36},{"id":"5b0e534595f723ed0b441cd9e6381f83327deafe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-05-21 19:23:09.000000000","message":"Uploaded patch set 37: Patch Set 36 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":37},{"id":"0ff7fdb710396eda08b0d05e3a056b512e3d13df","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-21 20:31:43.000000000","message":"Patch Set 37:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c7dfe4b6fd1e42f0be5dcfd18e0f7ef5\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9c4e754b6a9043788cce2d80ef45edbc : SUCCESS in 58m 22s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1601a661537d4a5bacbaf7548f15cc32 : SUCCESS in 1h 06m 39s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/9cdaa572770c49238664b7bef9228876 : SUCCESS in 56m 32s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/26f7133619dc4c4fadc573d6399d34ca : SUCCESS in 45m 55s","accounts_in_message":[],"_revision_number":37},{"id":"4791e52d82bca7fe86ef09b3ce8365194c49505d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-21 20:37:05.000000000","message":"Patch Set 37: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6e8e9daacc3d4c5dabee794cb476aac5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b59f6f832604bd0b1d1336dbcd66d11 : SUCCESS in 9m 29s\n- grenade https://zuul.opendev.org/t/openstack/build/6b7bba6008d442b3b3a829315a257d12 : SUCCESS in 1h 08m 13s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/ed5ce7f5df2b4f0e97af09221761287e : SUCCESS in 51m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/75537998dee34b72bb62776f7acd9b11 : SUCCESS in 50m 08s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5373817df10a44f3871f6bb07a1eadd3 : SUCCESS in 14m 57s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d217352e7ebb42ecb3515300eb92cf05 : SUCCESS in 16m 57s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e3125576d1cf4adc94efda68144aa727 : SUCCESS in 19m 32s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/81db5cf0b98741d8b72a32efc61f1cfb : SUCCESS in 10m 31s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/ba5022b8277345f3966b24a4ae85a009 : SUCCESS in 4m 57s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/761ba38a0bea469eacab60e14ab2d2e3 : SUCCESS in 20m 07s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/092b95f17be940d3a7c216789f8e688a : SUCCESS in 21m 33s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/42c8e98dd8334032a4145b94ab8b9dfa : SUCCESS in 14m 50s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/5c365e69df3b41f785a5fe15f4de9568 : SUCCESS in 17m 23s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/36c78fdc86e64de8892917fe9ceda641 : SUCCESS in 23m 02s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/e1b5a8fb2f4d4457b5b95c86efe6ba63 : SUCCESS in 18m 21s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/1d37d131974f4608a0330e0c7b77fd77 : SUCCESS in 19m 44s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/4a3ca306050f4927abc012298aa2b906 : SUCCESS in 18m 14s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/47e352471306405d899282914c616f50 : SUCCESS in 15m 27s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/5a40c20d733d4f84a6c1d1f44546106a : SUCCESS in 16m 15s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/fa0af720cb7441f9a3ac8c72556828ba : SUCCESS in 17m 09s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/775b979b991b431bbe52166b8ec22cef : SUCCESS in 8m 25s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/70fa56963c7b4166a40a04a9264e2dde : SUCCESS in 15m 20s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/7a75a6e4d2d343479909d3e0505ad695 : SUCCESS in 7m 14s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/04bdaccf959d4d00a895327f9c97f0b7 : SUCCESS in 1h 07m 40s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/0c59fb9946744267a2c547da64780060 : SUCCESS in 47m 40s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/89b97bd3f65345d1a174d627f91d0330 : SUCCESS in 50m 48s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6134d6ecb1bf43bb9736eee82a34c9be : SUCCESS in 18m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/19da3bf868eb4fefaf28d3cbf8f39380 : SUCCESS in 4m 07s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d0ee65533a8d4f9d95f93c1dd4fb4207 : SUCCESS in 24m 33s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bcb8905974c74cde99669ceb02686438 : SUCCESS in 43m 46s","accounts_in_message":[],"_revision_number":37},{"id":"26975a03164c57bc306492cc7481e0d003618386","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-06-03 22:24:09.000000000","message":"Uploaded patch set 38.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":38},{"id":"3f934badc6bbdc542622842029e6bd39d2109068","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-03 23:34:15.000000000","message":"Patch Set 38:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bd17cb395c1a4f5d8b31c6aec6e37469\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ba47f0721d0a4509b2c16c8f36314270 : SUCCESS in 53m 57s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f65bc40964994fa29bed460cb3e285a2 : SUCCESS in 20m 20s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/8d94497fbd194a629b2aee3cfb9e64b2 : SUCCESS in 1h 08m 15s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/f2f7130dd3fd4305a367458897a38bfb : SUCCESS in 21m 10s","accounts_in_message":[],"_revision_number":38},{"id":"34c8a1f8d8d56c01bd61151aee25269574977075","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-03 23:42:47.000000000","message":"Patch Set 38: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/933f7d4adb7849caadf19a44cf518e90\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a5c760a4ddd1432f885bff18a4f84df2 : SUCCESS in 7m 14s\n- grenade https://zuul.opendev.org/t/openstack/build/675da48488e14564bb8e5616639adec7 : SUCCESS in 54m 38s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a2ac202dfe7e499da8021598401b1d21 : SUCCESS in 47m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8baa5692044c4632871db3a818b917cb : SUCCESS in 1h 12m 48s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e2820a3823b541dfbd3821343a4c5861 : SUCCESS in 20m 21s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f543b6eb8b5e478c817dbe72e2fab18f : SUCCESS in 17m 11s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/eefe3ed4daac460ea3cfd82177611b14 : SUCCESS in 19m 26s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/91632681e5f94096ad7d75b622eb9a9b : SUCCESS in 12m 37s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/d54d3bb343c54e3280bf5f9caf18da4b : SUCCESS in 5m 12s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/37647cd8741c46db82ed35466328ccf1 : SUCCESS in 22m 08s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/2cf23b5b5c1246109b2f232e38963bc4 : SUCCESS in 12m 24s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/23c2a2ce186e430e904dc19d52d7eb67 : SUCCESS in 18m 53s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/4593db50c1b94fb3b3c19f4f4001d406 : SUCCESS in 27m 24s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/14b88eacb62d480f97139b7b5fed871a : SUCCESS in 17m 17s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/d18d997508b84ed9b89b64a338f04a54 : SUCCESS in 18m 20s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/acd3a0c74cfc4142a34d2a8bd73ad5e1 : SUCCESS in 22m 48s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/cb7f09681da942c9bdff328e31f53e72 : SUCCESS in 18m 32s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/8e71846c0b6642bda4649392cb827eb2 : SUCCESS in 15m 29s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/08659e845acd49c38ace78b89db25c8b : SUCCESS in 16m 16s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/ae0ac7626ad24228bcc607c31a8494f5 : SUCCESS in 16m 31s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/5cac09262e7e41c7b751e3adcea3fe94 : SUCCESS in 6m 33s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e508426e07624f07a4f51fcef07ed893 : RETRY_LIMIT in 1m 52s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/5e5220c34ff04d29afd9e4e9b326aab8 : RETRY_LIMIT in 1m 30s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/4a8b9a56f4a748f4a48f4d7f34081947 : SUCCESS in 53m 04s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c626805bf3c44a479deeb437deeb6c4c : SUCCESS in 55m 54s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/30e5523743a444a9b807fd023d9cbc22 : SUCCESS in 35m 17s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/425a08e05f2f420383160708ab14ee10 : SUCCESS in 17m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9d48bff58ff546f8ae4de037b375b7a9 : SUCCESS in 4m 03s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/dde2d6033a05469186a96a531eaa0fec : RETRY_LIMIT in 1m 57s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d463230cf0634f4393bce5df79cecde3 : SUCCESS in 54m 52s","accounts_in_message":[],"_revision_number":38},{"id":"05b6f72683e8b5862bbafeda6a58be25cfacb5da","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-06-04 18:11:10.000000000","message":"Uploaded patch set 39.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":39},{"id":"5d628bfc01b55516a97f4bdfd4122fd228b24093","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-04 19:15:15.000000000","message":"Patch Set 39:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e91890909fde4650a12d22fcff4ac2cb\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/50a3003f5abd4e25adac3f8395d2721e : SUCCESS in 17m 05s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ee9671c0c14d42069f5ada3b939c631e : SUCCESS in 1h 01m 51s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/77b34b0bd2564640941e20a4be6c34b1 : SUCCESS in 20m 16s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/aeb631d453ab4b8e97915eabeb27685b : SUCCESS in 43m 45s","accounts_in_message":[],"_revision_number":39},{"id":"a1efbbac62aebd22883ce6f209d2c8b585ddda25","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-04 19:26:51.000000000","message":"Patch Set 39: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a46eed5cc4a041f0b99294fc37016ea5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c25d0b15d1e94e15ab17768fad6335eb : SUCCESS in 8m 47s\n- grenade https://zuul.opendev.org/t/openstack/build/420b7adf53f64ae5af059ac8dcd4784f : SUCCESS in 1h 03m 29s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/7dd2bf1563674294a78dba8c6ff4f69e : SUCCESS in 57m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ab2c3965063648db9dde0f40a8808fed : SUCCESS in 1h 02m 13s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7ab68912bc9641c69487426e1aca282f : SUCCESS in 17m 25s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a1214a7008314f3a972abc8f3d0a074b : SUCCESS in 23m 24s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2ff36b4a463244ddb2861fb5e806a385 : SUCCESS in 18m 40s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/aca28fc16d3d42b6af6453cbe7a1fa6e : SUCCESS in 11m 11s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/06f6918c233d4c6eb261b53391a527fa : SUCCESS in 4m 48s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/50ea8022a32346d2b548d8786eb9d703 : SUCCESS in 18m 45s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/7930342328df4086a7253d5b3be5230f : SUCCESS in 18m 48s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/69707ec9752145fcaefc747588048246 : SUCCESS in 14m 59s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/241a52036de045e8bc4df36ec4c3c35d : SUCCESS in 15m 09s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/d0026bfc6eac4c52b8f64ee9d6561416 : SUCCESS in 21m 50s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/a5d925588c484bea953b068f5099da1d : SUCCESS in 16m 44s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/08cdb3af105a48199fcac5b7668f9f30 : SUCCESS in 18m 07s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/bfa3f8b05c754064acc75ca616d62a33 : SUCCESS in 17m 46s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/99fca9d8f9be4e0ab14e9b223e63dd49 : SUCCESS in 14m 44s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/5a3e5ec8501943608e28f947667d8253 : SUCCESS in 15m 19s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/1cfdea47e84e4091964abe981a84f865 : SUCCESS in 15m 49s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/2b2e5c47d9244500a53f723c7620a8fa : SUCCESS in 6m 36s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/45b1c12d56384d7ebc9c840b421f2222 : RETRY_LIMIT in 1m 53s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/e448c482df214985ae2807eb488a2163 : RETRY_LIMIT in 2m 50s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/3d42052176594fd995274201df50b9c1 : SUCCESS in 54m 51s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/0de9c7db14ee450991876c2359765c61 : SUCCESS in 53m 33s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/70e271659d394e608296f45ea205ea52 : SUCCESS in 1h 07m 33s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6089c020fae04e55a46dee616b2754f2 : SUCCESS in 15m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/076316d5073640a69a428d1ab97b9205 : SUCCESS in 4m 19s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/2d1927a550ab4af5b83d0570d37e6463 : RETRY_LIMIT in 2m 10s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/74449f9f48f64ca08b6b26b9f017aa7a : SUCCESS in 44m 26s","accounts_in_message":[],"_revision_number":39},{"id":"8aa9e917075e438063db4fa77b6b4e1435210c26","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-06-04 20:39:34.000000000","message":"Uploaded patch set 40.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":40},{"id":"6ae1792e323539e246b733015cf913fa12e9b355","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-06-04 20:42:36.000000000","message":"Patch Set 39:\n\n(8 comments)","accounts_in_message":[],"_revision_number":39},{"id":"d8b169d1dba69936b1edf58d65fbc2727774680c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-04 21:00:36.000000000","message":"Patch Set 40:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c15cc63431cd45778b1981bf7b5de90c\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c4ef06c175754aa68daf7aa95c0a5cad : FAILURE in 18m 54s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/445c680573b44025b6f85c7e07c2b559 : FAILURE in 17m 40s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/1c324e29a775421585fe93943aed232f : SUCCESS in 18m 20s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/f415066cee7340d2a5179637a226b3a9 : SUCCESS in 19m 29s","accounts_in_message":[],"_revision_number":40},{"id":"9fa980734e0135aee129fe34e50cfdf48e97952a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-06-04 21:05:15.000000000","message":"Uploaded patch set 41: Patch Set 40 was rebased.","accounts_in_message":[],"_revision_number":41},{"id":"42f3f95cce24cfba258313b6fcaa003db6dd1c79","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-04 22:09:09.000000000","message":"Patch Set 41:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/57d4c37bfe0242178fb99e8c55b3b1d1\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/638fa60aa2984ffca316c530d568e2ee : FAILURE in 18m 25s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2fbfa9565c344466b81a37084adf29ee : FAILURE in 49m 08s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/562fb4ac392c467cb6d7a5fc5b09e94a : SUCCESS in 22m 17s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/904063a1631b4a6e853ed266da94fcb3 : SUCCESS in 1h 01m 30s","accounts_in_message":[],"_revision_number":41},{"id":"08694c9d31bc4331c5b7a05b2ffc4a8961910f3a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-04 22:19:58.000000000","message":"Patch Set 41: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1cec463d6b5f40868fa23881ea4209de\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c5d268152b2f4ffe8a17f5750710d353 : SUCCESS in 8m 34s\n- grenade https://zuul.opendev.org/t/openstack/build/6645604f1f67472aba8ea1ea661ff322 : SUCCESS in 55m 42s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c188eaceb1f24112a2044837ac85f22f : SUCCESS in 1h 00m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/004aa44fb77b4c8b9715858347890a82 : FAILURE in 1h 08m 37s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2d38256d06884c3390c4382eec0f24e1 : SUCCESS in 19m 03s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f4fb8163fe8247c59c024a45bc52c68e : SUCCESS in 20m 59s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0f595ba328d84663ae053cce7f60e50e : SUCCESS in 20m 33s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/2b54ef5aa7d34f95a23a4cf46799ce93 : SUCCESS in 12m 21s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/d762b0f32a8743a7b958fe83ebc8cb63 : SUCCESS in 5m 19s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/2dc71b8a69c2497eb274538141ef5907 : FAILURE in 17m 13s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/5bb7273f8c574594a303b9ae5437abc7 : FAILURE in 16m 55s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/828401cc22f94a2fb2b16efa9515317c : FAILURE in 17m 26s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/f303eaefbd174385b612d502747ce898 : FAILURE in 18m 43s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/4602d0ddcbb748b7818e1ac93156eff6 : FAILURE in 21m 46s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/d0eb6a0749f84de3b01d8feb9b8fbcd8 : SUCCESS in 19m 15s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/24a40f0ccffb42339eaaec34be539332 : SUCCESS in 18m 18s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/95086cb4fa574d32bc2830ce82490636 : SUCCESS in 19m 04s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/f1e72f19d4524019914628446a8812c4 : SUCCESS in 20m 07s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/8476e28fe5664a8cac62aba1f7e80fff : SUCCESS in 19m 28s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/9cd29695c2694c488d1f8b2b4b7bdc93 : SUCCESS in 17m 23s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/ef657894706e4e7fa98885151dd2ca20 : SUCCESS in 6m 47s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f17f93f7c40b4bdc9ee80ad3f5e08cab : FAILURE in 17m 51s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/7210faaf4d0c41ae9f582d1594448343 : SUCCESS in 6m 37s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/e65e32b065ff44c0801bdfcc524c1a53 : SUCCESS in 50m 03s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a42f0bd7d0344c99bff0a72503831639 : SUCCESS in 34m 40s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/bf4d3925dc5148189e31282b9e20f3ee : SUCCESS in 51m 10s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d6cc8cde7f944de29a86baf4291e2906 : FAILURE in 20m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8ff0481da1174e4083cdf44f4e21c836 : SUCCESS in 4m 40s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/af8b6e8292a743cbaed6658004fda5e9 : SUCCESS in 18m 02s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3431c9b6cd9b4d13a5a086f0e4c528b2 : SUCCESS in 47m 56s","accounts_in_message":[],"_revision_number":41},{"id":"190e4fdf4a3423c78a0b9a6435b71afcbe09b1b1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-06-13 23:07:27.000000000","message":"Uploaded patch set 42.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":42},{"id":"2c2afbd7efd01826aa1c97f78de9ce8d1f06efed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-14 00:18:27.000000000","message":"Patch Set 42: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/71d27af573244ffdbfe0e8204e67789c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/20b313f7827b4a3084ac754acf2f7348 : SUCCESS in 7m 51s\n- grenade https://zuul.opendev.org/t/openstack/build/a44bfae1b89140b694446584c6e9d9e0 : SUCCESS in 1h 04m 40s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/9c91c8598a2c4ff6955728273bc839eb : SUCCESS in 48m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/47dc673cd2ad427b915709725a29ae10 : SUCCESS in 50m 28s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a0832d7a1a104b5b89d896c3e82263fe : SUCCESS in 22m 06s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/00632aafe0374fd889344e232915e450 : SUCCESS in 19m 14s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/81e3ae8cf55a4675aebd8861c5fa267d : SUCCESS in 19m 33s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f8fbe36361fb4519a8048d29a0fdb48c : SUCCESS in 10m 07s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/87eeae2c555242e0b57dc44dc0bf3ee5 : SUCCESS in 5m 29s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/6e558c1eb7e64b2f804cab32effa34e2 : SUCCESS in 16m 59s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/0c31e1b0f1ec4f93921a548adb21315e : SUCCESS in 15m 07s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/384eef8202234b73af7f06339cb9620a : SUCCESS in 18m 14s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/039cfa86233244748a09c98f64505a85 : SUCCESS in 17m 18s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/a60500278c374bbc83245a115b1fe4b6 : SUCCESS in 22m 39s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/6831620f1974475fa5ae9cb279241699 : SUCCESS in 17m 43s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/382d4002c7064e888d3264513c42c294 : SUCCESS in 16m 44s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/746f43880803445d9fe0cb80e1fb75ae : SUCCESS in 22m 25s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/503d46df95cf49b7b73ddf010af5fba0 : SUCCESS in 15m 02s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/c467de2c2f4c492db10ef101d04a6bc3 : SUCCESS in 15m 45s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/853ac6628ca24e11a3a039885bf7140d : SUCCESS in 15m 16s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/0435dee70c704471bbc7f3a5a2f0839e : SUCCESS in 6m 16s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/ea46bffcb3ff4a7ebbefdffa80a0afc5 : FAILURE in 18m 15s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/b044c9d346a74cfda9ce22d603cf6bf3 : SUCCESS in 5m 44s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/4d65e2000dd643cfa936e09b179477aa : SUCCESS in 49m 18s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/44180147acbc41d6ae5c24961f538681 : SUCCESS in 46m 58s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/53cb2daac8bd4d1fb4d43d0108d5dac4 : SUCCESS in 46m 06s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/51cf5e82490f4b89a3c0f66db2f4e742 : SUCCESS in 18m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/861ee1c636a74010957d38908d59a85e : SUCCESS in 3m 43s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/7f3811a04b264acf85cff0815993f067 : SUCCESS in 15m 10s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/706312e72d1b4506925b4bd335f07b46 : SUCCESS in 51m 08s","accounts_in_message":[],"_revision_number":42},{"id":"35aae6c4ebab9166fb49dffe3788954d2c140bd6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-14 00:41:13.000000000","message":"Patch Set 42:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/62565bb33c2a4aa5b0be7c06cf003bc7\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/df96b9a1239548ae85227e767a26ae0d : SUCCESS in 18m 03s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/00a0c73165db41bb8fa7fdee080d081c : FAILURE in 1h 25m 10s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/f9d778ac7326497dbc94415c69d8e23e : SUCCESS in 23m 18s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/2286e4f89d7c49f599446bfecabcc95d : TIMED_OUT in 1h 31m 00s","accounts_in_message":[],"_revision_number":42},{"id":"4b85016abb62425ed7beee97f07a0947b33250af","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-06-25 00:34:28.000000000","message":"Uploaded patch set 43.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":43},{"id":"194c77f09eefa0d1807b1959d64f43a563852994","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-25 00:59:21.000000000","message":"Patch Set 43:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/18338dff536a47b48d81971b542eba5e\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6d1a41a866c1455785159a218d0da132 : SUCCESS in 17m 27s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/576b5d02a05844b885a1f2e879be688c : SUCCESS in 18m 26s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/5b2438c25b4b40a081e7a14c01c0d958 : SUCCESS in 21m 58s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/1231a2a60f424e9bad1c97fd18de29ec : SUCCESS in 19m 38s","accounts_in_message":[],"_revision_number":43},{"id":"a2424c56ba8dc228953cded22583f9ba3f237d4c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-25 01:44:25.000000000","message":"Patch Set 43: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/675ee1ffebeb4db5ae4d13136143574e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e3fc5aea872475c838699b57b736b6b : SUCCESS in 7m 56s\n- grenade https://zuul.opendev.org/t/openstack/build/91408ebd031a476796b9a68becc83783 : FAILURE in 50m 24s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c55418ce0dbb4eecbfb3c64fa58e4785 : SUCCESS in 57m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2ee70ed5c54449f9aff3e521f7d99e2b : SUCCESS in 1h 03m 22s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/117ad6df411e454ebe6e73d042985e97 : SUCCESS in 18m 10s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6f16106f238c4ab1b70ab867d1f8b2aa : SUCCESS in 22m 20s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c09f56d2e9c5444ca7b04648ba1b9989 : SUCCESS in 22m 18s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/61248259ff7a4c3a8b3b0f10f9b621a6 : SUCCESS in 11m 30s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/b6d5562b144e4ea6bac80a6a14669ff9 : SUCCESS in 6m 08s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/691278c43cef4c83b27017cd7ca7e15d : SUCCESS in 15m 14s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/b326c81cc8614587b557f3374d17b564 : SUCCESS in 17m 34s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/ee5b3d6b413446ae858c931c5ff971e7 : SUCCESS in 16m 54s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/4aa304e648e14f47bef9b49d3810377e : SUCCESS in 16m 50s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/f456a5f9fc4b43e683bf97fc57086136 : SUCCESS in 21m 45s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/d275aec0a1f3452a974f1b953c33aa10 : SUCCESS in 17m 42s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/0514d0cf86ed44a7b93dbc2b491a6d5d : SUCCESS in 19m 19s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/8b86535e979e4aa79223b70d1f9bbe21 : SUCCESS in 15m 58s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/1da612eba49e4ade88520eb638b7eb22 : SUCCESS in 15m 47s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/68d4d3301a034edab78644c84cdc6765 : SUCCESS in 16m 29s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/29b2902f9c7c4b0f9d7077a2e469a071 : SUCCESS in 13m 04s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f43050cd4f8b47a284faf5b4886484e4 : SUCCESS in 9m 32s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/da5ca6b5269a49be87fca19406e301f4 : FAILURE in 4m 35s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/7a04105b8959468cbe771716a247e3c7 : SUCCESS in 6m 28s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/8b53612502514772bc2b0dae2f531ef3 : SUCCESS in 59m 16s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/335d143f680f48cead8b6eb4552bcc8a : SUCCESS in 49m 15s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/7541fe127a2e4bb386eb44a876878877 : SUCCESS in 51m 55s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c69e06848f4548dfa360ce48ad35d442 : SUCCESS in 19m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0f0292c614f04efb9e655973bbd6d504 : SUCCESS in 4m 24s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/2afb45755647446dbf1b704b48c58596 : SUCCESS in 14m 56s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/97ed26b8616c4c00817e6faa3a92ec5f : SUCCESS in 1h 04m 44s","accounts_in_message":[],"_revision_number":43},{"id":"05f32b06311d6a2427d141d84c8c0ba097c5937d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-07-09 23:58:46.000000000","message":"Uploaded patch set 44: Patch Set 43 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":44},{"id":"0d5cd5f486bc393d050cc0c057bb8527451c0b3f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-10 00:06:55.000000000","message":"Patch Set 44:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/446fb545af8c4d6cbc9647e1aa17a9b2\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/dccd6a47bed847ef8e1c484b4c28b024 : FAILURE in 3m 49s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c8396e011b7b4276a36b155e5b7469d5 : FAILURE in 3m 37s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/c5c84005f45e41168c7b69304c9daf43 : FAILURE in 3m 30s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/0d73ef75ee48402faf40fb921ed88c91 : FAILURE in 4m 05s","accounts_in_message":[],"_revision_number":44},{"id":"9984815e29c237145ba1b0cff8df913734a9aae2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-10 01:09:36.000000000","message":"Patch Set 44: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a6123f2dfdd040fa988a0303dd09bb79\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62fe9d1d5fb74a4ba130c3de7f7d3944 : SUCCESS in 8m 53s\n- grenade https://zuul.opendev.org/t/openstack/build/9d61656661fb4303b16072614bc8b369 : SUCCESS in 1h 05m 05s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a52afb72cea2498da26971163134b7c9 : SUCCESS in 54m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a37a8db7ddc7415f88cadaaa305e67ae : SUCCESS in 57m 04s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fd99e7528a234b988d35e384cc5b3ba2 : SUCCESS in 19m 42s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f022163fea4344f5b7de6a6ed030e712 : SUCCESS in 21m 27s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a4fbad76ccee4da6b757ae11d1c08845 : SUCCESS in 20m 47s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b17fbc94d43e41d8b2a352e1f9e62afa : SUCCESS in 5m 11s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/10bbdb96c818440caecb9c7a0dcf81d1 : SUCCESS in 19m 08s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/745e0437a127450cbe97d43ecfd2b3ad : SUCCESS in 18m 30s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/6188e1d012c742c5b58fb415758a1ff8 : SUCCESS in 18m 40s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/db4a731c647848c397adfdaaac418699 : SUCCESS in 18m 47s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/0399fb395e6144639b0114736736eda3 : SUCCESS in 18m 02s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/a5124365d3d5487297d7b16f691f96f6 : SUCCESS in 19m 09s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/aa6de389ba824898853b4c7870181c4e : SUCCESS in 18m 54s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/d96541bf852142f1ae44237b63503ffc : SUCCESS in 19m 54s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/787b69783baf465799e62e25373e6006 : SUCCESS in 16m 58s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/57660cf81fb1478981d86e866d845e2e : SUCCESS in 18m 31s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/d03e3a63e82c40f39af7703523a89bb2 : SUCCESS in 18m 33s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a578b4d8ab904dddb19aced08c395e5d : SUCCESS in 6m 25s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/3c65dc8b7ba341198796e1d1e279e3ad : SUCCESS in 13m 26s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/b15bc47d8d9549528763c8fcc5410c21 : SUCCESS in 6m 31s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/590d1d878689419eb9164484a9cfc51c : SUCCESS in 58m 43s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/2b627ca4d3684caaaf4e19980db177a4 : SUCCESS in 52m 25s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/90aee5723d6e4431a9bb6edb01e3e085 : SUCCESS in 54m 02s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/715ed9de81cb454cbee884e17a4f9717 : SUCCESS in 20m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/96c9bb4874cb4a68ae49f65a9844d10c : SUCCESS in 4m 21s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f6e5d4bc1c9141d0a2fec25f696cd7bc : SUCCESS in 17m 04s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/23ab93886c2541afa7e1e963b0010e96 : SUCCESS in 55m 31s","accounts_in_message":[],"_revision_number":44},{"id":"e0ad6c3647cf7fc013827264034b1df3c2c84ba0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-07-10 23:01:25.000000000","message":"Uploaded patch set 45: Patch Set 44 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":45},{"id":"325bb2cfdd08fe7dee54ba23864d170c28ec4ed2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-10 23:18:08.000000000","message":"Patch Set 45:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bfc95b6c9dc2426ea68237d707b5f03d\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ef6bd4d3bc6c451085e69e027c1769da : FAILURE in 4m 03s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/94b33a4699b74187b806946cedc1fa90 : FAILURE in 4m 34s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/b2bbf8bcbd444518955a0d8444e6b16c : FAILURE in 4m 16s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/794b1d0273df480796397be8a8e43b69 : FAILURE in 4m 55s","accounts_in_message":[],"_revision_number":45},{"id":"677855122e3587614b81af98ebcbb900f9b55450","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-11 00:13:53.000000000","message":"Patch Set 45: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c2567980021041f89e207b431233faf4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/997530ea66a5423396269d32f0b68912 : SUCCESS in 8m 10s\n- grenade https://zuul.opendev.org/t/openstack/build/6703f3b1a6b1402190254b19089a0586 : SUCCESS in 1h 01m 17s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/2e15439f77c6474ca09a75a2d3c2360e : SUCCESS in 45m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/03529b32f473434eab86d713a3207379 : SUCCESS in 51m 27s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2e213d1fcfa3451c95bb7c1847e48393 : SUCCESS in 20m 41s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fcf78e64c4eb4e4d94fa721039ceb3d1 : SUCCESS in 21m 38s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ca9c31d488fd4d02a2e7d69b68815c9f : SUCCESS in 21m 51s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/2a70d950b9374a60b0e0a0f03c42e785 : SUCCESS in 5m 23s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/fa48557c6740453b9d242705c6dd3632 : SUCCESS in 20m 14s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/476e1974c42745aa848302b219e80eab : SUCCESS in 19m 25s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/b3d1b05e7d4e466496541ce16aeed4d8 : SUCCESS in 23m 03s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/54ce32946f584812be075ff84a94d068 : SUCCESS in 22m 04s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/658e41a635dc450398a078dec5ca153b : SUCCESS in 23m 34s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/a112dc81b1ba470fa124ed0e913d3900 : SUCCESS in 20m 33s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/374f6b1523b24395a369ba5c50c13a22 : SUCCESS in 17m 54s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/860364a2403340b4bf21527bb0d25a07 : SUCCESS in 21m 39s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/40773fcb78584e2c8bcf1cda42a033ae : SUCCESS in 16m 31s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/a496648a55514e13a88bfbcc72fafe1d : SUCCESS in 17m 26s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/770f92262eab46ae963c627fe641e4bb : SUCCESS in 23m 07s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/9fdb59e878f64493936d2c859814dcbc : SUCCESS in 6m 27s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f3681bb811e64d95a01175a7cceed74f : SUCCESS in 14m 44s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/183a117264894d8d93c5efa037829483 : SUCCESS in 6m 40s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/90133a3d37e94a9d8bcc4191b4198448 : SUCCESS in 55m 25s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/68292e93e5054151a0d12b0878aa91fc : SUCCESS in 36m 44s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ff4343f1a8254d96ae8a6d9332deb5fb : SUCCESS in 52m 10s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fd73ddbbaed84a0cbdccff7bf4f55409 : SUCCESS in 18m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8e8a514b31a54c179a061a51bd35bac2 : SUCCESS in 4m 21s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6782d0e62f364706a8d5e36b2c242017 : SUCCESS in 17m 52s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b1ff0db1d9334f45b1ed396a345ec171 : SUCCESS in 46m 03s","accounts_in_message":[],"_revision_number":45},{"id":"521ffcffe1d172aef92c3bbfa2243e1e588f7e35","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-07-19 13:45:09.000000000","message":"Uploaded patch set 46: Patch Set 45 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":46},{"id":"f19f7a25c61f8454ab178fc646e37462f901a719","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-19 14:20:02.000000000","message":"Patch Set 46:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/85bd2deb12a443cc9d7f9b4a8cddb99e\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0007ccbbf5764de5ab445c2d10af60dd : SUCCESS in 27m 53s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2c24b58b61914b1cbbdd84d1888c5936 : RETRY_LIMIT in 1m 57s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/3b8ffb35e82e4e5fb5b5f2d69f0c806b : SUCCESS in 33m 47s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/aa33b6be916346a49958e9d2ed24b2e2 : SUCCESS in 29m 05s","accounts_in_message":[],"_revision_number":46},{"id":"132473d06564d8cbdb28afe588029d2d3519a5d4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-19 15:19:14.000000000","message":"Patch Set 46: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/88e697cc8df1471993c8619bf39b4782\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f028964da99a4af09c47240c79a89f1f : SUCCESS in 9m 43s\n- grenade https://zuul.opendev.org/t/openstack/build/6777ca7a66c840ecafbbc45d3342c249 : SUCCESS in 54m 30s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f26749a5e3924d55bcdf191d66ac23cf : SUCCESS in 52m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bd68cc9823b6442bbd4e063d780d45ee : SUCCESS in 1h 11m 17s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/af9a931afb9540298a77450b1bc817f2 : SUCCESS in 19m 37s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/bfa44096268e4e769b85c1b5317f7365 : SUCCESS in 24m 00s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e3a9a6e8f31448d7a29dffea2931d7e0 : SUCCESS in 23m 14s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d2533c71d7184f21932569627c3bedd1 : SUCCESS in 5m 51s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/c948ec4e0ee84792b81545f78b87e2da : SUCCESS in 18m 53s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/889a516adce9439a9a172e06d21f59c7 : SUCCESS in 19m 24s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/de113a1390674434b15b5d5ed4df0a8a : SUCCESS in 19m 41s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/1f8203f22755438984d166422ba701c9 : SUCCESS in 21m 48s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/1af00928db4c431799958e0af265d71e : SUCCESS in 26m 05s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/71fa0dc0baaa4a079730f23f1734ee7e : SUCCESS in 16m 28s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/4170311b367e46fb8c18043b95641931 : SUCCESS in 19m 07s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/6d7ed77bdc56425fbca6d9d359df9138 : SUCCESS in 19m 24s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/a2770b6b76cb4abf9f0cec650a94ec29 : SUCCESS in 19m 21s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/d1c257172de649c3baded4dd7ea5053a : SUCCESS in 16m 55s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/7a46b8da3ea545318b90871fdd5d7982 : SUCCESS in 17m 52s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/295ae9ce51bf4f12aa43101d0370c9ba : SUCCESS in 8m 16s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/24e12c5999954417b4c9cb0a378564a0 : SUCCESS in 15m 26s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/c8730085536f41c1a6f1712b5e096e6e : SUCCESS in 8m 08s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/1dfe325f500b4258a76f28be82014cbb : SUCCESS in 1h 10m 03s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ce1fe96de9d542b690ac3f52d941ae87 : SUCCESS in 46m 12s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/37236394d9ee4fc2847a520189313532 : SUCCESS in 42m 48s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4757903705d948708de8bbf6fae8aec4 : SUCCESS in 21m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c3cd289a0e14ba4836f236616544dcb : SUCCESS in 4m 06s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/3111f9dbf298429199d509269d5b18e1 : SUCCESS in 17m 48s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3ab4effc4a4a421db2c452e1d4ad87f0 : SUCCESS in 50m 35s","accounts_in_message":[],"_revision_number":46},{"id":"60203921a5055ca388d6658be91dbb4dfa47dbad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-07-24 10:24:51.000000000","message":"Uploaded patch set 47.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":47},{"id":"7b0653ef6da6d1c2a1d2a68f6095ab5743fb1944","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-24 11:37:09.000000000","message":"Patch Set 47: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bcc9ca582b4a47348edd55b909e22853\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/23fd9cc2b4e94bcabe8f94dea9a0b0a2 : SUCCESS in 8m 17s\n- grenade https://zuul.opendev.org/t/openstack/build/1c7a93d24087421d840d1ca538eeeced : SUCCESS in 1h 05m 50s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e177b7fbeb4b4a6a904e532f029c831a : SUCCESS in 47m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/081275f7a10443909f6a952731b467d2 : SUCCESS in 52m 03s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/30fda37806b740e985d05d07fb7788ee : SUCCESS in 19m 50s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/44972b08ecdd461f8b967baeac9280a4 : SUCCESS in 19m 05s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/cb7dc47808d04d9db07cf1c387b58748 : SUCCESS in 20m 30s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/463ba83562644dbf9829b2056a5a41ea : SUCCESS in 5m 05s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/bded84f5e43346cdac6e2d50db7c58ae : SUCCESS in 21m 24s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/f654a7866b3846ddb4ad31b796540cbe : SUCCESS in 20m 02s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/3ad0ae107ffc454db78fd65404192e02 : SUCCESS in 20m 42s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/f8844bbded5d4bf9b8d3859da6fec503 : SUCCESS in 18m 00s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/1233ea1381084c3ea2de48d921162fe7 : SUCCESS in 22m 30s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/7942cc5c4db1490cb6462e0faa825e9d : SUCCESS in 18m 24s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/f8360dccb5cc47fbbf80c587bbe7b033 : SUCCESS in 17m 46s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/9845690729264d23a5f1b817dfb58404 : SUCCESS in 21m 12s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/1256037342b34971997dba05c3b1e21d : SUCCESS in 16m 14s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/8208885574d54eb4a0834e8777ff94df : SUCCESS in 17m 16s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/2072edfb64bd41b0a497b603abc10e48 : SUCCESS in 20m 52s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e494aa0f5347476f83fa00dd4b47920c : SUCCESS in 7m 28s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/c1d027cadfbb4794a8e1115f133b67f0 : SUCCESS in 14m 39s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/640d7f5b26f24218b8c1f81370b14dc0 : SUCCESS in 8m 22s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/98a47d45d09d4aef94ad142279879d1d : SUCCESS in 59m 26s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/613183af14234dbf84835eb0b55b0642 : SUCCESS in 48m 04s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/70881cd575d34d4c8101646946e3c72d : SUCCESS in 51m 27s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/76b964d5f89441d1b596902357985488 : SUCCESS in 20m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f7b3ca378f4f4c9c9ac509cb61ff1319 : SUCCESS in 4m 23s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d662f6e67a9f4df3a3a9dc4459372195 : SUCCESS in 17m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7372d70651de4079a2f25337a3cddc85 : SUCCESS in 58m 54s","accounts_in_message":[],"_revision_number":47},{"id":"8a00e2c5a02a02474e3b59e4ef1bf50422d18fe4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-24 11:58:00.000000000","message":"Patch Set 47:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6c32fee9d68a4824860aafef1d1a1bfd\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5af316bb25784810bb9d103f93ebcd9b : SUCCESS in 25m 26s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4147fdc0eaab4f0195b1d5b434d4b020 : SUCCESS in 25m 09s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/70031f25c7e946298d31f7ed0b0f121f : SUCCESS in 26m 41s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/1d9cae9577e74377b5d6daae8da1778d : TIMED_OUT in 1h 30m 57s","accounts_in_message":[],"_revision_number":47},{"id":"1f26cd9d1c42685474dc9557fc748a84f3019aa6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-08-02 11:36:31.000000000","message":"Uploaded patch set 48.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":48},{"id":"3008e44626fe80e6d9d19b1b53169e41955abd13","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-02 13:09:23.000000000","message":"Patch Set 48:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8f02a5a7373a4da782ec85dbac113b39\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e57e2b15e55f46a39c56f88bcffd9ddc : TIMED_OUT in 1h 30m 59s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/84fb91b01a284e948e468fc2b6d21980 : SUCCESS in 1h 10m 21s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/540af4ef1e57461f96e52222028ba339 : SUCCESS in 57m 44s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/c82c9f5e0d2c47a191d52645353646a8 : SUCCESS in 49m 29s","accounts_in_message":[],"_revision_number":48},{"id":"c679b36897be599a3168e13dcedd528a5f911949","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-02 13:40:08.000000000","message":"Patch Set 48: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/74bcdf93ed494cbdbabc90d1d56f5ff4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e3acd6cc5c734930b211aa4ab2165ecc : SUCCESS in 8m 51s\n- grenade https://zuul.opendev.org/t/openstack/build/4a272143ebd64de399b0c17253df5de6 : SUCCESS in 1h 11m 24s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/1d99daca9881415eafb774fabf9fe560 : SUCCESS in 47m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a8ac706d6d1a4f0083361e11dca0b283 : SUCCESS in 1h 03m 58s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8b33b7a8ccb043dfafd60fcb8e4bbda5 : SUCCESS in 21m 11s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6da1a743e233472bbf64c1f3443f5067 : SUCCESS in 21m 47s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7879f30864dd4caca77d0854965a62ac : SUCCESS in 22m 59s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/4ab4e4eb84f342caad40adad9c67c1ed : SUCCESS in 5m 07s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/12f14d144c1849978a65297986133f19 : SUCCESS in 19m 01s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/4f6c056b938d4c68bb93508c0167bbc4 : SUCCESS in 19m 20s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/698ff7c82ed14382adad6506d504151e : SUCCESS in 21m 06s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/11ef1cd727774939a54c0d2ef1b045e1 : SUCCESS in 21m 33s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/4548f68774344de084bfd039c3fc2d40 : SUCCESS in 19m 52s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/927c842b111a4e1b8302e4ef14289466 : SUCCESS in 18m 39s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/94bbb75fd3c043e996d6587de58ce8bb : SUCCESS in 19m 36s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/358d76f2684d42e3b45eb1e401b77b1b : SUCCESS in 18m 39s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/e6ced61dd6b74472b60471033301d9ae : SUCCESS in 15m 13s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/048ba52ed542453d807e6212078afb80 : SUCCESS in 20m 11s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/593db04ef5ba4e25a09f537b2aefde31 : SUCCESS in 18m 00s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/1f868f78c83a48cd9c0bcfaec5d53e43 : SUCCESS in 7m 06s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/65290455e65b480a8b927c65a9ee7e3a : SUCCESS in 16m 26s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/b56ac9306c0c49aa8c90c82825057d8f : SUCCESS in 6m 12s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/07a92a7c9f274733aa281abd60caf9fb : TIMED_OUT in 2h 01m 11s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/0783072a6af94c198ddbad1c510caa33 : SUCCESS in 42m 35s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/47b51860acfb4b56b54ef7b71b7876dc : SUCCESS in 50m 28s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/13ff13b08fea4298a0d2a37a4cd15cfa : SUCCESS in 21m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3135f32309eb4f98b2c298b16643eb39 : SUCCESS in 4m 05s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/83a88bba6da24e7586e1193956d3a48f : SUCCESS in 21m 06s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/619bccf66809446494f1b9f60b08a584 : SUCCESS in 58m 55s","accounts_in_message":[],"_revision_number":48},{"id":"cd61af50e5c22ec797908c2d42ad27bf1bb246cc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-08-07 13:26:44.000000000","message":"Uploaded patch set 49: Patch Set 48 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":49},{"id":"92a15a6ccf563d2a41f5b8a0cf0d17d890f48b91","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-07 14:28:16.000000000","message":"Patch Set 49: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/47f9bff8c21445dc9fb9acc8fee56d8b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8443fd917e994e7a9d8fab0fab8d82ee : SUCCESS in 8m 01s\n- grenade https://zuul.opendev.org/t/openstack/build/43527557d64f4ec79b599bb3ecb95dfb : SUCCESS in 56m 20s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/5e960d6780104cf9b09930f32126c9aa : SUCCESS in 45m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fe0340277cc846a18fafb46c9775556b : SUCCESS in 55m 25s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ff89cee51e59429ba1766024783470de : SUCCESS in 19m 48s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/05eb66679e254fb4b416172352c961d3 : SUCCESS in 22m 19s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/feb14e4d33e14ababef835e6aa6345b2 : SUCCESS in 21m 55s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/cb681962cc2c406c9dc0531538e23137 : SUCCESS in 7m 18s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/60f718ec81c343068698e2cef6d87a73 : SUCCESS in 18m 03s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/1ce15090b91a49eb9e213e44e9d06ae4 : SUCCESS in 19m 28s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/6a2971c639d8490f93ee96a598c60b01 : SUCCESS in 21m 26s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/c9d0b20e86b74a4ab198dd47c488cc37 : SUCCESS in 19m 40s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/9810b187a77c4e6f96ebcef0a9a4717e : SUCCESS in 27m 56s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/fc4a6050aed6409f837c2e7360824b91 : SUCCESS in 18m 05s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/09b3a89f4e8345ffa78aeec14e766a00 : SUCCESS in 18m 32s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/61a9f85c9c2d4941bb79b83f5c7b3828 : SUCCESS in 20m 42s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/842ae456b9c54dcc87ea8e0d24d3f526 : SUCCESS in 16m 05s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/c7a81c26e7134182a09d7038a1768df3 : SUCCESS in 16m 44s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/d8c2fc83d3fe4c989f8ffe42c413a8ee : SUCCESS in 17m 53s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/cc8a871f81e34771ba43f254aa5957e9 : SUCCESS in 6m 20s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/cfea21988458453fa1d470c8faaf73a5 : SUCCESS in 13m 47s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/331bdb608959460b95e2add7b714d9d1 : SUCCESS in 6m 22s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/e7f734affb6243d4b1809a8f3d40626b : SUCCESS in 55m 39s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/483c18849b824d8ca7d61f2580bc2ec6 : SUCCESS in 50m 25s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a291af965b4f4db6a2c66be49c735063 : SUCCESS in 37m 43s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d65877c229374f8088adcb2531491ec1 : SUCCESS in 18m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2c6e4be577e8499b9e801aeba77f13f8 : SUCCESS in 4m 18s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/71a2d463c43941e58b7ef7d426ace2bc : SUCCESS in 19m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/52c4c46cbf6e4b3a8fd528595445c5af : SUCCESS in 47m 58s","accounts_in_message":[],"_revision_number":49},{"id":"46ee8a3d8a65bbe15900f6709c3f0e9da36126a2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-07 14:59:54.000000000","message":"Patch Set 49:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/856127fafb8b4b8897b1ce0aa9626f9b\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b230f76f25a84323bca83aeaf39abbe1 : SUCCESS in 1h 10m 02s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/747a735089f546308aee3e9bf13b1e78 : SUCCESS in 1h 13m 46s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/d80210180f074b64aa94b2e0d49765d9 : SUCCESS in 1h 22m 12s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/4063d40ffd3d4372905d27a7acfa330a : TIMED_OUT in 1h 31m 03s","accounts_in_message":[],"_revision_number":49},{"id":"012fe614ac0625f4e3609e9ea23190ba9e72ba29","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2024-09-05 06:05:13.000000000","message":"Uploaded patch set 50: Patch Set 49 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":50},{"id":"b329ea991007204e7d7a00d6ac6a0456afbcbbb8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-05 07:04:55.000000000","message":"Patch Set 50: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e8a64e3ebe6c40b9b0dc257f50a00dba\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e72bd72abfa34f05a746fd06dea2bba6 : SUCCESS in 9m 49s\n- grenade https://zuul.opendev.org/t/openstack/build/2f929ca76ad147dd8e622e7b6719909a : SUCCESS in 49m 17s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/76310e7f459a49e4a9dd49020ea28d8a : SUCCESS in 44m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1accb5cf81774a2091de36d65cd0e2ca : SUCCESS in 52m 19s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3733b8de9f1448de9d05b6bb8dd2289b : SUCCESS in 19m 55s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5b5737465bbb45c69b3b1269b82b280a : SUCCESS in 20m 17s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/16e358053209460bbae35a880456e157 : SUCCESS in 20m 59s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/134e8f1f0b1d4ab59bcd36b3ed751fb7 : SUCCESS in 5m 44s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/9df31737bc584d6d8f6ce5d91a295ec7 : SUCCESS in 18m 32s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/4a58a9e18ccb4b3a9c6c017b973354fe : SUCCESS in 16m 58s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/03a09689df1f4879afbab884a1a0e960 : SUCCESS in 17m 14s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/1c03b103fea04d1599d0e70adbc1e66f : SUCCESS in 18m 41s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/46ae4c5dc71541a3b6a0cd8c0fdcc6e7 : SUCCESS in 18m 38s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/e05e2e8c310a4045a19cc62059d1e2bd : SUCCESS in 19m 51s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/7742f546c861410fbe9b336b6809c208 : SUCCESS in 18m 56s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/2c368b67ec9e4496a20e8cbe9ce243a9 : SUCCESS in 20m 39s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/3fd25cff236e4f8686a59d427a3ac430 : SUCCESS in 14m 22s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/4a384c08a9b24c9994fca947f8a2e4da : SUCCESS in 16m 53s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/2ba9e988eca64c2da59e8b54806979f5 : SUCCESS in 19m 45s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f4e20e8b71514e85931deb759f230b76 : SUCCESS in 8m 06s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/76d726c919784a789ecd566b303009d0 : SUCCESS in 14m 53s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/a3ea2d3371474b7cbfe83f82166ec5a3 : SUCCESS in 9m 13s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/2dd56263f9ce4baea2b3834d1b68b053 : SUCCESS in 52m 24s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/0f75c00af4564859b265c7cd72d7098a : SUCCESS in 47m 39s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/88f5dd102db5406eb00f1a38c5ae48e3 : SUCCESS in 39m 12s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e54b7e6f0c0746818615335deca5f938 : SUCCESS in 20m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e36981b01ffe4a888db5cf518409ca20 : SUCCESS in 4m 54s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/8efc17748023422da62706c6e6607bae : SUCCESS in 19m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b9d14dd8c70b4710bf77476a72df14e3 : SUCCESS in 43m 16s","accounts_in_message":[],"_revision_number":50},{"id":"75f1a4d3ccbf02774575299d9c17dd33f002820c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-05 07:09:40.000000000","message":"Patch Set 50:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f5102f78d2c94a3a9feb21f154792fbb\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d1e9ca833a284962b27f5a3f9de91db9 : SUCCESS in 54m 56s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c6e2ba16be22472c96427b98dc1a125c : SUCCESS in 48m 06s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/ba86cbb249f940a48a7e8f5589710a2b : SUCCESS in 45m 13s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/07ed9685f3cd47eaa4fcc7656b6f8a7d : SUCCESS in 1h 02m 22s","accounts_in_message":[],"_revision_number":50},{"id":"ef61ec232190c5c5c5feed0c83b0c017cc23809a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-09-11 08:47:37.000000000","message":"Uploaded patch set 51: Patch Set 50 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":51},{"id":"c891d6e16cea1bdb4131d755265c841962b298b3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-11 10:06:42.000000000","message":"Patch Set 51:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/101037e00def4f1f8a68c78652fa65bc\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/90fd6e1a08f04a96bee1185cbe44a96e : SUCCESS in 50m 47s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e8cfb8a68e82411fbc5bc5f1642b4660 : SUCCESS in 1h 07m 12s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/812c36e519f6459888a6eb69eef24fc2 : SUCCESS in 1h 15m 50s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/c94e908bd595473ea0588cc341cc7622 : SUCCESS in 59m 17s","accounts_in_message":[],"_revision_number":51},{"id":"e34589d587630c09904536bc0c7a81a878f5908c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-11 10:07:16.000000000","message":"Patch Set 51: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6081104b65724071bc50106150efc75c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d82ed653e104a5f9b681c58354e64a6 : SUCCESS in 7m 24s\n- grenade https://zuul.opendev.org/t/openstack/build/ac88955e1fde4c3ba4f818171971d1c4 : SUCCESS in 42m 28s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/293c38da5a0a4208a674f44c696ae6d4 : SUCCESS in 47m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e4c6eb583cac4d9b8f35aa24098dd847 : SUCCESS in 45m 25s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5f8c9e6d8e564bdf8ab79cdf4c742f8c : SUCCESS in 19m 41s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5ebf14724de742b29ce8424fa2a60fc8 : SUCCESS in 21m 13s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3c0b8d0538834c178a7e4bab72a1e8ac : SUCCESS in 19m 02s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f1ebe40a51bd43ee8af0574de312143a : SUCCESS in 5m 27s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/21392e6c5b954bc4affe77e223bf5454 : SUCCESS in 16m 14s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/4b3d9a31d05540e894d2495b09c6c9b8 : SUCCESS in 16m 38s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/010bf7cfa2204cf58df1ded8da310ea9 : SUCCESS in 18m 53s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/386ac53233b5479aba64c0d1ea34a03e : SUCCESS in 21m 06s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/00ccaadeebbb4da89e04cac312176028 : SUCCESS in 22m 29s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/869fbbb8533f452ebfe66df7495ed005 : SUCCESS in 22m 43s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/202dbba34dba46b2a71368bb5555f71e : SUCCESS in 19m 21s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/c3d2096da6bc4428b94695ed90dc94ac : SUCCESS in 18m 51s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/efd959ca5e344096a7121953145a6988 : SUCCESS in 20m 00s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/829e5930d2a24ad9942882fcc10751e0 : SUCCESS in 9m 58s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/363f575de65b49e78bce54dbf8552aac : SUCCESS in 12m 48s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/10a9069a0b824d6a8c7bd598f5d6228a : SUCCESS in 4m 37s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/12a8a8c9bfaf40019430d8a84cb337e1 : SUCCESS in 9m 36s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/ade6ada627c64f95bc9c951b7e5b3718 : SUCCESS in 4m 47s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/e01b550a75364b0ea179d4bc0217e5f7 : SUCCESS in 36m 35s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/902116ae9ef7408ab3effda37306819b : SUCCESS in 33m 26s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/5faa32ef404c44f49636fd277eceafaf : SUCCESS in 37m 21s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c25eb89b050d4d63af7407382534d480 : SUCCESS in 18m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dacb851caef44575beb5e1e4d39469d6 : SUCCESS in 4m 20s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c1d7295d00664adb8cac10f7d610b672 : SUCCESS in 19m 00s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e624e122d8cb49e987c37da413fc7efa : SUCCESS in 52m 12s","accounts_in_message":[],"_revision_number":51},{"id":"8f2e7baf825ff5267ba8558417f42ecdde66ed70","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"real_author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-09-11 21:40:46.000000000","message":"Patch Set 52: Patch Set 51 was rebased on behalf of \u003cGERRIT_ACCOUNT_7847\u003e\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}],"_revision_number":52},{"id":"b773d89c2df4712000b378bcf6e4d2c17b33f7dc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-11 22:43:04.000000000","message":"Patch Set 52: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ab73769cff3542c6bbbb233774076382\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b6b48dc143614e118a2e782351990488 : SUCCESS in 10m 31s\n- grenade https://zuul.opendev.org/t/openstack/build/8c9353354cbd4ffe8efda9952af746a7 : SUCCESS in 59m 43s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/051f3e81230a4e15b735cc8bc2930839 : SUCCESS in 46m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3c991ac7c35e45ac850d30b324f5b462 : FAILURE in 37m 29s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2cf20162a8b443fd8ccc979ead8beaaf : SUCCESS in 22m 14s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/beb25e699fdd447786d654f36df90ca8 : SUCCESS in 13m 42s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2975d902ff8d4da5a9c3ab7391267e5b : SUCCESS in 19m 49s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f0ea46e758894d099230514826edd434 : SUCCESS in 5m 46s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/25b2a223e93847f2baf090871b8ce57b : SUCCESS in 20m 08s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a90e7cac12114d498c5153a9b58ad251 : SUCCESS in 18m 10s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/7897a0f0d1074225ba9e3aba9c4abc53 : SUCCESS in 21m 08s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/cdc523740ca54a46b713c2a4a3ce85c1 : SUCCESS in 21m 58s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/9f791238032e4dd48b6a655db3984684 : SUCCESS in 23m 06s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/f76e32eef42846dea4e6e7a751236f04 : SUCCESS in 17m 38s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/71ebeec054eb47419fc24ef35fe917af : SUCCESS in 20m 26s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/924a5315e6dd4ddea09f1f52193ceb0f : SUCCESS in 16m 36s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/bbe2efd991cf485f9bb618517e89b456 : SUCCESS in 13m 49s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/b2f716be9bda4aaa934b532a2175b89b : SUCCESS in 14m 21s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/2aee1374612744fe97ca22ce34982c85 : SUCCESS in 18m 53s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/92e41a4198cf4be8989c4333cf789a49 : SUCCESS in 7m 39s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/d70eed8e24f1469da0df9c4c7d40c404 : SUCCESS in 17m 16s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/df13fb1889e64b2381416ecf9d59b163 : SUCCESS in 4m 46s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/0f6939b3c8904bbca6bca499583476fa : SUCCESS in 50m 35s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/dedf34488f864a51a7e8fdf9d5decb30 : SUCCESS in 48m 39s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4f9eb9f9a34243e28b7ed881a70137fa : SUCCESS in 49m 25s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9bd9fa940a504130a1c6af7444d11e96 : SUCCESS in 23m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/56dfe058ea1649389bf54dc44cf6d863 : SUCCESS in 4m 48s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/dba42be19e1a422da208e990d5c5adeb : SUCCESS in 11m 04s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/858f15a1663640e3b907a9a9f0a6df80 : SUCCESS in 51m 16s","accounts_in_message":[],"_revision_number":52},{"id":"dc456181b1b600a6ef9f3ab27c8770c3efe9b493","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-11 23:44:19.000000000","message":"Patch Set 52:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8aa99a1e9e8f4e638aae1f0d91423526\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/72d5f8e1ffd94b6ca1173181fd0859ae : SUCCESS in 1h 04m 41s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3d1623c486bb4c5bb058d08117acce62 : SUCCESS in 1h 27m 12s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/e6604cf7d5af407c8ecc0b5baafae26e : SUCCESS in 1h 12m 20s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/be024b1aed3843c68b02df95950a4c5a : SUCCESS in 1h 17m 49s","accounts_in_message":[],"_revision_number":52},{"id":"16929c8aa70bc844cd9e750dde34c9e64ed2f3f2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2024-10-02 23:08:47.000000000","message":"Uploaded patch set 53: Patch Set 52 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":53},{"id":"76bfeedba27e9dc3d9a31b2f73f489d097fbd4e0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-03 01:19:46.000000000","message":"Patch Set 53: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e09f6d954bc6490d927ec2b707171387\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a314c7dd124d45ad8f15fa4e8b5a378a : SUCCESS in 8m 42s\n- grenade https://zuul.opendev.org/t/openstack/build/b2029d022e384fbf917090873bb4dfa2 : SUCCESS in 54m 19s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/242e4fc5b18c473d9610326c834db3d1 : SUCCESS in 57m 42s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/00de503268d14e208113923ada204d55 : SUCCESS in 58m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9c68fa670242484c930ce456ed2c7da2 : SUCCESS in 34m 07s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4920e7bc9ed74fe3b41f2f0e471950e1 : SUCCESS in 16m 49s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/81fb3c46bb294393b23ff400e6c98459 : SUCCESS in 17m 13s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5e5529dac7984faa8375d036644a58fe : SUCCESS in 19m 31s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/19d55056f7da4dfe8f7ce3aab98bd876 : SUCCESS in 5m 57s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/7e2a923cb39842f991c77805042a7f32 : SUCCESS in 19m 31s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/827f44bfcb344c6d8cc4717c758fdaa1 : SUCCESS in 9m 54s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/a35d4a97cbed4929bddceea0fcaa89e2 : SUCCESS in 13m 53s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/74bff0819b6948919cd3c3fa68e32300 : SUCCESS in 21m 29s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/2f548be10c45484dbb54f963d339101b : SUCCESS in 30m 54s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/811dabccb3104f6988cd14ac32c6843d : SUCCESS in 10m 41s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/c4460337d6fc4f12816bc153ff24e1bd : SUCCESS in 16m 17s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/386839cec3c243a7b225646625ef4673 : SUCCESS in 20m 26s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/cc47dce6a2e849a7a22075c75566138e : SUCCESS in 12m 57s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/73ac5bbed0f3429e923410e108e06c2c : SUCCESS in 9m 19s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/089f056c860a4a30ad7d44aa4a390df1 : SUCCESS in 19m 52s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/9cf5b7a453114fd98e98e2cf3de2d6ce : SUCCESS in 7m 56s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/a4dd972c521b46e7ad2eded064d63f0f : FAILURE in 5m 45s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/2f79c4281af148b1a0b6eb848543d468 : SUCCESS in 8m 52s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/82ae97690b9143de844b5fe2f8ef7778 : TIMED_OUT in 2h 00m 49s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/cf0f4e14e50746b5aa65dbffb76d1f2a : SUCCESS in 43m 25s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/37da7d3c5570471daf69999e808eaf37 : SUCCESS in 43m 30s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/db0e14610012412fa6ddd5db6d3a5437 : SUCCESS in 22m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/857577754e9246838415c6dff7261cee : SUCCESS in 4m 22s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/bc3439f9836a47e9a9a4dcb941b0fec9 : SUCCESS in 34m 00s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d3be12c42b084ceda56ea2f63aa76fed : SUCCESS in 1h 00m 29s","accounts_in_message":[],"_revision_number":53},{"id":"a27e759a473826dda95a88f806ff83407ef7a84f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-03 08:27:50.000000000","message":"Patch Set 53:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3f2a322980134cd98e173a1840d6dbca\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/121724914e8744e9a8bf0c9d2baeda09 : SUCCESS in 59m 32s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/91f19d977acb49f89d64a9153123390e : SUCCESS in 1h 24m 33s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/37af8ddb6fac47588e15f2bc03505036 : TIMED_OUT in 2h 00m 53s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/c199a230c47a44d9a748dfb5ae6ac19a : TIMED_OUT in 2h 01m 04s","accounts_in_message":[],"_revision_number":53},{"id":"26549b3e7433bafe4e7bf310bf3b065b0e3d2768","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2024-10-04 21:01:35.000000000","message":"Uploaded patch set 54: Patch Set 53 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":54},{"id":"a6230e6a64f41be1fec53e6e1c532e74759ac7c9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-04 22:07:00.000000000","message":"Patch Set 54: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5329de61a2a14bb7bb5522da2f3ba0fa\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a55573862bfe4f659ba6fae31f6620e9 : SUCCESS in 8m 24s\n- grenade https://zuul.opendev.org/t/openstack/build/245df9deb82a4b34be76f8c39efff2b4 : SUCCESS in 39m 37s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b0bc1c25d6d04d42aff0ff92f495812b : SUCCESS in 50m 09s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/74420d2de6864d1c95c7a598ccb57da4 : SUCCESS in 44m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8c46c77b1002481e80d489be7ffc927d : FAILURE in 37m 25s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d9d77414e1cb4d1c92f5034d7007177b : SUCCESS in 19m 54s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b79c279043ae49a2b1686770e76e0f23 : SUCCESS in 10m 38s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b0f15dba7f3e4961a63d44551c091031 : SUCCESS in 11m 00s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/dd56ae779bc44457bd70f4d4e64c82df : SUCCESS in 5m 07s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/723a12e36d9d4969b31e93eb8dc0c993 : SUCCESS in 10m 33s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d127fcd1181942b499a06d785f22ab8a : SUCCESS in 9m 49s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/108aea110d794ea5a8710c452bf521c8 : SUCCESS in 18m 09s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/683921893b48485c9b02449600fc37c5 : SUCCESS in 17m 38s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/3ee76736758c418690f5b8b90152f4e3 : SUCCESS in 29m 26s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/80c4031bbc954301a05a320107b5c360 : SUCCESS in 16m 05s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/5ac944aaa1d24f67995ca2cb14632a35 : SUCCESS in 16m 19s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/230aacc1b88a449c93b83e8b0a51a8ab : SUCCESS in 15m 10s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/26c08cdbd34e4148ad67356a9509d6e2 : SUCCESS in 13m 41s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/119e1f4af9384195b173aa3c7dce962c : SUCCESS in 15m 27s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/a4d3899b35ed4b0da6c1b93a9909a01c : SUCCESS in 14m 48s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/7aadbfe043b1442887f8c8ffca51a8b7 : SUCCESS in 6m 22s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/83b6e4d4a16a4ef6964cb0ee6e314513 : SUCCESS in 14m 15s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/cbdfacb455c94a849d457b8e46bbaa88 : SUCCESS in 6m 11s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/83e8234040a246d19122d68bde07d885 : SUCCESS in 50m 15s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/7db00ae2b0634001b37171317e06401f : SUCCESS in 40m 38s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/174da561ee544791ba6c47881c9c9ca9 : SUCCESS in 56m 45s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/39d2ee421c654010baa3fee0b310a05a : SUCCESS in 20m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d75936f1c48343f0ab5305bb76565a46 : SUCCESS in 4m 21s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/1bda42d27a9144bbb4d8877edc86c498 : SUCCESS in 23m 18s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/886044fed8a54c3d8f56b723c8e0e73b : SUCCESS in 42m 43s","accounts_in_message":[],"_revision_number":54},{"id":"abfd567e1d686ae463aac5d522b14dc2c8b98a9e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-04 23:19:55.000000000","message":"Patch Set 54:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/16be9857e18b4c549a97ec0e3ba50518\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4b4e0bfd985e459498cbb13174325281 : SUCCESS in 54m 14s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/abda76c7543a4a8aaf10ecebda8fa0fc : SUCCESS in 50m 07s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/8e35f34d59774efda846817807f3bf13 : SUCCESS in 1h 35m 00s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/514e31c6ab584925883f13edf9458fb9 : SUCCESS in 1h 42m 17s","accounts_in_message":[],"_revision_number":54},{"id":"7854d657c6e452ba3a9a1aa97f67e0b071ebc4a5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-10-24 21:11:26.000000000","message":"Uploaded patch set 55.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":55},{"id":"87a66cd7a380728df15f2ec7d0f172c65a8f7cd6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-24 22:20:44.000000000","message":"Patch Set 55: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9c8c0130b26447d1b997f37437dd9ba2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/12dadd1100dd46a7be44c3340dd6f64a : SUCCESS in 5m 05s\n- grenade https://zuul.opendev.org/t/openstack/build/a2357460965b4333b889e9d76fd1dcee : SUCCESS in 56m 09s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/324042ce16604cf787ef56d3094a55ff : SUCCESS in 1h 02m 43s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/4b2aa7f3ca9d4c6080b5cc5a8f487302 : SUCCESS in 55m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/12acd977010e4b789cb8d0c85c6bd944 : SUCCESS in 1h 00m 35s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/86421bee17a34464af77afb572ceb804 : SUCCESS in 17m 43s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/01b0a12e46294929b28b9f56b3ddd35d : SUCCESS in 17m 56s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7fcef6fa9a0a4757803e99da28c81920 : SUCCESS in 18m 39s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/cf90ac1226ff4117b115d67ace91ca18 : SUCCESS in 5m 44s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/5e27ff3d060840b6b921ef78675d1957 : FAILURE in 4m 55s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/ff2914cdd70e477aa0a27049ecebf536 : FAILURE in 13m 59s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/65e759b2249644d28ea931a821016a7d : FAILURE in 3m 02s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/9f16799f5d1845a686d76ca8d688ef0b : SUCCESS in 20m 58s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/5e51b47450b94a7aa41fac7458ed9c91 : FAILURE in 30m 59s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/58ac0651dfe94d0bab845b3018fa66ae : SUCCESS in 15m 44s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/00f8a71db10a4eabb946927522af4289 : SUCCESS in 14m 23s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/7fa732d5d8094f5496c70e41586caf52 : SUCCESS in 14m 46s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/a9f369f60e2c4af7aa8264ba9d539380 : SUCCESS in 14m 01s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/fbfa3fc4d6bc44a582dad1ac6801f919 : SUCCESS in 12m 27s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/940db3d153d64958a24b2cb41e06bb5a : SUCCESS in 12m 45s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/780eb90e345745c6929e9e74a1f7b0c9 : SUCCESS in 6m 05s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/c6f505d025c442899cc324953e8db9cf : SUCCESS in 13m 19s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/9b36d50fc4fc42ac9657ce633484ac79 : SUCCESS in 7m 05s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/5b773ad7dfe74a59bcd017dfeb35dbf1 : SUCCESS in 54m 36s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/17feb0ce377344d88f23348c3b6cc945 : SUCCESS in 36m 13s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/7120e1fc49b1408e9606f2237d5a439e : SUCCESS in 33m 28s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a7d97c4b504743b58af80a7e8848a796 : FAILURE in 4m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d367989b293541018176c20fa55b2282 : SUCCESS in 4m 10s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/0c1eb669e6be4f578c93016db49bbbff : SUCCESS in 17m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/367062cdd3da45a7acd0f09943e8a4dc : SUCCESS in 58m 00s","accounts_in_message":[],"_revision_number":55},{"id":"07129a850f8f9501a2b6d0ad5547b235e87e9993","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-24 22:33:12.000000000","message":"Patch Set 55:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/db3b85da9bd64a3ea52a38b9e82bc63a\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/46b349b01ff74d92912630d7d42eb3a5 : FAILURE in 1h 19m 35s","accounts_in_message":[],"_revision_number":55},{"id":"874ed487347e496061d934d8e4bb892cfbb74222","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-10-28 23:10:06.000000000","message":"Uploaded patch set 56: Patch Set 55 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":56},{"id":"447efddd38bb05ff572706cc55b5162b61d61900","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-29 00:17:01.000000000","message":"Patch Set 56: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/05e12645a5d14af9bbfb2c20b03aa5bf\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc300bdc74034809b7e5d4e5920ead33 : SUCCESS in 6m 00s\n- grenade https://zuul.opendev.org/t/openstack/build/54fbcf4dc2664f5db7389c3669631a55 : SUCCESS in 53m 46s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/a978b77c711d4c2eba2b8216d0a41471 : SUCCESS in 48m 40s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/6cb6793e754c42f0b3b54f5a5f396db0 : SUCCESS in 54m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d8da0f1bdd09432fb2888e5644302034 : SUCCESS in 39m 01s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4b2a0aee06ad4f0a9144caa6a63fcb30 : SUCCESS in 18m 05s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/92ddf08143344d6ba70a0f54054309aa : SUCCESS in 14m 14s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a1f33bfa1cad4d499be790da01c115c9 : SUCCESS in 20m 01s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/9334eff2fceb4691a8b532347d807f74 : POST_FAILURE in 34m 46s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/82115494c40344aa863e0565de7e35a9 : FAILURE in 5m 21s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/e35613f6675245f59b2a59564823d8d0 : FAILURE in 10m 43s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/6f911d5ba48a43288db7af9777f21836 : FAILURE in 14m 27s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/22ce72e283da49a58537b0fb2dc3c34d : FAILURE in 17m 15s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/954d3955ec504e468d7c3ff5d63af031 : FAILURE in 28m 27s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/281d45b98507486691472f67cc996d34 : SUCCESS in 14m 35s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/5841b3e5518145b8b05ceff4ff0b8022 : SUCCESS in 19m 15s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/87894c71b0424dc095f36008a9591dbd : SUCCESS in 19m 26s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/701a85e6d817439bb37a9f85f4e876f7 : SUCCESS in 14m 13s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/cd772467e1f249149b4d724bb8349d25 : SUCCESS in 9m 05s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/11079b44ad1c410e85ab3556690c7a11 : SUCCESS in 16m 27s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/4d972d953a054c6980311c80c4d3d955 : SUCCESS in 8m 18s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/2c58f17e03d2409eb02456503ecb327c : SUCCESS in 13m 15s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/f415ee609d414c39b9ea7b9958d8f298 : SUCCESS in 7m 17s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/28619114a61749479c862d5a87a951ad : SUCCESS in 49m 19s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/3da8dce9a2334ff89fcf41b0bda59caa : SUCCESS in 54m 56s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/aa780a84801648cf90807b875ad22588 : SUCCESS in 30m 19s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fd0240a956f143f19aeeaa96576d1852 : FAILURE in 5m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/51c5c0a64fe047049dba445658ac6092 : SUCCESS in 4m 15s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c3b90292754d4525bde8d89f672fce45 : SUCCESS in 21m 23s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c909b0284b824fdf98959b293aff9d92 : SUCCESS in 57m 04s","accounts_in_message":[],"_revision_number":56},{"id":"4cf22e0acb889d692b0b81450fb07b3359ae918d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-29 00:43:59.000000000","message":"Patch Set 56:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f15531e59e2b437ba76be35e214a3d3a\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/fe30372561754164843f37b900c04e9e : FAILURE in 1h 31m 39s","accounts_in_message":[],"_revision_number":56},{"id":"69ad4c5e759844baefd183d4ad66d4fe318835a2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-10-29 21:34:59.000000000","message":"Uploaded patch set 57: Patch Set 56 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":57},{"id":"bbd89ef93ffd5ddbe9948eb0c0e24d3a3e4af775","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-29 22:51:43.000000000","message":"Patch Set 57: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/73abe9e9a5e44819a90f344af780d6e0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/798db5aa6e434afd854d47f32d937f0e : SUCCESS in 8m 20s\n- grenade https://zuul.opendev.org/t/openstack/build/fc419e613a774949a348a99f17ad81b3 : SUCCESS in 1h 06m 39s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/9206ad22312a440ba9b2546263b41573 : SUCCESS in 56m 43s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f84a7906de0f461c950ff161fb55ee7b : SUCCESS in 51m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/699a787611df49e383eec7351fe14fcd : SUCCESS in 1h 01m 44s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/165fdd34532b42d2bf7511f57523070d : SUCCESS in 16m 09s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7f700308c02a40d6bd88665c8e60eaa8 : SUCCESS in 18m 11s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4e60c03ca7684128a0dede07e3ffcdb7 : SUCCESS in 18m 14s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/79fae135ef6f4ffabaf9dfc4922614c5 : SUCCESS in 9m 07s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/c2143f5c5a6b4730b07b25f37e095988 : FAILURE in 4m 47s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/3b29450549fb40b68370be449979968b : FAILURE in 15m 58s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/9bfe459643704f2d8d74931b65f7e29d : FAILURE in 17m 49s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/623098d14cb3416598fe81a5b441618d : FAILURE in 19m 10s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/92e99131e65c4636b28b7f8bbf1ac6c9 : FAILURE in 24m 24s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/8ae69c7ac865444e92b2cbc7652e7536 : SUCCESS in 15m 32s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/e13539c31310419e89d87c3babdc8089 : SUCCESS in 17m 09s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/4f376b3008c2470eb2157a2a25c22bb6 : SUCCESS in 15m 01s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/ba10ee4fda2f463eb2b83bdbccf57d1a : SUCCESS in 14m 11s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/364c2b47ff5d48bfab7751e7890f7029 : SUCCESS in 14m 05s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/65c0e98148b943aba15bcc09522eaf4b : SUCCESS in 16m 27s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/1372fbfa92f448c48c77a2205e712c73 : SUCCESS in 6m 03s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/977b3942e5bb4b53964d00a02b418420 : SUCCESS in 14m 41s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/dd0521dc7c8648de89d964101dd6784a : SUCCESS in 7m 02s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/6624dbc0ddb14836a5b6743d4cf94d9c : SUCCESS in 49m 44s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/cc22a2d9e7a0412198d775a8e07a7969 : SUCCESS in 51m 24s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4ae724647ca344c8a0e261a91f27f7e1 : SUCCESS in 47m 39s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3a48ad363c8e46feb8a83c540a938e84 : FAILURE in 5m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e1268330429b4065ba15740d7e32e3ee : SUCCESS in 4m 33s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f85ae311f43c403e942fe56fa33b4017 : SUCCESS in 14m 53s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4c9bfe9332fc4849b0013e46fb672237 : SUCCESS in 1h 00m 30s","accounts_in_message":[],"_revision_number":57},{"id":"2d8101b97db8859b6ab89c4f3a655f831ca1d899","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-30 01:46:41.000000000","message":"Patch Set 57:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cf7cf97966474815afdc89a102f36356\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/2167677e8dce449da7a6063683eca2c2 : RETRY_LIMIT in 13m 10s","accounts_in_message":[],"_revision_number":57},{"id":"83b133bac47016140c25f0f7593f1964dba1571e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-11-01 20:13:43.000000000","message":"Uploaded patch set 58: Patch Set 57 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":58},{"id":"74d0201f4ae18c071c2eba4561de28fb00ead824","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-01 21:40:17.000000000","message":"Patch Set 58: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/87e7d91407d14c8eac90627b2faf1543\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b65e854109da41eb872858aedbc7366b : SUCCESS in 8m 00s\n- grenade https://zuul.opendev.org/t/openstack/build/654493c0f6484f52a410f3ecc3bcd989 : FAILURE in 47m 26s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e2c1a8df23e0447dbaf47805f10de792 : SUCCESS in 44m 59s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/9d4b13fda9e34048ab3012313203ce38 : SUCCESS in 56m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9f00e7b2e9d54239b00bd7783faa9fbf : SUCCESS in 1h 02m 20s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d1cf7eb616a24dc18a68509873523549 : SUCCESS in 19m 53s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/36bd1affe5d64571bb192ee9ac70bdc0 : SUCCESS in 23m 05s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a56024244fbc4c36814063f69fd72552 : SUCCESS in 13m 09s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/80f1407fa18542e0b145a7ea0ab74c98 : SUCCESS in 5m 14s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/27293e88acdd47df891bc7e674bb878b : FAILURE in 5m 56s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/8bf9671cd5914c9eb8d68d766778fc14 : FAILURE in 19m 15s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/c73ad81124c94790947da6643e437a5a : FAILURE in 23m 48s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/5ed3b896d1964c48a3b1cd8a68f3463b : FAILURE in 33m 04s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/231859e7d25d4ad499c26aa74d6d768c : FAILURE in 13m 34s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/420c4aaefd134e998da4b09f35261b13 : SUCCESS in 13m 02s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/8a9ee393affc49f7a1db0dda26d3d8cd : SUCCESS in 14m 13s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/8c5b35209ef54317b0550d4fadfab449 : SUCCESS in 20m 18s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/4a98e6de94e6450cb385f48c5a1e1ca7 : SUCCESS in 11m 18s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/ee3bd237d23c4963b1df861e9b772dbe : SUCCESS in 17m 59s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/fb16cfdf320b416e9bf22de5d86f2383 : SUCCESS in 12m 39s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/58349e21fe654fd69512b98071f4cb6d : SUCCESS in 5m 51s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b0e24eca33e64d29a9b1c123dfd01d82 : SUCCESS in 11m 26s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/834099bf229c41e39d7d45acbeb1a5ca : SUCCESS in 6m 54s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/913b64bd95094077b2d071dfb815b3b1 : FAILURE in 39m 08s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4f2657efbe2a4391bbdc4c8767a4a92d : SUCCESS in 50m 03s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c3683d05b00743e282277be6879d38e4 : SUCCESS in 50m 20s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5534b2fcb87242078d19fb30bbc247d0 : FAILURE in 3m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2c97ab816d6c441691a33e0746c40511 : FAILURE in 4m 16s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/5d44ffb8180548d482cbb56338f8f522 : SUCCESS in 17m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ed47e9bb34ee4bbca05ac53cfd911525 : SUCCESS in 59m 14s","accounts_in_message":[],"_revision_number":58},{"id":"a01000543c594c6c0144772c4e0322b45923d205","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-01 22:40:34.000000000","message":"Patch Set 58:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0f44ea28a902470f8ca99358ddfb776f\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/f12ff0c07d9c44cebb7ad2a73e01e654 : FAILURE in 2h 12m 28s","accounts_in_message":[],"_revision_number":58},{"id":"c4c86eadd35857a57b3f42b17cff1df30220d7cd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2024-11-04 05:23:17.000000000","message":"Uploaded patch set 59: Patch Set 58 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":59},{"id":"5883e1194879ad636feb3a741d7ea5ccd53c1620","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-04 06:34:15.000000000","message":"Patch Set 59: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7dd3f1612c2a49a08e639aa9fa4a06ba\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c00ede6f80ff4706b1d0704c3779daaf : SUCCESS in 8m 57s\n- grenade https://zuul.opendev.org/t/openstack/build/7d69ed9eef41476d9d80f1680a2ecd2b : SUCCESS in 41m 30s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/fc796354f9424ea4a563b8443f40838e : SUCCESS in 1h 00m 43s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/43dc8a2cfc7649529694c0e5e496bf91 : SUCCESS in 51m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6d747475560344bf9926fd72b4d9512c : SUCCESS in 59m 24s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6e498aca006742749cbf5ecdae9cecab : SUCCESS in 17m 46s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f5f1150fb3474a8bb64d48d4ed10383e : SUCCESS in 12m 55s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/466b9798615a4ed598421b26a6fc9eda : SUCCESS in 16m 06s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/0d38e42b3b9f4d81b96d1ae3ae182391 : SUCCESS in 9m 03s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/5cc38eaad7064d498f87e9087288c45f : FAILURE in 5m 07s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c6563360bdcd48048c3d6af72ed31f36 : FAILURE in 18m 13s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/929b3866db0145e89d0f55cbcc0b8a08 : FAILURE in 18m 32s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/d4fca63efe0446fc8715cbe189af3f0d : FAILURE in 25m 23s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/c7d272752aa442788ab8a0640bd52c52 : SUCCESS in 12m 47s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/5210d6b7bbca41bd85826ecfa73025e8 : SUCCESS in 14m 57s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/2ad1e42f1058419e8fb0944ba9250a6a : SUCCESS in 17m 59s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/b83a68f4c23d4f158dc66e8652ff1d6d : SUCCESS in 9m 07s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/d41500e259094484bf4ebeaef6e03ead : SUCCESS in 14m 27s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/bb419975ba3441a7b06652c01ba53bf4 : SUCCESS in 14m 41s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/4efea439852c46ad88d01f1845327dd1 : SUCCESS in 7m 28s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/d23a466492514ef789cf88549912d62b : SUCCESS in 12m 01s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/59436c698aa645778e1272c3f80324f5 : SUCCESS in 6m 22s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/849e4e9b3fe84975b5481b28aa009894 : SUCCESS in 54m 45s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/edd20de558f14116ab73b4377da1b99e : SUCCESS in 32m 12s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/047388db830647ad8faecbb8c108880b : SUCCESS in 51m 47s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3d404c53b65244d0a9aca674cd653fb1 : FAILURE in 3m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/53e8043844404c41a4f9608cf738e799 : FAILURE in 4m 06s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/15230c2025664f9494293d929aa57339 : SUCCESS in 14m 37s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ffcd97818ed1448d93c9510f5c67d6fb : SUCCESS in 36m 49s","accounts_in_message":[],"_revision_number":59},{"id":"8c46fb04588e0518cf1d1220948e48a246ab1fb0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-04 07:00:06.000000000","message":"Patch Set 59:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/28030909c1b54a0285b9ba60e432be1e\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/972e62238dbb44d58d976b91546c8af6 : FAILURE in 1h 34m 24s","accounts_in_message":[],"_revision_number":59},{"id":"97c18d99048594f48589523fa4f04abc5cf8d597","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-11-05 00:21:14.000000000","message":"Uploaded patch set 60.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":60},{"id":"de44017ff15c19577dfcc4db1fe8a86c1014d6fc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-05 01:33:04.000000000","message":"Patch Set 60: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f741b460256d42c2be3098ab32220f16\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f308736cd02a49caa42dd908b400f19b : SUCCESS in 7m 21s\n- grenade https://zuul.opendev.org/t/openstack/build/1d6e9128c7d64dbc9e44b5e234c06775 : SUCCESS in 40m 56s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/2839e0c1e4fa45a3a017005bfd55fd87 : SUCCESS in 43m 09s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/356dc98df7a24ca580d8f1613ff6ce6b : SUCCESS in 52m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4fc3675106c949a3a33dc119342df37a : SUCCESS in 56m 31s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/42948fd7410048fe917ae8a3ef337f4d : SUCCESS in 12m 35s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ebe295b7d3f0402b8b2a600454e7d6fc : SUCCESS in 19m 37s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a7ab2cf56b9f43a9acb1dea3096cd0f6 : SUCCESS in 19m 41s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b526f24e57bc43c2bcbcc7e3822f4520 : SUCCESS in 6m 55s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/4c4c4c8931a644968752f80878bc293f : FAILURE in 5m 12s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/1fcedbea2acd495496952164f26e8506 : FAILURE in 15m 16s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/edab835bfdb9413094975b5cde092270 : SUCCESS in 17m 32s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/29012542e214455cbcb6aa7d7bb74d96 : SUCCESS in 34m 56s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/a584a81145234ad6b0b485980a210f8f : SUCCESS in 16m 29s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/b6a9e16ebaab4a96bbd3e1b11086297e : SUCCESS in 13m 52s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/18fe34f623b541b3b83c4767633e0f1d : SUCCESS in 17m 04s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/192f9187c9bd4be7b94c92e3db3f6557 : SUCCESS in 15m 13s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/fe6775b0c4e44edcb30515c164a65624 : SUCCESS in 9m 17s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/518ec89bacb44bd78d59b38ea3748334 : SUCCESS in 15m 14s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/1262994c18a04b8a9b17c53862d4678a : SUCCESS in 6m 19s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/500257cee8254eccaeb91aa650e45d17 : SUCCESS in 13m 30s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/e5b5ab049d2144c8811c7e77fb977b2e : SUCCESS in 6m 01s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/fd80dab56bdd4aa3b2341662631f3c3d : SUCCESS in 50m 58s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c416c40172ff46cab3e299c30296a8d4 : SUCCESS in 26m 41s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/584066dc5a4841aea1d264ca750e1ca9 : SUCCESS in 50m 13s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a740c272d119440b9f79bb5b65d2e758 : FAILURE in 5m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/799d23abb1e344849426b6a7b4ebd0a8 : SUCCESS in 3m 41s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/8dff1030e2474dac946002ff215e28e7 : SUCCESS in 14m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/39cadee93b9e483a9c40b6edaf9f1d7f : SUCCESS in 40m 47s","accounts_in_message":[],"_revision_number":60},{"id":"dc75e17f6e8e1a40bcdd13068477d6815f1c446b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-05 02:08:53.000000000","message":"Patch Set 60:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4193473a2cd24fbe971815a0762bc7e9\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/d3756659cf1046c4a19f3b3e2f71f6c4 : SUCCESS in 1h 45m 16s","accounts_in_message":[],"_revision_number":60},{"id":"4be097f4ea8ccf9212c6a43ea38d869c25b8306a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-11-05 22:52:54.000000000","message":"Uploaded patch set 61: Patch Set 60 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":61},{"id":"ee23c22c2ef4017fc0d2a20bb5de0d1172ce086f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-06 00:00:58.000000000","message":"Patch Set 61: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c2574fe733f14ea9a5a9d47cd4993d2c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b69a370233074de4b6098a1adc4d5e6e : SUCCESS in 9m 00s\n- grenade https://zuul.opendev.org/t/openstack/build/a3cddd162749465f927c784bcd491455 : SUCCESS in 50m 40s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/40772c5d2086491a87edb27ef3829f2f : SUCCESS in 47m 08s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d98cced1a40b424a825e372e4af3b282 : SUCCESS in 36m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6d03a17f5ac9435b96645656977b912d : SUCCESS in 50m 13s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5da91250401243cb93f983c8793e8d37 : SUCCESS in 19m 32s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1b647309e1ed4faf8bb48f94997aeab0 : SUCCESS in 19m 27s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1c0531f851ea48159fcb25523b48fa2b : SUCCESS in 21m 55s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/0ac707a9a1e04f5388adb2ccc007a51a : SUCCESS in 4m 59s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/9293cf55426d432380756f24013addfb : SUCCESS in 22m 22s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/fe6fde6674724b7292363e296512a1b5 : FAILURE in 19m 04s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/2120bd32683d428588e911a223daaf9c : SUCCESS in 25m 17s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/88af227e908a448da75287a1cb5f492f : SUCCESS in 37m 06s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/a513e05507a543c29591fea33612aeed : SUCCESS in 19m 39s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/f4ea56db867a4b859587ed0a50261caa : SUCCESS in 20m 00s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/aeab75e25b1d4a6091d3db79588046ac : SUCCESS in 17m 32s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/af3a627147e540a69e376e0d71c530c8 : SUCCESS in 13m 51s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/de9309227f35453ead1a9c978aa22c36 : SUCCESS in 14m 58s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/5614dfdefac148cfb753e625cf63402a : SUCCESS in 15m 35s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b17148928c1840a0846456aee418f10b : SUCCESS in 6m 35s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/fc235384f115481298e39b2404c52376 : SUCCESS in 13m 34s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/a57eec74be4b4824b83cc6d944f10c13 : SUCCESS in 6m 39s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/020dd041e174464388f1746629e73b7f : SUCCESS in 48m 28s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/75f2b3ba2e7b478cb921859e549989b5 : SUCCESS in 1h 00m 51s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/82b099af7acf45ae8360485e0cf4e388 : SUCCESS in 38m 35s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/52c86f24bc874176b7a6762233b8c32c : SUCCESS in 16m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/31bf592fa086467aa437d31a1d7da1c1 : SUCCESS in 5m 39s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/41be6232145b4854a2f21bc8ab0a4564 : SUCCESS in 19m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bda3161cceaa4eb7813058c4200d84c1 : SUCCESS in 1h 00m 23s","accounts_in_message":[],"_revision_number":61},{"id":"a4abf028935f91b89c6b5866c1b6b3aa8d42ce20","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-06 00:25:49.000000000","message":"Patch Set 61:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f0cbb642101144638ffdddf954f4d3c4\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/516e48d5f1ac4345bb6adf3c067d7dd0 : SUCCESS in 1h 30m 56s","accounts_in_message":[],"_revision_number":61},{"id":"5bb6a7a8b1cb18804cc9fd1b8fdd0fe699962677","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-11-12 17:02:51.000000000","message":"Uploaded patch set 62: Patch Set 61 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":62},{"id":"8ec510a08fd00ec06c1ed82920157e7b3bd2ed63","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-12 18:14:38.000000000","message":"Patch Set 62:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a05c3f6c53774ead8a00f4f07c99944e\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/d887be0f96cc49aeb2dd0f6de073fa17 : SUCCESS in 1h 08m 45s","accounts_in_message":[],"_revision_number":62},{"id":"081bda1c858fd17ef22d9e70d8e7c793362074dd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-12 18:19:49.000000000","message":"Patch Set 62: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a37da12f602a4fada404577e8e9eeb2d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e674e2d14164479db84315cfdbcca3b3 : SUCCESS in 8m 33s\n- grenade https://zuul.opendev.org/t/openstack/build/f3fe30621af04224884a3027a3b0c29c : SUCCESS in 1h 01m 03s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/29cf65d5b0534b3bab66ef9fc0d518c3 : SUCCESS in 57m 17s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/4f38f1b1272e46e7ba9c90bc3726b68b : POST_FAILURE in 40m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/95c726e1057343f29a68663bd21f45ef : SUCCESS in 1h 00m 51s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/45276c3002924ad0b4652da8149b47e3 : SUCCESS in 18m 58s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7d56badd0cbc4853b8045a707ce6044a : SUCCESS in 13m 57s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b45f03bf3c2e41e99e27744de5a5eb52 : SUCCESS in 21m 25s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/41528ad2a6d14904af0af6044aa47b37 : SUCCESS in 6m 12s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/3a29d95aa9cd4a91ad7ab1aa9620da2e : SUCCESS in 21m 08s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/8355b638fd934199ab4a3b5530aee579 : SUCCESS in 17m 35s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/22219408e63d4ac18aea540406063297 : SUCCESS in 22m 39s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/b79ff42e4fa34709b4cb569748a7cd97 : SUCCESS in 28m 19s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/e2f764cf103949c2a15b0fc483521505 : SUCCESS in 16m 43s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/afddfbba1fe44f84884147f455a95b7b : SUCCESS in 16m 39s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/b54d058c7bc54de7af68a08b650becfb : SUCCESS in 13m 38s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/564d63afb75a456a9c9e5e78c06bdfb5 : POST_FAILURE in 15m 10s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/fb15c86ae3dc4f2ebb8dc5f369387f40 : SUCCESS in 13m 38s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/5befa1bebfc34e6db6e97001c840c6b9 : POST_FAILURE in 14m 27s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/03c7ad8df9d643ae91aced140652757c : SUCCESS in 6m 06s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/1a3603ecaa194e6ba5ecb25e98bf4a22 : POST_FAILURE in 14m 43s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/8993c5c686804e10a40324874516ec88 : SUCCESS in 6m 33s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/e556bb1cdb794a0bb019f603ac85311a : SUCCESS in 1h 01m 11s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/fc5aa675ccc6472baa39690430f4f3a6 : SUCCESS in 52m 10s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f061a085f7fe419baf62504c356b3bcf : SUCCESS in 51m 19s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/28d8535e9c7940b18eaecda141d70ecc : SUCCESS in 20m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/94bf1b1b679a49a99d20ac2fdc21d4db : SUCCESS in 4m 23s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c53c6196995847c9978558b905a5c6a1 : SUCCESS in 17m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/432ae20fa55040b1b8fa994ff6eadf3e : SUCCESS in 54m 17s","accounts_in_message":[],"_revision_number":62},{"id":"41bd4e790b6a794c5c40e2ede5def210d708ef35","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2024-11-15 18:59:30.000000000","message":"Uploaded patch set 63: Patch Set 62 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":63},{"id":"3121c613cfe0d025bfe2ee2970832c5fae46c448","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-15 19:11:55.000000000","message":"Patch Set 63:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/58adb96cafb4454a93911ad48d8643fe\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/86259cc1daf04c488f41d47102360b88 : FAILURE in 9m 38s","accounts_in_message":[],"_revision_number":63},{"id":"c2f7a08351d7c06e7c0722dd5a22b1afcf434b3a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-15 20:10:56.000000000","message":"Patch Set 63: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/710c2bd5eebb41b0b7e8fe32f875da81\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c219b1db81e44c02b1fbdf638f756dca : SUCCESS in 9m 07s\n- grenade https://zuul.opendev.org/t/openstack/build/1151baf2973d44dbaac88ad4d7e6ada9 : SUCCESS in 50m 51s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/fbad194e16104953810d0ba86790532c : SUCCESS in 31m 07s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/18833159ca334823a261c0352ac852ea : SUCCESS in 52m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/43969cd5c2ba4764a2bd72904c70d150 : SUCCESS in 1h 03m 06s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9fbea19a22bc4167b2de8f21bc05c770 : SUCCESS in 18m 45s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/15374986640e491e9cebd66eb2eb59ad : SUCCESS in 17m 17s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f7c09a8b0a39455787cb32b4a01ebf44 : SUCCESS in 21m 09s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ef28fdd0e25e4968a7af6b052675d3c4 : SUCCESS in 5m 36s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/3ed0cea97c1346358d3fa92d3bc40b1f : FAILURE in 5m 53s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/aafd711c305f4864955ddb370b304daf : FAILURE in 3m 19s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/c6949e64c9a44c1bab98561d0bc0a8f9 : FAILURE in 5m 49s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/c69279b9628240dc83773c56d0ddd950 : FAILURE in 8m 35s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/d7bc98e9fcf444eea25be3e214525317 : SUCCESS in 17m 11s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/c5c3245d6f2e4ec5ae5fd21c32cf1f6f : SUCCESS in 21m 06s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/14466eba6f464ccea93e85b2305ad905 : SUCCESS in 16m 07s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/fc4404507898453b88a77ffbf083c617 : SUCCESS in 18m 31s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/7cf1e423611e4258acb515fa6886fadc : SUCCESS in 17m 59s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/e321eff849934f92a52ddee69d3b953f : SUCCESS in 13m 17s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/56f8079771874965b5e918991c061ff4 : SUCCESS in 9m 02s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/07b1319384934a31bf643a0f124dd7cd : SUCCESS in 17m 18s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/c34bbe7d59bf4d2a9257e3dee7f8b7d9 : SUCCESS in 7m 04s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/2ac533313efc4fe7a4a578e18218b39e : SUCCESS in 52m 42s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c130ebc70edf461bb590144a465029a7 : SUCCESS in 49m 04s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/306f2ab8c4a84845b1f3764a4daa0aa8 : SUCCESS in 55m 39s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9ed5c574413a4811aacb897ea93bad7c : FAILURE in 5m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/33dfdd98e068495982420da292f4561c : SUCCESS in 4m 35s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d84b1ea65f15497eb56d1a815db5a882 : SUCCESS in 10m 31s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b45c5893b6e848afafdfed78245b41fe : SUCCESS in 58m 33s","accounts_in_message":[],"_revision_number":63},{"id":"755712556f6bc9ab0ad28807a148597043cd8895","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-11-18 19:09:46.000000000","message":"Uploaded patch set 64: Patch Set 63 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":64},{"id":"bf69042252d3697505f6fd4f3506dd76c2103a9e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-18 20:40:00.000000000","message":"Patch Set 64:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9451da7ae29a42dcb351dc7cdfeb938d\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/219cfc507e3a44648bd34c0306589587 : SUCCESS in 1h 28m 02s","accounts_in_message":[],"_revision_number":64},{"id":"35b9ff92951f869e1b12cb3a9400abf7020edc0a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-18 20:57:13.000000000","message":"Patch Set 64: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/be84976c0ef9475486067098edaf387f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/555c70381d68467fb683d0a9d182dee4 : SUCCESS in 8m 34s\n- grenade https://zuul.opendev.org/t/openstack/build/a5a70b60a32f450dada0d97c9a39d209 : SUCCESS in 1h 00m 29s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/fb582af4ed7544429a3cfce7e916caca : SUCCESS in 1h 06m 48s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/8b95966d42e04b1a9d13cb15dcc69d7c : SUCCESS in 53m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c26881ee28f648fbae066104db6ebbaf : SUCCESS in 58m 37s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/986cff2641ff43598117de794c4deebf : SUCCESS in 19m 19s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4923f9d8f4064592a7a2987e4b8ccd43 : SUCCESS in 20m 57s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ffbebd9d68c04f009f0b2149806ed125 : SUCCESS in 20m 01s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e4506e38b8cb481cbd469ff15419259d : SUCCESS in 5m 55s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/4ea4723c499942218e4a0ca32eec2163 : SUCCESS in 19m 35s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/3c2fed182d2c4f909be50a55d212f7e7 : SUCCESS in 18m 53s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/7ffb740bfc1d4ccc957e8675cc8413a3 : SUCCESS in 21m 04s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/6dd5c1b455074eba99c1122007bd9d83 : SUCCESS in 36m 27s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/28b59b34c2464a62a70b382eff6f3d6c : SUCCESS in 17m 16s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/104f3789f356411d8ef2d613a8cd4f97 : SUCCESS in 16m 27s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/7d8d1330fd8f4b8ab6062714a46ce367 : SUCCESS in 17m 11s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/992e5c16011444e48ff004ebb4698287 : SUCCESS in 14m 29s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/c8849a77ab954ae7903503329e10b557 : SUCCESS in 14m 59s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/f6193e9587164f5da43ac0021767e808 : SUCCESS in 13m 21s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e464bdb9bd4e4438a5d81b87ea0819f1 : SUCCESS in 8m 40s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/bb6458963ec34092a7556eeddd6bd52d : SUCCESS in 13m 51s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/eec8e20b5321411e8b18aabc5b0ff686 : SUCCESS in 7m 04s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/422e930611f6417386db5e12525ecf05 : SUCCESS in 46m 57s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/248196e0b0864a8f9b0a0abb86b5152e : SUCCESS in 33m 38s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f053a502e0894627966945371cdf7715 : SUCCESS in 49m 21s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/332018411a8b420ea5ea6b0431f25623 : SUCCESS in 17m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ccf05d710ec84cfabca034f489f98e56 : SUCCESS in 7m 30s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/4d330c6ee08043ddb2f469daff9deebc : SUCCESS in 17m 37s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fb5bcabd20da4da6bd8d43c270a0e5ff : SUCCESS in 42m 41s","accounts_in_message":[],"_revision_number":64},{"id":"ffc17e87b8e9fcbd6dc7a5154c9ede4f574055b4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-11-29 14:47:58.000000000","message":"Uploaded patch set 65.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":65},{"id":"4c9a79f2c7ac20d3c7907c5373d872e26203ddb9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-11-29 15:10:59.000000000","message":"Uploaded patch set 66: Patch Set 65 was rebased.","accounts_in_message":[],"_revision_number":66},{"id":"e6874795af0793ba51481163c751d379ebae6f2b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-29 16:40:26.000000000","message":"Patch Set 66: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/088125902e1948dfab23463bb641c58f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f565e5ebb9e941b9a836a9da46da53a3 : SUCCESS in 9m 05s\n- grenade https://zuul.opendev.org/t/openstack/build/1095d343ab0f487abd25625a6657e5b5 : POST_FAILURE in 1h 19m 36s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/9703b762ff904d299b0479db7496e642 : SUCCESS in 1h 03m 22s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f4be2948691b4f54879c76ccb95ca89e : SUCCESS in 1h 08m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9ec3a9ccbc5648cc9ea3f72aad9269c7 : SUCCESS in 1h 07m 46s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/bd99f0c101c249c78916d1aec8a1859e : SUCCESS in 29m 20s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/88a7257b27fa410eb984abf0dca2d4ce : SUCCESS in 25m 31s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a91213c0048f470bb15f38ec844e0630 : SUCCESS in 29m 16s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d569347da1184a9ebee58bbfb2b4c290 : SUCCESS in 5m 20s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/8ab29657e9234e17ba4164fb795d91fb : SUCCESS in 30m 52s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c5634c50fc2049af88d0ac443c9f0be7 : SUCCESS in 26m 51s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/41f5d212e7e746b9a0b6520de7e344c9 : SUCCESS in 31m 09s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/7b44dd6e458e4bc485e60217209feb95 : SUCCESS in 39m 36s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/57054ae47d1c418383ca909346887c37 : SUCCESS in 26m 11s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/5e7443723dfe4b73b2c99bf385aeadc4 : SUCCESS in 27m 35s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/a147dd07c4b4419aa4d7149632033a77 : SUCCESS in 27m 19s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/92a95f37cc1f429dbcd4e49f576452a3 : SUCCESS in 23m 35s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/769ecac73b574076957bd304a15bd6f3 : SUCCESS in 15m 17s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/916688946cb746c2a3c72f2699b834b2 : SUCCESS in 23m 19s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/ad8bf0e0935b412db3323281f14a2285 : SUCCESS in 6m 25s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/ef2a1d6802f94a5a96e2661fdf90ac27 : SUCCESS in 21m 28s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/28ef0542e8a5436aa7de85aa6dc3b672 : SUCCESS in 7m 15s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/5c44526fd17246b5bfac071e79c0e2bc : SUCCESS in 56m 11s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/49829f5963314dce82e27cbb16e6285e : SUCCESS in 48m 02s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/911b9e7980754eefbbe16bfcd503d35a : SUCCESS in 50m 34s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/eede613e18204e4c99a1a68370a27f21 : SUCCESS in 22m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b907e74fd1c94bf9ba96cebc80bc90aa : SUCCESS in 4m 15s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/fca6c70754ff4a558afd40b9828f1915 : SUCCESS in 29m 24s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5360e29c79c3476cb4ef57c12bffc538 : SUCCESS in 59m 39s","accounts_in_message":[],"_revision_number":66},{"id":"9e9f672f44bf02d239ba7e99c002f067b05c31bf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-29 20:25:46.000000000","message":"Patch Set 66:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/84024a89219e4c84be6cf03e4446c23a\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b07bc00bdd70400a81bf3b221bcfc6c4 : TIMED_OUT in 3h 01m 03s","accounts_in_message":[],"_revision_number":66},{"id":"495eaea8549078424d529759b6f176b0375d2579","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-01-16 12:23:57.000000000","message":"Uploaded patch set 67: Patch Set 66 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":67},{"id":"0a614278352c27d7826600f0cfe255f637e2bf16","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-16 13:48:38.000000000","message":"Patch Set 67: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d64b5ec8395d4fa9b87778f6efb25a9c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b19faee2a2e4041b8af1bdb6e742b93 : SUCCESS in 5m 05s\n- grenade https://zuul.opendev.org/t/openstack/build/43dcc311b19f4b63a6571637918ac0b0 : SUCCESS in 54m 41s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/0b7999cb58bd4e04a122c44e7dcd74f3 : SUCCESS in 53m 41s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/eb85ee42dd1a4c15a418f04d26b26f58 : SUCCESS in 51m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5213d0358454416d8847421e3ec24d60 : SUCCESS in 1h 03m 07s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0c02e66b5dd545a4a5665abb0844c319 : SUCCESS in 18m 46s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/bf03cdc9e74f44e799c34e17eb863f35 : SUCCESS in 17m 11s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/70576443217446a18ffe529a2a780925 : SUCCESS in 19m 43s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f3379b00ec89481597a77de64c700444 : SUCCESS in 5m 13s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/8714d5b884e24e46ad0e626e1a9c002f : SUCCESS in 13m 50s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/9743d2b5c6cd4ec6995102084f920f52 : SUCCESS in 20m 29s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/f9dd398dc08e40b9a96817b450b5840a : SUCCESS in 34m 27s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/79d95b8c24b94f5c8c83cdefae76b9ea : SUCCESS in 16m 18s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/72129f6f3e6d413c862902fa9b55038d : SUCCESS in 13m 37s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/9934fe0a0e9e4e3d9905c97f0e527798 : SUCCESS in 16m 30s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/faa1a8c0ff4d47e2a1f9d2bb37cd3d3b : SUCCESS in 6m 11s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/72f475de20a542c1ab8f4bb1e604d1d4 : FAILURE in 15m 05s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/a7b203cde8da4d6592a9c484d18db54e : SUCCESS in 7m 54s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/b9302025b05345d9aa4cd3026184bf19 : SUCCESS in 53m 18s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/48d3b71a93684250be2643354ecb1321 : SUCCESS in 29m 01s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ff5a7e7fcadd483ca81f16d8e87958d5 : SUCCESS in 35m 44s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/407e3fed4a484c35bbee069c25ada935 : SUCCESS in 11m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cb3a113482d04ed79f054e43a535ab22 : FAILURE in 4m 44s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/0666807b542d4d969e4680803ef4b7fd : SUCCESS in 24m 49s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/546710d05994484b86cd30deeace9739 : SUCCESS in 39m 33s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b0a48444a8044e029d279c06c315ad41 : SUCCESS in 47m 33s","accounts_in_message":[],"_revision_number":67},{"id":"cc096558b33585f28a69a0ebc36e38e43dd7617f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-01-16 15:11:24.000000000","message":"Uploaded patch set 68: Patch Set 67 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":68},{"id":"facf3548b73157b41481a132f1f7fc0a666edf8b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-16 16:33:10.000000000","message":"Patch Set 68: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/91c8045c49aa4cffb03b7b0d4947b007\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/67cc13a86f2b4c81998ffe733289dacd : SUCCESS in 8m 58s\n- grenade https://zuul.opendev.org/t/openstack/build/8bc9168ea5c2429abefa4cf8f798bfcc : SUCCESS in 1h 05m 06s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e546b477f8da4bbbb4f0648c731e88d6 : SUCCESS in 54m 23s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/7570b5edd2d04151be585ec7b04180eb : SUCCESS in 1h 06m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/14c7f4f67fb543d0b527ae5df629b43b : SUCCESS in 1h 09m 56s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e59021c9698e4fc79fd1f424a6bfcb7f : SUCCESS in 21m 54s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5439f742eec0447a8141273c9f3edcf9 : SUCCESS in 18m 52s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/167fe902d96f4e1199a514e685b02484 : SUCCESS in 19m 14s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/1f0b07918942498ca8fce22b250d5fae : SUCCESS in 5m 46s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/2e944aba0c744ec1947b57adff9ce21d : SUCCESS in 15m 22s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/e76b5798a23e4cdeab63a925f84c5ac2 : SUCCESS in 22m 25s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/bd497c66608c4a3a9ddc6e3755e2e3c3 : SUCCESS in 26m 11s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/1022b129091f4acbaf1f204d73a97ceb : SUCCESS in 14m 05s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/8a1eec44469541d3b677bd2f25a6f7be : SUCCESS in 13m 49s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/cdd86d8a2ab2456caad1fac407ed7eba : SUCCESS in 16m 02s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/7dc0e70fdb6a4aed8ceb1add6264406d : SUCCESS in 6m 10s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/8337b932374349b695c16b332f7cab6c : FAILURE in 15m 52s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/f620664becd845b6a9d917bd00cbda4f : SUCCESS in 7m 00s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/a51e660014374fe19601318417ddeae0 : SUCCESS in 57m 55s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/11e2ab77697e4cc9a33b5a0d5b92aca4 : SUCCESS in 34m 33s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b2791f163786421bb72548943d1ede6e : SUCCESS in 29m 59s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6560b040018345c1b579ebd322415acc : SUCCESS in 17m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a9c6aad5a44542a4bfa60b7753d2a6ca : SUCCESS in 3m 56s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/535ae335164e4a9d96829185afc7b679 : SUCCESS in 17m 20s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ec5349c2de214e39b92cc7d24a223aff : SUCCESS in 58m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2325607dbb5240039d2c1917bf965b29 : SUCCESS in 1h 00m 30s","accounts_in_message":[],"_revision_number":68},{"id":"1ed457c7df1e429e1d83d0bf069361d211851472","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-16 21:00:34.000000000","message":"Patch Set 68:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c389322f76064251b26d34831e81654a\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ef77f1eb6db044bb83ade0b456ea5aac : SUCCESS in 2h 03m 41s","accounts_in_message":[],"_revision_number":68},{"id":"de8474c8c23d82fa1792d412d76d0f50efdf2001","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2025-02-07 19:43:36.000000000","message":"Uploaded patch set 69: Patch Set 68 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":69},{"id":"4c0cf395a711b53f37835e5bcd2659ca94d757f8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-07 20:56:20.000000000","message":"Patch Set 69: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6d820bca5e45461e90114ab51e9dc469\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/02ebb4daf2944f87945458f210210f2a : SUCCESS in 8m 25s\n- grenade https://zuul.opendev.org/t/openstack/build/79217ca0151d470bb088b75a15d88a8f : SUCCESS in 1h 04m 14s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/949e6e6b57e749fba16908bdb2b5a89c : SUCCESS in 34m 26s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e360b779e7ad408aa8ff2460adefeb95 : SUCCESS in 57m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d354c205d0c64132a5df28481b718888 : SUCCESS in 1h 06m 41s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e21f963ffb3b4057b948ca0c90f25083 : SUCCESS in 20m 00s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6c3787feef6148c4aa656c833bb30f1a : SUCCESS in 21m 20s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9c086ffee32841afbaffc1bb2b23453e : SUCCESS in 23m 40s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/0244687009d145bdbf457bc5a2e029a7 : SUCCESS in 5m 29s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/597759040b6246c1aa18ff698b883aa3 : SUCCESS in 20m 24s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/3e48b8827a9e4c2493cc71bc04c17799 : FAILURE in 21m 16s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/2d8c97c54e494c108a5fc34413da45b8 : SUCCESS in 21m 11s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/94b4483725604dc382d5a99daa802e36 : SUCCESS in 14m 09s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/7eb14e803dbc4088867da905252a8ea6 : SUCCESS in 15m 55s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/23954bb73afa4466aaca70231811b0d5 : SUCCESS in 18m 25s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/30e8ff7c019049c08f49387342ba6ef0 : SUCCESS in 6m 54s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/57c8c575a0124203b68144d2129ff941 : FAILURE in 18m 15s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/613385c25d2c487f844d3d5ca4ec5031 : SUCCESS in 8m 18s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/89369bdfe66c49958022f58729710d69 : SUCCESS in 42m 58s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/7c491c04ab3c40b0b5e922d411d41204 : SUCCESS in 35m 04s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/3b7a68e32a1c47caae95dccbbf9b7799 : SUCCESS in 25m 19s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f8762f824941431aa19893e87a39c87b : SUCCESS in 10m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4fb5d62c7b5b4645812812e4dca2d4c4 : SUCCESS in 4m 06s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/df2f30cf5bb14ac3a0797ee1a2a5518d : SUCCESS in 18m 05s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b7969879d8434f18970b1e92bbfd4621 : SUCCESS in 45m 00s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d8662f4eaa924b7f9f33fe86a5471fe4 : SUCCESS in 38m 49s","accounts_in_message":[],"_revision_number":69},{"id":"158ed04b62be07ab33669f313c7fbc75edd6187c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-07 21:15:10.000000000","message":"Patch Set 69:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2e23990ffadb460e81f7dbc8af32f563\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/f845075af19643eb934a3013c2eba188 : SUCCESS in 1h 28m 53s","accounts_in_message":[],"_revision_number":69},{"id":"8b3c5be339c4cec8be75bff58fca71f19dd91476","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2025-02-08 02:19:28.000000000","message":"Uploaded patch set 70.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":70},{"id":"de26fb6fcd9951b470095b52d8299855a30568f5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-08 04:04:08.000000000","message":"Patch Set 70: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8165e81dd15a4ba9835a62561f3dd80b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de6c69e123a94415b692c95b30414293 : SUCCESS in 10m 37s\n- grenade https://zuul.opendev.org/t/openstack/build/d486daac83ee49008f0491ca67c1e414 : SUCCESS in 1h 04m 23s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/00691179fd3d416fa1d3c144d0326794 : SUCCESS in 49m 52s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c1a89ccb9b514ada9f84eef9317127e6 : SUCCESS in 42m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a4895c38777d466385cf7d840e449bc5 : SUCCESS in 1h 05m 30s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/62b0e027e9a5495a93006f275539471c : SUCCESS in 24m 10s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2ca253b5c4794560b55bfaf05acad4a2 : SUCCESS in 21m 25s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/cdf416931c4b4ad78ec26b1083d3e92e : SUCCESS in 21m 41s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/70f88d39b81c4834b00c1caabc629e6f : SUCCESS in 5m 53s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/ea7c4ba63d1744649ebd1b415e1fb9fd : SUCCESS in 21m 14s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/d7914246382a41c791dcbf4bed976bdb : SUCCESS in 21m 04s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/2d9b8b71f7154fc6823bcba2e6f7803b : SUCCESS in 34m 03s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/472a1af85107452c847b6bad70167e39 : SUCCESS in 14m 57s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/0381e26dc12f4f7eb3920e1c9d1cf9f1 : SUCCESS in 12m 49s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/1cfb402e0a1c48e3bc986b564c029aef : SUCCESS in 18m 34s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b42a879e56494cf9bc572d60aae4f89c : SUCCESS in 9m 28s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/8a9df0ede1da478f828d79ee7960f93e : FAILURE in 20m 03s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/80207e08e5b84e689977e5794f89fb68 : SUCCESS in 8m 58s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/3cead3eebfae44798f5bbabe3d5c036a : FAILURE in 56m 00s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/8090a6df7f0d4e5daeaf44a5a457111f : SUCCESS in 31m 36s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f79785e99e864913a35ed72fbf1b6273 : SUCCESS in 40m 59s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a7e98a7ef7f54485badd6352cf6580c5 : SUCCESS in 22m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d2b9e25434a44a15b4680e8c2c6061c3 : FAILURE in 5m 22s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/67e2797cace744b28cd3bdded602c834 : SUCCESS in 21m 30s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7a139059e1754f2aaafca5287ca91484 : SUCCESS in 55m 23s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/299107e77628474c8b405c4af148d045 : SUCCESS in 57m 51s","accounts_in_message":[],"_revision_number":70},{"id":"d5c555c4e118089f4b13058d0d0fb7ce19719afc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-08 04:35:49.000000000","message":"Patch Set 70:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e6359c9f953c431a91f20a277cc88ff8\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/8f9840183b604f7fa85903840815a7bb : SUCCESS in 2h 12m 51s","accounts_in_message":[],"_revision_number":70},{"id":"5d295a5d5bc600840389a11470362fc6f4dd67e5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2025-02-10 19:59:11.000000000","message":"Uploaded patch set 71.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":71},{"id":"3c7e4cb1f5b253c4c6738eb8ec4b7a687ec19af0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-10 21:48:48.000000000","message":"Patch Set 71: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b03c75584be54c1d85912a2f696364ff\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0fb00e7864fc436c8863958424317f98 : SUCCESS in 26m 05s\n- grenade https://zuul.opendev.org/t/openstack/build/10b759f1d7434b0d8b5d2b7e8c04e2e6 : SUCCESS in 1h 24m 07s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/2c3dc2e9fa354fa58e94a97919f34ca3 : SUCCESS in 57m 51s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/2a7db4bec11a4bc6ad168041e9ae1800 : SUCCESS in 50m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ff4d6e2ec7bc41d08236ff8519435afa : SUCCESS in 53m 40s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5d039496731a456086b359b05a23e247 : SUCCESS in 18m 17s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b9a0c40e0bd540758d2e2a10ba72a76e : SUCCESS in 20m 25s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3c7f08cd64fa4b41811a940e2ade9419 : SUCCESS in 16m 51s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/7a225f434b7f46d0b2b31a8deda4c627 : SUCCESS in 5m 52s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/2860a464580c44b8ae89a6f71cc9f6e3 : SUCCESS in 20m 24s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/b8d4ec4d3f14479bbe680aca61398271 : SUCCESS in 16m 08s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/6ebf9830f39f45e5ba3e02d17f824f3b : SUCCESS in 19m 27s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/af879dd27adb4da4b01c8741b26fda70 : SUCCESS in 14m 25s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/cad9d1177cfd45ccbcf4256b6db91c4a : SUCCESS in 14m 42s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/8e7ed99f97e348b08a5d7e7fef92f329 : SUCCESS in 13m 44s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/1f9b71bac6be40f49b6062b44cf844a7 : SUCCESS in 11m 02s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/0443566fe7614cf194cb30b09ded603c : FAILURE in 13m 44s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/29e026638f8f41f48c29bd69606a43a8 : SUCCESS in 7m 48s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/899b5b83fcc44b4fa4ab69a5540096b7 : SUCCESS in 1h 09m 44s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/24c0fc6dae66412bb515f321337da683 : SUCCESS in 34m 04s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/3ad4f6f95d4f401594c9825b302d8dc7 : SUCCESS in 56m 02s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e3473ec5272a46ad993fc921a792807d : SUCCESS in 21m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/354f180abf474754b1d8934caddb352f : SUCCESS in 10m 26s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/baf4db34bc4f42c78c0ea78379719e8a : SUCCESS in 20m 54s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a15a2d37fe1148d1bc6df35b942b30a0 : SUCCESS in 40m 54s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/befe24013228440ab1b4c9d9c985ba9d : SUCCESS in 44m 42s","accounts_in_message":[],"_revision_number":71},{"id":"bacfb5d8fc496389cbfdd1b6222cd2749e00bbb3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-10 22:48:05.000000000","message":"Patch Set 71:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d761b56d5f9b438f8d37bcf3b4ef03a5\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e3ec5e09ded540e6a1f18e17d098be5e : SUCCESS in 1h 00m 35s","accounts_in_message":[],"_revision_number":71},{"id":"95a1fb0c64b592404a5b9395b9fef356eb439dd5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2025-02-18 19:17:44.000000000","message":"Uploaded patch set 72: Patch Set 71 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":72},{"id":"e9321b1ff40b486ad16091b2cad20856d5772152","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-18 20:37:40.000000000","message":"Patch Set 72: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3ee7c33f5ce2481d8cbe1f431fea7037\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee8ae0fa4a474c12a139e43573819f0d : SUCCESS in 8m 19s\n- grenade https://zuul.opendev.org/t/openstack/build/99266ae8d9254a03b57f5fd0b109004a : SUCCESS in 1h 03m 43s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/84244438946e47f2aaf94910897bb833 : SUCCESS in 1h 09m 52s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/385e74cf21e9481589b023c8c8bfd6f6 : SUCCESS in 57m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c5f3863e9e534d169ff1d89437d1ec83 : SUCCESS in 57m 14s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c9387442e2f44641bde60b3eb13eb50d : SUCCESS in 15m 18s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9d06868115984914b267bcb168e20184 : SUCCESS in 14m 58s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/bf7d48fd65d14b88bf8637ff1ce28644 : SUCCESS in 22m 06s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/25666ae143f748de843f2fd9c7b2cd2b : SUCCESS in 5m 46s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/e094f32daed240e8ae50f28f33ee27fd : SUCCESS in 18m 08s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/78e712cca064421180b74312bbb687b0 : SUCCESS in 19m 38s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/e756894239404111a489ebd73817e3cc : SUCCESS in 22m 33s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/7ecefd4e16714677ae1c4feb68cbe9f9 : SUCCESS in 15m 20s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/35ce45f8f46b4a5dbdb95d01efa641c1 : SUCCESS in 14m 39s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/1e7ac81fa8ed489ca4d0692d342c6e49 : SUCCESS in 16m 59s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e814d93d8c6749ea8ce82367d2cf0860 : SUCCESS in 6m 48s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/3b0ede1c90594938ac869468654e8e39 : SUCCESS in 18m 30s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/c2c8f4023b3b445d92ead78de6461761 : SUCCESS in 7m 54s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/b45e94dcb7cc458eb91a60f8cccc80b5 : SUCCESS in 59m 53s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/6e9ed71407fd4ac3a8e27aca5fb6d926 : SUCCESS in 1h 01m 04s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/9400fbe854544f079d540e32dec1a92c : SUCCESS in 57m 46s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/01bfa7d151f34d1fa46cd465a57618bf : SUCCESS in 21m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dd147a5d01c34242ae5f5a9042ba7356 : SUCCESS in 4m 26s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/3f81708780d945b9b023ef8a2b5a5cf5 : SUCCESS in 18m 31s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ac394946ca2d45dd939f8fa42fefb1e6 : SUCCESS in 1h 04m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/86c64ede46084443878169ad1c95dd4f : SUCCESS in 1h 04m 05s","accounts_in_message":[],"_revision_number":72},{"id":"c35693be90745044281507d1dc44000f5498c390","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-18 20:49:38.000000000","message":"Patch Set 72:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/02358fe8c9294ef3a37797fe6a02b4b8\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/4c8ecbf13ca1414b8cb149b124cc9b41 : SUCCESS in 1h 26m 36s","accounts_in_message":[],"_revision_number":72},{"id":"3ebad21d9b9ba105fd7eb599d1fa62f7b447a949","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"real_author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-04-10 21:43:25.000000000","message":"Patch Set 73: Patch Set 72 was rebased on behalf of \u003cGERRIT_ACCOUNT_34930\u003e\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"}],"_revision_number":73},{"id":"d3a948d526d271c739a9faac66188eaf0b06db24","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-10 21:55:34.000000000","message":"Patch Set 73:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f89ca441c83445fc8b196e71ea34411f\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/09ab3d19ca7142a1ac2bdbfde8f64f8a : FAILURE in 10m 19s","accounts_in_message":[],"_revision_number":73},{"id":"83acbfe2352ffc375d017b7bda37c4a631508cec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-10 22:46:40.000000000","message":"Patch Set 73: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/90ff488b96cd46d5991bba5d3a702a14\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f73daab6b7d489db17661f7b058b598 : SUCCESS in 9m 55s\n- grenade https://zuul.opendev.org/t/openstack/build/c205ca5e66df443c849bdda26f946158 : SUCCESS in 54m 50s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f68e37afba324dc586e2bf6595fb35c3 : SUCCESS in 58m 59s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/7215b50d34b843c0a4111ee6f46c9cf7 : SUCCESS in 48m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/40a1cb5608464bbd8a1fc9f99c2cd5bb : SUCCESS in 58m 21s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6dc18a3dd94540bfa029348b4802021b : SUCCESS in 20m 23s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/935e7bd7cf0e48eab2861cdfd2f1a8c9 : SUCCESS in 21m 11s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b1654b7c7ced441888a848ac447cb767 : SUCCESS in 21m 05s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/6c4c9dbf9779460d898f7f8e85a6f826 : SUCCESS in 5m 17s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/cd957ce2f390492f80127a501381b1e3 : FAILURE in 4m 33s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/962d169495f94d5d827437b09fc3df0f : FAILURE in 4m 31s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/557124108cff42af898195d845a47f89 : FAILURE in 6m 34s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/09b431f1849748298435d2a56f18ec19 : SUCCESS in 7m 29s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/5af006248dfc441f85f32a6f94e73580 : SUCCESS in 16m 08s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/98a0b2dca697400a8f99554cf98d65e3 : SUCCESS in 14m 27s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/48c53979189b4defbc133de2f9a8f505 : SUCCESS in 3m 16s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/6f82c9b204d842b88b10fd8d4dbfee70 : SUCCESS in 9m 19s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/bc1e10ab6e4c42d6ae377609bddb4f82 : SUCCESS in 8m 44s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/438839d3a6694d099ebd3a6273349654 : SUCCESS in 57m 39s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b8a0e5dc3be44bb9bbac7d95915e6a57 : SUCCESS in 42m 17s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ad5a79eb1da84814b2ea2747f3c4fdc3 : SUCCESS in 41m 11s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4082c08d074a4c00941f2e122999f119 : FAILURE in 3m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf82b17c63e248e39db837f4c2a8534c : SUCCESS in 4m 00s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/5b7b8c2f6aa2408bb6a128f0143d17b1 : SUCCESS in 21m 04s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/93d3cc941feb456c96ea54aca9d60971 : SUCCESS in 38m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a2822dc131f345469f38f55eeafdc922 : SUCCESS in 45m 19s","accounts_in_message":[],"_revision_number":73},{"id":"87b66329aecc71e4fa768b102b444e13c26f3def","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-04-10 23:07:01.000000000","message":"Uploaded patch set 74.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":74},{"id":"5dac8cc6154bb091f54694b5fcacb1290dec6803","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-10 23:16:50.000000000","message":"Patch Set 74:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f0e96ab10aee4bc19384b139f64a9124\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/0602796d102a42e8866a1b5a7c68b70d : FAILURE in 7m 42s","accounts_in_message":[],"_revision_number":74},{"id":"7055011f65ee7ee2ea0c3a73cf427897af18dd9f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-04-10 23:19:17.000000000","message":"Patch Set 75: Patch Set 74 was rebased","accounts_in_message":[],"_revision_number":75},{"id":"f9387783302ecc1c8c03a9717a079b842859b9d2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-10 23:29:51.000000000","message":"Patch Set 75:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/954ecc143823457e9159cadac4346d53\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/f363b5f780684d50a3d11bf0bea31b36 : FAILURE in 8m 39s","accounts_in_message":[],"_revision_number":75},{"id":"82a80dfffb02d025dc2ea1f5de7bf9490dc2daf2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-11 00:21:38.000000000","message":"Patch Set 75: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/720a5492a7704555acf3bb6ba8dad295\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b8a615803375450a9caacd00ee210a52 : SUCCESS in 4m 35s\n- grenade https://zuul.opendev.org/t/openstack/build/4a438ecabb9c4616af3c8e970b575db5 : SUCCESS in 54m 57s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e2d6c8f85a7e45a88c034e2934ea4e68 : SUCCESS in 53m 21s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/7f5efa37fef04db1981a417e66294ca4 : SUCCESS in 46m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2ce8eb68bc8e42978723074d8cc5d462 : SUCCESS in 53m 47s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4b84bc3366f34fe695022d7f9790ca11 : SUCCESS in 17m 21s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e84ccb71fd8e42979659e061e7af4e42 : SUCCESS in 11m 48s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0a0e03e76299423cac698d79396c07c1 : SUCCESS in 12m 05s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/18ea607d4943485ea31d03ea530a3ee2 : SUCCESS in 4m 00s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/8caf194ec1dd489c9973f477bb2953fb : FAILURE in 3m 21s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/bd8118f71f9945c3ad4183d0aa701fa9 : FAILURE in 3m 19s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/8d30154ec9114536b60be7944a979319 : FAILURE in 5m 24s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/e41ab4670d98421094a5b94eb37db885 : SUCCESS in 9m 15s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/a232680d8ba74e76b734ee7222cb04bc : SUCCESS in 12m 34s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/25ac490f9622433ca1d08c7ad052c3cb : SUCCESS in 13m 17s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/accb6aef14994a68b97d44d5e287bed4 : SUCCESS in 6m 33s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/2779facbe63a4c4da89b37981c3f2abb : SUCCESS in 13m 17s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/c08a83b9376843a79af928d338a05c7a : SUCCESS in 6m 47s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/baeea7ec450247c38bff8e1fd0e01d2c : SUCCESS in 57m 42s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f6696f8e632946a38b2da95c4e413010 : SUCCESS in 30m 07s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/9c82a46b8389442cbd5b3e85e7c359e1 : SUCCESS in 29m 52s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f593b157ba8c47b5a99aeccc713e078a : FAILURE in 4m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7fd45b2b30e140a180ad2b57eabb4c99 : SUCCESS in 3m 57s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b58cf4f2803a42518d4526dfc0eca5e5 : SUCCESS in 20m 59s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ec8235bc2b2c4070b2287285fb69f465 : SUCCESS in 27m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c5fc3e00e30648a49349e0ad5e46c117 : SUCCESS in 55m 12s","accounts_in_message":[],"_revision_number":75},{"id":"57d187870472e74b42fbd4dc4cdd303795c53003","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-04-11 09:53:56.000000000","message":"Uploaded patch set 76: Patch Set 75 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":76},{"id":"09cefd62c1aa6487b5fcf9bf241c3f430565a06c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-11 11:05:59.000000000","message":"Patch Set 76:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0fb2da7ceff14070837ce9a42884a3fa\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/4cb110a247654011a04b758feac4007c : SUCCESS in 1h 09m 56s","accounts_in_message":[],"_revision_number":76},{"id":"2bc8943ab720aec1d7adcbc0394f3c040f0a43c9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-11 11:08:39.000000000","message":"Patch Set 76: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fa20ab2241484fe8aa2289351afb8306\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/81e326d345584016b8320dbe3658d78d : SUCCESS in 8m 11s\n- grenade https://zuul.opendev.org/t/openstack/build/b56c82706fa4434a857df67e66f2b3ff : SUCCESS in 49m 32s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e6e2b77b6b704b679eeac02eefeddbce : SUCCESS in 28m 13s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/30c425b953fc4790afd85d2d9b6fe0ad : SUCCESS in 53m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0e82e293373044cc8da038bec522a7b1 : SUCCESS in 57m 43s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8ef6494b456e477980dbc8049e546d5d : SUCCESS in 10m 19s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d8f942385d994e7f9797c95fca66a259 : SUCCESS in 19m 01s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/87296f178df34672a2623054ee71d9e5 : SUCCESS in 10m 15s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/70dbd829feef49d488cf1191afab6a67 : SUCCESS in 5m 13s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/da11b6b269ce4cf6b290e2fd015fc7ab : SUCCESS in 17m 50s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/1909d92625514b3895dfcc705dc2e864 : SUCCESS in 19m 07s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/4fc7d2853a1b4cffac5723b88e8700f5 : SUCCESS in 19m 19s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/cc2ba0455d104dbe94f6b07cc3402ade : SUCCESS in 15m 27s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/1672a821f12e49f4adf55c9ba01bc638 : SUCCESS in 13m 12s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/4514442cfccb4ad99c6557ca55d27445 : SUCCESS in 15m 32s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e54df949a331495aafee3d4dbd2d8fec : SUCCESS in 6m 43s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/a381f3a9394448df9fffa34267b19b70 : SUCCESS in 19m 10s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/5ade52fffda44baea275d849862204bb : SUCCESS in 6m 48s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/47bf7906790f43ddb9ee7f5d6791b958 : SUCCESS in 49m 12s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/09ed0b904bcb41cb9d5c39e196f0841e : SUCCESS in 38m 25s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/23b9daf87e024d1da057a05092c15bb5 : SUCCESS in 41m 19s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7ad0473fc7204e41b709c9e248aab1cd : SUCCESS in 16m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d7b4cc69e39b4217b28763cb31dea4c5 : SUCCESS in 5m 21s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b10c65bb70f5409e8896961db8f8d0bd : SUCCESS in 17m 58s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b22ba274140c4dcfad49c8864aa82520 : SUCCESS in 55m 21s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3fe55c71b1d54dd180b11af252d9246a : SUCCESS in 51m 28s","accounts_in_message":[],"_revision_number":76},{"id":"26da781954b96d24136ad172244abadc4fbe2826","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"real_author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-05-14 18:38:33.000000000","message":"Patch Set 77: Patch Set 76 was rebased on behalf of \u003cGERRIT_ACCOUNT_7847\u003e\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}],"_revision_number":77},{"id":"78f8b519c6dbfabea91d47be3fe8120427d8d00a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-14 19:55:13.000000000","message":"Patch Set 77:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/acadd69f642041ffbaa2a310439437bd\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/0cfc3b49ed6d444797095889330558a6 : SUCCESS in 57m 02s","accounts_in_message":[],"_revision_number":77},{"id":"f7dde1bc06538d835f6fb8f78852e572c0eeafa0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-14 21:48:14.000000000","message":"Patch Set 77: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6243f291018048598d93e03e535b041d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bec9b33708514068ac2be10be947f60c : SUCCESS in 6m 59s\n- grenade https://zuul.opendev.org/t/openstack/build/b50b33828efc4a118c4d8e46666d8685 : SUCCESS in 50m 33s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e0a8d0d23be64d6c92605ad71d2fc257 : TIMED_OUT in 3h 02m 25s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/36449036c9974066a235b0843594c634 : SUCCESS in 47m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/382c2ced00bc4e6eae5982992207eed6 : SUCCESS in 54m 47s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/48ccc6c4068f4f4dbc793a471497313b : SUCCESS in 20m 04s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7ead53f62728479abdd68ba577c33f7d : SUCCESS in 18m 55s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/daeecb01fb7448a88530632995fd5118 : SUCCESS in 20m 15s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/1bef431245f54c918bc1c8d9a530b70c : SUCCESS in 4m 55s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/c3939c4e2ad74b7f9e7d4ccf750bc811 : SUCCESS in 22m 08s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/cb00b1a9ab0648f38afcd76665fa6b64 : SUCCESS in 21m 11s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/3a062bfa50ea444abc4a092eab201093 : SUCCESS in 26m 22s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/56c734d3afb24a889fe417aada0536bf : SUCCESS in 27m 19s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/92943ba589ba4c38ae08a81966c15ae2 : SUCCESS in 9m 27s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/cef0011ca2684a1baf644f17c8be6558 : SUCCESS in 14m 08s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/dc4ad87993cb47ec9931701f29df6039 : SUCCESS in 13m 15s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/37bdb3fff55e4c3e8004d1bab74e67aa : SUCCESS in 5m 57s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e2bb2230ba86444b84d4414b0e5b7f9d : SUCCESS in 14m 33s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/56ce81fad69e48bbb1af5e9cd8e78f51 : SUCCESS in 7m 02s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/56e6244cf70b440f9fa268c664e4d457 : SUCCESS in 55m 19s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ae714650454b48a7bd84aa8b2f788cb8 : SUCCESS in 46m 04s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4da1961fbdb24af3bdb6ad0028f2db2f : SUCCESS in 43m 45s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/483bb7f2d66a481a836f3f7a5cb962d0 : SUCCESS in 22m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5519dfbb547940288364625e1ce68140 : SUCCESS in 4m 11s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/01a3b1029efa442e84ac7b35635e8087 : SUCCESS in 17m 39s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/602fd6206067437cbbb81eeb517cfe11 : SUCCESS in 41m 06s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e601acd444534fec9dcc9f0b5ee2d44f : SUCCESS in 52m 26s","accounts_in_message":[],"_revision_number":77},{"id":"dde3062740a74e76066fcd23af5f134a7acc314a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"real_author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-05-20 01:24:51.000000000","message":"Patch Set 78: Patch Set 77 was rebased on behalf of \u003cGERRIT_ACCOUNT_7847\u003e\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}],"_revision_number":78},{"id":"c90fd4799283bf049ad4d6b1c87e2ae619e41202","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-20 02:23:50.000000000","message":"Patch Set 78:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/00e2bdc5d50d4d61bd56437d48b176bd\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e36503af26d5407f903c4c586c893847 : SUCCESS in 57m 18s","accounts_in_message":[],"_revision_number":78},{"id":"9f406ded6e3bcccbd37721797ae2ace69016d915","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-20 02:37:58.000000000","message":"Patch Set 78: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dfcbb0f277f84a6d87b38d709b42f843\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d951df4f67154c7c9bea48ff04c965c5 : SUCCESS in 8m 47s\n- grenade https://zuul.opendev.org/t/openstack/build/7b0adb070efd4bc49255ee0045291d81 : SUCCESS in 57m 39s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/4d56f1e0ca4347258351c31d9a750519 : SUCCESS in 56m 14s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/9e6c5b1b7d7f44f5a8f54354e77cea21 : SUCCESS in 1h 07m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c4c8207b5a404279bb386061a6808320 : SUCCESS in 56m 12s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/68402a12c6cd44a6b8c41bb8b1f57058 : SUCCESS in 18m 24s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/11e4a409e8c045619db15a1ca2495459 : SUCCESS in 19m 16s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/cab2426c852b4979bae1dc1ab33cde61 : SUCCESS in 10m 20s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e4d8e650095b4a0590678eaddd325899 : SUCCESS in 4m 50s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/aed78b4352d34b9b98ed3a5d2d7f404c : SUCCESS in 21m 18s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/02583a9958504dbd935c234eb747e13f : SUCCESS in 19m 10s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/4ca314de59f4481f98fadf6f45c534a4 : SUCCESS in 22m 08s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/958a18f30bef4b1bbb6673a667ae81f9 : SUCCESS in 24m 25s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/2388eb688d564a5e929bae4e32d2019c : SUCCESS in 15m 09s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/c258cd1e32114481bdaea4c74f8bcd70 : SUCCESS in 15m 06s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/5c6c81a81fdf48b999944bea7e7c7dc1 : SUCCESS in 13m 11s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d881cddfb8f24c46bff5abbc4c9f600f : SUCCESS in 6m 00s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/64ca643f26c64c0a8f9c6739527f3987 : SUCCESS in 19m 30s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/679042fca5454a969fbaf15c3748e7a9 : SUCCESS in 6m 50s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/70421ff1662647bc99c2bd525a936b8c : SUCCESS in 53m 38s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/bd2a138db9644f3392529ea4cf0a9f67 : SUCCESS in 41m 59s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/088d70231a6042cba136d957e6c34a07 : SUCCESS in 36m 35s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f78a827a956c4d91bf4fd7632cc1cc2b : SUCCESS in 22m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9b431e35bcf840a38061e39dead03011 : SUCCESS in 2m 16s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/dc9769f15a83451a962216f66f973a6a : SUCCESS in 8m 29s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bd5b0cd76590452dbd8c25616bc7c823 : SUCCESS in 42m 41s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/faabadf933ef43d799bbb4227c707624 : SUCCESS in 56m 23s","accounts_in_message":[],"_revision_number":78},{"id":"751a6091e7510430dd8f8bc41b1ec5e63d67243f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"real_author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-05-20 14:27:32.000000000","message":"Patch Set 79: Patch Set 78 was rebased on behalf of \u003cGERRIT_ACCOUNT_7847\u003e\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}],"_revision_number":79},{"id":"e83a16bee5defa817d4d906965409822863a64fe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-20 15:52:02.000000000","message":"Patch Set 79: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eccb1d42b1a945328a56f6a61081efcd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7826f914bb934c28b55b63095b3c5790 : SUCCESS in 9m 08s\n- grenade https://zuul.opendev.org/t/openstack/build/d6057d70fef54d7bb1f1d8817d783646 : SUCCESS in 57m 33s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/655ab0e00ae343b4b06bf8a955b47cd5 : SUCCESS in 58m 48s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/69cb957902634930b35151d0211cb33e : SUCCESS in 53m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1eb653dda5c84031b67d0489add3ed39 : SUCCESS in 57m 53s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/06ac9dc756f9435a939dac45619747a0 : SUCCESS in 11m 36s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/428ebcc705f643a7bc46083e6baabc88 : SUCCESS in 18m 17s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d8bcf3b7cbab4ad6af80b38b0488b31e : SUCCESS in 12m 35s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/3cb9d4a57a5d41abbac994bbaa1aefc8 : SUCCESS in 5m 09s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/36875028e125489eb1b2c44fca9df110 : SUCCESS in 20m 47s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/3b22cabc04e14dd7ad55b4b26533a185 : SUCCESS in 21m 57s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/68d65295694f4a19a3c4127ffb660919 : SUCCESS in 22m 34s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/521cf42f67d94158a6ee26f8d0de4650 : SUCCESS in 25m 29s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/20b087cfb8754f458aade5f801d90500 : SUCCESS in 14m 44s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/3e3685f81a8b435a82a8366333062c0e : SUCCESS in 18m 58s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/c0844ae77af04e499949a8a7f9eae67a : SUCCESS in 8m 18s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/45bfcec53c9541e6b55b139e382943ad : SUCCESS in 5m 48s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/5188e7b1cb19436dbe34ad05f706bb92 : SUCCESS in 19m 11s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/6cce1f5484144b1b82e9efbdef9ee1ab : SUCCESS in 9m 45s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/f3c76120ba1b4c80bec577a85eed86d5 : SUCCESS in 56m 46s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c22c9e3dbaaa4a8586c2651c2a06b49f : SUCCESS in 42m 16s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/51d9302aebbf4081b1651307e68be601 : SUCCESS in 36m 28s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b7e08dad5eed40538a3affa2323cc569 : SUCCESS in 20m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/930a3e4d6afd49859dd78279a545db99 : SUCCESS in 5m 16s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/49f0df9949e9449c9f2acc60b021e2cc : SUCCESS in 22m 31s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/720693df362d44a8a8d7f6952d2cac2e : SUCCESS in 46m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6da0b01af5994c0ca7fe66d66d9f4610 : SUCCESS in 54m 21s","accounts_in_message":[],"_revision_number":79},{"id":"9e962df0b22e6f3017934e47c1ca81e241303cc3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-05-20 15:56:54.000000000","message":"Uploaded patch set 80: Patch Set 79 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":80},{"id":"9c2f44f4d6aba8ed155de0dd3de0b211f8c67f90","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-05-20 16:47:05.000000000","message":"Uploaded patch set 81: Patch Set 80 was rebased.","accounts_in_message":[],"_revision_number":81},{"id":"e43287921ae1d0bc9313bdb45c0998575a64b1cc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-05-20 16:48:05.000000000","message":"Uploaded patch set 82: New patch set was added with same tree, parent tree, and commit message as Patch Set 81.","accounts_in_message":[],"_revision_number":82},{"id":"37e95f2d33ffa8b8198beaf3bdc0c41317d045d3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-20 18:09:37.000000000","message":"Patch Set 82: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/28faac529be94df89a7b81e90545e53f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ba3abae783294d95bd37c9f9f960fb89 : SUCCESS in 9m 23s\n- grenade https://zuul.opendev.org/t/openstack/build/6e3f40c1e181487494a004b594f1e57f : SUCCESS in 57m 23s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d5b735e62deb495a9c67103b364befd0 : SUCCESS in 36m 40s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/77a16225b4804a06a70c30965865b312 : SUCCESS in 52m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2a9de9d5e5ca469a89084adface118e5 : POST_FAILURE in 5m 14s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a9058ea926264b87985145fb394140ce : SUCCESS in 20m 32s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5383532a39d14910999b2eb5f9b31153 : SUCCESS in 21m 05s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/175bb18261004c6fa3991b4271416633 : SUCCESS in 20m 58s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/9185f79f8a4344558b669d4a8dec6f76 : SUCCESS in 4m 47s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/ce7ade610ac84c7c816522cb9e9df933 : SUCCESS in 23m 04s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/15e99369a68f49ceac827821566f177f : SUCCESS in 19m 50s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/71e7a578d9544316b5d230936fe7fd2c : SUCCESS in 24m 32s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/7f277c20cd044650a16600a874c20e09 : SUCCESS in 37m 34s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/7e138efa10634f94bddff925b2ffe42e : SUCCESS in 14m 33s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/22f3db6d3da74e8e8a8d9f3b20801164 : SUCCESS in 14m 24s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/8a56b3a89d3c4b7096caef1b05b4addd : SUCCESS in 16m 55s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/bd9348a0712244c3920623724cd809c7 : SUCCESS in 10m 27s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/8b2b5c1155e14cfab688298635f04ea1 : SUCCESS in 13m 56s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/1d5ac781a91b40c2a49dbe908e167363 : SUCCESS in 8m 31s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/5cb5cbf60e0843149cbf83a3dffadfa7 : SUCCESS in 1h 12m 22s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/dd55c3a7f3984b959f21b8d6c8e95e49 : SUCCESS in 35m 35s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/1197bfdac0934d719be7f68ce7aa78d7 : SUCCESS in 34m 48s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cdd4791ff90f48b18a7e6396ee909733 : SUCCESS in 21m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eb8bfb6bc4d4478685424c94b1bada49 : SUCCESS in 6m 57s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c00d889d40a04a70bede09c9782e5a67 : SUCCESS in 19m 07s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cf453b2e397a4fceb5fc57e8e073bd6b : SUCCESS in 51m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b353be35fd2748e8a6ebb4517d6e7b2b : SUCCESS in 27m 50s","accounts_in_message":[],"_revision_number":82},{"id":"3aea640318e9fd9e999b375f505face5b56edda2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-20 18:53:04.000000000","message":"Patch Set 82:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/76133d31b0734450b6e3558f161f4cc0\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/af06963e56a748b2960be6c6aa4514bc : SUCCESS in 1h 00m 23s","accounts_in_message":[],"_revision_number":82},{"id":"a53c7500f0363202f82320eb0b11426657f18900","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-05-20 19:15:17.000000000","message":"Patch Set 82:\n\n(1 comment)","accounts_in_message":[],"_revision_number":82},{"id":"2b417396cf03532b24370b59b93107dadf3a1396","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-20 20:23:46.000000000","message":"Patch Set 82: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/896cff024f6240d2a5cf0bf48200c56a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d31f68ffd454360be28d4dc8af1ffd5 : SUCCESS in 9m 01s\n- grenade https://zuul.opendev.org/t/openstack/build/182057315c5a4b31b8a1b8c77e7c1489 : SUCCESS in 1h 02m 21s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/82dc239ed16e401590d8558159a5f7e6 : SUCCESS in 1h 04m 39s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d31de55168794b9dadcb9dd245638b6a : SUCCESS in 56m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0ae889adef3d4ec1bc25a2b1c05e4836 : SUCCESS in 1h 00m 18s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/17e9802360c04965a62c29848b05e92a : SUCCESS in 19m 43s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/df09466fd7a447b393fede1fce390463 : SUCCESS in 18m 56s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/79a068d1b81941a0b69ed6052c1aeefe : SUCCESS in 23m 11s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e90f7635b57d449490b6ddbc202b4842 : SUCCESS in 5m 36s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/cb4591a4812d4eae9f939004dac407c5 : SUCCESS in 21m 42s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/cb1dcb06d80f47a8a643cfd0059fe5bd : SUCCESS in 16m 10s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/b71b0ab47869471393e48c17a21fd1fb : SUCCESS in 21m 28s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/6def4e7bc52f46c58c400f89463c3be8 : SUCCESS in 29m 32s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/d9cfe91e271c4cc690d9683e3cab23b3 : SUCCESS in 13m 01s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/b42ca71300c340618e6baf632fbfc130 : SUCCESS in 12m 42s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/6487f0807b984ccca696e1fb4dbe9164 : SUCCESS in 14m 25s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/cf3de4b29489446a853bdb7e629a5b40 : SUCCESS in 6m 32s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f1f107bcc3054f708922aa398bfbae24 : SUCCESS in 13m 52s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/bbd686a2715d41ffac38ecef38586da8 : SUCCESS in 8m 03s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/57dc310e4a834b27b22d336b4a1432bf : SUCCESS in 59m 22s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/046f00ad936746cea5d0d488e66ede30 : SUCCESS in 34m 33s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a8568c2982e44a1a9dd65f77e78d4bc8 : SUCCESS in 35m 08s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6627521242f043b6a606b00b3fe4479a : SUCCESS in 18m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4221b60474d1475e9bc758b52c129c09 : SUCCESS in 4m 12s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d613e9e32e11493aa435e91fb2a79f45 : SUCCESS in 18m 58s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f26ff9794c8f4b70bc8c4e1de95c05aa : SUCCESS in 46m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9ad66cd06da54c2e954b8a7d0972a287 : SUCCESS in 59m 00s","accounts_in_message":[],"_revision_number":82},{"id":"e5b649b56e238c8ab3ca8ca3d817ca0ee6547a15","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-20 21:21:27.000000000","message":"Patch Set 82:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d4f8040545a245dda50fef718067df76\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/d8ccd542db0d477d91ba0f2272e19a19 : SUCCESS in 42m 41s","accounts_in_message":[],"_revision_number":82},{"id":"26a998e42386f036599aefc97d5c0451bb689f6d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2025-07-14 18:58:34.000000000","message":"Uploaded patch set 83.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":83},{"id":"4d5509d7a64fdae3faf9cc6ac953dad62a4d696b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-14 19:52:37.000000000","message":"Patch Set 83:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4ab4b65dd4a9498fa4fad1a731370de5\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/7f5616b19dc149338cb5ffd1e3ffc256 : SUCCESS in 50m 48s","accounts_in_message":[],"_revision_number":83},{"id":"a65e58fcd8c3aa8953cd7702b8a05e41b1f37364","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-14 20:12:06.000000000","message":"Patch Set 83: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9fda39c27a2f4789b66278ee281eb1f3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/320a033d03704644b30dea89b69154a0 : SUCCESS in 9m 42s\n- grenade https://zuul.opendev.org/t/openstack/build/9c42a3d962b942b2a88d5aef78144b61 : SUCCESS in 1h 00m 15s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d528fce5260c45828a3a46c9307580fd : SUCCESS in 58m 31s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/dde04ba3513f4b3d9a8cc45a901b9157 : SUCCESS in 44m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ba928a6380db4d55a118ccf93e06610c : SUCCESS in 58m 08s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5653bfd22279453d9b71130d9e9d182f : SUCCESS in 20m 29s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7d7500b690a84307aa8ca0e9356d5cc3 : SUCCESS in 22m 42s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/334110bae24a4abb826a724359626827 : SUCCESS in 23m 17s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/779b0e53f04f42d382a6b240f2ffcaf8 : SUCCESS in 4m 51s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/4559cc1da08243d4bd47d01cd3caba62 : SUCCESS in 23m 20s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/96acf8288a5a4e31a97c9dc8946d4997 : SUCCESS in 19m 57s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/ba2f39fde00d47e0882f4050416ef437 : SUCCESS in 24m 14s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/beb18245961941d09fd436d960884d52 : SUCCESS in 30m 14s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/a790c2e017ef4ea9bbb6588179a95d67 : SUCCESS in 13m 53s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/21e087b5f5f046e6821f9ea3b55a4f0f : SUCCESS in 14m 43s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/ddac8a02101f42ad9876e2d0cc63e1b1 : SUCCESS in 16m 21s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/ad5b6a87dd334a8aa3c4ad77fb25acef : SUCCESS in 7m 32s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/4103d670a6f54697a0c745d9bbc95255 : SUCCESS in 15m 56s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/2af1fb0cabff46fba6449ea746c92601 : SUCCESS in 7m 18s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/9f44a82576df4f5eb7b0cae5e3f5b54c : SUCCESS in 58m 12s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/11247fba669b48588e138e40c44d1f42 : SUCCESS in 58m 14s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b5e41bc568254f5f9baef09bae51897c : SUCCESS in 41m 17s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1a5d458bc4d1494db3758e38e989ee32 : SUCCESS in 20m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3c0c6a55bd2a42b6a91cdbe43f0ebd5b : SUCCESS in 5m 29s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b719b69c60ba444db4264ed2bd0b8145 : SUCCESS in 16m 49s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ef7663f653e241b9a77e8eced20e43bd : SUCCESS in 51m 39s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e92b6fe0891c4539b8bc53ca31927351 : SUCCESS in 58m 27s","accounts_in_message":[],"_revision_number":83},{"id":"e2c28f7df43bd7d3710c9b4c3c11796a5322d502","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2025-08-20 05:28:06.000000000","message":"Uploaded patch set 84: Patch Set 83 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":84},{"id":"785efd8d329468c0bb606d65f0f9ada2784e6c3a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-20 06:34:48.000000000","message":"Patch Set 84: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f0b484fd95ae4f08a70737920a67f716\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2cae7e6cf00642d596e0044c6c705d01 : SUCCESS in 12m 26s\n- grenade https://zuul.opendev.org/t/openstack/build/e5047c473678476895d74e07dd0867ff : SUCCESS in 58m 47s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/bd6e0c8a04474414961bf01448463903 : SUCCESS in 58m 48s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/5a72f07d11b24f759ded9284ec956bf8 : SUCCESS in 51m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c164633ca31b42beb86af693235559bf : SUCCESS in 58m 27s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3a7a89f824ec4960a75478039dff5dbc : SUCCESS in 10m 28s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/934a0fccee0e421e93fa19800cd994f1 : SUCCESS in 21m 43s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1df08cc667e14f52abf24933f8f1b1a2 : SUCCESS in 20m 52s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/12ea99dfcad64d7ea1e41e933ccb7bda : SUCCESS in 6m 20s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/c3f40b9bf9d94f5b8e62ea3c69aa14e0 : SUCCESS in 11m 32s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/168b3141af4444d1ae1815fb28253346 : SUCCESS in 8m 31s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/44eb0b9b1b97448598f54f39a8446714 : SUCCESS in 28m 47s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/f77cd581b3a749eb9b725bf0bf128da0 : SUCCESS in 29m 35s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/9cd1f27aa82a4808865ce98c9b007ea8 : SUCCESS in 15m 12s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/63254fc06c0941588a6920677f01896b : SUCCESS in 9m 35s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/7ee18967b3c4493d8447bd5127f050cc : SUCCESS in 14m 53s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/fe12525ab1614905b7ac7bffe4ad41d3 : SUCCESS in 7m 21s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/0351a39d0b2d4c54a2c9dc9a5ce2d2d2 : SUCCESS in 14m 35s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/382a56e5bfd0445b91ecc7930b36daf2 : SUCCESS in 8m 21s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/8aa9d730cea44a7cb58af466c2d2533f : SUCCESS in 34m 15s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/bdb02c44a97648d5904b4dc22483ec19 : SUCCESS in 55m 00s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d1ab72ffbb2d43a2835b6c3367713506 : SUCCESS in 49m 42s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/31055f66440444989e7839c4bd2aedcf : SUCCESS in 28m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/38cba64eff2d4249873cd64101f70f81 : SUCCESS in 4m 19s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/0706aaf22ab74433815f77307473aba2 : SUCCESS in 16m 38s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e472b82f4cdc47dcba4343a1b55d08fd : SUCCESS in 30m 17s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5df540ff23a4485982718975c8ac14df : SUCCESS in 29m 09s","accounts_in_message":[],"_revision_number":84},{"id":"eacb38911f9e2543cd27d80a1b0d8bf0ce3bdd0f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-20 07:17:27.000000000","message":"Patch Set 84:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0a5697cdadd54318b8e02bf231dcaf4e\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/c53c37aad08f4d9e803247c6d3fa8571 : SUCCESS in 1h 31m 29s","accounts_in_message":[],"_revision_number":84},{"id":"88de7db84f48e4aba8e3007896d3216eb5944b56","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-09-05 19:46:54.000000000","message":"Uploaded patch set 85.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":85},{"id":"2b26c005c2e7992f02db8df6a190d8646ab5add7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-05 20:49:09.000000000","message":"Patch Set 85:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7c1c4df96ba84fd6a4c9d40f6b37999e\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/04ef606679f04af58b04df76d40e0328 : SUCCESS in 50m 45s","accounts_in_message":[],"_revision_number":85},{"id":"336c55e097859be2ab0d6511228e4af5a72700b3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-05 21:04:14.000000000","message":"Patch Set 85: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f9a9e756bbce411181f815eb451fc2a6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b844d4a75b942e6a0702dd1337a0f2c : SUCCESS in 5m 08s\n- grenade https://zuul.opendev.org/t/openstack/build/3eeec270ed6345369d941f4280387b8f : SUCCESS in 1h 02m 23s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b054f37cc3fc45a8a60d98fe4254b2da : SUCCESS in 34m 07s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e0ee2cf17ac147a795dbc9838e7d5eff : SUCCESS in 50m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ad1657819867436eaff409515c118730 : SUCCESS in 1h 02m 01s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b131d4772a9b4759915d87704aa3cbf9 : SUCCESS in 18m 28s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/84c2758939a849729706f6f6b5014a61 : SUCCESS in 21m 07s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/dc1c82c18a9c4babbd215d3fc2838a50 : SUCCESS in 19m 55s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/4d1ef8e6f5ef4ecfad4e4bfa5e9d7afb : SUCCESS in 6m 48s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/167a2a2a599845949e80c0125735dde2 : SUCCESS in 25m 09s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/5d85fbfe7f0b4a99942f153872fe2857 : SUCCESS in 21m 20s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/6befa04b607d46209bef977fb13935dd : SUCCESS in 30m 56s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/3e4b1f9ab80b4643b8537893dc445ea6 : SUCCESS in 29m 00s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/6eb77057cd1f46e681355f2ab96fee5f : SUCCESS in 14m 01s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/7cf283812d8e480b8c4e9670f52482f9 : SUCCESS in 15m 06s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/ed60ebe922b14a87854040d109d6b774 : SUCCESS in 18m 39s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e17ab44d69734860bd223d75f9500ee1 : SUCCESS in 6m 14s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/74428d4e2d984b92a0b24d722563d20e : SUCCESS in 16m 39s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/57099c25ffd144c2a00b3e8571b19eed : SUCCESS in 8m 28s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/c45337ae17e4429093f9738b590712f1 : SUCCESS in 55m 14s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/107225db413d42d4b621cf9b4eef2da7 : SUCCESS in 1h 03m 26s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/580cb92d522640e7b828c06044b99fda : SUCCESS in 58m 47s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/22ba2411f7774604ab2829a560441f13 : SUCCESS in 26m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d0b5961206de44ec98f20bf6b2285f3c : SUCCESS in 5m 34s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a868f844f4f04a62b3bdadf597a94240 : SUCCESS in 25m 20s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f5d6d3bf563d497ca4bd21a2e6323289 : SUCCESS in 53m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0aca0aee8a6b48b9b1a5ab7b2dae57d4 : SUCCESS in 30m 05s","accounts_in_message":[],"_revision_number":85},{"id":"6fd66ac279dd4700e65b2c510b17491cc03d0ed3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2025-11-10 05:18:45.000000000","message":"Uploaded patch set 86: Patch Set 85 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":86},{"id":"2d153613fb3199fbf88d024bf5f06c224f35d0c5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-10 06:18:42.000000000","message":"Patch Set 86:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d46afe3b89d5483881f963f1d0640c50\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/69e7a5a27e5548b393b286f579a3fb7b : SUCCESS in 57m 04s","accounts_in_message":[],"_revision_number":86},{"id":"aa1b54952566b727fd1fa5225eb7a8cc65de93c9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-10 06:30:45.000000000","message":"Patch Set 86: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0e8802187e174be1aa3438488032dced\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5cb99b7676a045d89aa3a8c66f9d3364 : SUCCESS in 9m 16s\n- grenade https://zuul.opendev.org/t/openstack/build/b83d99d900f14c6c8524668ab1d8f305 : SUCCESS in 59m 26s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/dc83a75e526d437b860da3128b1b1347 : SUCCESS in 54m 02s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/ab2c989c4c8b4ce39d0087fe3d0f9374 : SUCCESS in 39m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6c1f37393eb942d4b52e140343a8fa31 : SUCCESS in 45m 16s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ac47e66770e24868adbe0c740015f19c : SUCCESS in 19m 49s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2a22830eb3ff420dab7261b1d1eced67 : SUCCESS in 20m 21s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/377433bda4914719b1fec9fa1d32d6c2 : SUCCESS in 20m 38s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d6dee09f1d3541b7bab6fd9c266eb2c9 : SUCCESS in 8m 51s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/319142f6723d4db39c137db3d9b9ffce : SUCCESS in 28m 24s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/16e8e3f12e604439868d23ac6d8ca886 : SUCCESS in 21m 02s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/e3700bc227d44c78a5d6c98454771fb6 : SUCCESS in 38m 04s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/9929b66bf8724a49a13e886faaf8b087 : SUCCESS in 18m 28s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/f81ff08a7e4246b49d1ab39705b7c805 : SUCCESS in 9m 05s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/c74c16c35b474d8ea070302df37f5415 : SUCCESS in 15m 22s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/34bc5c6ca22241abb7694d656b143521 : SUCCESS in 14m 37s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/0a5674847c5d4c94b4f3ede6377950aa : SUCCESS in 6m 26s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b774cd8e0c7a4d4898493305fa5b7b62 : SUCCESS in 9m 10s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/9a3633cc6ac2469cbe521e0d30b71442 : SUCCESS in 8m 12s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/c94ec4679a7042ccbc00f273ad6edaba : SUCCESS in 53m 11s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/bed34d8b2d2d4e19abcd8d5128e3f329 : SUCCESS in 48m 23s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d841eda73e1441a0919325b606ba11be : SUCCESS in 46m 19s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e690b71ec0dd47c98e31a63ad64abf52 : SUCCESS in 25m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/71544038454e477094de8e5690cb42bd : SUCCESS in 4m 36s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b42fc7d0173f44cda747d23f36d7342b : SUCCESS in 24m 10s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/199d803476064e4f9a7769dd4e54b476 : SUCCESS in 50m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4fd1886181134b458762662cd8b3394c : SUCCESS in 58m 01s","accounts_in_message":[],"_revision_number":86},{"id":"8cf2ef4c332c50a4f2a9d7489bab29b58a7eb624","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2026-02-10 04:43:10.000000000","message":"Uploaded patch set 87.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":87},{"id":"b63f6856909d01728baf53fa4d4fb8305cf0303d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-10 05:37:14.000000000","message":"Patch Set 87:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/44ff8c1fc0c44ac597eacf3684c18d7a\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/81646c996dd94beb9b7c3ec5ae98c098 : SUCCESS in 51m 10s","accounts_in_message":[],"_revision_number":87},{"id":"d86c1726989b989f07f0b288629ed2b2b7c51239","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-10 05:54:54.000000000","message":"Patch Set 87: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/17b1af7ef2e54b2bb80e7beded71bd89\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/35509baa8ee04bf7af19a2d278793acf : SUCCESS in 9m 50s\n- grenade https://zuul.opendev.org/t/openstack/build/2eef7fe2f5a548ccb74f641854c4b61c : FAILURE in 14m 09s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e09f71ee09d64b959a69ef88ace65b63 : FAILURE in 12m 13s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/1fd7becfd1e0472b816d91a5ba981148 : SUCCESS in 47m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e50ebe3262fd4c4491f1d8926ad46833 : FAILURE in 34m 03s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a9d7a762ffe3408e837483ea8b773e14 : SUCCESS in 11m 44s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7c84768480634e48a2aac0553daa3a7a : SUCCESS in 16m 56s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/96a26b6cba3f487abfdfcfd340c3de38 : SUCCESS in 20m 31s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/4f5701b28f87484680f0be0d143cbcd0 : SUCCESS in 6m 42s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/9b8e34dbd28a4bb3a63c3eb9f7e37e1d : SUCCESS in 24m 42s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/6fc8f37fb0434da3ad37361598ae3f6d : SUCCESS in 15m 02s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/87a6560fd46e4e27b58f284c6f88e737 : SUCCESS in 19m 02s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/30e1efa8e59d488e9c3af4a4156276ad : SUCCESS in 30m 14s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/6a6ff20d7c7d44d786aef4a7f90e7c48 : SUCCESS in 15m 31s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/64b410d2791a44e1ab190ea006269f4f : SUCCESS in 15m 22s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/047094c9786443ac9f4136767366a73f : SUCCESS in 9m 36s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d7fe5f7ff68f4ab7877fa21ac2b20f00 : SUCCESS in 6m 10s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/68f1fdf76a474998b77c1a7ed09fef1d : SUCCESS in 15m 12s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/2a220f2281744c03bcf96cea4b272496 : SUCCESS in 8m 23s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/4d8bb3a46f4644bbb5665d7bc4c2f205 : SUCCESS in 34m 40s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/cd655dfa37254858a1a8f91dc1057230 : SUCCESS in 1h 03m 00s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/63b4e32f91704b16af65e7cd0b152262 : SUCCESS in 49m 58s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d1e644c0eaa3421c89316e43e530177e : SUCCESS in 29m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2bac83666d344db0973badaf7b34c842 : FAILURE in 3m 05s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/8a5328cc989346de8fc5f0fee8756bcf : SUCCESS in 18m 00s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3d6cd56ec8ac45ec9d07a875fa8a9aa6 : FAILURE in 42m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0837cbd6f7f344e287651e95b03ece2e : SUCCESS in 34m 13s","accounts_in_message":[],"_revision_number":87},{"id":"e989df9a58e05521504c9412f154f0140107658d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2026-03-03 05:39:12.000000000","message":"Uploaded patch set 88: Patch Set 87 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":88},{"id":"d0ee26b6512a4c5eb1a585e0fb2e7b5284cab42e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-03 06:11:25.000000000","message":"Patch Set 88:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ec3dcdeb3a5b469c9a8a141eb2b29edd\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/bee1a1ce09524ad1a8fd3f7e9a749835 : FAILURE in 29m 56s","accounts_in_message":[],"_revision_number":88},{"id":"2f52599b9b9482e9bfa6f063eb94a2b1dfaa63c4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-03 06:48:14.000000000","message":"Patch Set 88: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e5be0b84cc444099988a227e189d4030\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ecd12b8ab0844cefb1c9453025f3eb9c : SUCCESS in 13m 53s\n- grenade https://zuul.opendev.org/t/openstack/build/fce4ecc1e8b04ca195de6b2672a7c28e : SUCCESS in 1h 01m 12s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/24023b845cf54fb996f0287ccf2c2e27 : SUCCESS in 31m 03s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/89a10f7d567645a78edee1d023829713 : SUCCESS in 56m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f364cf57a2f54a3b99c8ca7f8ea836de : SUCCESS in 47m 53s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7391d9c0d2514bbf83c167bc8df8df4b : SUCCESS in 21m 50s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/76ad059a83c74c98b2b1fd70dfdd4788 : SUCCESS in 17m 16s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/51f68e39427640f988c16ccca0d5ceca : SUCCESS in 11m 36s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ce82471de6994163be874f44eb853402 : SUCCESS in 5m 46s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/d311f69892d5453a96371a578c7edc4c : FAILURE in 26m 16s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/e9b78f4e400746eda7a98491323f84df : FAILURE in 22m 21s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/a6bb250e2ace4606bd003fa4ab21f81b : FAILURE in 28m 49s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/0e41c7ed7b734c42b5e42812b4528ee8 : FAILURE in 31m 02s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/cf0842bf83824fd8a46edb43682c295f : SUCCESS in 14m 14s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/22421194415d49638b2ccfb33ede7471 : SUCCESS in 14m 16s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/86e9c7edd4ef465891d418e465966d10 : SUCCESS in 17m 04s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/418053c816df4cbb9d289fda62750e5e : SUCCESS in 3m 46s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/aa66dc7e0f474ecbbfeab0dda5bb480a : SUCCESS in 14m 49s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/7d58c95d17a4444c8331dbca158b5e7e : SUCCESS in 8m 21s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/c3627e3515494e9a99f0935ccf50dd86 : SUCCESS in 52m 57s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d0ab7b7bfd2f4dfdaf9053b2109b1abc : SUCCESS in 22m 32s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/1eeab37fabe9426883be65f0788bc272 : SUCCESS in 48m 02s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b6bf751b9cf149c5a8531cfff23be3ef : FAILURE in 27m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/87556274ec294f52b4df7bcd200aff85 : FAILURE in 4m 23s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/59445d15f33a4613845e2aa17974e788 : SUCCESS in 18m 46s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ace900dad21343fe9699d9d8e98e6d1a : SUCCESS in 56m 35s","accounts_in_message":[],"_revision_number":88},{"id":"ad33bfd159f355c7ceb1f45261fb5edd2a1b7b1d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-03-03 16:02:27.000000000","message":"Uploaded patch set 89.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":89},{"id":"e0d5a3b61e8842afc53b4894312cbf6ba60291a7","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2026-03-03 16:21:03.000000000","message":"Patch Set 89:\n\n(1 comment)","accounts_in_message":[],"_revision_number":89},{"id":"fc53ea70fcccff0aadf2a6b34bb59452c73b44aa","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-03 16:37:23.000000000","message":"Patch Set 89:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d8ed9d14ca104ed8bb017e207c3fbb21\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/053812c1ba7e4735b58f5f34ac5f84d8 : SUCCESS in 32m 54s","accounts_in_message":[],"_revision_number":89},{"id":"03db9ae5cd1828f9b95a8377bbf8680e78a48c3e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-03 17:34:53.000000000","message":"Patch Set 89: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ad9f230824734f3daf3b872edd50480f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1226bbf7bcd547b39cf244164947cb3b : SUCCESS in 10m 42s\n- grenade https://zuul.opendev.org/t/openstack/build/70e02d49a3de49dea448361179fa6e04 : SUCCESS in 34m 31s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/7166b9cc1c5a49c4b0d060570b3034e1 : SUCCESS in 1h 21m 47s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/2c3f2803694644af80d17b84dbe22c50 : SUCCESS in 46m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6af5a6b84ebb4f42b2ab779ab6f6e554 : SUCCESS in 1h 02m 07s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ed83bd48b3a848578318cc868c8fd7af : SUCCESS in 18m 05s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/41b51864079e4057bd1bdc1dfebdb310 : SUCCESS in 17m 48s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fbcc5f09d4674763ab6ebf598346ac20 : SUCCESS in 12m 18s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/07657b3afb5e499d9d0b8e92f754281e : SUCCESS in 8m 17s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/e6cdbf9825b44f9cb98633ae22594ac5 : SUCCESS in 13m 00s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/73668b799eb24f11a906aa3975802561 : SUCCESS in 19m 52s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/c672916bce6345d5adb0589c7ffd2bb6 : SUCCESS in 32m 49s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/69592963b7bc45428ed248216e8c12a9 : SUCCESS in 28m 52s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/7ab0390174db4dc196d9dc23a73310d3 : SUCCESS in 15m 07s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/83f27a64f4b54f8f8b3eb91ccdbbd419 : SUCCESS in 15m 15s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/6c964a78b9624032beffbb336a908f6d : SUCCESS in 8m 03s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/9da32439260345549bcb7b31fc89fd22 : SUCCESS in 8m 35s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/12f76ef4a68449d2a63b2c41089870a9 : SUCCESS in 9m 30s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/efe404992ea24fcc9c42cc7ac08159d6 : SUCCESS in 7m 13s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/3848af00556e46fa972794dbf41c305e : SUCCESS in 54m 42s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b9710929a1ba4805a71351ce96222b91 : SUCCESS in 56m 36s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e29f1118441948b6ab8292dcb50c6769 : SUCCESS in 41m 33s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/eaa660688cb044b9be23056f37eaa66d : SUCCESS in 28m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d9bcd517ac8946aabf9f921747117f78 : SUCCESS in 3m 36s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/cec1d430880a4b468a36b1827eea0a8e : SUCCESS in 18m 37s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/00ba6c15198b458f921ec886769a0f29 : SUCCESS in 59m 16s","accounts_in_message":[],"_revision_number":89},{"id":"01147a28ab3afb0d603df243fce2ead993d19fc0","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2026-03-03 17:42:11.000000000","message":"Patch Set 89: Code-Review+1","accounts_in_message":[],"_revision_number":89},{"id":"968820505959c384eb641e45fe0ba756688380e6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-03-18 13:35:59.000000000","message":"Uploaded patch set 90: Patch Set 89 was rebased.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":90},{"id":"4002ae3dc538f0434ba65bb3cabefc50d6552269","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-18 15:12:28.000000000","message":"Patch Set 90: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/84d8ff5ed6c649728399a9ef722e387f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3089a01be87047a69564d746ee516dca : SUCCESS in 11m 24s\n- grenade https://zuul.opendev.org/t/openstack/build/7ad0ceafb8a3484da01ecddb933ce3f5 : SUCCESS in 1h 00m 12s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/cb80685f57cd475689fc130a3ca5ac3f : SUCCESS in 34m 53s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/8bc4b6692d1e4ba7a0e98ebfb5f949cb : SUCCESS in 36m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8218011c3f2448178112b9209b92aab0 : SUCCESS in 54m 55s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/67efbc164664451a938d3bb4d631f886 : SUCCESS in 19m 58s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2dc8683c5f724600b1ea44bc901da215 : SUCCESS in 18m 43s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/33daf52e30564dab9c6f282afc57ccb3 : SUCCESS in 22m 05s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/337c1e1fa3104d8cba768e543ecae2fa : SUCCESS in 5m 38s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/f52bc85eebcf4c2a890339cfdc6c463d : SUCCESS in 25m 47s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/84eef897c9cb4ea7bab9c2132157300c : SUCCESS in 9m 43s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/9a78083e1004459c86cd99203e68144a : SUCCESS in 19m 30s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/56ff0e9e1aa44890b86a8dd6c7a7d28b : SUCCESS in 21m 13s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/d0381a4b70b54a60bff89b016a414120 : SUCCESS in 15m 10s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/49a2d0b110fa42afbedff2e51f9e95fd : SUCCESS in 15m 27s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/62245b2588eb4a6f85dd9c74964f9cb4 : SUCCESS in 16m 59s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/0f896b95fe0f43f9a099d52df5ea741f : SUCCESS in 6m 39s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/d298bbad0ad740fc9e99c66be4b51de2 : SUCCESS in 8m 42s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/7bf502f1e8fb4e4b8fdc74951b4e4314 : SUCCESS in 8m 21s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/988cdde962bc4a8096bedcf90c7ad047 : SUCCESS in 1h 01m 00s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/94917e13f152458caa379dbd95c6e999 : SUCCESS in 54m 45s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f89f62d76209468f86729b9fc34371b2 : SUCCESS in 50m 10s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dcb7d2a4aaa9478a979a2b479f17503c : SUCCESS in 28m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/319ae23f2e884afa9392aef0a176411f : SUCCESS in 3m 48s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/977d6b725d0d475a8bc70233d247c30a : SUCCESS in 19m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5d69789a8cea4aac9e4f7bbb6af75759 : SUCCESS in 50m 20s","accounts_in_message":[],"_revision_number":90},{"id":"67757868f435183422898ae046a8d7e416262727","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-03-18 15:18:47.000000000","message":"Uploaded patch set 91: Patch Set 90 was rebased.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":91},{"id":"38fd42909b921626197465a6bd219a55d2c823ba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-18 16:38:42.000000000","message":"Patch Set 91: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/122b282779b2455680bb262e7d008f6f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d6858202d6384b85872d9d395f1ebf4a : SUCCESS in 10m 53s\n- grenade https://zuul.opendev.org/t/openstack/build/f30b8348c0b143968090e4bd7a3e61e1 : SUCCESS in 40m 43s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/364e4b055783446ea63af5e079e1737e : SUCCESS in 1h 05m 32s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/9cfa3aa6741140c5a87524b4606abe02 : SUCCESS in 46m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ce349fb329154d9895285d660eff5a4a : SUCCESS in 1h 02m 54s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/64b206c3ca1c4a9d89f3148b3a3b4dfd : SUCCESS in 19m 49s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0e656e0690ac4fe891e660167c51ee5d : SUCCESS in 13m 25s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/852f0f8e91254214af538fde603f3ec7 : SUCCESS in 23m 48s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/062e27f80e6143a6b4d33bf5e076ec79 : SUCCESS in 7m 28s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/a25cba53b25b4ff8b5015cbbbf974d31 : SUCCESS in 30m 17s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/42a7953f81a44222ba21ca0f005a7bf3 : SUCCESS in 23m 35s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/87f3c231dffb45a497927787a94b5740 : SUCCESS in 13m 02s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/9663071937304beebcda2d0ce7769412 : SUCCESS in 23m 22s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/638d8692350547919752825d382691d8 : SUCCESS in 12m 42s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/36449e16f5f9459592bccccf9cc314ef : SUCCESS in 11m 43s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/1f971b29289c443e9d5ea0ab14ee6e58 : SUCCESS in 16m 11s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/7de0683247b140ddb9f63ab2e0c496b5 : SUCCESS in 6m 57s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/53ed79f0a61f4362b803811de31c098a : SUCCESS in 21m 03s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/60fe2edf57e14e92bc9891dcbf1b2afb : SUCCESS in 10m 28s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/52e94a0976d946f8a31a917af12000e1 : SUCCESS in 42m 05s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/07ba5de9f7344f1c8e3dfd2f0b532f9a : SUCCESS in 55m 33s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/3574e88261384828bb0b086243fd399a : SUCCESS in 59m 57s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e08adbc9825e42c2b2cf9dd4934eb2fb : SUCCESS in 33m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c4f50cb68f7e445ea009909788c6190f : SUCCESS in 5m 51s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/cfd08c3e57274ab88bd8beb563a341d7 : SUCCESS in 18m 48s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a4ff9399a9cc452f89a30eea4d81c01c : SUCCESS in 40m 29s","accounts_in_message":[],"_revision_number":91},{"id":"f10a259315828436b5f4318d51af5760e6a92535","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-18 18:58:33.000000000","message":"Patch Set 91:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/583428e9dec0438a8f6597ddea7fa32e\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/033e90f61f294ef690653ff06230cb5e : SUCCESS in 1h 15m 49s","accounts_in_message":[],"_revision_number":91},{"id":"02f96b696a11b0498260f69bca6ec392d306592f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-03-24 17:58:55.000000000","message":"Uploaded patch set 92: Patch Set 91 was rebased.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":92},{"id":"f2f2e4695ffe2a6c422f3decb33b1f9f964b22d3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-24 19:36:50.000000000","message":"Patch Set 92: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/625e0d5e78524e1d89501acb75e888a7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b0a100f6abc34a6abf5294615dfb1c72 : SUCCESS in 10m 29s\n- grenade https://zuul.opendev.org/t/openstack/build/5aa89728000940bab276e6d7d79590b7 : SUCCESS in 34m 43s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/cc60988b00044f958606afe8d32d3542 : SUCCESS in 53m 09s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/fc4a0b093c504feabf483f52f0fa157c : SUCCESS in 43m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a9c63f712c9242a4aaf1fc9bb0629122 : FAILURE in 1h 27m 35s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ae5a27cd0e8f4afdb367420f8400b1b8 : SUCCESS in 11m 52s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6e507f7b4d2745eba584f8bec1cf9069 : SUCCESS in 23m 48s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6ed40276dc974c2cab21401bdb650602 : SUCCESS in 11m 09s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/1836e78e16df4b4894e3e033e4322c4c : SUCCESS in 6m 26s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/767dcfcd54c74355be9c12ea344c63ea : SUCCESS in 26m 07s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/1032f5ddb91b44d398e0e17dd7ddd258 : SUCCESS in 10m 38s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/c450d012e3c1414db5c12e2c6000f35a : SUCCESS in 34m 45s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/b16e3d6165ba4013bb2dd8bf4ead761d : SUCCESS in 16m 55s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/6f6833b7d5bc40bf8ccca75bc627bc60 : SUCCESS in 9m 15s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/cc9c6ff9a0cf4d7484851335e374c619 : SUCCESS in 15m 55s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/8975d678a82f45f5a57967838d0585c1 : SUCCESS in 16m 20s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/0357f02a75b84c2b84ff94d2c766f8ce : SUCCESS in 6m 11s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/efd7c470ecbb43fea55c24eddd0f42a9 : SUCCESS in 9m 52s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/f3ff69d0477349ffbaf90c8934f58b6c : SUCCESS in 9m 08s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/50efafff4c62456da129258bde2e2ca5 : SUCCESS in 58m 44s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/3a6fb965bc4a49cf98dbb6b0ce10669d : SUCCESS in 23m 11s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/06ee3f690c0f4ed093a4e3f8bc60d147 : SUCCESS in 53m 10s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7ba652fac50e48118309671783f399a7 : SUCCESS in 13m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ce94facfff764cb582f829e39d98c2eb : SUCCESS in 5m 17s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/4f797eec0cf84de4bb71bbb835c38bcb : SUCCESS in 19m 08s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1aee78df03974ff6a8f60f5102a18e5e : SUCCESS in 1h 03m 00s","accounts_in_message":[],"_revision_number":92},{"id":"b0ec0136901197acbab9320e0dadd4495798a281","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-24 20:27:42.000000000","message":"Patch Set 92:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eb6888ea711d4c50b9587ea812008a52\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/8a5caab68f0d401dacd5237e4500b383 : SUCCESS in 34m 30s","accounts_in_message":[],"_revision_number":92},{"id":"fcfb06ba2e3f773598058f769de08abc5b9253c8","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-03-25 10:00:41.000000000","message":"Patch Set 92:\n\n(1 comment)","accounts_in_message":[],"_revision_number":92},{"id":"d810726c9d90af9dcc0a1fe88a95b90362b2a88b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-25 11:20:35.000000000","message":"Patch Set 92: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b43697535565444f90eab4d3f09e38fe\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a711df8d87f54dc0a1895e6c2940c334 : SUCCESS in 10m 45s\n- grenade https://zuul.opendev.org/t/openstack/build/8bc7ca869d7d49d3b8eecce060428555 : SUCCESS in 1h 19m 00s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/0498e323618d49aabe49fee8985942b1 : SUCCESS in 1h 05m 13s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f120d4cc280244c99e9d3d9d87aa8bcd : SUCCESS in 54m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a46412c40c234ee1833b01fba6a251b8 : SUCCESS in 1h 01m 29s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c1f12ca9c8704784a9979496e1efd570 : SUCCESS in 19m 59s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3536f5535c84443c920e9b7e8f90143f : SUCCESS in 21m 27s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/08f89936d30e4c4682bd153f5efa6e63 : SUCCESS in 22m 01s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/fd13285920f244cbbebffb33624a26c1 : SUCCESS in 7m 23s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/8345e5ff91ec4bd2bb69a9c488fe9eab : SUCCESS in 30m 35s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/f05d01a3f03147059610d7a1d61c8118 : SUCCESS in 23m 28s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/3b90c73096884e0a8b8306fa897dc142 : SUCCESS in 25m 39s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/7525a6b404f74a23956bfc815c711a67 : SUCCESS in 20m 36s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/58c47056f4ad4af38b4b858ab4a0cdd1 : SUCCESS in 15m 52s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/a81e6be4efa04b9a88b34e1e7b01e274 : SUCCESS in 17m 17s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/05bc15070d9c46e2ae60a8a5aeaf7455 : SUCCESS in 16m 12s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e1dd1854a94a4f138c56963f9e44c331 : SUCCESS in 8m 51s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/8292558287c8462bb481eceb7ab046f9 : SUCCESS in 20m 20s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/653374f207854eb2afbb6d9507b6cde4 : SUCCESS in 11m 18s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/017875696e4548d0880e2995a8a817b3 : SUCCESS in 1h 03m 52s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/adbf5f6094d0410db905f92195d054bf : SUCCESS in 44m 16s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b162e6fb817846fab85d2a30f1e26201 : SUCCESS in 51m 16s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a5c11c5db0a7476abd51307edb340b4d : SUCCESS in 29m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c8440d4d9a74d8da906bf37984cd7b8 : SUCCESS in 5m 45s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6374d414299e4cbfa65a3711adc4ae41 : SUCCESS in 19m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/85d8a77b587347b4a91cfee8e696ffaa : SUCCESS in 55m 23s","accounts_in_message":[],"_revision_number":92},{"id":"c9da15bf9ec8c2cca394dc908fdeb2abfe3406cc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-25 11:51:57.000000000","message":"Patch Set 92:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/88c5b65ad8b54e0aa49725181f589959\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b6df059fe9ce43829ee9ef2b9c264795 : SUCCESS in 59m 53s","accounts_in_message":[],"_revision_number":92},{"id":"41a04b595fa15014dcf7c87c4bc105875227eae3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-03-25 16:30:13.000000000","message":"Uploaded patch set 93: Patch Set 92 was rebased.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":93},{"id":"360e72cf24b384bb4b8337b61da919b2f4996446","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-25 17:44:50.000000000","message":"Patch Set 93: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a8549b0d643f4f91af56a1323d864399\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d952d2ad5974dea9edb200265a9ae58 : SUCCESS in 10m 37s\n- grenade https://zuul.opendev.org/t/openstack/build/23cf2e5feff1494fa6c1748834503b56 : SUCCESS in 33m 35s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/a5c1c11890ab4660b5c6ba8edbb0c9e6 : SUCCESS in 1h 01m 07s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/04600382428445eb9c6b62b98e9e8a51 : SUCCESS in 50m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/15715534cbe14c6b993f7fedfc521e74 : SUCCESS in 1h 05m 50s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0947b5a3327647629660b1cf5e0484c7 : SUCCESS in 11m 43s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c8c9083d286440d2ac9355e5ba7391d2 : SUCCESS in 20m 57s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7aa6587cc8a849e998a9edcddfd5d309 : SUCCESS in 11m 45s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/0d2971f5de424eda95bbf236230c4643 : SUCCESS in 6m 13s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/55b95d04ebd24bfcb2cd46801b181e45 : SUCCESS in 33m 13s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/9aebe7a6e7cb4cca8ee810d455b82470 : SUCCESS in 22m 39s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/1067a89c25b74d26869b8e50b67df425 : SUCCESS in 30m 42s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/4ed085206f594514a10fe01bb63025d1 : SUCCESS in 20m 19s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/b4df288a52ba410ba7c6217e3e1235e8 : SUCCESS in 14m 48s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/d2878b6342e64fc3970f216d23f2d142 : SUCCESS in 16m 39s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/7b4a8823861546c3a686b22bb061b09a : SUCCESS in 12m 54s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/844dddea2295429cb5fbe376de358ae7 : SUCCESS in 6m 53s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/2277a9e3e5e54740b6794c87117522fd : SUCCESS in 20m 19s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/4066e143b57d4addbfb0108477325444 : SUCCESS in 8m 35s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/d2a2d20d55934d73a33069e1dad92a5c : SUCCESS in 58m 55s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a3b779aa2b5d4d75973042ac48bd92d4 : SUCCESS in 50m 34s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a38bcaca57364853a896d6f1284273bd : SUCCESS in 34m 22s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b47c9c4edd8b44c790d613ae0412057b : SUCCESS in 12m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e35bc7cab6cf4367afd1fdb9a6c43750 : SUCCESS in 5m 27s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6e76c5824e5c4517a76c2f74697490fa : SUCCESS in 19m 15s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/625f20dcad9f4acfb6a4d2d96bbb7a71 : SUCCESS in 59m 52s","accounts_in_message":[],"_revision_number":93},{"id":"2913b610a112902bfb30cfff9fd686d88fcd27db","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-25 20:27:32.000000000","message":"Patch Set 93:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/52b0d2f7a6904dc88dd14c4235754517\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/fef00a67e766413eb1f19b0030736dc0 : SUCCESS in 41m 50s","accounts_in_message":[],"_revision_number":93},{"id":"cbdb435d1c805e215a25728f28fea0b6cfe09015","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-03-31 14:54:16.000000000","message":"Uploaded patch set 94: Patch Set 93 was rebased.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":94},{"id":"917958c230ca4f9a64f289d80c091dac5716b4f8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-31 16:12:53.000000000","message":"Patch Set 94: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ec7364021fe947ff85def97a005feada\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f6e1223fcbbd4140a13ea365cec47b2a : SUCCESS in 10m 00s\n- grenade https://zuul.opendev.org/t/openstack/build/4c3ad960c2114055886275b380c32780 : SUCCESS in 57m 41s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b0a8e2b13a064c45a1986d6e81b05634 : SUCCESS in 34m 16s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d5a5c4d75c8446c19a4444cbd2d941ba : SUCCESS in 33m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/83663577351a495b91bc51cd469ff271 : SUCCESS in 41m 00s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c2c2fb73cf554f97ba08b87de1228d33 : SUCCESS in 19m 00s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5f83a27c77f44654b7d6618e98bb7b51 : SUCCESS in 14m 44s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/04e10f28264b4520a1aafeddab3e064b : SUCCESS in 22m 31s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a76997cd10524b0f8ca20ba1a3c55228 : SUCCESS in 8m 42s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/5f37b13ac9264951b0c3e32e1c0bf1f6 : SUCCESS in 19m 46s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/9f06e9e37df2435a962ad31f77830cf3 : SUCCESS in 23m 22s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/c6cc9a77273d4b1e9274db9b6e03a793 : SUCCESS in 17m 04s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/1f196eff2c15441cb2a0014888538b7b : SUCCESS in 21m 37s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/1e3d800849fd419493162e0c1d163f77 : SUCCESS in 15m 30s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/21dd26634cc54f2385f424b51225273d : SUCCESS in 9m 22s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/79ee5caf73834988851ffd8641ac8f7d : SUCCESS in 11m 01s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/ed2340a17ac1469eb47ad87b3216436b : SUCCESS in 7m 40s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/ab6a483cd906422ab244e24075a58e2a : SUCCESS in 15m 55s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/800bbc88ae2e44b3957678f787a42ba3 : SUCCESS in 10m 37s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/8f915e439d88478c845508c63d80c59a : SUCCESS in 38m 11s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d0d5c242c5a444fc86b6f85b491d9a85 : SUCCESS in 42m 13s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ebcbfdd3aca0445e819a83f730254bf9 : SUCCESS in 43m 17s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fb99c6195c32493081bbf4cc449c5d44 : SUCCESS in 12m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/581fa704911f4df1a4b327c199ed649b : SUCCESS in 4m 31s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ee541334967348bc960cc036f01e14b2 : SUCCESS in 18m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b3f2bb0d86534164a7c3307afce60255 : SUCCESS in 1h 06m 09s","accounts_in_message":[],"_revision_number":94},{"id":"15fb2276c7c61a70a48e25aadfe6d5024596375e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-31 16:34:41.000000000","message":"Patch Set 94:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/395c318a96494a00b8812a2a4b4993d2\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/5023202164d844bfa606df60c613a6ef : SUCCESS in 34m 07s","accounts_in_message":[],"_revision_number":94},{"id":"615b955c71fa4876805c3edc1be206ca90a2291f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-04-09 09:21:56.000000000","message":"Uploaded patch set 95: Patch Set 94 was rebased.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":95},{"id":"171e938fe56f8698b6efd4e599b797e5df59b5bd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-09 10:36:15.000000000","message":"Patch Set 95: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2c1f4c4653cd467cb41481a7b52a71f0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b217553eebe0462aaa4cd9d6657f4dc0 : SUCCESS in 8m 55s\n- grenade https://zuul.opendev.org/t/openstack/build/92cd0eb54e9341c9acbdf78fca7466e9 : SUCCESS in 58m 15s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/a04ffcbd232a49b7a9a4f4825cd5bbab : SUCCESS in 1h 06m 59s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/35f6345554664fd796879f4ffb606fbb : SUCCESS in 34m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4ab0518a23d040e6a21be7fa991910d7 : SUCCESS in 1h 03m 30s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e9a8caf4c2c844d1908d8d35c2279a52 : SUCCESS in 19m 37s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/922276edf7fd4d4ca5707ac306c686f5 : SUCCESS in 19m 34s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7e600b258156453eafd2f0ad07321da0 : SUCCESS in 13m 59s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/84c440e5f9d8470bb1b5350cb6ccc59b : SUCCESS in 9m 08s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/3737b32c0cf649fe9ac954d1b29db7be : SUCCESS in 10m 27s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/4f140560bc9f45edb5ede0ec4e3d6161 : SUCCESS in 18m 01s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/4bc1abf5f9ce4284a01bda502a5d78ee : SUCCESS in 16m 30s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/6c7c05642ba647eba0054e75a4fb62ac : SUCCESS in 20m 02s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/188a6ae3a76f45d3a82391ea032d2f0b : SUCCESS in 8m 25s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/ab95cec643b24f8da01573d97bc98bf0 : SUCCESS in 10m 32s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/adff410e246242e7bbd244270dec5cae : SUCCESS in 10m 10s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/7194d24baf494101a256181687e621e4 : SUCCESS in 6m 29s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/5a25ba2602d44c64b9496d13503c1e3c : SUCCESS in 17m 12s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/625eaff62a0644d1aaa55145c05118ab : SUCCESS in 9m 13s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/9c63bf2259ae4ffeae7991cb28b8d2c7 : SUCCESS in 56m 16s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e979bf9a46ca43deb98b03816eb252a2 : SUCCESS in 39m 05s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/922e199519a846778a7224abae9e6fc8 : SUCCESS in 22m 59s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8a3ec644bdd74ce0b5941adb3bdedfd9 : SUCCESS in 25m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1aad5cb688914d4b9e2a77611588d062 : SUCCESS in 3m 03s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/5157d205ece04d0b8ef29f297432f804 : SUCCESS in 9m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c441e85bcc6949f1ad5e87cdd53bab0d : SUCCESS in 58m 52s","accounts_in_message":[],"_revision_number":95},{"id":"ac1f570be42c9c56a7e16649bce16dd7f45d581c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-09 11:49:13.000000000","message":"Patch Set 95:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/99f679b8db784444be60241142b5c97a\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/44871f21473d4ce6be1a5ffded629b89 : SUCCESS in 36m 57s","accounts_in_message":[],"_revision_number":95},{"id":"53398dbf4b2be5a7707586e263e8a309e84a1f3c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-04-13 09:47:13.000000000","message":"Uploaded patch set 96: Patch Set 95 was rebased.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":96},{"id":"6a080f59a910977e42557cd330457689efb11fca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-13 11:59:24.000000000","message":"Patch Set 96: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/129cdf41289f47c595d7d10019be7a37\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c3b4d3b853794e91a83261d0ca9c8c9d : SUCCESS in 11m 16s\n- grenade https://zuul.opendev.org/t/openstack/build/84756ae3f32546debe790d1f365e9113 : SUCCESS in 55m 25s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/9c9e67ab4114494f8f92bad5f8b5f29e : SUCCESS in 1h 01m 44s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f91012e942a54b2a84c294c20b7435e1 : SUCCESS in 39m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9650d2655e1341d4b7631cf4002959b8 : SUCCESS in 56m 43s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/763f2b760ac94d44a217a44289df4225 : SUCCESS in 20m 59s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/991b3741cd0c411d80da7a29c4ef29f0 : SUCCESS in 23m 13s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a625962b64a04ba9b3c319b3b0b75830 : SUCCESS in 22m 29s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/0b654b8ae381450c9a92b5a14b708f94 : SUCCESS in 7m 40s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/3c88bfdf43c44147b83aa335d57ffcf3 : SUCCESS in 25m 18s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/b694e3e8a2584cd0b36803269188203b : SUCCESS in 22m 50s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/451e673ed81049e7abc41fd1b43f0a77 : SUCCESS in 33m 10s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/70c2e2cc21c04771a7288f550e3d3eef : SUCCESS in 12m 22s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/62ee5df366054d9793f50018040b51e5 : SUCCESS in 14m 11s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/e700c28897aa4848ad8b86e6e057ab35 : SUCCESS in 8m 39s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/9257b388e269418a937c6ad412da5155 : SUCCESS in 16m 38s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e7d977d508b44ee18ec96d3859c7335a : SUCCESS in 5m 00s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/56c6aed68d104a0b99aa221910d8abd3 : SUCCESS in 12m 58s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/ff33d1e4565a421abfb1fbf581c75b6b : SUCCESS in 9m 28s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/31f745c037c14f7389b717d671579392 : SUCCESS in 1h 04m 52s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/1d58dd37f3934c748660dc571d06c5ca : SUCCESS in 59m 44s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f677a2510ad547e6806cf02970f7a193 : SUCCESS in 48m 38s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/311dee4561e740c39467dd6293e3e93e : SUCCESS in 11m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/723fee2036fe4a659e5b29773b7bcde5 : SUCCESS in 4m 25s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/30476cb704444a90a026b8d063163279 : SUCCESS in 18m 05s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1dad2e5224344845ab7b30eb537ecd8c : SUCCESS in 47m 20s","accounts_in_message":[],"_revision_number":96},{"id":"1fd7b4761313ad36c079c5ff3a7e7a34b2d55453","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-13 16:24:14.000000000","message":"Patch Set 96:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5a9c4c98bcad4e77910b46aae5dc13fa\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/44a00177060a49f7ab54e31799a25d2a : SUCCESS in 36m 16s","accounts_in_message":[],"_revision_number":96},{"id":"284235a5d6bfc4d7f402809562afc36089b39f60","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-04-14 15:09:20.000000000","message":"Uploaded patch set 97: Patch Set 96 was rebased.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":97},{"id":"fd66a575fb5f214917d76693a164ceb0912810b3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-14 16:35:11.000000000","message":"Patch Set 97: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/592ae64e80f14e56bc5c8cf872c75d44\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b886d83a06f54af1a094706b37d02202 : SUCCESS in 5m 31s\n- grenade https://zuul.opendev.org/t/openstack/build/eefbaf9f5e1e40c9824636be8b4c9fb5 : SUCCESS in 1h 07m 33s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/85481e92be7d43328e98a028a8b6164c : SUCCESS in 1h 06m 30s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/37a0b8ceb630465db962f1632a5a4dee : SUCCESS in 58m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/95dee273733a405687d33fbb6d118486 : SUCCESS in 1h 03m 35s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/158de25eb40840ab891c02e4f9647454 : SUCCESS in 20m 46s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e15da7f54a934ed0b0b2abf896a62157 : SUCCESS in 18m 09s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2637043be6954f6582ada68f05a70bd0 : SUCCESS in 17m 58s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c044619e5ce94404b33c8aab0a77763d : SUCCESS in 8m 29s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/b81b9ea7b5cf48edb9afd2acd36bc9a9 : SUCCESS in 14m 52s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/7b4b900008894f7cb4ffd6ac071916e7 : SUCCESS in 20m 46s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/a33f483b0b08475897b9e7d52f456555 : SUCCESS in 30m 54s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/5036b575bcdf4a2b954aca9166e13231 : SUCCESS in 22m 24s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/03e62effb94046aa9429b55af1541368 : SUCCESS in 10m 44s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/0061bb0ebba1494a8370ae45b6dea257 : SUCCESS in 16m 56s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/b011dc720c8242108ff0255bba1487e2 : SUCCESS in 9m 59s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/ab6861d207ae4419914ae765820c246e : SUCCESS in 6m 35s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f6ade5d8448545ea9d2fcdd60e5ed3bb : SUCCESS in 20m 12s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/1cd9b82f99594f85a30e186073ee30ea : SUCCESS in 8m 45s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/537ed6cdb7b144d5b976a3f5acfdde20 : SUCCESS in 52m 08s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/55466451c339499782761363906fd913 : SUCCESS in 30m 05s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/009e8820f46345b19dd148ec2725a9bf : SUCCESS in 48m 37s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3c8080a08822484abcd81e8a993123de : SUCCESS in 28m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/48f0b637b61344ac8c422e384f86fd46 : SUCCESS in 5m 51s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9362f91d3bca48638ae017e093320710 : SUCCESS in 28m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9d1f41d183bb401aa44ebe5c246f0c55 : SUCCESS in 49m 42s","accounts_in_message":[],"_revision_number":97},{"id":"fad7b4e49988fe308082eb816022ea616f6844d8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-14 21:22:03.000000000","message":"Patch Set 97:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/957ba9363ad34b3f9921bfb4071ab01b\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/4b47d2ddfb6f4d44a638f3774ccf5de7 : SUCCESS in 37m 10s","accounts_in_message":[],"_revision_number":97},{"id":"a25e7b72b5d499def471488aaf30c76cd350132c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-05-06 10:50:40.000000000","message":"Uploaded patch set 98: Patch Set 97 was rebased.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":98},{"id":"5074896fce0b9a98b4c21b4e502e79f3b89dca40","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-06 12:07:34.000000000","message":"Patch Set 98:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/52031b7ac5fd48a483b6c8f18f66a787\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/4c01f20b26fe4c0493565d6fed953222 : SUCCESS in 54m 42s","accounts_in_message":[],"_revision_number":98},{"id":"4064d4aabae5b456b21b179fed3b42c200e377de","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-06 12:14:40.000000000","message":"Patch Set 98: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/56cd2e013dfb4e179f54c4a08b0f7f50\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b417d3e16aa47e68a521e3f038835eb : SUCCESS in 5m 39s\n- grenade https://zuul.opendev.org/t/openstack/build/878034c2d04a4356a29d23ce59910d57 : SUCCESS in 1h 06m 39s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/a4f41ac6ddba4b279c23f85db708a0cd : SUCCESS in 42m 33s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/92fdeeba7b83451590f5ccd3f9830479 : SUCCESS in 56m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b29435af59db470bb1e6d6dffd685c0d : SUCCESS in 34m 13s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c00df9d6f1994cfb9af4457ba2260b1c : SUCCESS in 11m 17s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/bd1d58a065f0443891f4082c014b1d96 : SUCCESS in 20m 25s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c8f9b5d4fd2c42e9916feca903107740 : SUCCESS in 21m 44s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/9ff5765e27b04e7bb9c6c8f555309a63 : SUCCESS in 6m 54s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/2bdbe89b712f433a933d5950990cfdcb : SUCCESS in 25m 30s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/b87a82dc65264c2688b1c121297b3823 : SUCCESS in 22m 13s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/8ae68af5515c433ba5c2dbcc9f099817 : SUCCESS in 18m 09s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/f54bf2af5ee74c57bf8929f1b850dab8 : SUCCESS in 22m 47s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/e46ae8d817344145867b1381b07116f1 : SUCCESS in 15m 54s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/7d31d4b537d1451e87a1b2d04c821e6b : SUCCESS in 15m 37s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/b61d61b6de6d4779893971031fba5a23 : SUCCESS in 11m 13s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/7695bad7902641aea0fe7fc82d339ba0 : SUCCESS in 4m 16s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/9c8b1163ff2a4984bc1fff0109f3d087 : SUCCESS in 13m 57s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/1c2c70e6b3494ee0b08c081473b08d27 : SUCCESS in 9m 06s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/d7f493a7eb6143729058aa21b640618f : SUCCESS in 46m 33s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/10f9648637dc4714bed6a7cd476047a9 : SUCCESS in 57m 23s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/5bb55470e51142b7978a0918a8c4e2ca : SUCCESS in 33m 17s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7e43ed066af143ff9d5dd3018d9dec29 : SUCCESS in 29m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e54ad30f8a144ccebabeeda049bb67c9 : SUCCESS in 4m 36s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/baf615c94365487f9983f6732a2c22cf : SUCCESS in 19m 37s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/36fa92cd2afb4d389adff456502e3810 : SUCCESS in 59m 40s","accounts_in_message":[],"_revision_number":98},{"id":"e942231c4857e2f1152270ac83dadb1c546b0a48","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2026-07-22 23:58:01.000000000","message":"Uploaded patch set 99: Patch Set 98 was rebased. Commit message was updated.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":99},{"id":"b5b84585cbc39ff25271333828bb709a30a66e32","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-23 01:00:26.000000000","message":"Patch Set 99:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8ea066f111b548e1ba522ee3278df5af\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/2f6a6dd921414b33b66246ea5efd3990 : FAILURE in 57m 49s","accounts_in_message":[],"_revision_number":99},{"id":"aeae500ab990e88ff0219bb02c96791b764cc7c6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-23 01:24:11.000000000","message":"Patch Set 99: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/43459c406e584fcc9af08e8525ad551c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/47d60b990e96424f8442a5a845065c92 : SUCCESS in 13m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6420ab344b7e41a0a1bd366a553ebd77 : SUCCESS in 7m 06s\n- grenade https://zuul.opendev.org/t/openstack/build/569499f0daf64a4fbb856c1567c6fa1a : SUCCESS in 1h 03m 06s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/94aa8c98101a4288a5416302160ce906 : SUCCESS in 1h 09m 45s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/921c7a6794a74721be98a7b56ecc241c : SUCCESS in 51m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/765b60ad01174436a9949f7b255f9672 : SUCCESS in 35m 29s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2a3219e030d6461d92fe5ea30434f6eb : SUCCESS in 20m 00s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0933537bcb7048d993cc16636ff94338 : SUCCESS in 19m 48s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b99bef636ac64836a5abe02f709c36b5 : SUCCESS in 21m 15s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/8cda789ea1ee431aa5e57fa417f42e41 : SUCCESS in 7m 31s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/ead0b58a5e8f4cc7a763fd185fae76d4 : FAILURE in 15m 09s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/35a97f6a059e4cc99b65efdc4dbbec38 : FAILURE in 25m 41s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/090557073b2e4463bf65f692ae0c02b9 : FAILURE in 31m 06s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/d49b0a10f0854847b93243353737e7ef : FAILURE in 9m 57s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/744442a985e444a59822ca4f0a156569 : SUCCESS in 14m 38s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/fda98591232f43a7a1325be83b0e6920 : SUCCESS in 13m 58s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/97a425c4ab1746c5b7e2e89caf4f4886 : SUCCESS in 16m 40s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/07cf781ec00a4085af564d0aced21111 : SUCCESS in 16m 55s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/2424909dda6c4bc8a6c764c02448328c : SUCCESS in 4m 23s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/cc79d8f23cc84fd4b9a47b0df96fcaa5 : FAILURE in 10m 04s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/8e45d00dff0a4faa8af6dec68f10c081 : SUCCESS in 8m 33s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/3f2e3dfc64e74c8f9ef72beb3171ef99 : SUCCESS in 1h 01m 40s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/5bd7b14b82a542dcb32bb53cd72aceb1 : SUCCESS in 52m 01s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e0d6a0bfb4724debaaa971852924d7a7 : SUCCESS in 57m 09s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2e7a4e47416e4dcd91d0d4d69849d3f8 : FAILURE in 33m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d544ae37e8604c1189fea7d359ff67f5 : FAILURE in 4m 10s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/1022c386a0d04b55ba3b2cebfb099814 : SUCCESS in 9m 39s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2ecc3ab2e30048c98d6b7b3fd67c0a61 : SUCCESS in 53m 56s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":99},{"id":"eee483e771dc41328947489016d1e2c45577f6ff","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-07-23 10:27:09.000000000","message":"Uploaded patch set 100: Patch Set 99 was rebased.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":100},{"id":"7ec6d0f533c4d81a28b7e8fd3ce2dd3fad7e0e0f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-23 11:52:50.000000000","message":"Patch Set 100:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/17156c3f35304caab6bb1d2b6b83b750\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/69e8ec07015a48b094ddf316867317c9 : SUCCESS in 1h 06m 55s","accounts_in_message":[],"_revision_number":100},{"id":"bf0333f17326af8e52e14ad1b9bc48f51c73950f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-23 11:57:30.000000000","message":"Patch Set 100: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dacdf33ab2ac44d191d1d28a5986a1c2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ba7edda1cac482ab20f5b3b072d8bc9 : SUCCESS in 11m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c1bb548a70f84ac781f21b37ae64af49 : SUCCESS in 7m 26s\n- grenade https://zuul.opendev.org/t/openstack/build/b055a7a0bf31474f9a943be62c5a2638 : SUCCESS in 1h 18m 00s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/69456d996ffd48feb778b13afbfd7f5a : SUCCESS in 1h 02m 22s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/6dc6ae7374184aa2a6ae9524ca98dd13 : SUCCESS in 48m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4e25e7cbf98d4a6ba7127fc8a27cda9b : SUCCESS in 1h 03m 59s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/36a5c4e7b6284e4aa6dfd61cc97b8678 : SUCCESS in 19m 35s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9dbd8889c68040ddaaf09b818ce2d7b7 : SUCCESS in 18m 22s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/be146c720f6a4014bd4977536b2ff3b3 : SUCCESS in 21m 55s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/6f3f5499331b45bdb8653c9c2a54eca8 : SUCCESS in 7m 15s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/1bfaea427f324ceb93b84fb900f42cbf : SUCCESS in 27m 22s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/9f54bc785d18477091eaa9339aef7abb : SUCCESS in 23m 01s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/4e470d6877ac4aa6acdef2d8be92725b : SUCCESS in 25m 38s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/a1138ba6105a4fb3b78f6d31becd0f41 : SUCCESS in 19m 23s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/931bf171b294417399aebb0f9abf7201 : SUCCESS in 13m 38s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/7dec61ba61c14b6488b863b6210811f1 : SUCCESS in 13m 23s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/dfb01683a369446e88e92fba1a2e1784 : SUCCESS in 17m 30s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/8ce4f9e7d1aa4b1797a6b1a7036b2062 : SUCCESS in 16m 20s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b48a0e876cd04e5d82e44d94d832bcf3 : SUCCESS in 7m 14s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/83940d5314e04d9098c858b037f727fe : FAILURE in 10m 09s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/30e9ea17224243d3ad3d059e10ba8645 : SUCCESS in 9m 35s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/21e9ea3f693443aa8f0d8aa6d5706840 : SUCCESS in 51m 48s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f3db424185c94370907efe78e0f01fdd : SUCCESS in 1h 06m 17s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/12d634f70d1442a881a1cda5920eb98f : SUCCESS in 46m 11s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a14126da91b7430ba6d9032b85a76a5a : SUCCESS in 30m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/32582b9a2b314dfd8214ed325ea2bbc2 : SUCCESS in 6m 03s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/63207911ea4c427fbaeca20af672d948 : SUCCESS in 23m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/978e3db2145244c4ad2028d86e8c9732 : SUCCESS in 1h 13m 09s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":100},{"id":"c55d3f4753524de669eb17be76d4ccd2c0f80f05","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-09-11 20:37:09.000000000","message":"Uploaded patch set 101: Patch Set 100 was rebased.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":101},{"id":"597b9b5bb0d0782879f82b5917c3f8492c68719f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-11 21:19:06.000000000","message":"Patch Set 101:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0f01c13e81014dfabc9021f3bbcf6b0e\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/73d852915ab44a068ef64dced491088f : SUCCESS in 38m 16s","accounts_in_message":[],"_revision_number":101},{"id":"fa557a6226b3708c42b776b8f57fe47e440f592e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-11 21:58:04.000000000","message":"Patch Set 101: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e246f58005d046b1a51dae7f0236035a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7af298da85964d95a97397e5eb352d79 : SUCCESS in 6m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/40ad470246ee4194a6402307e4820579 : SUCCESS in 10m 54s\n- grenade https://zuul.opendev.org/t/openstack/build/79e99c49358a4b3c9ed884c8093871cf : SUCCESS in 37m 32s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/42bbc7a299bd481cadc0058c362ed679 : SUCCESS in 1h 05m 40s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/69137b32bb2c4062a8c9ad96ee87b6ec : SUCCESS in 52m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3a0ad084f4a44dd1a578e3c4a2177b2a : SUCCESS in 56m 56s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/89e35692b8b544c687fce21e8ddc72cc : SUCCESS in 20m 01s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/882e1a853bb84ac293846de7fa369e27 : SUCCESS in 20m 20s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/099cbf449baa43229060e106c295965b : SUCCESS in 20m 30s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e21abd791db6439e8d85fa81e1f9e797 : SUCCESS in 5m 04s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/e67c3ea1573541f68eb094b0127a54ac : SUCCESS in 24m 12s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/efa262d898a94501a892eeafe1e7915e : SUCCESS in 23m 57s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/6f1e4142b2454fe9b0d6d810cadfbfbb : SUCCESS in 34m 19s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/e39f1a5680ad4f75aedab3af4baa93d2 : SUCCESS in 20m 37s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/840c1900cbea443a8c6f71fe9e27e193 : SUCCESS in 16m 13s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/c0f2b8d55f224002868669586d955cf7 : SUCCESS in 15m 00s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/cda8383ebbfb490ca7fb417ead5f4bc4 : SUCCESS in 15m 43s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/ac2520a273274a5fb03a707540b6a546 : SUCCESS in 10m 29s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/5623ae1f8a8544acb593a0cca986d303 : SUCCESS in 6m 08s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/6f02585c522440ae8d804cf5fa53c560 : SUCCESS in 5m 55s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/21a2fe513bc44022a879b33969421064 : SUCCESS in 1h 07m 37s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b69bbea0a533440dbe57f663382116f3 : SUCCESS in 54m 02s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/bd889dfa20d94f5ca5a60b4194d971ef : SUCCESS in 54m 28s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/eb0627a0689d4f21ac6b4c3e566bbd64 : SUCCESS in 32m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9000b20e5af74b8b9ffba169870d1980 : SUCCESS in 6m 09s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e5324b76c8ef4f7fad5c0d2e638a4a00 : SUCCESS in 27m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/901f3306d60749668248c563ed5a76ed : SUCCESS in 1h 03m 10s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":101}],"current_revision_number":101,"current_revision":"46fa4894da0c489d746a11072c3fbc7b7ca95a7b","revisions":{"178674f074e20462cca411ea28b0f6a348685c5e":{"kind":"REWORK","_number":1,"created":"2021-07-13 17:47:53.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/01/800701/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/800701/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/800701/1"}}},"commit":{"parents":[{"commit":"88d87afda0bee3e5f3f8d2fccdcdeae063598223","subject":"Merge \"Sharder: Add root_audit overlap stats\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/88d87afda0bee3e5f3f8d2fccdcdeae063598223"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2021-07-13 17:34:48.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2021-07-13 17:34:48.000000000","tz":-300},"subject":"WIP: expirer the entire s3api MPU","message":"WIP: expirer the entire s3api MPU\n\nIt\u0027s not possible to reference s3api MPU segments from s3api - they may\nas well be in the reserved namespace.  Since X-Delete-At expiration will\nonly be set on the manifest; we should at least *try* to avoid creating\norphaned segments on purpose.\n\nThe use of SLO for s3api MPUs is an implementation detail.  Because SLO\nsegments are NOT in a reserved namespace it would be appropriate to\nclean up SLOs automatically in this mannor.\n\nThis is just a best effort cleanup, s3api MPUs are already known to\ngenerate orphaned segments [1].  A future feature \"Atomic Large Object\"\nexpiratoin will require processing new ALO tombstones rows in the\ncontainer such that we ALWAYS throw tombstones for segments under the\nreference of the manifest\u0027s reserved namespace (possibly even if *their*\nbest effort inline cleanup attempt had already done so)\n\n1. lp bug #1813202\n\nChange-Id: Ib5cec0e552cd51592653e9e100729fcf91f99415\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/178674f074e20462cca411ea28b0f6a348685c5e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/178674f074e20462cca411ea28b0f6a348685c5e"}]},"branch":"refs/heads/master"},"c784676a56c40db1454db30536c1c634922529ce":{"kind":"REWORK","_number":2,"created":"2021-12-17 18:21:06.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/01/800701/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/800701/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/800701/2"}}},"commit":{"parents":[{"commit":"4d48077a9882aff423e8530a6ff7dffbf4c317f8","subject":"DNM: add a probetest for MPU expiration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4d48077a9882aff423e8530a6ff7dffbf4c317f8"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2021-07-13 17:34:48.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2021-12-17 18:19:37.000000000","tz":-360},"subject":"WIP: expirer the entire s3api MPU","message":"WIP: expirer the entire s3api MPU\n\nIt\u0027s not possible to reference s3api MPU segments from s3api - they may\nas well be in the reserved namespace.  Since X-Delete-At expiration will\nonly be set on the manifest; we should at least *try* to avoid creating\norphaned segments on purpose.\n\nThe use of SLO for s3api MPUs is an implementation detail.  Because SLO\nsegments are NOT in a reserved namespace it would NOT be appropriate to\nclean up SLOs automatically in this mannor.\n\nThis is just a best effort cleanup, s3api MPUs are already known to\ngenerate orphaned segments [1].  A future feature \"Atomic Large Object\"\nexpiration will require processing new ALO tombstones rows in the\ncontainer such that we ALWAYS throw tombstones for segments under the\nreference of the manifest\u0027s reserved namespace (possibly even if *their*\nbest effort inline cleanup attempt had already done so)\n\n1. lp bug #1813202\n\nChange-Id: Ib5cec0e552cd51592653e9e100729fcf91f99415\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c784676a56c40db1454db30536c1c634922529ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c784676a56c40db1454db30536c1c634922529ce"}]},"branch":"refs/heads/master"},"3292127828b740737f169349e4fce4e8e8ec0316":{"kind":"REWORK","_number":3,"created":"2021-12-20 21:06:55.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/01/800701/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/800701/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/800701/3"}}},"commit":{"parents":[{"commit":"a2b3add9226b996c40be80a9274e4795887ee171","subject":"DNM: add a probetest for MPU expiration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a2b3add9226b996c40be80a9274e4795887ee171"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2021-07-13 17:34:48.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2021-12-20 21:06:43.000000000","tz":-360},"subject":"WIP: expirer the entire s3api MPU","message":"WIP: expirer the entire s3api MPU\n\nIt\u0027s not possible to reference s3api MPU segments from s3api - they may\nas well be in the reserved namespace.  Since X-Delete-At expiration will\nonly be set on the manifest; we should at least *try* to avoid creating\norphaned segments on purpose.\n\nThe use of SLO for s3api MPUs is an implementation detail.  Because SLO\nsegments are NOT in a reserved namespace it would be appropriate to\nclean up SLOs automatically in this mannor.\n\nThis is just a best effort cleanup, s3api MPUs are already known to\ngenerate orphaned segments [1].  A future feature \"Atomic Large Object\"\nexpiratoin will require processing new ALO tombstones rows in the\ncontainer such that we ALWAYS throw tombstones for segments under the\nreference of the manifest\u0027s reserved namespace (possibly even if *their*\nbest effort inline cleanup attempt had already done so)\n\n1. lp bug #1813202\n\nChange-Id: Ib5cec0e552cd51592653e9e100729fcf91f99415\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3292127828b740737f169349e4fce4e8e8ec0316"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3292127828b740737f169349e4fce4e8e8ec0316"}]},"branch":"refs/heads/master"},"a4c555a76e54350e3855e5df3f5fd6418383d898":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2021-12-21 02:30:44.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/01/800701/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/800701/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/800701/4"}}},"commit":{"parents":[{"commit":"768dc510d29295651bba1f1542cb2e36ada0d301","subject":"DNM: add a probetest for MPU expiration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/768dc510d29295651bba1f1542cb2e36ada0d301"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2021-07-13 17:34:48.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2021-12-21 02:30:44.000000000","tz":0},"subject":"WIP: expirer the entire s3api MPU","message":"WIP: expirer the entire s3api MPU\n\nIt\u0027s not possible to reference s3api MPU segments from s3api - they may\nas well be in the reserved namespace.  Since X-Delete-At expiration will\nonly be set on the manifest; we should at least *try* to avoid creating\norphaned segments on purpose.\n\nThe use of SLO for s3api MPUs is an implementation detail.  Because SLO\nsegments are NOT in a reserved namespace it would be appropriate to\nclean up SLOs automatically in this mannor.\n\nThis is just a best effort cleanup, s3api MPUs are already known to\ngenerate orphaned segments [1].  A future feature \"Atomic Large Object\"\nexpiratoin will require processing new ALO tombstones rows in the\ncontainer such that we ALWAYS throw tombstones for segments under the\nreference of the manifest\u0027s reserved namespace (possibly even if *their*\nbest effort inline cleanup attempt had already done so)\n\n1. lp bug #1813202\n\nChange-Id: Ib5cec0e552cd51592653e9e100729fcf91f99415\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a4c555a76e54350e3855e5df3f5fd6418383d898"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a4c555a76e54350e3855e5df3f5fd6418383d898"}]},"branch":"refs/heads/master","description":"Rebase"},"edff82be41225a2002ed177c2d1348e1d5873f72":{"kind":"REWORK","_number":5,"created":"2021-12-21 16:59:53.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/01/800701/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/800701/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/800701/5"}}},"commit":{"parents":[{"commit":"3dd11971dcfdc118b22ab2b5a7d703d2c0288d36","subject":"DNM: add a probetest for MPU expiration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3dd11971dcfdc118b22ab2b5a7d703d2c0288d36"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2021-07-13 17:34:48.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2021-12-21 16:59:39.000000000","tz":-360},"subject":"WIP: expirer the entire s3api MPU","message":"WIP: expirer the entire s3api MPU\n\nIt\u0027s not possible to reference s3api MPU segments from s3api - they may\nas well be in the reserved namespace.  Since X-Delete-At expiration will\nonly be set on the manifest; we should at least *try* to avoid creating\norphaned segments on purpose.\n\nThe use of SLO for s3api MPUs is an implementation detail.  Because SLO\nsegments are NOT in a reserved namespace it would be appropriate to\nclean up SLOs automatically in this mannor.\n\nThis is just a best effort cleanup, s3api MPUs are already known to\ngenerate orphaned segments [1].  A future feature \"Atomic Large Object\"\nexpiratoin will require processing new ALO tombstones rows in the\ncontainer such that we ALWAYS throw tombstones for segments under the\nreference of the manifest\u0027s reserved namespace (possibly even if *their*\nbest effort inline cleanup attempt had already done so)\n\n1. lp bug #1813202\n\nChange-Id: Ib5cec0e552cd51592653e9e100729fcf91f99415\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/edff82be41225a2002ed177c2d1348e1d5873f72"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/edff82be41225a2002ed177c2d1348e1d5873f72"}]},"branch":"refs/heads/master"},"699a89144abff95a77b98be5595978697022fa4f":{"kind":"REWORK","_number":6,"created":"2021-12-22 17:17:32.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/01/800701/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/800701/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/800701/6"}}},"commit":{"parents":[{"commit":"3dd11971dcfdc118b22ab2b5a7d703d2c0288d36","subject":"DNM: add a probetest for MPU expiration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3dd11971dcfdc118b22ab2b5a7d703d2c0288d36"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2021-07-13 17:34:48.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2021-12-21 22:56:47.000000000","tz":-360},"subject":"WIP: expirer the entire s3api MPU","message":"WIP: expirer the entire s3api MPU\n\nIt\u0027s not possible to reference s3api MPU segments from s3api - they may\nas well be in the reserved namespace.  Since X-Delete-At expiration will\nonly be set on the manifest; we should at least *try* to avoid creating\norphaned segments on purpose.\n\nThe use of SLO for s3api MPUs is an implementation detail.  Because SLO\nsegments are NOT in a reserved namespace it would be appropriate to\nclean up SLOs automatically in this mannor.\n\nThis is just a best effort cleanup, s3api MPUs are already known to\ngenerate orphaned segments [1].  A future feature \"Atomic Large Object\"\nexpiratoin will require processing new ALO tombstones rows in the\ncontainer such that we ALWAYS throw tombstones for segments under the\nreference of the manifest\u0027s reserved namespace (possibly even if *their*\nbest effort inline cleanup attempt had already done so)\n\n1. lp bug #1813202\n\nChange-Id: Ib5cec0e552cd51592653e9e100729fcf91f99415\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/699a89144abff95a77b98be5595978697022fa4f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/699a89144abff95a77b98be5595978697022fa4f"}]},"branch":"refs/heads/master"},"1d153f27be531b8c4602504c8fe18273bb6477a4":{"kind":"REWORK","_number":7,"created":"2022-01-06 22:16:28.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/01/800701/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/800701/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/800701/7"}}},"commit":{"parents":[{"commit":"e28abdeea18fb283f6d74d7b432aa37165c642c1","subject":"DNM: add a probetest for MPU expiration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e28abdeea18fb283f6d74d7b432aa37165c642c1"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2021-07-13 17:34:48.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2022-01-06 22:15:57.000000000","tz":-360},"subject":"WIP: expirer the entire s3api MPU","message":"WIP: expirer the entire s3api MPU\n\nIt\u0027s not possible to reference s3api MPU segments from s3api - they may\nas well be in the reserved namespace.  Since X-Delete-At expiration will\nonly be set on the manifest; we should at least *try* to avoid creating\norphaned segments on purpose.\n\nThe use of SLO for s3api MPUs is an implementation detail.  Because SLO\nsegments are NOT in a reserved namespace it would not be appropriate to\nclean up SLOs automatically in this mannor.\n\nThis is just a best effort cleanup, s3api MPUs already have a related\nbug to address orphaned segments.\n\nRelated-Bug: #1813202\nChange-Id: Ib5cec0e552cd51592653e9e100729fcf91f99415\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1d153f27be531b8c4602504c8fe18273bb6477a4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1d153f27be531b8c4602504c8fe18273bb6477a4"}]},"branch":"refs/heads/master"},"210807787136a603e7ed06b5bd87efedff60afe3":{"kind":"REWORK","_number":8,"created":"2022-01-07 22:11:31.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/01/800701/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/800701/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/800701/8"}}},"commit":{"parents":[{"commit":"3972b8b4baf20a36d140c90d6e38316e5d839104","subject":"DNM: add a probetest for MPU expiration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3972b8b4baf20a36d140c90d6e38316e5d839104"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2021-07-13 17:34:48.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2022-01-07 22:10:31.000000000","tz":-360},"subject":"Expire the entire s3api MPU","message":"Expire the entire s3api MPU\n\nIt\u0027s not possible to reference s3api MPU segments from s3api - they may\nas well be in the reserved namespace.  Since X-Delete-At expiration will\nonly be set on the manifest; we should at least *try* to avoid creating\norphaned segments on purpose.\n\nThe use of SLO for s3api MPUs is an implementation detail.  Because SLO\nsegments are NOT in a reserved namespace it would not be appropriate to\nclean up SLOs automatically in this mannor.\n\nThis is just a best effort cleanup, s3api MPUs already have a related\nbug to address orphaned segments.\n\nRelated-Bug: #1813202\nChange-Id: Ib5cec0e552cd51592653e9e100729fcf91f99415\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/210807787136a603e7ed06b5bd87efedff60afe3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/210807787136a603e7ed06b5bd87efedff60afe3"}]},"branch":"refs/heads/master"},"d333a1d67e8bf2214b28ac7788647b0ba5747aba":{"kind":"REWORK","_number":9,"created":"2022-01-11 20:40:16.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/01/800701/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/800701/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/800701/9"}}},"commit":{"parents":[{"commit":"96b798019602054382634b9d08e3bf26de90498d","subject":"Merge \"tests: Improve FakeMemcache call tracking\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/96b798019602054382634b9d08e3bf26de90498d"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2021-12-17 15:39:44.000000000","tz":-360},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2022-01-11 20:28:09.000000000","tz":-360},"subject":"Expire the entire s3api MPU","message":"Expire the entire s3api MPU\n\nIt\u0027s not possible to reference s3api MPU segments from s3api - they may\nas well be in the reserved namespace.  Since X-Delete-At expiration will\nonly be set on the manifest; we should at least *try* to avoid creating\norphaned segments on purpose.\n\nThe use of SLO for s3api MPUs is an implementation detail.  Because SLO\nsegments are NOT in a reserved namespace it would not be appropriate to\nclean up SLOs automatically in this mannor.\n\nThis is just a best effort cleanup, s3api MPUs already have a related\nbug to address orphaned segments.\n\nRelated-Bug: #1813202\nChange-Id: Ib5cec0e552cd51592653e9e100729fcf91f99415\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d333a1d67e8bf2214b28ac7788647b0ba5747aba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d333a1d67e8bf2214b28ac7788647b0ba5747aba"}]},"branch":"refs/heads/master"},"60cb24a6beaeb592b47fcbbd5bfab4b97a119d48":{"kind":"REWORK","_number":10,"created":"2022-01-13 06:13:03.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/01/800701/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/800701/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/800701/10"}}},"commit":{"parents":[{"commit":"96b798019602054382634b9d08e3bf26de90498d","subject":"Merge \"tests: Improve FakeMemcache call tracking\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/96b798019602054382634b9d08e3bf26de90498d"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2021-12-17 15:39:44.000000000","tz":-360},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2022-01-13 06:12:39.000000000","tz":-360},"subject":"Expire the entire s3api MPU","message":"Expire the entire s3api MPU\n\nIt\u0027s not possible to reference s3api MPU segments from s3api - they may\nas well be in the reserved namespace.  Since X-Delete-At expiration will\nonly be set on the manifest; we should at least *try* to avoid creating\norphaned segments on purpose.\n\nThe use of SLO for s3api MPUs is an implementation detail.  Because SLO\nsegments are NOT in a reserved namespace it would not be appropriate to\nclean up SLOs automatically in this manner.\n\nThis is just a best effort cleanup, s3api MPUs already have a related\nbug to address orphaned segments.\n\nRelated-Bug: #1813202\nChange-Id: Ib5cec0e552cd51592653e9e100729fcf91f99415\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/60cb24a6beaeb592b47fcbbd5bfab4b97a119d48"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/60cb24a6beaeb592b47fcbbd5bfab4b97a119d48"}]},"branch":"refs/heads/master"},"54ddec106774608414dc300cc30bed9d1d8e3c54":{"kind":"NO_CODE_CHANGE","_number":11,"created":"2022-01-13 20:13:06.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/01/800701/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/800701/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/800701/11"}}},"commit":{"parents":[{"commit":"96b798019602054382634b9d08e3bf26de90498d","subject":"Merge \"tests: Improve FakeMemcache call tracking\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/96b798019602054382634b9d08e3bf26de90498d"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2021-12-17 15:39:44.000000000","tz":-360},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2022-01-13 20:12:36.000000000","tz":-360},"subject":"Delete s3api MPU segments when the manifest expires","message":"Delete s3api MPU segments when the manifest expires\n\nThe object-expirer will delete s3api MPU segments automatically when it\nexpires an s3api manifest.\n\nThis is just a best effort cleanup of the segments. In some\ncircumstances (e.g. when manifest objects are not consistent) this may\nfail to delete all segments. However, s3api MPUs already have a related\nbug to address orphaned segments.  The automatic deletion is only\napplied to s3api segments, not to SLO segments in general, because:\n\n(a) s3api segments cannot be referenced via the S3 API, and are\ntherefore orphaned as soon as the manifest is deleted; SLO segments in\ngeneral are not necessarily orphaned when a manifest is deleted.\n\n(b) the names of s3api MPU segments can be easily inferred from the\nmanifest DELETE response headers without inspecting the manifest body.\n\nRelated-Bug: #1813202\nChange-Id: Ib5cec0e552cd51592653e9e100729fcf91f99415\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/54ddec106774608414dc300cc30bed9d1d8e3c54"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/54ddec106774608414dc300cc30bed9d1d8e3c54"}]},"branch":"refs/heads/master"},"b01be931d1cee2dd67e536ec2bf9a1485c912e40":{"kind":"REWORK","_number":12,"created":"2022-01-14 13:46:26.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/01/800701/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/800701/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/800701/12"}}},"commit":{"parents":[{"commit":"96b798019602054382634b9d08e3bf26de90498d","subject":"Merge \"tests: Improve FakeMemcache call tracking\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/96b798019602054382634b9d08e3bf26de90498d"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2021-12-17 15:39:44.000000000","tz":-360},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2022-01-14 13:45:56.000000000","tz":-360},"subject":"Delete s3api MPU segments when the manifest expires","message":"Delete s3api MPU segments when the manifest expires\n\nThe object-expirer will delete s3api MPU segments automatically when it\nexpires an s3api manifest.\n\nThis is just a best effort cleanup of the segments. In some\ncircumstances (e.g. when manifest objects are not consistent) this may\nfail to delete all segments. However, s3api MPUs already have a related\nbug to address orphaned segments.  The automatic deletion is only\napplied to s3api segments, not to SLO segments in general, because:\n\n(a) s3api segments cannot be referenced via the S3 API, and are\ntherefore orphaned as soon as the manifest is deleted; SLO segments in\ngeneral are not necessarily orphaned when a manifest is deleted.\n\n(b) the names of s3api MPU segments can be easily inferred from the\nmanifest DELETE response headers without inspecting the manifest body.\n\nRelated-Bug: #1813202\nChange-Id: Ib5cec0e552cd51592653e9e100729fcf91f99415\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b01be931d1cee2dd67e536ec2bf9a1485c912e40"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b01be931d1cee2dd67e536ec2bf9a1485c912e40"}]},"branch":"refs/heads/master"},"42d03756296c570f206080e9f99e9572e41d836c":{"kind":"REWORK","_number":13,"created":"2022-01-14 15:03:32.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/01/800701/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/800701/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/800701/13"}}},"commit":{"parents":[{"commit":"96b798019602054382634b9d08e3bf26de90498d","subject":"Merge \"tests: Improve FakeMemcache call tracking\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/96b798019602054382634b9d08e3bf26de90498d"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2021-12-17 15:39:44.000000000","tz":-360},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2022-01-14 15:03:19.000000000","tz":-360},"subject":"Delete s3api MPU segments when the manifest expires","message":"Delete s3api MPU segments when the manifest expires\n\nThe object-expirer will delete s3api MPU segments automatically when it\nexpires an s3api manifest.\n\nThis is just a best effort cleanup of the segments. In some\ncircumstances (e.g. when manifest objects are not consistent) this may\nfail to delete all segments. However, s3api MPUs already have a related\nbug to address orphaned segments.  The automatic deletion is only\napplied to s3api segments, not to SLO segments in general, because:\n\n(a) s3api segments cannot be referenced via the S3 API, and are\ntherefore orphaned as soon as the manifest is deleted; SLO segments in\ngeneral are not necessarily orphaned when a manifest is deleted.\n\n(b) the names of s3api MPU segments can be easily inferred from the\nmanifest DELETE response headers without inspecting the manifest body.\n\nRelated-Bug: #1813202\nChange-Id: Ib5cec0e552cd51592653e9e100729fcf91f99415\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/42d03756296c570f206080e9f99e9572e41d836c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/42d03756296c570f206080e9f99e9572e41d836c"}]},"branch":"refs/heads/master"},"baf8d307496c8105e9017ba559905cf2934badbb":{"kind":"REWORK","_number":14,"created":"2022-01-28 15:57:22.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/01/800701/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/800701/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/800701/14"}}},"commit":{"parents":[{"commit":"b0245f4eb5e625a0528e8ce607b1d604201bc6da","subject":"Proxy: override user_agent with backend_user_agent","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b0245f4eb5e625a0528e8ce607b1d604201bc6da"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2021-12-17 15:39:44.000000000","tz":-360},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2022-01-28 15:47:03.000000000","tz":-360},"subject":"Delete s3api MPU segments when the manifest expires","message":"Delete s3api MPU segments when the manifest expires\n\nThe object-expirer will delete s3api MPU segments automatically when it\nexpires an s3api manifest.\n\nThis is just a best effort cleanup of the segments. In some\ncircumstances (e.g. when manifest objects are not consistent) this may\nfail to delete all segments. However, s3api MPUs already have a related\nbug to address orphaned segments.  The automatic deletion is only\napplied to s3api segments, not to SLO segments in general, because:\n\n(a) s3api segments cannot be referenced via the S3 API, and are\ntherefore orphaned as soon as the manifest is deleted; SLO segments in\ngeneral are not necessarily orphaned when a manifest is deleted.\n\n(b) the names of s3api MPU segments can be easily inferred from the\nmanifest DELETE response headers without inspecting the manifest body.\n\nRelated-Bug: #1813202\nChange-Id: Ib5cec0e552cd51592653e9e100729fcf91f99415\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/baf8d307496c8105e9017ba559905cf2934badbb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/baf8d307496c8105e9017ba559905cf2934badbb"}]},"branch":"refs/heads/master"},"28553920d653d08fe0b06af0bda7808ec8b5a2a4":{"kind":"REWORK","_number":15,"created":"2022-02-14 18:18:32.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/01/800701/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/800701/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/800701/15"}}},"commit":{"parents":[{"commit":"b0245f4eb5e625a0528e8ce607b1d604201bc6da","subject":"Proxy: override user_agent with backend_user_agent","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b0245f4eb5e625a0528e8ce607b1d604201bc6da"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2021-12-17 15:39:44.000000000","tz":-360},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2022-02-14 18:16:32.000000000","tz":-360},"subject":"Delete s3api MPU segments when the manifest expires","message":"Delete s3api MPU segments when the manifest expires\n\nThe object-expirer will delete s3api MPU segments automatically when it\nexpires an s3api manifest.\n\nThis is just a best effort cleanup of the segments. In some\ncircumstances (e.g. when manifest objects are not consistent) this may\nfail to delete all segments. However, s3api MPUs already have a related\nbug to address orphaned segments.  The automatic deletion is only\napplied to s3api segments, not to SLO segments in general, because:\n\n(a) s3api segments cannot be referenced via the S3 API, and are\ntherefore orphaned as soon as the manifest is deleted; SLO segments in\ngeneral are not necessarily orphaned when a manifest is deleted.\n\n(b) the names of s3api MPU segments can be easily inferred from the\nmanifest DELETE response headers without inspecting the manifest body.\n\nRelated-Bug: #1813202\nChange-Id: Ib5cec0e552cd51592653e9e100729fcf91f99415\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/28553920d653d08fe0b06af0bda7808ec8b5a2a4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/28553920d653d08fe0b06af0bda7808ec8b5a2a4"}]},"branch":"refs/heads/master"},"f084cd291b306d73fc86786e00e54d4a0465d253":{"kind":"REWORK","_number":16,"created":"2022-10-17 18:21:53.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"ref":"refs/changes/01/800701/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/800701/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/800701/16"}}},"commit":{"parents":[{"commit":"cc033154ad4a4f345258457f3ceed9143fb3d46d","subject":"py2constraints: pin PasteDeploy version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cc033154ad4a4f345258457f3ceed9143fb3d46d"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2021-12-17 15:39:44.000000000","tz":-360},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2022-10-17 18:21:32.000000000","tz":-420},"subject":"Delete s3api MPU segments when the manifest expires","message":"Delete s3api MPU segments when the manifest expires\n\nThe object-expirer will delete s3api MPU segments automatically when it\nexpires an s3api manifest.\n\nThis is just a best effort cleanup of the segments. In some\ncircumstances (e.g. when manifest objects are not consistent) this may\nfail to delete all segments. However, s3api MPUs already have a related\nbug to address orphaned segments.  The automatic deletion is only\napplied to s3api segments, not to SLO segments in general, because:\n\n(a) s3api segments cannot be referenced via the S3 API, and are\ntherefore orphaned as soon as the manifest is deleted; SLO segments in\ngeneral are not necessarily orphaned when a manifest is deleted.\n\n(b) the names of s3api MPU segments can be easily inferred from the\nmanifest DELETE response headers without inspecting the manifest body.\n\nRelated-Bug: #1813202\nChange-Id: Ib5cec0e552cd51592653e9e100729fcf91f99415\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f084cd291b306d73fc86786e00e54d4a0465d253"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f084cd291b306d73fc86786e00e54d4a0465d253"}]},"branch":"refs/heads/master"},"96f3eb7a0aa1ba6cbb3425b419b9e2e181c7ecfc":{"kind":"REWORK","_number":17,"created":"2022-11-04 04:15:22.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/01/800701/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/800701/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/800701/17"}}},"commit":{"parents":[{"commit":"c4ad66b49418b569f920d26adad739a9102e19c9","subject":"Sharder: add more logs during container cleaving.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c4ad66b49418b569f920d26adad739a9102e19c9"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2021-12-17 15:39:44.000000000","tz":-360},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2022-11-04 04:15:14.000000000","tz":-300},"subject":"Delete s3api MPU segments when the manifest expires","message":"Delete s3api MPU segments when the manifest expires\n\nThe object-expirer will delete s3api MPU segments automatically when it\nexpires an s3api manifest.\n\nThis is just a best effort cleanup of the segments. In some\ncircumstances (e.g. when manifest objects are not consistent) this may\nfail to delete all segments. However, s3api MPUs already have a related\nbug to address orphaned segments.  The automatic deletion is only\napplied to s3api segments, not to SLO segments in general, because:\n\n(a) s3api segments cannot be referenced via the S3 API, and are\ntherefore orphaned as soon as the manifest is deleted; SLO segments in\ngeneral are not necessarily orphaned when a manifest is deleted.\n\n(b) the names of s3api MPU segments can be easily inferred from the\nmanifest DELETE response headers without inspecting the manifest body.\n\nRelated-Bug: #1813202\nChange-Id: Ib5cec0e552cd51592653e9e100729fcf91f99415\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/96f3eb7a0aa1ba6cbb3425b419b9e2e181c7ecfc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/96f3eb7a0aa1ba6cbb3425b419b9e2e181c7ecfc"}]},"branch":"refs/heads/master"},"52460eed7518ae432c44d3547fea9196b0dc026b":{"kind":"REWORK","_number":18,"created":"2023-01-03 18:31:03.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/01/800701/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/800701/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/800701/18"}}},"commit":{"parents":[{"commit":"2d7c1dc6dd8a672c1f89a46b2423898d33ff63d2","subject":"CI: Fix our usage of tox","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2d7c1dc6dd8a672c1f89a46b2423898d33ff63d2"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2021-12-17 15:39:44.000000000","tz":-360},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2023-01-03 18:30:54.000000000","tz":-360},"subject":"Delete s3api MPU segments when the manifest expires","message":"Delete s3api MPU segments when the manifest expires\n\nThe object-expirer will delete s3api MPU segments automatically when it\nexpires an s3api manifest.\n\nThis is just a best effort cleanup of the segments. In some\ncircumstances (e.g. when manifest objects are not consistent) this may\nfail to delete all segments. However, s3api MPUs already have a related\nbug to address orphaned segments.  The automatic deletion is only\napplied to s3api segments, not to SLO segments in general, because:\n\n(a) s3api segments cannot be referenced via the S3 API, and are\ntherefore orphaned as soon as the manifest is deleted; SLO segments in\ngeneral are not necessarily orphaned when a manifest is deleted.\n\n(b) the names of s3api MPU segments can be easily inferred from the\nmanifest DELETE response headers without inspecting the manifest body.\n\nRelated-Bug: #1813202\nChange-Id: Ib5cec0e552cd51592653e9e100729fcf91f99415\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/52460eed7518ae432c44d3547fea9196b0dc026b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/52460eed7518ae432c44d3547fea9196b0dc026b"}]},"branch":"refs/heads/master"},"d1a1821d2b32110377e6b8f3ba884f63510d27af":{"kind":"REWORK","_number":19,"created":"2023-02-09 15:11:45.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/01/800701/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/800701/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/800701/19"}}},"commit":{"parents":[{"commit":"231a67e1e9d4b752bb5ca8c465a887e36ef8c2bb","subject":"Merge \"object-replicator: Count sync failures in update_deleted\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/231a67e1e9d4b752bb5ca8c465a887e36ef8c2bb"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2021-12-17 15:39:44.000000000","tz":-360},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-02-09 15:11:29.000000000","tz":0},"subject":"Delete s3api MPU segments when the manifest expires","message":"Delete s3api MPU segments when the manifest expires\n\nThe object-expirer will delete s3api MPU segments automatically when it\nexpires an s3api manifest.\n\nThis is just a best effort cleanup of the segments. In some\ncircumstances (e.g. when manifest objects are not consistent) this may\nfail to delete all segments. However, s3api MPUs already have a related\nbug to address orphaned segments.  The automatic deletion is only\napplied to s3api segments, not to SLO segments in general, because:\n\n(a) s3api segments cannot be referenced via the S3 API, and are\ntherefore orphaned as soon as the manifest is deleted; SLO segments in\ngeneral are not necessarily orphaned when a manifest is deleted.\n\n(b) the names of s3api MPU segments can be easily inferred from the\nmanifest DELETE response headers without inspecting the manifest body.\n\nRelated-Bug: #1813202\nChange-Id: Ib5cec0e552cd51592653e9e100729fcf91f99415\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d1a1821d2b32110377e6b8f3ba884f63510d27af"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d1a1821d2b32110377e6b8f3ba884f63510d27af"}]},"branch":"refs/heads/master"},"001470f4e4ca141163298ee671faf4945a4a13a2":{"kind":"REWORK","_number":20,"created":"2023-07-25 01:41:14.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/01/800701/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/800701/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/800701/20"}}},"commit":{"parents":[{"commit":"8b900becd5bed2e0c0555cc55bcaef372910b88b","subject":"Merge \"proxy: remove client_chunk_size and skip_bytes from GetOrHeadHandler\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8b900becd5bed2e0c0555cc55bcaef372910b88b"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2021-12-17 15:39:44.000000000","tz":-360},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2023-07-25 01:41:01.000000000","tz":600},"subject":"Delete s3api MPU segments when the manifest expires","message":"Delete s3api MPU segments when the manifest expires\n\nThe object-expirer will delete s3api MPU segments automatically when it\nexpires an s3api manifest.\n\nThis is just a best effort cleanup of the segments. In some\ncircumstances (e.g. when manifest objects are not consistent) this may\nfail to delete all segments. However, s3api MPUs already have a related\nbug to address orphaned segments.  The automatic deletion is only\napplied to s3api segments, not to SLO segments in general, because:\n\n(a) s3api segments cannot be referenced via the S3 API, and are\ntherefore orphaned as soon as the manifest is deleted; SLO segments in\ngeneral are not necessarily orphaned when a manifest is deleted.\n\n(b) the names of s3api MPU segments can be easily inferred from the\nmanifest DELETE response headers without inspecting the manifest body.\n\nRelated-Bug: #1813202\nChange-Id: Ib5cec0e552cd51592653e9e100729fcf91f99415\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/001470f4e4ca141163298ee671faf4945a4a13a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/001470f4e4ca141163298ee671faf4945a4a13a2"}]},"branch":"refs/heads/master"},"817e4cdb89ff4d706efeae6edb1642c33397b68e":{"kind":"REWORK","_number":21,"created":"2023-09-28 22:43:53.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/01/800701/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/800701/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/800701/21"}}},"commit":{"parents":[{"commit":"8b900becd5bed2e0c0555cc55bcaef372910b88b","subject":"Merge \"proxy: remove client_chunk_size and skip_bytes from GetOrHeadHandler\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8b900becd5bed2e0c0555cc55bcaef372910b88b"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2021-12-17 15:39:44.000000000","tz":-360},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-09-28 22:43:46.000000000","tz":-420},"subject":"Delete s3api MPU segments when the manifest expires","message":"Delete s3api MPU segments when the manifest expires\n\nThe object-expirer will delete s3api MPU segments automatically when it\nexpires an s3api manifest.\n\nThis is just a best effort cleanup of the segments. In some\ncircumstances (e.g. when manifest objects are not consistent) this may\nfail to delete all segments. However, s3api MPUs already have a related\nbug to address orphaned segments.  The automatic deletion is only\napplied to s3api segments, not to SLO segments in general, because:\n\n(a) s3api segments cannot be referenced via the S3 API, and are\ntherefore orphaned as soon as the manifest is deleted; SLO segments in\ngeneral are not necessarily orphaned when a manifest is deleted.\n\n(b) the names of s3api MPU segments can be easily inferred from the\nmanifest DELETE response headers without inspecting the manifest body.\n\nRelated-Bug: #1813202\nChange-Id: Ib5cec0e552cd51592653e9e100729fcf91f99415\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/817e4cdb89ff4d706efeae6edb1642c33397b68e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/817e4cdb89ff4d706efeae6edb1642c33397b68e"}]},"branch":"refs/heads/master"},"096c5d4b4bc3160fc5e4b3ddea6bd745613f8155":{"kind":"REWORK","_number":22,"created":"2023-10-16 22:42:45.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/01/800701/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/800701/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/800701/22"}}},"commit":{"parents":[{"commit":"99e74415ff9e472f861f0d888cb6608e3ed03df8","subject":"s3api: Stop propagating storage policy to sub-requests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/99e74415ff9e472f861f0d888cb6608e3ed03df8"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2021-12-17 15:39:44.000000000","tz":-360},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2023-10-16 22:42:36.000000000","tz":-300},"subject":"Delete s3api MPU segments when the manifest expires","message":"Delete s3api MPU segments when the manifest expires\n\nThe object-expirer will delete s3api MPU segments automatically when it\nexpires an s3api manifest.\n\nThis is just a best effort cleanup of the segments. In some\ncircumstances (e.g. when manifest objects are not consistent) this may\nfail to delete all segments. However, s3api MPUs already have a related\nbug to address orphaned segments.  The automatic deletion is only\napplied to s3api segments, not to SLO segments in general, because:\n\n(a) s3api segments cannot be referenced via the S3 API, and are\ntherefore orphaned as soon as the manifest is deleted; SLO segments in\ngeneral are not necessarily orphaned when a manifest is deleted.\n\n(b) the names of s3api MPU segments can be easily inferred from the\nmanifest DELETE response headers without inspecting the manifest body.\n\nRelated-Bug: #1813202\nChange-Id: Ib5cec0e552cd51592653e9e100729fcf91f99415\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/096c5d4b4bc3160fc5e4b3ddea6bd745613f8155"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/096c5d4b4bc3160fc5e4b3ddea6bd745613f8155"}]},"branch":"refs/heads/master"},"96062198007262dd10655ceae5aa0e77cd16e708":{"kind":"TRIVIAL_REBASE","_number":23,"created":"2023-10-17 18:10:55.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/01/800701/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/800701/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/800701/23"}}},"commit":{"parents":[{"commit":"edcea8ae38533c1df4d7378bf19b7440907e4308","subject":"s3api: Stop propagating storage policy to sub-requests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/edcea8ae38533c1df4d7378bf19b7440907e4308"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2021-12-17 15:39:44.000000000","tz":-360},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-10-17 18:10:44.000000000","tz":60},"subject":"Delete s3api MPU segments when the manifest expires","message":"Delete s3api MPU segments when the manifest expires\n\nThe object-expirer will delete s3api MPU segments automatically when it\nexpires an s3api manifest.\n\nThis is just a best effort cleanup of the segments. In some\ncircumstances (e.g. when manifest objects are not consistent) this may\nfail to delete all segments. However, s3api MPUs already have a related\nbug to address orphaned segments.  The automatic deletion is only\napplied to s3api segments, not to SLO segments in general, because:\n\n(a) s3api segments cannot be referenced via the S3 API, and are\ntherefore orphaned as soon as the manifest is deleted; SLO segments in\ngeneral are not necessarily orphaned when a manifest is deleted.\n\n(b) the names of s3api MPU segments can be easily inferred from the\nmanifest DELETE response headers without inspecting the manifest body.\n\nRelated-Bug: #1813202\nChange-Id: Ib5cec0e552cd51592653e9e100729fcf91f99415\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/96062198007262dd10655ceae5aa0e77cd16e708"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/96062198007262dd10655ceae5aa0e77cd16e708"}]},"branch":"refs/heads/master"},"9a85f5a49c2ad06b1109b6471b4643ea30470794":{"kind":"TRIVIAL_REBASE","_number":24,"created":"2023-10-18 09:53:30.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/01/800701/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/800701/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/800701/24"}}},"commit":{"parents":[{"commit":"ea4770483640cfb0cc5ef39031f12ac050f41596","subject":"s3api: Stop propagating storage policy to sub-requests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ea4770483640cfb0cc5ef39031f12ac050f41596"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2021-12-17 15:39:44.000000000","tz":-360},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-10-18 09:53:17.000000000","tz":60},"subject":"Delete s3api MPU segments when the manifest expires","message":"Delete s3api MPU segments when the manifest expires\n\nThe object-expirer will delete s3api MPU segments automatically when it\nexpires an s3api manifest.\n\nThis is just a best effort cleanup of the segments. In some\ncircumstances (e.g. when manifest objects are not consistent) this may\nfail to delete all segments. However, s3api MPUs already have a related\nbug to address orphaned segments.  The automatic deletion is only\napplied to s3api segments, not to SLO segments in general, because:\n\n(a) s3api segments cannot be referenced via the S3 API, and are\ntherefore orphaned as soon as the manifest is deleted; SLO segments in\ngeneral are not necessarily orphaned when a manifest is deleted.\n\n(b) the names of s3api MPU segments can be easily inferred from the\nmanifest DELETE response headers without inspecting the manifest body.\n\nRelated-Bug: #1813202\nChange-Id: Ib5cec0e552cd51592653e9e100729fcf91f99415\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9a85f5a49c2ad06b1109b6471b4643ea30470794"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9a85f5a49c2ad06b1109b6471b4643ea30470794"}]},"branch":"refs/heads/master"},"d405c54ba8b117310ec199acac0be52e60d7dce4":{"kind":"TRIVIAL_REBASE","_number":25,"created":"2023-10-18 10:27:17.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/01/800701/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/800701/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/800701/25"}}},"commit":{"parents":[{"commit":"be33b4815fa4bdc4d8d49b12c420a6623104fd3d","subject":"s3api: Stop propagating storage policy to sub-requests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/be33b4815fa4bdc4d8d49b12c420a6623104fd3d"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2021-12-17 15:39:44.000000000","tz":-360},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-10-18 10:27:06.000000000","tz":60},"subject":"Delete s3api MPU segments when the manifest expires","message":"Delete s3api MPU segments when the manifest expires\n\nThe object-expirer will delete s3api MPU segments automatically when it\nexpires an s3api manifest.\n\nThis is just a best effort cleanup of the segments. In some\ncircumstances (e.g. when manifest objects are not consistent) this may\nfail to delete all segments. However, s3api MPUs already have a related\nbug to address orphaned segments.  The automatic deletion is only\napplied to s3api segments, not to SLO segments in general, because:\n\n(a) s3api segments cannot be referenced via the S3 API, and are\ntherefore orphaned as soon as the manifest is deleted; SLO segments in\ngeneral are not necessarily orphaned when a manifest is deleted.\n\n(b) the names of s3api MPU segments can be easily inferred from the\nmanifest DELETE response headers without inspecting the manifest body.\n\nRelated-Bug: #1813202\nChange-Id: Ib5cec0e552cd51592653e9e100729fcf91f99415\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d405c54ba8b117310ec199acac0be52e60d7dce4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d405c54ba8b117310ec199acac0be52e60d7dce4"}]},"branch":"refs/heads/master"},"0c21c9cb8626cbbcba8ea0f943ba3a570f90d092":{"kind":"TRIVIAL_REBASE","_number":26,"created":"2023-11-13 18:50:47.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/01/800701/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/800701/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/800701/26"}}},"commit":{"parents":[{"commit":"fb01d276171c52bcb08d291c35e77eff317643cc","subject":"s3api: Stop propagating storage policy to sub-requests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fb01d276171c52bcb08d291c35e77eff317643cc"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2021-12-17 15:39:44.000000000","tz":-360},"committer":{"name":"ASHWIN A NAIR","email":"nairashwin952013@gmail.com","date":"2023-11-13 18:50:47.000000000","tz":0},"subject":"Delete s3api MPU segments when the manifest expires","message":"Delete s3api MPU segments when the manifest expires\n\nThe object-expirer will delete s3api MPU segments automatically when it\nexpires an s3api manifest.\n\nThis is just a best effort cleanup of the segments. In some\ncircumstances (e.g. when manifest objects are not consistent) this may\nfail to delete all segments. However, s3api MPUs already have a related\nbug to address orphaned segments.  The automatic deletion is only\napplied to s3api segments, not to SLO segments in general, because:\n\n(a) s3api segments cannot be referenced via the S3 API, and are\ntherefore orphaned as soon as the manifest is deleted; SLO segments in\ngeneral are not necessarily orphaned when a manifest is deleted.\n\n(b) the names of s3api MPU segments can be easily inferred from the\nmanifest DELETE response headers without inspecting the manifest body.\n\nRelated-Bug: #1813202\nChange-Id: Ib5cec0e552cd51592653e9e100729fcf91f99415\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0c21c9cb8626cbbcba8ea0f943ba3a570f90d092"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0c21c9cb8626cbbcba8ea0f943ba3a570f90d092"}]},"branch":"refs/heads/master","description":"Rebase"},"cfd4cff87292d8e7133eb82eae5c8b882dadde01":{"kind":"TRIVIAL_REBASE","_number":27,"created":"2023-11-14 13:26:04.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/01/800701/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/800701/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/800701/27"}}},"commit":{"parents":[{"commit":"bcecbf1c5245a174cf0fd58178493a698f2b7c2f","subject":"s3api: Stop propagating storage policy to sub-requests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bcecbf1c5245a174cf0fd58178493a698f2b7c2f"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2021-12-17 15:39:44.000000000","tz":-360},"committer":{"name":"ASHWIN A NAIR","email":"nairashwin952013@gmail.com","date":"2023-11-14 13:26:04.000000000","tz":0},"subject":"Delete s3api MPU segments when the manifest expires","message":"Delete s3api MPU segments when the manifest expires\n\nThe object-expirer will delete s3api MPU segments automatically when it\nexpires an s3api manifest.\n\nThis is just a best effort cleanup of the segments. In some\ncircumstances (e.g. when manifest objects are not consistent) this may\nfail to delete all segments. However, s3api MPUs already have a related\nbug to address orphaned segments.  The automatic deletion is only\napplied to s3api segments, not to SLO segments in general, because:\n\n(a) s3api segments cannot be referenced via the S3 API, and are\ntherefore orphaned as soon as the manifest is deleted; SLO segments in\ngeneral are not necessarily orphaned when a manifest is deleted.\n\n(b) the names of s3api MPU segments can be easily inferred from the\nmanifest DELETE response headers without inspecting the manifest body.\n\nRelated-Bug: #1813202\nChange-Id: Ib5cec0e552cd51592653e9e100729fcf91f99415\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cfd4cff87292d8e7133eb82eae5c8b882dadde01"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cfd4cff87292d8e7133eb82eae5c8b882dadde01"}]},"branch":"refs/heads/master","description":"Rebase"},"2faf58c4cecc2f060717ac9a45c6449e03ef8e27":{"kind":"TRIVIAL_REBASE","_number":28,"created":"2023-11-14 15:20:57.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/01/800701/28","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/800701/28","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/28 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/28 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/28 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/800701/28"}}},"commit":{"parents":[{"commit":"60c04f116b41f97a5dcf9c07ef6e77ae9bdfe61e","subject":"s3api: Stop propagating storage policy to sub-requests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/60c04f116b41f97a5dcf9c07ef6e77ae9bdfe61e"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2021-12-17 15:39:44.000000000","tz":-360},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-11-14 15:13:14.000000000","tz":0},"subject":"Delete s3api MPU segments when the manifest expires","message":"Delete s3api MPU segments when the manifest expires\n\nThe object-expirer will delete s3api MPU segments automatically when it\nexpires an s3api manifest.\n\nThis is just a best effort cleanup of the segments. In some\ncircumstances (e.g. when manifest objects are not consistent) this may\nfail to delete all segments. However, s3api MPUs already have a related\nbug to address orphaned segments.  The automatic deletion is only\napplied to s3api segments, not to SLO segments in general, because:\n\n(a) s3api segments cannot be referenced via the S3 API, and are\ntherefore orphaned as soon as the manifest is deleted; SLO segments in\ngeneral are not necessarily orphaned when a manifest is deleted.\n\n(b) the names of s3api MPU segments can be easily inferred from the\nmanifest DELETE response headers without inspecting the manifest body.\n\nRelated-Bug: #1813202\nChange-Id: Ib5cec0e552cd51592653e9e100729fcf91f99415\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2faf58c4cecc2f060717ac9a45c6449e03ef8e27"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2faf58c4cecc2f060717ac9a45c6449e03ef8e27"}]},"branch":"refs/heads/master"},"af5254162122c8e1da03b05a0f5d39dc30bde8ec":{"kind":"TRIVIAL_REBASE","_number":29,"created":"2023-12-20 17:22:56.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"real_uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/01/800701/29","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/800701/29","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/29 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/29 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/29 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/800701/29"}}},"commit":{"parents":[{"commit":"365db20275d7798c30f10ee03221561a28bb7a79","subject":"FakeSwift: use HTTPMethodNotAllowed not HTTPNotImplemented","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/365db20275d7798c30f10ee03221561a28bb7a79"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2021-12-17 15:39:44.000000000","tz":-360},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-12-20 17:22:56.000000000","tz":0},"subject":"Delete s3api MPU segments when the manifest expires","message":"Delete s3api MPU segments when the manifest expires\n\nThe object-expirer will delete s3api MPU segments automatically when it\nexpires an s3api manifest.\n\nThis is just a best effort cleanup of the segments. In some\ncircumstances (e.g. when manifest objects are not consistent) this may\nfail to delete all segments. However, s3api MPUs already have a related\nbug to address orphaned segments.  The automatic deletion is only\napplied to s3api segments, not to SLO segments in general, because:\n\n(a) s3api segments cannot be referenced via the S3 API, and are\ntherefore orphaned as soon as the manifest is deleted; SLO segments in\ngeneral are not necessarily orphaned when a manifest is deleted.\n\n(b) the names of s3api MPU segments can be easily inferred from the\nmanifest DELETE response headers without inspecting the manifest body.\n\nRelated-Bug: #1813202\nChange-Id: Ib5cec0e552cd51592653e9e100729fcf91f99415\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/af5254162122c8e1da03b05a0f5d39dc30bde8ec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/af5254162122c8e1da03b05a0f5d39dc30bde8ec"}]},"branch":"refs/heads/master","description":"Rebase"},"e84831941e91c2011dd6ec87fd5e78201ae40b67":{"kind":"TRIVIAL_REBASE","_number":30,"created":"2023-12-20 18:21:31.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/01/800701/30","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/800701/30","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/30 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/30 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/30 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/800701/30"}}},"commit":{"parents":[{"commit":"2e9d68a0fd67db02696806b56fce1825041640a3","subject":"s3api: Support GET/HEAD request with ?partNumber","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2e9d68a0fd67db02696806b56fce1825041640a3"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2021-12-17 15:39:44.000000000","tz":-360},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-12-20 17:51:02.000000000","tz":0},"subject":"Delete s3api MPU segments when the manifest expires","message":"Delete s3api MPU segments when the manifest expires\n\nThe object-expirer will delete s3api MPU segments automatically when it\nexpires an s3api manifest.\n\nThis is just a best effort cleanup of the segments. In some\ncircumstances (e.g. when manifest objects are not consistent) this may\nfail to delete all segments. However, s3api MPUs already have a related\nbug to address orphaned segments.  The automatic deletion is only\napplied to s3api segments, not to SLO segments in general, because:\n\n(a) s3api segments cannot be referenced via the S3 API, and are\ntherefore orphaned as soon as the manifest is deleted; SLO segments in\ngeneral are not necessarily orphaned when a manifest is deleted.\n\n(b) the names of s3api MPU segments can be easily inferred from the\nmanifest DELETE response headers without inspecting the manifest body.\n\nRelated-Bug: #1813202\nChange-Id: Ib5cec0e552cd51592653e9e100729fcf91f99415\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e84831941e91c2011dd6ec87fd5e78201ae40b67"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e84831941e91c2011dd6ec87fd5e78201ae40b67"}]},"branch":"refs/heads/master"},"5fe7db6b865836ee54fd164d8b6babf8c11fd68e":{"kind":"TRIVIAL_REBASE","_number":31,"created":"2024-01-03 19:20:13.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"real_uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/01/800701/31","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/800701/31","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/31 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/31 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/31 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/800701/31"}}},"commit":{"parents":[{"commit":"67cdbb04d9bfa55f5218b92c8cde431152798298","subject":"s3api: Support GET/HEAD request with ?partNumber","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/67cdbb04d9bfa55f5218b92c8cde431152798298"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2021-12-17 15:39:44.000000000","tz":-360},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-01-03 19:20:13.000000000","tz":0},"subject":"Delete s3api MPU segments when the manifest expires","message":"Delete s3api MPU segments when the manifest expires\n\nThe object-expirer will delete s3api MPU segments automatically when it\nexpires an s3api manifest.\n\nThis is just a best effort cleanup of the segments. In some\ncircumstances (e.g. when manifest objects are not consistent) this may\nfail to delete all segments. However, s3api MPUs already have a related\nbug to address orphaned segments.  The automatic deletion is only\napplied to s3api segments, not to SLO segments in general, because:\n\n(a) s3api segments cannot be referenced via the S3 API, and are\ntherefore orphaned as soon as the manifest is deleted; SLO segments in\ngeneral are not necessarily orphaned when a manifest is deleted.\n\n(b) the names of s3api MPU segments can be easily inferred from the\nmanifest DELETE response headers without inspecting the manifest body.\n\nRelated-Bug: #1813202\nChange-Id: Ib5cec0e552cd51592653e9e100729fcf91f99415\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5fe7db6b865836ee54fd164d8b6babf8c11fd68e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5fe7db6b865836ee54fd164d8b6babf8c11fd68e"}]},"branch":"refs/heads/master","description":"Rebase"},"d7f8e8b98c3ac6d77a7ba91e24b5a46add9c7268":{"kind":"TRIVIAL_REBASE","_number":32,"created":"2024-01-11 14:30:16.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/01/800701/32","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/800701/32","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/32 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/32 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/32 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/800701/32"}}},"commit":{"parents":[{"commit":"212d5d438ac99288b32e68a03e288924f59cb262","subject":"s3api: Support GET/HEAD request with ?partNumber","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/212d5d438ac99288b32e68a03e288924f59cb262"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2021-12-17 15:39:44.000000000","tz":-360},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-01-11 14:30:05.000000000","tz":0},"subject":"Delete s3api MPU segments when the manifest expires","message":"Delete s3api MPU segments when the manifest expires\n\nThe object-expirer will delete s3api MPU segments automatically when it\nexpires an s3api manifest.\n\nThis is just a best effort cleanup of the segments. In some\ncircumstances (e.g. when manifest objects are not consistent) this may\nfail to delete all segments. However, s3api MPUs already have a related\nbug to address orphaned segments.  The automatic deletion is only\napplied to s3api segments, not to SLO segments in general, because:\n\n(a) s3api segments cannot be referenced via the S3 API, and are\ntherefore orphaned as soon as the manifest is deleted; SLO segments in\ngeneral are not necessarily orphaned when a manifest is deleted.\n\n(b) the names of s3api MPU segments can be easily inferred from the\nmanifest DELETE response headers without inspecting the manifest body.\n\nRelated-Bug: #1813202\nChange-Id: Ib5cec0e552cd51592653e9e100729fcf91f99415\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d7f8e8b98c3ac6d77a7ba91e24b5a46add9c7268"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d7f8e8b98c3ac6d77a7ba91e24b5a46add9c7268"}]},"branch":"refs/heads/master"},"9e713648524dd099ee9cfc94ed3a3b4c8e1e5f3b":{"kind":"TRIVIAL_REBASE","_number":33,"created":"2024-03-21 10:48:42.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/01/800701/33","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/800701/33","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/33 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/33 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/33 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/800701/33"}}},"commit":{"parents":[{"commit":"d3938c35471f12bc7f625c2bbbe44e980d30b25a","subject":"Merge \"CI: Move probe tests to centos 9 stream\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d3938c35471f12bc7f625c2bbbe44e980d30b25a"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2021-12-17 15:39:44.000000000","tz":-360},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-03-21 10:46:16.000000000","tz":0},"subject":"Delete s3api MPU segments when the manifest expires","message":"Delete s3api MPU segments when the manifest expires\n\nThe object-expirer will delete s3api MPU segments automatically when it\nexpires an s3api manifest.\n\nThis is just a best effort cleanup of the segments. In some\ncircumstances (e.g. when manifest objects are not consistent) this may\nfail to delete all segments. However, s3api MPUs already have a related\nbug to address orphaned segments.  The automatic deletion is only\napplied to s3api segments, not to SLO segments in general, because:\n\n(a) s3api segments cannot be referenced via the S3 API, and are\ntherefore orphaned as soon as the manifest is deleted; SLO segments in\ngeneral are not necessarily orphaned when a manifest is deleted.\n\n(b) the names of s3api MPU segments can be easily inferred from the\nmanifest DELETE response headers without inspecting the manifest body.\n\nRelated-Bug: #1813202\nChange-Id: Ib5cec0e552cd51592653e9e100729fcf91f99415\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9e713648524dd099ee9cfc94ed3a3b4c8e1e5f3b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9e713648524dd099ee9cfc94ed3a3b4c8e1e5f3b"}]},"branch":"refs/heads/master"},"92be61cb7f653e51998524a59fa385f59d9f9168":{"kind":"REWORK","_number":34,"created":"2024-04-01 14:48:16.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/01/800701/34","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/800701/34","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/34 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/34 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/34 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/800701/34"}}},"commit":{"parents":[{"commit":"d3938c35471f12bc7f625c2bbbe44e980d30b25a","subject":"Merge \"CI: Move probe tests to centos 9 stream\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d3938c35471f12bc7f625c2bbbe44e980d30b25a"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2021-12-17 15:39:44.000000000","tz":-360},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-04-01 13:58:04.000000000","tz":-300},"subject":"Delete s3api MPU segments when expiring the manifest","message":"Delete s3api MPU segments when expiring the manifest\n\ns3api MPU segments will be automatically deleted by the expirer when it\nexpires an s3api manifest.\n\nThis is just a best effort cleanup of the segments. In some\ncircumstances (e.g. when manifest objects are not consistent) this may\nfail to delete all segments. However, s3api MPUs already have a related\nbug to address orphaned segments.\n\nThe automatic deletion is only applied to s3api segments, not to SLO\nsegments in general, because:\n\n(a) s3api segments cannot be referenced via the S3 API, and are\ntherefore orphaned as soon as the manifest is deleted; SLO segments in\ngeneral are not necessarily orphaned when a manifest is deleted.\n\n(b) the names of s3api MPU segments can be easily inferred from the\nmanifest DELETE response headers without inspecting the manifest body.\n\nRelated-Bug: #1813202\nChange-Id: Ib5cec0e552cd51592653e9e100729fcf91f99415\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/92be61cb7f653e51998524a59fa385f59d9f9168"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/92be61cb7f653e51998524a59fa385f59d9f9168"}]},"branch":"refs/heads/master"},"b83741d2d356f7aa6a5c46e35ea6b3a725a5c7db":{"kind":"REWORK","_number":35,"created":"2024-04-09 19:31:47.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/01/800701/35","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/800701/35","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/35 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/35 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/35 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/800701/35"}}},"commit":{"parents":[{"commit":"4107f58699688085fc521b06a476969d834e3d0b","subject":"expirer: randomize task_container iteration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4107f58699688085fc521b06a476969d834e3d0b"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2021-12-17 15:39:44.000000000","tz":-360},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-04-09 19:31:22.000000000","tz":-300},"subject":"Delete s3api MPU segments when expiring the manifest","message":"Delete s3api MPU segments when expiring the manifest\n\ns3api MPU segments will be automatically deleted by the expirer when it\nexpires an s3api manifest.\n\nThis is just a best effort cleanup of the segments. In some\ncircumstances (e.g. when manifest objects are not consistent) this may\nfail to delete all segments. However, s3api MPUs already have a related\nbug to address orphaned segments.\n\nThe automatic deletion is only applied to s3api segments, not to SLO\nsegments in general, because:\n\n(a) s3api segments cannot be referenced via the S3 API, and are\ntherefore orphaned as soon as the manifest is deleted; SLO segments in\ngeneral are not necessarily orphaned when a manifest is deleted.\n\n(b) the names of s3api MPU segments can be easily inferred from the\nmanifest DELETE response headers without inspecting the manifest body.\n\nRelated-Bug: #1813202\nChange-Id: Ib5cec0e552cd51592653e9e100729fcf91f99415\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b83741d2d356f7aa6a5c46e35ea6b3a725a5c7db"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b83741d2d356f7aa6a5c46e35ea6b3a725a5c7db"}]},"branch":"refs/heads/master"},"491ffa8b3438183ac7a69bf690a49bd28eda484e":{"kind":"TRIVIAL_REBASE","_number":36,"created":"2024-05-01 18:42:11.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/01/800701/36","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/800701/36","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/36 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/36 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/36 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/800701/36"}}},"commit":{"parents":[{"commit":"38e331570633496a7c6fdfa3139d5e89407e9d9b","subject":"expirer: new options to control task iteration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/38e331570633496a7c6fdfa3139d5e89407e9d9b"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2021-12-17 15:39:44.000000000","tz":-360},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-05-01 18:41:55.000000000","tz":-300},"subject":"Delete s3api MPU segments when expiring the manifest","message":"Delete s3api MPU segments when expiring the manifest\n\ns3api MPU segments will be automatically deleted by the expirer when it\nexpires an s3api manifest.\n\nThis is just a best effort cleanup of the segments. In some\ncircumstances (e.g. when manifest objects are not consistent) this may\nfail to delete all segments. However, s3api MPUs already have a related\nbug to address orphaned segments.\n\nThe automatic deletion is only applied to s3api segments, not to SLO\nsegments in general, because:\n\n(a) s3api segments cannot be referenced via the S3 API, and are\ntherefore orphaned as soon as the manifest is deleted; SLO segments in\ngeneral are not necessarily orphaned when a manifest is deleted.\n\n(b) the names of s3api MPU segments can be easily inferred from the\nmanifest DELETE response headers without inspecting the manifest body.\n\nRelated-Bug: #1813202\nChange-Id: Ib5cec0e552cd51592653e9e100729fcf91f99415\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/491ffa8b3438183ac7a69bf690a49bd28eda484e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/491ffa8b3438183ac7a69bf690a49bd28eda484e"}]},"branch":"refs/heads/master"},"9f653df27aaa18774ca50426e4f39c8d0f039e67":{"kind":"TRIVIAL_REBASE","_number":37,"created":"2024-05-21 19:23:09.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/01/800701/37","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/800701/37","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/37 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/37 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/37 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/800701/37"}}},"commit":{"parents":[{"commit":"e113e4ed3af20387e498dca84c478378ec2c649c","subject":"DNM expirer: new options to control task iteration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e113e4ed3af20387e498dca84c478378ec2c649c"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2021-12-17 15:39:44.000000000","tz":-360},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-05-21 19:22:56.000000000","tz":-300},"subject":"Delete s3api MPU segments when expiring the manifest","message":"Delete s3api MPU segments when expiring the manifest\n\ns3api MPU segments will be automatically deleted by the expirer when it\nexpires an s3api manifest.\n\nThis is just a best effort cleanup of the segments. In some\ncircumstances (e.g. when manifest objects are not consistent) this may\nfail to delete all segments. However, s3api MPUs already have a related\nbug to address orphaned segments.\n\nThe automatic deletion is only applied to s3api segments, not to SLO\nsegments in general, because:\n\n(a) s3api segments cannot be referenced via the S3 API, and are\ntherefore orphaned as soon as the manifest is deleted; SLO segments in\ngeneral are not necessarily orphaned when a manifest is deleted.\n\n(b) the names of s3api MPU segments can be easily inferred from the\nmanifest DELETE response headers without inspecting the manifest body.\n\nRelated-Bug: #1813202\nChange-Id: Ib5cec0e552cd51592653e9e100729fcf91f99415\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9f653df27aaa18774ca50426e4f39c8d0f039e67"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9f653df27aaa18774ca50426e4f39c8d0f039e67"}]},"branch":"refs/heads/master"},"ab5533ca169fcbf8d4671f6925256e3175670bc7":{"kind":"REWORK","_number":38,"created":"2024-06-03 22:24:09.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/01/800701/38","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/800701/38","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/38 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/38 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/38 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/800701/38"}}},"commit":{"parents":[{"commit":"689ae8e8a1e7223ff20b44afa5c1790187751bfd","subject":"DNM expirer: new options to control task iteration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/689ae8e8a1e7223ff20b44afa5c1790187751bfd"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-06-03 22:21:59.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-06-03 22:22:12.000000000","tz":-300},"subject":"Delete s3api MPU segments when expiring the manifest","message":"Delete s3api MPU segments when expiring the manifest\n\ns3api MPU segments will be automatically deleted by the expirer when it\nexpires an s3api manifest.\n\nThis is just a best effort cleanup of the segments. In some\ncircumstances (e.g. when manifest objects are not consistent) this may\nfail to delete all segments. However, s3api MPUs already have a related\nbug to address orphaned segments.\n\nThe automatic deletion is only applied to s3api segments, not to SLO\nsegments in general, because:\n\n(a) s3api segments cannot be referenced via the S3 API, and are\ntherefore orphaned as soon as the manifest is deleted; SLO segments in\ngeneral are not necessarily orphaned when a manifest is deleted.\n\n(b) the names of s3api MPU segments can be easily inferred from the\nmanifest DELETE response headers without inspecting the manifest body.\n\nDrive-by: also update expirer.embed_expirer_bytes_in_ctype to report the\nslo_size instead of the manifest content-length (since we\u0027ll reap the\nwhole thing)\n\nRelated-Bug: #1813202\nChange-Id: Ib5cec0e552cd51592653e9e100729fcf91f99415\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ab5533ca169fcbf8d4671f6925256e3175670bc7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ab5533ca169fcbf8d4671f6925256e3175670bc7"}]},"branch":"refs/heads/master"},"4b729c66577f9e4d7b6679715e17cdd166b6b33b":{"kind":"REWORK","_number":39,"created":"2024-06-04 18:11:10.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/01/800701/39","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/800701/39","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/39 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/39 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/39 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/800701/39"}}},"commit":{"parents":[{"commit":"cee4450fb53a44f0e7f062c89e19b41f29a94953","subject":"DNM expirer: new options to control task iteration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cee4450fb53a44f0e7f062c89e19b41f29a94953"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-06-03 22:21:59.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-06-04 18:07:16.000000000","tz":-300},"subject":"Delete s3api MPU segments when expiring the manifest","message":"Delete s3api MPU segments when expiring the manifest\n\ns3api MPU segments will be automatically deleted by the expirer when it\nexpires an s3api manifest.\n\nThis is just a best effort cleanup of the segments. In some\ncircumstances (e.g. when manifest objects are not consistent) this may\nfail to delete all segments. However, s3api MPUs already have a related\nbug to address orphaned segments.\n\nThe automatic deletion is only applied to s3api segments, not to SLO\nsegments in general, because:\n\n(a) s3api segments cannot be referenced via the S3 API, and are\ntherefore orphaned as soon as the manifest is deleted; SLO segments in\ngeneral are not necessarily orphaned when a manifest is deleted.\n\n(b) the names of s3api MPU segments can be easily inferred from the\nmanifest DELETE response headers without inspecting the manifest body.\n\nDrive-by: also update expirer.embed_expirer_bytes_in_ctype to report the\nslo_size instead of the manifest content-length (since we\u0027ll reap the\nwhole thing)\n\nRelated-Bug: #1813202\nChange-Id: Ib5cec0e552cd51592653e9e100729fcf91f99415\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4b729c66577f9e4d7b6679715e17cdd166b6b33b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4b729c66577f9e4d7b6679715e17cdd166b6b33b"}]},"branch":"refs/heads/master"},"c12758f39ab4a70d4ac4fa3940f4149794130169":{"kind":"REWORK","_number":40,"created":"2024-06-04 20:39:34.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/01/800701/40","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/800701/40","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/40 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/40 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/40 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/800701/40"}}},"commit":{"parents":[{"commit":"cee4450fb53a44f0e7f062c89e19b41f29a94953","subject":"DNM expirer: new options to control task iteration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cee4450fb53a44f0e7f062c89e19b41f29a94953"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-06-03 22:21:59.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-06-04 20:38:32.000000000","tz":-300},"subject":"Delete s3api MPU segments when expiring the manifest","message":"Delete s3api MPU segments when expiring the manifest\n\ns3api MPU segments will be automatically deleted by the expirer when it\nexpires an s3api manifest.\n\nThis is just a best effort cleanup of the segments. In some\ncircumstances (e.g. when manifest objects are not consistent) this may\nfail to delete all segments. However, s3api MPUs already have a related\nbug to address orphaned segments.\n\nThe automatic deletion is only applied to s3api segments, not to SLO\nsegments in general, because:\n\n(a) s3api segments cannot be referenced via the S3 API, and are\ntherefore orphaned as soon as the manifest is deleted; SLO segments in\ngeneral are not necessarily orphaned when a manifest is deleted.\n\n(b) the names of s3api MPU segments can be easily inferred from the\nmanifest DELETE response headers without inspecting the manifest body.\n\nDrive-by: also update expirer.embed_expirer_bytes_in_ctype to report the\nslo_size instead of the manifest content-length (since we\u0027ll reap the\nwhole thing)\n\nRelated-Bug: #1813202\nChange-Id: Ib5cec0e552cd51592653e9e100729fcf91f99415\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c12758f39ab4a70d4ac4fa3940f4149794130169"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c12758f39ab4a70d4ac4fa3940f4149794130169"}]},"branch":"refs/heads/master"},"e8812eaf70b5ae4a70a75e17c8133fdab9d5ec5c":{"kind":"TRIVIAL_REBASE","_number":41,"created":"2024-06-04 21:05:15.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/01/800701/41","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/800701/41","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/41 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/41 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/41 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/800701/41"}}},"commit":{"parents":[{"commit":"193cffa6863064d8452351e314cbee69bdb9cd3f","subject":"DNM expirer: new options to control task iteration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/193cffa6863064d8452351e314cbee69bdb9cd3f"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-06-03 22:21:59.000000000","tz":-300},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-06-04 21:04:02.000000000","tz":-420},"subject":"Delete s3api MPU segments when expiring the manifest","message":"Delete s3api MPU segments when expiring the manifest\n\ns3api MPU segments will be automatically deleted by the expirer when it\nexpires an s3api manifest.\n\nThis is just a best effort cleanup of the segments. In some\ncircumstances (e.g. when manifest objects are not consistent) this may\nfail to delete all segments. However, s3api MPUs already have a related\nbug to address orphaned segments.\n\nThe automatic deletion is only applied to s3api segments, not to SLO\nsegments in general, because:\n\n(a) s3api segments cannot be referenced via the S3 API, and are\ntherefore orphaned as soon as the manifest is deleted; SLO segments in\ngeneral are not necessarily orphaned when a manifest is deleted.\n\n(b) the names of s3api MPU segments can be easily inferred from the\nmanifest DELETE response headers without inspecting the manifest body.\n\nDrive-by: also update expirer.embed_expirer_bytes_in_ctype to report the\nslo_size instead of the manifest content-length (since we\u0027ll reap the\nwhole thing)\n\nRelated-Bug: #1813202\nChange-Id: Ib5cec0e552cd51592653e9e100729fcf91f99415\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e8812eaf70b5ae4a70a75e17c8133fdab9d5ec5c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e8812eaf70b5ae4a70a75e17c8133fdab9d5ec5c"}]},"branch":"refs/heads/master"},"84a253375b4e48b23774848576308e98456bc938":{"kind":"REWORK","_number":42,"created":"2024-06-13 23:07:27.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/01/800701/42","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/800701/42","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/42 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/42 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/42 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/800701/42"}}},"commit":{"parents":[{"commit":"cdab2f1294fabddb35e3ee71f3e8a2e13197c7f5","subject":"DNM expirer: new options to control task iteration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cdab2f1294fabddb35e3ee71f3e8a2e13197c7f5"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-06-03 22:21:59.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-06-13 22:07:02.000000000","tz":-300},"subject":"Delete s3api MPU segments when expiring the manifest","message":"Delete s3api MPU segments when expiring the manifest\n\ns3api MPU segments will be automatically deleted by the expirer when it\nexpires an s3api manifest.\n\nThis is just a best effort cleanup of the segments. In some\ncircumstances (e.g. when manifest objects are not consistent) this may\nfail to delete all segments. However, s3api MPUs already have a related\nbug to address orphaned segments.\n\nThe automatic deletion is only applied to s3api segments, not to SLO\nsegments in general, because:\n\n(a) s3api segments cannot be referenced via the S3 API, and are\ntherefore orphaned as soon as the manifest is deleted; SLO segments in\ngeneral are not necessarily orphaned when a manifest is deleted.\n\n(b) the names of s3api MPU segments can be easily inferred from the\nmanifest DELETE response headers without inspecting the manifest body.\n\nDrive-by: also update expirer.embed_expirer_bytes_in_ctype to report the\nslo_size instead of the manifest content-length (since we\u0027ll reap the\nwhole thing)\n\nRelated-Bug: #1813202\nChange-Id: Ib5cec0e552cd51592653e9e100729fcf91f99415\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/84a253375b4e48b23774848576308e98456bc938"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/84a253375b4e48b23774848576308e98456bc938"}]},"branch":"refs/heads/master"},"dc9d869ca7324f67afca45f9f26230970e286817":{"kind":"REWORK","_number":43,"created":"2024-06-25 00:34:28.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/01/800701/43","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/800701/43","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/43 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/43 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/43 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/800701/43"}}},"commit":{"parents":[{"commit":"ff08a649e0bc657e13f346e963ae2d40d0e2f001","subject":"Merge \"Add get_statsd_client function\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ff08a649e0bc657e13f346e963ae2d40d0e2f001"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-06-24 19:51:51.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-06-24 19:59:05.000000000","tz":-300},"subject":"Delete s3api MPU segments when expiring the manifest","message":"Delete s3api MPU segments when expiring the manifest\n\ns3api MPU segments will be automatically deleted by the expirer when it\nexpires an s3api manifest.\n\nThis is just a best effort cleanup of the segments. In some\ncircumstances (e.g. when manifest objects are not consistent) this may\nfail to delete all segments. However, s3api MPUs already have a related\nbug to address orphaned segments.\n\nThe automatic deletion is only applied to s3api segments, not to SLO\nsegments in general, because:\n\n(a) s3api segments cannot be referenced via the S3 API, and are\ntherefore orphaned as soon as the manifest is deleted; SLO segments in\ngeneral are not necessarily orphaned when a manifest is deleted.\n\n(b) the names of s3api MPU segments can be easily inferred from the\nmanifest DELETE response headers without inspecting the manifest body.\n\nDrive-by: also update expirer.embed_expirer_bytes_in_ctype to report the\nslo_size instead of the manifest content-length (since we\u0027ll reap the\nwhole thing)\n\nRelated-Bug: #1813202\nChange-Id: Ib5cec0e552cd51592653e9e100729fcf91f99415\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/dc9d869ca7324f67afca45f9f26230970e286817"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/dc9d869ca7324f67afca45f9f26230970e286817"}]},"branch":"refs/heads/master"},"b607f845a857ce1c05284a9740d36121fc94a852":{"kind":"TRIVIAL_REBASE","_number":44,"created":"2024-07-09 23:58:46.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/01/800701/44","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/800701/44","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/44 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/44 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/44 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/800701/44"}}},"commit":{"parents":[{"commit":"0fd36e395dbe05fb77a4e76bc6ea4c83597d86ff","subject":"common: move memcached exceptions to the base file.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0fd36e395dbe05fb77a4e76bc6ea4c83597d86ff"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-06-24 19:51:51.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-07-09 14:12:50.000000000","tz":-300},"subject":"Delete s3api MPU segments when expiring the manifest","message":"Delete s3api MPU segments when expiring the manifest\n\ns3api MPU segments will be automatically deleted by the expirer when it\nexpires an s3api manifest.\n\nThis is just a best effort cleanup of the segments. In some\ncircumstances (e.g. when manifest objects are not consistent) this may\nfail to delete all segments. However, s3api MPUs already have a related\nbug to address orphaned segments.\n\nThe automatic deletion is only applied to s3api segments, not to SLO\nsegments in general, because:\n\n(a) s3api segments cannot be referenced via the S3 API, and are\ntherefore orphaned as soon as the manifest is deleted; SLO segments in\ngeneral are not necessarily orphaned when a manifest is deleted.\n\n(b) the names of s3api MPU segments can be easily inferred from the\nmanifest DELETE response headers without inspecting the manifest body.\n\nDrive-by: also update expirer.embed_expirer_bytes_in_ctype to report the\nslo_size instead of the manifest content-length (since we\u0027ll reap the\nwhole thing)\n\nRelated-Bug: #1813202\nChange-Id: Ib5cec0e552cd51592653e9e100729fcf91f99415\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b607f845a857ce1c05284a9740d36121fc94a852"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b607f845a857ce1c05284a9740d36121fc94a852"}]},"branch":"refs/heads/master"},"8920934c6663bedb721291df9d8478239646e4a7":{"kind":"TRIVIAL_REBASE","_number":45,"created":"2024-07-10 23:01:25.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/01/800701/45","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/800701/45","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/45 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/45 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/45 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/800701/45"}}},"commit":{"parents":[{"commit":"9c38d756e0233eaa5f9e26066756eea9ce6a11ad","subject":"Merge \"Use entry_points for server executables\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9c38d756e0233eaa5f9e26066756eea9ce6a11ad"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-06-24 19:51:51.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-07-10 20:33:54.000000000","tz":-300},"subject":"Delete s3api MPU segments when expiring the manifest","message":"Delete s3api MPU segments when expiring the manifest\n\ns3api MPU segments will be automatically deleted by the expirer when it\nexpires an s3api manifest.\n\nThis is just a best effort cleanup of the segments. In some\ncircumstances (e.g. when manifest objects are not consistent) this may\nfail to delete all segments. However, s3api MPUs already have a related\nbug to address orphaned segments.\n\nThe automatic deletion is only applied to s3api segments, not to SLO\nsegments in general, because:\n\n(a) s3api segments cannot be referenced via the S3 API, and are\ntherefore orphaned as soon as the manifest is deleted; SLO segments in\ngeneral are not necessarily orphaned when a manifest is deleted.\n\n(b) the names of s3api MPU segments can be easily inferred from the\nmanifest DELETE response headers without inspecting the manifest body.\n\nDrive-by: also update expirer.embed_expirer_bytes_in_ctype to report the\nslo_size instead of the manifest content-length (since we\u0027ll reap the\nwhole thing)\n\nRelated-Bug: #1813202\nChange-Id: Ib5cec0e552cd51592653e9e100729fcf91f99415\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8920934c6663bedb721291df9d8478239646e4a7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8920934c6663bedb721291df9d8478239646e4a7"}]},"branch":"refs/heads/master"},"1576630496a8050710d07fc70a99a6d670f378c7":{"kind":"TRIVIAL_REBASE","_number":46,"created":"2024-07-19 13:45:09.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/01/800701/46","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/800701/46","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/46 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/46 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/46 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/800701/46"}}},"commit":{"parents":[{"commit":"7c12870068e283344f750b1a16d0b1f91cc9045c","subject":"Merge \"add object-count quota for accounts in middleware\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7c12870068e283344f750b1a16d0b1f91cc9045c"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-06-24 19:51:51.000000000","tz":-300},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-07-19 12:10:20.000000000","tz":60},"subject":"Delete s3api MPU segments when expiring the manifest","message":"Delete s3api MPU segments when expiring the manifest\n\ns3api MPU segments will be automatically deleted by the expirer when it\nexpires an s3api manifest.\n\nThis is just a best effort cleanup of the segments. In some\ncircumstances (e.g. when manifest objects are not consistent) this may\nfail to delete all segments. However, s3api MPUs already have a related\nbug to address orphaned segments.\n\nThe automatic deletion is only applied to s3api segments, not to SLO\nsegments in general, because:\n\n(a) s3api segments cannot be referenced via the S3 API, and are\ntherefore orphaned as soon as the manifest is deleted; SLO segments in\ngeneral are not necessarily orphaned when a manifest is deleted.\n\n(b) the names of s3api MPU segments can be easily inferred from the\nmanifest DELETE response headers without inspecting the manifest body.\n\nDrive-by: also update expirer.embed_expirer_bytes_in_ctype to report the\nslo_size instead of the manifest content-length (since we\u0027ll reap the\nwhole thing)\n\nRelated-Bug: #1813202\nChange-Id: Ib5cec0e552cd51592653e9e100729fcf91f99415\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1576630496a8050710d07fc70a99a6d670f378c7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1576630496a8050710d07fc70a99a6d670f378c7"}]},"branch":"refs/heads/master"},"9e581d7466f4fb1f58f45d183ae216e86192ffef":{"kind":"REWORK","_number":47,"created":"2024-07-24 10:24:51.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/01/800701/47","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/800701/47","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/47 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/47 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/47 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/800701/47"}}},"commit":{"parents":[{"commit":"ac5bec8889f131c6ba9a763f9cde6fcf8f88c514","subject":"Merge \"AUTHORS: update Zaitcev\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ac5bec8889f131c6ba9a763f9cde6fcf8f88c514"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-06-24 19:51:51.000000000","tz":-300},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-07-24 10:02:17.000000000","tz":60},"subject":"Delete s3api MPU segments when expiring the manifest","message":"Delete s3api MPU segments when expiring the manifest\n\ns3api MPU segments will be automatically deleted by the expirer when it\nexpires an s3api manifest.\n\nThis is just a best effort cleanup of the segments. In some\ncircumstances (e.g. when manifest objects are not consistent) this may\nfail to delete all segments. However, s3api MPUs already have a related\nbug to address orphaned segments.\n\nThe automatic deletion is only applied to s3api segments, not to SLO\nsegments in general, because:\n\n(a) s3api segments cannot be referenced via the S3 API, and are\ntherefore orphaned as soon as the manifest is deleted; SLO segments in\ngeneral are not necessarily orphaned when a manifest is deleted.\n\n(b) the names of s3api MPU segments can be easily inferred from the\nmanifest DELETE response headers without inspecting the manifest body.\n\nDrive-by: also update expirer.embed_expirer_bytes_in_ctype to report the\nslo_size instead of the manifest content-length (since we\u0027ll reap the\nwhole thing)\n\nRelated-Bug: #1813202\nChange-Id: Ib5cec0e552cd51592653e9e100729fcf91f99415\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9e581d7466f4fb1f58f45d183ae216e86192ffef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9e581d7466f4fb1f58f45d183ae216e86192ffef"}]},"branch":"refs/heads/master"},"c3adcb7f1159634b9633fe2c9ef90938d6ac7089":{"kind":"REWORK","_number":48,"created":"2024-08-02 11:36:31.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/01/800701/48","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/800701/48","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/48 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/48 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/48 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/800701/48"}}},"commit":{"parents":[{"commit":"b400a1fdb3ae609d5f86d4d0f1b590640337b8f0","subject":"Expirer: add delete task iteration related metrics","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b400a1fdb3ae609d5f86d4d0f1b590640337b8f0"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-06-24 19:51:51.000000000","tz":-300},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-08-02 11:20:42.000000000","tz":60},"subject":"Delete s3api MPU segments when expiring the manifest","message":"Delete s3api MPU segments when expiring the manifest\n\ns3api MPU segments will be automatically deleted by the expirer when it\nexpires an s3api manifest.\n\nThis is just a best effort cleanup of the segments. In some\ncircumstances (e.g. when manifest objects are not consistent) this may\nfail to delete all segments. However, s3api MPUs already have a related\nbug to address orphaned segments.\n\nThe automatic deletion is only applied to s3api segments, not to SLO\nsegments in general, because:\n\n(a) s3api segments cannot be referenced via the S3 API, and are\ntherefore orphaned as soon as the manifest is deleted; SLO segments in\ngeneral are not necessarily orphaned when a manifest is deleted.\n\n(b) the names of s3api MPU segments can be easily inferred from the\nmanifest DELETE response headers without inspecting the manifest body.\n\nDrive-by: also update expirer.embed_expirer_bytes_in_ctype to report the\nslo_size instead of the manifest content-length (since we\u0027ll reap the\nwhole thing)\n\nRelated-Bug: #1813202\nChange-Id: Ib5cec0e552cd51592653e9e100729fcf91f99415\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c3adcb7f1159634b9633fe2c9ef90938d6ac7089"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c3adcb7f1159634b9633fe2c9ef90938d6ac7089"}]},"branch":"refs/heads/master"},"f062c749ba4bfeb19f4c82d2851a3f230a9c09c3":{"kind":"TRIVIAL_REBASE","_number":49,"created":"2024-08-07 13:26:44.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/01/800701/49","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/800701/49","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/49 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/49 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/49 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/800701/49"}}},"commit":{"parents":[{"commit":"9d14c7c830ba92dfa9bcb59efee367c66001479b","subject":"expirer stats: add test coverage for \u0027errors\u0027 stat","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9d14c7c830ba92dfa9bcb59efee367c66001479b"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-06-24 19:51:51.000000000","tz":-300},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-08-07 13:25:28.000000000","tz":60},"subject":"Delete s3api MPU segments when expiring the manifest","message":"Delete s3api MPU segments when expiring the manifest\n\ns3api MPU segments will be automatically deleted by the expirer when it\nexpires an s3api manifest.\n\nThis is just a best effort cleanup of the segments. In some\ncircumstances (e.g. when manifest objects are not consistent) this may\nfail to delete all segments. However, s3api MPUs already have a related\nbug to address orphaned segments.\n\nThe automatic deletion is only applied to s3api segments, not to SLO\nsegments in general, because:\n\n(a) s3api segments cannot be referenced via the S3 API, and are\ntherefore orphaned as soon as the manifest is deleted; SLO segments in\ngeneral are not necessarily orphaned when a manifest is deleted.\n\n(b) the names of s3api MPU segments can be easily inferred from the\nmanifest DELETE response headers without inspecting the manifest body.\n\nDrive-by: also update expirer.embed_expirer_bytes_in_ctype to report the\nslo_size instead of the manifest content-length (since we\u0027ll reap the\nwhole thing)\n\nRelated-Bug: #1813202\nChange-Id: Ib5cec0e552cd51592653e9e100729fcf91f99415\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f062c749ba4bfeb19f4c82d2851a3f230a9c09c3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f062c749ba4bfeb19f4c82d2851a3f230a9c09c3"}]},"branch":"refs/heads/master"},"e68f0514c226325ac469352addaf785017f1ce1f":{"kind":"TRIVIAL_REBASE","_number":50,"created":"2024-09-05 06:05:13.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/01/800701/50","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/800701/50","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/50 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/50 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/50 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/800701/50"}}},"commit":{"parents":[{"commit":"48e087b1cc7ea564d0f205baf5d625cad34be810","subject":"Merge \"object-expirer: refactor creating InternalClient\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/48e087b1cc7ea564d0f205baf5d625cad34be810"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-06-24 19:51:51.000000000","tz":-300},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2024-09-05 05:53:14.000000000","tz":600},"subject":"Delete s3api MPU segments when expiring the manifest","message":"Delete s3api MPU segments when expiring the manifest\n\ns3api MPU segments will be automatically deleted by the expirer when it\nexpires an s3api manifest.\n\nThis is just a best effort cleanup of the segments. In some\ncircumstances (e.g. when manifest objects are not consistent) this may\nfail to delete all segments. However, s3api MPUs already have a related\nbug to address orphaned segments.\n\nThe automatic deletion is only applied to s3api segments, not to SLO\nsegments in general, because:\n\n(a) s3api segments cannot be referenced via the S3 API, and are\ntherefore orphaned as soon as the manifest is deleted; SLO segments in\ngeneral are not necessarily orphaned when a manifest is deleted.\n\n(b) the names of s3api MPU segments can be easily inferred from the\nmanifest DELETE response headers without inspecting the manifest body.\n\nDrive-by: also update expirer.embed_expirer_bytes_in_ctype to report the\nslo_size instead of the manifest content-length (since we\u0027ll reap the\nwhole thing)\n\nRelated-Bug: #1813202\nChange-Id: Ib5cec0e552cd51592653e9e100729fcf91f99415\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e68f0514c226325ac469352addaf785017f1ce1f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e68f0514c226325ac469352addaf785017f1ce1f"}]},"branch":"refs/heads/master"},"4177fcd150b1253d7517c53ace5e6b155c2d5f4a":{"kind":"TRIVIAL_REBASE","_number":51,"created":"2024-09-11 08:47:37.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/01/800701/51","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/800701/51","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/51 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/51 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/51 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/800701/51"}}},"commit":{"parents":[{"commit":"6f7687b23c43b616b1402898e6657e28c6a49fbe","subject":"Merge \"trivial: Use already-parsed a/c/o\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6f7687b23c43b616b1402898e6657e28c6a49fbe"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-06-24 19:51:51.000000000","tz":-300},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-09-11 08:46:19.000000000","tz":60},"subject":"Delete s3api MPU segments when expiring the manifest","message":"Delete s3api MPU segments when expiring the manifest\n\ns3api MPU segments will be automatically deleted by the expirer when it\nexpires an s3api manifest.\n\nThis is just a best effort cleanup of the segments. In some\ncircumstances (e.g. when manifest objects are not consistent) this may\nfail to delete all segments. However, s3api MPUs already have a related\nbug to address orphaned segments.\n\nThe automatic deletion is only applied to s3api segments, not to SLO\nsegments in general, because:\n\n(a) s3api segments cannot be referenced via the S3 API, and are\ntherefore orphaned as soon as the manifest is deleted; SLO segments in\ngeneral are not necessarily orphaned when a manifest is deleted.\n\n(b) the names of s3api MPU segments can be easily inferred from the\nmanifest DELETE response headers without inspecting the manifest body.\n\nDrive-by: also update expirer.embed_expirer_bytes_in_ctype to report the\nslo_size instead of the manifest content-length (since we\u0027ll reap the\nwhole thing)\n\nRelated-Bug: #1813202\nChange-Id: Ib5cec0e552cd51592653e9e100729fcf91f99415\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4177fcd150b1253d7517c53ace5e6b155c2d5f4a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4177fcd150b1253d7517c53ace5e6b155c2d5f4a"}]},"branch":"refs/heads/master"},"6a0d716725eee958e19f3488f287dae6a30cff10":{"kind":"TRIVIAL_REBASE","_number":52,"created":"2024-09-11 21:40:46.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"real_uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/01/800701/52","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/800701/52","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/52 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/52 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/52 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/800701/52"}}},"commit":{"parents":[{"commit":"ffdf96259814e554dc16ea29b3de6b4e37173158","subject":"object-expirer: fix unused _make_internal_client arg","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ffdf96259814e554dc16ea29b3de6b4e37173158"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-06-24 19:51:51.000000000","tz":-300},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-09-11 21:40:46.000000000","tz":0},"subject":"Delete s3api MPU segments when expiring the manifest","message":"Delete s3api MPU segments when expiring the manifest\n\ns3api MPU segments will be automatically deleted by the expirer when it\nexpires an s3api manifest.\n\nThis is just a best effort cleanup of the segments. In some\ncircumstances (e.g. when manifest objects are not consistent) this may\nfail to delete all segments. However, s3api MPUs already have a related\nbug to address orphaned segments.\n\nThe automatic deletion is only applied to s3api segments, not to SLO\nsegments in general, because:\n\n(a) s3api segments cannot be referenced via the S3 API, and are\ntherefore orphaned as soon as the manifest is deleted; SLO segments in\ngeneral are not necessarily orphaned when a manifest is deleted.\n\n(b) the names of s3api MPU segments can be easily inferred from the\nmanifest DELETE response headers without inspecting the manifest body.\n\nDrive-by: also update expirer.embed_expirer_bytes_in_ctype to report the\nslo_size instead of the manifest content-length (since we\u0027ll reap the\nwhole thing)\n\nRelated-Bug: #1813202\nChange-Id: Ib5cec0e552cd51592653e9e100729fcf91f99415\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6a0d716725eee958e19f3488f287dae6a30cff10"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6a0d716725eee958e19f3488f287dae6a30cff10"}]},"branch":"refs/heads/master","description":"Rebase"},"6f7791352aa5b37a5b3f9cb04af1315a6237004d":{"kind":"TRIVIAL_REBASE","_number":53,"created":"2024-10-02 23:08:47.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"ref":"refs/changes/01/800701/53","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/800701/53","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/53 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/53 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/53 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/800701/53"}}},"commit":{"parents":[{"commit":"b74d022574dbb662caf829e977e3220e0bbb8939","subject":"Object-expirer: continue to process next container on listing errors","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b74d022574dbb662caf829e977e3220e0bbb8939"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-06-24 19:51:51.000000000","tz":-300},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2024-10-02 23:05:09.000000000","tz":-420},"subject":"Delete s3api MPU segments when expiring the manifest","message":"Delete s3api MPU segments when expiring the manifest\n\ns3api MPU segments will be automatically deleted by the expirer when it\nexpires an s3api manifest.\n\nThis is just a best effort cleanup of the segments. In some\ncircumstances (e.g. when manifest objects are not consistent) this may\nfail to delete all segments. However, s3api MPUs already have a related\nbug to address orphaned segments.\n\nThe automatic deletion is only applied to s3api segments, not to SLO\nsegments in general, because:\n\n(a) s3api segments cannot be referenced via the S3 API, and are\ntherefore orphaned as soon as the manifest is deleted; SLO segments in\ngeneral are not necessarily orphaned when a manifest is deleted.\n\n(b) the names of s3api MPU segments can be easily inferred from the\nmanifest DELETE response headers without inspecting the manifest body.\n\nDrive-by: also update expirer.embed_expirer_bytes_in_ctype to report the\nslo_size instead of the manifest content-length (since we\u0027ll reap the\nwhole thing)\n\nRelated-Bug: #1813202\nChange-Id: Ib5cec0e552cd51592653e9e100729fcf91f99415\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6f7791352aa5b37a5b3f9cb04af1315a6237004d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6f7791352aa5b37a5b3f9cb04af1315a6237004d"}]},"branch":"refs/heads/master"},"ceddba6649535df0dc2cabdb9581d25961dc7961":{"kind":"TRIVIAL_REBASE","_number":54,"created":"2024-10-04 21:01:35.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"ref":"refs/changes/01/800701/54","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/800701/54","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/54 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/54 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/54 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/800701/54"}}},"commit":{"parents":[{"commit":"7980b6a0d371c97dd3a381d6cd645d8b19f7dcb7","subject":"Object-expirer: continue to process next container on listing errors","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7980b6a0d371c97dd3a381d6cd645d8b19f7dcb7"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-06-24 19:51:51.000000000","tz":-300},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2024-10-04 20:47:55.000000000","tz":-420},"subject":"Delete s3api MPU segments when expiring the manifest","message":"Delete s3api MPU segments when expiring the manifest\n\ns3api MPU segments will be automatically deleted by the expirer when it\nexpires an s3api manifest.\n\nThis is just a best effort cleanup of the segments. In some\ncircumstances (e.g. when manifest objects are not consistent) this may\nfail to delete all segments. However, s3api MPUs already have a related\nbug to address orphaned segments.\n\nThe automatic deletion is only applied to s3api segments, not to SLO\nsegments in general, because:\n\n(a) s3api segments cannot be referenced via the S3 API, and are\ntherefore orphaned as soon as the manifest is deleted; SLO segments in\ngeneral are not necessarily orphaned when a manifest is deleted.\n\n(b) the names of s3api MPU segments can be easily inferred from the\nmanifest DELETE response headers without inspecting the manifest body.\n\nDrive-by: also update expirer.embed_expirer_bytes_in_ctype to report the\nslo_size instead of the manifest content-length (since we\u0027ll reap the\nwhole thing)\n\nRelated-Bug: #1813202\nChange-Id: Ib5cec0e552cd51592653e9e100729fcf91f99415\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ceddba6649535df0dc2cabdb9581d25961dc7961"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ceddba6649535df0dc2cabdb9581d25961dc7961"}]},"branch":"refs/heads/master"},"96af1ef4c52f2b533ebdc1dbf0df38ecd18f8994":{"kind":"REWORK","_number":55,"created":"2024-10-24 21:11:26.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/01/800701/55","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/800701/55","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/55 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/55 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/55 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/800701/55"}}},"commit":{"parents":[{"commit":"913a6fa0d776f518cb83394e23d3b95c5fb93575","subject":"Parallel distributed task container iteration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/913a6fa0d776f518cb83394e23d3b95c5fb93575"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-06-24 19:51:51.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-10-24 18:06:50.000000000","tz":-300},"subject":"Delete s3api MPU segments when expiring the manifest","message":"Delete s3api MPU segments when expiring the manifest\n\ns3api MPU segments will be automatically deleted by the expirer when it\nexpires an s3api manifest.\n\nThis is just a best effort cleanup of the segments. In some\ncircumstances (e.g. when manifest objects are not consistent) this may\nfail to delete all segments. However, s3api MPUs already have a related\nbug to address orphaned segments.\n\nThe automatic deletion is only applied to s3api segments, not to SLO\nsegments in general, because:\n\n(a) s3api segments cannot be referenced via the S3 API, and are\ntherefore orphaned as soon as the manifest is deleted; SLO segments in\ngeneral are not necessarily orphaned when a manifest is deleted.\n\n(b) the names of s3api MPU segments can be easily inferred from the\nmanifest DELETE response headers without inspecting the manifest body.\n\nDrive-by: also update expirer.embed_expirer_bytes_in_ctype to report the\nslo_size instead of the manifest content-length (since we\u0027ll reap the\nwhole thing)\n\nRelated-Bug: #1813202\nChange-Id: Ib5cec0e552cd51592653e9e100729fcf91f99415\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/96af1ef4c52f2b533ebdc1dbf0df38ecd18f8994"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/96af1ef4c52f2b533ebdc1dbf0df38ecd18f8994"}]},"branch":"refs/heads/master"},"a7241c791aa47eb6f81feadfaca6762b5cae0cd9":{"kind":"TRIVIAL_REBASE","_number":56,"created":"2024-10-28 23:10:06.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/01/800701/56","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/800701/56","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/56 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/56 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/56 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/800701/56"}}},"commit":{"parents":[{"commit":"4e1e39ac506db4769aff102241420ec0fd631272","subject":"Parallel distributed task container iteration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4e1e39ac506db4769aff102241420ec0fd631272"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-06-24 19:51:51.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-10-28 23:09:54.000000000","tz":-300},"subject":"Delete s3api MPU segments when expiring the manifest","message":"Delete s3api MPU segments when expiring the manifest\n\ns3api MPU segments will be automatically deleted by the expirer when it\nexpires an s3api manifest.\n\nThis is just a best effort cleanup of the segments. In some\ncircumstances (e.g. when manifest objects are not consistent) this may\nfail to delete all segments. However, s3api MPUs already have a related\nbug to address orphaned segments.\n\nThe automatic deletion is only applied to s3api segments, not to SLO\nsegments in general, because:\n\n(a) s3api segments cannot be referenced via the S3 API, and are\ntherefore orphaned as soon as the manifest is deleted; SLO segments in\ngeneral are not necessarily orphaned when a manifest is deleted.\n\n(b) the names of s3api MPU segments can be easily inferred from the\nmanifest DELETE response headers without inspecting the manifest body.\n\nDrive-by: also update expirer.embed_expirer_bytes_in_ctype to report the\nslo_size instead of the manifest content-length (since we\u0027ll reap the\nwhole thing)\n\nRelated-Bug: #1813202\nChange-Id: Ib5cec0e552cd51592653e9e100729fcf91f99415\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a7241c791aa47eb6f81feadfaca6762b5cae0cd9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a7241c791aa47eb6f81feadfaca6762b5cae0cd9"}]},"branch":"refs/heads/master"},"b93db0289fe9b841ff95324bcf97ef4fabaa07ff":{"kind":"TRIVIAL_REBASE","_number":57,"created":"2024-10-29 21:34:59.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/01/800701/57","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/800701/57","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/57 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/57 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/57 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/800701/57"}}},"commit":{"parents":[{"commit":"e628af24314a3e57e419b2112c5629f8a37dbdf6","subject":"Parallel distributed task container iteration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e628af24314a3e57e419b2112c5629f8a37dbdf6"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-06-24 19:51:51.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-10-29 21:34:42.000000000","tz":-300},"subject":"Delete s3api MPU segments when expiring the manifest","message":"Delete s3api MPU segments when expiring the manifest\n\ns3api MPU segments will be automatically deleted by the expirer when it\nexpires an s3api manifest.\n\nThis is just a best effort cleanup of the segments. In some\ncircumstances (e.g. when manifest objects are not consistent) this may\nfail to delete all segments. However, s3api MPUs already have a related\nbug to address orphaned segments.\n\nThe automatic deletion is only applied to s3api segments, not to SLO\nsegments in general, because:\n\n(a) s3api segments cannot be referenced via the S3 API, and are\ntherefore orphaned as soon as the manifest is deleted; SLO segments in\ngeneral are not necessarily orphaned when a manifest is deleted.\n\n(b) the names of s3api MPU segments can be easily inferred from the\nmanifest DELETE response headers without inspecting the manifest body.\n\nDrive-by: also update expirer.embed_expirer_bytes_in_ctype to report the\nslo_size instead of the manifest content-length (since we\u0027ll reap the\nwhole thing)\n\nRelated-Bug: #1813202\nChange-Id: Ib5cec0e552cd51592653e9e100729fcf91f99415\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b93db0289fe9b841ff95324bcf97ef4fabaa07ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b93db0289fe9b841ff95324bcf97ef4fabaa07ff"}]},"branch":"refs/heads/master"},"490dadae8f047fdc48daa7e27069ca2b8655573b":{"kind":"TRIVIAL_REBASE","_number":58,"created":"2024-11-01 20:13:43.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/01/800701/58","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/800701/58","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/58 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/58 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/58 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/800701/58"}}},"commit":{"parents":[{"commit":"75fd0ddd1798398d20ee37b6e9c80596b1192e76","subject":"Parallel distributed task container iteration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/75fd0ddd1798398d20ee37b6e9c80596b1192e76"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-06-24 19:51:51.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-11-01 20:13:09.000000000","tz":-300},"subject":"Delete s3api MPU segments when expiring the manifest","message":"Delete s3api MPU segments when expiring the manifest\n\ns3api MPU segments will be automatically deleted by the expirer when it\nexpires an s3api manifest.\n\nThis is just a best effort cleanup of the segments. In some\ncircumstances (e.g. when manifest objects are not consistent) this may\nfail to delete all segments. However, s3api MPUs already have a related\nbug to address orphaned segments.\n\nThe automatic deletion is only applied to s3api segments, not to SLO\nsegments in general, because:\n\n(a) s3api segments cannot be referenced via the S3 API, and are\ntherefore orphaned as soon as the manifest is deleted; SLO segments in\ngeneral are not necessarily orphaned when a manifest is deleted.\n\n(b) the names of s3api MPU segments can be easily inferred from the\nmanifest DELETE response headers without inspecting the manifest body.\n\nDrive-by: also update expirer.embed_expirer_bytes_in_ctype to report the\nslo_size instead of the manifest content-length (since we\u0027ll reap the\nwhole thing)\n\nRelated-Bug: #1813202\nChange-Id: Ib5cec0e552cd51592653e9e100729fcf91f99415\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/490dadae8f047fdc48daa7e27069ca2b8655573b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/490dadae8f047fdc48daa7e27069ca2b8655573b"}]},"branch":"refs/heads/master"},"5fc7dbf0f9d7ee44b3ba74793417c4f240bce9e3":{"kind":"TRIVIAL_REBASE","_number":59,"created":"2024-11-04 05:23:17.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"ref":"refs/changes/01/800701/59","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/800701/59","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/59 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/59 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/59 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/800701/59"}}},"commit":{"parents":[{"commit":"b4a718c597e55236b0bd8ca961af75108cbf2515","subject":"Parallel distributed task container iteration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b4a718c597e55236b0bd8ca961af75108cbf2515"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-06-24 19:51:51.000000000","tz":-300},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2024-11-04 05:22:57.000000000","tz":-480},"subject":"Delete s3api MPU segments when expiring the manifest","message":"Delete s3api MPU segments when expiring the manifest\n\ns3api MPU segments will be automatically deleted by the expirer when it\nexpires an s3api manifest.\n\nThis is just a best effort cleanup of the segments. In some\ncircumstances (e.g. when manifest objects are not consistent) this may\nfail to delete all segments. However, s3api MPUs already have a related\nbug to address orphaned segments.\n\nThe automatic deletion is only applied to s3api segments, not to SLO\nsegments in general, because:\n\n(a) s3api segments cannot be referenced via the S3 API, and are\ntherefore orphaned as soon as the manifest is deleted; SLO segments in\ngeneral are not necessarily orphaned when a manifest is deleted.\n\n(b) the names of s3api MPU segments can be easily inferred from the\nmanifest DELETE response headers without inspecting the manifest body.\n\nDrive-by: also update expirer.embed_expirer_bytes_in_ctype to report the\nslo_size instead of the manifest content-length (since we\u0027ll reap the\nwhole thing)\n\nRelated-Bug: #1813202\nChange-Id: Ib5cec0e552cd51592653e9e100729fcf91f99415\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5fc7dbf0f9d7ee44b3ba74793417c4f240bce9e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5fc7dbf0f9d7ee44b3ba74793417c4f240bce9e3"}]},"branch":"refs/heads/master"},"4fae2fc80bad65bc4ebe4a7cc2736fa05cc59450":{"kind":"REWORK","_number":60,"created":"2024-11-05 00:21:14.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/01/800701/60","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/800701/60","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/60 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/60 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/60 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/800701/60"}}},"commit":{"parents":[{"commit":"4046619a74cb320a4617dce93c20899dd3252eae","subject":"Parallel distributed task container iteration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4046619a74cb320a4617dce93c20899dd3252eae"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-06-24 19:51:51.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-11-05 00:18:06.000000000","tz":-360},"subject":"Delete s3api MPU segments when expiring the manifest","message":"Delete s3api MPU segments when expiring the manifest\n\ns3api MPU segments will be automatically deleted by the expirer when it\nexpires an s3api manifest.\n\nThis is just a best effort cleanup of the segments. In some\ncircumstances (e.g. when manifest objects are not consistent) this may\nfail to delete all segments. However, s3api MPUs already have a related\nbug to address orphaned segments.\n\nThe automatic deletion is only applied to s3api segments, not to SLO\nsegments in general, because:\n\n(a) s3api segments cannot be referenced via the S3 API, and are\ntherefore orphaned as soon as the manifest is deleted; SLO segments in\ngeneral are not necessarily orphaned when a manifest is deleted.\n\n(b) the names of s3api MPU segments can be easily inferred from the\nmanifest DELETE response headers without inspecting the manifest body.\n\nDrive-by: also update expirer.embed_expirer_bytes_in_ctype to report the\nslo_size instead of the manifest content-length (since we\u0027ll reap the\nwhole thing)\n\nRelated-Bug: #1813202\nChange-Id: Ib5cec0e552cd51592653e9e100729fcf91f99415\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4fae2fc80bad65bc4ebe4a7cc2736fa05cc59450"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4fae2fc80bad65bc4ebe4a7cc2736fa05cc59450"}]},"branch":"refs/heads/master"},"d512e22e4b915e2793932604367c0ae33d9ac805":{"kind":"TRIVIAL_REBASE","_number":61,"created":"2024-11-05 22:52:54.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/01/800701/61","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/800701/61","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/61 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/61 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/61 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/800701/61"}}},"commit":{"parents":[{"commit":"d9ff5532075663a3d6684d8bdcdf26df843845a9","subject":"Parallel distributed task container iteration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d9ff5532075663a3d6684d8bdcdf26df843845a9"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-06-24 19:51:51.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-11-05 22:52:27.000000000","tz":-360},"subject":"Delete s3api MPU segments when expiring the manifest","message":"Delete s3api MPU segments when expiring the manifest\n\ns3api MPU segments will be automatically deleted by the expirer when it\nexpires an s3api manifest.\n\nThis is just a best effort cleanup of the segments. In some\ncircumstances (e.g. when manifest objects are not consistent) this may\nfail to delete all segments. However, s3api MPUs already have a related\nbug to address orphaned segments.\n\nThe automatic deletion is only applied to s3api segments, not to SLO\nsegments in general, because:\n\n(a) s3api segments cannot be referenced via the S3 API, and are\ntherefore orphaned as soon as the manifest is deleted; SLO segments in\ngeneral are not necessarily orphaned when a manifest is deleted.\n\n(b) the names of s3api MPU segments can be easily inferred from the\nmanifest DELETE response headers without inspecting the manifest body.\n\nDrive-by: also update expirer.embed_expirer_bytes_in_ctype to report the\nslo_size instead of the manifest content-length (since we\u0027ll reap the\nwhole thing)\n\nRelated-Bug: #1813202\nChange-Id: Ib5cec0e552cd51592653e9e100729fcf91f99415\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d512e22e4b915e2793932604367c0ae33d9ac805"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d512e22e4b915e2793932604367c0ae33d9ac805"}]},"branch":"refs/heads/master"},"4ddf8ecec99b55368ba63e2ddc8b86e4e173f6db":{"kind":"TRIVIAL_REBASE","_number":62,"created":"2024-11-12 17:02:51.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/01/800701/62","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/800701/62","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/62 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/62 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/62 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/800701/62"}}},"commit":{"parents":[{"commit":"6ef360fc6c50791aa4e91a3c760a74d028604c81","subject":"Parallel distributed task container iteration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6ef360fc6c50791aa4e91a3c760a74d028604c81"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-06-24 19:51:51.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-11-12 16:58:24.000000000","tz":-360},"subject":"Delete s3api MPU segments when expiring the manifest","message":"Delete s3api MPU segments when expiring the manifest\n\ns3api MPU segments will be automatically deleted by the expirer when it\nexpires an s3api manifest.\n\nThis is just a best effort cleanup of the segments. In some\ncircumstances (e.g. when manifest objects are not consistent) this may\nfail to delete all segments. However, s3api MPUs already have a related\nbug to address orphaned segments.\n\nThe automatic deletion is only applied to s3api segments, not to SLO\nsegments in general, because:\n\n(a) s3api segments cannot be referenced via the S3 API, and are\ntherefore orphaned as soon as the manifest is deleted; SLO segments in\ngeneral are not necessarily orphaned when a manifest is deleted.\n\n(b) the names of s3api MPU segments can be easily inferred from the\nmanifest DELETE response headers without inspecting the manifest body.\n\nDrive-by: also update expirer.embed_expirer_bytes_in_ctype to report the\nslo_size instead of the manifest content-length (since we\u0027ll reap the\nwhole thing)\n\nRelated-Bug: #1813202\nChange-Id: Ib5cec0e552cd51592653e9e100729fcf91f99415\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4ddf8ecec99b55368ba63e2ddc8b86e4e173f6db"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4ddf8ecec99b55368ba63e2ddc8b86e4e173f6db"}]},"branch":"refs/heads/master"},"6964ca5ff8a727779367e963d285e914c98515a2":{"kind":"TRIVIAL_REBASE","_number":63,"created":"2024-11-15 18:59:30.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"ref":"refs/changes/01/800701/63","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/800701/63","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/63 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/63 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/63 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/800701/63"}}},"commit":{"parents":[{"commit":"71c41e1f2691cc2e245310843f8ba9f840397dda","subject":"Parallel distributed task container iteration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/71c41e1f2691cc2e245310843f8ba9f840397dda"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-06-24 19:51:51.000000000","tz":-300},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2024-11-15 18:58:51.000000000","tz":-480},"subject":"Delete s3api MPU segments when expiring the manifest","message":"Delete s3api MPU segments when expiring the manifest\n\ns3api MPU segments will be automatically deleted by the expirer when it\nexpires an s3api manifest.\n\nThis is just a best effort cleanup of the segments. In some\ncircumstances (e.g. when manifest objects are not consistent) this may\nfail to delete all segments. However, s3api MPUs already have a related\nbug to address orphaned segments.\n\nThe automatic deletion is only applied to s3api segments, not to SLO\nsegments in general, because:\n\n(a) s3api segments cannot be referenced via the S3 API, and are\ntherefore orphaned as soon as the manifest is deleted; SLO segments in\ngeneral are not necessarily orphaned when a manifest is deleted.\n\n(b) the names of s3api MPU segments can be easily inferred from the\nmanifest DELETE response headers without inspecting the manifest body.\n\nDrive-by: also update expirer.embed_expirer_bytes_in_ctype to report the\nslo_size instead of the manifest content-length (since we\u0027ll reap the\nwhole thing)\n\nRelated-Bug: #1813202\nChange-Id: Ib5cec0e552cd51592653e9e100729fcf91f99415\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6964ca5ff8a727779367e963d285e914c98515a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6964ca5ff8a727779367e963d285e914c98515a2"}]},"branch":"refs/heads/master"},"38814414c775320caa5492e02d4b170ff697ded4":{"kind":"TRIVIAL_REBASE","_number":64,"created":"2024-11-18 19:09:46.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/01/800701/64","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/800701/64","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/64 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/64 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/64 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/800701/64"}}},"commit":{"parents":[{"commit":"1d3b2932927a350fe083f7731b304296aca195d2","subject":"Parallel distributed task container iteration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1d3b2932927a350fe083f7731b304296aca195d2"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-06-24 19:51:51.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-11-18 19:09:13.000000000","tz":-360},"subject":"Delete s3api MPU segments when expiring the manifest","message":"Delete s3api MPU segments when expiring the manifest\n\ns3api MPU segments will be automatically deleted by the expirer when it\nexpires an s3api manifest.\n\nThis is just a best effort cleanup of the segments. In some\ncircumstances (e.g. when manifest objects are not consistent) this may\nfail to delete all segments. However, s3api MPUs already have a related\nbug to address orphaned segments.\n\nThe automatic deletion is only applied to s3api segments, not to SLO\nsegments in general, because:\n\n(a) s3api segments cannot be referenced via the S3 API, and are\ntherefore orphaned as soon as the manifest is deleted; SLO segments in\ngeneral are not necessarily orphaned when a manifest is deleted.\n\n(b) the names of s3api MPU segments can be easily inferred from the\nmanifest DELETE response headers without inspecting the manifest body.\n\nDrive-by: also update expirer.embed_expirer_bytes_in_ctype to report the\nslo_size instead of the manifest content-length (since we\u0027ll reap the\nwhole thing)\n\nRelated-Bug: #1813202\nChange-Id: Ib5cec0e552cd51592653e9e100729fcf91f99415\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/38814414c775320caa5492e02d4b170ff697ded4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/38814414c775320caa5492e02d4b170ff697ded4"}]},"branch":"refs/heads/master"},"828eb3bd4fb70db567dfa5d5dff3e319478008cd":{"kind":"REWORK","_number":65,"created":"2024-11-29 14:47:58.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/01/800701/65","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/800701/65","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/65 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/65 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/65 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/800701/65"}}},"commit":{"parents":[{"commit":"366a94491a61e5aac86e23a28790b295c789f5b9","subject":"Parallel distributed task container iteration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/366a94491a61e5aac86e23a28790b295c789f5b9"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-06-24 19:51:51.000000000","tz":-300},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-11-29 14:40:42.000000000","tz":0},"subject":"Delete s3api MPU segments when expiring the manifest","message":"Delete s3api MPU segments when expiring the manifest\n\ns3api MPU segments will be automatically deleted by the expirer when it\nexpires an s3api manifest.\n\nThis is just a best effort cleanup of the segments. In some\ncircumstances (e.g. when manifest objects are not consistent) this may\nfail to delete all segments. However, s3api MPUs already have a related\nbug to address orphaned segments.\n\nThe automatic deletion is only applied to s3api segments, not to SLO\nsegments in general, because:\n\n(a) s3api segments cannot be referenced via the S3 API, and are\ntherefore orphaned as soon as the manifest is deleted; SLO segments in\ngeneral are not necessarily orphaned when a manifest is deleted.\n\n(b) the names of s3api MPU segments can be easily inferred from the\nmanifest DELETE response headers without inspecting the manifest body.\n\nDrive-by: also update expirer.embed_expirer_bytes_in_ctype to report the\nslo_size instead of the manifest content-length (since we\u0027ll reap the\nwhole thing)\n\nRelated-Bug: #1813202\nChange-Id: Ib5cec0e552cd51592653e9e100729fcf91f99415\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/828eb3bd4fb70db567dfa5d5dff3e319478008cd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/828eb3bd4fb70db567dfa5d5dff3e319478008cd"}]},"branch":"refs/heads/master"},"b01d4baa4aeaa67ed0ff89331da9f0207b7849f9":{"kind":"TRIVIAL_REBASE","_number":66,"created":"2024-11-29 15:10:59.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/01/800701/66","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/800701/66","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/66 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/66 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/66 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/800701/66"}}},"commit":{"parents":[{"commit":"89f0e804c6c5785efec7d9940aa892528e058493","subject":"Parallel distributed task container iteration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/89f0e804c6c5785efec7d9940aa892528e058493"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-06-24 19:51:51.000000000","tz":-300},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-11-29 15:10:18.000000000","tz":0},"subject":"Delete s3api MPU segments when expiring the manifest","message":"Delete s3api MPU segments when expiring the manifest\n\ns3api MPU segments will be automatically deleted by the expirer when it\nexpires an s3api manifest.\n\nThis is just a best effort cleanup of the segments. In some\ncircumstances (e.g. when manifest objects are not consistent) this may\nfail to delete all segments. However, s3api MPUs already have a related\nbug to address orphaned segments.\n\nThe automatic deletion is only applied to s3api segments, not to SLO\nsegments in general, because:\n\n(a) s3api segments cannot be referenced via the S3 API, and are\ntherefore orphaned as soon as the manifest is deleted; SLO segments in\ngeneral are not necessarily orphaned when a manifest is deleted.\n\n(b) the names of s3api MPU segments can be easily inferred from the\nmanifest DELETE response headers without inspecting the manifest body.\n\nDrive-by: also update expirer.embed_expirer_bytes_in_ctype to report the\nslo_size instead of the manifest content-length (since we\u0027ll reap the\nwhole thing)\n\nRelated-Bug: #1813202\nChange-Id: Ib5cec0e552cd51592653e9e100729fcf91f99415\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b01d4baa4aeaa67ed0ff89331da9f0207b7849f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b01d4baa4aeaa67ed0ff89331da9f0207b7849f9"}]},"branch":"refs/heads/master"},"103f8a73a64c11ceffaf97284726c7a7ebe50b22":{"kind":"TRIVIAL_REBASE","_number":67,"created":"2025-01-16 12:23:57.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/01/800701/67","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/800701/67","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/67 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/67 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/67 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/800701/67"}}},"commit":{"parents":[{"commit":"865ad47be755e128e051f85c8a35e3386be0392c","subject":"Parallel distributed task container iteration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/865ad47be755e128e051f85c8a35e3386be0392c"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-06-24 19:51:51.000000000","tz":-300},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-01-16 12:16:53.000000000","tz":0},"subject":"Delete s3api MPU segments when expiring the manifest","message":"Delete s3api MPU segments when expiring the manifest\n\ns3api MPU segments will be automatically deleted by the expirer when it\nexpires an s3api manifest.\n\nThis is just a best effort cleanup of the segments. In some\ncircumstances (e.g. when manifest objects are not consistent) this may\nfail to delete all segments. However, s3api MPUs already have a related\nbug to address orphaned segments.\n\nThe automatic deletion is only applied to s3api segments, not to SLO\nsegments in general, because:\n\n(a) s3api segments cannot be referenced via the S3 API, and are\ntherefore orphaned as soon as the manifest is deleted; SLO segments in\ngeneral are not necessarily orphaned when a manifest is deleted.\n\n(b) the names of s3api MPU segments can be easily inferred from the\nmanifest DELETE response headers without inspecting the manifest body.\n\nDrive-by: also update expirer.embed_expirer_bytes_in_ctype to report the\nslo_size instead of the manifest content-length (since we\u0027ll reap the\nwhole thing)\n\nRelated-Bug: #1813202\nChange-Id: Ib5cec0e552cd51592653e9e100729fcf91f99415\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/103f8a73a64c11ceffaf97284726c7a7ebe50b22"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/103f8a73a64c11ceffaf97284726c7a7ebe50b22"}]},"branch":"refs/heads/master"},"49de5c41e561504a9187493f9e38ae6e6ac33412":{"kind":"TRIVIAL_REBASE","_number":68,"created":"2025-01-16 15:11:24.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/01/800701/68","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/800701/68","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/68 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/68 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/68 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/800701/68"}}},"commit":{"parents":[{"commit":"883a65beff6b49caf5b085abb42f8339885f5a6f","subject":"Parallel distributed task container iteration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/883a65beff6b49caf5b085abb42f8339885f5a6f"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-06-24 19:51:51.000000000","tz":-300},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-01-16 15:11:10.000000000","tz":0},"subject":"Delete s3api MPU segments when expiring the manifest","message":"Delete s3api MPU segments when expiring the manifest\n\ns3api MPU segments will be automatically deleted by the expirer when it\nexpires an s3api manifest.\n\nThis is just a best effort cleanup of the segments. In some\ncircumstances (e.g. when manifest objects are not consistent) this may\nfail to delete all segments. However, s3api MPUs already have a related\nbug to address orphaned segments.\n\nThe automatic deletion is only applied to s3api segments, not to SLO\nsegments in general, because:\n\n(a) s3api segments cannot be referenced via the S3 API, and are\ntherefore orphaned as soon as the manifest is deleted; SLO segments in\ngeneral are not necessarily orphaned when a manifest is deleted.\n\n(b) the names of s3api MPU segments can be easily inferred from the\nmanifest DELETE response headers without inspecting the manifest body.\n\nDrive-by: also update expirer.embed_expirer_bytes_in_ctype to report the\nslo_size instead of the manifest content-length (since we\u0027ll reap the\nwhole thing)\n\nRelated-Bug: #1813202\nChange-Id: Ib5cec0e552cd51592653e9e100729fcf91f99415\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/49de5c41e561504a9187493f9e38ae6e6ac33412"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/49de5c41e561504a9187493f9e38ae6e6ac33412"}]},"branch":"refs/heads/master"},"be2342354db35a8bdf51acd9ef90006bc676d304":{"kind":"TRIVIAL_REBASE","_number":69,"created":"2025-02-07 19:43:36.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"ref":"refs/changes/01/800701/69","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/800701/69","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/69 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/69 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/69 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/800701/69"}}},"commit":{"parents":[{"commit":"df249434186b9f173edf8f5aa93d1c46643912fa","subject":"Parallel distributed task container iteration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/df249434186b9f173edf8f5aa93d1c46643912fa"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-06-24 19:51:51.000000000","tz":-300},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2025-02-07 19:43:25.000000000","tz":-480},"subject":"Delete s3api MPU segments when expiring the manifest","message":"Delete s3api MPU segments when expiring the manifest\n\ns3api MPU segments will be automatically deleted by the expirer when it\nexpires an s3api manifest.\n\nThis is just a best effort cleanup of the segments. In some\ncircumstances (e.g. when manifest objects are not consistent) this may\nfail to delete all segments. However, s3api MPUs already have a related\nbug to address orphaned segments.\n\nThe automatic deletion is only applied to s3api segments, not to SLO\nsegments in general, because:\n\n(a) s3api segments cannot be referenced via the S3 API, and are\ntherefore orphaned as soon as the manifest is deleted; SLO segments in\ngeneral are not necessarily orphaned when a manifest is deleted.\n\n(b) the names of s3api MPU segments can be easily inferred from the\nmanifest DELETE response headers without inspecting the manifest body.\n\nDrive-by: also update expirer.embed_expirer_bytes_in_ctype to report the\nslo_size instead of the manifest content-length (since we\u0027ll reap the\nwhole thing)\n\nRelated-Bug: #1813202\nChange-Id: Ib5cec0e552cd51592653e9e100729fcf91f99415\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/be2342354db35a8bdf51acd9ef90006bc676d304"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/be2342354db35a8bdf51acd9ef90006bc676d304"}]},"branch":"refs/heads/master"},"651547ee33a5570df3b5b398d70fc31ec7c233a0":{"kind":"REWORK","_number":70,"created":"2025-02-08 02:19:28.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"ref":"refs/changes/01/800701/70","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/800701/70","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/70 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/70 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/70 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/800701/70"}}},"commit":{"parents":[{"commit":"5b70606cc8d5734af7f88a297921514498a0f982","subject":"Object-server: return 409 if x-delete-at is newer than queue entry","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5b70606cc8d5734af7f88a297921514498a0f982"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-06-24 19:51:51.000000000","tz":-300},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2025-02-08 02:16:53.000000000","tz":-480},"subject":"Delete s3api MPU segments when expiring the manifest","message":"Delete s3api MPU segments when expiring the manifest\n\ns3api MPU segments will be automatically deleted by the expirer when it\nexpires an s3api manifest.\n\nThis is just a best effort cleanup of the segments. In some\ncircumstances (e.g. when manifest objects are not consistent) this may\nfail to delete all segments. However, s3api MPUs already have a related\nbug to address orphaned segments.\n\nThe automatic deletion is only applied to s3api segments, not to SLO\nsegments in general, because:\n\n(a) s3api segments cannot be referenced via the S3 API, and are\ntherefore orphaned as soon as the manifest is deleted; SLO segments in\ngeneral are not necessarily orphaned when a manifest is deleted.\n\n(b) the names of s3api MPU segments can be easily inferred from the\nmanifest DELETE response headers without inspecting the manifest body.\n\nDrive-by: also update expirer.embed_expirer_bytes_in_ctype to report the\nslo_size instead of the manifest content-length (since we\u0027ll reap the\nwhole thing)\n\nRelated-Bug: #1813202\nChange-Id: Ib5cec0e552cd51592653e9e100729fcf91f99415\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/651547ee33a5570df3b5b398d70fc31ec7c233a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/651547ee33a5570df3b5b398d70fc31ec7c233a0"}]},"branch":"refs/heads/master"},"bfc8876d4b54be531a354de45981f0f849d7f588":{"kind":"REWORK","_number":71,"created":"2025-02-10 19:59:11.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"ref":"refs/changes/01/800701/71","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/800701/71","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/71 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/71 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/71 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/800701/71"}}},"commit":{"parents":[{"commit":"5b70606cc8d5734af7f88a297921514498a0f982","subject":"Object-server: return 409 if x-delete-at is newer than queue entry","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5b70606cc8d5734af7f88a297921514498a0f982"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-06-24 19:51:51.000000000","tz":-300},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2025-02-10 19:58:46.000000000","tz":-480},"subject":"Delete s3api MPU segments when expiring the manifest","message":"Delete s3api MPU segments when expiring the manifest\n\ns3api MPU segments will be automatically deleted by the expirer when it\nexpires an s3api manifest.\n\nThis is just a best effort cleanup of the segments. In some\ncircumstances (e.g. when manifest objects are not consistent) this may\nfail to delete all segments. However, s3api MPUs already have a related\nbug to address orphaned segments.\n\nThe automatic deletion is only applied to s3api segments, not to SLO\nsegments in general, because:\n\n(a) s3api segments cannot be referenced via the S3 API, and are\ntherefore orphaned as soon as the manifest is deleted; SLO segments in\ngeneral are not necessarily orphaned when a manifest is deleted.\n\n(b) the names of s3api MPU segments can be easily inferred from the\nmanifest DELETE response headers without inspecting the manifest body.\n\nDrive-by: also update expirer.embed_expirer_bytes_in_ctype to report the\nslo_size instead of the manifest content-length (since we\u0027ll reap the\nwhole thing)\n\nRelated-Bug: #1813202\nChange-Id: Ib5cec0e552cd51592653e9e100729fcf91f99415\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bfc8876d4b54be531a354de45981f0f849d7f588"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bfc8876d4b54be531a354de45981f0f849d7f588"}]},"branch":"refs/heads/master"},"2c07e96f4a2a19eeadb7d978b09772216402c335":{"kind":"TRIVIAL_REBASE","_number":72,"created":"2025-02-18 19:17:44.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"ref":"refs/changes/01/800701/72","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/800701/72","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/72 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/72 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/72 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/800701/72"}}},"commit":{"parents":[{"commit":"995b0990b2ef560b71fe3e13cc358394460f8aa0","subject":"Object-server: return 409 if x-delete-at is newer than queue entry","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/995b0990b2ef560b71fe3e13cc358394460f8aa0"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-06-24 19:51:51.000000000","tz":-300},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2025-02-18 19:17:31.000000000","tz":-480},"subject":"Delete s3api MPU segments when expiring the manifest","message":"Delete s3api MPU segments when expiring the manifest\n\ns3api MPU segments will be automatically deleted by the expirer when it\nexpires an s3api manifest.\n\nThis is just a best effort cleanup of the segments. In some\ncircumstances (e.g. when manifest objects are not consistent) this may\nfail to delete all segments. However, s3api MPUs already have a related\nbug to address orphaned segments.\n\nThe automatic deletion is only applied to s3api segments, not to SLO\nsegments in general, because:\n\n(a) s3api segments cannot be referenced via the S3 API, and are\ntherefore orphaned as soon as the manifest is deleted; SLO segments in\ngeneral are not necessarily orphaned when a manifest is deleted.\n\n(b) the names of s3api MPU segments can be easily inferred from the\nmanifest DELETE response headers without inspecting the manifest body.\n\nDrive-by: also update expirer.embed_expirer_bytes_in_ctype to report the\nslo_size instead of the manifest content-length (since we\u0027ll reap the\nwhole thing)\n\nRelated-Bug: #1813202\nChange-Id: Ib5cec0e552cd51592653e9e100729fcf91f99415\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2c07e96f4a2a19eeadb7d978b09772216402c335"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2c07e96f4a2a19eeadb7d978b09772216402c335"}]},"branch":"refs/heads/master"},"54c88a7392ce57d8d4d8ea6fe4450b9d5bd8ebb4":{"kind":"TRIVIAL_REBASE","_number":73,"created":"2025-04-10 21:43:25.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"real_uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/01/800701/73","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/800701/73","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/73 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/73 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/73 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/800701/73"}}},"commit":{"parents":[{"commit":"0756f43f6eed5b8af3ebada452baed120b37a136","subject":"Object-server: return 409 if x-delete-at is newer than queue entry","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0756f43f6eed5b8af3ebada452baed120b37a136"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-06-24 19:51:51.000000000","tz":-300},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2025-04-10 21:43:25.000000000","tz":0},"subject":"Delete s3api MPU segments when expiring the manifest","message":"Delete s3api MPU segments when expiring the manifest\n\ns3api MPU segments will be automatically deleted by the expirer when it\nexpires an s3api manifest.\n\nThis is just a best effort cleanup of the segments. In some\ncircumstances (e.g. when manifest objects are not consistent) this may\nfail to delete all segments. However, s3api MPUs already have a related\nbug to address orphaned segments.\n\nThe automatic deletion is only applied to s3api segments, not to SLO\nsegments in general, because:\n\n(a) s3api segments cannot be referenced via the S3 API, and are\ntherefore orphaned as soon as the manifest is deleted; SLO segments in\ngeneral are not necessarily orphaned when a manifest is deleted.\n\n(b) the names of s3api MPU segments can be easily inferred from the\nmanifest DELETE response headers without inspecting the manifest body.\n\nDrive-by: also update expirer.embed_expirer_bytes_in_ctype to report the\nslo_size instead of the manifest content-length (since we\u0027ll reap the\nwhole thing)\n\nRelated-Bug: #1813202\nChange-Id: Ib5cec0e552cd51592653e9e100729fcf91f99415\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/54c88a7392ce57d8d4d8ea6fe4450b9d5bd8ebb4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/54c88a7392ce57d8d4d8ea6fe4450b9d5bd8ebb4"}]},"branch":"refs/heads/master","description":"Rebase"},"8f54a5d43a22dd30add9f8016b8f4a8abfc3a005":{"kind":"REWORK","_number":74,"created":"2025-04-10 23:07:01.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/01/800701/74","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/800701/74","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/74 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/74 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/74 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/800701/74"}}},"commit":{"parents":[{"commit":"0756f43f6eed5b8af3ebada452baed120b37a136","subject":"Object-server: return 409 if x-delete-at is newer than queue entry","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0756f43f6eed5b8af3ebada452baed120b37a136"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-06-24 19:51:51.000000000","tz":-300},"committer":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2025-04-10 23:06:48.000000000","tz":-420},"subject":"Delete s3api MPU segments when expiring the manifest","message":"Delete s3api MPU segments when expiring the manifest\n\ns3api MPU segments will be automatically deleted by the expirer when it\nexpires an s3api manifest.\n\nThis is just a best effort cleanup of the segments. In some\ncircumstances (e.g. when manifest objects are not consistent) this may\nfail to delete all segments. However, s3api MPUs already have a related\nbug to address orphaned segments.\n\nThe automatic deletion is only applied to s3api segments, not to SLO\nsegments in general, because:\n\n(a) s3api segments cannot be referenced via the S3 API, and are\ntherefore orphaned as soon as the manifest is deleted; SLO segments in\ngeneral are not necessarily orphaned when a manifest is deleted.\n\n(b) the names of s3api MPU segments can be easily inferred from the\nmanifest DELETE response headers without inspecting the manifest body.\n\nDrive-by: also update expirer.embed_expirer_bytes_in_ctype to report the\nslo_size instead of the manifest content-length (since we\u0027ll reap the\nwhole thing)\n\nRelated-Bug: #1813202\nChange-Id: Ib5cec0e552cd51592653e9e100729fcf91f99415\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8f54a5d43a22dd30add9f8016b8f4a8abfc3a005"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8f54a5d43a22dd30add9f8016b8f4a8abfc3a005"}]},"branch":"refs/heads/master"},"f1d68affbdf56d1d4d0f3f1db606149399c94c38":{"kind":"TRIVIAL_REBASE","_number":75,"created":"2025-04-10 23:19:17.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/01/800701/75","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/800701/75","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/75 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/75 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/75 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/800701/75"}}},"commit":{"parents":[{"commit":"f557fc86ae8b7532d23c6c00e2acef908f057a81","subject":"Object-server: return 409 if x-delete-at is newer than queue entry","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f557fc86ae8b7532d23c6c00e2acef908f057a81"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-06-24 19:51:51.000000000","tz":-300},"committer":{"name":"ASHWIN A NAIR","email":"ashnair@nvidia.com","date":"2025-04-10 23:19:17.000000000","tz":0},"subject":"Delete s3api MPU segments when expiring the manifest","message":"Delete s3api MPU segments when expiring the manifest\n\ns3api MPU segments will be automatically deleted by the expirer when it\nexpires an s3api manifest.\n\nThis is just a best effort cleanup of the segments. In some\ncircumstances (e.g. when manifest objects are not consistent) this may\nfail to delete all segments. However, s3api MPUs already have a related\nbug to address orphaned segments.\n\nThe automatic deletion is only applied to s3api segments, not to SLO\nsegments in general, because:\n\n(a) s3api segments cannot be referenced via the S3 API, and are\ntherefore orphaned as soon as the manifest is deleted; SLO segments in\ngeneral are not necessarily orphaned when a manifest is deleted.\n\n(b) the names of s3api MPU segments can be easily inferred from the\nmanifest DELETE response headers without inspecting the manifest body.\n\nDrive-by: also update expirer.embed_expirer_bytes_in_ctype to report the\nslo_size instead of the manifest content-length (since we\u0027ll reap the\nwhole thing)\n\nRelated-Bug: #1813202\nChange-Id: Ib5cec0e552cd51592653e9e100729fcf91f99415\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f1d68affbdf56d1d4d0f3f1db606149399c94c38"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f1d68affbdf56d1d4d0f3f1db606149399c94c38"}]},"branch":"refs/heads/master","description":"Rebase"},"94014af4e85b732e28421c401ba78adaf1065ef1":{"kind":"TRIVIAL_REBASE","_number":76,"created":"2025-04-11 09:53:56.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/01/800701/76","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/800701/76","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/76 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/76 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/76 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/800701/76"}}},"commit":{"parents":[{"commit":"2b2a77a26caaaa34e7b193a868b4b07ab3edc953","subject":"Object-server: return 409 if x-delete-at is newer than queue entry","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2b2a77a26caaaa34e7b193a868b4b07ab3edc953"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-06-24 19:51:51.000000000","tz":-300},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-04-11 09:23:52.000000000","tz":60},"subject":"Delete s3api MPU segments when expiring the manifest","message":"Delete s3api MPU segments when expiring the manifest\n\ns3api MPU segments will be automatically deleted by the expirer when it\nexpires an s3api manifest.\n\nThis is just a best effort cleanup of the segments. In some\ncircumstances (e.g. when manifest objects are not consistent) this may\nfail to delete all segments. However, s3api MPUs already have a related\nbug to address orphaned segments.\n\nThe automatic deletion is only applied to s3api segments, not to SLO\nsegments in general, because:\n\n(a) s3api segments cannot be referenced via the S3 API, and are\ntherefore orphaned as soon as the manifest is deleted; SLO segments in\ngeneral are not necessarily orphaned when a manifest is deleted.\n\n(b) the names of s3api MPU segments can be easily inferred from the\nmanifest DELETE response headers without inspecting the manifest body.\n\nDrive-by: also update expirer.embed_expirer_bytes_in_ctype to report the\nslo_size instead of the manifest content-length (since we\u0027ll reap the\nwhole thing)\n\nRelated-Bug: #1813202\nChange-Id: Ib5cec0e552cd51592653e9e100729fcf91f99415\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/94014af4e85b732e28421c401ba78adaf1065ef1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/94014af4e85b732e28421c401ba78adaf1065ef1"}]},"branch":"refs/heads/master"},"8471fe341a985dfb1aac025b8b567a5ab454b5de":{"kind":"TRIVIAL_REBASE","_number":77,"created":"2025-05-14 18:38:33.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"real_uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/01/800701/77","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/800701/77","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/77 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/77 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/77 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/800701/77"}}},"commit":{"parents":[{"commit":"4a7f8cf4d955f73b947c0d5c9c5a2b544f833844","subject":"Object-server: return 409 if x-delete-at is newer than queue entry","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4a7f8cf4d955f73b947c0d5c9c5a2b544f833844"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-06-24 19:51:51.000000000","tz":-300},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-05-14 18:38:33.000000000","tz":0},"subject":"Delete s3api MPU segments when expiring the manifest","message":"Delete s3api MPU segments when expiring the manifest\n\ns3api MPU segments will be automatically deleted by the expirer when it\nexpires an s3api manifest.\n\nThis is just a best effort cleanup of the segments. In some\ncircumstances (e.g. when manifest objects are not consistent) this may\nfail to delete all segments. However, s3api MPUs already have a related\nbug to address orphaned segments.\n\nThe automatic deletion is only applied to s3api segments, not to SLO\nsegments in general, because:\n\n(a) s3api segments cannot be referenced via the S3 API, and are\ntherefore orphaned as soon as the manifest is deleted; SLO segments in\ngeneral are not necessarily orphaned when a manifest is deleted.\n\n(b) the names of s3api MPU segments can be easily inferred from the\nmanifest DELETE response headers without inspecting the manifest body.\n\nDrive-by: also update expirer.embed_expirer_bytes_in_ctype to report the\nslo_size instead of the manifest content-length (since we\u0027ll reap the\nwhole thing)\n\nRelated-Bug: #1813202\nChange-Id: Ib5cec0e552cd51592653e9e100729fcf91f99415\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8471fe341a985dfb1aac025b8b567a5ab454b5de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8471fe341a985dfb1aac025b8b567a5ab454b5de"}]},"branch":"refs/heads/master","description":"Rebase"},"2ec0acdcfb0e91da79ceffd502374aa069d77b19":{"kind":"TRIVIAL_REBASE","_number":78,"created":"2025-05-20 01:24:51.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"real_uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/01/800701/78","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/800701/78","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/78 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/78 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/78 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/800701/78"}}},"commit":{"parents":[{"commit":"884c65f80ab1a0a3f9406b8b91e71397ba634346","subject":"Object-server: return 409 if x-delete-at is newer than queue entry","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/884c65f80ab1a0a3f9406b8b91e71397ba634346"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-06-24 19:51:51.000000000","tz":-300},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-05-20 01:24:51.000000000","tz":0},"subject":"Delete s3api MPU segments when expiring the manifest","message":"Delete s3api MPU segments when expiring the manifest\n\ns3api MPU segments will be automatically deleted by the expirer when it\nexpires an s3api manifest.\n\nThis is just a best effort cleanup of the segments. In some\ncircumstances (e.g. when manifest objects are not consistent) this may\nfail to delete all segments. However, s3api MPUs already have a related\nbug to address orphaned segments.\n\nThe automatic deletion is only applied to s3api segments, not to SLO\nsegments in general, because:\n\n(a) s3api segments cannot be referenced via the S3 API, and are\ntherefore orphaned as soon as the manifest is deleted; SLO segments in\ngeneral are not necessarily orphaned when a manifest is deleted.\n\n(b) the names of s3api MPU segments can be easily inferred from the\nmanifest DELETE response headers without inspecting the manifest body.\n\nDrive-by: also update expirer.embed_expirer_bytes_in_ctype to report the\nslo_size instead of the manifest content-length (since we\u0027ll reap the\nwhole thing)\n\nRelated-Bug: #1813202\nChange-Id: Ib5cec0e552cd51592653e9e100729fcf91f99415\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2ec0acdcfb0e91da79ceffd502374aa069d77b19"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2ec0acdcfb0e91da79ceffd502374aa069d77b19"}]},"branch":"refs/heads/master","description":"Rebase"},"397871c9147c8dd2abbd9264ce3c91e270a87861":{"kind":"TRIVIAL_REBASE","_number":79,"created":"2025-05-20 14:27:32.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"real_uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/01/800701/79","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/800701/79","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/79 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/79 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/79 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/800701/79"}}},"commit":{"parents":[{"commit":"094f5b86fd6496f09fe079f14832f89808489224","subject":"Object-server: return 409 if x-delete-at is newer than queue entry","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/094f5b86fd6496f09fe079f14832f89808489224"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-06-24 19:51:51.000000000","tz":-300},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-05-20 14:27:32.000000000","tz":0},"subject":"Delete s3api MPU segments when expiring the manifest","message":"Delete s3api MPU segments when expiring the manifest\n\ns3api MPU segments will be automatically deleted by the expirer when it\nexpires an s3api manifest.\n\nThis is just a best effort cleanup of the segments. In some\ncircumstances (e.g. when manifest objects are not consistent) this may\nfail to delete all segments. However, s3api MPUs already have a related\nbug to address orphaned segments.\n\nThe automatic deletion is only applied to s3api segments, not to SLO\nsegments in general, because:\n\n(a) s3api segments cannot be referenced via the S3 API, and are\ntherefore orphaned as soon as the manifest is deleted; SLO segments in\ngeneral are not necessarily orphaned when a manifest is deleted.\n\n(b) the names of s3api MPU segments can be easily inferred from the\nmanifest DELETE response headers without inspecting the manifest body.\n\nDrive-by: also update expirer.embed_expirer_bytes_in_ctype to report the\nslo_size instead of the manifest content-length (since we\u0027ll reap the\nwhole thing)\n\nRelated-Bug: #1813202\nChange-Id: Ib5cec0e552cd51592653e9e100729fcf91f99415\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/397871c9147c8dd2abbd9264ce3c91e270a87861"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/397871c9147c8dd2abbd9264ce3c91e270a87861"}]},"branch":"refs/heads/master","description":"Rebase"},"3b3bb1376f56fa65080e996f072a945a4138d92b":{"kind":"TRIVIAL_REBASE","_number":80,"created":"2025-05-20 15:56:54.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/01/800701/80","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/800701/80","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/80 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/80 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/80 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/800701/80"}}},"commit":{"parents":[{"commit":"f2e4642120b131e25ee29f5edbe26a02442466dc","subject":"Object-server: return 409 if x-delete-at is newer than queue entry","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f2e4642120b131e25ee29f5edbe26a02442466dc"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-06-24 19:51:51.000000000","tz":-300},"committer":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2025-05-20 15:06:13.000000000","tz":-420},"subject":"Delete s3api MPU segments when expiring the manifest","message":"Delete s3api MPU segments when expiring the manifest\n\ns3api MPU segments will be automatically deleted by the expirer when it\nexpires an s3api manifest.\n\nThis is just a best effort cleanup of the segments. In some\ncircumstances (e.g. when manifest objects are not consistent) this may\nfail to delete all segments. However, s3api MPUs already have a related\nbug to address orphaned segments.\n\nThe automatic deletion is only applied to s3api segments, not to SLO\nsegments in general, because:\n\n(a) s3api segments cannot be referenced via the S3 API, and are\ntherefore orphaned as soon as the manifest is deleted; SLO segments in\ngeneral are not necessarily orphaned when a manifest is deleted.\n\n(b) the names of s3api MPU segments can be easily inferred from the\nmanifest DELETE response headers without inspecting the manifest body.\n\nDrive-by: also update expirer.embed_expirer_bytes_in_ctype to report the\nslo_size instead of the manifest content-length (since we\u0027ll reap the\nwhole thing)\n\nRelated-Bug: #1813202\nChange-Id: Ib5cec0e552cd51592653e9e100729fcf91f99415\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3b3bb1376f56fa65080e996f072a945a4138d92b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3b3bb1376f56fa65080e996f072a945a4138d92b"}]},"branch":"refs/heads/master"},"6b35e38c6c965f9e66cb97cfe73a50a32e6cda52":{"kind":"TRIVIAL_REBASE","_number":81,"created":"2025-05-20 16:47:05.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/01/800701/81","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/800701/81","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/81 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/81 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/81 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/800701/81"}}},"commit":{"parents":[{"commit":"094f5b86fd6496f09fe079f14832f89808489224","subject":"Object-server: return 409 if x-delete-at is newer than queue entry","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/094f5b86fd6496f09fe079f14832f89808489224"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-06-24 19:51:51.000000000","tz":-300},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-05-20 16:46:45.000000000","tz":60},"subject":"Delete s3api MPU segments when expiring the manifest","message":"Delete s3api MPU segments when expiring the manifest\n\ns3api MPU segments will be automatically deleted by the expirer when it\nexpires an s3api manifest.\n\nThis is just a best effort cleanup of the segments. In some\ncircumstances (e.g. when manifest objects are not consistent) this may\nfail to delete all segments. However, s3api MPUs already have a related\nbug to address orphaned segments.\n\nThe automatic deletion is only applied to s3api segments, not to SLO\nsegments in general, because:\n\n(a) s3api segments cannot be referenced via the S3 API, and are\ntherefore orphaned as soon as the manifest is deleted; SLO segments in\ngeneral are not necessarily orphaned when a manifest is deleted.\n\n(b) the names of s3api MPU segments can be easily inferred from the\nmanifest DELETE response headers without inspecting the manifest body.\n\nDrive-by: also update expirer.embed_expirer_bytes_in_ctype to report the\nslo_size instead of the manifest content-length (since we\u0027ll reap the\nwhole thing)\n\nRelated-Bug: #1813202\nChange-Id: Ib5cec0e552cd51592653e9e100729fcf91f99415\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6b35e38c6c965f9e66cb97cfe73a50a32e6cda52"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6b35e38c6c965f9e66cb97cfe73a50a32e6cda52"}]},"branch":"refs/heads/master"},"8a6628acfb13af9ac717620f4357a234882a81c3":{"kind":"NO_CHANGE","_number":82,"created":"2025-05-20 16:48:05.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/01/800701/82","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/800701/82","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/82 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/82 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/82 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/800701/82"}}},"commit":{"parents":[{"commit":"77ba4df1d9442195e4a844b89e90d4c72dd1e9f5","subject":"Object-server: return 409 if x-delete-at is newer than queue entry","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/77ba4df1d9442195e4a844b89e90d4c72dd1e9f5"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-06-24 19:51:51.000000000","tz":-300},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-05-20 16:47:53.000000000","tz":60},"subject":"Delete s3api MPU segments when expiring the manifest","message":"Delete s3api MPU segments when expiring the manifest\n\ns3api MPU segments will be automatically deleted by the expirer when it\nexpires an s3api manifest.\n\nThis is just a best effort cleanup of the segments. In some\ncircumstances (e.g. when manifest objects are not consistent) this may\nfail to delete all segments. However, s3api MPUs already have a related\nbug to address orphaned segments.\n\nThe automatic deletion is only applied to s3api segments, not to SLO\nsegments in general, because:\n\n(a) s3api segments cannot be referenced via the S3 API, and are\ntherefore orphaned as soon as the manifest is deleted; SLO segments in\ngeneral are not necessarily orphaned when a manifest is deleted.\n\n(b) the names of s3api MPU segments can be easily inferred from the\nmanifest DELETE response headers without inspecting the manifest body.\n\nDrive-by: also update expirer.embed_expirer_bytes_in_ctype to report the\nslo_size instead of the manifest content-length (since we\u0027ll reap the\nwhole thing)\n\nRelated-Bug: #1813202\nChange-Id: Ib5cec0e552cd51592653e9e100729fcf91f99415\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8a6628acfb13af9ac717620f4357a234882a81c3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8a6628acfb13af9ac717620f4357a234882a81c3"}]},"branch":"refs/heads/master"},"9ffecb7d1945d2eb8b2d53555d8304fa8937aa1b":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":83,"created":"2025-07-14 18:58:34.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"ref":"refs/changes/01/800701/83","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/800701/83","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/83 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/83 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/83 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/800701/83"}}},"commit":{"parents":[{"commit":"624fd7a139e9d7c85ab435ddf1efae6bb455ebc9","subject":"Object-server: return 409 if x-delete-at is newer than queue entry","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/624fd7a139e9d7c85ab435ddf1efae6bb455ebc9"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-06-24 19:51:51.000000000","tz":-300},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2025-07-14 18:57:47.000000000","tz":-420},"subject":"Delete s3api MPU segments when expiring the manifest","message":"Delete s3api MPU segments when expiring the manifest\n\ns3api MPU segments will be automatically deleted by the expirer when it\nexpires an s3api manifest.\n\nThis is just a best effort cleanup of the segments. In some\ncircumstances (e.g. when manifest objects are not consistent) this may\nfail to delete all segments. However, s3api MPUs already have a related\nbug to address orphaned segments.\n\nThe automatic deletion is only applied to s3api segments, not to SLO\nsegments in general, because:\n\n(a) s3api segments cannot be referenced via the S3 API, and are\ntherefore orphaned as soon as the manifest is deleted; SLO segments in\ngeneral are not necessarily orphaned when a manifest is deleted.\n\n(b) the names of s3api MPU segments can be easily inferred from the\nmanifest DELETE response headers without inspecting the manifest body.\n\nDrive-by: also update expirer.embed_expirer_bytes_in_ctype to report the\nslo_size instead of the manifest content-length (since we\u0027ll reap the\nwhole thing)\n\nRelated-Bug: #1813202\nSigned-off-by: Jianjian Huo \u003cjhuo@nvidia.com\u003e\nChange-Id: Ib5cec0e552cd51592653e9e100729fcf91f99415\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9ffecb7d1945d2eb8b2d53555d8304fa8937aa1b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9ffecb7d1945d2eb8b2d53555d8304fa8937aa1b"}]},"branch":"refs/heads/master"},"797adfa88803127a372d3e47ef7ad5f3e625413d":{"kind":"TRIVIAL_REBASE","_number":84,"created":"2025-08-20 05:28:06.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"ref":"refs/changes/01/800701/84","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/800701/84","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/84 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/84 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/84 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/800701/84"}}},"commit":{"parents":[{"commit":"5ae343b4f0726efa3d273ff522e1262eacdf0e60","subject":"Object-server: return 409 if x-delete-at is newer than queue entry","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5ae343b4f0726efa3d273ff522e1262eacdf0e60"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-06-24 19:51:51.000000000","tz":-300},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2025-08-20 05:18:31.000000000","tz":-420},"subject":"Delete s3api MPU segments when expiring the manifest","message":"Delete s3api MPU segments when expiring the manifest\n\ns3api MPU segments will be automatically deleted by the expirer when it\nexpires an s3api manifest.\n\nThis is just a best effort cleanup of the segments. In some\ncircumstances (e.g. when manifest objects are not consistent) this may\nfail to delete all segments. However, s3api MPUs already have a related\nbug to address orphaned segments.\n\nThe automatic deletion is only applied to s3api segments, not to SLO\nsegments in general, because:\n\n(a) s3api segments cannot be referenced via the S3 API, and are\ntherefore orphaned as soon as the manifest is deleted; SLO segments in\ngeneral are not necessarily orphaned when a manifest is deleted.\n\n(b) the names of s3api MPU segments can be easily inferred from the\nmanifest DELETE response headers without inspecting the manifest body.\n\nDrive-by: also update expirer.embed_expirer_bytes_in_ctype to report the\nslo_size instead of the manifest content-length (since we\u0027ll reap the\nwhole thing)\n\nRelated-Bug: #1813202\nSigned-off-by: Jianjian Huo \u003cjhuo@nvidia.com\u003e\nChange-Id: Ib5cec0e552cd51592653e9e100729fcf91f99415\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/797adfa88803127a372d3e47ef7ad5f3e625413d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/797adfa88803127a372d3e47ef7ad5f3e625413d"}]},"branch":"refs/heads/master"},"71654eb035267cc53220c6b121512e573827f027":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":85,"created":"2025-09-05 19:46:54.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/01/800701/85","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/800701/85","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/85 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/85 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/85 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/800701/85"}}},"commit":{"parents":[{"commit":"8f36683b564abeeb07e4887a203c1654b16ede31","subject":"Object-server: return 409 if x-delete-at is newer than queue entry","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8f36683b564abeeb07e4887a203c1654b16ede31"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-06-24 19:51:51.000000000","tz":-300},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-09-05 19:46:42.000000000","tz":-420},"subject":"Delete s3api MPU segments when expiring the manifest","message":"Delete s3api MPU segments when expiring the manifest\n\ns3api MPU segments will be automatically deleted by the expirer when it\nexpires an s3api manifest.\n\nThis is just a best effort cleanup of the segments. In some\ncircumstances (e.g. when manifest objects are not consistent) this may\nfail to delete all segments. However, s3api MPUs already have a related\nbug to address orphaned segments.\n\nThe automatic deletion is only applied to s3api segments, not to SLO\nsegments in general, because:\n\n(a) s3api segments cannot be referenced via the S3 API, and are\ntherefore orphaned as soon as the manifest is deleted; SLO segments in\ngeneral are not necessarily orphaned when a manifest is deleted.\n\n(b) the names of s3api MPU segments can be easily inferred from the\nmanifest DELETE response headers without inspecting the manifest body.\n\nDrive-by: also update expirer.embed_expirer_bytes_in_ctype to report the\nslo_size instead of the manifest content-length (since we\u0027ll reap the\nwhole thing)\n\nRelated-Bug: #1813202\nSigned-off-by: Jianjian Huo \u003cjhuo@nvidia.com\u003e\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\nChange-Id: Ib5cec0e552cd51592653e9e100729fcf91f99415\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/71654eb035267cc53220c6b121512e573827f027"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/71654eb035267cc53220c6b121512e573827f027"}]},"branch":"refs/heads/master"},"8eaff20843c8365982dee67ad2640dadf5b0be54":{"kind":"TRIVIAL_REBASE","_number":86,"created":"2025-11-10 05:18:45.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"ref":"refs/changes/01/800701/86","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/800701/86","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/86 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/86 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/86 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/800701/86"}}},"commit":{"parents":[{"commit":"d06072563f71e12b91512756282e2b62f38efaa2","subject":"Object-server: return 409 if x-delete-at is newer than queue entry","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d06072563f71e12b91512756282e2b62f38efaa2"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-06-24 19:51:51.000000000","tz":-300},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2025-11-10 05:14:10.000000000","tz":-480},"subject":"Delete s3api MPU segments when expiring the manifest","message":"Delete s3api MPU segments when expiring the manifest\n\ns3api MPU segments will be automatically deleted by the expirer when it\nexpires an s3api manifest.\n\nThis is just a best effort cleanup of the segments. In some\ncircumstances (e.g. when manifest objects are not consistent) this may\nfail to delete all segments. However, s3api MPUs already have a related\nbug to address orphaned segments.\n\nThe automatic deletion is only applied to s3api segments, not to SLO\nsegments in general, because:\n\n(a) s3api segments cannot be referenced via the S3 API, and are\ntherefore orphaned as soon as the manifest is deleted; SLO segments in\ngeneral are not necessarily orphaned when a manifest is deleted.\n\n(b) the names of s3api MPU segments can be easily inferred from the\nmanifest DELETE response headers without inspecting the manifest body.\n\nDrive-by: also update expirer.embed_expirer_bytes_in_ctype to report the\nslo_size instead of the manifest content-length (since we\u0027ll reap the\nwhole thing)\n\nRelated-Bug: #1813202\nSigned-off-by: Jianjian Huo \u003cjhuo@nvidia.com\u003e\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\nChange-Id: Ib5cec0e552cd51592653e9e100729fcf91f99415\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8eaff20843c8365982dee67ad2640dadf5b0be54"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8eaff20843c8365982dee67ad2640dadf5b0be54"}]},"branch":"refs/heads/master"},"6f6b334111b5ad6932c6732abd4a57660d495ace":{"kind":"REWORK","_number":87,"created":"2026-02-10 04:43:10.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"ref":"refs/changes/01/800701/87","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/800701/87","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/87 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/87 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/87 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/800701/87"}}},"commit":{"parents":[{"commit":"3309915f025f460ae09a7810f20bd9f37ae10643","subject":"Object-server: return 409 if x-delete-at is newer than queue entry","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3309915f025f460ae09a7810f20bd9f37ae10643"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-06-24 19:51:51.000000000","tz":-300},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2026-02-10 04:41:55.000000000","tz":-480},"subject":"Delete s3api MPU segments when expiring the manifest","message":"Delete s3api MPU segments when expiring the manifest\n\ns3api MPU segments will be automatically deleted by the expirer when it\nexpires an s3api manifest.\n\nThis is just a best effort cleanup of the segments. In some\ncircumstances (e.g. when manifest objects are not consistent) this may\nfail to delete all segments. However, s3api MPUs already have a related\nbug to address orphaned segments.\n\nThe automatic deletion is only applied to s3api segments, not to SLO\nsegments in general, because:\n\n(a) s3api segments cannot be referenced via the S3 API, and are\ntherefore orphaned as soon as the manifest is deleted; SLO segments in\ngeneral are not necessarily orphaned when a manifest is deleted.\n\n(b) the names of s3api MPU segments can be easily inferred from the\nmanifest DELETE response headers without inspecting the manifest body.\n\nDrive-by: also update expirer.embed_expirer_bytes_in_ctype to report the\nslo_size instead of the manifest content-length (since we\u0027ll reap the\nwhole thing)\n\nRelated-Bug: #1813202\nSigned-off-by: Jianjian Huo \u003cjhuo@nvidia.com\u003e\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\nChange-Id: Ib5cec0e552cd51592653e9e100729fcf91f99415\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6f6b334111b5ad6932c6732abd4a57660d495ace"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6f6b334111b5ad6932c6732abd4a57660d495ace"}]},"branch":"refs/heads/master"},"cd00281c7986eda76c8e16e089c8f5b11b6be0c4":{"kind":"TRIVIAL_REBASE","_number":88,"created":"2026-03-03 05:39:12.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"ref":"refs/changes/01/800701/88","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/800701/88","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/88 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/88 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/88 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/800701/88"}}},"commit":{"parents":[{"commit":"bab67a58aeac54ab652a2fcdbf03f75d00dbd4cd","subject":"Object-server: return 409 if x-delete-at is newer than queue entry","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bab67a58aeac54ab652a2fcdbf03f75d00dbd4cd"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-06-24 19:51:51.000000000","tz":-300},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2026-03-03 05:38:51.000000000","tz":-480},"subject":"Delete s3api MPU segments when expiring the manifest","message":"Delete s3api MPU segments when expiring the manifest\n\ns3api MPU segments will be automatically deleted by the expirer when it\nexpires an s3api manifest.\n\nThis is just a best effort cleanup of the segments. In some\ncircumstances (e.g. when manifest objects are not consistent) this may\nfail to delete all segments. However, s3api MPUs already have a related\nbug to address orphaned segments.\n\nThe automatic deletion is only applied to s3api segments, not to SLO\nsegments in general, because:\n\n(a) s3api segments cannot be referenced via the S3 API, and are\ntherefore orphaned as soon as the manifest is deleted; SLO segments in\ngeneral are not necessarily orphaned when a manifest is deleted.\n\n(b) the names of s3api MPU segments can be easily inferred from the\nmanifest DELETE response headers without inspecting the manifest body.\n\nDrive-by: also update expirer.embed_expirer_bytes_in_ctype to report the\nslo_size instead of the manifest content-length (since we\u0027ll reap the\nwhole thing)\n\nRelated-Bug: #1813202\nSigned-off-by: Jianjian Huo \u003cjhuo@nvidia.com\u003e\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\nChange-Id: Ib5cec0e552cd51592653e9e100729fcf91f99415\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cd00281c7986eda76c8e16e089c8f5b11b6be0c4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cd00281c7986eda76c8e16e089c8f5b11b6be0c4"}]},"branch":"refs/heads/master"},"329daeb18680e142af1c1f5c1e898d0c7d515f6d":{"kind":"REWORK","_number":89,"created":"2026-03-03 16:02:27.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/01/800701/89","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/800701/89","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/89 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/89 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/89 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/800701/89"}}},"commit":{"parents":[{"commit":"842de095b24deb5a207961ae53313187553ca49a","subject":"Object-server: return 409 if x-delete-at is newer than queue entry","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/842de095b24deb5a207961ae53313187553ca49a"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-06-24 19:51:51.000000000","tz":-300},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-03-03 16:02:10.000000000","tz":0},"subject":"Delete s3api MPU segments when expiring the manifest","message":"Delete s3api MPU segments when expiring the manifest\n\ns3api MPU segments will be automatically deleted by the expirer when it\nexpires an s3api manifest.\n\nThis is just a best effort cleanup of the segments. In some\ncircumstances (e.g. when manifest objects are not consistent) this may\nfail to delete all segments. However, s3api MPUs already have a related\nbug to address orphaned segments.\n\nThe automatic deletion is only applied to s3api segments, not to SLO\nsegments in general, because:\n\n(a) s3api segments cannot be referenced via the S3 API, and are\ntherefore orphaned as soon as the manifest is deleted; SLO segments in\ngeneral are not necessarily orphaned when a manifest is deleted.\n\n(b) the names of s3api MPU segments can be easily inferred from the\nmanifest DELETE response headers without inspecting the manifest body.\n\nDrive-by: also update expirer.embed_expirer_bytes_in_ctype to report the\nslo_size instead of the manifest content-length (since we\u0027ll reap the\nwhole thing)\n\nRelated-Bug: #1813202\nSigned-off-by: Jianjian Huo \u003cjhuo@nvidia.com\u003e\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\nChange-Id: Ib5cec0e552cd51592653e9e100729fcf91f99415\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/329daeb18680e142af1c1f5c1e898d0c7d515f6d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/329daeb18680e142af1c1f5c1e898d0c7d515f6d"}]},"branch":"refs/heads/master"},"dafdd80b84c1c0952c6fc1aca7d3b273c625251a":{"kind":"TRIVIAL_REBASE","_number":90,"created":"2026-03-18 13:35:59.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/01/800701/90","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/800701/90","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/90 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/90 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/90 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/800701/90"}}},"commit":{"parents":[{"commit":"fdc9c5b7e2ba6a287d9196699332b6b9f76fd351","subject":"Object-server: return 409 if x-delete-at is newer than queue entry","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fdc9c5b7e2ba6a287d9196699332b6b9f76fd351"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-06-24 19:51:51.000000000","tz":-300},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-03-18 13:35:39.000000000","tz":0},"subject":"Delete s3api MPU segments when expiring the manifest","message":"Delete s3api MPU segments when expiring the manifest\n\ns3api MPU segments will be automatically deleted by the expirer when it\nexpires an s3api manifest.\n\nThis is just a best effort cleanup of the segments. In some\ncircumstances (e.g. when manifest objects are not consistent) this may\nfail to delete all segments. However, s3api MPUs already have a related\nbug to address orphaned segments.\n\nThe automatic deletion is only applied to s3api segments, not to SLO\nsegments in general, because:\n\n(a) s3api segments cannot be referenced via the S3 API, and are\ntherefore orphaned as soon as the manifest is deleted; SLO segments in\ngeneral are not necessarily orphaned when a manifest is deleted.\n\n(b) the names of s3api MPU segments can be easily inferred from the\nmanifest DELETE response headers without inspecting the manifest body.\n\nDrive-by: also update expirer.embed_expirer_bytes_in_ctype to report the\nslo_size instead of the manifest content-length (since we\u0027ll reap the\nwhole thing)\n\nRelated-Bug: #1813202\nSigned-off-by: Jianjian Huo \u003cjhuo@nvidia.com\u003e\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\nChange-Id: Ib5cec0e552cd51592653e9e100729fcf91f99415\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/dafdd80b84c1c0952c6fc1aca7d3b273c625251a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/dafdd80b84c1c0952c6fc1aca7d3b273c625251a"}]},"branch":"refs/heads/master"},"8a1fae991bace45cb201fb095c37dd51bbf71445":{"kind":"TRIVIAL_REBASE","_number":91,"created":"2026-03-18 15:18:47.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/01/800701/91","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/800701/91","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/91 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/91 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/91 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/800701/91"}}},"commit":{"parents":[{"commit":"daaf3c07e2cdf1bf830b7aab77cf3700cf6b6253","subject":"Object-server: return 409 if x-delete-at is newer than queue entry","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/daaf3c07e2cdf1bf830b7aab77cf3700cf6b6253"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-06-24 19:51:51.000000000","tz":-300},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-03-18 15:18:13.000000000","tz":0},"subject":"Delete s3api MPU segments when expiring the manifest","message":"Delete s3api MPU segments when expiring the manifest\n\ns3api MPU segments will be automatically deleted by the expirer when it\nexpires an s3api manifest.\n\nThis is just a best effort cleanup of the segments. In some\ncircumstances (e.g. when manifest objects are not consistent) this may\nfail to delete all segments. However, s3api MPUs already have a related\nbug to address orphaned segments.\n\nThe automatic deletion is only applied to s3api segments, not to SLO\nsegments in general, because:\n\n(a) s3api segments cannot be referenced via the S3 API, and are\ntherefore orphaned as soon as the manifest is deleted; SLO segments in\ngeneral are not necessarily orphaned when a manifest is deleted.\n\n(b) the names of s3api MPU segments can be easily inferred from the\nmanifest DELETE response headers without inspecting the manifest body.\n\nDrive-by: also update expirer.embed_expirer_bytes_in_ctype to report the\nslo_size instead of the manifest content-length (since we\u0027ll reap the\nwhole thing)\n\nRelated-Bug: #1813202\nSigned-off-by: Jianjian Huo \u003cjhuo@nvidia.com\u003e\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\nChange-Id: Ib5cec0e552cd51592653e9e100729fcf91f99415\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8a1fae991bace45cb201fb095c37dd51bbf71445"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8a1fae991bace45cb201fb095c37dd51bbf71445"}]},"branch":"refs/heads/master"},"a9482f7a05f289c5df828db65435e213945ec898":{"kind":"TRIVIAL_REBASE","_number":92,"created":"2026-03-24 17:58:55.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/01/800701/92","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/800701/92","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/92 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/92 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/92 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/800701/92"}}},"commit":{"parents":[{"commit":"ef7fa29a74bfe46758573b591be4d4a0a4af522b","subject":"Object-server: return 409 if x-delete-at is newer than queue entry","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ef7fa29a74bfe46758573b591be4d4a0a4af522b"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-06-24 19:51:51.000000000","tz":-300},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-03-24 17:40:01.000000000","tz":0},"subject":"Delete s3api MPU segments when expiring the manifest","message":"Delete s3api MPU segments when expiring the manifest\n\ns3api MPU segments will be automatically deleted by the expirer when it\nexpires an s3api manifest.\n\nThis is just a best effort cleanup of the segments. In some\ncircumstances (e.g. when manifest objects are not consistent) this may\nfail to delete all segments. However, s3api MPUs already have a related\nbug to address orphaned segments.\n\nThe automatic deletion is only applied to s3api segments, not to SLO\nsegments in general, because:\n\n(a) s3api segments cannot be referenced via the S3 API, and are\ntherefore orphaned as soon as the manifest is deleted; SLO segments in\ngeneral are not necessarily orphaned when a manifest is deleted.\n\n(b) the names of s3api MPU segments can be easily inferred from the\nmanifest DELETE response headers without inspecting the manifest body.\n\nDrive-by: also update expirer.embed_expirer_bytes_in_ctype to report the\nslo_size instead of the manifest content-length (since we\u0027ll reap the\nwhole thing)\n\nRelated-Bug: #1813202\nSigned-off-by: Jianjian Huo \u003cjhuo@nvidia.com\u003e\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\nChange-Id: Ib5cec0e552cd51592653e9e100729fcf91f99415\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a9482f7a05f289c5df828db65435e213945ec898"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a9482f7a05f289c5df828db65435e213945ec898"}]},"branch":"refs/heads/master"},"ef6b5633393c66273a894e1c5b10904b7ba43d09":{"kind":"TRIVIAL_REBASE","_number":93,"created":"2026-03-25 16:30:13.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/01/800701/93","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/800701/93","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/93 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/93 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/93 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/800701/93"}}},"commit":{"parents":[{"commit":"9c8ce3a7d5be8e1d940bb9682341560fb2c90c43","subject":"Object-server: return 409 if x-delete-at is newer than queue entry","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9c8ce3a7d5be8e1d940bb9682341560fb2c90c43"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-06-24 19:51:51.000000000","tz":-300},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-03-25 16:24:46.000000000","tz":0},"subject":"Delete s3api MPU segments when expiring the manifest","message":"Delete s3api MPU segments when expiring the manifest\n\ns3api MPU segments will be automatically deleted by the expirer when it\nexpires an s3api manifest.\n\nThis is just a best effort cleanup of the segments. In some\ncircumstances (e.g. when manifest objects are not consistent) this may\nfail to delete all segments. However, s3api MPUs already have a related\nbug to address orphaned segments.\n\nThe automatic deletion is only applied to s3api segments, not to SLO\nsegments in general, because:\n\n(a) s3api segments cannot be referenced via the S3 API, and are\ntherefore orphaned as soon as the manifest is deleted; SLO segments in\ngeneral are not necessarily orphaned when a manifest is deleted.\n\n(b) the names of s3api MPU segments can be easily inferred from the\nmanifest DELETE response headers without inspecting the manifest body.\n\nDrive-by: also update expirer.embed_expirer_bytes_in_ctype to report the\nslo_size instead of the manifest content-length (since we\u0027ll reap the\nwhole thing)\n\nRelated-Bug: #1813202\nSigned-off-by: Jianjian Huo \u003cjhuo@nvidia.com\u003e\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\nChange-Id: Ib5cec0e552cd51592653e9e100729fcf91f99415\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ef6b5633393c66273a894e1c5b10904b7ba43d09"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ef6b5633393c66273a894e1c5b10904b7ba43d09"}]},"branch":"refs/heads/master"},"085c7afdbc5a17369396fd66ce529bf55a408629":{"kind":"TRIVIAL_REBASE","_number":94,"created":"2026-03-31 14:54:16.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/01/800701/94","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/800701/94","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/94 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/94 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/94 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/800701/94"}}},"commit":{"parents":[{"commit":"7f69047e3382592791e6037a0fdd7980ffa8909b","subject":"Object-server: return 409 if x-delete-at is newer than queue entry","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7f69047e3382592791e6037a0fdd7980ffa8909b"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-06-24 19:51:51.000000000","tz":-300},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-03-31 14:53:55.000000000","tz":60},"subject":"Delete s3api MPU segments when expiring the manifest","message":"Delete s3api MPU segments when expiring the manifest\n\ns3api MPU segments will be automatically deleted by the expirer when it\nexpires an s3api manifest.\n\nThis is just a best effort cleanup of the segments. In some\ncircumstances (e.g. when manifest objects are not consistent) this may\nfail to delete all segments. However, s3api MPUs already have a related\nbug to address orphaned segments.\n\nThe automatic deletion is only applied to s3api segments, not to SLO\nsegments in general, because:\n\n(a) s3api segments cannot be referenced via the S3 API, and are\ntherefore orphaned as soon as the manifest is deleted; SLO segments in\ngeneral are not necessarily orphaned when a manifest is deleted.\n\n(b) the names of s3api MPU segments can be easily inferred from the\nmanifest DELETE response headers without inspecting the manifest body.\n\nDrive-by: also update expirer.embed_expirer_bytes_in_ctype to report the\nslo_size instead of the manifest content-length (since we\u0027ll reap the\nwhole thing)\n\nRelated-Bug: #1813202\nSigned-off-by: Jianjian Huo \u003cjhuo@nvidia.com\u003e\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\nChange-Id: Ib5cec0e552cd51592653e9e100729fcf91f99415\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/085c7afdbc5a17369396fd66ce529bf55a408629"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/085c7afdbc5a17369396fd66ce529bf55a408629"}]},"branch":"refs/heads/master"},"631a4cfb5e91dff80c83048b28b09a47feab4959":{"kind":"TRIVIAL_REBASE","_number":95,"created":"2026-04-09 09:21:56.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/01/800701/95","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/800701/95","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/95 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/95 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/95 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/800701/95"}}},"commit":{"parents":[{"commit":"7b17aad69ee9b200b8e127c26a7790207de41d74","subject":"Object-server: return 409 if x-delete-at is newer than queue entry","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7b17aad69ee9b200b8e127c26a7790207de41d74"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-06-24 19:51:51.000000000","tz":-300},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-04-09 09:16:32.000000000","tz":60},"subject":"Delete s3api MPU segments when expiring the manifest","message":"Delete s3api MPU segments when expiring the manifest\n\ns3api MPU segments will be automatically deleted by the expirer when it\nexpires an s3api manifest.\n\nThis is just a best effort cleanup of the segments. In some\ncircumstances (e.g. when manifest objects are not consistent) this may\nfail to delete all segments. However, s3api MPUs already have a related\nbug to address orphaned segments.\n\nThe automatic deletion is only applied to s3api segments, not to SLO\nsegments in general, because:\n\n(a) s3api segments cannot be referenced via the S3 API, and are\ntherefore orphaned as soon as the manifest is deleted; SLO segments in\ngeneral are not necessarily orphaned when a manifest is deleted.\n\n(b) the names of s3api MPU segments can be easily inferred from the\nmanifest DELETE response headers without inspecting the manifest body.\n\nDrive-by: also update expirer.embed_expirer_bytes_in_ctype to report the\nslo_size instead of the manifest content-length (since we\u0027ll reap the\nwhole thing)\n\nRelated-Bug: #1813202\nSigned-off-by: Jianjian Huo \u003cjhuo@nvidia.com\u003e\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\nChange-Id: Ib5cec0e552cd51592653e9e100729fcf91f99415\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/631a4cfb5e91dff80c83048b28b09a47feab4959"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/631a4cfb5e91dff80c83048b28b09a47feab4959"}]},"branch":"refs/heads/master"},"e9215f3619261452cbf3c4948efa81a84778dc0c":{"kind":"TRIVIAL_REBASE","_number":96,"created":"2026-04-13 09:47:13.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/01/800701/96","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/800701/96","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/96 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/96 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/96 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/800701/96"}}},"commit":{"parents":[{"commit":"c577f964700ad947c0d018f82958c2f27ed38373","subject":"Object-server: return 409 if x-delete-at is newer than queue entry","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c577f964700ad947c0d018f82958c2f27ed38373"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-06-24 19:51:51.000000000","tz":-300},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-04-13 09:46:54.000000000","tz":60},"subject":"Delete s3api MPU segments when expiring the manifest","message":"Delete s3api MPU segments when expiring the manifest\n\ns3api MPU segments will be automatically deleted by the expirer when it\nexpires an s3api manifest.\n\nThis is just a best effort cleanup of the segments. In some\ncircumstances (e.g. when manifest objects are not consistent) this may\nfail to delete all segments. However, s3api MPUs already have a related\nbug to address orphaned segments.\n\nThe automatic deletion is only applied to s3api segments, not to SLO\nsegments in general, because:\n\n(a) s3api segments cannot be referenced via the S3 API, and are\ntherefore orphaned as soon as the manifest is deleted; SLO segments in\ngeneral are not necessarily orphaned when a manifest is deleted.\n\n(b) the names of s3api MPU segments can be easily inferred from the\nmanifest DELETE response headers without inspecting the manifest body.\n\nDrive-by: also update expirer.embed_expirer_bytes_in_ctype to report the\nslo_size instead of the manifest content-length (since we\u0027ll reap the\nwhole thing)\n\nRelated-Bug: #1813202\nSigned-off-by: Jianjian Huo \u003cjhuo@nvidia.com\u003e\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\nChange-Id: Ib5cec0e552cd51592653e9e100729fcf91f99415\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e9215f3619261452cbf3c4948efa81a84778dc0c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e9215f3619261452cbf3c4948efa81a84778dc0c"}]},"branch":"refs/heads/master"},"e62a16ddf71aa0b620b73e65f19758a80c73dfc6":{"kind":"TRIVIAL_REBASE","_number":97,"created":"2026-04-14 15:09:20.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/01/800701/97","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/800701/97","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/97 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/97 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/97 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/800701/97"}}},"commit":{"parents":[{"commit":"6eeda9276892eced269040a9123a4f054d599924","subject":"Object-server: return 409 if x-delete-at is newer than queue entry","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6eeda9276892eced269040a9123a4f054d599924"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-06-24 19:51:51.000000000","tz":-300},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-04-14 15:08:59.000000000","tz":60},"subject":"Delete s3api MPU segments when expiring the manifest","message":"Delete s3api MPU segments when expiring the manifest\n\ns3api MPU segments will be automatically deleted by the expirer when it\nexpires an s3api manifest.\n\nThis is just a best effort cleanup of the segments. In some\ncircumstances (e.g. when manifest objects are not consistent) this may\nfail to delete all segments. However, s3api MPUs already have a related\nbug to address orphaned segments.\n\nThe automatic deletion is only applied to s3api segments, not to SLO\nsegments in general, because:\n\n(a) s3api segments cannot be referenced via the S3 API, and are\ntherefore orphaned as soon as the manifest is deleted; SLO segments in\ngeneral are not necessarily orphaned when a manifest is deleted.\n\n(b) the names of s3api MPU segments can be easily inferred from the\nmanifest DELETE response headers without inspecting the manifest body.\n\nDrive-by: also update expirer.embed_expirer_bytes_in_ctype to report the\nslo_size instead of the manifest content-length (since we\u0027ll reap the\nwhole thing)\n\nRelated-Bug: #1813202\nSigned-off-by: Jianjian Huo \u003cjhuo@nvidia.com\u003e\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\nChange-Id: Ib5cec0e552cd51592653e9e100729fcf91f99415\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e62a16ddf71aa0b620b73e65f19758a80c73dfc6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e62a16ddf71aa0b620b73e65f19758a80c73dfc6"}]},"branch":"refs/heads/master"},"2a1cd82a07ed647d5be38bb96b4f82d20c2f3f82":{"kind":"TRIVIAL_REBASE","_number":98,"created":"2026-05-06 10:50:40.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/01/800701/98","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/800701/98","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/98 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/98 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/98 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/800701/98"}}},"commit":{"parents":[{"commit":"8f53e42068b063c20c43de087fdd671374d86b43","subject":"Object-server: return 409 if x-delete-at is newer than queue entry","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8f53e42068b063c20c43de087fdd671374d86b43"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-06-24 19:51:51.000000000","tz":-300},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-05-06 10:50:07.000000000","tz":60},"subject":"Delete s3api MPU segments when expiring the manifest","message":"Delete s3api MPU segments when expiring the manifest\n\ns3api MPU segments will be automatically deleted by the expirer when it\nexpires an s3api manifest.\n\nThis is just a best effort cleanup of the segments. In some\ncircumstances (e.g. when manifest objects are not consistent) this may\nfail to delete all segments. However, s3api MPUs already have a related\nbug to address orphaned segments.\n\nThe automatic deletion is only applied to s3api segments, not to SLO\nsegments in general, because:\n\n(a) s3api segments cannot be referenced via the S3 API, and are\ntherefore orphaned as soon as the manifest is deleted; SLO segments in\ngeneral are not necessarily orphaned when a manifest is deleted.\n\n(b) the names of s3api MPU segments can be easily inferred from the\nmanifest DELETE response headers without inspecting the manifest body.\n\nDrive-by: also update expirer.embed_expirer_bytes_in_ctype to report the\nslo_size instead of the manifest content-length (since we\u0027ll reap the\nwhole thing)\n\nRelated-Bug: #1813202\nSigned-off-by: Jianjian Huo \u003cjhuo@nvidia.com\u003e\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\nChange-Id: Ib5cec0e552cd51592653e9e100729fcf91f99415\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2a1cd82a07ed647d5be38bb96b4f82d20c2f3f82"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2a1cd82a07ed647d5be38bb96b4f82d20c2f3f82"}]},"branch":"refs/heads/master"},"5b317a7a7cca2f9fd692d93b0438caae6c507a71":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":99,"created":"2026-07-22 23:58:01.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/01/800701/99","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/800701/99","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/99 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/99 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/99 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/800701/99"}}},"commit":{"parents":[{"commit":"50742f41bf2b0444385e1fb03cb9d0e6dbd76023","subject":"Object-server: return 409 if x-delete-at is newer than queue entry","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/50742f41bf2b0444385e1fb03cb9d0e6dbd76023"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-06-24 19:51:51.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2026-07-22 23:37:13.000000000","tz":-300},"subject":"Delete s3api MPU segments when expiring the manifest","message":"Delete s3api MPU segments when expiring the manifest\n\ns3api MPU segments will be automatically deleted by the expirer when it\nexpires an s3api manifest.\n\nThis is just a best effort cleanup of the segments. In some\ncircumstances (e.g. when manifest objects are not consistent) this may\nfail to delete all segments. However, s3api MPUs already have a related\nbug to address orphaned segments.\n\nThe automatic deletion is only applied to s3api segments, not to SLO\nsegments in general, because:\n\n(a) s3api segments cannot be referenced via the S3 API, and are\ntherefore orphaned as soon as the manifest is deleted; SLO segments in\ngeneral are not necessarily orphaned when a manifest is deleted.\n\n(b) the names of s3api MPU segments can be easily inferred from the\nmanifest DELETE response headers without inspecting the manifest body.\n\nDrive-by: also update expirer.embed_expirer_bytes_in_ctype to report the\nslo_size instead of the manifest content-length (since we\u0027ll reap the\nwhole thing)\n\nRelated-Bug: #1813202\nSigned-off-by: Jianjian Huo \u003cjhuo@nvidia.com\u003e\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\nChange-Id: Ib5cec0e552cd51592653e9e100729fcf91f99415\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\nSigned-off-by: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5b317a7a7cca2f9fd692d93b0438caae6c507a71"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5b317a7a7cca2f9fd692d93b0438caae6c507a71"}]},"branch":"refs/heads/master"},"99ba5dcbe37a9b7b051773d4ed1bbe2cb6dc6191":{"kind":"TRIVIAL_REBASE","_number":100,"created":"2026-07-23 10:27:09.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/01/800701/100","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/800701/100","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/100 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/100 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/100 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/800701/100"}}},"commit":{"parents":[{"commit":"a65c31894987ad39b2968658ccd59dc56afe3a54","subject":"Object-server: return 409 if x-delete-at is newer than queue entry","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a65c31894987ad39b2968658ccd59dc56afe3a54"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-06-24 19:51:51.000000000","tz":-300},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-07-23 09:55:23.000000000","tz":60},"subject":"Delete s3api MPU segments when expiring the manifest","message":"Delete s3api MPU segments when expiring the manifest\n\ns3api MPU segments will be automatically deleted by the expirer when it\nexpires an s3api manifest.\n\nThis is just a best effort cleanup of the segments. In some\ncircumstances (e.g. when manifest objects are not consistent) this may\nfail to delete all segments. However, s3api MPUs already have a related\nbug to address orphaned segments.\n\nThe automatic deletion is only applied to s3api segments, not to SLO\nsegments in general, because:\n\n(a) s3api segments cannot be referenced via the S3 API, and are\ntherefore orphaned as soon as the manifest is deleted; SLO segments in\ngeneral are not necessarily orphaned when a manifest is deleted.\n\n(b) the names of s3api MPU segments can be easily inferred from the\nmanifest DELETE response headers without inspecting the manifest body.\n\nDrive-by: also update expirer.embed_expirer_bytes_in_ctype to report the\nslo_size instead of the manifest content-length (since we\u0027ll reap the\nwhole thing)\n\nRelated-Bug: #1813202\nSigned-off-by: Jianjian Huo \u003cjhuo@nvidia.com\u003e\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\nChange-Id: Ib5cec0e552cd51592653e9e100729fcf91f99415\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\nSigned-off-by: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/99ba5dcbe37a9b7b051773d4ed1bbe2cb6dc6191"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/99ba5dcbe37a9b7b051773d4ed1bbe2cb6dc6191"}]},"branch":"refs/heads/master"},"46fa4894da0c489d746a11072c3fbc7b7ca95a7b":{"kind":"TRIVIAL_REBASE","_number":101,"created":"2026-09-11 20:37:09.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/01/800701/101","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/800701/101","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/101 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/101 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/800701/101 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/800701/101"}}},"commit":{"parents":[{"commit":"c799e3e851875a9de282a1b29b2f46a39eebdf5b","subject":"Object-server: return 409 if x-delete-at is newer than queue entry","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c799e3e851875a9de282a1b29b2f46a39eebdf5b"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-06-24 19:51:51.000000000","tz":-300},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-09-11 20:35:37.000000000","tz":-420},"subject":"Delete s3api MPU segments when expiring the manifest","message":"Delete s3api MPU segments when expiring the manifest\n\ns3api MPU segments will be automatically deleted by the expirer when it\nexpires an s3api manifest.\n\nThis is just a best effort cleanup of the segments. In some\ncircumstances (e.g. when manifest objects are not consistent) this may\nfail to delete all segments. However, s3api MPUs already have a related\nbug to address orphaned segments.\n\nThe automatic deletion is only applied to s3api segments, not to SLO\nsegments in general, because:\n\n(a) s3api segments cannot be referenced via the S3 API, and are\ntherefore orphaned as soon as the manifest is deleted; SLO segments in\ngeneral are not necessarily orphaned when a manifest is deleted.\n\n(b) the names of s3api MPU segments can be easily inferred from the\nmanifest DELETE response headers without inspecting the manifest body.\n\nDrive-by: also update expirer.embed_expirer_bytes_in_ctype to report the\nslo_size instead of the manifest content-length (since we\u0027ll reap the\nwhole thing)\n\nRelated-Bug: #1813202\nSigned-off-by: Jianjian Huo \u003cjhuo@nvidia.com\u003e\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\nChange-Id: Ib5cec0e552cd51592653e9e100729fcf91f99415\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\nSigned-off-by: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/46fa4894da0c489d746a11072c3fbc7b7ca95a7b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/46fa4894da0c489d746a11072c3fbc7b7ca95a7b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
