)]}'
{"id":"openstack%2Fswift~800748","triplet_id":"openstack%2Fswift~master~Ib1d6197caa15d0974334be3d7eff24d4342ac378","project":"openstack/swift","branch":"master","topic":"fix-update-shard-spi","hashtags":[],"change_id":"Ib1d6197caa15d0974334be3d7eff24d4342ac378","subject":"sharder: update shard storage_policy_index if roots changes","status":"NEW","created":"2021-07-14 06:28:29.000000000","updated":"2024-05-09 17:47:08.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":59,"unresolved_comment_count":29,"has_review_started":true,"meta_rev_id":"349154ee54851c988ea25e844cd2c5a41eef697f","_number":800748,"virtual_id_number":800748,"owner":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:check","value":1,"date":"2024-05-09 17:47:08.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-14 07:15:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-07-14 07:36:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-14 08:50:27.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"},{"updated":"2021-07-29 21:36:19.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"}],"messages":[{"id":"12711658aafe92f09fbd0dff9813ea5f9b34e044","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2021-07-14 06:28:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"18d2456c4e031b89fbb9174d1b410c88a72ae1bf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-14 07:15:11.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2ee60db88ffa4b8993e4cbe96def53b8 : SUCCESS in 40m 02s (non-voting)\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/de1d25b17642473a908387a464d1189e : SUCCESS in 44m 20s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c72e157fd508e26d132cbfd382b5d9a05d71b9fb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-14 07:36:09.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf83a24e3f444a298733a8d96fbb118d : SUCCESS in 11m 22s\n- grenade https://zuul.opendev.org/t/openstack/build/95fd4c975b2943c39ab564ee29d6fd05 : SUCCESS in 59m 36s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/420b339fcaf24db6af9d6f586b3f5de1 : SUCCESS in 45m 41s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b223f47d9dcb49a8aaae7f1f73353051 : SUCCESS in 9m 19s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/871093096ae140d7847baacf46965fd8 : SUCCESS in 9m 54s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/c648318a5f054569a4ebc30047355b4f : SUCCESS in 17m 22s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/255b10b3adb84a77a4ccc83266e4a6a7 : SUCCESS in 17m 44s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/625f05f6cc824742a7a55468f5441fa2 : SUCCESS in 20m 07s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/ff12b352f35245249eb1f404be154d17 : SUCCESS in 18m 34s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/6f614ec672f64c5085a1f06682dbd616 : SUCCESS in 17m 29s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/5db45321ec3a4526950d03b517de1153 : SUCCESS in 18m 13s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/02c2d47635f94daeaf71b71f7e636c07 : SUCCESS in 23m 14s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/ff8e245f7e77416cad186fb4e5dc1d1c : SUCCESS in 23m 57s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/130536bca94342fc9789eece1a0e69dd : SUCCESS in 17m 24s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/d82ff05941d34992a0fa78e71f41ca9a : SUCCESS in 18m 15s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/2fbc9ea829044b8fa554779eea4f3c0a : SUCCESS in 19m 26s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f532cf93005640d0b93727ced3efcf43 : SUCCESS in 8m 09s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/2c2d3150532843d4a0b38b46a2ebe73f : SUCCESS in 21m 41s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/0e8289022f4a46909afbc275cc8568e3 : SUCCESS in 53m 21s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/fc03944992cf4418966a81dc2f31f8ad : SUCCESS in 51m 24s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4e8606815e6f44f2bed2aa315746832d : SUCCESS in 45m 04s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/9a254d10bd194280880bd30537679e7b : SUCCESS in 40m 07s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/97e9df377e0745fc97f558e1fe6b93cb : SUCCESS in 18m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/701b691d4d0e4eb1b5a781f2a1667b0a : FAILURE in 7m 30s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b91b709450f4409e87eedb810d20856e : SUCCESS in 31m 21s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8a431a6ce39e40099890db562a5abf51 : SUCCESS in 51m 06s\n\nWarning:\n  Comments left for invalid file bin/swift-container-migrate","accounts_in_message":[],"_revision_number":1},{"id":"6dbdc63e6be3ada928d8c5db262528b0eeab127a","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-07-14 08:50:27.000000000","message":"Patch Set 1: Code-Review-1\n\n(3 comments)\n\ngood start - we\u0027re going to need tests obviously 😊\n\nI\u0027m curious, did you manage to play with a sharded container and observe the bad behaviour? It does seem that there has to be a bug. In fact, do we have a launchpad bug for this?\n\nI\u0027d vote for pulling the method from the parent patch and basing this patch on master - this is a bug fix, the parent is a new feature, so this patch should be the priority. \n\nThere\u0027s still a window of time when the shard has not yet learnt about the policy change: two issues then:\n1. objects may be pouring into it but presumably in the new policy (i.e. the wrong policy w.r.t. the shard policy index) and the replicator generates reconciler work that isn\u0027t actually needed\n2. objects already in the shard do not get migrated to the new policy\n\nI wonder if we need to pull some of the audit code up into a replicator method which updates shard state, and have the replicator *and* sharder both call that method when they visit shards, so that the shard policy would be up to date before the replicator generates reconciler work?","accounts_in_message":[],"_revision_number":1},{"id":"2b803f2c6e128537763509deb0316867d2c23a08","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2021-07-19 01:58:56.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"e403638083867775bf1e125bafa161e286509f82","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2021-07-19 07:00:27.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ed2d2bb98ed6e7d3d5944c707cdf4ceed6fb0fb8","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2021-07-19 07:01:54.000000000","message":"Patch Set 2:\n\nAddressing some of Al\u0027s comments. Still need to add tests.","accounts_in_message":[],"_revision_number":2},{"id":"6181db593b5899d9b2495a8cafad24fb1c8fd7c2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-19 08:05:25.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e6836428ff247329fb1a54983a79702 : SUCCESS in 9m 12s\n- grenade https://zuul.opendev.org/t/openstack/build/6ae7d8da05574e5db5bc02e98e347b71 : SUCCESS in 1h 03m 09s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/07789641f9d7439b9c870a33c646c599 : SUCCESS in 42m 03s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d625a647dc0b4920a9b2dc15eab80e9f : SUCCESS in 8m 15s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/ae6600ca282342cc86466f823476db0c : SUCCESS in 9m 36s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/19cba08ae9324348809e6e57b65619cd : SUCCESS in 16m 53s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/48a9a19207e240af9695a12fecbe6522 : SUCCESS in 14m 44s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/01684b1fd9dc4e1b9ebc5da80bf8cf26 : SUCCESS in 28m 35s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/50acb2692bc04848b7e486215d47161e : SUCCESS in 17m 24s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/6822f46edcc24afd923eb3bc935e30bf : SUCCESS in 17m 31s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/3f63697a3e674508a9c013dbefb69729 : SUCCESS in 16m 54s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/c88ecdc7c79740098a03bcb0539d0619 : SUCCESS in 17m 45s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/b49a0a1675464af095f0e4437c64af10 : SUCCESS in 20m 51s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/4d810efd631e410aa2a8a4ed3e7aa5b4 : SUCCESS in 16m 32s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/71734d40ecf5431bac2177ade22f76ec : SUCCESS in 19m 28s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/441f7c13c6e8448b84905ce29d449744 : SUCCESS in 19m 40s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f4f34cf6f9534a4fb02ee6b9620ef2b2 : SUCCESS in 7m 14s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/1984947e0ee9470492d3459832787f30 : SUCCESS in 21m 55s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/88b047b61f7543eca0c880d24e41c146 : SUCCESS in 52m 59s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/ffb69c3a2b8d40c5a969a035a16d0e71 : SUCCESS in 47m 55s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/3ae1b0ce684442538a161a2ef21f3770 : SUCCESS in 42m 49s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/40b7abf00d4e4e2288135cecca705ed2 : SUCCESS in 41m 42s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/254a85da26e443d4816c40dda8caeb6c : SUCCESS in 17m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dc5ecfdc4d124944abc48715afe8996b : SUCCESS in 5m 38s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ee5e9f73ebf3498394a2673cbef28b80 : SUCCESS in 30m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d604446d0b764ce6bea3ec915e746ba0 : SUCCESS in 42m 53s","accounts_in_message":[],"_revision_number":2},{"id":"c7744036b9a8542f0e5e33db4f3b038917e8bc36","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-19 09:15:49.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7a06b661f8c141b990d0664880ea74e9 : SUCCESS in 1h 09m 32s (non-voting)\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/dfdf7fb4d48f466184a230d93e1b7842 : TIMED_OUT in 1h 31m 14s (non-voting)\n- swift-probetests-centos-8-arm64 https://zuul.opendev.org/t/openstack/build/9b06f3820e7c43f59313f9ee488cde03 : SUCCESS in 1h 49m 27s (non-voting)\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/bdc23316ba3743f5b1cef2af63def9b1 : SUCCESS in 1h 10m 47s (non-voting)\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/cfdf0196202c4cb38f9660f535208a7e : SUCCESS in 1h 29m 44s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"fe98c2609a9dff36b93e2d6c08cefd08934725e3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2021-07-21 07:08:58.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"f472079cbfb68dd69c01da6da44c51c8b938c1fc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-21 08:19:20.000000000","message":"Patch Set 3: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8cf6d3e1c7cf4805993221acddb1db71 : SUCCESS in 8m 43s\n- grenade https://zuul.opendev.org/t/openstack/build/234ea7e56c094856b70a2e8dacfdd674 : SUCCESS in 1h 04m 48s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d86127667dc74234a14d42dfdbde7500 : SUCCESS in 47m 37s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/73e828b3dc5d46d59a50a8c891dfa389 : SUCCESS in 8m 03s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/63dcb061812448b287b7d80d83237a97 : SUCCESS in 8m 25s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/348580e03e3d49078f612478b954c1e3 : SUCCESS in 17m 27s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/cea9a21f5c3241eeb155ce8c27fc05ed : SUCCESS in 17m 01s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/c0169636b40c4301bd141c63b6ed9c37 : SUCCESS in 16m 35s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/7216ed2767cf4531b0ca73152415b05e : SUCCESS in 16m 48s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/e3a63a7dd08b4104a1502e91f2463e46 : SUCCESS in 17m 36s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/53a8df19efb944b8aa71674f65c27dc0 : SUCCESS in 18m 14s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/a0af034cac9d4728811bbfb43e1f7924 : SUCCESS in 17m 23s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/2b1b755aaa4e4b9f983fb81369d9f04a : SUCCESS in 19m 41s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/69fb31650e0440ce87ef2f14a3427074 : SUCCESS in 18m 00s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/254f73ced5064dde8db7152008814861 : SUCCESS in 19m 33s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/da0c5cadc69a46bcb711b94b032b8d89 : SUCCESS in 18m 53s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/3eb05d23999d47f6a2287a2c89479d87 : SUCCESS in 7m 55s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/3c3b307372e8490282db97620b6ed4e8 : SUCCESS in 20m 11s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/d5aec169d3cf4f44b7fd9335010fb104 : SUCCESS in 49m 42s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/52b5db53add2438eb3df95e79866a63e : SUCCESS in 52m 27s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/12f0e571560d406cb86b3225f69ad2c9 : SUCCESS in 41m 16s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/270d081d39a145a58c7001afb87e8b54 : SUCCESS in 36m 16s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/19afc73dc0544315a2ff9cebd2cc2fef : SUCCESS in 17m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/553f7341df20417eb772be18cf905fb7 : FAILURE in 5m 36s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/75aa49aa04ef473786460a0a070c6341 : FAILURE in 25m 50s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/548ce2c4fd54482ab2eb2ef738b99f69 : FAILURE in 6m 25s","accounts_in_message":[],"_revision_number":3},{"id":"69b008b8f92e2830b9b5a3bffbe0f641fa9a4827","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-21 08:40:45.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9406522198374762b23926500e0b6f99 : SUCCESS in 30m 22s (non-voting)\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7dcc6f63642746ad84217aafa5e9e833 : SUCCESS in 29m 48s (non-voting)\n- swift-probetests-centos-8-arm64 https://zuul.opendev.org/t/openstack/build/ca893b40d6c44c2d8a7a2ff0d48d84ad : SUCCESS in 1h 29m 38s (non-voting)\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/552d28ac3e2d4c9ebf106a6e4a27ab97 : SUCCESS in 39m 54s (non-voting)\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/fb1505c606304c4a995e2e5a31ff57d5 : SUCCESS in 37m 42s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"bb4e6b2061f51a5cf0ccfc61a3b43d17dfeb09d8","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-07-22 11:08:55.000000000","message":"Patch Set 3:\n\n(4 comments)\n\nWe need to avoid the shard policy ping-ponging on audit cycles (and when shard replicates). So shard\u0027s should not always update their policy of the policy from root is different, only if the policy from root is more up to date than the last policy from root.\n\nAlso, I think the shard\u0027s put timestamp should remain independent of the root\u0027s, otherwise shard-to-shard policy replication could cause ping-ponging.\n\nDoes the strategy need to be more like replication sync of policy index? i.e.\nhttps://github.com/openstack/swift/blob/bc5f4c061122aee5965cfb055beb3cb079073536/swift/container/replicator.py#L359-L364\n\nBut, it\u0027s a little different because, when sync\u0027ing shard with root, we can\u0027t compare the root and shard put timestamps to determine which has the most up to date policy. So we need a way to capture \u0027policy-timestamp\u0027 independently of put timestamp (unless status_changed_at can do that, but like put timestamp I would prefer not to couple the root and shard status_changed_at times).\n\nI\u0027m thinking along the lines of:\n\nshard creation:\n - shard is given root policy X and root put timestamp t0\n\naudit cycle 1:\n - shard learns that root has policy Z at put timestamp t1\n - shard has policy X at put timestamp t10\n - shard compares new root policy info with stashed root put timestamp and sees that policy Z at t1 supersedes policy X at stashed root put timestamp t0\n - shard updates its policy to Z and updates its put timestamp to t11\n - shard remembers that root was Z at root put timestamp t1\n\nreplication cycle (existing behaviour IIUC):\n - shard peers learn that shard has policy Z at put timestamp t11\n - shard peers update policy from X at t10 to Z\n\naudit cycle 2:\n - shard learns that root has policy X at put timestamp t0\n - shard has policy Z at put timestamp t10 + a + b\n - shard compares new root policy info with stashed root put timestamp and sees that policy X at t0 is older than stashed policy Z at t1\n - shard makes no changes to its policy: root will eventually become consistent with policy Z at t1\n\nNote:\n1. the shard needs to remember what root info timestamp it based its policy on, as \"policy timestamp\" meta\n2. shard updates its policy only when new root info supersedes stashed root info policy timestamp \n3. the shard put timestamp remains independent from the root put timestamp.\n4. after shard updates its policy, shard replication between peers should propagate the latest policy because the shard put timestamp always advances when policy changes","accounts_in_message":[],"_revision_number":3},{"id":"e8186b811739a67eb688773cd8b451b15c34057f","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2021-07-23 03:07:21.000000000","message":"Patch Set 3:\n\n(3 comments)\n\nWill just push up a new version with a probe test. then want to go back and take a closer look at the potential timestamp issues.","accounts_in_message":[],"_revision_number":3},{"id":"dd1171e685705d8de2e6b07150e876a0f13ca192","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2021-07-23 03:09:55.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"a3e2dcf2e70e590902425641edc6d47b0d819cff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-23 04:13:25.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/891e001ccd704f95b9837198cb6035d3 : SUCCESS in 8m 47s\n- grenade https://zuul.opendev.org/t/openstack/build/80cb3900d05c4e30a62c2e440e946c0e : SUCCESS in 1h 02m 17s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/56d13fa5f2bc47a9af6b437bae2ce819 : SUCCESS in 40m 06s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b00d7b2495ec4c8290ba5549e49ebdb3 : SUCCESS in 8m 26s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/92ae2d98d33e4fe0907119822a2b56c9 : SUCCESS in 8m 40s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/b62fa71ddd8f4bcd9177ccd07dda688b : SUCCESS in 20m 50s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/6be7d3a5d0364b6c89c5263b89f07538 : SUCCESS in 23m 06s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/ffc499045d43490fb7bcaea81602f2cf : SUCCESS in 20m 22s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/c65d6d4cd99141c1ba59c307dc1cb6e0 : SUCCESS in 20m 42s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/f872b3dffd59482980953e83e621ece7 : SUCCESS in 18m 26s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/7b7401c34af54d3bb5a08771d92467cb : SUCCESS in 20m 58s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/96b052d1fa534b24939d27f54d572e59 : SUCCESS in 18m 35s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/68af45e455e44b10b8c583b146d529b0 : SUCCESS in 21m 45s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/5be6019359ab4170964c3340d985c764 : SUCCESS in 19m 29s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/a8a1dc80acec4d6f9c4819305951723f : SUCCESS in 18m 48s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/fec974bb2e0c4b7486868c3cf6a187d4 : SUCCESS in 20m 57s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/2d54049ec22e48b0b678355db44fda16 : SUCCESS in 7m 02s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/93f531d111574bcca00aa56b90683418 : SUCCESS in 24m 32s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/33435c9be08445c1a001eb0aa026ecc8 : FAILURE in 52m 45s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/2283e0b0007040cf91e4c62e28b82457 : FAILURE in 51m 16s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/67535fdce5e0469092ce6ef3f2af596b : SUCCESS in 39m 31s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/64dd9096148943ccbc6677696bf5546d : SUCCESS in 40m 29s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2b23940b19c249dcb5d024efc4947cc1 : SUCCESS in 18m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/151aa2a9621e4159bcfaebbabb2d15f8 : SUCCESS in 6m 14s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/dc745eb603ff419a8ee55fcc00fa540c : SUCCESS in 31m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/aa50c3305d1540c298df48946c8d3611 : SUCCESS in 46m 46s","accounts_in_message":[],"_revision_number":4},{"id":"6cebbf023ad1767f58b7cd5f1548eec01d56ea6e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-23 04:59:49.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7e92416cda464e12a5f2a9fd29ded835 : SUCCESS in 48m 00s (non-voting)\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/923132001b604d5abe1bfb23f5aa3cbf : SUCCESS in 38m 14s (non-voting)\n- swift-probetests-centos-8-arm64 https://zuul.opendev.org/t/openstack/build/eec5cf7760354a4bab8490e7c6c9281a : FAILURE in 1h 47m 39s (non-voting)\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/5dbe838ca9434c729a5eac2524a0bd25 : SUCCESS in 51m 47s (non-voting)\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/321dc1bbaed34b1cba35c9eb9b2de011 : FAILURE in 40m 25s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"dfb5ab095dda954f1a72a92d58939e379f643f3f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2021-07-23 05:00:57.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"3c14d395c376b14b7ff1008ab44e301ce079c4f8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-23 05:59:17.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cea04f7198b542abb454b5e4dfbc1a4c : SUCCESS in 9m 43s\n- grenade https://zuul.opendev.org/t/openstack/build/062ae5849db64b019b96da8d5360f66a : SUCCESS in 57m 06s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/5a3fc5d6bdca4016b018172c32f4f5a7 : SUCCESS in 40m 59s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ec3702d98d9a45bc878f570a113a0f1b : SUCCESS in 7m 49s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/dd76f3eec5f14d6781be39ce2c59e829 : SUCCESS in 8m 25s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/07b7a24a6a2649559aaafb81c4ad95cf : SUCCESS in 14m 44s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/5ab701715fba4be3aeca143352055852 : SUCCESS in 20m 19s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/fbd80e1d978c467196f463e6738b2111 : SUCCESS in 17m 09s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/c8c0ebb061c741c98619321d19ea2af7 : SUCCESS in 20m 36s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/330afb9d4edf4acb85424755b5e0cec0 : SUCCESS in 16m 07s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/eafd67d590504df7a62f5c04720789a2 : SUCCESS in 21m 07s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/bd185ac1ebe9436e90081afe30ed4a73 : SUCCESS in 21m 34s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/e9e232fb19a14159bef92fe94156cb0b : SUCCESS in 20m 47s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/0b301a63699e4c11828472ab16bb3f7d : SUCCESS in 20m 51s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/6c25daf7e8ed44949c03c1d6a72aec23 : SUCCESS in 20m 53s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/68fcd216d0d44cb89109b0b4268dea11 : SUCCESS in 21m 42s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/3e846026a7cd4e8b805e71611c11bda3 : SUCCESS in 6m 13s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e65e6b0068e24d6c9f8ba30ddba23bcf : SUCCESS in 25m 00s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/f561c8620c38421386ebd405916be7cf : SUCCESS in 48m 13s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/6b22c6b1ff614cfb9e71e3ba1320ab8c : SUCCESS in 50m 44s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a9496ef671cf4e90b7c355522243b33f : SUCCESS in 39m 22s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/065ab0fd4b8044f99beeb07c2fa15277 : SUCCESS in 41m 12s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4e76ebf83c784844a82b5b1c575ba1f6 : SUCCESS in 15m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0f38707654584a2882228221b1574fc6 : SUCCESS in 6m 30s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/913a3f09e2844b9fbde2df7362de3b05 : SUCCESS in 21m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/487723ffef68403fad4b87f8c4d921af : SUCCESS in 43m 05s","accounts_in_message":[],"_revision_number":5},{"id":"f7d3a42a928ad970f9adc42b6be4f15a07ff532c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-23 06:38:36.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e22733803b844dcf92ac10e2e7ba65d7 : SUCCESS in 31m 37s (non-voting)\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ddac5b821bd744c6a1eed37424936057 : SUCCESS in 42m 16s (non-voting)\n- swift-probetests-centos-8-arm64 https://zuul.opendev.org/t/openstack/build/9b0bf7e4c9534096b6975fc78180ab31 : SUCCESS in 1h 25m 52s (non-voting)\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/ce6084c16fdf48ceb1baf998b3019080 : SUCCESS in 32m 25s (non-voting)\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/078178bd4b36415ca15bcdaa958bc0af : SUCCESS in 31m 38s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"7fc2f12007203c9736618aead7fee877683265b4","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2021-07-23 06:54:26.000000000","message":"Patch Set 5:\n\n\u003e Patch Set 3:\n\u003e \n\u003e (4 comments)\n\u003e \n\u003e We need to avoid the shard policy ping-ponging on audit cycles (and when shard replicates). So shard\u0027s should not always update their policy of the policy from root is different, only if the policy from root is more up to date than the last policy from root.\n\u003e \n\u003e Also, I think the shard\u0027s put timestamp should remain independent of the root\u0027s, otherwise shard-to-shard policy replication could cause ping-ponging.\n\u003e \n\u003e Does the strategy need to be more like replication sync of policy index? i.e.\n\u003e https://github.com/openstack/swift/blob/bc5f4c061122aee5965cfb055beb3cb079073536/swift/container/replicator.py#L359-L364\n\u003e \n\u003e But, it\u0027s a little different because, when sync\u0027ing shard with root, we can\u0027t compare the root and shard put timestamps to determine which has the most up to date policy. So we need a way to capture \u0027policy-timestamp\u0027 independently of put timestamp (unless status_changed_at can do that, but like put timestamp I would prefer not to couple the root and shard status_changed_at times).\n\u003e \n\u003e I\u0027m thinking along the lines of:\n\u003e \n\u003e shard creation:\n\u003e  - shard is given root policy X and root put timestamp t0\n\u003e \n\u003e audit cycle 1:\n\u003e  - shard learns that root has policy Z at put timestamp t1\n\u003e  - shard has policy X at put timestamp t10\n\u003e  - shard compares new root policy info with stashed root put timestamp and sees that policy Z at t1 supersedes policy X at stashed root put timestamp t0\n\u003e  - shard updates its policy to Z and updates its put timestamp to t11\n\u003e  - shard remembers that root was Z at root put timestamp t1\n\u003e \n\u003e replication cycle (existing behaviour IIUC):\n\u003e  - shard peers learn that shard has policy Z at put timestamp t11\n\u003e  - shard peers update policy from X at t10 to Z\n\u003e \n\u003e audit cycle 2:\n\u003e  - shard learns that root has policy X at put timestamp t0\n\u003e  - shard has policy Z at put timestamp t10 + a + b\n\u003e  - shard compares new root policy info with stashed root put timestamp and sees that policy X at t0 is older than stashed policy Z at t1\n\u003e  - shard makes no changes to its policy: root will eventually become consistent with policy Z at t1\n\u003e \n\u003e Note:\n\u003e 1. the shard needs to remember what root info timestamp it based its policy on, as \"policy timestamp\" meta\n\u003e 2. shard updates its policy only when new root info supersedes stashed root info policy timestamp \n\u003e 3. the shard put timestamp remains independent from the root put timestamp.\n\u003e 4. after shard updates its policy, shard replication between peers should propagate the latest policy because the shard put timestamp always advances when policy changes\n\nYup, we need to do something like this. Current thinking:\n  - Shards will be a Timestamp.now() not based on root\u0027s put_timestamp. \n  - When migrating a root container we\u0027ll add a policy_change (or something) metadata. Which can be used in the shards to move to the latest (if different) policy.\n\nWill push up a new change around this.","accounts_in_message":[],"_revision_number":5},{"id":"0cf8fe2c422126f908e7d8953be25313afae00ce","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2021-07-23 07:36:17.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"ef703864f9ebecfc0ad78cea5acdf00951d593bc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-23 08:43:27.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/221222a5d5d247da8b0c2995683d589e : SUCCESS in 9m 13s\n- grenade https://zuul.opendev.org/t/openstack/build/5559e293dc1c4a33a8032d3e702319bb : SUCCESS in 1h 03m 06s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/949c4cd2740147ec92c9b26dad30ca51 : FAILURE in 32m 32s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d0ec64e4fa0d458b951f7e524528d0b7 : SUCCESS in 8m 20s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/569eec00196e4754b55413dda155d73a : SUCCESS in 8m 54s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/b38642ec169848778ac8103701e4916b : FAILURE in 17m 49s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/03d494769b654a0e8ead7ffe634e3851 : FAILURE in 19m 40s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/c6a5ea7f9bec41a0b3ab75890f8aaed9 : FAILURE in 16m 29s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/441df3414d99421da64b8df68491dd71 : FAILURE in 18m 21s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/dbf50cdcc26e496c8db3f7415cac1fdf : FAILURE in 16m 20s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/a61df8d86f56409e8a7074c85455ee5f : SUCCESS in 16m 16s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/5b3f91703fdc43649a17ff22c1c75402 : SUCCESS in 17m 21s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/be27a750962c40158f22b93eaffc23b1 : SUCCESS in 17m 37s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/4d67347183704e2280d1fd6ebcf189d2 : SUCCESS in 16m 10s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/559fceb463ce4314b0da3ef17b87eb1e : SUCCESS in 17m 05s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/a86f2b37748e44abbca37ab955e9effd : SUCCESS in 20m 22s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/371e4c9db1174069aa8b8d54b47f6fec : SUCCESS in 6m 33s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/c175cfeb956545ba943505604c213d82 : SUCCESS in 19m 15s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/4ff91def30ac4cabb438a04f843a4b15 : SUCCESS in 52m 55s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/33aad9f6ca354eaa81543ad3460c7e63 : FAILURE in 48m 36s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/99eb4ac70e4f4fa1b76ccc7f1c2a08a1 : SUCCESS in 39m 41s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/6a6beed44e6c47cb938b351db3495368 : SUCCESS in 38m 33s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/997c3f86a57240e193cc44d0688d5e44 : FAILURE in 16m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9c7881326ff94cd4a09674e16b3c49e8 : SUCCESS in 6m 06s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e0aa557c187b42c59cf6b29fc9080e3b : SUCCESS in 31m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7ae5085eca954923bad9f64bd5c02528 : SUCCESS in 53m 29s","accounts_in_message":[],"_revision_number":6},{"id":"f7f0f8e81b39f87a1ed8ef932b8c85937fd04ed5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-23 09:04:19.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e8b6761c49d843a488507c013db391f2 : FAILURE in 40m 10s (non-voting)\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a2ecd8a145dc4c5999c747662af5099d : FAILURE in 32m 47s (non-voting)\n- swift-probetests-centos-8-arm64 https://zuul.opendev.org/t/openstack/build/458a4835e19b44daa30842bffe35a743 : FAILURE in 1h 19m 50s (non-voting)\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/a57085b75c5d4a88bc1638cf52272355 : SUCCESS in 34m 18s (non-voting)\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/0fe5e653d3894fbd89b18d31003c07cb : SUCCESS in 32m 30s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"d5586a3211778e37ef2cfe35760e30e68ae8eb55","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2021-07-26 08:33:45.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"dd8509f0fd8c65b0ae4df0415101cb7b7f306587","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-26 09:33:26.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e8c7473046a43359f23eff43a928646 : SUCCESS in 12m 00s\n- grenade https://zuul.opendev.org/t/openstack/build/a8bc703902f9435a8ede89a2cc354b81 : SUCCESS in 58m 09s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/318b57a77e2749a58467b5168fee9a6c : SUCCESS in 43m 12s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/1927e644d4b14b8e83885e983f15417d : SUCCESS in 9m 29s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/924604ebc8db4f1299e9187c2bed1fb5 : SUCCESS in 10m 02s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/54a28582873543c08c74155051c93b2b : SUCCESS in 17m 40s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c5f1c69a249243e89994b4299cdd2f69 : SUCCESS in 22m 51s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/f1c663d77b724ee5b5e92e946599ce79 : SUCCESS in 17m 02s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/e0068bb549a04859adf39d4c54199058 : SUCCESS in 23m 09s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/c902a5d06cbb45a08607f2fc119df85c : SUCCESS in 18m 09s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/216a9f7431e7421c89302d613b4033c5 : SUCCESS in 23m 17s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/44953e82ae244ab4a3be9bd629c42af2 : SUCCESS in 18m 31s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/adb7a71321ff4243878e165026f0e261 : SUCCESS in 24m 19s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/d42e8c6de7664ae7900f4016f9bca70d : SUCCESS in 16m 07s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/46a525aef74d449fb90bf4e2de7348cf : SUCCESS in 21m 12s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/9b6639dfd6254e5b8ac4ae749f1bf788 : SUCCESS in 18m 01s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/bec4ce713864492994b11e2957a82b1a : SUCCESS in 8m 43s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/457ad883c9f1466881f82f483d87d0e1 : SUCCESS in 23m 45s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/63ff6677fa154a499e2a0bc2d90f86de : FAILURE in 54m 31s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/0e047e3f9576495383d5af28a84de1a7 : FAILURE in 53m 55s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/79c59e773a2647cfbb6b93ed336ec5cf : SUCCESS in 43m 59s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/3f6dc75dbec34c34acb56027a4151e9c : SUCCESS in 36m 26s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2517bb4f239e469c96294de24e152f35 : SUCCESS in 17m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c114412b9a834976969e31712fba79fc : SUCCESS in 6m 20s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6e170f3fbbb743e9acf2feeb2028301f : SUCCESS in 31m 41s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d68dd884674c4d92a7911f181ffd8c63 : SUCCESS in 53m 13s","accounts_in_message":[],"_revision_number":7},{"id":"a2ce602eb87b93602950f79d5cce9274d523e648","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-26 10:33:59.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b22470034a3a4222b4fae2eb4e6e4662 : SUCCESS in 51m 37s (non-voting)\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0e6a8ac7bcf04aeca949b40c5acbaf96 : SUCCESS in 50m 57s (non-voting)\n- swift-probetests-centos-8-arm64 https://zuul.opendev.org/t/openstack/build/ce85b5b2a0ac4953b71df2e57cb988f8 : FAILURE in 1h 58m 09s (non-voting)\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/7bc61812bc2d4f4c98a1953979383e76 : SUCCESS in 48m 38s (non-voting)\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/9c55ff434c8c4e558ca1a4ed3093ff22 : SUCCESS in 46m 09s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"d5ecf85c03406c40b5a2442ca0471d77ce7a6dc6","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2021-07-27 00:15:44.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"1908dc8c56fc8bd538004802dd3f9f0dd7599b54","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2021-07-27 06:56:08.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"13d30b869b6dac478387daeef9c87763ae0fff82","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-27 08:14:30.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/371dc76038a0495eb0f6fb91c2cf47db : SUCCESS in 9m 28s\n- grenade https://zuul.opendev.org/t/openstack/build/0fd010efff3a4f479efd1a0116dc4ac7 : SUCCESS in 1h 00m 19s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/4811485f668442918ccde04aa98f30ab : SUCCESS in 48m 15s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/149740081fcd4830bc54851beddecdca : SUCCESS in 8m 25s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/c67c0bfc9329419f9f15df03c41ea444 : SUCCESS in 9m 34s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/a3c5bf503c8041c58c5d4dc4918e8f12 : SUCCESS in 16m 50s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/52353ae072d04c3eb2f7dffa9d8ed9c6 : SUCCESS in 17m 22s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/b508aa42d54d413d89f524e5152da628 : SUCCESS in 18m 38s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/67472cedf42d48fbb5591e80dbf285bd : SUCCESS in 16m 01s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/0fe2a3cd33ce4122a2006f9b5d730ce2 : SUCCESS in 16m 10s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/42faf646605d49bc871c1d90858395d4 : SUCCESS in 16m 40s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/4e0d25306e8f4dbc9d947229caacbf37 : SUCCESS in 17m 15s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/bad54d1001664a75a28e8c79995fafef : SUCCESS in 19m 41s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/f60104e915964f2094d9dd70c01b7cb5 : SUCCESS in 28m 51s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/53379e8c51074dfb893dc07b09ad017a : SUCCESS in 17m 03s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/419f3c620cf04c5292126829b5dfa770 : SUCCESS in 18m 48s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/4ba3c127cd404a7493472b313ae14463 : SUCCESS in 7m 49s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/968037709f754342a737e798f8f8b30c : SUCCESS in 28m 36s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/e963419c3d5341ddbe168be69bde7759 : SUCCESS in 52m 56s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/3eb22763a6a04e8792505cbf82290873 : FAILURE in 1h 09m 37s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/9396bf2357f44433a2e1f03dcb60772d : SUCCESS in 40m 19s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/8a1de49e433c47338d199a9045bd7ff2 : SUCCESS in 41m 49s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/da70d36e826442a1b8a99ff0e7a0d691 : SUCCESS in 19m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/17d50b14ef8342aa8dc56ede92ae997b : SUCCESS in 6m 09s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/04bf4e5c607848caa1e6ed387c80d355 : FAILURE in 24m 21s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8db11d8191bf4cf2bbde1d29ee40345e : SUCCESS in 46m 18s","accounts_in_message":[],"_revision_number":8},{"id":"b061b97e25a270527703aba6e82617af44023337","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-27 09:08:17.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/74489f093be34576b543af81e240c196 : TIMED_OUT in 1h 31m 49s (non-voting)\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/593c9a03ef474aaba74183431c064ff5 : SUCCESS in 47m 45s (non-voting)\n- swift-probetests-centos-8-arm64 https://zuul.opendev.org/t/openstack/build/deceaa036cd24012a01ce923fa233710 : SUCCESS in 1h 44m 30s (non-voting)\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/bcfcf24a6a8a447f8d9c2deeca71ca31 : SUCCESS in 48m 09s (non-voting)\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/a3537bf59edc452d8a3bc8eee6e93713 : TIMED_OUT in 1h 31m 14s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"4962a3cabeb2f5c57f16fc9d88e461b51c2e58ad","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-07-28 14:24:28.000000000","message":"Patch Set 8: Code-Review-1\n\n(4 comments)\n\nMaybe I am missing it, but I don\u0027t see where \u0027X-Container-Sysmeta-Policy-Change\u0027 comes from?","accounts_in_message":[],"_revision_number":8},{"id":"d81b38a3fa8f3cbc871632097464d4d118cbabec","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2021-07-29 00:43:40.000000000","message":"Patch Set 8:\n\n(3 comments)","accounts_in_message":[],"_revision_number":8},{"id":"b5cc1bf1ca0d570b76844ff645f17d97b9f6b05d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2021-07-29 05:59:34.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"3a0002ba560aee616e0d375723f37201c14a7ab9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-29 06:59:04.000000000","message":"Patch Set 9: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e19d39cf5bf24cf7aaa0bcffaa09d29d : SUCCESS in 11m 35s\n- grenade https://zuul.opendev.org/t/openstack/build/2683496f73784f9e83215fceb653a6ef : SUCCESS in 57m 03s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/caf8a34e4d6e4901ad021d9bd6e59af9 : SUCCESS in 38m 17s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e41535174fdf4392b31a414dd40f09be : SUCCESS in 8m 47s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/d123ca28d6784c308a3615da346d3f53 : SUCCESS in 8m 24s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/95ea70cc2d834ca3bb0397498879427c : SUCCESS in 16m 44s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/dff26b31dd4f4bbebb7ec250f73fd96b : SUCCESS in 15m 58s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/078222c8e6b944abbe53e78da4fcc468 : SUCCESS in 17m 11s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/2d0c163b0d7a4c6a935cd4c3f83a6587 : SUCCESS in 16m 23s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/ea9533df1f6f4151a63d6a3784e3875d : SUCCESS in 16m 35s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/04545005f2ed4168ad11603fa3f4ac74 : SUCCESS in 31m 30s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/94e52c8aea2f4c66bdb4aaedfef9cd94 : SUCCESS in 19m 01s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/47f5fa808e1b4f01aae05dbdb11bf350 : SUCCESS in 32m 23s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/49d798dc9430401f8ab0e4fc44ec57d7 : SUCCESS in 30m 51s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/5e2d183282104445be18269ef70d799c : SUCCESS in 18m 27s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/9704f7e8e33d488f9a19fb75237f26c5 : SUCCESS in 19m 21s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/fab5c705761541c79f686f551a985b6a : SUCCESS in 6m 59s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/8ff733a4258a4f5aa81a3e48e4799088 : SUCCESS in 26m 08s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/6d9fa713fede45b9baa14ea57f37a1f3 : FAILURE in 51m 05s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/8832be36eb8e4ca3b8bc8f88d854cecc : FAILURE in 51m 15s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/8aacacb4a44a4c2386920153f75a293c : SUCCESS in 38m 56s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a6035ea5dbb94198bb51adef6abc49ee : SUCCESS in 38m 10s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e52294cbfb9a405c9d99a55d36b42b11 : SUCCESS in 17m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5ebff592f0914293b68df0cd729aa56c : FAILURE in 5m 59s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/443bde7ac6f54497968fa1f58a7081a7 : SUCCESS in 29m 09s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b707342b7db342ed8ef5ec071571f37b : SUCCESS in 49m 53s","accounts_in_message":[],"_revision_number":9},{"id":"632f6297eaf3f8844859a6c75ec97b3c5d29b935","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-29 08:03:42.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/036c26e6ff3b4a7194eff7cf12edf82d : SUCCESS in 49m 56s (non-voting)\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/586c584e81614b7ca73bbf295824d9c7 : SUCCESS in 1h 06m 52s (non-voting)\n- swift-probetests-centos-8-arm64 https://zuul.opendev.org/t/openstack/build/5fcd3e0638ca473d9143bf66b2415f83 : TIMED_OUT in 2h 01m 29s (non-voting)\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/82b2bf5626dc4812aae7ae2e616a21e8 : SUCCESS in 1h 02m 34s (non-voting)\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/fc4922e960014eb1a8287b40cd47b294 : SUCCESS in 58m 13s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"ce97765f208019a9c380ffe86e4e9d707a2f7820","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-07-29 17:57:22.000000000","message":"Patch Set 9:\n\n(5 comments)","accounts_in_message":[],"_revision_number":9},{"id":"2530ecd44ea6cd9175ed69b10a5398a7dc58bd2b","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2021-07-29 21:36:19.000000000","message":"Patch Set 9: Code-Review+1\n\nI have not been able to duplicate the probetest failure on my vsaio","accounts_in_message":[],"_revision_number":9},{"id":"29932970f21a07ac3eca27622d2d9730b733a64c","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2021-07-29 21:57:24.000000000","message":"Patch Set 9: Code-Review-1\n\n(4 comments)\n\nwe need to figure out which thing we want to do first and split up this change:\n\n1) make .shards inhert a change of policy from the root\n2) make containers able to change storage policy in response to new sysmeta\n\nwe already know you CAN change a containers policy by doing a delete and re-create (which results in a bug where the old .shards are confused) but I don\u0027t see a reconciler test that deletes and recreates a sharded container.\n\nTHIS change seems to mostly focus on changing the storage policy with the new query - which will probably be quite useful - but may need to be seperate...","accounts_in_message":[],"_revision_number":9},{"id":"4794ef821b64d85ba74809f90b46bbd2ec472941","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2021-07-29 23:43:26.000000000","message":"Patch Set 9:\n\n(6 comments)\n\nThanks guys, new better patch in coming 😊","accounts_in_message":[],"_revision_number":9},{"id":"7e066f8c4d2b95d129be346490bbdd34f1dfe86a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2021-07-30 04:56:04.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"a2859b08d8bcf065d3d1c78e947e73f6a70ba5d7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-30 05:58:10.000000000","message":"Patch Set 10: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9be05981be5240459d0c0813791ef569 : SUCCESS in 9m 08s\n- grenade https://zuul.opendev.org/t/openstack/build/2596c42da3e747e49a0f99dea0036247 : SUCCESS in 1h 00m 48s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d6fa1c821231417fb69a2ea6aa29cddc : SUCCESS in 46m 20s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/83f816bf808848c29edf65abb9136639 : SUCCESS in 7m 53s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/c1d9d5bbafbc4652816a5f36a5fead49 : SUCCESS in 8m 25s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/7740fd9949074976bf28b7fde36d090e : SUCCESS in 15m 49s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c3d13452e53b4ab58f223e43c9f73bdf : SUCCESS in 15m 50s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/13ecca0b3bb64915bfef003ef843a117 : SUCCESS in 16m 29s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/a210ccf386704a0cadb11e6b67186437 : SUCCESS in 19m 23s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/7f71f7d69fcf4121aa76f15991bdaee4 : SUCCESS in 16m 14s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/5ee2c2c687c64cb3bf666e65bcffa195 : SUCCESS in 16m 02s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/db1f0a4889984764b5862a93c60ddcc1 : SUCCESS in 16m 42s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/e0902961ff4d4666b9e74ef138d3712c : SUCCESS in 16m 56s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/f07ba8c103964526b0bc5e6b216fcbbb : SUCCESS in 16m 32s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/11591217eaea4349bbeec8204030c096 : SUCCESS in 17m 01s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/57d5f1e1659343bea8f36a73d4c884ce : SUCCESS in 17m 08s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/13f59fcb3a75474ea98a0ffdf7e249ee : SUCCESS in 6m 29s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/7291e720ac68489fa5413e657a89e201 : SUCCESS in 19m 15s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/8f4e2837f37947cfa24cc95e6126c4ae : SUCCESS in 51m 50s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/a8e4520292dd4d3e981cff9e367fab54 : SUCCESS in 47m 08s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/3cfc350bd5354751a607e42f701550cc : SUCCESS in 39m 06s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/265f165029a642ccac73fc0742e00bd9 : SUCCESS in 37m 49s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/446a17fb125945f294c85bad74b4ed1d : SUCCESS in 14m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bca8d1fad2d84c80bff453956a6bdaec : FAILURE in 6m 58s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/59ab2517c4004296aaaad871ac25abe5 : SUCCESS in 22m 35s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5848deb876d4443e95f421da1c0955ef : SUCCESS in 56m 22s","accounts_in_message":[],"_revision_number":10},{"id":"51723c18a4e7c66845fe3e79e1801d334b269299","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-30 06:24:00.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4cdf3d46f7df4ae5a88e35d0a1bdda5d : SUCCESS in 29m 48s (non-voting)\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/36cb20a0cf10478e938c0b37ee8c3968 : SUCCESS in 30m 13s (non-voting)\n- swift-probetests-centos-8-arm64 https://zuul.opendev.org/t/openstack/build/9a137c09743b42adbdbc37a3c92c841e : SUCCESS in 1h 17m 32s (non-voting)\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/75035ad070474ca2ba14fef66ca68725 : SUCCESS in 36m 30s (non-voting)\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/4ad40cf4af514196a02baed5bf74c57a : SUCCESS in 35m 15s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"beb03913b16cd35a5d26eecf9227cb2df52bb773","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2021-07-30 07:26:55.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"0f132cc3761169c8da6e7e16186ca8b77c92d978","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-30 08:50:20.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dfcf49f844bc404ca5aca4ef78e3c249 : SUCCESS in 9m 01s\n- grenade https://zuul.opendev.org/t/openstack/build/85ad964e460c4e67b39695a853d3ded0 : SUCCESS in 1h 16m 36s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f11920c6412d464a95dcde073c3238a9 : SUCCESS in 47m 57s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ad5960ac3b8e4065935ed84923cb1475 : SUCCESS in 9m 37s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/3cf10525dbbf491fadfc67af1f92ecd5 : SUCCESS in 8m 39s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/e4d73953a040495f87ff8392839df267 : SUCCESS in 17m 09s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/2376a069e0cf4a409b5dcf4f9d5b9859 : SUCCESS in 17m 42s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/d85ee3fd061a46b1819f175c336b481f : SUCCESS in 17m 50s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/16fbd07b741e414ab9e03cf922b7ba99 : SUCCESS in 18m 49s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/eb39e447d7884052b07d5126e598a3d9 : SUCCESS in 20m 07s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/0d429d4283e848c6b769c22a0aa9092c : SUCCESS in 19m 22s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/fce5ce416b4e45e6b0ed3bcbe24d5dd9 : SUCCESS in 18m 14s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/3b284f2d9d334f24ab972e6b6b083cc7 : SUCCESS in 19m 07s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/c831403bf40645d193d97d54125ba4f8 : SUCCESS in 17m 37s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/2fadd980c3f5496cbd7a15c48c3d2a71 : SUCCESS in 17m 12s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/1b57629a4ff54836a1d0135d9d0430a0 : SUCCESS in 18m 57s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/8a965f2ac2684d8e9365fa5a953e5f81 : SUCCESS in 6m 33s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/9e276f40baf7430f966cca884202fbdb : SUCCESS in 25m 18s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/4ca48b101c22415caff464c2a08eeacb : SUCCESS in 54m 09s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/1ac7878b917e4689821d1b990d3f94ae : SUCCESS in 50m 35s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e13567c252bf478394d5bce35db6496f : SUCCESS in 42m 08s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/281c2535d8394453bec8a46e6c471095 : SUCCESS in 43m 01s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e2464df993ab481ca03dbacdaba56ee3 : SUCCESS in 18m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/01723bd3a0af4981b8e63865c990350b : SUCCESS in 8m 29s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/514d2f88e62d42109b440d67381efe1d : SUCCESS in 34m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ca482186acaa4bd983b1b5042894d257 : SUCCESS in 50m 32s","accounts_in_message":[],"_revision_number":11},{"id":"d5bd82cafd69f50fbf59a27b3c4e5d5b73d9dbf8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-30 09:06:49.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1924c94b9a10437087a5a1813ca7599e : SUCCESS in 52m 09s (non-voting)\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8736a6ea022841cb80ad18af5970b47d : SUCCESS in 47m 15s (non-voting)\n- swift-probetests-centos-8-arm64 https://zuul.opendev.org/t/openstack/build/f23d60e92d67475e875d9b0a889171bc : FAILURE in 1h 31m 30s (non-voting)\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/e769e344cdba4e969ae6674c11813e86 : SUCCESS in 47m 24s (non-voting)\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/5e2f6ba941d24db6948a35b22b6cce2e : SUCCESS in 45m 34s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"ef2c4903c0ad5e37a2f4d54052001d376836495c","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-07-30 16:30:53.000000000","message":"Patch Set 11: Code-Review-1\n\n(6 comments)\n\nfeels like this is getting closer, -1 for possible None[0]","accounts_in_message":[],"_revision_number":11},{"id":"13f8dd7b634733a0b858a76baa112af589a67471","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2021-07-30 20:53:42.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"838ab97145ad48678e6f6f07b80024b0d9b31afe","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2021-07-30 20:58:56.000000000","message":"Patch Set 12:\n\n(3 comments)\n\nI\u0027m warming on this change with the new probe test that asserts it fixed the issue when a container is recreated\n\nmaybe for followup: I think there\u0027s still some nasty race while we\u0027re waiting on the sharder to propogate the spi update - we might want to see if we can get the SPI update inline with the container replicator or maybe even the recreate PUT request (I was a little surprised we \"re-use\" shards after a delete honestly)","accounts_in_message":[],"_revision_number":12},{"id":"de0a09ab77c66cad88122558c070fe839f09f96d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-30 22:05:49.000000000","message":"Patch Set 12: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f08192998a44902a4f5f4cad7ee5a3b : SUCCESS in 9m 15s\n- grenade https://zuul.opendev.org/t/openstack/build/0ec0abe516ee49ba9ff2a9602926a867 : SUCCESS in 1h 07m 41s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a2c929ee3ea74cc8a0bad3fc9faa6060 : SUCCESS in 44m 07s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/4055d82000394e969dcf1fbb2d19bf2c : SUCCESS in 8m 27s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/ba63561d62be4264911e2c4e86f8906f : SUCCESS in 8m 43s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/23ad543b30e54173aaef04f608ff707a : SUCCESS in 17m 26s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/f985aa4db3354c8c8c4b54229b7625d2 : SUCCESS in 16m 32s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/2671234ecd754ac99c546befc351c0e8 : SUCCESS in 16m 53s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/39034dd8543e4ac3b2acb4e73d9293c5 : SUCCESS in 16m 29s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/e7adf2f46eb24062b033855119fde21f : SUCCESS in 16m 19s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/f156a5f093cb45de846a6f0c0a758963 : SUCCESS in 16m 43s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/a1497d704d734229bdc180c48054d939 : SUCCESS in 22m 02s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/752c4dfeba6c470588c92d3c42f343e5 : SUCCESS in 19m 55s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/5d428b0e93b14c42aacc1bb141816406 : SUCCESS in 18m 06s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/6aab361327a647028d42b01f66b07717 : SUCCESS in 16m 43s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/9254a721504e4bea9ad082ac1595fee8 : SUCCESS in 22m 30s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/8d228a4e9ff2467fb04b8bc986fb5dfe : SUCCESS in 6m 40s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/4e454f1a26de4ea7b2dbd4f0ecf1f230 : SUCCESS in 26m 08s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/72da4ff5265f470e9559a00d34c55d45 : FAILURE in 57m 55s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/d973ec71e5ba4bc3ac4e58921a61d529 : SUCCESS in 57m 40s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d466e6e490df438ea6447f22f50841b2 : SUCCESS in 40m 02s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f1af7cd1f560481ca0a9e4ecb10cf866 : SUCCESS in 43m 42s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3bc32aa89051495ab82ff3c23a507d3a : SUCCESS in 21m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c459fa0f1854767807cd99cdea473bf : FAILURE in 7m 48s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/2228e15c4aaf4a319246897cc00d2a40 : SUCCESS in 31m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e7540d520cc34280b0a3e21bd68e6191 : SUCCESS in 46m 13s","accounts_in_message":[],"_revision_number":12},{"id":"555bf89f917b7910853b0485c3a2ce26d8d94458","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-30 22:50:14.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4fc6bc8185a1498abf31a98cece88a3a : SUCCESS in 29m 57s (non-voting)\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8303772255de4886a3c9d20c9dc23bde : SUCCESS in 1h 07m 52s (non-voting)\n- swift-probetests-centos-8-arm64 https://zuul.opendev.org/t/openstack/build/07905093226f4acea790820f3980ba23 : FAILURE in 1h 54m 12s (non-voting)\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/5ea5b497b36a440cb68862fcaa73a0d3 : SUCCESS in 1h 07m 28s (non-voting)\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/b43db34fabee4e4181cd9ff41aca3647 : SUCCESS in 57m 59s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"edad5f3826b26116a5751b77b4a13d25a73f13a6","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2021-08-02 00:42:13.000000000","message":"Patch Set 12:\n\n(2 comments)","accounts_in_message":[],"_revision_number":12},{"id":"7f19c0a752f9ad4c5ec267f3f0ad270f7c339dbf","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2021-08-02 00:54:56.000000000","message":"Patch Set 12:\n\n(4 comments)","accounts_in_message":[],"_revision_number":12},{"id":"6c3667d5af4bf692f33c61b8da40b69020a2c37e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2021-08-02 00:56:09.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"b0cbf75eef5d3b1ec167a9e5e4eefcabc764d73f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-02 01:58:39.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e24c104242b465e96f296e79d7bd24b : SUCCESS in 14m 04s\n- grenade https://zuul.opendev.org/t/openstack/build/e214756dae654e8294ba33bf6e854f3b : SUCCESS in 58m 37s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/763dd2084b27474589ba9bbb8e266d22 : SUCCESS in 45m 52s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/61f7585c2be04e4485aba9ae02737854 : SUCCESS in 8m 36s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/f7042b8a1def481fa7d25eeb0aa1785a : SUCCESS in 9m 01s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/e0237469f31c4e0d87926bd8b00a87c5 : SUCCESS in 21m 38s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/56933c2990e64018bbd9fff87f439ae5 : SUCCESS in 21m 51s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/d389ac13b8894c44b02a144dd09143bc : SUCCESS in 17m 39s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/b02d74adfa7e4355bbea7a6b491a16b7 : SUCCESS in 22m 04s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/e54e20a958ff4cb3830742427aeb12d1 : SUCCESS in 18m 14s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/b32e465a26b240e38217c44915d624be : SUCCESS in 17m 05s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/bb14753a5759408dac030a02dd33db08 : SUCCESS in 21m 38s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/4a8c6bbf55384df4981cd8f6d514c96c : SUCCESS in 22m 06s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/5b3890bd4fbd4b8992066ff148b4f3a6 : SUCCESS in 20m 10s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/80a344d59eb04635af09d0389253fe1f : SUCCESS in 20m 52s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/2305334813804488ba32a5b2a7353960 : SUCCESS in 18m 04s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/35a249c762c040479158f25cf1457879 : SUCCESS in 8m 33s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/64e5b8a03c5346449024258fd5d4a678 : SUCCESS in 24m 39s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/796f9bfb7aae42beab838160e44f9e67 : FAILURE in 59m 23s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/0602aec6eb9344eeb1febdd3dc8603a2 : SUCCESS in 51m 38s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/acd1c87f5d3f4c6aa1e5c13a1cb618fc : SUCCESS in 40m 44s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/747428f43b624c0eb3524ac1d5334b29 : SUCCESS in 40m 13s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cdfbc560220c4d14a3e42d91aab6c251 : SUCCESS in 17m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9aef26024f5841c79d0fa4f4462eaadd : SUCCESS in 6m 10s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/8cb5c19c530c44cfb00ed538c28a65bc : SUCCESS in 30m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/72807b11656147579e15b2e39be78c28 : SUCCESS in 46m 15s","accounts_in_message":[],"_revision_number":13},{"id":"615e28bd77d2c6a6e2f7468e1a6f0db13c895f2f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-02 02:16:59.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a846014848564ff6888b9ab63bfd3141 : SUCCESS in 29m 29s (non-voting)\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3e14b9980aa74e2e9d2ebe66e4bc730f : SUCCESS in 29m 57s (non-voting)\n- swift-probetests-centos-8-arm64 https://zuul.opendev.org/t/openstack/build/d55c6d61346244a99e0c5ec1e5e9004b : SUCCESS in 1h 14m 10s (non-voting)\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/b2c2f4153e714c37965401a6b68cbee9 : SUCCESS in 29m 55s (non-voting)\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/7dc1316cc2cc4ae39e9753affc916ea7 : SUCCESS in 28m 51s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"19a99a6f7ec89961731663742296b355bc2eb303","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-08-02 07:37:41.000000000","message":"Patch Set 13:\n\na couple of thoughts following a conversations with Matt:\n\n- it might be useful as a separate patch to have the proxy specify a policy index when fetching shard listings.\n\n- can the replicator be modified to find misplaced objects based on the *root* storage policy index? either by first updating the shard policy index from root, or by simply always requesting the root policy index before querying for misplaced shard objects?","accounts_in_message":[],"_revision_number":13},{"id":"3d90e2a929ecc711a21d34ceedb7ad0ce3432f04","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2021-08-03 10:42:42.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"335f3b54b3044bd3527ee54941b894f332de7806","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-03 11:38:10.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb8e1aebe31b46389cfa25636577b63e : SUCCESS in 9m 22s\n- grenade https://zuul.opendev.org/t/openstack/build/b486f14ba3634b1eb6bf5e7bc9797ea6 : FAILURE in 49m 08s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/0ea202b891974b56bb0cb9365ee7ebbe : SUCCESS in 47m 33s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/21828514d8424642887379797aa88643 : SUCCESS in 8m 45s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/c826a778f5514ced944457b42cc0bba0 : SUCCESS in 9m 12s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/44fc8bc128ca473683ec5fc56a62c5cb : FAILURE in 16m 15s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/508d5e357ba147e583bbcc940fb8abc1 : FAILURE in 17m 34s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/3c26e70cad1141cb9b3b8e4cc552381f : FAILURE in 16m 52s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/9bb9c68211014ecdb25402c0c3996474 : FAILURE in 16m 24s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/9e76462bc1da480ab66e9794e888d8b8 : FAILURE in 17m 13s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/594fef61f5ec4f6b89f7cc9285a23bce : SUCCESS in 17m 33s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/0d2b67fd79df4c6aae69db8245a7e0fe : SUCCESS in 20m 43s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/20a7b617ca0d4b0583e3f706b150e911 : SUCCESS in 18m 33s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/2431ab78f3f94ab5baa872534aa88109 : SUCCESS in 17m 01s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/6ef61d1bf54d429eb8e6600d1d735f5c : SUCCESS in 17m 19s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/0db31d92dbed4b9a89111d3ee561dabb : SUCCESS in 17m 33s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e3e56a8251ba4d479cbb5e04cc215e60 : SUCCESS in 6m 52s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/ad13a78890d440909768abf646b8bc51 : SUCCESS in 20m 18s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/7b152a97a27142409cbcf18974ad1d16 : FAILURE in 51m 34s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/4401be59ea6f47e8ad77505562282785 : SUCCESS in 49m 51s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/31d41262dc9f40c2ab7d5d8ceb4aedf9 : SUCCESS in 39m 19s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/301171a80db342f5b775bcd399bbae16 : SUCCESS in 39m 39s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d9f45db6774441feb98f268e4b5c0659 : FAILURE in 18m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d86fd72800714a66a3107fbb579c354d : SUCCESS in 6m 06s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/4ac5e424707b42f2975c54990a487780 : SUCCESS in 25m 48s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6748f693c8a14af183793f1e1dfe52d0 : SUCCESS in 51m 11s","accounts_in_message":[],"_revision_number":14},{"id":"39b17586d63ac60f0599c88662e101ad4edc1097","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-03 12:27:52.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/bddfc69e31a5477f8df2a407cdedb82f : FAILURE in 50m 39s (non-voting)\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7a2adfe8cf3841468cfc8982b250e3db : FAILURE in 51m 13s (non-voting)\n- swift-probetests-centos-8-arm64 https://zuul.opendev.org/t/openstack/build/789b30579cbc4e988ef5b86252c4f6cd : FAILURE in 1h 42m 51s (non-voting)\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/95ae1220615a48f48c557c9435655d8c : SUCCESS in 53m 13s (non-voting)\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/d1bf96dcd1b644fcb24678de1a22ae49 : SUCCESS in 37m 17s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"3d2f154796244ab36c75eb1b955bf7562de9d969","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2021-08-10 05:37:13.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"d8d27539147bcfeab2169d2c28c5477e2bfbcf96","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-10 06:44:49.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b8c6d3ecc334cdc93e1e3f674b1f294 : SUCCESS in 8m 53s\n- grenade https://zuul.opendev.org/t/openstack/build/1835164a08494080b10db05b5eb41363 : SUCCESS in 1h 03m 32s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/ab13d85dab25466fae1a2b0a49ac0121 : SUCCESS in 40m 41s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/3eb501d851164ab9a7c2eec1261ec0cd : SUCCESS in 7m 31s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/988d277031144addbc8206a8954adfb6 : SUCCESS in 9m 27s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/a058724a1aac4a0db34cfeb214258edf : SUCCESS in 15m 56s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a022567dad404a38be8a022b21d27c94 : SUCCESS in 16m 43s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/5d6e1869a59e47a8b49ff0aaaa55fb6c : SUCCESS in 16m 16s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/d2034595ab09460b8aad3ac2ee7fe878 : SUCCESS in 16m 10s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/f08ff6a70d6240dbb0981d022b6b0c3f : SUCCESS in 16m 40s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/bce11bb3c0a04d0ba31092fcd487840f : SUCCESS in 16m 07s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/296dc61956aa42f9b4eb4edd831894ee : SUCCESS in 16m 28s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/c4ccaf3e09124a698c3bef4d93d906ac : SUCCESS in 17m 45s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/d914b5d0c2784adda3e85640e8d60260 : SUCCESS in 16m 13s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/0c786e81d0684c9e9cf8ff7235db3a9a : SUCCESS in 16m 32s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/0c18f38261684af1a7f2c9458ae9b3da : SUCCESS in 16m 52s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/924f3a954abb432b9af24b8d69cc2891 : SUCCESS in 6m 14s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/2a7f909f7f7f4a8589504bcf31a133da : SUCCESS in 21m 11s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/00f100758931456f892450e2e08ce28f : FAILURE in 54m 07s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/62e09bd7df50401f96a2f51d042d21bb : SUCCESS in 50m 27s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/3efdb2422b864d2489b9e449025116c2 : SUCCESS in 39m 12s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/311a988ec068423886616ffa2fd96e30 : SUCCESS in 38m 43s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/04df4044fe3143abb290d1f447d6e28a : SUCCESS in 16m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6928b3c70cdd4dae94b615f135ab14be : SUCCESS in 5m 57s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a999ed9c908d42318d67436f39e4f366 : SUCCESS in 22m 06s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a27eefc3e08f4f56868e1f2adeaeaa26 : SUCCESS in 59m 27s","accounts_in_message":[],"_revision_number":15},{"id":"a9b0dc3125cb634189953606a5fccd68c93d453c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-10 07:17:55.000000000","message":"Patch Set 15:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e6cc515af4c94ec9bf1074cfd3abe4fa : SUCCESS in 55m 42s (non-voting)\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b24e1a16c6fe4896b685b4e46f6869a6 : SUCCESS in 37m 36s (non-voting)\n- swift-probetests-centos-8-arm64 https://zuul.opendev.org/t/openstack/build/3c693bd72ca1491b90193cf0db08f207 : SUCCESS in 1h 33m 14s (non-voting)\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/5558d6d6243b466e87354f54e925fbd5 : SUCCESS in 41m 20s (non-voting)\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/43bf86a4265040b5b94162d5cb465d44 : SUCCESS in 39m 19s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"67534c8ed594bb230bdecf5ebe8a7b1694250c4a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-08-13 14:24:41.000000000","message":"Uploaded patch set 16: Patch Set 15 was rebased.","accounts_in_message":[],"_revision_number":16},{"id":"f54f59d44de64036b3cd72944451fc71d9cc3b04","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-08-13 14:26:33.000000000","message":"Patch Set 16:\n\n(1 comment)\n\nAdded some more probe test coverage/assertions here: https://review.opendev.org/c/openstack/swift/+/804527","accounts_in_message":[],"_revision_number":16},{"id":"254dd91d652288d19c2583d81eb6e58e26ccad79","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-13 15:36:13.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/876024517c76461fa90ebf14c863e59e : SUCCESS in 9m 31s\n- grenade https://zuul.opendev.org/t/openstack/build/511bc822f4d34f64abaf64029b3255ee : SUCCESS in 59m 49s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c2ea072f5b014c7e9464cb5d0e64478b : SUCCESS in 39m 46s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/9a5d2556dfdb461ab72ef780dc471378 : SUCCESS in 7m 53s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/c4effac955124e3f80c4ae3b943413dc : SUCCESS in 8m 35s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/3601a24bbf774a10885fbaab5d74b5e1 : SUCCESS in 18m 14s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/7c2dd540889e4557a4fe29821a7d3073 : SUCCESS in 16m 26s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/7da810544ee647edb403d9024df5c428 : SUCCESS in 18m 08s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/be2154a5799641b698ecfba477e0a0d7 : SUCCESS in 18m 07s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/a12386aafff24f66a4213a8d65a6b6cf : SUCCESS in 18m 06s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/8cd6f3dfdb8b4c18b3f34c4f551c15fe : SUCCESS in 18m 55s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/fe56eb1969b14022a99ef1078b18ecb7 : SUCCESS in 18m 48s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/cd13619ef17d487d8b06a218dfc01bc8 : SUCCESS in 22m 42s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/7932614481ff483f811e788b45026c6b : SUCCESS in 21m 04s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/a5ddc7cde99041899fb3e674df1ada47 : SUCCESS in 21m 46s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/3c024f7e4e8340118472f01a28800708 : SUCCESS in 22m 32s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/3b851e74bed44375a20d17f6b2de42e6 : SUCCESS in 7m 10s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/715d4e05bd5a4e5d9507d75d7b6b311a : SUCCESS in 24m 58s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/65b8ac48b6fe4677b215b23a9be02a64 : FAILURE in 1h 00m 38s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/c763848cbc09445b9b4130b701e29e12 : SUCCESS in 56m 39s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/1bdafa4c44734bf2b403e834d8bc3ffe : SUCCESS in 40m 27s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/6d786feedb674e5686d8941372d965b4 : SUCCESS in 41m 27s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b1aab24278ad46439f7285ab84fab687 : SUCCESS in 20m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6ea138596a5a45158dc8152540ebdb44 : SUCCESS in 6m 33s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/26a6dd5fcad844fb959b35d0376b09c8 : SUCCESS in 30m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9e73c7fa96ee448b869fa52b3b5a836c : SUCCESS in 46m 27s","accounts_in_message":[],"_revision_number":16},{"id":"14c648577e06e3c9b0fef9c84fbc51340cb43620","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-13 16:01:59.000000000","message":"Patch Set 16:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4dc844736300458882d20aabdbc6353a : SUCCESS in 37m 20s (non-voting)\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0e1aa834ca254570b7cf62009e9b7db3 : SUCCESS in 40m 41s (non-voting)\n- swift-probetests-centos-8-arm64 https://zuul.opendev.org/t/openstack/build/c72b7ce4fad8457c82cf32628e2b19b6 : SUCCESS in 1h 27m 08s (non-voting)\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/7637141a0e504eccac16f4ea62a4847d : SUCCESS in 37m 00s (non-voting)\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/bf30081c5279448795648dead60bcb3d : SUCCESS in 39m 59s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"e506da43d0549345e68894eb5a76c008791b7530","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2021-08-16 02:57:06.000000000","message":"Uploaded patch set 17.","accounts_in_message":[],"_revision_number":17},{"id":"d51e41ce3aadd045dbf858ac559de5f927a0910a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-16 04:09:09.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/34c8ab7049994dc697a5d5c1de10457b : SUCCESS in 10m 20s\n- grenade https://zuul.opendev.org/t/openstack/build/8a1a7be71c6a4298ad475580ef06f548 : SUCCESS in 1h 07m 12s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/b1ad7601cb5445d28f527d60a61729a0 : SUCCESS in 44m 32s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/57ed38df9201453a850184f6af238521 : SUCCESS in 7m 55s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/71bf747a1c964cc9814979dd2c1261e6 : SUCCESS in 8m 36s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/bb2ba3e7b6a54a7da76275d35832e2bc : SUCCESS in 18m 16s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/4928deafae4f4c72a4349a8d75d92b6b : SUCCESS in 22m 01s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/6a16896f1a30471582b391d85a4b6f8f : SUCCESS in 16m 50s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/0dbfad95148a4da3ba3a1f0914ddc6a5 : SUCCESS in 17m 20s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/896bdd265b224bd39ec492caea4c2e40 : SUCCESS in 16m 02s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/93cee7cfb60d43f5ac619693d43e4957 : SUCCESS in 16m 43s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/b7dc775745e142e59b34d8b7ab479b84 : SUCCESS in 16m 20s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/8334a73de4ba422a84c45314320f9bcb : SUCCESS in 17m 02s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/f9d62857363440e09d9fe307c3fd2539 : SUCCESS in 17m 17s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/8a4ec4221c36493ab1569ce4d9655fd9 : SUCCESS in 16m 31s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/cf34810ce5484ffca420d682d661baa3 : SUCCESS in 18m 39s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/38a6a0bb089b4ad1916f6910a358dc4c : SUCCESS in 6m 21s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/c3677e760f96478bba1030d02145f49e : SUCCESS in 20m 54s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/c09a66a7582e4a8683d029d1570a0776 : SUCCESS in 53m 19s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/f87099461e5f4a37a9ad463b64c893f8 : SUCCESS in 47m 51s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/8c509ef8c57f4d61b3afa6c380925329 : SUCCESS in 33m 22s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/8f7d7e0858a84c4e9c52a60d98b53c13 : SUCCESS in 40m 44s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/352a9458756e46108464f6c08d446cfa : SUCCESS in 16m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4008c4c7ac3341c690d9790a5af79c19 : SUCCESS in 5m 51s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ef606f0407d84917a212e3083dec8fae : SUCCESS in 30m 50s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f14f9fc9258d4e19b4ab517232dc66f2 : SUCCESS in 48m 15s","accounts_in_message":[],"_revision_number":17},{"id":"683fdca90477a2ce2506be54d9e87ddd48d971b9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-16 04:31:16.000000000","message":"Patch Set 17:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8a1b77a614924798884062df9ad455e5 : SUCCESS in 32m 56s (non-voting)\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/cb24c194951d452eb4ff0f9ecc8e4f62 : SUCCESS in 31m 50s (non-voting)\n- swift-probetests-centos-8-arm64 https://zuul.opendev.org/t/openstack/build/2443193f1c3a4c4e9a873e5416a9ea9c : SUCCESS in 1h 27m 09s (non-voting)\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/e42d1ea7985f4b58a6da14a484065b15 : SUCCESS in 32m 06s (non-voting)\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/fa7078cf94b84214a672b5df06281381 : SUCCESS in 39m 14s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"e46061a6ebefac224c0a51bca019ac0fac7b68f9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2021-08-17 06:36:09.000000000","message":"Uploaded patch set 18: Patch Set 17 was rebased.","accounts_in_message":[],"_revision_number":18},{"id":"5dace6b7b604e0eaac79482e3895081588c25abe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-17 08:16:12.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a06a7aa8685e4b6abfddb2b70432d02a : SUCCESS in 12m 10s\n- grenade https://zuul.opendev.org/t/openstack/build/2ea2e9ad69dc4cf48dd7166552210325 : SUCCESS in 1h 01m 50s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/207e32d6fd514d52adbb945eae7050e5 : SUCCESS in 39m 09s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/9231268ecf4549579870029f60c47b05 : SUCCESS in 9m 26s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/ca6a3c75d2dd44eab30aaf716a789302 : SUCCESS in 10m 16s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/3896d2ccb16745f18e84c05646cde6c9 : SUCCESS in 16m 05s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d0579380d28d430482cdb9ef73a121d9 : SUCCESS in 15m 31s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/cec9892f0bb14d39b8c20a872daf40e6 : SUCCESS in 17m 53s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/bb39c244c8014f5bbada1df3a8633687 : SUCCESS in 16m 27s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/f608db12aeaa4ddfb4678238793e42fa : SUCCESS in 23m 14s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/7f7f525661164ad99224322364e7c9d7 : SUCCESS in 16m 54s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/09fa713c81734f2891d40631fe2b672e : SUCCESS in 16m 58s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/b0059a275ed14bc698ff74da553cc7bc : SUCCESS in 17m 48s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/9a489b10e65a4956baa9c164c20ceaa5 : SUCCESS in 17m 32s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/4e00c6395f8449c0950a4dae65089b30 : SUCCESS in 18m 18s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/8077778e20344540a2c6c3e8fb7dc7e4 : SUCCESS in 17m 30s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/2a595620ada644d08bbb213e69a23604 : SUCCESS in 8m 49s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/13ab51c6cf0f4069b000c8cd3b1ba54e : SUCCESS in 21m 43s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/d1f812ac7ec24220bec214683f2ebfd4 : SUCCESS in 53m 26s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/3b3b0d5523e9498fba6ec2566384a5f5 : SUCCESS in 50m 45s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b3b69d793c63409c97b84b48996dbaab : SUCCESS in 41m 02s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/2362d577d6b140a08713008564872ae4 : SUCCESS in 40m 15s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/51f4d9565d9746df8dd71ebc4458a4ed : SUCCESS in 19m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e989de52948f4c87917c878c93104a70 : SUCCESS in 5m 45s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/971e559a8a32434fa924976b7619b185 : SUCCESS in 21m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3b6de0f494ba452d86a0472f97c01a1f : SUCCESS in 50m 09s","accounts_in_message":[],"_revision_number":18},{"id":"9f1d20f749b8b910d6548486cb2cfe6f907a23e0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-17 08:40:12.000000000","message":"Patch Set 18:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/51f36117d8d64180aa76851578990019 : SUCCESS in 29m 53s (non-voting)\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f255ff1275384c4cbf0ed71a4f158dbb : SUCCESS in 35m 05s (non-voting)\n- swift-probetests-centos-8-arm64 https://zuul.opendev.org/t/openstack/build/2528192aa68846f99479b43ed0c29fb0 : TIMED_OUT in 2h 01m 33s (non-voting)\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/de34ef65c0414c1bad86c14caf05b549 : SUCCESS in 35m 11s (non-voting)\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/8c415a1314714c71b4495e468722bdda : SUCCESS in 50m 40s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"d624c3974b1dd8ca8893b4a610d32af8b1a122e1","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-08-20 14:10:59.000000000","message":"Patch Set 18: Code-Review-1\n\n(6 comments)\n\nApart from the bugs around str vs int types, I think there is a fundamental problem with how the shard decides to migrate vs how the reconciler and root replication resolve policy mismatch. The replicator/reconciler do not favour the most recent put-timestamp, but rather then last recreated container. This patch has the shard favouring the most recent put timestamp. That can lead to shards mismatching roots after everything is replicated.\n\nI think we need the shards to perhaps stash both the root put timestamp and the root delete timestamp and then use something like reconciler.py:incorrect_policy_index to decide when a migration is needed.\n\nI\u0027ve pushed a few follow ups, 2 have fixes, the last demonstrates the problem:\n\nhttps://review.opendev.org/c/openstack/swift/+/805387 \nhttps://review.opendev.org/c/openstack/swift/+/805388 \nhttps://review.opendev.org/c/openstack/swift/+/805389","accounts_in_message":[],"_revision_number":18},{"id":"a56353c40e5d9cf0ace223e67f3d3a9baf17247a","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2021-08-20 18:41:56.000000000","message":"Patch Set 18: Workflow-1\n\n(1 comment)\n\nthe str/int fixups are a no brainer - we should squash everything in","accounts_in_message":[],"_revision_number":18},{"id":"da5548d4c77838e4489b276bbf8ae2d8335c4ab1","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2021-08-23 04:36:23.000000000","message":"Patch Set 18:\n\n\u003e Patch Set 18: Workflow-1\n\u003e \n\u003e (1 comment)\n\u003e \n\u003e the str/int fixups are a no brainer - we should squash everything in\n\nAgreed! Will squash the 2 follow ups. And then we can discuss clays and the probetest. As if we just use the reconciler code to check (which makes perfect sense in hindsight) we can figure out what needs to be in and out of the patch.. but let\u0027s just have 1 patch to kull stuff out of 😊\n\nGreat work Al!","accounts_in_message":[],"_revision_number":18},{"id":"cb43ad302dd8fc40b68e6d8f9cfa8a567c92d163","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2021-08-23 04:41:55.000000000","message":"Uploaded patch set 19.","accounts_in_message":[],"_revision_number":19},{"id":"a77d260a8c5b244f4227a7eb60e402cb849bc6f7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-23 05:55:34.000000000","message":"Patch Set 19: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e53f8df6c1804ac3a069022bcab3329d : SUCCESS in 9m 05s\n- grenade https://zuul.opendev.org/t/openstack/build/ec315a9796674241aeff23b7efdd6ae7 : SUCCESS in 1h 08m 13s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d05d30f852824843b2a0d6fbc8a1b794 : SUCCESS in 47m 39s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/5830ac41804e4e0b9478bb6b319be49b : FAILURE in 7m 26s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/520b02da62da49d5af2fd892bf218891 : FAILURE in 7m 50s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/17add831e5434fccb7c7edb60aad7e57 : SUCCESS in 16m 29s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/8bffbf8ac8a743729180547e6332ec93 : FAILURE in 16m 32s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/348ec50c50834dc694becebd0e0fa2c2 : FAILURE in 17m 05s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/79d59bf1a89743f1b859864512e54b4c : FAILURE in 17m 04s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/f236bc54c6c04bb597991f510493fe54 : FAILURE in 17m 32s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/47183edbdb11496194d99885465fb7fb : SUCCESS in 18m 58s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/7d1123861fd1432abc54dc47f8fc07be : SUCCESS in 18m 34s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/83ff3759b6e84ab2ba618036ad48ad4e : SUCCESS in 18m 23s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/baccb411fc4645b7bdca0ba9391f1bb5 : SUCCESS in 16m 04s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/5509253b42114ad49a7a8f7f59e7131b : SUCCESS in 17m 02s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/de99a640b4944ad7baf7fd395d52ec07 : SUCCESS in 17m 46s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/51c0e08b2e9d4cc6bb548f8f2d545b35 : SUCCESS in 6m 51s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/79bbfdc8e3f7482ab4341de7384c753e : SUCCESS in 23m 21s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/2824f79e87bd45ab88db1eca4cb2e986 : SUCCESS in 59m 53s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/7f997b66dcd647f791869f9925e6f57d : SUCCESS in 59m 39s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/acb5f24cbb684b1082edc9c187ea5a42 : SUCCESS in 39m 04s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a91f24f36291490ab6a1b59247916f2b : SUCCESS in 47m 11s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e5caaa3e89b541be88ee10f985178990 : SUCCESS in 16m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bfd8b624e51a4fd3998baeabf00729ec : SUCCESS in 6m 07s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6329c27e7b1848c5a7e7f43062a5cfce : SUCCESS in 22m 17s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d64283f1781049cc920a4e2846d1805e : SUCCESS in 45m 15s","accounts_in_message":[],"_revision_number":19},{"id":"3ee03e830cb2584ac3821075b49842b9887cd0aa","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-23 06:13:56.000000000","message":"Patch Set 19:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2c9506224ff04a868af03cb77bb1dcf6 : FAILURE in 37m 44s (non-voting)\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/32ee372523ea415b83effc8babdbb6a7 : FAILURE in 34m 12s (non-voting)\n- swift-probetests-centos-8-arm64 https://zuul.opendev.org/t/openstack/build/35b3ee2c69d64b5ea14ec2c96c825fd5 : SUCCESS in 1h 25m 23s (non-voting)\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/ccfc543ba9564e73bede5d7cba4c2ee7 : SUCCESS in 48m 19s (non-voting)\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/76ca838472b747e1bf00d07753d20e4f : SUCCESS in 45m 44s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"664aecf1ac7d6dbd2e7bcee963163331736a24ba","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-08-23 09:24:05.000000000","message":"Uploaded patch set 20.","accounts_in_message":[],"_revision_number":20},{"id":"ca81cf46796dfde94aab7333a5b44d7ea2f1abc8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-23 10:25:04.000000000","message":"Patch Set 20: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed2c48ae234c4c5289f88dca6fdee16d : SUCCESS in 11m 04s\n- grenade https://zuul.opendev.org/t/openstack/build/3d644cbca40f49d7826d5258a092e9ce : SUCCESS in 59m 35s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f25d2b7ed1984783ba2ab837bf4c40d4 : SUCCESS in 39m 56s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e993993c554e4024ac67d44c00875d5c : FAILURE in 7m 34s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/5977df451159476ca2f5f02b162efb2f : FAILURE in 9m 24s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/930275a6050247b5ba1afe5f596aa795 : SUCCESS in 26m 27s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/96c5f746c8b7477dbb6499b5ce4f4dd3 : SUCCESS in 16m 07s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/c4fbf4bd70da48909e84b9f189e6a69b : SUCCESS in 16m 34s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/f8557e7fb33c460ea8e391f23ef57cb2 : SUCCESS in 18m 45s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/3a981505183942e4a8f35ecf2a92767b : SUCCESS in 17m 10s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/da5edf9318b345cba2cf80fabdb96e0b : SUCCESS in 18m 02s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/cfd258d509c34818826e86fc5bd60078 : SUCCESS in 18m 05s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/17cfd6ed98664af1a54e46b5a6a4e2fd : SUCCESS in 18m 20s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/cf710ba98e174a01964a3d9fb59c5fc1 : SUCCESS in 26m 46s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/b3847756e1f44c12a582be065437f74c : SUCCESS in 20m 44s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/e745a5a882de4aa5a865098965974a58 : SUCCESS in 18m 12s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/0e32502004064591bf95cadda5ae7ce0 : SUCCESS in 9m 08s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/2f0713b05b2449d1a488eb4185101ca8 : SUCCESS in 22m 35s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/3415ca4102f2494e9a37058d1dffd3a4 : SUCCESS in 50m 19s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/255b97c3aa5a493392f0d869a17740a9 : SUCCESS in 48m 23s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/723ef9774935426eb10291899552589d : SUCCESS in 38m 14s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/05ef9e7354b24c01b49e793b9557c7c8 : SUCCESS in 39m 39s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a8f3e511ec6a4d3f86c41dbbc79bfb8c : SUCCESS in 16m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4e4de6a5befe419ab168ddc7663f035a : SUCCESS in 6m 23s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d6980dbbc408478f8bcb02326491a72b : SUCCESS in 26m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/45eac8f7bf0c4d13b61ffb0f8188fb1f : SUCCESS in 49m 00s","accounts_in_message":[],"_revision_number":20},{"id":"159266e65babb865668f3ab84b5d96b9b6b96503","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-23 11:49:46.000000000","message":"Patch Set 20:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/20b259bf94924abd9ff4c2d108f48da8 : SUCCESS in 39m 48s (non-voting)\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5ff8683391a6475082618d09975bc691 : SUCCESS in 46m 14s (non-voting)\n- swift-probetests-centos-8-arm64 https://zuul.opendev.org/t/openstack/build/7fed5ab2994647afbaf3176c93718440 : TIMED_OUT in 2h 01m 28s (non-voting)\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/16e2393f125b4fd89436dfcdc8ea30a3 : SUCCESS in 41m 29s (non-voting)\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/554d35baa7dc490b8e85fefbbd30c067 : SUCCESS in 45m 50s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"a66d3a36748a07fed4a526da3a3c82385dbf6bba","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-05-09 16:29:08.000000000","message":"Uploaded patch set 21.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":21},{"id":"afb0aa56383110a9980f60b4848b3c597ac2e9a9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-09 16:53:26.000000000","message":"Patch Set 21:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b6a51243cc914ee09bffeb8f5cc8b8f3\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7b1bdaaf1da34aa6b12bcd8ce84a412c : SUCCESS in 15m 06s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1cfb925259494bf9afa45f5b7271db5f : SUCCESS in 16m 10s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/9963912e956e41c18c46743c925ef5a0 : SUCCESS in 22m 59s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/ffa72f43669c40a49eed681c61e84d88 : SUCCESS in 18m 10s","accounts_in_message":[],"_revision_number":21},{"id":"349154ee54851c988ea25e844cd2c5a41eef697f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-09 17:47:08.000000000","message":"Patch Set 21: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d7b93dee73b844219b983e8375a71391\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5fa0d487bd03492a8feb579ff658fd93 : SUCCESS in 8m 59s\n- grenade https://zuul.opendev.org/t/openstack/build/fceeba946ff54f8b97596a6ac1923bff : SUCCESS in 55m 22s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f69420d6d2f84a4789c80a1529a9e2d7 : SUCCESS in 36m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6976d02106984b4498814fc640428d7b : SUCCESS in 1h 11m 37s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3abfd74a09d44668a829d2f471b6f363 : SUCCESS in 14m 28s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1010629e73b74915964f680cdfe48178 : SUCCESS in 19m 00s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d00b053832054fae8530dc77f92ea205 : SUCCESS in 18m 23s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c4470f61655548a7b4d5de02784d2967 : SUCCESS in 10m 37s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/e763ccbf491e4cd3b658745cd5fb38fe : SUCCESS in 5m 11s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/68f95e6c90db42f4b37f4b3e26157670 : SUCCESS in 22m 26s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/db06e5cbde64481ca62329e16fd2497e : SUCCESS in 19m 05s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/5fca4d71b4184601af6cc7c6d196276c : SUCCESS in 18m 36s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/96713d61b2244da28bf26d96cae5c5fc : SUCCESS in 18m 16s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/2a7046c0ec194cdea095fff8d461c958 : SUCCESS in 22m 52s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/74e3a0c0f5374ac0a16ac544b71d6c8a : SUCCESS in 21m 01s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/3b73e73f23a74c78bbca9985837cbcd3 : SUCCESS in 18m 50s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/84f0aa13f2b5402d918ed492e8bccde8 : SUCCESS in 20m 43s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/e115139ffb49422cb54ce331c51b593c : SUCCESS in 18m 37s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/f9845e2b2f6546b39d23f3d458a7700b : SUCCESS in 17m 50s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f362b111ea8f442ca183a0de3d432c6e : SUCCESS in 7m 12s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e29c74d6ee4c44219acb10d39f9357f3 : SUCCESS in 16m 31s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/a79093ea61c74c389294f09c73c4c5b2 : SUCCESS in 6m 59s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/8ac50a411c6c487694992577450bfad1 : SUCCESS in 1h 07m 52s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c85a62b58206480497621519bfa7b9a0 : SUCCESS in 51m 55s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b28bc169d9004b0ca339c26d650dff7c : SUCCESS in 1h 07m 51s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/466a1599ad474939807be1f4e0f15860 : SUCCESS in 19m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/84967309657d4849a995294185f49fe4 : SUCCESS in 4m 12s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6fa6eebe0abf40e1bef9be7b621f21ca : SUCCESS in 19m 41s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8d589055bc4348bdb0abbe614e987c86 : SUCCESS in 1h 08m 17s","accounts_in_message":[],"_revision_number":21}],"current_revision_number":21,"current_revision":"086842a0712a2fc07918aa6e70a6a51b38065695","revisions":{"94f5fe2955849fce65115caefdff083953581d1e":{"kind":"REWORK","_number":1,"created":"2021-07-14 06:28:29.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/48/800748/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/48/800748/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/800748/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/800748/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/800748/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/48/800748/1"}}},"commit":{"parents":[{"commit":"e6116ce0d30047eaa81209127305f558e969a316","subject":"wip: Cluster assisted Storage Policy Migration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e6116ce0d30047eaa81209127305f558e969a316"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2021-07-14 06:20:00.000000000","tz":600},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2021-07-14 06:20:00.000000000","tz":600},"subject":"sharder: update shard storage_policy_index if roots changes","message":"sharder: update shard storage_policy_index if roots changes\n\nIf there is a change to the root container\u0027s storage_policy_index, the\nreconciler will kick in and migrate objects to the new/correct storage\npolicy. However, a policy_index change on the root isn\u0027t propagated to\nit\u0027s shards.\n\nTurns out when we go and fetch the shard range from the root container\nin the shard container audit, the response includes the roots policy\nindex. So in this patch, if it sees that the roots is different from the\nshards it\u0027ll migrate to the new storage policy. This will kick of the\nreconciler on the next container-replication run as it enqueues objects\nto reconcile.\n\nThis is the first version and is based on clays migration work. And\nseems to be working when testing on my SAIO.\nBut we might decide we don\u0027t want it based on Clay\u0027s parent patch so\nmight have to strip it back and just update the policy_stat table\ndirectly to change the storage_policy.\n\nChange-Id: Ib1d6197caa15d0974334be3d7eff24d4342ac378\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/94f5fe2955849fce65115caefdff083953581d1e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/94f5fe2955849fce65115caefdff083953581d1e"}]},"branch":"refs/heads/master"},"bdb44596a41902347af4ee7c64f4e1e4f4358001":{"kind":"REWORK","_number":2,"created":"2021-07-19 07:00:27.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/48/800748/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/48/800748/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/800748/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/800748/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/800748/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/48/800748/2"}}},"commit":{"parents":[{"commit":"902557e1cddce3f8935d806dfab35c057b8fa97e","subject":"Merge \"reconciler: Tolerate 503s on HEAD\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/902557e1cddce3f8935d806dfab35c057b8fa97e"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2021-07-14 06:20:00.000000000","tz":600},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2021-07-19 02:22:04.000000000","tz":600},"subject":"sharder: update shard storage_policy_index if roots changes","message":"sharder: update shard storage_policy_index if roots changes\n\nIf there is a change to the root container\u0027s storage_policy_index, the\nreconciler will kick in and migrate objects to the new/correct storage\npolicy. However, a policy_index change on the root isn\u0027t propagated to\nit\u0027s shards.\n\nTurns out when we go and fetch the shard range from the root container\nin the shard container audit, the response includes the roots policy\nindex. So in this patch, if it sees that the roots is different from the\nshards it\u0027ll migrate to the new storage policy. This will kick of the\nreconciler on the next container-replication run as it enqueues objects\nto reconcile.\n\nThis is the first version and is based on clays migration work. And\nseems to be working when testing on my SAIO.\nBut we might decide we don\u0027t want it based on Clay\u0027s parent patch so\nmight have to strip it back and just update the policy_stat table\ndirectly to change the storage_policy.\n\nChange-Id: Ib1d6197caa15d0974334be3d7eff24d4342ac378\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bdb44596a41902347af4ee7c64f4e1e4f4358001"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bdb44596a41902347af4ee7c64f4e1e4f4358001"}]},"branch":"refs/heads/master"},"2d91ef276229ce88c7cdc2d6bfc6e5cdbe8d4bae":{"kind":"REWORK","_number":3,"created":"2021-07-21 07:08:58.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/48/800748/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/48/800748/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/800748/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/800748/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/800748/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/48/800748/3"}}},"commit":{"parents":[{"commit":"902557e1cddce3f8935d806dfab35c057b8fa97e","subject":"Merge \"reconciler: Tolerate 503s on HEAD\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/902557e1cddce3f8935d806dfab35c057b8fa97e"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2021-07-14 06:20:00.000000000","tz":600},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2021-07-21 07:08:28.000000000","tz":600},"subject":"sharder: update shard storage_policy_index if roots changes","message":"sharder: update shard storage_policy_index if roots changes\n\nIf there is a change to the root container\u0027s storage_policy_index, the\nreconciler will kick in and migrate objects to the new/correct storage\npolicy. However, a policy_index change on the root isn\u0027t propagated to\nit\u0027s shards.\n\nTurns out when we go and fetch the shard range from the root container\nin the shard container audit, the response includes the roots policy\nindex. So in this patch, if it sees that the roots is different from the\nshards it\u0027ll migrate to the new storage policy. This will kick of the\nreconciler on the next container-replication run as it enqueues objects\nto reconcile.\n\nThis is the first version and is based on clays migration work. And\nseems to be working when testing on my SAIO.\nBut we might decide we don\u0027t want it based on Clay\u0027s parent patch so\nmight have to strip it back and just update the policy_stat table\ndirectly to change the storage_policy.\n\nChange-Id: Ib1d6197caa15d0974334be3d7eff24d4342ac378\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2d91ef276229ce88c7cdc2d6bfc6e5cdbe8d4bae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2d91ef276229ce88c7cdc2d6bfc6e5cdbe8d4bae"}]},"branch":"refs/heads/master"},"11b8cd34faf5e6e12faa80145b224d3ef91c3dba":{"kind":"REWORK","_number":4,"created":"2021-07-23 03:09:55.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/48/800748/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/48/800748/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/800748/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/800748/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/800748/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/48/800748/4"}}},"commit":{"parents":[{"commit":"b53a9d811413840d03840c3d9d0287e2a8aff717","subject":"Merge \"trivial: Use own_shard_range instead of doing an extra query\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b53a9d811413840d03840c3d9d0287e2a8aff717"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2021-07-14 06:20:00.000000000","tz":600},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2021-07-23 03:09:31.000000000","tz":600},"subject":"sharder: update shard storage_policy_index if roots changes","message":"sharder: update shard storage_policy_index if roots changes\n\nIf there is a change to the root container\u0027s storage_policy_index, the\nreconciler will kick in and migrate objects to the new/correct storage\npolicy. However, a policy_index change on the root isn\u0027t propagated to\nit\u0027s shards.\n\nTurns out when we go and fetch the shard range from the root container\nin the shard container audit, the response includes the roots policy\nindex. So in this patch, if it sees that the roots is different from the\nshards it\u0027ll migrate to the new storage policy. This will kick of the\nreconciler on the next container-replication run as it enqueues objects\nto reconcile.\n\nThis is the first version and is based on clays migration work. And\nseems to be working when testing on my SAIO.\nBut we might decide we don\u0027t want it based on Clay\u0027s parent patch so\nmight have to strip it back and just update the policy_stat table\ndirectly to change the storage_policy.\n\nChange-Id: Ib1d6197caa15d0974334be3d7eff24d4342ac378\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/11b8cd34faf5e6e12faa80145b224d3ef91c3dba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/11b8cd34faf5e6e12faa80145b224d3ef91c3dba"}]},"branch":"refs/heads/master"},"292a5d519787f0152151cd782b9302ac8ae2ee55":{"kind":"REWORK","_number":5,"created":"2021-07-23 05:00:57.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/48/800748/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/48/800748/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/800748/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/800748/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/800748/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/48/800748/5"}}},"commit":{"parents":[{"commit":"b53a9d811413840d03840c3d9d0287e2a8aff717","subject":"Merge \"trivial: Use own_shard_range instead of doing an extra query\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b53a9d811413840d03840c3d9d0287e2a8aff717"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2021-07-14 06:20:00.000000000","tz":600},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2021-07-23 05:00:29.000000000","tz":600},"subject":"sharder: update shard storage_policy_index if roots changes","message":"sharder: update shard storage_policy_index if roots changes\n\nIf there is a change to the root container\u0027s storage_policy_index, the\nreconciler will kick in and migrate objects to the new/correct storage\npolicy. However, a policy_index change on the root isn\u0027t propagated to\nit\u0027s shards.\n\nTurns out when we go and fetch the shard range from the root container\nin the shard container audit, the response includes the roots policy\nindex. So in this patch, if it sees that the roots is different from the\nshards it\u0027ll migrate to the new storage policy. This will kick of the\nreconciler on the next container-replication run as it enqueues objects\nto reconcile.\n\nThis is the first version and is based on clays migration work. And\nseems to be working when testing on my SAIO.\nBut we might decide we don\u0027t want it based on Clay\u0027s parent patch so\nmight have to strip it back and just update the policy_stat table\ndirectly to change the storage_policy.\n\nChange-Id: Ib1d6197caa15d0974334be3d7eff24d4342ac378\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/292a5d519787f0152151cd782b9302ac8ae2ee55"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/292a5d519787f0152151cd782b9302ac8ae2ee55"}]},"branch":"refs/heads/master"},"0c59884a14edf122e63dc5e2c363530b0c6c6a38":{"kind":"REWORK","_number":6,"created":"2021-07-23 07:36:17.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/48/800748/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/48/800748/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/800748/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/800748/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/800748/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/48/800748/6"}}},"commit":{"parents":[{"commit":"b53a9d811413840d03840c3d9d0287e2a8aff717","subject":"Merge \"trivial: Use own_shard_range instead of doing an extra query\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b53a9d811413840d03840c3d9d0287e2a8aff717"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2021-07-14 06:20:00.000000000","tz":600},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2021-07-23 07:35:23.000000000","tz":600},"subject":"sharder: update shard storage_policy_index if roots changes","message":"sharder: update shard storage_policy_index if roots changes\n\nIf there is a change to the root container\u0027s storage_policy_index, the\nreconciler will kick in and migrate objects to the new/correct storage\npolicy. However, a policy_index change on the root isn\u0027t propagated to\nit\u0027s shards.\n\nTurns out when we go and fetch the shard range from the root container\nin the shard container audit, the response includes the roots policy\nindex. So in this patch, if it sees that the roots is different from the\nshards it\u0027ll migrate to the new storage policy. This will kick of the\nreconciler on the next container-replication run as it enqueues objects\nto reconcile.\n\nChange-Id: Ib1d6197caa15d0974334be3d7eff24d4342ac378\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0c59884a14edf122e63dc5e2c363530b0c6c6a38"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0c59884a14edf122e63dc5e2c363530b0c6c6a38"}]},"branch":"refs/heads/master"},"5b6648951ba637eafcf4fcda47df89acdb5b9c20":{"kind":"REWORK","_number":7,"created":"2021-07-26 08:33:45.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/48/800748/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/48/800748/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/800748/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/800748/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/800748/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/48/800748/7"}}},"commit":{"parents":[{"commit":"b53a9d811413840d03840c3d9d0287e2a8aff717","subject":"Merge \"trivial: Use own_shard_range instead of doing an extra query\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b53a9d811413840d03840c3d9d0287e2a8aff717"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2021-07-14 06:20:00.000000000","tz":600},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2021-07-26 08:33:07.000000000","tz":600},"subject":"sharder: update shard storage_policy_index if roots changes","message":"sharder: update shard storage_policy_index if roots changes\n\nIf there is a change to the root container\u0027s storage_policy_index, the\nreconciler will kick in and migrate objects to the new/correct storage\npolicy. However, a policy_index change on the root isn\u0027t propagated to\nit\u0027s shards.\n\nTurns out when we go and fetch the shard range from the root container\nin the shard container audit, the response includes the roots policy\nindex. So in this patch, if it sees that the roots is different from the\nshards it\u0027ll migrate to the new storage policy. This will kick of the\nreconciler on the next container-replication run as it enqueues objects\nto reconcile.\n\nChange-Id: Ib1d6197caa15d0974334be3d7eff24d4342ac378\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5b6648951ba637eafcf4fcda47df89acdb5b9c20"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5b6648951ba637eafcf4fcda47df89acdb5b9c20"}]},"branch":"refs/heads/master"},"da376375df2d7cf6c1d3d7ead09cb7e945de5aa9":{"kind":"REWORK","_number":8,"created":"2021-07-27 06:56:08.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/48/800748/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/48/800748/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/800748/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/800748/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/800748/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/48/800748/8"}}},"commit":{"parents":[{"commit":"b53a9d811413840d03840c3d9d0287e2a8aff717","subject":"Merge \"trivial: Use own_shard_range instead of doing an extra query\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b53a9d811413840d03840c3d9d0287e2a8aff717"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2021-07-14 06:20:00.000000000","tz":600},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2021-07-27 06:55:28.000000000","tz":600},"subject":"sharder: update shard storage_policy_index if roots changes","message":"sharder: update shard storage_policy_index if roots changes\n\nIf there is a change to the root container\u0027s storage_policy_index, the\nreconciler will kick in and migrate objects to the new/correct storage\npolicy. However, a policy_index change on the root isn\u0027t propagated to\nit\u0027s shards.\n\nTurns out when we go and fetch the shard range from the root container\nin the shard container audit, the response includes the roots policy\nindex. So in this patch, if it sees that the roots is different from the\nshards it\u0027ll migrate to the new storage policy. This will kick of the\nreconciler on the next container-replication run as it enqueues objects\nto reconcile.\n\nChange-Id: Ib1d6197caa15d0974334be3d7eff24d4342ac378\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/da376375df2d7cf6c1d3d7ead09cb7e945de5aa9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/da376375df2d7cf6c1d3d7ead09cb7e945de5aa9"}]},"branch":"refs/heads/master"},"12fa27551b76fa5bacaf9e4211ae5958a7e3420e":{"kind":"REWORK","_number":9,"created":"2021-07-29 05:59:34.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/48/800748/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/48/800748/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/800748/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/800748/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/800748/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/48/800748/9"}}},"commit":{"parents":[{"commit":"b53a9d811413840d03840c3d9d0287e2a8aff717","subject":"Merge \"trivial: Use own_shard_range instead of doing an extra query\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b53a9d811413840d03840c3d9d0287e2a8aff717"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2021-07-14 06:20:00.000000000","tz":600},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2021-07-29 05:59:08.000000000","tz":600},"subject":"sharder: update shard storage_policy_index if roots changes","message":"sharder: update shard storage_policy_index if roots changes\n\nIf there is a change to the root container\u0027s storage_policy_index, the\nreconciler will kick in and migrate objects to the new/correct storage\npolicy. However, a policy_index change on the root isn\u0027t propagated to\nit\u0027s shards.\n\nTurns out when we go and fetch the shard range from the root container\nin the shard container audit, the response includes the roots policy\nindex. So in this patch, if it sees that the roots is different from the\nshards it\u0027ll migrate to the new storage policy. This will kick of the\nreconciler on the next container-replication run as it enqueues objects\nto reconcile.\n\nChange-Id: Ib1d6197caa15d0974334be3d7eff24d4342ac378\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/12fa27551b76fa5bacaf9e4211ae5958a7e3420e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/12fa27551b76fa5bacaf9e4211ae5958a7e3420e"}]},"branch":"refs/heads/master"},"b9c692273de072bece4f2c633a36a91341bd6f6a":{"kind":"REWORK","_number":10,"created":"2021-07-30 04:56:04.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/48/800748/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/48/800748/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/800748/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/800748/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/800748/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/48/800748/10"}}},"commit":{"parents":[{"commit":"b53a9d811413840d03840c3d9d0287e2a8aff717","subject":"Merge \"trivial: Use own_shard_range instead of doing an extra query\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b53a9d811413840d03840c3d9d0287e2a8aff717"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2021-07-14 06:20:00.000000000","tz":600},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2021-07-30 04:40:19.000000000","tz":600},"subject":"sharder: update shard storage_policy_index if roots changes","message":"sharder: update shard storage_policy_index if roots changes\n\nIf there is a change to the root container\u0027s storage_policy_index, the\nreconciler will kick in and migrate objects to the new/correct storage\npolicy. However, a policy_index change on the root isn\u0027t propagated to\nit\u0027s shards.\n\nTurns out when we go and fetch the shard range from the root container\nin the shard container audit, the response includes the roots policy\nindex. So in this patch, if it sees that the roots is different from the\nshards it\u0027ll migrate to the new storage policy. This will kick of the\nreconciler on the next container-replication run as it enqueues objects\nto reconcile.\n\nChange-Id: Ib1d6197caa15d0974334be3d7eff24d4342ac378\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b9c692273de072bece4f2c633a36a91341bd6f6a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b9c692273de072bece4f2c633a36a91341bd6f6a"}]},"branch":"refs/heads/master"},"10f1dd8e5f311a7e97f01917503eb951a2591f4e":{"kind":"REWORK","_number":11,"created":"2021-07-30 07:26:55.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/48/800748/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/48/800748/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/800748/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/800748/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/800748/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/48/800748/11"}}},"commit":{"parents":[{"commit":"b53a9d811413840d03840c3d9d0287e2a8aff717","subject":"Merge \"trivial: Use own_shard_range instead of doing an extra query\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b53a9d811413840d03840c3d9d0287e2a8aff717"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2021-07-14 06:20:00.000000000","tz":600},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2021-07-30 07:26:28.000000000","tz":600},"subject":"sharder: update shard storage_policy_index if roots changes","message":"sharder: update shard storage_policy_index if roots changes\n\nIf there is a change to the root container\u0027s storage_policy_index, the\nreconciler will kick in and migrate objects to the new/correct storage\npolicy. However, a policy_index change on the root isn\u0027t propagated to\nit\u0027s shards.\n\nTurns out when we go and fetch the shard range from the root container\nin the shard container audit, the response includes the roots policy\nindex. So in this patch, if it sees that the roots is different from the\nshards it\u0027ll migrate to the new storage policy. This will kick of the\nreconciler on the next container-replication run as it enqueues objects\nto reconcile.\n\nChange-Id: Ib1d6197caa15d0974334be3d7eff24d4342ac378\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/10f1dd8e5f311a7e97f01917503eb951a2591f4e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/10f1dd8e5f311a7e97f01917503eb951a2591f4e"}]},"branch":"refs/heads/master"},"ed8c78d95ed425a2d47c93f229fa93b2520fce8a":{"kind":"REWORK","_number":12,"created":"2021-07-30 20:53:42.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/48/800748/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/48/800748/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/800748/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/800748/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/800748/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/48/800748/12"}}},"commit":{"parents":[{"commit":"b53a9d811413840d03840c3d9d0287e2a8aff717","subject":"Merge \"trivial: Use own_shard_range instead of doing an extra query\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b53a9d811413840d03840c3d9d0287e2a8aff717"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2021-07-14 06:20:00.000000000","tz":600},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2021-07-30 20:51:25.000000000","tz":-300},"subject":"sharder: update shard storage_policy_index if roots changes","message":"sharder: update shard storage_policy_index if roots changes\n\nIf there is a change to the root container\u0027s storage_policy_index (e.g.\nafter it\u0027s deleted and recreated), the reconciler will kick in and\nmigrate objects (or tombstone rows) to the new/correct storage policy.\n\nTo ensure this works correctly on sharded containers, we must propagate\na policy_index change on the root to it\u0027s shards.\n\nTurns out when we go and fetch the shard range from the root container\nin the shard container audit, the response includes the roots policy\nindex. So in this patch, if it sees that the roots is different from the\nshards it\u0027ll migrate to the new storage policy. This will kick off the\nreconciler on the next container-replication run as it enqueues objects\nto reconcile.\n\nChange-Id: Ib1d6197caa15d0974334be3d7eff24d4342ac378\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ed8c78d95ed425a2d47c93f229fa93b2520fce8a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ed8c78d95ed425a2d47c93f229fa93b2520fce8a"}]},"branch":"refs/heads/master"},"2a74d952052ca5f79effa0b706eea7d58899aca4":{"kind":"REWORK","_number":13,"created":"2021-08-02 00:56:09.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/48/800748/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/48/800748/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/800748/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/800748/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/800748/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/48/800748/13"}}},"commit":{"parents":[{"commit":"b53a9d811413840d03840c3d9d0287e2a8aff717","subject":"Merge \"trivial: Use own_shard_range instead of doing an extra query\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b53a9d811413840d03840c3d9d0287e2a8aff717"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2021-07-14 06:20:00.000000000","tz":600},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2021-08-02 00:55:51.000000000","tz":600},"subject":"sharder: update shard storage_policy_index if roots changes","message":"sharder: update shard storage_policy_index if roots changes\n\nIf there is a change to the root container\u0027s storage_policy_index (e.g.\nafter it\u0027s deleted and recreated), the reconciler will kick in and\nmigrate objects (or tombstone rows) to the new/correct storage policy.\n\nTo ensure this works correctly on sharded containers, we must propagate\na policy_index change on the root to it\u0027s shards.\n\nTurns out when we go and fetch the shard range from the root container\nin the shard container audit, the response includes the roots policy\nindex. So in this patch, if it sees that the roots is different from the\nshards it\u0027ll migrate to the new storage policy. This will kick off the\nreconciler on the next container-replication run as it enqueues objects\nto reconcile.\n\nChange-Id: Ib1d6197caa15d0974334be3d7eff24d4342ac378\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2a74d952052ca5f79effa0b706eea7d58899aca4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2a74d952052ca5f79effa0b706eea7d58899aca4"}]},"branch":"refs/heads/master"},"6053a65c63340cafa661fa4379ae85d035615f53":{"kind":"REWORK","_number":14,"created":"2021-08-03 10:42:42.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/48/800748/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/48/800748/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/800748/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/800748/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/800748/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/48/800748/14"}}},"commit":{"parents":[{"commit":"b53a9d811413840d03840c3d9d0287e2a8aff717","subject":"Merge \"trivial: Use own_shard_range instead of doing an extra query\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b53a9d811413840d03840c3d9d0287e2a8aff717"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2021-07-14 06:20:00.000000000","tz":600},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2021-08-03 10:40:02.000000000","tz":600},"subject":"sharder: update shard storage_policy_index if roots changes","message":"sharder: update shard storage_policy_index if roots changes\n\nIf there is a change to the root container\u0027s storage_policy_index (e.g.\nafter it\u0027s deleted and recreated), the reconciler will kick in and\nmigrate objects (or tombstone rows) to the new/correct storage policy.\n\nTo ensure this works correctly on sharded containers, we must propagate\na policy_index change on the root to it\u0027s shards.\n\nThis patch actually can do the migration in 2 places, either form the\nsharder when we go and fetch the shard range from the root container\nin the shard container audit, the response includes the roots policy\nindex. So we can do the migration. And also in the\npost_replicate_hook of the container replicator before we do any\nreconciler enqueueing.\n\nChange-Id: Ib1d6197caa15d0974334be3d7eff24d4342ac378\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6053a65c63340cafa661fa4379ae85d035615f53"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6053a65c63340cafa661fa4379ae85d035615f53"}]},"branch":"refs/heads/master"},"04febd6590d80a3c389cd8fa4f0d8f85bed9fd30":{"kind":"REWORK","_number":15,"created":"2021-08-10 05:37:13.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/48/800748/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/48/800748/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/800748/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/800748/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/800748/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/48/800748/15"}}},"commit":{"parents":[{"commit":"63c6933f5d938bea47c365407ab7cea0b015b033","subject":"container-server: return objects of a given policy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/63c6933f5d938bea47c365407ab7cea0b015b033"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2021-07-14 06:20:00.000000000","tz":600},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2021-08-10 05:36:26.000000000","tz":600},"subject":"sharder: update shard storage_policy_index if roots changes","message":"sharder: update shard storage_policy_index if roots changes\n\nIf there is a change to the root container\u0027s storage_policy_index (e.g.\nafter it\u0027s deleted and recreated), the reconciler will kick in and\nmigrate objects (or tombstone rows) to the new/correct storage policy.\n\nTo ensure this works correctly on sharded containers, we must propagate\na policy_index change on the root to it\u0027s shards.\n\nTurns out when we go and fetch the shard range from the root container\nin the shard container audit, the response includes the roots policy\nindex. So in this patch, if it sees that the roots is different from the\nshards it\u0027ll migrate to the new storage policy. This will kick off the\nreconciler on the next container-replication run as it enqueues objects\nto reconcile.\n\nChange-Id: Ib1d6197caa15d0974334be3d7eff24d4342ac378\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/04febd6590d80a3c389cd8fa4f0d8f85bed9fd30"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/04febd6590d80a3c389cd8fa4f0d8f85bed9fd30"}]},"branch":"refs/heads/master"},"5d26f58de70a4bc2ac5c87e02433d92415927f3a":{"kind":"TRIVIAL_REBASE","_number":16,"created":"2021-08-13 14:24:41.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/48/800748/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/48/800748/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/800748/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/800748/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/800748/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/48/800748/16"}}},"commit":{"parents":[{"commit":"91e1f3047bd616db61a75e8947bc26b7365a3552","subject":"container-server: return objects of a given policy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/91e1f3047bd616db61a75e8947bc26b7365a3552"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2021-07-14 06:20:00.000000000","tz":600},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-08-13 14:13:40.000000000","tz":60},"subject":"sharder: update shard storage_policy_index if roots changes","message":"sharder: update shard storage_policy_index if roots changes\n\nIf there is a change to the root container\u0027s storage_policy_index (e.g.\nafter it\u0027s deleted and recreated), the reconciler will kick in and\nmigrate objects (or tombstone rows) to the new/correct storage policy.\n\nTo ensure this works correctly on sharded containers, we must propagate\na policy_index change on the root to it\u0027s shards.\n\nTurns out when we go and fetch the shard range from the root container\nin the shard container audit, the response includes the roots policy\nindex. So in this patch, if it sees that the roots is different from the\nshards it\u0027ll migrate to the new storage policy. This will kick off the\nreconciler on the next container-replication run as it enqueues objects\nto reconcile.\n\nChange-Id: Ib1d6197caa15d0974334be3d7eff24d4342ac378\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5d26f58de70a4bc2ac5c87e02433d92415927f3a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5d26f58de70a4bc2ac5c87e02433d92415927f3a"}]},"branch":"refs/heads/master"},"af58dff5c9256f5afec361f89863ca74c78398dd":{"kind":"REWORK","_number":17,"created":"2021-08-16 02:57:06.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/48/800748/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/48/800748/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/800748/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/800748/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/800748/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/48/800748/17"}}},"commit":{"parents":[{"commit":"91e1f3047bd616db61a75e8947bc26b7365a3552","subject":"container-server: return objects of a given policy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/91e1f3047bd616db61a75e8947bc26b7365a3552"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2021-07-14 06:20:00.000000000","tz":600},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2021-08-16 02:56:25.000000000","tz":600},"subject":"sharder: update shard storage_policy_index if roots changes","message":"sharder: update shard storage_policy_index if roots changes\n\nIf there is a change to the root container\u0027s storage_policy_index (e.g.\nafter it\u0027s deleted and recreated), the reconciler will kick in and\nmigrate objects (or tombstone rows) to the new/correct storage policy.\n\nTo ensure this works correctly on sharded containers, we must propagate\na policy_index change on the root to it\u0027s shards.\n\nTurns out when we go and fetch the shard range from the root container\nin the shard container audit, the response includes the roots policy\nindex. So in this patch, if it sees that the roots is different from the\nshards it\u0027ll migrate to the new storage policy. This will kick off the\nreconciler on the next container-replication run as it enqueues objects\nto reconcile.\n\nChange-Id: Ib1d6197caa15d0974334be3d7eff24d4342ac378\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/af58dff5c9256f5afec361f89863ca74c78398dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/af58dff5c9256f5afec361f89863ca74c78398dd"}]},"branch":"refs/heads/master"},"f41b02dd953dc5ef10be177cb3d5ced961784e65":{"kind":"TRIVIAL_REBASE","_number":18,"created":"2021-08-17 06:36:09.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/48/800748/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/48/800748/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/800748/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/800748/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/800748/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/48/800748/18"}}},"commit":{"parents":[{"commit":"3cca6a11be695a2b8fa393fad638674203a157f3","subject":"Merge \"Add a project scope read-only role to keystoneauth\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3cca6a11be695a2b8fa393fad638674203a157f3"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2021-07-14 06:20:00.000000000","tz":600},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2021-08-17 06:34:37.000000000","tz":600},"subject":"sharder: update shard storage_policy_index if roots changes","message":"sharder: update shard storage_policy_index if roots changes\n\nIf there is a change to the root container\u0027s storage_policy_index (e.g.\nafter it\u0027s deleted and recreated), the reconciler will kick in and\nmigrate objects (or tombstone rows) to the new/correct storage policy.\n\nTo ensure this works correctly on sharded containers, we must propagate\na policy_index change on the root to it\u0027s shards.\n\nTurns out when we go and fetch the shard range from the root container\nin the shard container audit, the response includes the roots policy\nindex. So in this patch, if it sees that the roots is different from the\nshards it\u0027ll migrate to the new storage policy. This will kick off the\nreconciler on the next container-replication run as it enqueues objects\nto reconcile.\n\nChange-Id: Ib1d6197caa15d0974334be3d7eff24d4342ac378\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f41b02dd953dc5ef10be177cb3d5ced961784e65"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f41b02dd953dc5ef10be177cb3d5ced961784e65"}]},"branch":"refs/heads/master"},"3e1c6746c916cc6afdbb86f95fc58982cd5b14fb":{"kind":"REWORK","_number":19,"created":"2021-08-23 04:41:55.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/48/800748/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/48/800748/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/800748/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/800748/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/800748/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/48/800748/19"}}},"commit":{"parents":[{"commit":"3cca6a11be695a2b8fa393fad638674203a157f3","subject":"Merge \"Add a project scope read-only role to keystoneauth\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3cca6a11be695a2b8fa393fad638674203a157f3"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2021-07-14 06:20:00.000000000","tz":600},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2021-08-23 04:41:01.000000000","tz":600},"subject":"sharder: update shard storage_policy_index if roots changes","message":"sharder: update shard storage_policy_index if roots changes\n\nIf there is a change to the root container\u0027s storage_policy_index (e.g.\nafter it\u0027s deleted and recreated), the reconciler will kick in and\nmigrate objects (or tombstone rows) to the new/correct storage policy.\n\nTo ensure this works correctly on sharded containers, we must propagate\na policy_index change on the root to it\u0027s shards.\n\nTurns out when we go and fetch the shard range from the root container\nin the shard container audit, the response includes the roots policy\nindex. So in this patch, if it sees that the roots is different from the\nshards it\u0027ll migrate to the new storage policy. This will kick off the\nreconciler on the next container-replication run as it enqueues objects\nto reconcile.\n\nChange-Id: Ib1d6197caa15d0974334be3d7eff24d4342ac378\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3e1c6746c916cc6afdbb86f95fc58982cd5b14fb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3e1c6746c916cc6afdbb86f95fc58982cd5b14fb"}]},"branch":"refs/heads/master"},"e743ecc81a2160df5817547bb20b5a67cf03c565":{"kind":"REWORK","_number":20,"created":"2021-08-23 09:24:05.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/48/800748/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/48/800748/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/800748/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/800748/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/800748/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/48/800748/20"}}},"commit":{"parents":[{"commit":"3cca6a11be695a2b8fa393fad638674203a157f3","subject":"Merge \"Add a project scope read-only role to keystoneauth\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3cca6a11be695a2b8fa393fad638674203a157f3"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2021-07-14 06:20:00.000000000","tz":600},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-08-23 07:36:35.000000000","tz":60},"subject":"sharder: update shard storage_policy_index if roots changes","message":"sharder: update shard storage_policy_index if roots changes\n\nIf there is a change to the root container\u0027s storage_policy_index (e.g.\nafter it\u0027s deleted and recreated), the reconciler will kick in and\nmigrate objects (or tombstone rows) to the new/correct storage policy.\n\nTo ensure this works correctly on sharded containers, we must propagate\na policy_index change on the root to it\u0027s shards.\n\nTurns out when we go and fetch the shard range from the root container\nin the shard container audit, the response includes the roots policy\nindex. So in this patch, if it sees that the roots is different from the\nshards it\u0027ll migrate to the new storage policy. This will kick off the\nreconciler on the next container-replication run as it enqueues objects\nto reconcile.\n\nChange-Id: Ib1d6197caa15d0974334be3d7eff24d4342ac378\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e743ecc81a2160df5817547bb20b5a67cf03c565"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e743ecc81a2160df5817547bb20b5a67cf03c565"}]},"branch":"refs/heads/master"},"086842a0712a2fc07918aa6e70a6a51b38065695":{"kind":"REWORK","_number":21,"created":"2024-05-09 16:29:08.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/48/800748/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/48/800748/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/800748/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/800748/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/800748/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/48/800748/21"}}},"commit":{"parents":[{"commit":"9da22bb5fedd900bbe004aa76fb163abfb769bd7","subject":"Move statsd testing to its own module","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9da22bb5fedd900bbe004aa76fb163abfb769bd7"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2021-07-14 06:20:00.000000000","tz":600},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-05-09 16:29:00.000000000","tz":-420},"subject":"sharder: update shard storage_policy_index if roots changes","message":"sharder: update shard storage_policy_index if roots changes\n\nIf there is a change to the root container\u0027s storage_policy_index (e.g.\nafter it\u0027s deleted and recreated), the reconciler will kick in and\nmigrate objects (or tombstone rows) to the new/correct storage policy.\n\nTo ensure this works correctly on sharded containers, we must propagate\na policy_index change on the root to it\u0027s shards.\n\nTurns out when we go and fetch the shard range from the root container\nin the shard container audit, the response includes the roots policy\nindex. So in this patch, if it sees that the roots is different from the\nshards it\u0027ll migrate to the new storage policy. This will kick off the\nreconciler on the next container-replication run as it enqueues objects\nto reconcile.\n\nChange-Id: Ib1d6197caa15d0974334be3d7eff24d4342ac378\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/086842a0712a2fc07918aa6e70a6a51b38065695"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/086842a0712a2fc07918aa6e70a6a51b38065695"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
