)]}'
{"id":"openstack%2Fswift~800842","triplet_id":"openstack%2Fswift~master~I21ad3c6377d15e931f737d8bec48ad6c887be52e","project":"openstack/swift","branch":"master","topic":"priority-reconciler","hashtags":[],"change_id":"I21ad3c6377d15e931f737d8bec48ad6c887be52e","subject":"reconciler: concurreny follow-up","status":"MERGED","created":"2021-07-14 21:54:54.000000000","updated":"2021-07-15 00:16:20.000000000","submitted":"2021-07-15 00:14:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"800842-priority-reconciler","meta_rev_id":"5d40fd6328756863cd351f3756189712c3a3f14c","_number":800842,"virtual_id_number":800842,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-07-15 00:14:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-07-14 22:01:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-07-14 22:01:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-14 22:01:11.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2021-07-14 22:28:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-07-14 22:55:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"651e41f632c30c45b599a3f9306b9094ef96d85c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2021-07-14 21:54:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"552636e537d09c23adc39e8fa9ca1340cf3bc480","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-07-14 22:01:11.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"abcac5e96a9a64846562e34466a6129a60cd9364","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-14 22:28:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1b8bf56d44124698b1d773372b6dfc5d : SUCCESS in 29m 24s (non-voting)\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a0dd1609ddac4424b9b0ddf0f9b00746 : SUCCESS in 29m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"87a5d92b0993ae21b04e3699145ec20f489d6c98","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-14 22:55:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2bc4ac27bd554666898d53e92200c75f : SUCCESS in 9m 37s\n- grenade https://zuul.opendev.org/t/openstack/build/c4356abdfa7740d2ac070edfb420518d : SUCCESS in 59m 11s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/53ff4dad90ca4c049aa53d52abf04d96 : SUCCESS in 53m 59s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/304353eecb2345c2b5d29892ad801a64 : SUCCESS in 9m 15s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/83b682f7540144be910f2abf63f70481 : SUCCESS in 11m 49s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/dadac8c6e5d142aab47fad13c57f816f : SUCCESS in 16m 02s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/38286ffd7dbd4b19a7ea75c72575f133 : SUCCESS in 18m 25s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/a882225fea404b24aa33bb5cd26f5b16 : SUCCESS in 15m 45s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/79c59c22457a44038bb004256fd0a1f8 : SUCCESS in 25m 06s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/0b01aa6e5621421ca21c2495e61fbffb : SUCCESS in 17m 29s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/1408b9fbf0fe4177837bea1f5c095946 : SUCCESS in 20m 03s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/0ea72d8e983f4d0686e591e2e93991e2 : SUCCESS in 18m 09s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/ac996c03d92a4e1cadbd9d85d7c5cec1 : SUCCESS in 28m 22s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/e96813aab6d14aff8927a9a713a2dd24 : SUCCESS in 20m 29s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/ddc176d9d3974ed4a5be74d4b014c270 : SUCCESS in 22m 17s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/971831188c174437824e06c1f1a09b92 : SUCCESS in 18m 38s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/962db3de2df1420fb4244b36c5919b3b : SUCCESS in 8m 14s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/2b5860bec1024153aa3b307d9065de9f : SUCCESS in 24m 15s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/9640f92f689f4eeab967c9c8a7ff81fc : SUCCESS in 49m 35s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/c4e90863364448a0900270fa0d1e812d : SUCCESS in 53m 33s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d5efbed9cad14859b6afd761138efbcc : SUCCESS in 37m 23s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/8cd9a1a54d984468a6d4fd9f6738869e : SUCCESS in 40m 17s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f7b755be886748239e0db66ff907d1b8 : SUCCESS in 16m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/303667f3bdbe498099daec0c72ab97d5 : SUCCESS in 6m 37s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9d4b71e1ba4a49ac8d1ce2e9d28167e3 : SUCCESS in 27m 56s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/124564dd9e4347f0bc8cdbaaeb291be5 : SUCCESS in 46m 06s","accounts_in_message":[],"_revision_number":1},{"id":"a3f31c5a6b80346fc4ec148376556ce34e6f179b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-14 22:55:36.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fb1e264900fd21b0f6980e0b035aea8366ccb6ed","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-15 00:14:52.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e7a507fee23d4e2caa9f09a11efb6986 : SUCCESS in 8m 52s\n- grenade https://zuul.opendev.org/t/openstack/build/3f56c616d86f4443bfdf8c5c77fbeeac : SUCCESS in 1h 19m 09s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/41e29e99dad347d6a98a98cb3eb0d737 : SUCCESS in 42m 02s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/9ce4801a7416455391800611e4b0a4fb : SUCCESS in 8m 24s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/988ea078b1af4c3a81901595a5598475 : SUCCESS in 8m 59s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/96509ff3ee0e49308aba2ca7cb3403d8 : SUCCESS in 17m 11s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/2e4e7954e3d34916b4b6f7c3913bce3b : SUCCESS in 18m 56s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/8e333c54fc0240e6b35d0e4d14f64127 : SUCCESS in 16m 00s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/d6e6ddeba2f545eb8970914848fc522e : SUCCESS in 16m 04s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/53c4739ba56d4c528d67c1866f5fc65f : SUCCESS in 16m 08s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/b52a543a61844ab384eec3ba993f9349 : SUCCESS in 16m 09s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/af6d6c062afe4d7882352cd0bf0bac48 : SUCCESS in 16m 39s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/fe23d8717a07487099cc49aa44feb5a6 : SUCCESS in 17m 04s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/f5749700ed214653b035bd08c5c82bad : SUCCESS in 16m 12s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/6b1fb23a84c44b52bb6b5b29868a12d3 : SUCCESS in 16m 21s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/7bc8a440ee08403896e82e6e198bfa21 : SUCCESS in 17m 09s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/dd1b606be88e4da4a25cad6017151439 : SUCCESS in 6m 39s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/6fc64a309cb848278395aac78ddcc501 : SUCCESS in 48m 28s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/259c0b8ee6774ce7a3e85331463a7941 : SUCCESS in 45m 34s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/0f44dce72dc0400b977301ad3ec18093 : SUCCESS in 38m 05s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/663951d9b421493b9980014be303e23e : SUCCESS in 38m 29s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f8307bbc244c47f691a3a9f944185428 : SUCCESS in 15m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/db486625dc5d476583bcf924651e4727 : SUCCESS in 6m 09s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/99520af4271e4c838f566ff8aeaac51d : SUCCESS in 21m 52s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fc7f81532055497d90da52ce2ea06187 : SUCCESS in 42m 52s","accounts_in_message":[],"_revision_number":1},{"id":"f7296307ceb70419bef84bb979e4d69c9b02982a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-15 00:14:53.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"5d40fd6328756863cd351f3756189712c3a3f14c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-15 00:16:20.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/17fb90fe84f5431d8bb45a20753d766a : SUCCESS in 1m 07s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/e36d42cb40c64a54b6598243224efe28 : SUCCESS in 50s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/8adc659d8f224a348579e616fb224e17 : SUCCESS in 45s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f29ea9d04a613a39511ce6be3f21aaae428af20d","revisions":{"f29ea9d04a613a39511ce6be3f21aaae428af20d":{"kind":"REWORK","_number":1,"created":"2021-07-14 21:54:54.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/42/800842/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/42/800842/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/800842/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/800842/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/800842/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/42/800842/1"}}},"commit":{"parents":[{"commit":"f739cadf979de25eeffc9800e1b2ecc4ace2324d","subject":"Merge \"Add concurrency to reconciler\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f739cadf979de25eeffc9800e1b2ecc4ace2324d"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2021-07-14 18:49:50.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2021-07-14 21:54:45.000000000","tz":-300},"subject":"reconciler: concurreny follow-up","message":"reconciler: concurreny follow-up\n\nRelated-Change-Id: I72e9601b58c2f20bb1294876bb39f2c78827d5f8\nChange-Id: I21ad3c6377d15e931f737d8bec48ad6c887be52e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f29ea9d04a613a39511ce6be3f21aaae428af20d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f29ea9d04a613a39511ce6be3f21aaae428af20d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
