)]}'
{"id":"openstack%2Fswift~800844","triplet_id":"openstack%2Fswift~master~Icb17346b70714cc318b216586326cd57ee27150d","project":"openstack/swift","branch":"master","hashtags":[],"change_id":"Icb17346b70714cc318b216586326cd57ee27150d","subject":"Remove support for --link-check-limit","status":"MERGED","created":"2021-07-14 21:58:18.000000000","updated":"2021-07-20 08:47:09.000000000","submitted":"2021-07-20 08:45:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"800844","meta_rev_id":"6613cf52f421399dbf136e4f941b4def239589aa","_number":800844,"virtual_id_number":800844,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-07-20 08:45:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":2,"date":"2021-07-20 07:45:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":1,"date":"2021-07-20 07:45:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-14 22:33:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-07-14 23:15:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-20 07:45:12.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"}],"messages":[{"id":"2550373b00dbceaa7f87b5f4c402100c0fca0d6e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-07-14 21:58:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2d7ff10628d947a836af25fc5b32f7972bbb92c8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-14 22:33:23.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/11729363fca34501ba5cd1228d1dcc29 : SUCCESS in 28m 42s (non-voting)\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a955bcdba27e4d588b90c5b15695e7a3 : SUCCESS in 29m 45s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6631d27a43d038e4ae63e237c7a9c188caa40b5f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-14 23:15:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/32b15f57979d4be6af67d1b2044ac0f9 : SUCCESS in 8m 42s\n- grenade https://zuul.opendev.org/t/openstack/build/20fbbf2f0aa0482393fa9c2de560ae2a : SUCCESS in 57m 02s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/0b2dddcf4b094e35a614a67a30cd6a9e : SUCCESS in 39m 29s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c64397bbabf94d408f8ef09897ae9bf5 : SUCCESS in 9m 36s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/6c174e8d81024063a25a605f14d4e3b0 : SUCCESS in 9m 47s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/0f0c3c56d899471795523f02e6a136fb : SUCCESS in 20m 14s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/885f353ee09645229cf70bce15c22061 : SUCCESS in 15m 29s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/c146b32702f543e2ba0c4b637e77d48e : SUCCESS in 17m 02s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/fb064d17f33d4db8a789be23d8e18164 : SUCCESS in 17m 02s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/f92ecfdbfce2401d985a9099ba1dd29f : SUCCESS in 20m 40s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/0c0863d763434bf484428f1951cf9ca4 : SUCCESS in 17m 43s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/bfd859212000411d99dbd0eae3dc8deb : SUCCESS in 21m 18s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/d7c1e3474b634dab8d7d8dd7cec7def3 : SUCCESS in 18m 02s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/799bcd51d0ea4f52bb188cd64f775f1c : SUCCESS in 21m 05s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/c6b21bd5d566472297ecb77c67228252 : SUCCESS in 19m 00s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/3e078d025c394226bcd6013b70e073ac : SUCCESS in 22m 12s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/439513d93a51405cb9bea63855907c5b : SUCCESS in 7m 37s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f32b01cae4724145ba835ac6802d690c : SUCCESS in 22m 55s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/fecff7cf99304a6abc4868ce7d00c382 : SUCCESS in 1h 05m 24s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/3f78ab27c42f4ee39fa7dfb4a8d99f03 : SUCCESS in 50m 30s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/efdde3d52b2144e0bdf4d6a610013111 : SUCCESS in 45m 30s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/9b9b275c209c48b48e98e6a523adc491 : SUCCESS in 38m 27s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cc92345ddf95459a8e89cd294a9f8d0e : SUCCESS in 18m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ff6ae6a62f7940e2a06f318f22f438d1 : SUCCESS in 6m 20s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/1f075092df254b4c97e1566bf37482c5 : SUCCESS in 30m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/144a00db0e11452ebdbf02a597b10535 : SUCCESS in 45m 09s","accounts_in_message":[],"_revision_number":1},{"id":"fa2ce799d9947a339b8f78f77ca38125b2aa77ec","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-07-20 07:45:12.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f6cacc42618ebdba82c10d1d82730ea18ffa1341","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-20 07:45:40.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"aafd15543132a124352f291eceaa5c30cae83a40","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-20 08:45:36.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/46161bef448846908c2b6be5be4dc2ed : SUCCESS in 9m 18s\n- grenade https://zuul.opendev.org/t/openstack/build/39c992cbc21c458a83e4330ec5e46dc1 : SUCCESS in 58m 59s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/075b562aa2534b1bb2c71ea2acf3183c : SUCCESS in 42m 34s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/41c4c49829f94315972a337b6af4a05a : SUCCESS in 8m 49s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/a11c55d4b23f4467b4d51e7483bcb637 : SUCCESS in 9m 07s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/d3456b9a92a54258b783673456500264 : SUCCESS in 18m 19s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c6a753953b0343ebae546c9d3a35543d : SUCCESS in 20m 22s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/c9548912c7a74253b89f930073b96d92 : SUCCESS in 20m 49s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/4c783c25858744deaeb0231a6a4bde94 : SUCCESS in 21m 05s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/d489c24fe3ad4763a05e3157f3460dd1 : SUCCESS in 16m 43s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/17c2579ff77a4a7e936699c6a758c849 : SUCCESS in 22m 06s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/565274d4891b4f01b4d974176850ee19 : SUCCESS in 21m 41s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/c2f646d14d714b5181d3d886193d55e6 : SUCCESS in 19m 12s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/dc74c59e051842989db323bd85351156 : SUCCESS in 16m 37s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/44e98b15b64d4fab82ea1b55418abea8 : SUCCESS in 18m 52s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/4517c921825d4f5ba08b8b27ee98d32d : SUCCESS in 21m 32s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d43214e1b903450381f39bcedb24b6f9 : SUCCESS in 6m 51s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/e5a2a4578ec0449b9372c0cde923c56c : SUCCESS in 50m 19s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/0ed9fde892d84887b1833d78efb96496 : SUCCESS in 49m 57s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/34a083fc2de34f6a86dff8ddc7568e70 : SUCCESS in 39m 58s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/9f6610f692bc4920880a6720d880c412 : SUCCESS in 37m 27s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3f244996d6b14628af101ee31ee1ea00 : SUCCESS in 17m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/20b21106bd044996b46f1088dcb68925 : SUCCESS in 6m 53s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/0762e5320bc449f2b8f1a7a5b69d6b32 : SUCCESS in 30m 50s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/314de5f17e3046d3a768239558fdfbc0 : SUCCESS in 47m 04s","accounts_in_message":[],"_revision_number":1},{"id":"c8283e71026208e5d0acc867f3dfadae5e7730ca","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-20 08:45:36.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"6613cf52f421399dbf136e4f941b4def239589aa","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-20 08:47:09.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7885a5f1eb104c2683d3615b15f20bda : SUCCESS in 1m 13s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/9efeb256bef24b3980d7c56339360db7 : SUCCESS in 54s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/763796f84e60481181799bacdde93fdd : SUCCESS in 55s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"02dd0187e9931fda01741b6b2619e53a8ff94d67","revisions":{"02dd0187e9931fda01741b6b2619e53a8ff94d67":{"kind":"REWORK","_number":1,"created":"2021-07-14 21:58:18.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/44/800844/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/44/800844/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/44/800844/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/44/800844/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/44/800844/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/44/800844/1"}}},"commit":{"parents":[{"commit":"a5fc6a82113c036be716e4ac088b17243c821e48","subject":"Merge \"reconciler: PPI aware reconciler\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a5fc6a82113c036be716e4ac088b17243c821e48"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-07-14 21:57:08.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-07-14 21:57:11.000000000","tz":-420},"subject":"Remove support for --link-check-limit","message":"Remove support for --link-check-limit\n\nThis was added after 2.27.0 and deprecated before 2.28.0; let\u0027s just\nremove it.\n\nChange-Id: Icb17346b70714cc318b216586326cd57ee27150d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/02dd0187e9931fda01741b6b2619e53a8ff94d67"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/02dd0187e9931fda01741b6b2619e53a8ff94d67"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
