)]}'
{"id":"openstack%2Fswift~800850","triplet_id":"openstack%2Fswift~master~Ie46bda37ca3f6e692ec31a4ddcd46f343fb1aeca","project":"openstack/swift","branch":"master","topic":"multi-reconciler","hashtags":[],"change_id":"Ie46bda37ca3f6e692ec31a4ddcd46f343fb1aeca","subject":"container-reconciler: support multiple processes","status":"MERGED","created":"2021-07-14 22:35:02.000000000","updated":"2021-07-22 03:25:31.000000000","submitted":"2021-07-22 03:24:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":4,"has_review_started":true,"submission_id":"800850-multi-reconciler","meta_rev_id":"b9d5615635d46d5d2592729850e0ed51716522e7","_number":800850,"virtual_id_number":800850,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-07-22 03:24:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-07-21 18:56:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-07-22 00:26:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-14 23:21:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-07-14 23:37:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-15 17:05:18.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"b0575fd4030ecc5f9a99ee5467641166fbd605e1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2021-07-14 22:35:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dd0acc7b22c32ae224e24bf8a7e26fe49af4318f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-14 23:21:58.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f699b678519f43218addae4561f51d3f : FAILURE in 35m 51s (non-voting)\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e397846379b942eb9df25af37f7a5ee1 : FAILURE in 45m 03s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"dd6514891e6f93f25eee7c2c413346684629689b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-14 23:37:04.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2fe189dd8991449fbf2aa20418a208b1 : SUCCESS in 8m 43s\n- grenade https://zuul.opendev.org/t/openstack/build/444690f04c474bb6bd3c37d3c672f2d7 : SUCCESS in 1h 00m 35s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/beb1ff2aa7804fa4ae31a8fad10c1734 : SUCCESS in 59m 07s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/baf6e9ec822f45eb9558de2a04f8f37e : SUCCESS in 9m 00s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/a675550a8ef84d9f89f82d5d1e388419 : SUCCESS in 8m 10s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/667719f703404db0ac6290defcc28b33 : FAILURE in 28m 22s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c2b9f51c730b4859b5d935fbe438704b : FAILURE in 20m 26s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/bb116e25903a4870a4875628eb633f70 : FAILURE in 27m 08s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/0389c7146eb94f06965559920e41e4e4 : FAILURE in 27m 49s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/7a648d559cf34fcfab632a7e226aa77a : FAILURE in 15m 55s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/313c178aa8014320bec5689fd9b3759d : SUCCESS in 16m 55s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/b27cdbb83e8b4c47b3753809b9d06a02 : SUCCESS in 18m 35s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/6fa56ff6f623482e9fb15ab197133057 : SUCCESS in 22m 48s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/08806a8c3f9042e2bd1193fb7eb9552e : SUCCESS in 16m 52s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/e28561342db1412dbf1c3a31588827dc : SUCCESS in 22m 02s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/3820923fe7184df09698cba88e8becf0 : SUCCESS in 18m 27s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/5a05a52f7ab14f8c960f1af26e3bda02 : SUCCESS in 6m 30s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f716d91ea999438682eaa5746c9abc60 : SUCCESS in 21m 33s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/4a818fb4e17f4a1287ab07517672b093 : SUCCESS in 59m 11s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/b569ddea52c64bc0b3a7d787a2741851 : SUCCESS in 51m 20s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/06a92753965f45a0a4971b556e42b819 : SUCCESS in 40m 52s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/33ece1778ebf45c2add7e1e27845046a : SUCCESS in 54m 50s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2482fa5a5ee94aa5ae93fde3ff6c2252 : FAILURE in 16m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/456701ef262c46fbbf023845da90b5a3 : SUCCESS in 6m 33s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b98f170425f94ab1862ae64607e941e8 : SUCCESS in 30m 52s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1c123cfc80d544688cd3bf7981d1eefa : SUCCESS in 46m 05s","accounts_in_message":[],"_revision_number":1},{"id":"c4898e5de6e7d01a6128310f54a9c59c712b6d8f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2021-07-15 16:01:19.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"80b6196bb0b2ef45471b251c4fca2f2089cc1050","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-15 16:56:19.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/78dc402dbcd846d6b035894c882f6031 : FAILURE in 49m 30s (non-voting)\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/546f87f64a484997bcde36ad66005750 : FAILURE in 48m 21s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d7d7c4aea342f434c610990e796ebd83af23dd98","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-07-15 17:05:18.000000000","message":"Patch Set 2: Code-Review+2\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"f66473e8a677c582836bf8a34536ea7e624048dd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-15 17:19:24.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/913861aefcf749bcadb7b2d15ec5ed2c : SUCCESS in 9m 51s\n- grenade https://zuul.opendev.org/t/openstack/build/7696209c641a460488d7fe3f299d4602 : SUCCESS in 1h 11m 52s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/21d9bd19aed449f79987e4eeca473133 : SUCCESS in 44m 12s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d446f03633304f148edff57e94843de2 : SUCCESS in 9m 38s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/934320d58b2d468c83c76180849ed01e : SUCCESS in 10m 01s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/e1e68689a87c416d8ef25f1feeaa1139 : SUCCESS in 15m 51s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/37c78d0c3bb643e8bc12f1c73e91cc30 : FAILURE in 18m 09s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/083667e5e4d24958ae3b7d543d20a18d : FAILURE in 19m 10s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/5774ba51d2724a20b11fbfdd9910c976 : FAILURE in 16m 28s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/73e84f4f3a6b4bb391b2defcf1b4a20a : FAILURE in 18m 37s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/9e2ba0fbc2374d0ab3be2d62c019ce3a : SUCCESS in 17m 02s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/2b3c74e05ba149999e0c14e752cd7b5e : SUCCESS in 17m 00s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/37a0af0f506040acb8ace6194cf1c1f0 : SUCCESS in 19m 08s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/9ea207a377ee4994ab47f2cf55f12b0e : SUCCESS in 16m 38s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/a85704a086154bda8dbf928fc0c463ea : SUCCESS in 18m 23s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/9b24270e601245df9b23f03c21b69ba8 : SUCCESS in 17m 37s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/288b7ac5e6c34619b56b7132defb909d : SUCCESS in 6m 55s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/293648d8784f4779a6c9e41d6d05b748 : SUCCESS in 19m 55s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/47993cb7bf8643b5a8f28859fef6cd14 : SUCCESS in 54m 07s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/9828c83a66754c589a4bb0c69976c3e9 : SUCCESS in 46m 32s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/1a8e1f0363e041b0a42f8179bcd31b8b : SUCCESS in 46m 34s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/34fe68ef6b54403caf92f7623b066baf : SUCCESS in 38m 52s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/627551a839be4c128299e6c1e01eefba : SUCCESS in 17m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6fc747c6da96415faa1f6e4cf08d4fbd : SUCCESS in 6m 21s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/91901af5c2e34d80b10869253a164400 : SUCCESS in 26m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/16cfae58785b4ecd8c7bece186e8b4ac : SUCCESS in 51m 22s","accounts_in_message":[],"_revision_number":2},{"id":"93b9fa23ba250a6f02f7c2cff2c1eb596dd69d7e","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-07-15 17:40:19.000000000","message":"Patch Set 2: Code-Review-1\n\nBah! hash_path needs bytes, and the parsed items are native strings (so unicode on py3).\n\nI was a little surprised that the centos8 probe tests still passed, but then I remembered it only has the one reconciler. Should we update the SAIO docs to have a reconciler per \"node\"?","accounts_in_message":[],"_revision_number":2},{"id":"51852671829f5174b3587e9b8373696ed38d6dd1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-07-21 18:46:06.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"b5671acb7a96b771b9415feb1f6bc75c2da51ef2","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-07-21 18:46:14.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a6ec1c78ec8a511603cc5f25dfe070176d11df47","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-07-21 18:56:05.000000000","message":"Patch Set 3: Code-Review+2\n\nLGTM -- tests were failing on the last patchset because of a bug in tests, not real code. Submitted a couple follow-ups to consider:\n\n* https://review.opendev.org/c/openstack/swift/+/801648 - Make SAIO reconciler multiprocess (so we\u0027re using it in the gate)\n* https://review.opendev.org/c/openstack/swift/+/801649 - Change default processes to 1 instead of 0","accounts_in_message":[],"_revision_number":3},{"id":"5cf4f36ff5b6675a46b5aeabf0a78b08132c61f4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-21 19:54:40.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a4961b735e78497f83dc242fac4b9f17 : SUCCESS in 8m 57s\n- grenade https://zuul.opendev.org/t/openstack/build/4734dab639304371b0a5e6d43bbaf884 : SUCCESS in 1h 02m 53s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/dab5c3d3ca284a35a26e5b3a12ef57a0 : SUCCESS in 44m 13s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d30c2656f46a4b4ab897f1ecc36a2d05 : SUCCESS in 8m 12s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/50e072aff52c41829229e6c0c99ade59 : SUCCESS in 8m 10s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/7cee91804572450f8d78b30e756f1129 : SUCCESS in 16m 17s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/72cc46d336524c88891ec3dfffe71c93 : SUCCESS in 16m 33s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/77bc981e7b2d4704a9b6c66c0ee0b8bd : SUCCESS in 15m 44s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/c8845df0206c421e9d40022e9b2deb24 : SUCCESS in 18m 02s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/086572f8acea4196b6094c569f79d281 : SUCCESS in 15m 17s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/6fd337e63b654ee7bfcea1c3952f73a8 : SUCCESS in 18m 06s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/ad1a944ec54a49bb8054bc8c9298be7a : SUCCESS in 18m 52s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/2e5a84021ae148f8ac78db6f44c1379d : SUCCESS in 17m 00s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/1fbc9a09e4b24fdb9ba5ba6befc01ecb : SUCCESS in 16m 49s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/bbafdea4b48b4de59b32d816c04ed3f8 : SUCCESS in 16m 54s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/8e27b5021b084ce88985951d5397787d : SUCCESS in 22m 37s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/2d4868d13b404655bc1c3a2d71407e8b : SUCCESS in 6m 46s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/4ce1e195f2254a78ac813c03dcab77e1 : SUCCESS in 19m 59s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/bb17b6462fa04333894b5e17988737e5 : SUCCESS in 1h 03m 12s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/ddfb2f6b7e274a2ba1fd0b4fae56af64 : SUCCESS in 50m 30s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a26a2517be3e4a3ab336b790e1506438 : SUCCESS in 50m 14s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/31f14c67263343b9b76061005377df8a : SUCCESS in 37m 31s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/967499ededc34d60865d9f52b50c8967 : SUCCESS in 16m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b64f2ec9989949aaae89e2b9259eef69 : SUCCESS in 5m 40s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/fe111698b0d74d40bbe695b02a865fc3 : FAILURE in 25m 33s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/07ae50021f074000a95191fb52472171 : SUCCESS in 45m 15s","accounts_in_message":[],"_revision_number":3},{"id":"3d1072f7b7ec0151f2ee2450686feb26632137da","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-21 20:20:56.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f4daaa74839d4b349c3c7f4296751f28 : SUCCESS in 47m 11s (non-voting)\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/31c251d7f9054d4196dff33b1c97c3e0 : SUCCESS in 59m 23s (non-voting)\n- swift-probetests-centos-8-arm64 https://zuul.opendev.org/t/openstack/build/c2715c6a664840da85436022fa3bd3f8 : SUCCESS in 1h 32m 35s (non-voting)\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/db4f9fef3bd04b30aff39978c6c6b182 : SUCCESS in 35m 43s (non-voting)\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/d055e60ee84b4b3f8884f575c49e69b3 : SUCCESS in 48m 29s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"16a0ceec646d60d5439cbc42460760e9cdde72ca","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-07-21 22:09:35.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"5ce777c4d4723394a7b76304deb267b11af87313","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-07-22 00:26:08.000000000","message":"Patch Set 3: Workflow+1\n\nI\u0027m less and less convinced of the wisdom of my follow-ups.","accounts_in_message":[],"_revision_number":3},{"id":"629ca50e8289d5ceea270a2baf3c021f68b7fbc9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-22 00:26:40.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c42fedac0a5b48dba4fd7ab0af3439f1 : SUCCESS in 9m 52s\n- grenade https://zuul.opendev.org/t/openstack/build/396fc3d6ac3c4817a333f0d35a07c7a5 : SUCCESS in 1h 07m 31s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c250c4d20cd8406ea3c1a98e7c386261 : SUCCESS in 49m 31s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ea31fd568fd84c108a316942d38204df : SUCCESS in 8m 25s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/34b34913821842ccb41dd954810584c5 : SUCCESS in 8m 48s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/441772da355a4ebb99d1675370b4c470 : SUCCESS in 17m 43s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/6929b39d9fd54714b92803d9d6631e69 : SUCCESS in 17m 50s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/d3fb6ada6d9a4bd7887ec1a4f245a2a3 : SUCCESS in 17m 22s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/5bccb8900958401cb75c8ee47e7ea5d1 : SUCCESS in 16m 39s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/8156cdb50ab9444d94cb81806f889f69 : SUCCESS in 18m 06s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/fb7ec0ab9c71428c95e19a75824a4eaf : SUCCESS in 20m 01s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/30293946b52d4a7eb27ae66a30d09930 : SUCCESS in 18m 01s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/c17c822a239844ca95c3ebf659048926 : SUCCESS in 18m 11s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/0123c36139b442aebd768ef3ac89fb70 : SUCCESS in 17m 22s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/7f7a08a91cb24f29800f5d039438729b : SUCCESS in 16m 59s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/05d7a778a9664b9896d8d4d13d3c7a01 : SUCCESS in 19m 21s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/bad0aa521888431b87202497f6308fec : SUCCESS in 7m 05s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/0ca0c8127dea46889d7097ba1931be69 : SUCCESS in 23m 02s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/6852d91f0ad346bc94f7f04f3d445fee : SUCCESS in 46m 59s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/dfda4bcd7f00416dba09c59eabc3fe86 : SUCCESS in 54m 09s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/04340d2c677d429cbbfc53baca4e2312 : SUCCESS in 42m 21s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b2c200c5b53e4256ae032608ebe3a2da : SUCCESS in 43m 39s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1750d251667042b19e90d95d62fa962d : SUCCESS in 16m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9d8e977003da49f1b647ac7f79dffb95 : SUCCESS in 6m 49s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/8fcdb7c0da014d1984f3d41d7d0e885d : SUCCESS in 21m 53s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5100419957034442886390e4bb771ea8 : SUCCESS in 45m 13s","accounts_in_message":[],"_revision_number":3},{"id":"3449415a5e07116e55b4115f2800651c3791c462","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-22 00:27:13.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"e84022119a0197b92b25b058e5c9bd335e8d7ca6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-22 00:35:34.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/739f593f927b4ea7b3531ac796c84458 : SUCCESS in 30m 53s (non-voting)\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1cffc28477f34e4e8463a036c1487613 : SUCCESS in 38m 32s (non-voting)\n- swift-probetests-centos-8-arm64 https://zuul.opendev.org/t/openstack/build/0a86ca3d434d43349cf8fc3a631b3c53 : SUCCESS in 1h 12m 40s (non-voting)\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/a98322dc9c8e4e76a1dca15616909e27 : SUCCESS in 52m 57s (non-voting)\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/131cfb9b08c24313a8a506b2a236453d : SUCCESS in 30m 21s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"73f6c162804ddd47da2724172d14ae080675cb4a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-22 03:24:12.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a5be114f70af414fb1474842a51c7940 : SUCCESS in 13m 23s\n- grenade https://zuul.opendev.org/t/openstack/build/145ba4053ac04c41b8cd77573194f6e1 : SUCCESS in 57m 10s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/bb44f0f4c21c4d8d8c6d3affad1933fd : SUCCESS in 38m 21s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/7265209140ee43a38dbdce79c81c516e : SUCCESS in 10m 00s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/5151a3f03f4a4f3f94d581125601fbed : SUCCESS in 9m 15s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/5ba1315881a54fb38c11075b88402d74 : SUCCESS in 18m 43s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/debbfd97098344aebd9b53470bead1ca : SUCCESS in 16m 11s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/3f5eb97d4936445f8ef15d96e8e08116 : SUCCESS in 17m 46s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/bde538d3d76e4e568a19fa9657b1cf95 : SUCCESS in 14m 29s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/4fcad9e0b99042aa964136431185608b : SUCCESS in 16m 02s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/ad9a8e6c8bf24bcea86be28c4e6ea90b : SUCCESS in 20m 41s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/f0a6753c3dff444ab110643b8e181b4b : SUCCESS in 21m 07s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/9b9a1237dcef4cb5a474d1cb70f63841 : SUCCESS in 16m 27s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/2539aeacf78b4e1a8717fef7ac5d68a1 : SUCCESS in 42m 58s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/e4116fcc155b4d17839e817d290cfd9b : SUCCESS in 16m 23s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/4d8bd57c46fe4b0a9f3ec0a6446b7ecd : SUCCESS in 25m 25s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/4a4edb3eb5934f79a3bb9c86e184566d : SUCCESS in 6m 04s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/b12a770cbeb64f15ba1f09b39d2d85e7 : SUCCESS in 49m 25s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/646a3dc6dcd74822858b820463179399 : SUCCESS in 51m 05s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b4f312542a374ef0a8021170c6f10712 : SUCCESS in 37m 24s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a5603b6319d74f7d89c3830ccc0dfb09 : SUCCESS in 45m 57s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/54df11cd5c9440d986f8d0a0bc7fd8c9 : SUCCESS in 15m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bc4e634b90e64cb4b24926bab335f3a4 : SUCCESS in 6m 15s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/017737a349754917b84068920018372f : SUCCESS in 21m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a804c69de4e84f15945f84b81e5a366e : SUCCESS in 38m 18s","accounts_in_message":[],"_revision_number":3},{"id":"c6f73d556927e526ba0ff05f531f6a0671a1a4f1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-22 03:24:12.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"b9d5615635d46d5d2592729850e0ed51716522e7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-22 03:25:31.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2758f530e3f64856850ce8ccf97fc39a : SUCCESS in 59s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/bf6c2bc71a91459e8d63e51828528bd2 : SUCCESS in 44s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/7aee1502f7ef45678b014eebd70edebc : SUCCESS in 44s (non-voting)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"eb969fdeea8f529aec733607ca5f6edb4e7661e7","revisions":{"823c6c3cb5b4acd63855d39097deb2ecf674c876":{"kind":"REWORK","_number":1,"created":"2021-07-14 22:35:02.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/50/800850/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/50/800850/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/800850/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/800850/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/800850/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/50/800850/1"}}},"commit":{"parents":[{"commit":"f29ea9d04a613a39511ce6be3f21aaae428af20d","subject":"reconciler: concurreny follow-up","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f29ea9d04a613a39511ce6be3f21aaae428af20d"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2021-07-14 22:31:05.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2021-07-14 22:31:05.000000000","tz":-300},"subject":"container-reconciler: support multiple processes","message":"container-reconciler: support multiple processes\n\nThis follows the same pattern of configuration used in the\nobject-expirer.  When the container-recociler has a configuration value\nfor processes it expects that many instances of the reconciler to be\nconfigured with a process value from [0, processes).\n\nChange-Id: Ie46bda37ca3f6e692ec31a4ddcd46f343fb1aeca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/823c6c3cb5b4acd63855d39097deb2ecf674c876"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/823c6c3cb5b4acd63855d39097deb2ecf674c876"}]},"branch":"refs/heads/master"},"92d2fd16a0ad2fcec7b947084521674fa71abc37":{"kind":"REWORK","_number":2,"created":"2021-07-15 16:01:19.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/50/800850/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/50/800850/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/800850/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/800850/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/800850/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/50/800850/2"}}},"commit":{"parents":[{"commit":"f29ea9d04a613a39511ce6be3f21aaae428af20d","subject":"reconciler: concurreny follow-up","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f29ea9d04a613a39511ce6be3f21aaae428af20d"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2021-07-14 22:31:05.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2021-07-15 16:00:55.000000000","tz":-300},"subject":"container-reconciler: support multiple processes","message":"container-reconciler: support multiple processes\n\nThis follows the same pattern of configuration used in the\nobject-expirer.  When the container-recociler has a configuration value\nfor processes it expects that many instances of the reconciler to be\nconfigured with a process value from [0, processes).\n\nChange-Id: Ie46bda37ca3f6e692ec31a4ddcd46f343fb1aeca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/92d2fd16a0ad2fcec7b947084521674fa71abc37"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/92d2fd16a0ad2fcec7b947084521674fa71abc37"}]},"branch":"refs/heads/master"},"eb969fdeea8f529aec733607ca5f6edb4e7661e7":{"kind":"REWORK","_number":3,"created":"2021-07-21 18:46:06.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/50/800850/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/50/800850/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/800850/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/800850/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/800850/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/50/800850/3"}}},"commit":{"parents":[{"commit":"f29ea9d04a613a39511ce6be3f21aaae428af20d","subject":"reconciler: concurreny follow-up","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f29ea9d04a613a39511ce6be3f21aaae428af20d"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2021-07-14 22:31:05.000000000","tz":-300},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-07-21 18:45:01.000000000","tz":-420},"subject":"container-reconciler: support multiple processes","message":"container-reconciler: support multiple processes\n\nThis follows the same pattern of configuration used in the\nobject-expirer.  When the container-recociler has a configuration value\nfor processes it expects that many instances of the reconciler to be\nconfigured with a process value from [0, processes).\n\nChange-Id: Ie46bda37ca3f6e692ec31a4ddcd46f343fb1aeca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/eb969fdeea8f529aec733607ca5f6edb4e7661e7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/eb969fdeea8f529aec733607ca5f6edb4e7661e7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
