)]}'
{"id":"openstack%2Fswift~800974","triplet_id":"openstack%2Fswift~master~I5f3318a44af64b77a63713e6ff8d0fd3b6144f13","project":"openstack/swift","branch":"master","topic":"p-diskfile-commit-window","hashtags":[],"change_id":"I5f3318a44af64b77a63713e6ff8d0fd3b6144f13","subject":"diskfile: don\u0027t remove recently written non-durables","status":"MERGED","created":"2021-07-15 17:10:15.000000000","updated":"2021-07-20 00:59:34.000000000","submitted":"2021-07-20 00:58:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":20,"unresolved_comment_count":11,"has_review_started":true,"submission_id":"800974-p-diskfile-commit-window","meta_rev_id":"02bb81b8fa134cbd89f9640e5de0b4c987fc6967","_number":800974,"virtual_id_number":800974,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-07-20 00:58:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-07-19 20:44:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-07-19 20:44:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-15 17:32:16.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2021-07-15 18:04:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-07-15 18:26:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-15 21:08:56.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"}],"messages":[{"id":"e1fbd9619544b3db3340a292d89174f386dd2de9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-07-15 17:10:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"26779ad0418eea5e72e9d4525fdc8c4761b87781","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-07-15 17:32:16.000000000","message":"Patch Set 1: Code-Review+1\n\n(3 comments)\n\nYeah, seems reasonable. There\u0027s an obvious place to turn it into a config option, but hopefully 60s is enough for most object-servers.\n\nMight be worth an explicit\n\n Related-Change: I0d519ebaaade35249fb7b17bd5f419ffdaa616c0\n\nin the commit message; we could probably even replace nondurable_purge_delay with a single reclaim_delay (or maybe commit_window?) option that covers both aspects.","accounts_in_message":[],"_revision_number":1},{"id":"eff41697d8399cf6b518a17c6119f43df0b6c319","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-15 18:04:19.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3f9ef069ab194ca68f0df17c7d9510ec : FAILURE in 47m 54s (non-voting)\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9b099912aea043c593a638f4c10aa39e : FAILURE in 43m 38s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"59b40fa46bcdbfd4bf5e8ac88da6433c4222e847","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-15 18:26:19.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4fb3258b14d24f5e88dc31030076773b : SUCCESS in 9m 23s\n- grenade https://zuul.opendev.org/t/openstack/build/38ec3c5211d1417c87c674f36968b359 : SUCCESS in 1h 06m 02s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/00471605d0f64accbb5489a46a5beb31 : SUCCESS in 46m 45s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/6eb3ebc56479480fa1ef99d5127bdbaa : SUCCESS in 8m 16s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/0031dcdf4ce74d299487b60214d7bdd1 : SUCCESS in 9m 00s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/f8fce61f45514006b4326407896a06ff : FAILURE in 17m 07s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/89e2f0b39a3f46ea99fbaf8751618dc0 : FAILURE in 17m 00s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/a55a63b6dee543cc867ea914fef3d5cb : FAILURE in 19m 29s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/8f269dc0c3c54a42b4a5602ec9417cc5 : FAILURE in 16m 53s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/01893e3f477640b389bf3d1634a8a9f9 : FAILURE in 18m 43s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/d050dc37fe1c440cb788c6ddfbe6781b : SUCCESS in 18m 35s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/eda125afc1654f4fb76c43cd9fcab5d2 : SUCCESS in 19m 15s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/78a8551cc78c4d62ae59c1fd75442b56 : SUCCESS in 19m 15s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/a72d131ade6a472c83fd360bfa289a41 : SUCCESS in 16m 33s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/45802f04cfd6497f9ffd0dd01fbb7368 : SUCCESS in 18m 00s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/ea57745d1bf8475b9137584fcd7f2945 : SUCCESS in 18m 40s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e4d799f4e6a1459ebb074bdd54526243 : SUCCESS in 6m 52s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/7ceca62a74fb4818b68df11c7daa026b : SUCCESS in 21m 00s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/c6c0b87dba144f25bb73b42900667794 : SUCCESS in 55m 30s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/5e05e8da7e194348aedd4ef9ababae6b : SUCCESS in 54m 41s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/1bd1ed9acf5b4ef59beec263cfbfdbbb : SUCCESS in 45m 39s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d227016354684cf2a89db3b78b108cf5 : SUCCESS in 41m 47s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c41dd710ea0248ea8e530635b5f5e33b : FAILURE in 17m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f74c852844894c9eb4f7612fffb7abc0 : SUCCESS in 6m 42s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/0600e36532a444a5b751d55dbde33e02 : SUCCESS in 24m 21s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a3679e7f4a0a44f3a2ffe9988d52deb1 : SUCCESS in 51m 47s","accounts_in_message":[],"_revision_number":1},{"id":"070636c735f28026b42273ee1b7d68f5fa1fdaa1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-07-15 18:31:56.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"87bdf33d520f08375b4d3e3e5f52b94bb3c9bec3","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-07-15 18:38:38.000000000","message":"Patch Set 2: Code-Review-1\n\n(4 comments)\n\nI\u0027d like to find the right bug to link or write up a new one.\n\nI\u0027d also like to understand why Kota wrote on this older patch https://review.opendev.org/c/openstack/swift/+/289756:\n\n\"We shouldn\u0027t use st_mtime because it may cause flaky statement due to replication timing. i.e. if each node has too different st_mtime for same .data/.durable, it may cause \"reclaim-\u003ereplication-\u003ereclaim-\u003ereplication...\" loop.\"","accounts_in_message":[],"_revision_number":2},{"id":"5a8f0f82ee97a8e275b01f89a05b4d73b22d9b4e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-15 19:33:08.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/405a5cd8a4bb4ed38c875db4f533603c : SUCCESS in 10m 34s\n- grenade https://zuul.opendev.org/t/openstack/build/b4bbba503ba642dd8a9739efdafc2973 : SUCCESS in 1h 00m 12s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/94ee50fddc104cdca23800bc5ff3194e : SUCCESS in 48m 07s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/0a85350ae85649b98cfd1a242f3e2588 : SUCCESS in 7m 52s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/5a21fe04d39a49bc8e1db7f55fb42334 : SUCCESS in 8m 39s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/46f01c89eca643209d89112cf9d61c3b : SUCCESS in 16m 28s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/622868378e2b4968b81892c4e5de0c3d : SUCCESS in 18m 00s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/c30dadde457d4fa28296798029b2b4a3 : SUCCESS in 18m 01s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/c51bbdd4b7254f568616df76468af95b : SUCCESS in 16m 26s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/a0c80c4551284b4da555f7f2693fc5ef : SUCCESS in 16m 24s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/f0666a14f35743d18188df41ce75724d : SUCCESS in 18m 18s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/7b09b61d4f0f4b7d98f6674311b82010 : SUCCESS in 19m 01s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/131d0414f7ad4b46b0fa169cbb7bb57a : SUCCESS in 19m 48s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/4863c7d9913a448ab150669ce7801ce1 : SUCCESS in 16m 33s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/5bfc9807f7054f5685c99089c711e1c4 : SUCCESS in 18m 40s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/f6d92c4abb5447bc826f854f7f307e7b : SUCCESS in 19m 23s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/fdd8c2aeb1a54befbf9664cdebd6e6ae : SUCCESS in 6m 22s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/af920f9a0a1a4ea8a8904f81780bf199 : SUCCESS in 24m 01s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/a4a59f247f91449f9942e7baaf6e3a49 : SUCCESS in 53m 59s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/0e5ecb00384e4d4a9f0f7f4cf8fdd481 : SUCCESS in 48m 04s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f97f36116a83460887be578b3f968e3c : SUCCESS in 42m 15s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/53524b3292b04604be9b73e69ce87737 : SUCCESS in 39m 39s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6bb7d00d7d7d466dbf29e75e1aae2f1a : SUCCESS in 17m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e06a5646f00243c3b83dbfae9c254ddd : SUCCESS in 6m 16s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b9b9573bed9a49c7bd9c9b33864e6840 : SUCCESS in 28m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/406ebc227f354d2db1aaf342a57dfbd5 : SUCCESS in 49m 43s","accounts_in_message":[],"_revision_number":2},{"id":"4a86761388fcaac344d63134c60d3c3e71141467","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-15 19:53:53.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e094c4ef5690446085f11daca9cd96cd : SUCCESS in 44m 39s (non-voting)\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/733d1ef1f9064c2aab2568efd410f261 : SUCCESS in 1h 17m 31s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d50065bc6ffbb1b50efc597298282b52c953e153","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2021-07-15 21:08:56.000000000","message":"Patch Set 2: Code-Review+1\n\n(3 comments)\n\nthanks for putting together such great tests Al!!!","accounts_in_message":[],"_revision_number":2},{"id":"f7e5c6386598e1a87bbb1db9f73d08171f508a85","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-07-16 10:18:43.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"a9a7c507079e382d18e557dabd50027541166864","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-16 11:17:50.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1fede86c5e7848988f6c67f0b89c7b9c : SUCCESS in 8m 32s\n- grenade https://zuul.opendev.org/t/openstack/build/9b821d5efeb8494da3c6b73ca59abc0d : SUCCESS in 57m 56s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/3834413f6a964860b7ad2f7b583c9570 : SUCCESS in 50m 29s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/6965102dfe634da0a9ac6dd79f462191 : SUCCESS in 8m 20s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/14372cfdad3d4d65a93ea4f28678c38b : SUCCESS in 8m 19s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/76d720efc7014930ad0bdc18bea14ece : SUCCESS in 17m 04s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/862f00d1b1b045c7a1746d185272e553 : SUCCESS in 15m 48s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/54d20a2e04a74b5686e8afb9b4f599de : SUCCESS in 17m 35s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/1515440fcb6b4d49b1475793bb564e5e : SUCCESS in 15m 56s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/64345a0d837544cdbbbdf6fd088419fe : SUCCESS in 20m 07s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/ee613f1cc4174d5da14ff204f2e136b7 : SUCCESS in 18m 03s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/50420130f71a4819859d8f6e8305d11c : SUCCESS in 21m 04s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/99b5ab11cbc64a5e80e043762e09eae2 : SUCCESS in 17m 46s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/9d21fdbf69774d3dbae5ddbb20f0abf6 : SUCCESS in 20m 58s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/6eeb5ed0af1b4e1a8d6413f83c8abbd7 : SUCCESS in 20m 48s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/27625c68b2264da9856af8a3a86572e7 : SUCCESS in 22m 16s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/4137cc5ee22041fe8899694e4c7e1b34 : SUCCESS in 8m 28s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/4c97c7aedb4f4e14a1ad311803efb222 : SUCCESS in 25m 02s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/007401433572489d9dd2256fc261a724 : SUCCESS in 47m 11s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/690c3874d59a46bb8b01dab80ed8d3b4 : SUCCESS in 52m 27s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f6937557ec4a46b79be9577095ffff11 : SUCCESS in 36m 16s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b764e151d1884f27923664ae33854801 : SUCCESS in 37m 01s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6a263796d7b744cdbaf683be34c9f3a5 : SUCCESS in 15m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bacb590545ff48aabee692527ab29ce8 : SUCCESS in 6m 37s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/90c96647557a4e49a7d899fd361f27c3 : SUCCESS in 25m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f4944b9e2ee94263bd4415b0ce5cb16f : SUCCESS in 43m 37s","accounts_in_message":[],"_revision_number":3},{"id":"331cdfdbb7346559f562d237d1ae26fa7bfb278c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-16 11:40:39.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6d9675002fe448db98b21b42df58f33f : SUCCESS in 28m 37s (non-voting)\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/cd10ab25f812427b9ded0ac4f16e8cfc : SUCCESS in 29m 54s (non-voting)\n- swift-probetests-centos-8-arm64 https://zuul.opendev.org/t/openstack/build/7a36466d0d2840c28601c249f67c2f3d : SUCCESS in 1h 15m 42s (non-voting)\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/d67498cc69bf46fb9d3fc63e2d56ee51 : SUCCESS in 30m 34s (non-voting)\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/d90e8247740d4e039b3b5f63b05f500b : SUCCESS in 29m 45s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"00254a19a18f7a604d1a623db505a2a0ad2ced82","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-07-16 12:47:20.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"98421e9c09336f89b7d12ecef18c75b7eaaa13ff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-16 14:12:00.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df0e8f6ef1d242bd93ea2889825d66eb : SUCCESS in 9m 30s\n- grenade https://zuul.opendev.org/t/openstack/build/c47fbf1b98874cc69aa5232d0f32ab6e : SUCCESS in 1h 22m 16s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/49a11475da494150a3cd4290707bd7c6 : SUCCESS in 47m 07s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/234b318efd4847abaeb1e4c86e8d6e85 : SUCCESS in 10m 22s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/f884a5caeb254d418c9091d9503e8acb : SUCCESS in 9m 53s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/53a01ea62f1c40fb91778b4dacaffa35 : SUCCESS in 18m 26s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/387407cdcdfa4182bab6b9172650ad47 : SUCCESS in 18m 04s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/9cca90e766eb4fae96dbc455409f9df4 : SUCCESS in 18m 31s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/0dd0eeb026ba462c99680451ddaf11d5 : SUCCESS in 18m 23s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/2855fdf9f3984d11aa20cd80ece38505 : SUCCESS in 17m 54s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/2727990a36974fa6a0f48c7ec69777bf : SUCCESS in 18m 31s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/e1b13b015580446fa75059eaadc15ef1 : SUCCESS in 19m 13s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/ce850845b7634571a88ea13ef74013c2 : SUCCESS in 20m 15s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/520bc327dcd7408a9b83dd67a7b2fde7 : SUCCESS in 18m 47s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/a81128d8bfa1467d8f4d7bad31e0beeb : SUCCESS in 18m 39s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/af535371b0b541eca621bb980c55d830 : SUCCESS in 19m 03s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/8fef62db05324f149195981d47e3ecd2 : SUCCESS in 13m 43s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/155a12dafd0449d594af06eeda46003d : SUCCESS in 25m 18s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/2d8eda0f67364024bc796c1918547b87 : SUCCESS in 51m 20s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/caeb2b6b389a46fea773f08bf704fe93 : SUCCESS in 53m 34s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e043dcc8342e42c9bea1cd7061a1df7d : SUCCESS in 47m 25s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d5afe728bd2e471e9fb2f4ed4b7ea2ba : SUCCESS in 47m 28s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/306ca8c8d5c64893bee56fe0a4af8463 : SUCCESS in 17m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/19ce5b8e53ba4de983231d67dd64da55 : SUCCESS in 7m 22s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/aa8202bed4224418af069d7700ed3685 : FAILURE in 33m 57s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6adfe0b383d34a75a2b0fbf42196c84a : SUCCESS in 55m 32s","accounts_in_message":[],"_revision_number":4},{"id":"059f6a06e4269fffd00132014d1eaebb4d8220ab","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-16 14:31:23.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b165e6ff0b904bc6a49050f07608b1bc : SUCCESS in 33m 42s (non-voting)\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/da795f432a1d469d8644ab2a3e265e2a : SUCCESS in 44m 35s (non-voting)\n- swift-probetests-centos-8-arm64 https://zuul.opendev.org/t/openstack/build/c364ff033c904f52a4345478088a858e : SUCCESS in 1h 41m 32s (non-voting)\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/d0962719340b48d2906de132f309c947 : SUCCESS in 42m 52s (non-voting)\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/c2508e3e9d694df7ad79756dd18b305d : SUCCESS in 54m 47s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"883b3f72b5e6c26dd229a477b26ca698bb1fa2aa","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-07-16 16:23:24.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"2ea648c8aa437343f66d3f799adb34333258de8e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-16 17:29:34.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d738e7f64c8420d9ab29c07a3e04d0e : SUCCESS in 9m 51s\n- grenade https://zuul.opendev.org/t/openstack/build/bc8698f4c6f148df925ab7052cff8ec0 : SUCCESS in 1h 05m 38s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/ba535dae103b45ed9272c4d399a4205d : SUCCESS in 43m 46s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ff550796852d44daa69775a39f994a67 : SUCCESS in 9m 28s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/6578e473cc4945ffac3725cc69566ce4 : SUCCESS in 8m 29s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/44ca9483fc334b5f9ffc13b6776eb841 : SUCCESS in 24m 25s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d27339728ddf439fb58d6bd2b68c476b : SUCCESS in 17m 08s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/6a01e6be7f9e4f47858bee55bc56b0eb : SUCCESS in 16m 55s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/92d39ac0d341432da4d90b83da875d5c : SUCCESS in 17m 16s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/c044a7bcf4bf4a97a1da8ab07dee58ed : SUCCESS in 18m 26s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/21d551e3ab9b4c34b139e50a74805b1c : SUCCESS in 17m 58s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/8ddf7b07fe714b0b90cdf54e92e9abe7 : SUCCESS in 19m 47s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/6f78d300fa5a42a3b526cbd57a8d4001 : SUCCESS in 19m 07s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/8907bf8b53d14efdae402d61b28080ff : SUCCESS in 19m 20s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/1bd0ff7c375749ecb993491690d32c0e : SUCCESS in 17m 30s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/8686a2f505c54e80b48f91754b605dc9 : SUCCESS in 20m 14s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/05f3b7d6cf9f4894b74fa3004adf4b75 : SUCCESS in 7m 32s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/8a3cfd0dedf24901b4d8a61dff58ac22 : SUCCESS in 21m 51s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/b812553664cf465a961d09cec7f86ae5 : SUCCESS in 54m 42s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/c93064f41eb246fb8969b4693c6e5130 : SUCCESS in 53m 33s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e2e8d2e662084b5c948e737945e5e7c8 : SUCCESS in 49m 03s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/99835f3a83164cea8611d1163a15418b : SUCCESS in 39m 36s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a72116e866a9417caf30b51bd6da8442 : SUCCESS in 18m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b0d5e791aa6d49cdb90b5a88469a0e75 : SUCCESS in 6m 22s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9186708d0d2144cdbe7ac15ac7b93a52 : SUCCESS in 27m 05s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3da3b6227d814766af595197c6247b71 : SUCCESS in 50m 36s","accounts_in_message":[],"_revision_number":4},{"id":"6e27e0825cb7b58227579368d5a6e122e40b718f","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-07-16 17:46:30.000000000","message":"Patch Set 4: Code-Review+1\n\n(4 comments)\n\nYeah, this all seems reasonable.\n\nI think Kota\u0027s concern with https://review.opendev.org/c/openstack/swift/+/289756 was based on re-using reclaim_age. Operators ought to configure reclaim_age \u003e reconstructor cycle time (and likely something like 10-100x cycle time) -- if commit_window \u003d reclaim_age, you run into a situation where reconstructors complete, operator sees it\u0027s safe to rebalance (based either purely on cycle time or considering handoff counts), new ring gets pushed out, so the non-durable/lonely-meta gets moved again and (literally :P) gets a new lease on life. If the cluster\u0027s turning over fast enough relative to reclaim_age, it\u0027ll *never* get deleted.\n\nI definitely like the separate commit_window configurable, and we might want to call out that you want commit_window \u003c\u003c reclaim_age in the sample conf/docs. Speaking of, I suppose we probably ought to update doc/source/config/object_server_config.rst and doc/manpages/object-server.conf.5","accounts_in_message":[],"_revision_number":4},{"id":"92277fda8845bdccdf427dac7132d8e81beb616c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-16 18:33:21.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7ed0036049f746e7861c3f3601bd452c : SUCCESS in 37m 22s (non-voting)\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4a300f09a8844825beea637373e4d760 : SUCCESS in 38m 13s (non-voting)\n- swift-probetests-centos-8-arm64 https://zuul.opendev.org/t/openstack/build/15447c0bb8794c13a17811eae86b9f74 : TIMED_OUT in 2h 01m 35s (non-voting)\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/04d3d271a29f41b1974508bb54e872f9 : SUCCESS in 35m 39s (non-voting)\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/879677b1015a41e7b3bd588e542ed9bc : SUCCESS in 47m 23s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"92321bad589667a396203499c7a59c91ddd0fc99","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-07-19 10:00:38.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"6aba297a000b2de4b685d4edc427097cf93132a3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-07-19 10:06:09.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"a9ccd80b530d6e64755158eac9b462ef41776137","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-19 11:14:48.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/447acbae74e245d3805b9ef9f2c1a8fb : SUCCESS in 9m 45s\n- grenade https://zuul.opendev.org/t/openstack/build/ca0830d8e6cd4b26a806c60d48f964fc : SUCCESS in 1h 03m 10s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/86ebbc8df9884f9db0138d20ca564ad5 : SUCCESS in 36m 41s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/78746ada8f20484dae26709662c1f014 : SUCCESS in 8m 16s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/6d06007b34434a6ba89e63b6d68a2f69 : SUCCESS in 8m 46s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/be79d22024ee4191ba46842e995a75e8 : SUCCESS in 18m 18s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/2b3a5c193a964bd7ae3d961f43df66fd : SUCCESS in 18m 36s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/6087400936ac447cbe1ca813fced6e93 : SUCCESS in 20m 03s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/a78707d96e1b48688d24dcd1376cbcd5 : SUCCESS in 16m 31s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/e5da6c41df0d496ab843d0bea21bcdc3 : SUCCESS in 17m 28s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/a7c5f808f0434a98a3d0f6016b89177e : SUCCESS in 19m 24s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/8744e88f67364e5e9b9b8a20ee9d01ab : SUCCESS in 17m 32s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/7c93dec5e2de4371a90e1f199230b3d3 : SUCCESS in 20m 45s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/c9cabcccb9b34f5eb9aad4a80cbf491d : SUCCESS in 44m 34s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/3c6ac6cf192d41cdb513fdc2a1adf06e : SUCCESS in 16m 33s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/7b209f1e53c34a168dd58ccb7ff72c37 : SUCCESS in 21m 26s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/86520a63975e410484dfebe3d195367b : SUCCESS in 7m 51s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/658d4887aae34260acc4a5452be1f7a9 : SUCCESS in 25m 23s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/27424967c4a24838a577b0624498350e : SUCCESS in 46m 41s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/a010343a85d44c53a05340864f71875f : SUCCESS in 50m 46s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/0b2b43edac054ab2be51f72b479e7aff : SUCCESS in 35m 16s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/7c75d084182644529ea7ef9980c619ae : SUCCESS in 40m 00s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/791145ed310b457c97c4e30d3d484f14 : SUCCESS in 15m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af27529eecae4f6da0fe1d0166b692f7 : SUCCESS in 6m 12s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/3b45bd30d9854e39a024acc780710879 : FAILURE in 30m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/45939a68961a4963931718fd5144765a : SUCCESS in 44m 34s","accounts_in_message":[],"_revision_number":5},{"id":"7f30e1961ed0623802b9d959229c3a92c8ebb761","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-19 11:49:52.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/49964448c5ad4403bb62ff0e7a64291f : SUCCESS in 35m 45s (non-voting)\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8ac466be92eb47148be1847bef3e0a93 : SUCCESS in 35m 41s (non-voting)\n- swift-probetests-centos-8-arm64 https://zuul.opendev.org/t/openstack/build/10da687f96af4daf8ed9eb2fcba90ad0 : SUCCESS in 1h 41m 21s (non-voting)\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/f8c18568a62043bab203ff814aed3aae : SUCCESS in 38m 48s (non-voting)\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/96f8f10cf99345f7abd6a5df8a881ee0 : SUCCESS in 30m 33s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"e4d091391f80d751e7ba0506cdae8e3aec63f282","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-07-19 12:00:05.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"85898de717480a75971cd87628d61ecc2011ca8f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-19 12:57:34.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/11c81b777c344d5f8254af9162b956bc : SUCCESS in 9m 04s\n- grenade https://zuul.opendev.org/t/openstack/build/62f2de647c794ae6a3ec6b98d58b8015 : SUCCESS in 56m 55s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/dbcfd4da2ecc484fbeb25abdeda32d3f : SUCCESS in 45m 58s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/203c2cbef49d41f49314d5ab4f2bcf56 : SUCCESS in 8m 27s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/e920bdb19fd245a286fa49ba26626385 : SUCCESS in 8m 21s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/69cddd30b4904063b5c593acc0aa0bf0 : SUCCESS in 20m 09s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/b77b65406842413ba01013b194566654 : SUCCESS in 27m 51s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/48d4d52315974c669dadb474d3fd85aa : SUCCESS in 20m 48s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/d401bfa734e04449a8c82eb7b18820f7 : SUCCESS in 14m 26s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/775ce93bea534195b42503c9ead44a51 : SUCCESS in 17m 34s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/2bf4f38ad5904061882f6e2f82e7648b : SUCCESS in 15m 43s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/c57220ad00504c218ea1d2cec033441b : SUCCESS in 22m 01s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/b76f807319f847d9bbcf4a80cacfbf72 : SUCCESS in 15m 35s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/517a600e55f04ceb92e1d516424eafda : SUCCESS in 21m 00s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/8fe1ac5149e44c0a8d0ff1d673fe6f00 : SUCCESS in 17m 25s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/a396da51130d4ff6a7428e68396f79a8 : SUCCESS in 17m 05s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e68d6358ef2b4d4681c56512fe8507f0 : SUCCESS in 6m 52s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/bed53c6e2f4240e5b90ecc51be04cba1 : SUCCESS in 20m 25s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/7ac0c08539fb467fafe9e8fa24e130e9 : SUCCESS in 48m 49s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/99419d4c6b064ace8624841dc78bd66d : SUCCESS in 52m 18s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c9ba1471ae554cdc8f41c05d910befdb : SUCCESS in 34m 18s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e3d0e4e1676f4a3c937b108f45780d38 : SUCCESS in 36m 52s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/505f1c8f4865460c8436fc6b032a5d80 : SUCCESS in 18m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eab58676b1cf4e028b176aed02b14911 : SUCCESS in 6m 52s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/8a9da05841914329b8267eb71979281d : SUCCESS in 24m 41s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/02e45a537e4f43dbaef463f34e0e1906 : SUCCESS in 49m 21s","accounts_in_message":[],"_revision_number":5},{"id":"7a80aab09dda2a8b67e59fc9a16b8e24cd601171","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-19 13:17:43.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8b98cd405ea141fa936a9fa9e9fbc7ef : SUCCESS in 42m 45s (non-voting)\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/55d6be99e37a4713ac0397ad30f73d02 : SUCCESS in 30m 00s (non-voting)\n- swift-probetests-centos-8-arm64 https://zuul.opendev.org/t/openstack/build/540296fea6104b49a908eaa78c9e1a17 : SUCCESS in 1h 11m 33s (non-voting)\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/c5f5134b8ada4ae6a9e207fe18bc0f8b : SUCCESS in 38m 18s (non-voting)\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/78f1de62f2c14fb0a211fb16c4d43184 : SUCCESS in 29m 39s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"f4a4f2437503783cdfbedbf6e28e8d3215488691","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-07-19 19:58:40.000000000","message":"Patch Set 5:\n\n(2 comments)\n\nCouple minor things about comments in tests, but otherwise LGTM. I\u0027ll plan on pushing up a fix later today, assuming you don\u0027t get to it.","accounts_in_message":[],"_revision_number":5},{"id":"26617e51b319ce9dc458e1450a0860a7d233efe0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-07-19 20:18:20.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"f12c72a5f0cb9dddd0a15887f00365aa2480b376","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-07-19 20:19:06.000000000","message":"Patch Set 6:\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"1d09759d325e6643dd9681612c17dfcbc3c60a7f","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-07-19 20:44:53.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"25b6c4979a4a2685b251e37eb8887c224a7389ba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-19 21:14:05.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/641bf99f6f114a6f8972d98364a5d02d : SUCCESS in 9m 34s\n- grenade https://zuul.opendev.org/t/openstack/build/4f8a81a419374917a939a58de4c6e864 : SUCCESS in 53m 57s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/2431029c398f439cafe22995e7dc2a9d : SUCCESS in 36m 57s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/927807421b6a4354bc8c6c1f5c58aebc : SUCCESS in 8m 52s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/d586864af7574af7be53abc0935de944 : SUCCESS in 9m 13s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/d2ba234f8de44f97869de3569e851510 : SUCCESS in 16m 15s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/dd75bb57fca04e468bf6bc0fae71fc81 : SUCCESS in 16m 16s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/e9fc1bd9502040a7ac6e5651266836c9 : SUCCESS in 16m 14s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/4c48ef5d63a448febe35e46bcac2b694 : SUCCESS in 15m 56s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/72b8b459abab48828e32925bb3decb1f : SUCCESS in 16m 31s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/9ce2dbd5bf384d2a85ff15349f880c91 : SUCCESS in 16m 53s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/1159eb6e010a422eba7f23835bd82aaa : SUCCESS in 18m 03s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/7c0e02afcf8f4f40b7f707b16105921f : SUCCESS in 17m 36s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/0ea57f8878a64987a5d0240cdb1f6eb4 : SUCCESS in 17m 01s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/41d5779c62104e64accc75859bb24863 : SUCCESS in 17m 58s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/4a28feef334a4ed5bee8d4892c3b164c : SUCCESS in 17m 42s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/60153fbe707e45f9b3afbaebd3134767 : SUCCESS in 7m 30s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/1244a4a972fc47c9b9453458eb2734ee : SUCCESS in 22m 19s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/e3c631a5d7504a05917aeec0867678d7 : SUCCESS in 48m 45s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/5482c3d739604b10981a807c5bdc1314 : SUCCESS in 48m 51s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/6c5be7719f164326b6ede4ec68cb5a6c : SUCCESS in 38m 10s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4fe9a9be5602463287462cecba258818 : SUCCESS in 38m 10s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/72469d13e5c8419fb90e272f819005df : SUCCESS in 15m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a229685f0706412f938706cd781b82e5 : SUCCESS in 6m 25s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/89cacf5e53e743c3a38cebf88fc77e15 : SUCCESS in 26m 21s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e6c6f5909cae45c4bd94751aa21f2ecb : SUCCESS in 41m 53s","accounts_in_message":[],"_revision_number":6},{"id":"2deb6e6d73c9edd2237f4f48f572d2b82dd34418","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-19 21:14:33.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"f44f10a1503d0e0720ab78472a1472e6b9a59d42","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-19 22:13:36.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9602594794bc4dcc8851ec523b1451aa : SUCCESS in 40m 48s (non-voting)\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5b58ba4067e7453ab9e11cd8ee04a5f7 : SUCCESS in 29m 06s (non-voting)\n- swift-probetests-centos-8-arm64 https://zuul.opendev.org/t/openstack/build/f0bff1cbaae542149ffb9c2880cc23e9 : SUCCESS in 1h 52m 59s (non-voting)\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/7b8d811a1d464a05aa1a7e4ce47b7095 : SUCCESS in 48m 21s (non-voting)\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/27e0b0f008464752a4d56a49c936fbde : SUCCESS in 43m 38s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"5cc52d6fd66995ad70f205f1f378a48a9a0fe26e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-19 22:29:16.000000000","message":"Patch Set 6: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7927f8724a144a24b474d4b02e824613 : SUCCESS in 10m 10s\n- grenade https://zuul.opendev.org/t/openstack/build/396a743909c145f095d9eb01892c993a : SUCCESS in 55m 32s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/001a74f59e76421292a2773a4f7b5df5 : FAILURE in 17m 25s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/e95e8f124d6141a7b7242d276d242d12 : SUCCESS in 8m 18s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/b07ee3b980114f7ba93227093e4691ae : SUCCESS in 9m 05s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/e0c7f597c16f474db530a09212bd7de2 : SUCCESS in 16m 53s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d916b5a63a1c4cc0960bf4311c408b21 : SUCCESS in 17m 46s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/fb8a75d38a3840a38008e29b8b85c8c3 : SUCCESS in 18m 21s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/649047e739f348f0bcf83711f8a87abb : SUCCESS in 16m 55s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/9eb9500402ec4978a36d6de4325ac011 : SUCCESS in 15m 45s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/d14583c0683948eea05f37a639e9f835 : SUCCESS in 17m 47s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/5b0d92e843d2439a9dae59474e6a823e : SUCCESS in 17m 43s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/869e9687ce054348be2f8dca99344cb2 : SUCCESS in 17m 53s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/d5ad64045d644f7da2cb17ca3f785844 : SUCCESS in 16m 51s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/e1fddf9af7a643e199aa9218e61a2f24 : SUCCESS in 16m 15s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/25711992c354473aa12ebd294ca9ff36 : SUCCESS in 17m 30s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/7e49b1b5cf1c4bbc8c185ca60dffe2b6 : SUCCESS in 6m 00s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/8a77d641f8264e84a5a226f70ae33345 : SUCCESS in 50m 28s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/285368f924204e13bb87b1406c71ce82 : SUCCESS in 46m 40s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e2c684378fe4417f868c8eba735373d9 : SUCCESS in 37m 45s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f184590596a5461fbc2756fdf208ee58 : SUCCESS in 42m 27s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2667813ba74f43c986b595e9da16e327 : SUCCESS in 18m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b84f773c6791493487125dcaf89e1f15 : SUCCESS in 5m 53s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/84c7313342014a3ebdab77395c59ef21 : SUCCESS in 23m 24s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bff9118f8df844c3b2de97482b0fa88a : SUCCESS in 50m 17s","accounts_in_message":[],"_revision_number":6},{"id":"542f19d365c3abbef979efde7e49e76f6315ea26","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-07-19 22:44:26.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"19be7a5ba56dd2d97a175ce76e5ebfa2c59d2704","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-19 23:52:12.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56480aaa637f4893a4811e060274fb0a : SUCCESS in 10m 43s\n- grenade https://zuul.opendev.org/t/openstack/build/ad02d5a5476e4239bda8fe2227f989df : SUCCESS in 58m 07s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/29173b9ef5fb454b80e0ed2c1eb46653 : SUCCESS in 40m 27s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e5431e771a5e4c369d8226502204d50d : SUCCESS in 10m 25s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/c807ae4f49e94b1eaf795625ef63226a : SUCCESS in 11m 20s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/ff12d81d4a5d4b1488480403f94cb324 : SUCCESS in 17m 14s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/69b76f58c77a4995ae4f204d88906a83 : SUCCESS in 17m 06s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/9ce0fb2000b94786b01412df043a5d2b : SUCCESS in 16m 36s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/025559032ff04a8682612ec2a593c4f9 : SUCCESS in 16m 05s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/1948d3d5bf5f4cfd84bfebf861aafd25 : SUCCESS in 16m 57s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/4b5dd2bd18954ddba16d1dfba7273ac3 : SUCCESS in 17m 55s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/12944b075302451386a24ae04897b1b7 : SUCCESS in 17m 18s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/47269627b6c345408337e6acba3118cd : SUCCESS in 18m 12s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/1dc97e95a0654db69e7738cbb97c2909 : SUCCESS in 18m 04s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/cfc01ff81022444eafb54322150d969a : SUCCESS in 18m 25s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/7651992d687f47319ef96bcdc6aa2435 : SUCCESS in 17m 41s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b76bdaa8cdb44629934e4d5118604550 : SUCCESS in 7m 59s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/9cd6b3c603fb49c6b9bb0307850582c2 : SUCCESS in 22m 15s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/b8c57c1c8b5f47daa347120d0e8de34b : SUCCESS in 48m 24s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/fc148c5a6a05479594ebc2104349a31f : SUCCESS in 48m 56s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/81b3aaf3e7be427db2e92e0ea1a44962 : SUCCESS in 40m 15s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/13304f505db24589a3e953e217cd85db : SUCCESS in 41m 46s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/760b0c70301547e3875128694d84d061 : SUCCESS in 17m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/026adabc6c1b4c4bb9b9348f6b55ca85 : SUCCESS in 6m 22s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/3938f31e7ef044138f0ecdd828934097 : SUCCESS in 25m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a5b4f36d8a9b43099d28ce15678bd4ec : SUCCESS in 1h 00m 10s","accounts_in_message":[],"_revision_number":6},{"id":"ed2ce5d0fdf3e8238f9c64a10fb2ee89c9df6785","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-19 23:52:43.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"55c4a55da6a4a3d5657e75105ee07014233cece1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-20 00:28:01.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c266eadcdfce4d97a311c533285f9349 : SUCCESS in 57m 09s (non-voting)\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/54d1da1c245f419d9d81d60e876eabdd : SUCCESS in 29m 52s (non-voting)\n- swift-probetests-centos-8-arm64 https://zuul.opendev.org/t/openstack/build/9c11f8e0ea114402923e85e8e7ba2836 : SUCCESS in 1h 41m 02s (non-voting)\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/1020f6f6762b4d2586c26f08064450a7 : SUCCESS in 58m 39s (non-voting)\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/038aa35fc03142629fc31eb3fa4ce3c2 : SUCCESS in 1h 00m 16s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"49f627354b98e2ded80b402530792405bc97c6f7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-20 00:58:17.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d4c86e99091548f6b5b1b40158fcf963 : SUCCESS in 9m 29s\n- grenade https://zuul.opendev.org/t/openstack/build/ee9dcd526d874d3585560fa04f5233ea : SUCCESS in 1h 02m 14s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/eef08b48d7964e3cb3de2fd62eef14a5 : SUCCESS in 37m 17s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/16b18718df254ef4a6c20321c020eb2e : SUCCESS in 8m 50s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/ca44305f6f4b4a27a935c54eca7afa8f : SUCCESS in 8m 59s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/14c30d0556eb468a9b3a4b85386c1f95 : SUCCESS in 20m 03s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/84c604af8f5f4554ad04404141cd991f : SUCCESS in 21m 35s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/45fe8e3bd1d8403f90b6b4cd0970c438 : SUCCESS in 16m 03s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/205fea77ad124d5d92a6d697b5daf828 : SUCCESS in 21m 21s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/fe195bade0594698939809ee5f9a2cb1 : SUCCESS in 21m 53s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/d09431dff4504af6814a62a1bb103e8c : SUCCESS in 20m 17s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/ee52f1b561bb4fdc93879d0424f76172 : SUCCESS in 20m 45s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/3601032826f947708efdc1a1321bb495 : SUCCESS in 21m 44s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/394a6b62454f4d09a73597c8dc72d0d3 : SUCCESS in 21m 40s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/a17163b0183c4725b4f6788ffa2db6be : SUCCESS in 17m 00s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/0f3e7c189dfe4ec5812c716b314f7476 : SUCCESS in 22m 29s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/0e5026a4deae4f178f7cfb0e8f258dcc : SUCCESS in 6m 55s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/3cfdac9a7f4b43b8aeaa5794b5cd0e1f : SUCCESS in 55m 30s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/ed2ea975fbf7406d9fe31240f84a5a7f : SUCCESS in 56m 59s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/0b1b5d82c23a42688df8623d1d859a87 : SUCCESS in 40m 21s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/7a28857fd35c43b48962cd7ce5b82e1a : SUCCESS in 40m 39s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c64a1a7271d24f44ba5299b376fcb4d3 : SUCCESS in 18m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/457bbd00bb034bc6ad099b60298e7f1a : SUCCESS in 6m 29s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f385facbb09d43dfaa3260fda0544b19 : SUCCESS in 30m 31s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bc0ccbb9c24b46ff9b22367e978ca6ad : SUCCESS in 42m 20s","accounts_in_message":[],"_revision_number":6},{"id":"d79e94e5045d92a77384507bf3b558b0c489064c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-20 00:58:17.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6},{"id":"02bb81b8fa134cbd89f9640e5de0b4c987fc6967","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-20 00:59:34.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/913767c7bf3148818986a597a474c8ba : SUCCESS in 59s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/5126795d30664b7d948d4ae3c602eedd : SUCCESS in 48s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/2e2e70cd16df4c68b1145f2dc19256fe : SUCCESS in 44s (non-voting)","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"bbaed18e9b681ce9cf26ffa6a5d5292f5cb219b7","revisions":{"7ff1dc876729f9c5709306326ee854d2b8ae40a9":{"kind":"REWORK","_number":1,"created":"2021-07-15 17:10:15.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/74/800974/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/74/800974/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/74/800974/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/74/800974/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/74/800974/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/74/800974/1"}}},"commit":{"parents":[{"commit":"dfd020d11cb9b5aad0027f245e49026b91125749","subject":"Merge \"sharder: add more validation checks on config\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/dfd020d11cb9b5aad0027f245e49026b91125749"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-07-15 17:08:01.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-07-15 17:09:26.000000000","tz":60},"subject":"WIP diskfile: don\u0027t remove recently written non-durables","message":"WIP diskfile: don\u0027t remove recently written non-durables\n\nChange-Id: I5f3318a44af64b77a63713e6ff8d0fd3b6144f13\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7ff1dc876729f9c5709306326ee854d2b8ae40a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7ff1dc876729f9c5709306326ee854d2b8ae40a9"}]},"branch":"refs/heads/master"},"e95305df4e24c14e50622a02d0003248bc84bc87":{"kind":"REWORK","_number":2,"created":"2021-07-15 18:31:56.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/74/800974/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/74/800974/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/74/800974/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/74/800974/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/74/800974/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/74/800974/2"}}},"commit":{"parents":[{"commit":"dfd020d11cb9b5aad0027f245e49026b91125749","subject":"Merge \"sharder: add more validation checks on config\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/dfd020d11cb9b5aad0027f245e49026b91125749"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-07-15 17:08:01.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-07-15 18:29:42.000000000","tz":60},"subject":"WIP diskfile: don\u0027t remove recently written non-durables","message":"WIP diskfile: don\u0027t remove recently written non-durables\n\nDiskFileManager will remove any stale files during\ncleanup_ondisk_files(): these include tombstones and nondurable EC\ndata fragments whose timestamps are older than reclaim_age. It can\nusually be safely assumed that a non-durable data fragment older than\nreclaim_age is not going to become durable. However, if an agent PUTs\nobjects with specified older X-Timestamps (for example the reconciler\nor container-sync) then there is a window of time during which the\nobject server has written an old non-durable data file but has not yet\ncommitted it to make it durable.\n\nPreviously, if another process (for example the reconstructor) called\ncleanup_ondisk_files during this window then the non-durable data file\nwould be removed. The subsequent attempt to commit the data file would\nthen result in a traceback due to there no longer being a data file to\nrename, and of course the data file is lost.\n\nThis patch modifies cleanup_ondisk_files to not remove old, otherwise\nstale, non-durable data files that were only written to disk in the\npreceding \u0027commit_window\u0027 seconds. \u0027commit_window\u0027 is configurable for\nthe object server and defaults to 60 seconds.\n\nRelated-Change: I0d519ebaaade35249fb7b17bd5f419ffdaa616c0\nChange-Id: I5f3318a44af64b77a63713e6ff8d0fd3b6144f13\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e95305df4e24c14e50622a02d0003248bc84bc87"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e95305df4e24c14e50622a02d0003248bc84bc87"}]},"branch":"refs/heads/master"},"e87a10533ea7f1923ef159ce053f7169d503b077":{"kind":"REWORK","_number":3,"created":"2021-07-16 10:18:43.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/74/800974/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/74/800974/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/74/800974/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/74/800974/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/74/800974/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/74/800974/3"}}},"commit":{"parents":[{"commit":"dfd020d11cb9b5aad0027f245e49026b91125749","subject":"Merge \"sharder: add more validation checks on config\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/dfd020d11cb9b5aad0027f245e49026b91125749"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-07-15 17:08:01.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-07-16 10:18:30.000000000","tz":60},"subject":"diskfile: don\u0027t remove recently written non-durables","message":"diskfile: don\u0027t remove recently written non-durables\n\nDiskFileManager will remove any stale files during\ncleanup_ondisk_files(): these include tombstones and nondurable EC\ndata fragments whose timestamps are older than reclaim_age. It can\nusually be safely assumed that a non-durable data fragment older than\nreclaim_age is not going to become durable. However, if an agent PUTs\nobjects with specified older X-Timestamps (for example the reconciler\nor container-sync) then there is a window of time during which the\nobject server has written an old non-durable data file but has not yet\ncommitted it to make it durable.\n\nPreviously, if another process (for example the reconstructor) called\ncleanup_ondisk_files during this window then the non-durable data file\nwould be removed. The subsequent attempt to commit the data file would\nthen result in a traceback due to there no longer being a data file to\nrename, and of course the data file is lost.\n\nThis patch modifies cleanup_ondisk_files to not remove old, otherwise\nstale, non-durable data files that were only written to disk in the\npreceding \u0027commit_window\u0027 seconds. \u0027commit_window\u0027 is configurable for\nthe object server and defaults to 60 seconds.\n\nCloses-Bug: #1936508\nRelated-Change: I0d519ebaaade35249fb7b17bd5f419ffdaa616c0\nChange-Id: I5f3318a44af64b77a63713e6ff8d0fd3b6144f13\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e87a10533ea7f1923ef159ce053f7169d503b077"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e87a10533ea7f1923ef159ce053f7169d503b077"}]},"branch":"refs/heads/master"},"808bfc72969c7ccffe7f1d4f1b8328cd95f0b417":{"kind":"REWORK","_number":4,"created":"2021-07-16 12:47:20.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/74/800974/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/74/800974/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/74/800974/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/74/800974/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/74/800974/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/74/800974/4"}}},"commit":{"parents":[{"commit":"dfd020d11cb9b5aad0027f245e49026b91125749","subject":"Merge \"sharder: add more validation checks on config\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/dfd020d11cb9b5aad0027f245e49026b91125749"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-07-15 17:08:01.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-07-16 12:46:25.000000000","tz":60},"subject":"diskfile: don\u0027t remove recently written non-durables","message":"diskfile: don\u0027t remove recently written non-durables\n\nDiskFileManager will remove any stale files during\ncleanup_ondisk_files(): these include tombstones and nondurable EC\ndata fragments whose timestamps are older than reclaim_age. It can\nusually be safely assumed that a non-durable data fragment older than\nreclaim_age is not going to become durable. However, if an agent PUTs\nobjects with specified older X-Timestamps (for example the reconciler\nor container-sync) then there is a window of time during which the\nobject server has written an old non-durable data file but has not yet\ncommitted it to make it durable.\n\nPreviously, if another process (for example the reconstructor) called\ncleanup_ondisk_files during this window then the non-durable data file\nwould be removed. The subsequent attempt to commit the data file would\nthen result in a traceback due to there no longer being a data file to\nrename, and of course the data file is lost.\n\nThis patch modifies cleanup_ondisk_files to not remove old, otherwise\nstale, non-durable data files that were only written to disk in the\npreceding \u0027commit_window\u0027 seconds. \u0027commit_window\u0027 is configurable for\nthe object server and defaults to 60.0 seconds.\n\nCloses-Bug: #1936508\nRelated-Change: I0d519ebaaade35249fb7b17bd5f419ffdaa616c0\nChange-Id: I5f3318a44af64b77a63713e6ff8d0fd3b6144f13\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/808bfc72969c7ccffe7f1d4f1b8328cd95f0b417"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/808bfc72969c7ccffe7f1d4f1b8328cd95f0b417"}]},"branch":"refs/heads/master"},"9656b85fd8bf5dc2f8f380485d61d47e646d6567":{"kind":"REWORK","_number":5,"created":"2021-07-19 10:06:09.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/74/800974/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/74/800974/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/74/800974/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/74/800974/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/74/800974/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/74/800974/5"}}},"commit":{"parents":[{"commit":"dfd020d11cb9b5aad0027f245e49026b91125749","subject":"Merge \"sharder: add more validation checks on config\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/dfd020d11cb9b5aad0027f245e49026b91125749"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-07-15 17:08:01.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-07-19 09:58:36.000000000","tz":60},"subject":"diskfile: don\u0027t remove recently written non-durables","message":"diskfile: don\u0027t remove recently written non-durables\n\nDiskFileManager will remove any stale files during\ncleanup_ondisk_files(): these include tombstones and nondurable EC\ndata fragments whose timestamps are older than reclaim_age. It can\nusually be safely assumed that a non-durable data fragment older than\nreclaim_age is not going to become durable. However, if an agent PUTs\nobjects with specified older X-Timestamps (for example the reconciler\nor container-sync) then there is a window of time during which the\nobject server has written an old non-durable data file but has not yet\ncommitted it to make it durable.\n\nPreviously, if another process (for example the reconstructor) called\ncleanup_ondisk_files during this window then the non-durable data file\nwould be removed. The subsequent attempt to commit the data file would\nthen result in a traceback due to there no longer being a data file to\nrename, and of course the data file is lost.\n\nThis patch modifies cleanup_ondisk_files to not remove old, otherwise\nstale, non-durable data files that were only written to disk in the\npreceding \u0027commit_window\u0027 seconds. \u0027commit_window\u0027 is configurable for\nthe object server and defaults to 60.0 seconds.\n\nCloses-Bug: #1936508\nRelated-Change: I0d519ebaaade35249fb7b17bd5f419ffdaa616c0\nChange-Id: I5f3318a44af64b77a63713e6ff8d0fd3b6144f13\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9656b85fd8bf5dc2f8f380485d61d47e646d6567"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9656b85fd8bf5dc2f8f380485d61d47e646d6567"}]},"branch":"refs/heads/master"},"bbaed18e9b681ce9cf26ffa6a5d5292f5cb219b7":{"kind":"REWORK","_number":6,"created":"2021-07-19 20:18:20.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/74/800974/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/74/800974/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/74/800974/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/74/800974/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/74/800974/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/74/800974/6"}}},"commit":{"parents":[{"commit":"dfd020d11cb9b5aad0027f245e49026b91125749","subject":"Merge \"sharder: add more validation checks on config\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/dfd020d11cb9b5aad0027f245e49026b91125749"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-07-15 17:08:01.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-07-19 20:18:02.000000000","tz":60},"subject":"diskfile: don\u0027t remove recently written non-durables","message":"diskfile: don\u0027t remove recently written non-durables\n\nDiskFileManager will remove any stale files during\ncleanup_ondisk_files(): these include tombstones and nondurable EC\ndata fragments whose timestamps are older than reclaim_age. It can\nusually be safely assumed that a non-durable data fragment older than\nreclaim_age is not going to become durable. However, if an agent PUTs\nobjects with specified older X-Timestamps (for example the reconciler\nor container-sync) then there is a window of time during which the\nobject server has written an old non-durable data file but has not yet\ncommitted it to make it durable.\n\nPreviously, if another process (for example the reconstructor) called\ncleanup_ondisk_files during this window then the non-durable data file\nwould be removed. The subsequent attempt to commit the data file would\nthen result in a traceback due to there no longer being a data file to\nrename, and of course the data file is lost.\n\nThis patch modifies cleanup_ondisk_files to not remove old, otherwise\nstale, non-durable data files that were only written to disk in the\npreceding \u0027commit_window\u0027 seconds. \u0027commit_window\u0027 is configurable for\nthe object server and defaults to 60.0 seconds.\n\nCloses-Bug: #1936508\nRelated-Change: I0d519ebaaade35249fb7b17bd5f419ffdaa616c0\nChange-Id: I5f3318a44af64b77a63713e6ff8d0fd3b6144f13\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bbaed18e9b681ce9cf26ffa6a5d5292f5cb219b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bbaed18e9b681ce9cf26ffa6a5d5292f5cb219b7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
