)]}'
{"id":"openstack%2Fswift~801056","triplet_id":"openstack%2Fswift~master~I1589a7517b7375fcc21472e2d514f26986bf5079","project":"openstack/swift","branch":"master","topic":"p-diskfile-commit-window","hashtags":[],"change_id":"I1589a7517b7375fcc21472e2d514f26986bf5079","subject":"reconstructor: retire nondurable_purge_delay option","status":"MERGED","created":"2021-07-16 12:47:20.000000000","updated":"2021-07-20 09:53:34.000000000","submitted":"2021-07-20 09:51:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"801056-p-diskfile-commit-window","meta_rev_id":"025eacec46342583093a170022bcc11e802810c5","_number":801056,"virtual_id_number":801056,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-07-20 09:51:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-07-20 09:53:34.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-07-19 20:46:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-16 13:48:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-16 17:51:24.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"9c03cd64ef8c8237dcb7256a0d9642d2aeb996fb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-07-16 12:47:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b2d90166d210266ba65e73b9faab4ed3dd6a8edf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-16 13:48:56.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/628106fcd26f4972a637246b8aa0f7b5 : SUCCESS in 12m 28s\n- grenade https://zuul.opendev.org/t/openstack/build/2efd44aae02947cea948af70b60dadf6 : SUCCESS in 58m 50s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c299b36545ca4850b1eb9dabb060705b : SUCCESS in 56m 27s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/dc93e7f04a9044ea85df98237c1feea5 : SUCCESS in 7m 51s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/12430aa2344f49b692ab412e2e39850c : SUCCESS in 8m 48s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/5133825edf3f450d9860675e6bbde3d4 : SUCCESS in 18m 26s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/cd40b993009e4156a1cd8b2ea3845874 : SUCCESS in 17m 06s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/958d720f13ac4d2f902b6c728e386637 : SUCCESS in 29m 26s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/2026e2ef3a2247c7a1fc3605d16d48ea : SUCCESS in 31m 24s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/c16580f745744a0c87f77d821d987690 : SUCCESS in 15m 11s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/b0ff3aa5329846cba40cbf646c11e096 : SUCCESS in 17m 03s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/0001ddfdfbc243ff95c472003fbf44de : SUCCESS in 17m 32s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/1faf6623cea641be9321d1a95811b336 : SUCCESS in 36m 07s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/05c9220a710c4cb98b62634f0cc539c4 : SUCCESS in 18m 15s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/60006c33830442d9be3bab8d393179a4 : SUCCESS in 17m 22s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/96eaf6a463884e6b8223965db4006a21 : SUCCESS in 36m 40s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/7c9dd98845434d298e94f069ba394001 : SUCCESS in 6m 41s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f76317420cdf4e3fabf076f0bfe5e86b : SUCCESS in 20m 55s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/f8ab954c0abf412d8c34bb4764af16f7 : SUCCESS in 50m 21s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/6fccea6548494428bb532d81f415f1bc : SUCCESS in 49m 51s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ab71b2a9344b41d1aa1d396a0d398f61 : SUCCESS in 53m 31s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/03a087366de04f438269ce9e94c7a347 : SUCCESS in 40m 09s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/be90836f761b4a45a842a786ce3f5178 : SUCCESS in 16m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e01d53d1dfa34690935fee2ec8c27ef0 : SUCCESS in 5m 55s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/26a0ffaf2f014d8e97c81ac729de3c5d : FAILURE in 26m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b3b02c9570a44a6ab14028a2d1f6e091 : SUCCESS in 55m 04s","accounts_in_message":[],"_revision_number":1},{"id":"7935a6e3d57fa52eecdbe11f6fdd21985993be74","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-16 14:25:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/742d379207244fdf8ad71b11dd3a4bb6 : SUCCESS in 41m 53s (non-voting)\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/cd779ba2637b4b4aaae27efcf5f0f9fa : SUCCESS in 56m 38s (non-voting)\n- swift-probetests-centos-8-arm64 https://zuul.opendev.org/t/openstack/build/356495a589b540249f4630fd874a0e87 : FAILURE in 1h 31m 59s (non-voting)\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/49b0f90231104079bb0750eadb5ba718 : SUCCESS in 33m 29s (non-voting)\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/4f17a9c91eb44a1f939b957dc7559b05 : SUCCESS in 32m 31s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"02f73b5cc3ac33e9031906228579dc2dd6ba42ca","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-07-16 16:23:48.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"5474ad2f089eb53f8118b727a520a200f523219b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-16 17:27:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d486e6ad080430c886cd73ce6da7144 : SUCCESS in 13m 15s\n- grenade https://zuul.opendev.org/t/openstack/build/949d97947a52427c8b73f9d24fcfed0a : SUCCESS in 1h 02m 10s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/577a90730d634c588eaebf1daee575ed : SUCCESS in 47m 25s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/efa7d11954df4359be85068cb749e463 : SUCCESS in 9m 35s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/6e2077ac25294566859306a2781d5dc0 : SUCCESS in 9m 38s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/80732de7cc134a118a731c473f44a44c : SUCCESS in 17m 38s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/e67b09d010944fbe924d7aae2338fbb7 : SUCCESS in 16m 24s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/1e2f2d70cc3243a9a3cf105772a4484e : SUCCESS in 24m 22s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/b56adc69850f47359a4fedd4e713d603 : SUCCESS in 15m 53s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/3bd02dec0026436a887363541801292b : SUCCESS in 17m 25s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/71b9f6870ef742648c9f4f63b19e5568 : SUCCESS in 18m 36s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/3e9113b8756e4a04b92025c16be82bb6 : SUCCESS in 18m 13s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/92e550435b3b4d5d98735e1b61deaccd : SUCCESS in 19m 35s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/3ace5b23f4bf47c2a8f801df693399f1 : SUCCESS in 17m 45s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/24ed1898656f43d691404d68f7116664 : SUCCESS in 18m 19s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/cdbcdd7717ec4c099fb6fd13f6a5ee9c : SUCCESS in 19m 14s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f76d363c8b45403ebe2d5ef339dc4b60 : SUCCESS in 9m 08s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/07883bc8dd7c42b2ad5d6d6372ca4e38 : SUCCESS in 26m 11s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/1d3b324f1d20476f9f995336d8d31769 : SUCCESS in 53m 00s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/146188b368ce4c1ca163f21be92332dd : SUCCESS in 54m 29s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e64abb018aff40f79fc1ddd4c6e82b03 : SUCCESS in 42m 19s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/8edeccd4bc784848a29346c07743c9d0 : SUCCESS in 38m 41s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ae4a0bb6c2f04d32ba316dc2b302b9e0 : SUCCESS in 17m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f1046f2698a14bfdb0a6e3ce0dd1f251 : SUCCESS in 7m 36s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b066642d0dfe4c4d9f1dc0ffa14fefd7 : SUCCESS in 25m 57s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bb8b3f00022341b7924d5279c5a58aff : SUCCESS in 47m 12s","accounts_in_message":[],"_revision_number":1},{"id":"01035f6d48009b5021ac41ecf630f854b6ba3d33","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-07-16 17:51:24.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)\n\nCouple minor issues, but I\u0027d love to have this before a 2.28.0.","accounts_in_message":[],"_revision_number":1},{"id":"3eabc81ba8dbd8f8d854e7fb2b872b22f4792fe2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-16 18:33:01.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1e628141245a4dda9a59705f57f9d815 : FAILURE in 1h 02m 06s (non-voting)\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/66900fa2a1aa4a1a9468189c27646614 : FAILURE in 59m 06s (non-voting)\n- swift-probetests-centos-8-arm64 https://zuul.opendev.org/t/openstack/build/edd4870417e448bcbb7d6fb3853e1da4 : TIMED_OUT in 2h 01m 14s (non-voting)\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/45005d6125e44acd9a9a448263d77b2c : SUCCESS in 1h 09m 06s (non-voting)\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/8d19b27754ad45b7aa33b0ab176eedc4 : SUCCESS in 1h 08m 20s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6eb8b3003e47e8705f52d7c20d679381ba41d8b5","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-07-19 10:03:45.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"9749b54d9ac777537117d25aa34c264e26022867","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-07-19 10:06:09.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"15c7ed3334b44ba5e6ab6eaa2dfc7caf29e0eb64","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-19 11:21:34.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2f8989b9e4b443f9385cb421bc73829 : SUCCESS in 10m 40s\n- grenade https://zuul.opendev.org/t/openstack/build/c361a0d42b774482b669e28a159ec69f : SUCCESS in 1h 08m 57s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d05f327f125945c08bfdf62070ce27b2 : SUCCESS in 47m 32s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e3fe842a30a94c40abb8129d1616342b : SUCCESS in 9m 30s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/ebf2481f0fe04dd68d93dc8951960c52 : SUCCESS in 9m 13s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/d26ff174ba1846eb8141cdab6af343e9 : SUCCESS in 16m 39s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/72c34a8c285f4dea87ad52bdb7a52861 : SUCCESS in 18m 11s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/c9baafeb64284dc890388820bb4f1f02 : SUCCESS in 18m 19s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/09c047be8b154adaabe79e7d887bf8a9 : SUCCESS in 18m 22s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/1759e2e959e74a01bd203a67fc9379fb : SUCCESS in 17m 09s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/08748b62f6d94a34ac0f375589c062f3 : SUCCESS in 19m 53s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/d15f5327441b43e79224d7e7dcd1607f : SUCCESS in 19m 33s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/b0e3e400ab004f128e4f15345124d236 : SUCCESS in 17m 53s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/d2acd20afad248af9125551287db2124 : SUCCESS in 16m 34s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/674f884db2b04e7d9cd57305ab83d6e8 : SUCCESS in 17m 15s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/9db5d34583054aa5add9ab699af38fb5 : SUCCESS in 19m 20s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/2a3505f73c6b4880a3730779948c6592 : SUCCESS in 7m 13s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/96e6745121004f23aa18ce0f1e0b5a21 : SUCCESS in 22m 25s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/a01fecb41a044058bb0b38679d433bd2 : SUCCESS in 48m 11s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/7a5736ff649d4c029021a62edca4e13b : SUCCESS in 45m 46s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/75179404dd294e2c86a5708b5d71bd6b : SUCCESS in 38m 51s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/dd89b173f60e49b08e7321b0a2e4fafa : SUCCESS in 38m 48s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/60605c0d8c4d408c9446e4c117d8d46b : SUCCESS in 17m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/406164b921a3408ea0c8922c46eabc96 : SUCCESS in 5m 27s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e9f271f8a00f44e396fa302b7cd715c0 : SUCCESS in 25m 31s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/57e826a8789f4a5282283fae1e550fad : SUCCESS in 54m 10s","accounts_in_message":[],"_revision_number":2},{"id":"5d7cc5bf31fa04e3d2fd26e362f6c458ffb7fd99","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-19 11:31:28.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/800a87a8c7944f049c766c5c974bac57 : SUCCESS in 30m 01s (non-voting)\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/418b9becd41547cb9b0025a3eb35ee5d : SUCCESS in 32m 25s (non-voting)\n- swift-probetests-centos-8-arm64 https://zuul.opendev.org/t/openstack/build/c2c42a64789a458fbf2bc4e4f6b984c3 : SUCCESS in 1h 15m 51s (non-voting)\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/72a5006d5eb64321a1b4b14de0f00179 : SUCCESS in 30m 19s (non-voting)\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/4583d324a851467e9a6a96c47c527b2f : SUCCESS in 29m 52s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d95de0d0fff6922d579ebe6f424cc3496017684f","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-07-19 19:59:23.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"237835085683bb07b3d81629582ad6e1b65375f7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-07-19 20:18:20.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"5eea08726abb924a7d1cbc3332adaf47322ef09d","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-07-19 20:46:59.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"a268b27e4175b78036eec23e3adc5ede9987607f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-19 21:35:43.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8e1f1cbbde7645c7b2c03e01089e1a8e : SUCCESS in 48m 29s (non-voting)\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/df697381be8444b6b45f2b7e90e52d88 : SUCCESS in 29m 16s (non-voting)\n- swift-probetests-centos-8-arm64 https://zuul.opendev.org/t/openstack/build/0e5dfe6546e04e0091446d16487b3738 : SUCCESS in 1h 10m 59s (non-voting)\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/1351e04d04b647eeb3cbf4fef388e2b2 : SUCCESS in 30m 04s (non-voting)\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/a0f7597c67ea4c7f8486b17ec19f3339 : SUCCESS in 29m 12s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"415019c204c3b6695f070fa3e1e39f1f08358459","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-19 21:38:17.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a2bb10d4806491f845c7fe8b7f42ee3 : SUCCESS in 10m 57s\n- grenade https://zuul.opendev.org/t/openstack/build/d0d515baf8fc4e319ed683a6fae6bbbb : SUCCESS in 1h 00m 44s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/903d0c2850ee41ed8a60867ceecedca6 : SUCCESS in 41m 42s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e0e04aad3cd24e2eba4594844c1505eb : SUCCESS in 8m 26s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/d51c9e12e14e4350b83ebe84cf7cb3a8 : SUCCESS in 8m 55s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/d27f1f3c5b3b4af49b73baa286a0b9ca : SUCCESS in 20m 14s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a9e53019ea714ad4b1f82fca9afe26f7 : SUCCESS in 22m 28s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/0564ed092d3847af989183787aad2841 : SUCCESS in 17m 06s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/841c284fb88d4326bb7e6a094fadcc90 : SUCCESS in 18m 01s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/b9d5498e21354d4993d5b9e9263cb2be : SUCCESS in 18m 47s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/669bd72441984f3eba3a28bcf453106f : SUCCESS in 18m 33s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/9b08bcae35484e1dac618bf9fd252ddf : SUCCESS in 20m 40s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/f278a4f2179e400f9d471fd17bdcf328 : SUCCESS in 22m 07s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/e5d4b6c9948f441b940e42bcc3a3dcb1 : SUCCESS in 16m 32s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/3f786f451f214489a68ede888b9cb72a : SUCCESS in 16m 35s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/901e0e8dca7c4878b23ce32afe60bfbb : SUCCESS in 21m 52s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/6defaecb6b254e469b839e449bd3edf3 : SUCCESS in 6m 31s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/7439be46d24840ecb8e25ac0ad00424c : SUCCESS in 19m 35s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/5d9c3e499702448a87eb4fcbcdbe0588 : SUCCESS in 47m 37s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/d1c363069d164537a5ff47094f38ee27 : SUCCESS in 50m 10s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e426a45f948e4934a519fe01508b12cd : SUCCESS in 40m 49s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/39d8ccfcb2ef49fa8cf95491cd55e549 : SUCCESS in 40m 49s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b4d2a4b5acde4352abc833b19c34f70b : SUCCESS in 16m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/865bf96f12a7485b85f8628579130cc2 : SUCCESS in 6m 19s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/80222b52dd924b979b6578e71624bd51 : SUCCESS in 23m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/320f6e1a6f0d4ee9a50dfea8c30b6856 : SUCCESS in 1h 19m 15s","accounts_in_message":[],"_revision_number":3},{"id":"5af35cd9c95894194737ba58075908b6be5a1680","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-19 22:29:17.000000000","message":"Patch Set 3: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":3},{"id":"13bf99686f01256d63b851bb5ca082b81e168dd9","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-07-20 07:37:25.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"691521747f773fdf633df14f63bf6ac782112c84","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-20 08:40:09.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fdcd003af70b4db4bb2a72b1d67228ea : SUCCESS in 9m 10s\n- grenade https://zuul.opendev.org/t/openstack/build/f07e1449c1ae49c99fb4315ad6542f2b : SUCCESS in 1h 01m 15s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/be00cf8db4b14a75bcc280d385944ee0 : SUCCESS in 43m 54s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/39c86b8e694f41aab351cd9c72ef2ada : SUCCESS in 8m 13s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/0f74868847f543ddb62755d4a1b51566 : SUCCESS in 9m 00s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/b8b43deca3a24ab28d2e4ddf5e1cd320 : SUCCESS in 16m 55s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/4a158d53d3b649ddac305050cf99db5d : SUCCESS in 16m 51s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/f30511abba6a4214b932701d0630c769 : SUCCESS in 17m 16s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/17a7fb2519134f0daa0a9b0749602e99 : SUCCESS in 16m 43s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/bac5c77b0c0f46f8b0b9b4a82fa80fff : SUCCESS in 16m 33s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/8445047bdf3f4edba4e4dace14a55c04 : SUCCESS in 16m 44s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/03379d60ddbc4ae698c196b3a6851d24 : SUCCESS in 16m 51s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/2e6b8b42a1c0459085d5bfb57a53814a : SUCCESS in 17m 29s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/ada5a92fb9af4cedac1aa60e8e6c9331 : SUCCESS in 16m 33s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/dc3c474ae8714e7f81399509a47cb33e : SUCCESS in 16m 53s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/0a5bdd37c4c64d048f96ea1bd04658ff : SUCCESS in 17m 33s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/5d07d76440584ee48c590327ec15a1be : SUCCESS in 7m 02s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/ce3c3fb5e3bb470fa80dc273394dfb40 : SUCCESS in 19m 42s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/bd9154f83e584a5ba2a82ca406b12bd4 : SUCCESS in 48m 17s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/c0832043c744405fa3975feb48cb9e89 : SUCCESS in 47m 48s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/79906d24d3d341baa71a10d50a071c6f : SUCCESS in 38m 08s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/615493a7ac0a46839cafe3075f0e9a61 : SUCCESS in 38m 35s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5aa6d0c764dd486c91a845fb351d89f3 : SUCCESS in 16m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/de5fb2734cb642538ffa437c0fe34d34 : SUCCESS in 6m 15s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/fe27c988c7a04d9b80242b81bc78bbb5 : SUCCESS in 23m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ba38755c5e3743d39978b9bda80f44d1 : SUCCESS in 49m 22s","accounts_in_message":[],"_revision_number":3},{"id":"031f8b8802398f494bd1e65ecba11e48e8afca76","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-20 08:40:37.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"b637f23615ee1e6799035fdcb8851433b30f4c9e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-20 09:42:44.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/cb2d57772315444e8033f2fdde5fa39e : SUCCESS in 34m 43s (non-voting)\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1230e71a4c824d0bbeb5daa1d3982fd2 : SUCCESS in 31m 09s (non-voting)\n- swift-probetests-centos-8-arm64 https://zuul.opendev.org/t/openstack/build/74e148a6a56d4a8e83033d2a28a112dc : TIMED_OUT in 2h 02m 26s (non-voting)\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/70e60780dfaa439eb0e76e406721b6ce : SUCCESS in 40m 49s (non-voting)\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/d4c320f208ab428bbecb701bcf0877cf : SUCCESS in 29m 42s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"b65a77424ba30dd011bc2256e1cd008a77d56cc3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-20 09:51:49.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f1205d541864ca1a825fa48adb5b394 : SUCCESS in 10m 00s\n- grenade https://zuul.opendev.org/t/openstack/build/06ff4bf52c854efabb147a336624d558 : SUCCESS in 1h 05m 46s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d33a19d6fdef43d298c134c7c4ef127a : SUCCESS in 48m 18s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/5916743380ad405da7485953091495c9 : SUCCESS in 8m 32s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/819fd1fcab684da7945edf77d141f4e5 : SUCCESS in 9m 49s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/36e6b0e3e0354301afb38d572fe47f58 : SUCCESS in 16m 06s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a9756dc2bec14a1183561d4c08702dc3 : SUCCESS in 17m 15s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/3f61554f5ac640e8bc7f305499f9e3bd : SUCCESS in 17m 05s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/259c5004ec404a849380471c286e8c32 : SUCCESS in 17m 47s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/7672ae99818c4aa180e90de9e6795f71 : SUCCESS in 16m 46s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/7894deac944240358dd0bb39c08aa01d : SUCCESS in 16m 49s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/edb26b4739f24ab6a4e79ae148bcf766 : SUCCESS in 18m 53s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/6670e8f9ad714027a8b9e97429b1c4f9 : SUCCESS in 18m 03s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/b7adc1af7efa4167a16f3b94cd35fc9e : SUCCESS in 17m 58s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/9089a72e0d2f4edd8ac49be768f124f0 : SUCCESS in 16m 48s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/353a0c70e38b4a219e2317207decb5b4 : SUCCESS in 18m 52s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/4974f85d0c57422c9c56f8e2d020fda9 : SUCCESS in 6m 56s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/792d2b2a0b19479aac384b6bad3005fc : SUCCESS in 53m 26s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/2a4fa7ea9acc4290ab297a2cb5f49a4c : SUCCESS in 45m 44s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d4bf4e049f424481b1f5db4112c9ec7c : SUCCESS in 44m 32s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/11e9c9ea22b441eb9d28454cefa62aaa : SUCCESS in 42m 29s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/44d12f3bce744ba68cae7e4fa0912021 : SUCCESS in 16m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bcba8287681b47f287f2d8d3309da386 : SUCCESS in 7m 31s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/083dff8422e9499da1c9430faa9134de : SUCCESS in 23m 54s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b96300d91abb410e99cdd9ccee928095 : SUCCESS in 47m 03s","accounts_in_message":[],"_revision_number":3},{"id":"bf534da4381603c9ef18c568cc141912e96b4c2d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-20 09:51:49.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"46bbaab6c36b695adaffe651f7294cf0887282f7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-20 09:53:34.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/479527a5e92047b3810a53bdc756495c : SUCCESS in 1m 16s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/870ab04711a14861ab0dfd95c38d36a3 : SUCCESS in 49s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/cb092ab6c0d645d4a12d4842968676cc : SUCCESS in 57s (non-voting)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"2696a79f098b02988136b13caf1c2565ec09481f","revisions":{"65e9ce037854f235075f5da7e07e4ff539a214da":{"kind":"REWORK","_number":1,"created":"2021-07-16 12:47:20.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/56/801056/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/56/801056/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/801056/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/801056/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/801056/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/56/801056/1"}}},"commit":{"parents":[{"commit":"808bfc72969c7ccffe7f1d4f1b8328cd95f0b417","subject":"diskfile: don\u0027t remove recently written non-durables","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/808bfc72969c7ccffe7f1d4f1b8328cd95f0b417"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-07-16 11:27:25.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-07-16 12:47:01.000000000","tz":60},"subject":"reconstructor: retire nondurable_purge_delay option","message":"reconstructor: retire nondurable_purge_delay option\n\nThe nondurable_purge_delay option was introduced in [1] to prevent the\nreconstructor removing non-durable data files on handoffs that were\nabout to be made durable. The DiskFileManager commit_window option has\nsince been introduced [2] which specifies a similar time window during\nwhich non-durable data files should not be removed. The commit_window\noption can be re-used by the reconstructor, making the\nnondurable_purge_delay option redundant.\n\nThe nondurable_purge_delay option has not been available in any tagged\nrelease and is therefore removed with no backwards compatibility.\n\n[1] Related-Change: I0d519ebaaade35249fb7b17bd5f419ffdaa616c0\n[2] Related-Change: I5f3318a44af64b77a63713e6ff8d0fd3b6144f13\nChange-Id: I1589a7517b7375fcc21472e2d514f26986bf5079\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/65e9ce037854f235075f5da7e07e4ff539a214da"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/65e9ce037854f235075f5da7e07e4ff539a214da"}]},"branch":"refs/heads/master"},"d5df279b19c2dbc113223d065072f91aa7947c4b":{"kind":"REWORK","_number":2,"created":"2021-07-19 10:06:09.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/56/801056/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/56/801056/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/801056/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/801056/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/801056/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/56/801056/2"}}},"commit":{"parents":[{"commit":"9656b85fd8bf5dc2f8f380485d61d47e646d6567","subject":"diskfile: don\u0027t remove recently written non-durables","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9656b85fd8bf5dc2f8f380485d61d47e646d6567"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-07-16 11:27:25.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-07-19 10:03:03.000000000","tz":60},"subject":"reconstructor: retire nondurable_purge_delay option","message":"reconstructor: retire nondurable_purge_delay option\n\nThe nondurable_purge_delay option was introduced in [1] to prevent the\nreconstructor removing non-durable data files on handoffs that were\nabout to be made durable. The DiskFileManager commit_window option has\nsince been introduced [2] which specifies a similar time window during\nwhich non-durable data files should not be removed. The commit_window\noption can be re-used by the reconstructor, making the\nnondurable_purge_delay option redundant.\n\nThe nondurable_purge_delay option has not been available in any tagged\nrelease and is therefore removed with no backwards compatibility.\n\n[1] Related-Change: I0d519ebaaade35249fb7b17bd5f419ffdaa616c0\n[2] Related-Change: I5f3318a44af64b77a63713e6ff8d0fd3b6144f13\nChange-Id: I1589a7517b7375fcc21472e2d514f26986bf5079\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d5df279b19c2dbc113223d065072f91aa7947c4b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d5df279b19c2dbc113223d065072f91aa7947c4b"}]},"branch":"refs/heads/master"},"2696a79f098b02988136b13caf1c2565ec09481f":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-07-19 20:18:20.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/56/801056/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/56/801056/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/801056/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/801056/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/801056/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/56/801056/3"}}},"commit":{"parents":[{"commit":"bbaed18e9b681ce9cf26ffa6a5d5292f5cb219b7","subject":"diskfile: don\u0027t remove recently written non-durables","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bbaed18e9b681ce9cf26ffa6a5d5292f5cb219b7"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-07-16 11:27:25.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-07-19 20:18:06.000000000","tz":60},"subject":"reconstructor: retire nondurable_purge_delay option","message":"reconstructor: retire nondurable_purge_delay option\n\nThe nondurable_purge_delay option was introduced in [1] to prevent the\nreconstructor removing non-durable data files on handoffs that were\nabout to be made durable. The DiskFileManager commit_window option has\nsince been introduced [2] which specifies a similar time window during\nwhich non-durable data files should not be removed. The commit_window\noption can be re-used by the reconstructor, making the\nnondurable_purge_delay option redundant.\n\nThe nondurable_purge_delay option has not been available in any tagged\nrelease and is therefore removed with no backwards compatibility.\n\n[1] Related-Change: I0d519ebaaade35249fb7b17bd5f419ffdaa616c0\n[2] Related-Change: I5f3318a44af64b77a63713e6ff8d0fd3b6144f13\nChange-Id: I1589a7517b7375fcc21472e2d514f26986bf5079\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2696a79f098b02988136b13caf1c2565ec09481f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2696a79f098b02988136b13caf1c2565ec09481f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
