)]}'
{"id":"openstack%2Fswift~802768","triplet_id":"openstack%2Fswift~master~Ia952303a16457a5075e9a4e060806855c21577f5","project":"openstack/swift","branch":"master","hashtags":[],"change_id":"Ia952303a16457a5075e9a4e060806855c21577f5","subject":"diskfile: Stop raising ENOENT when we race to mark durable","status":"MERGED","created":"2021-07-28 18:22:23.000000000","updated":"2021-07-29 18:38:46.000000000","submitted":"2021-07-29 18:37:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"802768","meta_rev_id":"994e1d31a55d6ee24356c80a337b3ab96d9c4e8b","_number":802768,"virtual_id_number":802768,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-07-29 18:37:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-07-29 16:38:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-07-29 16:38:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-28 18:26:50.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"},{"updated":"2021-07-28 19:27:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-28 21:16:15.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"}],"messages":[{"id":"79e49c9a96cc9a869cec42d88433c9a5d135aced","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-07-28 18:22:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"28409c343c4894ca3db6eeb7879543f7e4b5130d","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2021-07-28 18:26:50.000000000","message":"Patch Set 1: Code-Review+2\n\n(2 comments)\n\nawesome","accounts_in_message":[],"_revision_number":1},{"id":"1714f1250a73d878e7de7547b6f58196a2476d11","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-28 19:27:46.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc565dfce43f4d5da127714dce6e16e0 : SUCCESS in 9m 29s\n- grenade https://zuul.opendev.org/t/openstack/build/39d1116b911f4d328bba6a90cb507154 : SUCCESS in 1h 00m 03s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/3a05f1c928744a238b557a384121980d : SUCCESS in 41m 26s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ccaf14696a594f1999df8c7a8f2c4eca : SUCCESS in 8m 05s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/ad4b154d8ac046e89f5fb3979a72b1ea : SUCCESS in 9m 17s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/c998e8311d4b4dadb15299aa642c29e7 : FAILURE in 14m 50s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/7b37f278e78245fbae9f43d347948603 : FAILURE in 16m 03s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/aa43c0a3b74844018e387d3885ef808c : FAILURE in 15m 09s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/ec8d18ec67b4446086cbe8bc172b40ec : FAILURE in 17m 03s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/31e4eb5268ee496e9dc9a91d1e48b3d4 : FAILURE in 17m 11s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/861004ea4ea7407d822ae9b85d9f5f62 : SUCCESS in 18m 11s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/cebf7a8b0dc54165949bd2b342f36687 : SUCCESS in 20m 32s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/ed68c1269179463f8d44c1d0d123fc6d : SUCCESS in 19m 34s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/1e68659029e047e584224a4196446ed3 : SUCCESS in 16m 45s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/b7650673306e4639be276fb95430b736 : SUCCESS in 19m 28s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/5ceda65e60ca4df8a947d90b2baefb4c : SUCCESS in 17m 59s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/7280f2a2f82449e787c94e3b601d4f0b : SUCCESS in 6m 29s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/780e8ccf877b43829845e3d8bf97972b : SUCCESS in 21m 30s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/aee41188d2ce45baa86feb33c24fbcbb : SUCCESS in 47m 43s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/72632e90113f448db7a6256fe3f9cf3e : SUCCESS in 51m 27s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/0648c7331d114e4888123ac67cf185e0 : SUCCESS in 36m 54s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/de42686bd7dc4ac39c3de260a2b87df5 : SUCCESS in 41m 37s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7b33702b274d4de29b856a3ee8173d40 : FAILURE in 15m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b3b15426c954b03a9371f87fc2eac5d : SUCCESS in 6m 00s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/20071af2c50747fba8a89dd8ed8e9028 : SUCCESS in 22m 06s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/05458d46849f4d3f9a1110ed9af3a66b : SUCCESS in 50m 50s","accounts_in_message":[],"_revision_number":1},{"id":"b830a376b9eac1d36a0b096d2101054463faac2a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-07-28 20:22:14.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"bc7dc5c8d31ea11ab6a77766467ba51f91b62a3b","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-07-28 21:16:15.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)\n\nmakes sense, good catch, I\u0027ll review the test when I am more awake so just +1 for now","accounts_in_message":[],"_revision_number":2},{"id":"f342befb0f1b5641dba899bb7c3e7f76482313fb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-28 21:18:52.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a0e3e5972ddf4f91979ee4e26a5fad70 : SUCCESS in 8m 51s\n- grenade https://zuul.opendev.org/t/openstack/build/8084994866ed4f538098d626c18e4011 : SUCCESS in 56m 03s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/82dae166f8d048ac9393fc917043bd12 : SUCCESS in 40m 28s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d490169d0ca84cb181b44ec463c211cc : SUCCESS in 8m 05s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/e2bd8fb161be41d59291d879e641dff8 : SUCCESS in 9m 49s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/e0c795b0c39547afb0821156c856e925 : SUCCESS in 15m 48s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/e4fbfc573d2a4dbd86a3065c9cababe8 : SUCCESS in 16m 15s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/9cc815fa42d24a97a54eb6c216436ba4 : SUCCESS in 37m 47s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/567b805caf424bb3969d2ffc22f82a65 : SUCCESS in 18m 34s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/520ab1bb95f844afa212d74d2ccdb0e0 : SUCCESS in 39m 15s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/19a5606b2a094ead803d02832deabd49 : SUCCESS in 18m 58s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/9ac717873ee04e5381b2751ca435dd21 : SUCCESS in 43m 04s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/ef9139b5afd74a3995c7d9737e8ae7e4 : SUCCESS in 17m 56s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/fcf7e0f6166a4ec687f41f6e13c5abb0 : SUCCESS in 41m 48s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/aec5b1abe84b4e8bb707417ff1aa5cbb : SUCCESS in 43m 02s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/1e73cd2c247d4535b670a338a5699ebe : SUCCESS in 18m 37s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d27ab653cf714e1aa609242f82e43261 : SUCCESS in 8m 12s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/d4623ccf936945bea632ae4d1e8de698 : SUCCESS in 22m 16s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/27bc354ef59042468c5429b325ca992a : SUCCESS in 51m 06s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/205f752725d948c4abdd5c756f9aeef4 : SUCCESS in 50m 31s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e317938d3bea4fa18859e54066ec25a3 : SUCCESS in 41m 16s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/6872381db7644c159ae4a4a35ad8059f : SUCCESS in 39m 50s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b1358d29c0cf4f9587a1551e83a61430 : SUCCESS in 19m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/72ee698690644a2f8437aede16c37f64 : SUCCESS in 6m 15s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/23e482ca6b5e4bb9ad976df8b2eef09d : SUCCESS in 27m 38s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/26fd7324e441476f8900937f845d2bac : SUCCESS in 41m 51s","accounts_in_message":[],"_revision_number":2},{"id":"3a79d806cc510e902ac06e2be17362f196e7bf99","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-28 22:16:22.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e561e6fb049b4fb9a4ee67a109e921aa : SUCCESS in 53m 31s (non-voting)\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c25bf678748647c384be79879d1bff4b : SUCCESS in 43m 34s (non-voting)\n- swift-probetests-centos-8-arm64 https://zuul.opendev.org/t/openstack/build/72d90fc100ce40dabac4f62bc6ca0126 : FAILURE in 1h 51m 50s (non-voting)\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/a4a02c0c8c894d1886ce934c7c68a32c : SUCCESS in 50m 25s (non-voting)\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/4251def55a6943ecb6c19e046cb68199 : SUCCESS in 48m 19s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"91854ba5cfa34a749d96c9fa6b2c08015c83bd9e","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-07-29 10:17:02.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nThis deals with, at least, the case when an EC frag is being PUT, the client transaction has written the non-durable data file, then an ssync process renames it to be a durable data file. The client transaction then tries to do the same rename and an ENOENT is raised, but the durable datafile is intact.","accounts_in_message":[],"_revision_number":2},{"id":"68c1da256ff878bba75950bc33bef1bb41786a15","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-29 10:17:32.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"3aa38c7d7fc4ccc6607fe577fa6cc8553a2348b3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-29 11:31:28.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b1cc06c39696490a8d6a145df0037ae8 : SUCCESS in 9m 53s\n- grenade https://zuul.opendev.org/t/openstack/build/2d709e7b65034bcfa63ee2b35b5a2a5e : SUCCESS in 1h 08m 03s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/fc175b822e3d4fef9d51043772e3a038 : SUCCESS in 41m 27s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/c796513bf30b4e7086775ba1ae668c3a : SUCCESS in 12m 21s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/5c44d25c29af4c64b6615817631e0857 : SUCCESS in 13m 20s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/ca7903d72955440ebd5ef17d5039450c : SUCCESS in 17m 02s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/2613c7ce24d443f696b93a7f1ed52f3e : SUCCESS in 18m 46s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/8875b1c9cbbb46db9e928dff4e5a85da : SUCCESS in 16m 54s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/27153748a38f4207ac0926989fe612c3 : SUCCESS in 17m 55s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/8cd3047243204d39bd69c299c7ff8ab5 : SUCCESS in 19m 55s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/7084fdc6d690449685f6b3b1cb0c6dc5 : SUCCESS in 19m 04s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/04613bcad1d54f949b6b1b896ae55ac2 : SUCCESS in 18m 29s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/4f0f3d4a45364c0bbf3797e83e7d3e53 : SUCCESS in 29m 26s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/e12633f01ccc417fab0946de94e6e462 : SUCCESS in 17m 07s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/63b1da359aa44db38b786ae954c05aa5 : SUCCESS in 20m 27s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/99fd5ed7966b422c84732fe3569a8ebf : SUCCESS in 18m 57s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/0c7b6e587db542fda0e9a1edef2280da : SUCCESS in 7m 46s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/6fee6385691847d3944c62cf17c8c43e : SUCCESS in 51m 41s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/6449701ffe5748b19bd42efd42d9053c : SUCCESS in 50m 56s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4b281d5f9aae4564be48a85d328e4942 : SUCCESS in 45m 25s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f1ea66764c3744a397b8c08607b4d80b : SUCCESS in 44m 48s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9e212b7d5ded43da8afb07e7c22184f9 : SUCCESS in 19m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c41c9a9368d483c9477d7a75045cd6b : SUCCESS in 8m 14s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/8610e8c102d644229207835f237b9a3b : FAILURE in 29m 43s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7f1d4b867c7b4eb7a3b14dc8841052fc : SUCCESS in 53m 08s","accounts_in_message":[],"_revision_number":2},{"id":"41e8af0aa4beb3c31a69638e78b2b1b1baba0e44","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-07-29 16:21:27.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"549d8ff71e767075f7f8395b432c8e05a71f3fe7","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-07-29 16:38:59.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)\n\nJust addressing the nit, since CI failed anyway. Approving based on prior approval.","accounts_in_message":[],"_revision_number":3},{"id":"e5a7ebd80087e7098877ad0ffce53924d459167a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-29 17:35:50.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e9395d7ef9404012800a73630f4862b9 : SUCCESS in 9m 02s\n- grenade https://zuul.opendev.org/t/openstack/build/0e187f977b244c28bd97c5bd799fc2c6 : SUCCESS in 1h 08m 29s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/9c636ad839204dbc924df215522462f5 : SUCCESS in 46m 26s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ac00d41e04464fceb9cd218f40610386 : SUCCESS in 7m 48s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/eb8cddfe8fc34096bb19f4e1a7d87795 : SUCCESS in 8m 57s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/d419e45acc734f9a8cc55a2f5bfb5484 : SUCCESS in 19m 27s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/1ac93a5c94864b5cac49041f490a2e2e : SUCCESS in 16m 00s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/ed5b6aa1c8b7429f9533bd81d3c57ebe : SUCCESS in 16m 43s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/612c4799707b446297f9637af4f5e21d : SUCCESS in 17m 27s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/d5f0faa3806040d48fecb2612bf0ab56 : SUCCESS in 16m 08s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/7a8aee7241044e8aa8dc42326330e8d8 : SUCCESS in 18m 33s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/cd2d1513743d4e6d999d6c12475db06b : SUCCESS in 18m 46s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/60a691a0e16d4f09be1d7a930b6d6d62 : SUCCESS in 17m 35s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/295ed1ac1e634793b8a497fea0bde226 : SUCCESS in 22m 46s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/9de9a760b90a48998e0d66c5a00cc647 : SUCCESS in 18m 51s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/ab599e0994fc4ef488fc42f292feba99 : SUCCESS in 17m 54s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/30434888d42a4c89ac5630f416bbb838 : SUCCESS in 6m 25s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/57665534fff049818342106a39a78061 : SUCCESS in 53m 26s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/629d06eec48542f7a32db1e6638ba4c5 : SUCCESS in 49m 21s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/931efe5c2e7144b88ef0ae243cddb439 : SUCCESS in 47m 08s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/cbc095354bba44729ec1120f6a0d117c : SUCCESS in 42m 14s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/fd7b4239917b4d02a8e0e21f21717fc7 : SUCCESS in 38m 06s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d1be5945e9764b49b5c0754f80fbe732 : SUCCESS in 16m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/852fd8cfd6964fa9a6cc2cd12aaa9612 : SUCCESS in 5m 44s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/8180ec94144b4f04be806db6b0dd9a88 : SUCCESS in 27m 57s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b9b2d8c133b04615ae669d947504f9a4 : SUCCESS in 52m 03s","accounts_in_message":[],"_revision_number":3},{"id":"f29e84b72ac98bd3dcbc4f54a2b13d2fd307801f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-29 17:36:26.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"7656c74fdd546a9bec394c271f292aea771077a8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-29 17:46:49.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e142feee9965416a93121f23ca546e61 : SUCCESS in 34m 40s (non-voting)\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/bbab2a5b723e4fedaef48a39a94daaf9 : SUCCESS in 34m 45s (non-voting)\n- swift-probetests-centos-8-arm64 https://zuul.opendev.org/t/openstack/build/dedf5375db5b4834bcec4f04a7655bfe : SUCCESS in 1h 19m 06s (non-voting)\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/743051d9cfe04fd194dc4d0eff5ce248 : SUCCESS in 37m 14s (non-voting)\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/aeaaa4d11abf4b04b96bd9d0db9fd8e9 : SUCCESS in 1h 05m 21s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"b66e4fdbdb9d75e2e2772ea58ed0214d200c044b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-29 18:37:15.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a1203983d12448fdb141c36c06ca425c : SUCCESS in 10m 10s\n- grenade https://zuul.opendev.org/t/openstack/build/a01ad6ebd8ca4f9f92360bc84bfeff63 : SUCCESS in 59m 50s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d605c7c584074baca7cf48d97edf8663 : SUCCESS in 40m 58s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/0b581477466d4de7b7039068a0675c39 : SUCCESS in 8m 58s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/8425aa7ac60b468599da6edce0af9801 : SUCCESS in 9m 37s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/858c1bdd53fb4b9e83fcbcb4d9858b2c : SUCCESS in 16m 59s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/290164f61ff34b8d844acbe809b69451 : SUCCESS in 17m 55s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/a8cc1e7d22944a3a99ce0acdb9ea0d2b : SUCCESS in 20m 39s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/3b5239c1d4444595bf12f89c5efe75fc : SUCCESS in 20m 33s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/ac6e662491dd48079179a511d31eaf06 : SUCCESS in 20m 47s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/9264e97c9b284fee998a65d7f9317e2c : SUCCESS in 20m 49s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/78b60c425e324e6082d6550475ca2c36 : SUCCESS in 23m 12s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/dcca8d50dcba465eb2f8f65df88c908c : SUCCESS in 22m 05s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/7c64b438f4784d78bceba1ab0d5f2578 : SUCCESS in 19m 44s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/11352bfc657f4bbc9d6fc545f7066331 : SUCCESS in 19m 30s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/817efc5f5a674848b9db5d0eaf01bb11 : SUCCESS in 24m 06s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/73f99a5452ea441791d20bd71fd51a08 : SUCCESS in 7m 03s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/f97af5701fe743b89381d3be8af306f5 : SUCCESS in 50m 51s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/c9d53fc1e8f7449ea34837de447eeb4c : SUCCESS in 51m 02s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/72ad4f3c81cd49d9ab9eba28d8072223 : SUCCESS in 42m 43s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/bb66f0f1409e4d45a775c91a20e5f10e : SUCCESS in 40m 28s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/986dfc7f2d454029b488590d899d0168 : SUCCESS in 17m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/85fb02484ba243aab2246ef221f55221 : SUCCESS in 6m 40s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9df2d52f3666468292763872ae0db32a : SUCCESS in 33m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/556d1afd66cd466b971cdeb4b1079657 : SUCCESS in 46m 11s","accounts_in_message":[],"_revision_number":3},{"id":"860f7873448c023b0f56d68236b61d0ed179993a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-29 18:37:15.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"994e1d31a55d6ee24356c80a337b3ab96d9c4e8b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-29 18:38:46.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e5834e0387ee46b399828a4f2505bcd5 : SUCCESS in 1m 10s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/86f3bf0a093a44e19f4295e81cf629e8 : SUCCESS in 47s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/ba1a22b6d8b54fe89311d4a05c808f05 : SUCCESS in 51s (non-voting)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"c9b0795b9ecc430839eb3783327eb9c7c4aad5f8","revisions":{"884790119b15a32051e80f8e4b9ec09ce8ad5d9d":{"kind":"REWORK","_number":1,"created":"2021-07-28 18:22:23.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/68/802768/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/68/802768/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/68/802768/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/68/802768/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/68/802768/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/68/802768/1"}}},"commit":{"parents":[{"commit":"660e8386ddd602c52b451495832330cc22c51f2a","subject":"Merge \"s3api: Return KeyTooLongError when upload name exceeds constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/660e8386ddd602c52b451495832330cc22c51f2a"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-07-28 18:21:44.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-07-28 18:21:44.000000000","tz":-420},"subject":"diskfile: Stop raising ENOENT when we race to mark durable","message":"diskfile: Stop raising ENOENT when we race to mark durable\n\nChange-Id: Ia952303a16457a5075e9a4e060806855c21577f5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/884790119b15a32051e80f8e4b9ec09ce8ad5d9d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/884790119b15a32051e80f8e4b9ec09ce8ad5d9d"}]},"branch":"refs/heads/master"},"00a3e47578e951cdb7d335bbd745c772d4b365b1":{"kind":"REWORK","_number":2,"created":"2021-07-28 20:22:14.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/68/802768/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/68/802768/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/68/802768/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/68/802768/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/68/802768/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/68/802768/2"}}},"commit":{"parents":[{"commit":"660e8386ddd602c52b451495832330cc22c51f2a","subject":"Merge \"s3api: Return KeyTooLongError when upload name exceeds constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/660e8386ddd602c52b451495832330cc22c51f2a"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-07-28 18:21:44.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-07-28 20:22:07.000000000","tz":-420},"subject":"diskfile: Stop raising ENOENT when we race to mark durable","message":"diskfile: Stop raising ENOENT when we race to mark durable\n\nChange-Id: Ia952303a16457a5075e9a4e060806855c21577f5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/00a3e47578e951cdb7d335bbd745c772d4b365b1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/00a3e47578e951cdb7d335bbd745c772d4b365b1"}]},"branch":"refs/heads/master"},"c9b0795b9ecc430839eb3783327eb9c7c4aad5f8":{"kind":"REWORK","_number":3,"created":"2021-07-29 16:21:27.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/68/802768/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/68/802768/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/68/802768/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/68/802768/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/68/802768/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/68/802768/3"}}},"commit":{"parents":[{"commit":"660e8386ddd602c52b451495832330cc22c51f2a","subject":"Merge \"s3api: Return KeyTooLongError when upload name exceeds constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/660e8386ddd602c52b451495832330cc22c51f2a"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-07-28 18:21:44.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-07-29 16:21:16.000000000","tz":-420},"subject":"diskfile: Stop raising ENOENT when we race to mark durable","message":"diskfile: Stop raising ENOENT when we race to mark durable\n\nChange-Id: Ia952303a16457a5075e9a4e060806855c21577f5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c9b0795b9ecc430839eb3783327eb9c7c4aad5f8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c9b0795b9ecc430839eb3783327eb9c7c4aad5f8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
