)]}'
{"id":"openstack%2Fswift~803638","triplet_id":"openstack%2Fswift~master~I2761b37cdb57d4a0884c37ee749c688277ff4b37","project":"openstack/swift","branch":"master","hashtags":[],"change_id":"I2761b37cdb57d4a0884c37ee749c688277ff4b37","subject":"Kill coros when shutting down ECAppIters","status":"NEW","created":"2021-08-05 17:55:46.000000000","updated":"2021-08-09 16:28:00.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"ecb4c75005702769a37ad0ffc67a14e468335158","_number":803638,"virtual_id_number":803638,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2021-08-05 18:52:25.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-05 18:52:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8ca9b24b9c3c8ba60b0c46563fb1d5bf3db0a382","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-08-05 17:55:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7c5f3fbc1714968aff70286184b4dabd0658602f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-05 18:52:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/86863b81713444528f637f8141a1e18c : SUCCESS in 8m 49s\n- grenade https://zuul.opendev.org/t/openstack/build/787512779c9349c881c5ab79db722b0c : SUCCESS in 48m 14s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/abdcfdda8163416ba9e342ccc1e4a92d : SUCCESS in 41m 18s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/93c19a79285740fbb5e3a057e26d2329 : SUCCESS in 8m 16s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/2f58746bb38f42f99809fa765ca02ef1 : SUCCESS in 8m 36s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/fd7a20f2928b4dc3a9a69324c5d362fd : SUCCESS in 14m 12s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/b198bb0f09284912b80879cc247ebafb : SUCCESS in 15m 16s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/a82f6115202f4076a8d40af65fa1ddd4 : SUCCESS in 20m 26s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/06a8aefd17f543ff96356cfdd5b38573 : SUCCESS in 26m 19s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/5b1706501c344a77b1c7c8ecdea607ff : SUCCESS in 14m 43s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/ea198e140dc54945bbf16c4edf2b72f6 : SUCCESS in 21m 04s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/48373e6e20134af7a04174e4ec28054a : SUCCESS in 17m 09s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/788ba1de653244509b092f054a31dfbe : SUCCESS in 16m 43s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/98d77ad789d44e9d9659b825a5ed6559 : SUCCESS in 20m 47s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/f0b4bd089e5a4fa1a08177e6dc6f37e9 : SUCCESS in 15m 04s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/83316cb8fc4e44b98181aa52a553000c : SUCCESS in 17m 28s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/0c34bc8f87bf489da468b81a48cacdf6 : SUCCESS in 6m 45s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/fd7775efc02e454fb988636e59351580 : SUCCESS in 22m 00s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/ae30931c74644744817bf0c302c134a5 : SUCCESS in 46m 55s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/33dc2e22a32c4dddaa32502bf02368ee : SUCCESS in 54m 43s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/58fb513804b74b6e83c25aa880790aac : SUCCESS in 37m 23s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a92a5eab393f4016aa2335735a0af9fc : SUCCESS in 47m 22s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/407be29072224ee590c661b0b381de98 : SUCCESS in 18m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/14f04deea2054611af239eb8138ff8fa : SUCCESS in 6m 14s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/35f9370678fe4e7cb8a74d808cf05074 : SUCCESS in 24m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e3a006a7fa3347a79c45d2de41c4faa2 : SUCCESS in 43m 03s","accounts_in_message":[],"_revision_number":1},{"id":"7c71cd7f79a2313a2b22f3a9dfb9dcf7d5518b9f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-05 19:36:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c661caf6c75841859a4f8f4dfe8d5e5b : SUCCESS in 42m 01s (non-voting)\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f981069b279140f89d812efba10146ea : SUCCESS in 42m 34s (non-voting)\n- swift-probetests-centos-8-arm64 https://zuul.opendev.org/t/openstack/build/b8a4a36de482450cb64e64164d84e266 : SUCCESS in 1h 23m 04s (non-voting)\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/853ecdf3cc824109bd032d37786e3439 : SUCCESS in 42m 37s (non-voting)\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/6733e717b4e644d6ab54d97d561be374 : SUCCESS in 39m 03s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ecb4c75005702769a37ad0ffc67a14e468335158","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-08-09 16:28:00.000000000","message":"Patch Set 1:\n\nClay points out that this likely relates to https://review.opendev.org/c/openstack/swift/+/796735 -- we should try his repro script against just this patch and see if we can abandon one or the other.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"cbc54331845f8eb7b4db058ad2949638411e50fb","revisions":{"cbc54331845f8eb7b4db058ad2949638411e50fb":{"kind":"REWORK","_number":1,"created":"2021-08-05 17:55:46.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/38/803638/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/38/803638/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/38/803638/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/38/803638/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/38/803638/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/38/803638/1"}}},"commit":{"parents":[{"commit":"b3c5a9a9e01aafa2a0819636b2eb20640cf7b529","subject":"Merge \"s3api: Fix (async) multi-delete of MPUs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b3c5a9a9e01aafa2a0819636b2eb20640cf7b529"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-08-05 17:08:50.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-08-05 17:54:52.000000000","tz":-420},"subject":"Kill coros when shutting down ECAppIters","message":"Kill coros when shutting down ECAppIters\n\nOtherwise you can find yourself in a situation where you try to close\nthe stashed iter, which eventually closes the part_iter in ECFragGetter,\nbut one of the coros is still consuming the part_iter and you get a\n\"generator already executing\" error.\n\nChange-Id: I2761b37cdb57d4a0884c37ee749c688277ff4b37\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cbc54331845f8eb7b4db058ad2949638411e50fb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cbc54331845f8eb7b4db058ad2949638411e50fb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
