)]}'
{"id":"openstack%2Fswift~803729","triplet_id":"openstack%2Fswift~master~I1b745e2ef8d8f295b2a8a083b3f6717286039f57","project":"openstack/swift","branch":"master","topic":"container-server-return-specified-api-objs","hashtags":[],"change_id":"I1b745e2ef8d8f295b2a8a083b3f6717286039f57","subject":"Preserve X-Backend-Storage-Policy-Index when recursively listing shards","status":"ABANDONED","created":"2021-08-06 10:48:04.000000000","updated":"2021-08-09 11:41:42.000000000","total_comment_count":2,"unresolved_comment_count":2,"has_review_started":true,"meta_rev_id":"5c92d404efea9fa3bdac38f01f6a3f5e64a66842","_number":803729,"virtual_id_number":803729,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2021-08-06 11:46:05.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-06 11:46:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-09 05:10:48.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"CC"}],"messages":[{"id":"186126db5c42443f88c12aaad9564fa02a0afa9b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-08-06 10:48:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"baab27cfbe3ea13c2e25a86f26a975873d900f64","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-06 11:46:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ca99fca2fcd45bdbadaa1ecb3c086e0 : SUCCESS in 8m 34s\n- grenade https://zuul.opendev.org/t/openstack/build/574b2ec38afa410d9d42b2e42ac9d717 : SUCCESS in 56m 07s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/b9d22fa2af7c4235bafbe818b086ebd4 : SUCCESS in 44m 45s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/064c5989602645d0b2afe365d2252726 : SUCCESS in 9m 28s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/8a53e88af97c43faa01a7edb9d04b07f : SUCCESS in 8m 42s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/fac4d4a8b83c469bb092ac9144db0d7d : SUCCESS in 17m 19s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/8289ae629bc5461ca4a3e697e38890cb : SUCCESS in 18m 10s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/b6aab62c56c1449b80d0dc6c21ccda0e : SUCCESS in 15m 43s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/ab8ad46ea3454d3f9f9fdf7ada949667 : SUCCESS in 21m 12s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/3113eb5a65374bcdb1d2bc3be8ec5e9f : SUCCESS in 17m 04s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/c6c90fdb91814ac9b74e237756ea6c0f : SUCCESS in 15m 43s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/9952ea2ef2564611a723d4fa5bd39a53 : SUCCESS in 19m 02s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/1bbb108bcfca4f49a9af93b2a936fdfb : SUCCESS in 20m 24s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/75112e02a29c4598b83025590fd960a9 : SUCCESS in 16m 29s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/c40db498a3b145ac9e2d278304af177b : SUCCESS in 21m 50s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/753279db99794906bf3316dd92af68e8 : SUCCESS in 19m 13s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/3d1b221275294d2cbdf4ccbca0052426 : SUCCESS in 7m 10s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b078c93b42394ce0b5aa97cb7c703add : SUCCESS in 26m 13s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/15759c80f1994c689e8b03cd804cafdd : SUCCESS in 51m 45s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/f6f7fbe344164d1dafbfd055a3690637 : SUCCESS in 47m 09s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/2ee1cbe3f8f04e18a8543cc217e231fb : SUCCESS in 40m 43s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/54527db95c434ef19ed50f8af5b63093 : SUCCESS in 37m 48s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4589d3f475d24b4d9071132f219ad856 : SUCCESS in 15m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aef4955f302244089b971bb6414a175f : SUCCESS in 7m 21s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/46d3ba8f92884979856f7560ad08d377 : SUCCESS in 22m 35s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/48b66435e1c04b3b819857ca0d3812ba : SUCCESS in 40m 52s","accounts_in_message":[],"_revision_number":1},{"id":"0381deed5fce3ccba0f0369692969605f6586763","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-06 12:51:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/89c1137f5f2a481b961f0639f53f392d : SUCCESS in 41m 12s (non-voting)\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c30d0fc7bb074160a7eaf4e9a8252dee : SUCCESS in 55m 05s (non-voting)\n- swift-probetests-centos-8-arm64 https://zuul.opendev.org/t/openstack/build/8f2d04dee049432ebd074204264cf105 : TIMED_OUT in 2h 01m 16s (non-voting)\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/7439091b08b64b768161f5d5168cf416 : SUCCESS in 33m 03s (non-voting)\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/dc46594553fc4d4c86d18bef2803408a : SUCCESS in 1h 10m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"081e9768ba8344a315c5108126d873b7373bddfa","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2021-08-09 05:10:48.000000000","message":"Patch Set 1:\n\n(2 comments)\n\nMerged into parent","accounts_in_message":[],"_revision_number":1},{"id":"5c92d404efea9fa3bdac38f01f6a3f5e64a66842","tag":"autogenerated:gerrit:abandon","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-08-09 11:41:42.000000000","message":"Abandoned\n\nsquashed to parent","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"76d7a8be59926ecbd301b602356cc87c53d246e6","revisions":{"76d7a8be59926ecbd301b602356cc87c53d246e6":{"kind":"REWORK","_number":1,"created":"2021-08-06 10:48:04.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/29/803729/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/29/803729/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/803729/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/803729/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/803729/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/29/803729/1"}}},"commit":{"parents":[{"commit":"1e13e3fd0321f13bca2dd0be5bef5c917cb65e7e","subject":"container-server: return objects of a given policy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1e13e3fd0321f13bca2dd0be5bef5c917cb65e7e"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-08-06 10:47:16.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-08-06 10:47:16.000000000","tz":60},"subject":"Preserve X-Backend-Storage-Policy-Index when recursively listing shards","message":"Preserve X-Backend-Storage-Policy-Index when recursively listing shards\n\nChange-Id: I1b745e2ef8d8f295b2a8a083b3f6717286039f57\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/76d7a8be59926ecbd301b602356cc87c53d246e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/76d7a8be59926ecbd301b602356cc87c53d246e6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
