)]}'
{"id":"openstack%2Fswift~804234","triplet_id":"openstack%2Fswift~master~I6990748b894598f35a5b6ee08d5c2b6ff1c6d0ef","project":"openstack/swift","branch":"master","topic":"container-server-return-specified-api-objs","hashtags":[],"change_id":"I6990748b894598f35a5b6ee08d5c2b6ff1c6d0ef","subject":"Set X-Backend-Storage-Policy in original listing request","status":"ABANDONED","created":"2021-08-11 13:20:21.000000000","updated":"2021-08-12 12:59:27.000000000","total_comment_count":7,"unresolved_comment_count":5,"has_review_started":true,"meta_rev_id":"cc2028feb84a34856a74f9ec65e6b437542a5d96","_number":804234,"virtual_id_number":804234,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:check","value":1,"date":"2021-08-12 11:19:32.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-11 14:24:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-11 16:55:40.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"CC"},{"updated":"2021-08-11 18:19:46.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"},{"updated":"2021-08-12 03:41:19.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"CC"}],"messages":[{"id":"706c7ee142ff672a7ffbdfa9a250d19f79645317","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-08-11 13:20:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"485d09351c7c51117a70f00bf8dd5354184c497c","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-08-11 13:23:57.000000000","message":"Patch Set 1: Workflow-1\n\n(2 comments)\n\nDNM without addressing the TODO","accounts_in_message":[],"_revision_number":1},{"id":"d621a85bccd87bbd8089bcce161352164e76a47b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-11 14:24:13.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6c866b8fa24249ce83e2658dea91016f : SUCCESS in 9m 43s\n- grenade https://zuul.opendev.org/t/openstack/build/4911dab6dd6f42e484a13088ae1da3d8 : SUCCESS in 1h 02m 10s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/8cea9cd393ef4115b6ebe889eb3b046a : FAILURE in 13m 58s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/8d639e6847424908bc251ae415f08e81 : SUCCESS in 9m 21s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/16deef0c52db4b2db91be9bb8de97993 : SUCCESS in 10m 46s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/d0b06105c36d4c37bf12ce554f409a77 : SUCCESS in 17m 33s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/e76217438bbe4b528b77a7d02f96fead : SUCCESS in 18m 30s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/e4742134e732495480582943abe1b7e7 : SUCCESS in 16m 15s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/456a946661c4477783375a9cee972a8b : SUCCESS in 17m 27s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/038bf231d00b4784a9dca025e9cac264 : SUCCESS in 17m 19s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/92f09fa02198485fb89a55f3511314f1 : SUCCESS in 19m 01s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/fd6fd0462e0a4c988465b3ca978c674d : SUCCESS in 18m 12s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/d7aa42a8210248b6971f888a2fb594f5 : SUCCESS in 18m 55s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/55e4ce367cb1451f9e6cea4b274f70d0 : SUCCESS in 17m 49s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/b06ec727641846cd98dd6175e2e40da6 : SUCCESS in 19m 27s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/45f41a32d9794cc8bd6b76431711dbe2 : SUCCESS in 22m 40s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/9cc9f65be2d240cab44805c9b1f448f3 : SUCCESS in 10m 04s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/813e6685a67245e19f71412e77d62956 : SUCCESS in 18m 55s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/6941500593d047d683d80d0432274eed : SUCCESS in 55m 26s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/865cc2009bb74a29a07687f42d328566 : SUCCESS in 46m 40s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f6044919ba8540bdaa246aa3a8425d50 : SUCCESS in 38m 04s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/257e55673e5a43a9b8b86ec18ed17f5f : SUCCESS in 43m 37s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b8aca65bdda146b9849472fa8bf7b6bb : SUCCESS in 17m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ffd6d247bef44723b4cf778112009f92 : SUCCESS in 6m 21s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e867e706e5614d70b9c700a4265c46c1 : SUCCESS in 23m 23s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ed84dcc182d8466583dbd7e245f22772 : SUCCESS in 50m 12s","accounts_in_message":[],"_revision_number":1},{"id":"4978e7ca540ddf1fef9658cf90cef54ea396ea09","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-11 14:51:20.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/df0272d361b64dbb9c27de0d2d6a75c1 : SUCCESS in 34m 01s (non-voting)\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/dd3551e892d5497d8307a4a10494235c : SUCCESS in 33m 02s (non-voting)\n- swift-probetests-centos-8-arm64 https://zuul.opendev.org/t/openstack/build/6113c18df69f4e249161a455460724a7 : SUCCESS in 1h 23m 17s (non-voting)\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/ee997bdfc250407eb128e4bea54b4ae5 : SUCCESS in 33m 48s (non-voting)\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/a10b625a545a4cbc8edb7bf86656be85 : SUCCESS in 33m 04s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"796b6aefca832f1a963ee40ad63538c5dfbc4b33","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-08-11 16:55:40.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6a935eb4794182a3c99bf9cfe1057eaae6d3ad6a","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2021-08-11 18:19:46.000000000","message":"Patch Set 1: Code-Review+1\n\n(2 comments)\n\nI agree with all the other comments, and would love to see this squashed","accounts_in_message":[],"_revision_number":1},{"id":"acf6d8e94c1cb24d236f3cf62480732f66ca4886","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2021-08-12 03:41:19.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3b36ada7e59bf7e9dfd9dc452fb21671430ea63c","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-08-12 09:50:25.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"61c439d410a601cfb6d80717e1b55aaae8045208","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-08-12 10:07:42.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f4baec0359eb07eaee4f8793bb2fb1513eccbc7c","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-08-12 10:14:22.000000000","message":"Patch Set 2:\n\nUpdated but planning to squash into parent.","accounts_in_message":[],"_revision_number":2},{"id":"1d182f294813f8893cabd48675cd88a1ed4ebe0d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-08-12 10:16:11.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"5fb5717e4a74d74b2a90da6b7f5ce6def636864b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-12 11:19:32.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0fd465331d0445dd89b33b29339cf857 : SUCCESS in 8m 43s\n- grenade https://zuul.opendev.org/t/openstack/build/457684a292cc4e00b9c1c48e01800b47 : SUCCESS in 1h 01m 18s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/833223286e4d4be8b4f68c75f96347b9 : SUCCESS in 41m 10s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f4c333cbed884eebafdc245c4dad81f1 : SUCCESS in 8m 28s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/ad51f264863148bbb4fc2941ab5fa7f0 : SUCCESS in 9m 14s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/9eea63685b4e46babe0cd24a07ddcd42 : SUCCESS in 15m 27s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/13edf834e816445fbbedc04c1d65d882 : SUCCESS in 16m 45s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/d1afed3ff4a646b3a8c7b28e4b6cae76 : SUCCESS in 15m 56s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/7f7eb5ca1df24d159357ae607ce99b6a : SUCCESS in 18m 45s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/fb0cd5fd5c6c4ac1a1890fbaf0eaefcc : SUCCESS in 17m 36s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/7182e53a7be8439cb35b8ae41eec9503 : SUCCESS in 18m 22s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/5cc142fe38dd4ed9a1e573e4b04a7830 : SUCCESS in 16m 42s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/231ea5fe8bce48f685836ad2b0835b95 : SUCCESS in 19m 34s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/3eabe5d7d01b4e7b8522ccbdf122ecd4 : SUCCESS in 17m 55s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/67a309cc65294974b04c94c320bf2a11 : SUCCESS in 19m 34s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/59163bf73bc0480c87237e5b58965e85 : SUCCESS in 17m 36s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/85f57d07c1fb45e4a3c5b871cd5e8ea4 : SUCCESS in 8m 28s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/db96ea62867f4804b03b69c5346cbd9d : SUCCESS in 22m 11s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/719a3957b46440de963271facdc6c930 : SUCCESS in 52m 26s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/04e8308b79a14319a0b1f1c918f0ba41 : SUCCESS in 45m 19s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/0fd89c12f00f40158cbfcd070f3f9fce : SUCCESS in 41m 12s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/697eb8be128a44998822bb193dc005e7 : SUCCESS in 37m 52s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b68303f930fd4556b33adb2d119dc65f : SUCCESS in 15m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7a81a8fab5ae43409d243fe99208a7f4 : SUCCESS in 6m 13s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/4152393871f0471ab88d2ca50c84be6e : SUCCESS in 27m 14s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/23ccef6deed142b6ba548ed3105fa65a : SUCCESS in 40m 19s","accounts_in_message":[],"_revision_number":3},{"id":"92dd4ec04cee0ecc8f1fdad7cb46db3c00b31d64","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-12 11:41:04.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6dd2a5db182a441d86af15ce73387396 : SUCCESS in 32m 17s (non-voting)\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6285feee67084208b1c6c7a877b25ac6 : SUCCESS in 40m 32s (non-voting)\n- swift-probetests-centos-8-arm64 https://zuul.opendev.org/t/openstack/build/f86def5b963f42e0b75496d0ee1069eb : SUCCESS in 1h 18m 57s (non-voting)\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/6cd349c020a347a4b383d113bd7b4616 : SUCCESS in 54m 31s (non-voting)\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/bbbeb734a1ff4d93a4b5eb72bbe27f89 : SUCCESS in 32m 15s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"cc2028feb84a34856a74f9ec65e6b437542a5d96","tag":"autogenerated:gerrit:abandon","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-08-12 12:59:27.000000000","message":"Abandoned\n\nsquashed into parent","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"d8ca5f298fa93f79f845cd97828d09fd73fea44c","revisions":{"8815fbcec0e66c5693d806e87d6684d9daa07c57":{"kind":"REWORK","_number":1,"created":"2021-08-11 13:20:21.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/34/804234/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/34/804234/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/34/804234/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/34/804234/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/34/804234/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/34/804234/1"}}},"commit":{"parents":[{"commit":"63c6933f5d938bea47c365407ab7cea0b015b033","subject":"container-server: return objects of a given policy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/63c6933f5d938bea47c365407ab7cea0b015b033"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-08-11 13:18:50.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-08-11 13:18:50.000000000","tz":60},"subject":"Set X-Backend-Storage-Policy in original listing request","message":"Set X-Backend-Storage-Policy in original listing request\n\nChange-Id: I6990748b894598f35a5b6ee08d5c2b6ff1c6d0ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8815fbcec0e66c5693d806e87d6684d9daa07c57"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8815fbcec0e66c5693d806e87d6684d9daa07c57"}]},"branch":"refs/heads/master"},"9b65a819fa012c1599ca1085cf96248f5dc78300":{"kind":"REWORK","_number":2,"created":"2021-08-12 10:07:42.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/34/804234/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/34/804234/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/34/804234/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/34/804234/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/34/804234/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/34/804234/2"}}},"commit":{"parents":[{"commit":"5f519176ed2bad028cb272203886ae2c758cef93","subject":"container-server: return objects of a given policy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5f519176ed2bad028cb272203886ae2c758cef93"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-08-11 13:18:50.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-08-12 10:07:00.000000000","tz":60},"subject":"Set X-Backend-Storage-Policy in original listing request","message":"Set X-Backend-Storage-Policy in original listing request\n\nChange-Id: I6990748b894598f35a5b6ee08d5c2b6ff1c6d0ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9b65a819fa012c1599ca1085cf96248f5dc78300"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9b65a819fa012c1599ca1085cf96248f5dc78300"}]},"branch":"refs/heads/master"},"d8ca5f298fa93f79f845cd97828d09fd73fea44c":{"kind":"REWORK","_number":3,"created":"2021-08-12 10:16:11.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/34/804234/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/34/804234/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/34/804234/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/34/804234/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/34/804234/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/34/804234/3"}}},"commit":{"parents":[{"commit":"5f519176ed2bad028cb272203886ae2c758cef93","subject":"container-server: return objects of a given policy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5f519176ed2bad028cb272203886ae2c758cef93"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-08-11 13:18:50.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-08-12 10:16:01.000000000","tz":60},"subject":"Set X-Backend-Storage-Policy in original listing request","message":"Set X-Backend-Storage-Policy in original listing request\n\nChange-Id: I6990748b894598f35a5b6ee08d5c2b6ff1c6d0ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d8ca5f298fa93f79f845cd97828d09fd73fea44c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d8ca5f298fa93f79f845cd97828d09fd73fea44c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
