)]}'
{"id":"openstack%2Fswift~804695","triplet_id":"openstack%2Fswift~master~Ieab6238030e8c264ee90186012be6e9da937b42e","project":"openstack/swift","branch":"master","topic":"container-server-return-specified-api-objs","hashtags":[],"change_id":"Ieab6238030e8c264ee90186012be6e9da937b42e","subject":"container GET: return 503 if policy index mismatches","status":"MERGED","created":"2021-08-16 10:33:01.000000000","updated":"2021-08-17 08:08:57.000000000","submitted":"2021-08-16 23:04:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":3,"has_review_started":true,"submission_id":"804695-container-server-return-specified-api-objs","meta_rev_id":"c208e0a893e57cc9ba80e0f707854287b6a496ee","_number":804695,"virtual_id_number":804695,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-16 23:04:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":2,"date":"2021-08-16 21:56:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-08-16 18:31:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":1,"date":"2021-08-16 21:56:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-16 11:59:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-16 18:31:35.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2021-08-16 21:56:16.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"}],"messages":[{"id":"ef6efa31548cefcfe6202423bb16b7a83a0eb6c3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-08-16 10:33:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7f4051e28d8d91d1f2fc75493ac8287b3cdeda54","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-08-16 10:35:00.000000000","message":"Uploaded patch set 2: New patch set was added with same tree, parent, and commit message as Patch Set 1.","accounts_in_message":[],"_revision_number":2},{"id":"7d8f4da89f7e82cd49dfd32343c6652404d7163b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-08-16 10:57:06.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"b8f596ab1c4b8c358de86fd1101926f4378a2ba7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-16 11:59:40.000000000","message":"Patch Set 3: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/32441cbd54e9485391b5a72e5727d998 : SUCCESS in 8m 49s\n- grenade https://zuul.opendev.org/t/openstack/build/96933794b7354c8aa408e9a744dc2fbe : SUCCESS in 57m 24s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e7c981468df04d58ba605ce9474c63c0 : SUCCESS in 33m 49s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d5e3ff06e25048e095a6d624ec4fb719 : SUCCESS in 8m 47s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/8bd8eb6748344fc3a07b47f19ed78585 : SUCCESS in 8m 58s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/d16e365682944a3d856b3875fb7a3859 : SUCCESS in 15m 54s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/f2e39a777df949fc9ac7d80f92201851 : SUCCESS in 16m 46s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/b68798d80f2a41ec88c71d6769a73caa : SUCCESS in 16m 11s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/b742a0f4019347d496866b14194c4309 : SUCCESS in 16m 20s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/10705324d56d4717a4af93ba460ab349 : SUCCESS in 16m 53s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/bdcfb5426cb54b039460f1b95e5980af : SUCCESS in 16m 48s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/10f24ef3ac884e0e801d5b21107bfcf7 : SUCCESS in 17m 03s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/1ccccfb4de564d6d94f67037c2a6230f : SUCCESS in 18m 10s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/0f2e7a9fc6c24b538ed3547f21131678 : SUCCESS in 16m 36s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/2be6501b031c4553b6d325881f24963c : SUCCESS in 17m 17s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/733da1af436f486e9e7f5de785b580b1 : SUCCESS in 17m 38s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f90a5f72b24a4d8c95ad38d41327edf5 : SUCCESS in 7m 22s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/5c5df88c1894460493e9c5c48d36eb39 : SUCCESS in 19m 28s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/44f9b739026c412fb62f4d4355b43c77 : SUCCESS in 57m 27s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/ce48b885bfb941a8bb9e730afb23dc7c : SUCCESS in 50m 11s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/8a83d04bd581457996cefabd3d7350b2 : SUCCESS in 38m 40s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ffa4a4fbe8bf4e17b4243fdebbdb7bff : SUCCESS in 40m 34s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fcf821c96e0a4c1985cdaf738e3070d8 : SUCCESS in 15m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/974a86be25a04504af0be2d032374020 : FAILURE in 5m 40s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e681fb4faa1644969cc3bc59f1da2dd5 : SUCCESS in 22m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8da39cd1c21440aeb8554571b7cd1ec3 : SUCCESS in 53m 38s","accounts_in_message":[],"_revision_number":3},{"id":"4ef85691e5c9e65931c903a0dd1f3a5d1662e6c3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-16 12:36:22.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9bf9feb2985448478db0e0105f2c2b57 : SUCCESS in 37m 02s (non-voting)\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/fddfaa387007447ebe6714b6b580fa32 : SUCCESS in 36m 43s (non-voting)\n- swift-probetests-centos-8-arm64 https://zuul.opendev.org/t/openstack/build/1f3a11fc9eed4d2a82c4cceb8ea1a9c0 : SUCCESS in 1h 30m 29s (non-voting)\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/6fc351dbc64548229c19284afd42c0b7 : SUCCESS in 36m 34s (non-voting)\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/b965740aea4e43f091cd748fdfb48ff8 : SUCCESS in 46m 43s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"d9a8b00b1f1dda0c7b3a0e081281ab4f4d26fa99","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-08-16 12:50:54.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"15c16b2c3451df9ae12c4fbc7b61ab5ed15945ba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-16 13:53:09.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a931acf70a704614a5dd7170ff1db8da : SUCCESS in 10m 54s\n- grenade https://zuul.opendev.org/t/openstack/build/0835a587db784cd3817ad8f0338eacae : SUCCESS in 1h 01m 02s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/80631c948a9d47b19e57a1c9c701ee7d : SUCCESS in 36m 58s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/15e300f3e7eb4ec8a6a8e7363193ec9e : SUCCESS in 8m 25s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/0d62f86a807e4be5ab4f7feee4273de6 : SUCCESS in 9m 39s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/a7df55f7272744dba0ac86b791c13573 : SUCCESS in 19m 49s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d5ff3f1c2c8f433a8d415dc43b291012 : SUCCESS in 17m 53s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/16a11b4ae54b489cbeab1fef468fe0ca : SUCCESS in 21m 03s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/68f000bb2a424089a34fa7bfd6e9bcd7 : SUCCESS in 21m 00s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/b88e94540770423e9e377ce6f1976ba5 : SUCCESS in 17m 23s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/302e817b10974572ac67aaaef9f5860d : SUCCESS in 18m 56s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/9f45c663dd4b441e8c367668a5b5c60b : SUCCESS in 18m 26s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/b4e05990b1744238a08ab0b20b284d86 : SUCCESS in 19m 08s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/416f26cfbd8c44729cfd0dc643642794 : SUCCESS in 17m 48s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/a1039ee5ac0448abba9e4f4cab4321e9 : SUCCESS in 19m 18s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/965fe73aa12741f18edf802aacb57bb5 : SUCCESS in 19m 10s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/90983fa2d031400ca54ebf63c026dad5 : SUCCESS in 7m 56s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/9482301c4c5a4e968519722c8381f6f3 : SUCCESS in 24m 12s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/57e759e7bcf04340a7fd09a4fdfe2170 : SUCCESS in 56m 24s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/d06f1b0ed8c348f08106a9ea5975ebed : SUCCESS in 53m 07s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e9f2dec134884ce6b1a66d337e0e39ff : SUCCESS in 42m 38s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e9323332ef5b4a27b6ff4fa825c20807 : SUCCESS in 40m 35s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ea79e59b8ef74379813e5cf0d8794654 : SUCCESS in 19m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/79f9ce8ca1864f12b38b950b06b1bc3e : SUCCESS in 6m 37s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/0ee096585f314ed3855ae54938d89138 : SUCCESS in 27m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8fe00887568340bdae3bd5d59bdbb238 : SUCCESS in 47m 40s","accounts_in_message":[],"_revision_number":4},{"id":"e4cd5fb63ce6e8763c88a8d7d2ce78d4f902b20f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-16 14:37:06.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8fb17789b480436d9e0fe552398b619d : SUCCESS in 35m 15s (non-voting)\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0da031104b6a4de09fb4f7952dc62714 : SUCCESS in 40m 01s (non-voting)\n- swift-probetests-centos-8-arm64 https://zuul.opendev.org/t/openstack/build/c7848d115b64490cab849399f9c6912b : FAILURE in 1h 33m 25s (non-voting)\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/50bbb32787974826b9e84939b40753ef : SUCCESS in 44m 03s (non-voting)\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/313ab8ca679f49968e6ad594500b6b81 : SUCCESS in 43m 05s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"6e7ef15c77a6f3e473094d68002f4d462ec25e4c","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-08-16 18:31:35.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"a245bca42988e8c5a8c68985d88e50c330f34bfc","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2021-08-16 21:56:16.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\n(3 comments)\n\nseems helpful","accounts_in_message":[],"_revision_number":4},{"id":"4c7a4d2a3c1b592c4e9b42138d4a61b82d7d95c7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-16 21:56:53.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"1f5c1438897c923082c88b39a5cb6cc269d7f354","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-16 23:04:24.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2121f7b43cc14c9f9450d97eb4b44951 : SUCCESS in 10m 51s\n- grenade https://zuul.opendev.org/t/openstack/build/ff6b57de5a304a8bb9fa04035b878018 : SUCCESS in 1h 02m 21s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a401a9ef5bf84561802cb92561b6e228 : SUCCESS in 46m 44s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/3a708532e7114d40b3b3d941a44a2406 : SUCCESS in 8m 48s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/1d6294191c084803adbdde217d547853 : SUCCESS in 9m 13s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/06c86f7958ba44519a060567e2c1d8f6 : SUCCESS in 16m 46s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/60530c1b308a4ba8a0f35c806d5ed908 : SUCCESS in 17m 16s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/3fbcade1a16f442d91f25007d9c5314b : SUCCESS in 20m 08s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/dc53a67eadea4d87bdeb74ae0b0deb49 : SUCCESS in 15m 53s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/29b51d9f3eb24578b2f39a3f8d8561dc : SUCCESS in 15m 43s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/01ed54059c2749a3a129883ac7c6f1d4 : SUCCESS in 16m 41s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/b3407737c1ea4e749764207a27897ac7 : SUCCESS in 17m 00s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/ffe1976e319c42529ac93b3728669f30 : SUCCESS in 18m 08s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/d78f15ea5849429e9595b93747518df7 : SUCCESS in 16m 11s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/fc093c84fda543ea8813fa7fe28647c3 : SUCCESS in 16m 47s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/69cebd2436f94cfcac005f56591db660 : SUCCESS in 17m 26s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/6a455820cc5a4160bf7fd17fa994cc97 : SUCCESS in 6m 28s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/857a49eaa9114bb4aab93e2cb1437213 : SUCCESS in 46m 35s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/e5aa9cba61a741a0a432a29a7482026c : SUCCESS in 46m 36s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b2ac8ffb438b4b34a7974a2f8df42955 : SUCCESS in 39m 30s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/062385195aeb49a29d6f2a743317e59f : SUCCESS in 37m 29s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3af0ab53ec134f04b19b4722eae9005e : SUCCESS in 15m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/79c62d126f404032928ad4f10b90659e : SUCCESS in 5m 54s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/fd97bc1a90bc4bacad61fe3372775727 : SUCCESS in 22m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2179edb912994ab8923cd5c0ba5ca033 : SUCCESS in 48m 09s","accounts_in_message":[],"_revision_number":4},{"id":"9a3b521f416e08a4ee5629af868bd2f1895a13e6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-16 23:04:25.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"82408af5ea5bee0a0e9b7c20509a045500f2f8d7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-16 23:05:52.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bbf95ca866e847ae895c59d6e7e79ed8 : SUCCESS in 1m 04s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/0eb6a1de102b40ba9056baaa9443f664 : SUCCESS in 50s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/61078dc4bfa7474987bca57ba32d60bf : SUCCESS in 47s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"c208e0a893e57cc9ba80e0f707854287b6a496ee","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-08-17 08:08:57.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"2992f14176ca31918a02cbf5f52fe2a2852750db","revisions":{"fb6a1acfc249ee56ae4a37cfdb1af4ffd5ebfc42":{"kind":"REWORK","_number":1,"created":"2021-08-16 10:33:01.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/95/804695/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/95/804695/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/95/804695/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/95/804695/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/95/804695/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/95/804695/1"}}},"commit":{"parents":[{"commit":"82e1563eec98e94d273a31ed20c71b71eaf5abf7","subject":"tmp","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/82e1563eec98e94d273a31ed20c71b71eaf5abf7"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-08-16 09:53:26.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-08-16 09:53:26.000000000","tz":60},"subject":"container GET: return 503 if policy index mismatches","message":"container GET: return 503 if policy index mismatches\n\nReturn a 503 to the original container listing request if a GET from a\nshard returns objects from a different policy than requested (e.g. due\nto the shard container server not being upgraded).\n\nChange-Id: Ieab6238030e8c264ee90186012be6e9da937b42e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fb6a1acfc249ee56ae4a37cfdb1af4ffd5ebfc42"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fb6a1acfc249ee56ae4a37cfdb1af4ffd5ebfc42"}]},"branch":"refs/heads/master"},"20459d9877bec42982895fcf403ac22a2731a1ac":{"kind":"NO_CHANGE","_number":2,"created":"2021-08-16 10:35:00.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/95/804695/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/95/804695/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/95/804695/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/95/804695/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/95/804695/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/95/804695/2"}}},"commit":{"parents":[{"commit":"89945a6ae9929eeae91946f212751e6482e52565","subject":"container-server: return objects of a given policy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/89945a6ae9929eeae91946f212751e6482e52565"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-08-16 09:53:26.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-08-16 10:34:52.000000000","tz":60},"subject":"container GET: return 503 if policy index mismatches","message":"container GET: return 503 if policy index mismatches\n\nReturn a 503 to the original container listing request if a GET from a\nshard returns objects from a different policy than requested (e.g. due\nto the shard container server not being upgraded).\n\nChange-Id: Ieab6238030e8c264ee90186012be6e9da937b42e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/20459d9877bec42982895fcf403ac22a2731a1ac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/20459d9877bec42982895fcf403ac22a2731a1ac"}]},"branch":"refs/heads/master"},"777288d224c31c6f4f2e219006818cd7bfe20e02":{"kind":"REWORK","_number":3,"created":"2021-08-16 10:57:06.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/95/804695/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/95/804695/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/95/804695/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/95/804695/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/95/804695/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/95/804695/3"}}},"commit":{"parents":[{"commit":"56510ab3c31070a1b8720a33d26c8a220e6ce5c9","subject":"container-server: return objects of a given policy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/56510ab3c31070a1b8720a33d26c8a220e6ce5c9"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-08-16 09:53:26.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-08-16 10:56:54.000000000","tz":60},"subject":"container GET: return 503 if policy index mismatches","message":"container GET: return 503 if policy index mismatches\n\nReturn a 503 to the original container listing request if a GET from a\nshard returns objects from a different policy than requested (e.g. due\nto the shard container server not being upgraded).\n\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nChange-Id: Ieab6238030e8c264ee90186012be6e9da937b42e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/777288d224c31c6f4f2e219006818cd7bfe20e02"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/777288d224c31c6f4f2e219006818cd7bfe20e02"}]},"branch":"refs/heads/master"},"2992f14176ca31918a02cbf5f52fe2a2852750db":{"kind":"REWORK","_number":4,"created":"2021-08-16 12:50:54.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/95/804695/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/95/804695/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/95/804695/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/95/804695/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/95/804695/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/95/804695/4"}}},"commit":{"parents":[{"commit":"56510ab3c31070a1b8720a33d26c8a220e6ce5c9","subject":"container-server: return objects of a given policy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/56510ab3c31070a1b8720a33d26c8a220e6ce5c9"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-08-16 09:53:26.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-08-16 12:50:21.000000000","tz":60},"subject":"container GET: return 503 if policy index mismatches","message":"container GET: return 503 if policy index mismatches\n\nReturn a 503 to the original container listing request if a GET from a\nshard returns objects from a different policy than requested (e.g. due\nto the shard container server not being upgraded).\n\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nChange-Id: Ieab6238030e8c264ee90186012be6e9da937b42e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2992f14176ca31918a02cbf5f52fe2a2852750db"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2992f14176ca31918a02cbf5f52fe2a2852750db"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
