)]}'
{"id":"openstack%2Fswift~805388","triplet_id":"openstack%2Fswift~master~Id67707b2fe69b4a8284c4c8adbe337727f13ba33","project":"openstack/swift","branch":"master","topic":"container-migrate","hashtags":[],"change_id":"Id67707b2fe69b4a8284c4c8adbe337727f13ba33","subject":"sharder _update_storage_policy_index: coerce policy index to int","status":"ABANDONED","created":"2021-08-20 14:09:36.000000000","updated":"2021-08-23 07:31:51.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"afe1b18f3d8856b14869e7481727973c29ef86d0","_number":805388,"virtual_id_number":805388,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-08-20 15:20:09.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-20 15:20:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7ee57651e5fd13613f9ce9bd89f01d74d005a1a6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-08-20 14:09:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9f0a683ea960eb9eb561e1b1ab9eb24e947d6194","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-20 15:20:09.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e085f6b3e9744ba9ab25fd00ad3da14e : SUCCESS in 10m 16s\n- grenade https://zuul.opendev.org/t/openstack/build/a38a9c3db0544b52a9084c6c86c556f1 : SUCCESS in 1h 03m 14s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/174382e599af44f0868500a4daf1773f : SUCCESS in 39m 45s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/70db1a7c71c64c9b8f93738ffabf0568 : FAILURE in 9m 23s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/ba5706304a11449da4df9ee20c42bae7 : FAILURE in 8m 58s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/ed72b79ab6ad42939d0b362247527dd7 : SUCCESS in 17m 24s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/227f0f1bae9d4030a9009f5800294402 : FAILURE in 23m 52s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/133ae3217c694135a3f954b51ae2ec33 : FAILURE in 19m 24s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/628ae0c7e8144cbfb08a4d4106f398bd : FAILURE in 19m 16s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/3139f0a7f47e46849ce10f995718f584 : FAILURE in 28m 37s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/749cde32d4234bba89ca712ea518ec97 : SUCCESS in 18m 49s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/f11853e996b04ea48a437163e6f46ab4 : SUCCESS in 18m 27s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/d1f4ea7f65444a8b83a4dbd5877ec14a : SUCCESS in 20m 31s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/19eaf4815086411f93f4666f2e625392 : SUCCESS in 18m 04s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/23fa143424c546f59daa0bb402e71c67 : SUCCESS in 20m 53s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/3a48907d3e8e4037b39e5272f1e5d790 : SUCCESS in 19m 51s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/003936ec35e144bda31cba365f6cda78 : SUCCESS in 10m 12s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f675038c0e104f7598a9abe1207d1792 : SUCCESS in 21m 06s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/ed66608f8d4b4e01948d9ddd607edfa5 : SUCCESS in 54m 03s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/190276c5f9c040e8914bbeba2f18c121 : SUCCESS in 52m 17s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/66dc45b3a4544d898b8fe1222e0cbdfb : SUCCESS in 42m 02s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ec688f0b8bc446f7987bc5c6d680126f : SUCCESS in 44m 02s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4f0fd3ab02e74d9e97bfc9f846bab31b : SUCCESS in 16m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/35a18b5293454e11acc6575813405678 : SUCCESS in 7m 21s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ddc946d113de45cfa77ad739fccef011 : SUCCESS in 25m 17s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fdc9d2a7eac54c64a626a367dcfe9496 : SUCCESS in 47m 43s","accounts_in_message":[],"_revision_number":1},{"id":"81d697f0346028d3783a70328997fbd2e61074ad","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-20 16:06:17.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9a7c961b8a2c475da88122f33e2c0433 : FAILURE in 44m 37s (non-voting)\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/fef86f5880d6431c8112aa0747585288 : FAILURE in 49m 17s (non-voting)\n- swift-probetests-centos-8-arm64 https://zuul.opendev.org/t/openstack/build/2e18ce3fd2454d0fae46b4f642e22c1f : SUCCESS in 1h 50m 55s (non-voting)\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/7016dc47a2754bbd942f376fa8ce4861 : SUCCESS in 55m 51s (non-voting)\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/bda8fd7621054477baff97f70ef2447b : SUCCESS in 46m 14s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"afe1b18f3d8856b14869e7481727973c29ef86d0","tag":"autogenerated:gerrit:abandon","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-08-23 07:31:51.000000000","message":"Abandoned\n\nsquashed to parent","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"45fcb06c161f029656c73809a1c60024a4c24f49","revisions":{"45fcb06c161f029656c73809a1c60024a4c24f49":{"kind":"REWORK","_number":1,"created":"2021-08-20 14:09:36.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/88/805388/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/88/805388/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/88/805388/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/88/805388/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/88/805388/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/88/805388/1"}}},"commit":{"parents":[{"commit":"7a116bf8c9c2f12300c308e11cc8a43a8a1baab2","subject":"migrate_to_storage_policy: reset _storage_policy_index after migrating","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7a116bf8c9c2f12300c308e11cc8a43a8a1baab2"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-08-20 12:43:22.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-08-20 13:49:48.000000000","tz":60},"subject":"sharder _update_storage_policy_index: coerce policy index to int","message":"sharder _update_storage_policy_index: coerce policy index to int\n\nChange-Id: Id67707b2fe69b4a8284c4c8adbe337727f13ba33\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/45fcb06c161f029656c73809a1c60024a4c24f49"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/45fcb06c161f029656c73809a1c60024a4c24f49"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
