)]}'
{"id":"openstack%2Fswift~805389","triplet_id":"openstack%2Fswift~master~I86cfe1ee7c39366d8a0a43bb4f4f47b9840c4e7a","project":"openstack/swift","branch":"master","topic":"fix-update-shard-spi","hashtags":[],"change_id":"I86cfe1ee7c39366d8a0a43bb4f4f47b9840c4e7a","subject":"test demonstrating working shard policy migration","status":"NEW","created":"2021-08-20 14:09:36.000000000","updated":"2021-09-20 08:41:57.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":4,"unresolved_comment_count":3,"has_review_started":true,"meta_rev_id":"ced625c3c9852772feee25a04990162fefd00282","_number":805389,"virtual_id_number":805389,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:check","value":1,"date":"2021-09-20 08:32:54.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":-2,"date":"2021-09-20 08:41:57.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-20 14:56:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-20 18:40:17.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"CC"},{"updated":"2021-08-23 04:41:55.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"REVIEWER"},{"updated":"2021-09-17 15:48:18.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"}],"messages":[{"id":"3d10080c6109811468dfc9e1521b7827af296034","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-08-20 14:09:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cf4563dcf870d55bcc5d2392d5ea5ad5d11fa0a2","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-08-20 14:11:20.000000000","message":"Patch Set 1: Code-Review-2","accounts_in_message":[],"_revision_number":1},{"id":"17f83f354167a10064d9a48b48edcfe0a6ecd6ec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-20 14:56:49.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b316418f9ebe4ca7bcb5c778456a4dc4 : SUCCESS in 10m 34s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e56d35d6386b461eaab46050a64b8e51 : FAILURE in 7m 48s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/3778803285484772a1f911a2d2023a04 : FAILURE in 10m 13s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/7e0c8c9501454864a56198e6cc84d407 : SUCCESS in 19m 49s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/b535781c4b22425dbb58bee8666ca599 : FAILURE in 19m 49s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/d14b28b51c894b5fb363680a81db3fa2 : FAILURE in 19m 01s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/1921fcd6cac64d9e8e31f1d8885a24bb : FAILURE in 20m 31s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/f1103914a91e4f12b31e56cd6095c1b4 : FAILURE in 28m 02s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/ccbb12c006504a83a526c42203ec035f : SUCCESS in 19m 24s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/ce4138a741034a649b25b1c9b1ba66c6 : SUCCESS in 19m 31s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/81dd305bc84541988154fd5bfd46250c : SUCCESS in 24m 39s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/911a2b366bb14ac991a13e3acf681ac8 : SUCCESS in 18m 13s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/be19b902a0ad40f9a6e25e68da2f05fd : SUCCESS in 21m 57s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/60b0e69403f84859ab1085116fc5f4b9 : SUCCESS in 20m 06s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a49cd96bfcc540748aeeae751c8b1a23 : SUCCESS in 39m 50s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ccc9b90523d54092893622e8243b3ffb : SUCCESS in 41m 18s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e2070900c9f34283b92a9b9456ff9a80 : SUCCESS in 21m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9de268b2c3c048fb92e3ef3bcf762a02 : SUCCESS in 6m 56s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b209c4a026ad4f10a340d1a48f4ef0a4 : SUCCESS in 25m 38s","accounts_in_message":[],"_revision_number":1},{"id":"c9586ede8fa019db1456c8f593b3a72872277c3c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-20 16:10:54.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1e33dade1bd945808412b2eb0c774268 : FAILURE in 40m 22s (non-voting)\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7a5e5abf39444d66abe758f2451c768a : FAILURE in 38m 46s (non-voting)\n- swift-probetests-centos-8-arm64 https://zuul.opendev.org/t/openstack/build/c9337d42486c4dadbe6d02c4bca79e03 : FAILURE in 1h 52m 55s (non-voting)\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/29cb06ade88c4eab9566866347d92198 : SUCCESS in 52m 22s (non-voting)\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/ab03e407e19a4ad8bb49af6e2be06225 : SUCCESS in 51m 19s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7fcc9e6e0ed5565a22063b6bc469682c5b5a6fb6","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2021-08-20 18:40:17.000000000","message":"Patch Set 1:\n\n(2 comments)\n\nthis certainly demonstrates to me we should re-use the reconciler stuff and squash everything down:\n\nhttps://review.opendev.org/c/openstack/swift/+/805450","accounts_in_message":[],"_revision_number":1},{"id":"cb73e1159e77646d0ac4adbfc8aec49b11547803","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2021-08-23 04:41:55.000000000","message":"Uploaded patch set 2: New patch set was added with same tree, parent, and commit message as Patch Set 1.","accounts_in_message":[],"_revision_number":2},{"id":"c5e429fab9c73e8a70b231798a4b70f36b11f967","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-23 05:44:27.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a08c693c2a64bb3be0179851374c5c9 : SUCCESS in 8m 39s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/7c470677ff2f49afa29e628ec51a0bef : FAILURE in 6m 38s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/76175acf880341adaeb1244acc3c7681 : FAILURE in 7m 37s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/57ffda3dced5451da0196bfa9e9acf92 : SUCCESS in 16m 13s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/679e446806f34a7b9773cc81bb33d73d : FAILURE in 16m 17s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/5b025dc78ee84345817a9a4a5a94376e : FAILURE in 16m 34s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/3d029253755e40118ba085eeead2ed23 : FAILURE in 15m 48s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/48a926e1f40b4d05a2775a408c94410b : FAILURE in 15m 54s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/c70633f6f93f4963975574264c473f1e : SUCCESS in 17m 28s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/9142464780914dbdbb88ff00165b44d4 : SUCCESS in 16m 38s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/2b8b74cbaf8b4330957ada9f31221406 : SUCCESS in 16m 27s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/6e1ac5ea6819465da52a22d108ef42cc : SUCCESS in 16m 18s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/7569170d497d48e08fac3a4fa41fdf88 : SUCCESS in 16m 37s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/ec04fab5d9a34773babf576f2fa60ff4 : SUCCESS in 17m 29s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/24c06375703442bfab80e924ef606747 : SUCCESS in 37m 30s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e5530dff5eb543a9977007d0e04c5641 : SUCCESS in 40m 07s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e7127e8d3f49477398f5899776fbc066 : SUCCESS in 15m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/256f46d2d2ec4b8b927ab210db8cdc38 : SUCCESS in 5m 42s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/cf2ed90defb84f2c9cab18dac79e5bd4 : SUCCESS in 25m 40s","accounts_in_message":[],"_revision_number":2},{"id":"27161e7ab7014108d8cbd72214ae1cf0159956a0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-23 06:21:11.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5df7d2030c27454e8fdd6059d0f69bc2 : FAILURE in 33m 12s (non-voting)\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/746ff9c7e3ff4061a3fb51fa69c59176 : FAILURE in 40m 27s (non-voting)\n- swift-probetests-centos-8-arm64 https://zuul.opendev.org/t/openstack/build/313f95f7ea9a4dc0b074141325e73e62 : SUCCESS in 1h 31m 15s (non-voting)\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/e0218d3308fe426281fbc21f337d2a8b : SUCCESS in 49m 18s (non-voting)\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/313327e861134a909d3260a91ee99250 : SUCCESS in 34m 56s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6e179ed312c726ad8d54a9825b2ccf9aef1b825c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-08-23 09:24:05.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"35ef36596843836fd7428323d695df5c8ff6a8fe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-23 10:08:58.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ac38c8010a54494aa36ada206009595 : SUCCESS in 9m 39s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/8c77f75362de467da1df26b3ffd659e0 : FAILURE in 7m 24s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/1f8a9770e606451f96d6b7c5f84e5788 : FAILURE in 8m 06s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/ea28007c704e4bc8bcdc4be7577eacf3 : SUCCESS in 20m 55s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/5e0f2ff62be34af4a408df9e6975b0a5 : SUCCESS in 21m 56s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/e6ffc053999b4470a31898edf44c0b4e : SUCCESS in 18m 40s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/e606cab9ace14847b0a400bb006a7825 : SUCCESS in 21m 10s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/19f8caf53941496aa3e05b17685098bc : SUCCESS in 16m 38s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/a05173af4caa454ba273697b95baa3d0 : SUCCESS in 22m 12s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/c543cb0182184099bd1e3b645664b553 : SUCCESS in 21m 29s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/9c2ceb271c8943369a8069a20ab4423e : SUCCESS in 22m 32s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/32b2a962665749deab110b526f64a16f : SUCCESS in 18m 13s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/3799242d416e4a389f1346365c095a76 : SUCCESS in 18m 21s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/3bf23709e5f74282a4e12588d0b9a94a : SUCCESS in 17m 26s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f23156ab2edd449293a54e9344615e62 : SUCCESS in 43m 21s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4fb4b2979ad446a4a3af990d83bcdf5a : SUCCESS in 38m 43s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/26800ee6814449fba936b1a463419e4f : SUCCESS in 20m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c214b2c7b7b848a5873d4102669c643e : SUCCESS in 6m 15s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a3b7f27a67a34a99a1b1344de28744a0 : SUCCESS in 30m 07s","accounts_in_message":[],"_revision_number":3},{"id":"b593fc650db99d123299d18fa6109e54c3a0dfb3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-23 11:28:23.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ded3dbe601a247dea2188dec219d5609 : SUCCESS in 45m 19s (non-voting)\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8ee3b16db9244eb1b4fb6aaf1a2ff6e7 : SUCCESS in 55m 08s (non-voting)\n- swift-probetests-centos-8-arm64 https://zuul.opendev.org/t/openstack/build/d5511d81548c4f1881ce2a8799e7d9af : TIMED_OUT in 2h 01m 34s (non-voting)\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/641fbb0f7fd4498aa754ffd53525d521 : SUCCESS in 53m 36s (non-voting)\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/62c4a3e1009f4d618eb7da322deca58d : SUCCESS in 55m 42s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"1dc130a0f355b3af7984df9cf261e12e6a0c9339","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2021-08-26 03:04:37.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"26631abf19c91dbb1d181058fb397d089e8a0d84","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-08-26 08:56:39.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"89a2093d48a2262b611149bbde3464165b8bd10a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2021-09-17 15:48:18.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"63dce7dc5ed43d7f3000a67fec13a82dcefc6b50","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-17 16:48:41.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/edc94c04b6844004a6696591a34b94d6 : SUCCESS in 10m 38s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/27fb0328fed748108219c09402ecaa24 : SUCCESS in 10m 04s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/dcf409810ac140faa1204a86bcead295 : SUCCESS in 10m 37s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/beb1b0b707924ea1ae8de2abf91055b8 : SUCCESS in 21m 59s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/66b93df8d0fb405e8f0428af12fe1f70 : SUCCESS in 20m 40s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/875bb6ad27db4c4b9cd14329cdd62aed : SUCCESS in 18m 49s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/7f3b9d2b86f44125ba3d360ca33191f6 : SUCCESS in 21m 09s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/6253605c34174c96adb8c6aa5e677b05 : SUCCESS in 17m 43s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/8b638d24bd8d4e4399477be9b19a34ad : SUCCESS in 20m 32s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/4d65a2cf24d5490d9a574db4aca2e0b2 : SUCCESS in 17m 48s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/37e40ffa33434596a4d6790e0bca1807 : SUCCESS in 23m 28s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/16a271b74b0446c3a4e4ff6f1e1703e9 : SUCCESS in 19m 19s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/69ea67b77cf243edb86ec1cd4cdb7fa0 : SUCCESS in 17m 15s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/25c95bb57f4746c4a77a751c3c6e4bd8 : SUCCESS in 18m 48s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c4d7965d97d0474ebbd16a940e5d8c73 : SUCCESS in 43m 21s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4bdd7145a50147c79e1e758aedcf4ae3 : SUCCESS in 44m 49s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5d5a707c5b014858853a2f630c0fc6c9 : SUCCESS in 16m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/df8c1916a3c64a189e4de53061b8b79f : SUCCESS in 6m 52s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/46f4a341b8e94f03acceba4798d17a35 : SUCCESS in 28m 35s","accounts_in_message":[],"_revision_number":4},{"id":"6f2ccf47d66b59ee16d787c80b29701495f89939","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-17 17:43:47.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e6bf4ce751dd4d63a367b2a04f13dc18 : SUCCESS in 38m 49s (non-voting)\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/947f4b1abf234f62be7698cdf9a82435 : SUCCESS in 56m 00s (non-voting)\n- swift-probetests-centos-8-arm64 https://zuul.opendev.org/t/openstack/build/e8e094ec63ee4d12b9cc2c4322f7feb9 : SUCCESS in 1h 41m 48s (non-voting)\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/28a15ba0c42540158a7e36406acaa3b2 : SUCCESS in 41m 43s (non-voting)\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/a0ed9616899e4b969256aa59fee6a9a3 : SUCCESS in 39m 38s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"6e04608ab18e059ed134e2ce0dc7a86d2104a26c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2021-09-20 06:37:37.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"74ab9ea9bf03525ee2039aa0ac26cb2d2918a1db","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-20 08:32:54.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/33c937a387dc4aa1b6ae8396baa30f79 : SUCCESS in 10m 54s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b6cc3fab1f4b4da086af60b34e362857 : SUCCESS in 11m 26s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/4980c4af428641e4afb5d6d8dda2255d : SUCCESS in 11m 34s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/3c4a666f59384833ad5d42aa15fd14ef : SUCCESS in 22m 58s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/225dc6bc0a0c4cc192afd4f7ea50a881 : SUCCESS in 21m 11s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/0a51ba843c35492485666938bbb5151c : SUCCESS in 23m 37s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/188bfe6e5cd4434c979e735de9b5eb5e : SUCCESS in 21m 42s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/aaa6bdc2fb7c4a5b83d7fca6a3c6baea : SUCCESS in 21m 24s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/4b313ad5a992445bbd6fe217ccc7d8b3 : SUCCESS in 20m 06s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/76b5a4aa3fad41eeae20f7bae841b544 : SUCCESS in 18m 17s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/62c4a9b5bab04a0087488eee31324814 : SUCCESS in 19m 57s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/b0109d30badb441ea3280d36d814d2af : SUCCESS in 17m 47s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/8de67de6ae424acb87b996f0b560ac72 : SUCCESS in 18m 13s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/287d50fa97554ee1a80453830327761d : SUCCESS in 21m 08s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/58378c74329f4c55a946d4d18acd9668 : SUCCESS in 39m 22s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/132fd8b8cb204b91b5ff1afa9e4b3675 : SUCCESS in 38m 15s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0f721a6990e9435f8dffa6e82a47d6d7 : SUCCESS in 14m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b30f35b532b4c80b15997b984495ea6 : SUCCESS in 7m 30s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c7af72741dbd4e7988cb5df61b0e0eb3 : SUCCESS in 27m 21s","accounts_in_message":[],"_revision_number":5},{"id":"105922d0b2809df1f6cadebb24e785961d03c1a6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-20 08:41:57.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1d47e6296439437595a6cf9d25a7ac05 : SUCCESS in 57m 36s (non-voting)\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f83bdd596fc4431188cccfa272621f43 : SUCCESS in 40m 03s (non-voting)\n- swift-probetests-centos-8-arm64 https://zuul.opendev.org/t/openstack/build/47bbf05ec9a54da3a8c4103fd37229cc : TIMED_OUT in 2h 01m 29s (non-voting)\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/cbd5f93f47c2429ca3884837e42588c6 : SUCCESS in 55m 18s (non-voting)\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/d0fa89ff873d49fdbe8c552519f9bcbf : SUCCESS in 50m 54s (non-voting)","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"99c2753e1c9b9eaf96870e34896ce0da7fbeb1fe","revisions":{"0d9e43f1693536346a5eeb409246fde3e4ce4090":{"kind":"REWORK","_number":1,"created":"2021-08-20 14:09:36.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/89/805389/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/89/805389/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/805389/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/805389/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/805389/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/89/805389/1"}}},"commit":{"parents":[{"commit":"45fcb06c161f029656c73809a1c60024a4c24f49","subject":"sharder _update_storage_policy_index: coerce policy index to int","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/45fcb06c161f029656c73809a1c60024a4c24f49"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-08-20 12:44:24.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-08-20 14:05:42.000000000","tz":60},"subject":"DNM: test demonstrating broken shard policy migration","message":"DNM: test demonstrating broken shard policy migration\n\nThe shard policy migration is using different criteria than the root\nfor deciding when to migrate to a new policy. This can result in\npermanent policy mismatch between the root and shard.\n\nChange-Id: I86cfe1ee7c39366d8a0a43bb4f4f47b9840c4e7a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0d9e43f1693536346a5eeb409246fde3e4ce4090"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0d9e43f1693536346a5eeb409246fde3e4ce4090"}]},"branch":"refs/heads/master"},"cb80eff29128c902c07931d1cd766397681a081a":{"kind":"NO_CHANGE","_number":2,"created":"2021-08-23 04:41:55.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/89/805389/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/89/805389/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/805389/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/805389/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/805389/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/89/805389/2"}}},"commit":{"parents":[{"commit":"3e1c6746c916cc6afdbb86f95fc58982cd5b14fb","subject":"sharder: update shard storage_policy_index if roots changes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3e1c6746c916cc6afdbb86f95fc58982cd5b14fb"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-08-20 12:44:24.000000000","tz":60},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2021-08-23 04:41:16.000000000","tz":600},"subject":"DNM: test demonstrating broken shard policy migration","message":"DNM: test demonstrating broken shard policy migration\n\nThe shard policy migration is using different criteria than the root\nfor deciding when to migrate to a new policy. This can result in\npermanent policy mismatch between the root and shard.\n\nChange-Id: I86cfe1ee7c39366d8a0a43bb4f4f47b9840c4e7a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cb80eff29128c902c07931d1cd766397681a081a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cb80eff29128c902c07931d1cd766397681a081a"}]},"branch":"refs/heads/master"},"0084c753a4410b77d4ab7ed3316a90771f571def":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-08-23 09:24:05.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/89/805389/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/89/805389/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/805389/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/805389/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/805389/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/89/805389/3"}}},"commit":{"parents":[{"commit":"e743ecc81a2160df5817547bb20b5a67cf03c565","subject":"sharder: update shard storage_policy_index if roots changes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e743ecc81a2160df5817547bb20b5a67cf03c565"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-08-20 12:44:24.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-08-23 09:23:25.000000000","tz":60},"subject":"DNM: test demonstrating broken shard policy migration","message":"DNM: test demonstrating broken shard policy migration\n\nThe shard policy migration is using different criteria than the root\nfor deciding when to migrate to a new policy. This can result in\npermanent policy mismatch between the root and shard.\n\nChange-Id: I86cfe1ee7c39366d8a0a43bb4f4f47b9840c4e7a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0084c753a4410b77d4ab7ed3316a90771f571def"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0084c753a4410b77d4ab7ed3316a90771f571def"}]},"branch":"refs/heads/master"},"c62ffd98e4d894baaf2a57ce7be63f022e8c1275":{"kind":"REWORK","_number":4,"created":"2021-09-17 15:48:18.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/89/805389/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/89/805389/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/805389/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/805389/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/805389/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/89/805389/4"}}},"commit":{"parents":[{"commit":"b1c3b8976884d22aa18f98956a4f8b6a9d6bb48f","subject":"sharder: update shard storage_policy_index if roots changes - Simple","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b1c3b8976884d22aa18f98956a4f8b6a9d6bb48f"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-08-20 12:44:24.000000000","tz":60},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2021-09-17 15:46:09.000000000","tz":-300},"subject":"test demonstrating working shard policy migration","message":"test demonstrating working shard policy migration\n\nThe shard policy migration is using the same criteria as the root for\ndeciding when to migrate to a new policy. This can only result in\npermanent policy agreement between the root and shard.\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nRelated-Change-Id: I86cfe1ee7c39366d8a0a43bb4f4f47b9840c4e7a\nRelated-Change-Id: Id6541d0deabb9f7c3cfd7523833b77218448c990\nChange-Id: I86cfe1ee7c39366d8a0a43bb4f4f47b9840c4e7a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c62ffd98e4d894baaf2a57ce7be63f022e8c1275"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c62ffd98e4d894baaf2a57ce7be63f022e8c1275"}]},"branch":"refs/heads/master"},"99c2753e1c9b9eaf96870e34896ce0da7fbeb1fe":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2021-09-20 06:37:37.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/89/805389/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/89/805389/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/805389/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/805389/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/805389/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/89/805389/5"}}},"commit":{"parents":[{"commit":"a6a92b66232254b923ab0fb6563a74bc01bcad8d","subject":"sharder: update shard SPI when roots changes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a6a92b66232254b923ab0fb6563a74bc01bcad8d"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-08-20 12:44:24.000000000","tz":60},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2021-09-20 06:37:37.000000000","tz":0},"subject":"test demonstrating working shard policy migration","message":"test demonstrating working shard policy migration\n\nThe shard policy migration is using the same criteria as the root for\ndeciding when to migrate to a new policy. This can only result in\npermanent policy agreement between the root and shard.\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nRelated-Change-Id: I86cfe1ee7c39366d8a0a43bb4f4f47b9840c4e7a\nRelated-Change-Id: Id6541d0deabb9f7c3cfd7523833b77218448c990\nChange-Id: I86cfe1ee7c39366d8a0a43bb4f4f47b9840c4e7a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/99c2753e1c9b9eaf96870e34896ce0da7fbeb1fe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/99c2753e1c9b9eaf96870e34896ce0da7fbeb1fe"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Code-Review\u003dMIN"],"failing_atoms":["label:Code-Review\u003dMAX"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
