)]}'
{"id":"openstack%2Fswift~805711","triplet_id":"openstack%2Fswift~master~I3f42fbaae5f246836024f95b549b487b41bd6f1d","project":"openstack/swift","branch":"master","topic":"fix-docker-image-builds","hashtags":[],"change_id":"I3f42fbaae5f246836024f95b549b487b41bd6f1d","subject":"Fix docker image builds","status":"MERGED","created":"2021-08-23 21:22:52.000000000","updated":"2021-08-24 01:07:21.000000000","submitted":"2021-08-24 01:05:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"805711-fix-docker-image-builds","meta_rev_id":"51638fe9a62f563a79ffb263dd3c4512ce83ec5c","_number":805711,"virtual_id_number":805711,"owner":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-24 01:05:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-08-23 22:10:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-08-23 22:10:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-23 22:10:46.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2021-08-23 22:55:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"eabe9019ae9e1d9798b47ea54d3d34826a4d85c7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2021-08-23 21:22:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5125467ce14758baf367c3ce2aed3087760c90a6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2021-08-23 21:56:26.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1d8f4e9a1421410d40bf04b031bd2804b277eb18","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-08-23 22:10:46.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nHash on the issue (6101B2783B2FD161) matches the hash from a recent build (https://storage.bhs.cloud.ovh.net/v1/AUTH_dcaab5e32b234d56b626f72581e3644c/zuul_opendev_logs_8a1/648263/26/check/swift-build-image/8a1ab05/job-output.txt) so at least we\u0027re no *worse off* than before the build broke ;-)\n\nThanks!","accounts_in_message":[],"_revision_number":2},{"id":"d09e818d13052cf6e9051fd83b4b488d427e09ec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-23 22:55:17.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ff55946155447edb3f0fdfed9643310 : SUCCESS in 9m 32s\n- grenade https://zuul.opendev.org/t/openstack/build/7a5dd8285acf40db9222f3dc5f1a0f47 : SUCCESS in 57m 32s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c4c4f56b77d343f69f7dbafb8c9eb60f : SUCCESS in 33m 49s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/704ba698e1fe4620b6a44a761b5bed93 : SUCCESS in 9m 01s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/5613d92a979b4d5ba1827751c415327b : SUCCESS in 8m 31s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/b8cbd72847f840bb8835b6a7b330a17a : SUCCESS in 20m 41s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/34acaac437e3452891ab97f62338ad41 : SUCCESS in 16m 24s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/79d54855f8594bbb82e14c96581d0e3f : SUCCESS in 15m 39s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/6e5db9ba9f134e2d8765f45283c0110a : SUCCESS in 14m 01s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/ba77cf1edfcf4e90995968f48e5d8f9c : SUCCESS in 15m 59s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/23e09394b2a7458c8028f6d7b8514255 : SUCCESS in 21m 27s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/6b41ce58964048d49e0c259ee0b9c59f : SUCCESS in 16m 53s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/39f58592719d4414b05e4f68093d474c : SUCCESS in 17m 43s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/2eba58bb226d44b1a146c907ec97fa2f : SUCCESS in 16m 04s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/7e145f1c9c3a4855a03833ef384df7c7 : SUCCESS in 16m 40s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/3cd0bcc934964fcfaf1f0d8abc60984e : SUCCESS in 22m 32s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/da67ce7556e34560a69e03d912ff707d : SUCCESS in 6m 47s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/4fa10001910c4947981f4cb4025c2da3 : SUCCESS in 34m 17s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/8bc9f424f60747419bf20b4ec636aafb : SUCCESS in 45m 41s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/f201735dd5b745f297948ca2c8ffbabe : SUCCESS in 51m 28s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/eec275f55187421d8d28f8d20a2fc730 : SUCCESS in 37m 15s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/0f713e32aa584315a8d8fdec5e229f1f : SUCCESS in 38m 42s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/939d592c17de4f3992e166806a0cc1aa : SUCCESS in 15m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b2208673c1a846e78fbd273ac6d60b7c : SUCCESS in 6m 27s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/306a20fb25774651b6b288d4e24f0ae6 : SUCCESS in 23m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/73f8b655f80443a08316488b573959c2 : SUCCESS in 39m 11s","accounts_in_message":[],"_revision_number":2},{"id":"b19684b90baff126606d4bc791c49cf671d161b8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-23 22:56:01.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"62c977d88714a638f01fb0aca823e55b2dc94080","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-23 23:32:53.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/861e21bc290a4b709fb8e3f840effcf0 : SUCCESS in 34m 17s (non-voting)\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/336c73ba09734f0d8b9b953f77a9d1c3 : SUCCESS in 52m 45s (non-voting)\n- swift-probetests-centos-8-arm64 https://zuul.opendev.org/t/openstack/build/b8f3d66335074dc583f31cb9519d2983 : FAILURE in 1h 29m 51s (non-voting)\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/754d6d629ef64eef881ab0d933546e82 : SUCCESS in 35m 52s (non-voting)\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/e34f65f4730543b28476f933e6a8a7ed : SUCCESS in 51m 48s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"33a54e1d808513786b21427a8560e8d056db507e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-24 01:05:51.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d04f95f2ad34b99a080a3dc12229b66 : SUCCESS in 9m 43s\n- grenade https://zuul.opendev.org/t/openstack/build/fceef481e5c8421dbd8df51fac3162ba : SUCCESS in 1h 02m 26s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/5b9666bd44034c0cb11bc00719c2bc41 : SUCCESS in 36m 22s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/ca6f9a879ce44de594688a90e5781692 : SUCCESS in 11m 14s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/f7abc4a2a1954dfa9a8db0f152827264 : SUCCESS in 9m 53s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/f008c48dd9e7497d994370432e3284ef : SUCCESS in 15m 59s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/0cf9a4fc05a74c2ab6fd91220b4d26da : SUCCESS in 16m 51s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/2bb51d23b86045b08ec3c56dcabed370 : SUCCESS in 21m 41s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/533a6ab99cf8489da8c75acf117ca067 : SUCCESS in 16m 46s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/e0fbf2c39eb548e5afae24056e70cd7d : SUCCESS in 15m 17s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/7cea5ed9602341a3842cadbc73149be6 : SUCCESS in 19m 28s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/383ce42775ca4ef286a0bfb66087c514 : SUCCESS in 17m 04s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/9cac4077cd20402ba0b1c6c152540594 : SUCCESS in 18m 02s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/d4158060319d4bf8ad47a2ea63b8c322 : SUCCESS in 16m 39s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/b9cce3562861408b9a1b273ada0fcaea : SUCCESS in 17m 16s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/646294774a8b411fadd19363a571f029 : SUCCESS in 23m 01s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/89c5da90834b4cfa9482fa8ef058b626 : SUCCESS in 6m 54s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/ab33e07238ed48338a80c288be2a5fd2 : SUCCESS in 47m 13s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/57ada868f60445cd8f27d94be3004b47 : SUCCESS in 49m 37s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/755b70ec19d344019f3dd5cea0d373d7 : SUCCESS in 39m 15s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4ce0da5757464d6e9e5a74a3916048ab : SUCCESS in 44m 41s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9149d9f9ae6348f6ab4e20fc8aeb5772 : SUCCESS in 16m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b62ce3081f474045b5ba4f352ee4c75c : SUCCESS in 6m 23s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/3900948dde154f15af8f9f0c0064b2c8 : SUCCESS in 25m 58s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2bdb6be362cd43c4a4e82738eca66a5c : SUCCESS in 49m 47s","accounts_in_message":[],"_revision_number":2},{"id":"70e098ef4ec00711fd9c62341e496cf10c031319","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-24 01:05:51.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"51638fe9a62f563a79ffb263dd3c4512ce83ec5c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-24 01:07:21.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/14b7fc2740d8491abdba49e7286c552c : SUCCESS in 1m 09s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/dc5d560c9ab3458891b3dab1548229c7 : SUCCESS in 47s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/2ddb41cdfd2340a29b280c1c5e8e9083 : SUCCESS in 49s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"930ac1114b9d2c91c2d6c277202a80b729ff9903","revisions":{"999d7d06b12b44d6634a35b01a5a51fe4bb0acdb":{"kind":"REWORK","_number":1,"created":"2021-08-23 21:22:52.000000000","uploader":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"ref":"refs/changes/11/805711/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/11/805711/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/805711/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/805711/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/805711/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/11/805711/1"}}},"commit":{"parents":[{"commit":"f190ba2b7d02693ee2d7b59043b0baf9fddd203b","subject":"Merge \"use a context manager to apply --no-auto-shard option\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f190ba2b7d02693ee2d7b59043b0baf9fddd203b"}]}],"author":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2021-08-23 21:20:51.000000000","tz":-420},"committer":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2021-08-23 21:22:47.000000000","tz":-420},"subject":"Fix docker image builds","message":"Fix docker image builds\n\nThe image builds are trying to fetch a gpg pubkey from keybase which\ndoes not exist. This causes the builds to fail. Fix this by vendoring\nthe key in the swift repo directly. This should be more reliable.\n\nUpstream bug at https://github.com/just-containers/s6-overlay/issues/352\n\nNote I looked up the pubkey given the hash on that issue and copied its\ncontents into this change. I can\u0027t vouch for the validity of this key\nbeyond that. However, if it is listed in that issue and validates the\npackages from s6 then it is probably good.\n\nChange-Id: I3f42fbaae5f246836024f95b549b487b41bd6f1d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/999d7d06b12b44d6634a35b01a5a51fe4bb0acdb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/999d7d06b12b44d6634a35b01a5a51fe4bb0acdb"}]},"branch":"refs/heads/master"},"930ac1114b9d2c91c2d6c277202a80b729ff9903":{"kind":"REWORK","_number":2,"created":"2021-08-23 21:56:26.000000000","uploader":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"ref":"refs/changes/11/805711/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/11/805711/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/805711/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/805711/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/805711/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/11/805711/2"}}},"commit":{"parents":[{"commit":"f190ba2b7d02693ee2d7b59043b0baf9fddd203b","subject":"Merge \"use a context manager to apply --no-auto-shard option\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f190ba2b7d02693ee2d7b59043b0baf9fddd203b"}]}],"author":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2021-08-23 21:20:51.000000000","tz":-420},"committer":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2021-08-23 21:56:17.000000000","tz":-420},"subject":"Fix docker image builds","message":"Fix docker image builds\n\nThe image builds are trying to fetch a gpg pubkey from keybase which\ndoes not exist. This causes the builds to fail. Fix this by vendoring\nthe key in the swift repo directly. This should be more reliable.\n\nUpstream bug at https://github.com/just-containers/s6-overlay/issues/352\n\nNote I looked up the pubkey given the hash on that issue and copied its\ncontents into this change. I can\u0027t vouch for the validity of this key\nbeyond that. However, if it is listed in that issue and validates the\npackages from s6 then it is probably good.\n\nChange-Id: I3f42fbaae5f246836024f95b549b487b41bd6f1d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/930ac1114b9d2c91c2d6c277202a80b729ff9903"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/930ac1114b9d2c91c2d6c277202a80b729ff9903"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
