)]}'
{"id":"openstack%2Fswift~805808","triplet_id":"openstack%2Fswift~master~Iab53c574cc226e9505f28d3443fd7972b90a463a","project":"openstack/swift","branch":"master","topic":"bug/1704200","hashtags":[],"change_id":"Iab53c574cc226e9505f28d3443fd7972b90a463a","subject":"Add documentation for DELETE method for Swift Object Store API.","status":"MERGED","created":"2021-08-24 11:00:22.000000000","updated":"2021-08-31 20:49:24.000000000","submitted":"2021-08-31 20:47:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":10,"unresolved_comment_count":7,"has_review_started":true,"submission_id":"805808-bug/1704200","meta_rev_id":"e787045b23070d984d0653c9e06d663c32b6b245","_number":805808,"virtual_id_number":805808,"owner":{"_account_id":33856,"name":"Nicholas Njihia","display_name":"nicknjihian","email":"nicholas.njihia@canonical.com","username":"nicknjihian"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-31 20:47:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-08-31 18:42:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-08-31 18:42:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-24 11:11:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-24 16:30:26.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"CC"},{"updated":"2021-08-31 18:42:01.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"0ebd6b0e384820bfbdb3d80efab13bf66870ae98","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33856,"name":"Nicholas Njihia","display_name":"nicknjihian","email":"nicholas.njihia@canonical.com","username":"nicknjihian"},"date":"2021-08-24 11:00:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1f541b598e1c44ed04a80645f052cbc4e7de0ec6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-24 11:11:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/60f6e97b2b47448dbbfc5dbcd6d6798f : SUCCESS in 5m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be245eb4f7dd4aefb934a3d811a2dedc : SUCCESS in 10m 23s","accounts_in_message":[],"_revision_number":1},{"id":"55fdb88517d26a47a1b90be7cf88353ad1e50d3c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-24 12:38:54.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-probetests-centos-8-arm64 https://zuul.opendev.org/t/openstack/build/338f3ffb1efe45eaa84bdf5c5626f691 : SUCCESS in 1h 33m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"64d4e4d7aa07e05bd469f9b9e33c2de45a7f9420","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-08-24 16:30:26.000000000","message":"Patch Set 1:\n\n(7 comments)\n\nThanks for picking this up!","accounts_in_message":[],"_revision_number":1},{"id":"c183d3b517dc4449c184e5adf2967980368fef77","author":{"_account_id":33856,"name":"Nicholas Njihia","display_name":"nicknjihian","email":"nicholas.njihia@canonical.com","username":"nicknjihian"},"date":"2021-08-27 12:19:33.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"5a3743b92a976086ae03cdc2e3803a3b77fa4578","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33856,"name":"Nicholas Njihia","display_name":"nicknjihian","email":"nicholas.njihia@canonical.com","username":"nicknjihian"},"date":"2021-08-27 12:20:21.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"7a6125d61f97d38c4728bd394734f1c045cbb67b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33856,"name":"Nicholas Njihia","display_name":"nicknjihian","email":"nicholas.njihia@canonical.com","username":"nicknjihian"},"date":"2021-08-27 12:23:45.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"5564bee6733580a61ae0b2577dacb3d0bb59b464","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-27 12:36:54.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/e5bfe757d2934588b2f03da0603bed8d : SUCCESS in 5m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/384862d8473c4865b73aa1a5bc1905ae : SUCCESS in 9m 06s","accounts_in_message":[],"_revision_number":3},{"id":"3d0757cc33b9ff5bfaddfc2aca808755651e3478","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-27 14:27:46.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-probetests-centos-8-arm64 https://zuul.opendev.org/t/openstack/build/47f9e5368ae747148450d8c86185010f : TIMED_OUT in 2h 01m 48s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"41f22598eedbb4eeadf6aece40328f8c1bf60fe8","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-08-31 18:42:01.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\nLooks great, thanks!","accounts_in_message":[],"_revision_number":3},{"id":"e327d1352d7da2293820fb82dcb7a241a609d346","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-31 18:42:55.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"c14f829d55550aa5375b41bc04a0643f5fd07415","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-31 20:47:23.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/522d1863446540a6ada912969ee3a042 : SUCCESS in 6m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/426d4ce47a7a4fda85f69a47d6cd43ad : SUCCESS in 11m 10s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/83007c361a204b18993be79af1bca0f3 : SUCCESS in 10m 19s","accounts_in_message":[],"_revision_number":3},{"id":"4a080cdd610c139d9135ed50868796f2b0fb9253","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-31 20:47:23.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"e787045b23070d984d0653c9e06d663c32b6b245","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-31 20:49:24.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/6566587f381145eea6466f7179e829a0 : SUCCESS in 1m 02s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b93d8f48823946f2ab7ac7415c4c6c56 : SUCCESS in 1m 16s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/87326a1f94e247c181c73f9f96285e7c : FAILURE in 46s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/f73f779496654b56b858ac5ee5476063 : FAILURE in 1m 00s (non-voting)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"dfd1bdc9e2eae0a31cc5abbbf8e2584e1545598a","revisions":{"065c4ee4811d64ba1933e5fbbe3fb9369b99a3d0":{"kind":"REWORK","_number":1,"created":"2021-08-24 11:00:22.000000000","uploader":{"_account_id":33856,"name":"Nicholas Njihia","display_name":"nicknjihian","email":"nicholas.njihia@canonical.com","username":"nicknjihian"},"ref":"refs/changes/08/805808/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/08/805808/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/08/805808/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/08/805808/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/08/805808/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/08/805808/1"}}},"commit":{"parents":[{"commit":"f190ba2b7d02693ee2d7b59043b0baf9fddd203b","subject":"Merge \"use a context manager to apply --no-auto-shard option\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f190ba2b7d02693ee2d7b59043b0baf9fddd203b"}]}],"author":{"name":"Nicholas Njihia","email":"nicholas.njihia@canonical.com","date":"2021-08-24 09:55:35.000000000","tz":180},"committer":{"name":"Nicholas Njihia","email":"nicholas.njihia@canonical.com","date":"2021-08-24 09:55:35.000000000","tz":180},"subject":" Add documentation for DELETE method for Swift Object Store API.","message":" Add documentation for DELETE method for Swift Object Store API.\n\n   Account API does not document \u0027DELETE\u0027 verb, which is a valid request to\ndelete an account.\n   This is a doc addition request.\n\n Closes-Bug: 1704200\n\nChange-Id: Iab53c574cc226e9505f28d3443fd7972b90a463a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/065c4ee4811d64ba1933e5fbbe3fb9369b99a3d0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/065c4ee4811d64ba1933e5fbbe3fb9369b99a3d0"}]},"branch":"refs/heads/master"},"252d35e7bef1ef879a6c42cd89044ff0daf40258":{"kind":"REWORK","_number":2,"created":"2021-08-27 12:20:21.000000000","uploader":{"_account_id":33856,"name":"Nicholas Njihia","display_name":"nicknjihian","email":"nicholas.njihia@canonical.com","username":"nicknjihian"},"ref":"refs/changes/08/805808/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/08/805808/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/08/805808/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/08/805808/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/08/805808/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/08/805808/2"}}},"commit":{"parents":[{"commit":"f190ba2b7d02693ee2d7b59043b0baf9fddd203b","subject":"Merge \"use a context manager to apply --no-auto-shard option\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f190ba2b7d02693ee2d7b59043b0baf9fddd203b"}]}],"author":{"name":"Nicholas Njihia","email":"nicholas.njihia@canonical.com","date":"2021-08-24 09:55:35.000000000","tz":180},"committer":{"name":"Nicholas Njihia","email":"nicholas.njihia@canonical.com","date":"2021-08-27 12:15:43.000000000","tz":180},"subject":" Add documentation for DELETE method for Swift Object Store API.","message":" Add documentation for DELETE method for Swift Object Store API.\n\n   Account API does not document \u0027DELETE\u0027 verb, which is a valid request to\ndelete an account.\n   This is a doc addition request.\n\n Closes-Bug: 1704200\n\nChange-Id: Iab53c574cc226e9505f28d3443fd7972b90a463a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/252d35e7bef1ef879a6c42cd89044ff0daf40258"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/252d35e7bef1ef879a6c42cd89044ff0daf40258"}]},"branch":"refs/heads/master"},"dfd1bdc9e2eae0a31cc5abbbf8e2584e1545598a":{"kind":"REWORK","_number":3,"created":"2021-08-27 12:23:45.000000000","uploader":{"_account_id":33856,"name":"Nicholas Njihia","display_name":"nicknjihian","email":"nicholas.njihia@canonical.com","username":"nicknjihian"},"ref":"refs/changes/08/805808/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/08/805808/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/08/805808/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/08/805808/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/08/805808/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/08/805808/3"}}},"commit":{"parents":[{"commit":"f190ba2b7d02693ee2d7b59043b0baf9fddd203b","subject":"Merge \"use a context manager to apply --no-auto-shard option\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f190ba2b7d02693ee2d7b59043b0baf9fddd203b"}]}],"author":{"name":"Nicholas Njihia","email":"nicholas.njihia@canonical.com","date":"2021-08-24 09:55:35.000000000","tz":180},"committer":{"name":"Nicholas Njihia","email":"nicholas.njihia@canonical.com","date":"2021-08-27 12:23:08.000000000","tz":180},"subject":" Add documentation for DELETE method for Swift Object Store API.","message":" Add documentation for DELETE method for Swift Object Store API.\n\n   Account API does not document \u0027DELETE\u0027 verb, which is a valid request to\ndelete an account.\n   This is a doc addition request.\n\n Closes-Bug: 1704200\n\nChange-Id: Iab53c574cc226e9505f28d3443fd7972b90a463a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/dfd1bdc9e2eae0a31cc5abbbf8e2584e1545598a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/dfd1bdc9e2eae0a31cc5abbbf8e2584e1545598a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
