)]}'
{"id":"openstack%2Fswift~809114","triplet_id":"openstack%2Fswift~master~I8f2f247dd113ad637b17d241133b14c35cadecae","project":"openstack/swift","branch":"master","hashtags":[],"change_id":"I8f2f247dd113ad637b17d241133b14c35cadecae","subject":"func test improvements","status":"MERGED","created":"2021-09-15 04:51:20.000000000","updated":"2021-09-15 10:01:30.000000000","submitted":"2021-09-15 09:59:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"809114","meta_rev_id":"517f14c3b58a17941a39f9f36d62eef02346c4f3","_number":809114,"virtual_id_number":809114,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-15 09:59:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"all":[{"value":2,"date":"2021-09-15 07:26:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"all":[{"value":1,"date":"2021-09-15 07:26:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-15 04:53:15.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"},{"updated":"2021-09-15 04:53:15.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"REVIEWER"},{"updated":"2021-09-15 04:53:15.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"},{"updated":"2021-09-15 05:31:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f03d6177ad676264106c65ea0129fe1e1113cb55","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-09-15 04:51:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ae94a741ff48a0a58198ff4492a9af4d80259b5d","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-09-15 04:53:15.000000000","message":"Patch Set 1:\n\nStill got something like\n\n Ran 878 tests in 3815.515s\n \n FAILED (SKIP\u003d144, errors\u003d1, failures\u003d34)\n\nrunning against staging, but it\u0027s an improvement for sure. Eyeballing it, looked like a lot of the failures were due to listing inconsistencies.","accounts_in_message":[],"_revision_number":1},{"id":"850633d31820d5da98fc9cbc23b4afb5792a9da8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-15 05:31:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/065bfe95dcf94adba4af4d2125078ef8 : SUCCESS in 9m 11s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/8603b1cb9d8241f4bf0be6723e18b659 : SUCCESS in 17m 17s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/944a666df70f4b2d906a69f12fe4373f : SUCCESS in 16m 40s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/ccd75e43996f405d92003ae53d7d07e8 : SUCCESS in 21m 50s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/09461027bfdc478d9e2300e62a0e6555 : SUCCESS in 16m 47s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/ff345396e56041ad942f90660982d185 : SUCCESS in 18m 03s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/f2bc7831fb094393b4a5de6ea26866e3 : SUCCESS in 23m 19s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/9bab737d4d3f473580e74f0c411a5c4f : SUCCESS in 23m 01s (non-voting)\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/1803ddbbe66048cfa1a748af0e5c1c20 : SUCCESS in 38m 16s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/1daef0b6993b4a4bb68ac75c526dc402 : SUCCESS in 36m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/42aca5cf97b444eaae280907649eee31 : SUCCESS in 7m 38s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/135b3a11b89a49d4aa9c7b33bda305f6 : SUCCESS in 27m 35s","accounts_in_message":[],"_revision_number":1},{"id":"babb8a3381ab95e45c7f1240f5c80c7ac2ecdd77","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-15 06:54:20.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d0f7de94dee648118761d1a7cef88109 : SUCCESS in 42m 15s (non-voting)\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/59a8ea4eb63d4f5db8dd72bed9629139 : SUCCESS in 33m 40s (non-voting)\n- swift-probetests-centos-8-arm64 https://zuul.opendev.org/t/openstack/build/d20a16aeb3774c109a21b601041542eb : FAILURE in 1h 55m 12s (non-voting)\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/27a91c29a6a6438898c2c46511aaae65 : SUCCESS in 34m 56s (non-voting)\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/6cdeac3975aa4dd3b1ee85aa81977741 : SUCCESS in 34m 45s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8b239de835cafaf03c2378245aee92a9a288bed7","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2021-09-15 07:26:47.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nSeems sane to me.","accounts_in_message":[],"_revision_number":1},{"id":"df34a2ff1b1db7ff0abbd05a022177c399c48c92","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-15 07:27:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4bd1323a5999de830de0347639f16280abf03b5d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-15 09:59:37.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/39e0da0f5f6d46b4aea6f6f14f712311 : SUCCESS in 14m 14s\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/a976d36dbd6b499a991a9728cae14e92 : SUCCESS in 23m 13s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/2c7d0f05883c42a1ae6cfcabca8786eb : SUCCESS in 24m 44s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/300ed710645c4ac8bc5e3aa998118877 : SUCCESS in 22m 34s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/b51f8f955d714f0fbadaa577ff1c4dd7 : SUCCESS in 18m 09s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/99b97124d8e54d9b81a2a1c819e366f6 : SUCCESS in 21m 34s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/fed01c14b4374f8caed554f4982d064b : SUCCESS in 22m 49s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/1c96f47ff595421294512423e88027ff : SUCCESS in 20m 54s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/e278302d3ab14c6e9124c692794d93c7 : SUCCESS in 18m 47s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/69edd0f97bbf48aaa1e55e025807de1b : SUCCESS in 24m 17s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/792b3d0ed4134bc59048f9ba70303c7f : SUCCESS in 22m 28s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/4743400daa3f48f0a3ba2d3564f8428c : SUCCESS in 20m 58s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/ec329a9eec1c46c2aa390edc9e6ba10b : SUCCESS in 8m 01s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f72d67b073e547a0bde6420a7270b38a : SUCCESS in 48m 48s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a29a96beae3a4c92915b531927901437 : SUCCESS in 41m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9ce05bb1d8c8486c8b42ad6c540ba856 : SUCCESS in 7m 22s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6cfc356d01194144a8351334abc52811 : SUCCESS in 31m 39s","accounts_in_message":[],"_revision_number":1},{"id":"a1472c960b57fe32d4e7928c5c15c1808aa11f53","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-15 09:59:37.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"517f14c3b58a17941a39f9f36d62eef02346c4f3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-15 10:01:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/170619c1302a44fbb26c59be9291a453 : SUCCESS in 1m 14s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/35aa9afffa1a465bbca3c6704a724c2e : FAILURE in 59s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/cc66e9a6c2b44015b1c4f15627b339a8 : FAILURE in 52s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b4e532a46f27b8e757206d64b1238618eea6bdf6","revisions":{"b4e532a46f27b8e757206d64b1238618eea6bdf6":{"kind":"REWORK","_number":1,"created":"2021-09-15 04:51:20.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/14/809114/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/14/809114/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/809114/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/809114/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/809114/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/14/809114/1"}}},"commit":{"parents":[{"commit":"5d52afbe4c968cbd9094df207bc9072a1b201250","subject":"Merge \"s3api: Allow CORS preflights for pre-signed URLs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5d52afbe4c968cbd9094df207bc9072a1b201250"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-09-15 04:49:03.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-09-15 04:49:32.000000000","tz":-420},"subject":"func test improvements","message":"func test improvements\n\nNot all v1 auth systems use an acct:user format; s3api tests should not require it.\n\nBe a little more tolerant of listing consistency issues when resetting.\n\nTolerate s3api /info results returning strings instead of ints.\n\nRelated-Change: I4a46bd650a53f88c642d402e697869df28bd2fd3\nChange-Id: I8f2f247dd113ad637b17d241133b14c35cadecae\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b4e532a46f27b8e757206d64b1238618eea6bdf6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b4e532a46f27b8e757206d64b1238618eea6bdf6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
