)]}'
{"id":"openstack%2Fswift~810825","triplet_id":"openstack%2Fswift~stable%2Fxena~I82e94c86a7ef78627567297e80f71b176c633f69","project":"openstack/swift","branch":"stable/xena","topic":"create-xena","hashtags":[],"change_id":"I82e94c86a7ef78627567297e80f71b176c633f69","subject":"Update .gitreview for stable/xena","status":"MERGED","created":"2021-09-24 07:23:48.000000000","updated":"2021-09-30 02:52:55.000000000","submitted":"2021-09-30 02:51:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"810825-create-xena","meta_rev_id":"8c05d4272c7a31a850e617fcef0ee34d46d43663","_number":810825,"virtual_id_number":810825,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-30 02:51:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-09-30 02:13:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-09-30 02:13:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-24 07:50:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-30 02:13:14.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"802401d9590181db6bf614b55f5d2f1989149366","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2021-09-24 07:23:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0ab8d8e122560975fc7ea21bddfc3af07e3d661f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-24 07:50:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c25134e5ce0e4c69849127c3443462dc : SUCCESS in 10m 24s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/fc51e976b240418bbd7ca1dd87e122f8 : FAILURE in 8m 24s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/93a1674ff85540299728dbda678d4500 : FAILURE in 9m 58s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/dcec5fe146884277a77ffed436dee358 : SUCCESS in 16m 42s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/41f198898867486c85e924d711452a1f : SUCCESS in 21m 41s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/67e0cc5523744378aac0ded1843f538f : SUCCESS in 20m 19s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/f4f40303f5bc4afca99ee122d1712ee7 : SUCCESS in 22m 26s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/55bd3940e3a34f3eba625e852e0eeae0 : SUCCESS in 20m 25s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1d79514178214f20815cbc415daa4be2 : SUCCESS in 22m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ab1297ecf2104380a51605fe6e515d58 : SUCCESS in 9m 07s","accounts_in_message":[],"_revision_number":1},{"id":"90a64b82d19f038b63036b5ea6c452920936393e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-24 09:29:02.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9a6e9613e2a546da9b19a546aa1f3b1c : SUCCESS in 50m 26s (non-voting)\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c526836a81b1432d9a43de03b5a2ac67 : SUCCESS in 43m 38s (non-voting)\n- swift-probetests-centos-8-arm64 https://zuul.opendev.org/t/openstack/build/21509c8599524f2fa23b2bd6862ff348 : FAILURE in 1h 59m 55s (non-voting)\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/f504a30a5d454809bce0291b657045f1 : SUCCESS in 57m 33s (non-voting)\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/c544954c2a514b57860a870f96e48707 : SUCCESS in 58m 36s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4cc17cd627bee7337681cfcbe2219ed4548eeeef","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-09-30 02:13:14.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"fc700f9c7f7889b0e8af2fe2a5d978f3d0dad013","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-30 02:13:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"25135bad724a1768df3bf0478c80149b15ed31dd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-30 02:51:25.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ba06797f88f247e6a6a9f9d3a28a5fdb : SUCCESS in 10m 38s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/45375b6327ce460ca425cb01a039b24d : POST_FAILURE in 9m 58s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/5946573a979845429f702d6e71c08bed : POST_FAILURE in 9m 23s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/2cd39612675d47feb7b0951fa543fff0 : SUCCESS in 20m 58s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/1a50bad9ab7c4706a216d0b7951a2b01 : SUCCESS in 17m 30s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/818e4f8b3bbc4a28bf5bfe1e73c038dc : SUCCESS in 30m 23s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/ca7290a9cd3641bb9eace307f7365592 : SUCCESS in 22m 06s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/1f2848fb57304e088f5966653d0dfcd8 : SUCCESS in 24m 24s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/f7851b94604446679176fb6df59ac9b6 : SUCCESS in 21m 42s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/a4e29f0888ae498d9ec256e1dbd1a1b8 : SUCCESS in 28m 27s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/19b455e8afcf43aa9dc2b40444e61fed : SUCCESS in 22m 50s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/90afc9b0d39340ba982c8183056167a2 : SUCCESS in 26m 05s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/805d34c916f94f77a1398dbb87c83338 : SUCCESS in 19m 13s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/9d32809b512b48fe9124c948b9a27697 : SUCCESS in 34m 02s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/05ddbcf32e9b4f8499400f5c2b458b93 : SUCCESS in 7m 11s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2e64213c1f1c4db6b99121bdea6af314 : SUCCESS in 17m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8d30a0e4c0344cd5b84bbb4f1afe5048 : SUCCESS in 6m 19s","accounts_in_message":[],"_revision_number":1},{"id":"c985b2e87b4377366e5af67c61ba357f21198399","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-30 02:51:25.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"8c05d4272c7a31a850e617fcef0ee34d46d43663","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-30 02:52:55.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f690f1bff6f44148a2ed8c969018a1ae : SUCCESS in 1m 00s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/53b21564ad54473c96a5861892dbad97 : FAILURE in 51s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/0059bb4ecbbe44dba1825a063883d9e8 : FAILURE in 51s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"29c314ea7ee63e66522dbaa6424c07bf02fbc803","revisions":{"29c314ea7ee63e66522dbaa6424c07bf02fbc803":{"kind":"REWORK","_number":1,"created":"2021-09-24 07:23:48.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/25/810825/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/25/810825/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/25/810825/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/25/810825/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/25/810825/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/25/810825/1"}}},"commit":{"parents":[{"commit":"a8f15128639f7e62a3e8d51c607b45c496206191","subject":"AUTHORS/CHANGELOG for 2.28.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a8f15128639f7e62a3e8d51c607b45c496206191"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2021-09-24 07:23:47.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2021-09-24 07:23:47.000000000","tz":0},"subject":"Update .gitreview for stable/xena","message":"Update .gitreview for stable/xena\n\nChange-Id: I82e94c86a7ef78627567297e80f71b176c633f69\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/29c314ea7ee63e66522dbaa6424c07bf02fbc803"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/29c314ea7ee63e66522dbaa6424c07bf02fbc803"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
